)]}'
{"id":"openstack%2Fkayobe~976214","triplet_id":"openstack%2Fkayobe~master~I122168270505d1d804fe3463426a2e74d8aa5011","project":"openstack/kayobe","branch":"master","topic":"fix/setuptools_pkg_resources","attention_set":{},"removed_from_attention_set":{"38626":{"account":{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"},"last_update":"2026-02-19 08:59:10.000000000","reason":"\u003cGERRIT_ACCOUNT_38626\u003e replied on the change","reason_account":{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"}}},"hashtags":[],"change_id":"I122168270505d1d804fe3463426a2e74d8aa5011","subject":"Workaround setuptools 82 removed pkg_resources","status":"ABANDONED","created":"2026-02-10 09:48:33.000000000","updated":"2026-02-19 09:00:02.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"85580e8af521eb5c1849d4f332e8c7e9e1519c03","_number":976214,"virtual_id_number":976214,"owner":{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-12 12:39:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 09:53:12.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-02-10 10:14:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2026-02-10 11:10:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-10 11:27:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-12 11:01:52.000000000","updated_by":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"reviewer":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"state":"CC"},{"updated":"2026-02-12 11:14:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-02-12 14:17:56.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"837357f6d05876bd2118e03fef77a157979597cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"},"date":"2026-02-10 09:48:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6817cb65473b17bf2c7b530fc347a562e977cba","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-10 09:53:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5594572cb7f461fd8982c491d15bd9d5d4daea0","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 11:10:04.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/447fef46c1fc4d7bba12d4d5c9b9942c\n\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/136be95863004f29bc7cbac72a59353c : SUCCESS in 9m 14s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/ab0ef1cd28934b16abb1dc9a539bb1eb : SUCCESS in 15m 56s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/ca586bf05f29458eb8f4624272d617c9 : SUCCESS in 13m 05s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/19fa6cbcf54e4691bf73500189ab5e18 : SUCCESS in 22m 25s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/50ee6cc1e8e6420696b5f8873017e25d : SUCCESS in 50m 12s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/b0f35f1bbcee45dc96f24f1139f3a0b0 : SUCCESS in 11m 11s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/3f06383774ee46b487f90ed5ecf9b8c4 : SUCCESS in 50m 08s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/ab84a33ab20a47598e4846fa7659bdc2 : SUCCESS in 9m 14s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/d0b49551cfef4432bca7f23dc3679ba1 : FAILURE in 9m 38s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/85206430046c443f918d99d7c353c739 : FAILURE in 9m 22s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/58c75460a80544f287c67e61dbca263a : FAILURE in 9m 38s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e23b5af4567e4613b776ecaa9c1be589 : SUCCESS in 1h 12m 07s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d2c5ffa8d1054eef963927831f60c0d4 : FAILURE in 19m 28s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/17287bf1d685480c9b75bc3233b014f3 : SUCCESS in 18m 47s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/a3b8a0fd1575497a8354a9219bb6f47d : SUCCESS in 11m 58s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/96bc536d363047a59a2b8315f69232cd : SUCCESS in 8m 28s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/c8ebbb77cb2847c8828d0500be4706c2 : SUCCESS in 11m 56s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/7e7aa216d2324a3cabd5c68195002f2f : SUCCESS in 9m 19s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/ebbce8f0a5ea4af6b8f46b7c681bf8e9 : SUCCESS in 22m 01s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/c1cf0bfaaaf64475891d3b644401cd45 : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bb62cd5bfeb549312b5955a74f5332d353dc6fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 11:27:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42df4291832d47c195940837c317beb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46dd9d043fae41efab0012db4cb8e746 : SUCCESS in 5m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/716d0ee688fd4892a4cebd44a349b71a : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c24bdcd5fdc4466da956c92b3c83b4bf : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2746b84ab33d4f73b00762272188ccf8 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abe6440f7a914413ab08d15c470ff8f7 : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dfad7ab6e87846adbe1c3532ae9bb8b7 : SUCCESS in 2m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17d6765e80004594a72c637e8659ce62 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e6dacefe9ae24d80acf83047b370cbc3 : SUCCESS in 10m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4dd6bfdf03fe44b086e1ea15096be1e8 : SUCCESS in 13m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df889630e0cd43d8858eed3e0c37b697 : SUCCESS in 12m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b09efe1f87d4450d9902b3763162123d : SUCCESS in 15m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da030af946304a2ebc0db43d2219dce3 : SUCCESS in 13m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6670c39031764e6fa2e35bfb418d6e23 : SUCCESS in 52m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/38c722f1efac4a0c8179016c86ebb3a2 : SUCCESS in 53m 30s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7bd50034993f488a9b592eab61680969 : SUCCESS in 56m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98acaa1ef9db420f933653a46ba8db78 : SUCCESS in 54m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6fda1c6b0e694273accc4c9baeb65186 : SUCCESS in 54m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b1bd88f7885a4a3cb18c3c62ee7ace1f : SUCCESS in 1h 38m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c712b3607b4a4dc1b2492f311c7846fc : SUCCESS in 1h 27m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4007fa31b97a47f6b6c03b7aedd2ad92 : SUCCESS in 1h 22m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1083c5e97fee4a379bf635a2d128b60d : SUCCESS in 1h 35m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/010f3bf1ef1443c69884d842b69d7c72 : SUCCESS in 17m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/baf31eb108f542a2b610d25aa5e2d393 : SUCCESS in 9m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34f8fdb7ac464febbe69e6ae8aacde3a : SUCCESS in 16m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fb9f56674ad74723b39b49261e0d7368 : SUCCESS in 15m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5590955743fd40a18480b96b3756b6a6 : SUCCESS in 29m 26s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6bae279dd2c6490999d87fbbbfcdf87d : SUCCESS in 16m 59s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a54a53941c9444eb2c1624e12f0de79 : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d99fd9ce03f24dac9813b17692a72f57 : SUCCESS in 24m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/32059a077ff24d01abf5d9bf0e96e794 : SUCCESS in 8m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dae9d8657e33475195a6ffc5f7ee2f92 : SUCCESS in 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4fc80d49dc9ae01c7d7b6431f6ae436c380bae0d","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2026-02-12 11:01:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ed102ebc865085e0ec4c229b67077fef08c517d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"},"date":"2026-02-12 11:03:30.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dad6f64a2f939037cf92edd8130df8a22e19901c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-12 11:14:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b10c07b7d1e805ea5e6fdbbcce7daaf886970a7f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 12:19:43.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/942f276ecb874e96914c0e2b3b393846\n\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/d013ca0df73741dfb776b7ce491ec308 : SUCCESS in 7m 28s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/05fecf57dd144b31a4a3d1b09882ad0b : SUCCESS in 15m 38s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/1c84c03ab898455dac34bd7b4e544626 : SUCCESS in 16m 55s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/39504cca75964fa0b12b500f19673472 : FAILURE in 22m 22s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/8dea40e152be4d6b92f4b3fade16d25f : SUCCESS in 48m 47s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/2e00954c3a4e4a6d9bfd90eefad95f90 : SUCCESS in 17m 28s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/aa055b0e9d70459e9541fff657152ca9 : SUCCESS in 52m 22s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/e045d88a57f84f3389a0e57bc342bd14 : SUCCESS in 9m 39s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/6f7faf7b8ad24dc98d7ab6421064548a : FAILURE in 16m 18s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/fd8be722f37244fc83f04ecab425261f : FAILURE in 14m 26s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/c6768b9a7b57426eb1da7cf8a9a25d38 : FAILURE in 14m 09s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff2173e1e3614d83ac776bb3d9fbf641 : SUCCESS in 59m 16s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ac17e2fb6c1448f8ce3980f67ac91f0 : FAILURE in 14m 11s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/84b56335f38c4642b40f56bcbffe1404 : SUCCESS in 12m 26s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/27dc6be0f0ed41dca8e6d5ef38062b67 : SUCCESS in 11m 09s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/1df0736f93f34e6aa4a03099f5b79593 : SUCCESS in 8m 57s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/0c6a9ab6afe449c8b4f59b607ec5d6ea : SUCCESS in 12m 49s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/ebe5db5bfd5d4284aeda2e8edb7cc4ad : SUCCESS in 8m 43s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/2299be99fc6c4d1db20ac8f04cb29117 : SUCCESS in 19m 00s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/e9b95c8a3ff0493c999e1237c8912424 : SUCCESS in 12m 10s","accounts_in_message":[],"_revision_number":2},{"id":"93b240662935da2e81cb27a700f58610760d2be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 12:39:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d24d4c303aa4076a8575f85a15da854\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a35648bc2f40cc9915568cc6e5289e : SUCCESS in 6m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2a185ece457f4a7e83ffab5fe415b27a : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be093b69d1744a3d8393555ce663ec2c : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/521ebde88adf43c188e49195cef486ad : SUCCESS in 5m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78ce6e7cc87a406aa066a56cea30bfb7 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2a8382b208344297a8edefacd5db23d6 : SUCCESS in 3m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d0cc8036eff04f35bb5d51288c0a7fff : SUCCESS in 13m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f5f93c35a252425eac9544f5d600e3a2 : SUCCESS in 10m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aada92e5137344e18d15ea2bec4c2a49 : SUCCESS in 13m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6edebc82bb4e4e51bb08d879ff18c944 : SUCCESS in 10m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/610f66b951ac4b4797185d0912d7e4de : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7aa475e70c3147bb871d197a7a12beca : SUCCESS in 16m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1dc1a03d87c6432ba39899ff35fede0f : SUCCESS in 55m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/aeb0626786d8447ba77a6c8001bb22d0 : SUCCESS in 52m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2981507a787d4e78ab4ecc79c2976bcf : SUCCESS in 59m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b149a1fbef24fa681efc0eafd4b746e : SUCCESS in 58m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3f9e65d0374408da4def19c83bb486c : SUCCESS in 57m 28s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5a996f5b530f435083e0cc38ab84f8bc : SUCCESS in 1h 32m 01s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e43565198eca4aa79e1c870d4f87e523 : SUCCESS in 1h 27m 29s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ab0077e845249538767c39a89d564f4 : SUCCESS in 1h 25m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dccb3dee5db64d89a0fe2c63ca11a116 : SUCCESS in 1h 34m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/03b3d3433116463f84b1095af430f263 : SUCCESS in 22m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/86b26cbd15f94a228dff9f53bc1e0c55 : SUCCESS in 21m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92958f84a59c4ddf8f41d0ca5e275f7f : SUCCESS in 19m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/71c8db5db10543a69831b1ea01b9f4e4 : SUCCESS in 17m 49s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/54f9f0c6066245d9826b29bd31a97f2b : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5d8bffe503254607a4e6a59112c48998 : SUCCESS in 37m 48s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19b9fe3917724dafaf8d1c897a8f2c00 : SUCCESS in 28m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3def59f235d4a6eb392c71fb971878c : SUCCESS in 29m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a20317f2df0d4e0293ebd0e341cc9e9d : SUCCESS in 9m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/965c024ccfa24ce6bc92ad74ac87643b : SUCCESS in 15m 14s","accounts_in_message":[],"_revision_number":2},{"id":"b35f12bf0a0872322525deb6dd35e54156dfa5f6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-12 14:17:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21bf30c929c4ee2cb263b4233a6a4c126adc17fa","author":{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"},"date":"2026-02-19 08:59:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"85580e8af521eb5c1849d4f332e8c7e9e1519c03","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"},"date":"2026-02-19 09:00:02.000000000","message":"Abandoned\n\nIt\u0027s better to bump the pbr package to 7.0.x","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d2e5e5fadd823a121c6e5c4a15f17cb4c2821946","revisions":{"da24a5bdfaf5838613c256489bdcae704af5e321":{"kind":"REWORK","_number":1,"created":"2026-02-10 09:48:33.000000000","uploader":{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"},"ref":"refs/changes/14/976214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/14/976214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/976214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/976214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/976214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/14/976214/1"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Eric Le Lay","email":"eric@stackhpc.com","date":"2026-02-10 09:41:47.000000000","tz":60},"committer":{"name":"Eric Le Lay","email":"eric@stackhpc.com","date":"2026-02-10 09:48:21.000000000","tz":60},"subject":"Workaround setuptool 82 removed pkg_resources","message":"Workaround setuptool 82 removed pkg_resources\n\nsetuptools v82.0.0 removed pkg_resources, causing\nkolla-build to fail with this stacktrace:\n\nTraceback (most recent call last):\n  File \"/bin/kolla-build\", line 3, in \u003cmodule\u003e\n    from kolla.cmd.build import main\n  File \"/lib/python3.12/site-packages/kolla/cmd/build.py\", line 28, in \u003cmodule\u003e\n    from kolla.image import build  # noqa\n    ^^^^^^^^^^^^^^^^^^^^^^^^^^^^^\n  File \"/lib/python3.12/site-packages/kolla/image/build.py\", line 21, in \u003cmodule\u003e\n    from kolla.common import config as common_config\n  File \"/lib/python3.12/site-packages/kolla/common/config.py\", line 21, in \u003cmodule\u003e\n    from kolla.version import version_info as version\n  File \"/lib/python3.12/site-packages/kolla/version.py\", line 14, in \u003cmodule\u003e\n    import pbr.git\n  File \"/lib/python3.12/site-packages/pbr/git.py\", line 28, in \u003cmodule\u003e\n    import pkg_resources\nModuleNotFoundError: No module named \u0027pkg_resources\u0027\n\nChange-Id: I122168270505d1d804fe3463426a2e74d8aa5011\nSigned-off-by: Eric Le Lay \u003ceric@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da24a5bdfaf5838613c256489bdcae704af5e321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da24a5bdfaf5838613c256489bdcae704af5e321"}]},"branch":"refs/heads/master"},"d2e5e5fadd823a121c6e5c4a15f17cb4c2821946":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-12 11:03:30.000000000","uploader":{"_account_id":38626,"name":"Eric Le Lay","display_name":"Eric Le Lay","email":"eric@stackhpc.com","username":"elelaysh","status":"Proud to work for StackHPC"},"ref":"refs/changes/14/976214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/14/976214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/976214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/976214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/976214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/14/976214/2"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Eric Le Lay","email":"eric@stackhpc.com","date":"2026-02-10 09:41:47.000000000","tz":60},"committer":{"name":"Eric Le Lay","email":"eric@stackhpc.com","date":"2026-02-12 11:03:06.000000000","tz":0},"subject":"Workaround setuptools 82 removed pkg_resources","message":"Workaround setuptools 82 removed pkg_resources\n\nsetuptools v82.0.0 removed pkg_resources, causing\nkolla-build to fail with this stacktrace:\n\nTraceback (most recent call last):\n  File \"/bin/kolla-build\", line 3, in \u003cmodule\u003e\n    from kolla.cmd.build import main\n  File \"/lib/python3.12/site-packages/kolla/cmd/build.py\", line 28, in \u003cmodule\u003e\n    from kolla.image import build  # noqa\n    ^^^^^^^^^^^^^^^^^^^^^^^^^^^^^\n  File \"/lib/python3.12/site-packages/kolla/image/build.py\", line 21, in \u003cmodule\u003e\n    from kolla.common import config as common_config\n  File \"/lib/python3.12/site-packages/kolla/common/config.py\", line 21, in \u003cmodule\u003e\n    from kolla.version import version_info as version\n  File \"/lib/python3.12/site-packages/kolla/version.py\", line 14, in \u003cmodule\u003e\n    import pbr.git\n  File \"/lib/python3.12/site-packages/pbr/git.py\", line 28, in \u003cmodule\u003e\n    import pkg_resources\nModuleNotFoundError: No module named \u0027pkg_resources\u0027\n\nChange-Id: I122168270505d1d804fe3463426a2e74d8aa5011\nSigned-off-by: Eric Le Lay \u003ceric@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2e5e5fadd823a121c6e5c4a15f17cb4c2821946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2e5e5fadd823a121c6e5c4a15f17cb4c2821946"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
