)]}'
{"id":"openstack%2Fkayobe~976524","triplet_id":"openstack%2Fkayobe~master~Idac833e94e566e169460980dbe58b4aeac520765","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-02-16 09:25:55.000000000","reason":"Change was marked work in progress"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-02-16 09:25:55.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Idac833e94e566e169460980dbe58b4aeac520765","subject":"Force refresh of facts on bootstrap","status":"NEW","created":"2026-02-11 18:04:07.000000000","updated":"2026-02-16 09:25:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"cec0a9bcdd52c4a1d21d7bf147b701686f7a6a78","_number":976524,"virtual_id_number":976524,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-11 19:45:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-11 18:20:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-11 19:45:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02ed2fdb21f7205e4f61df943814533be02f2584","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 18:04:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9e7e9580711299cc8966f1cb6079bb39468ea43","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 18:07:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c76befb3c0923d93d4a8346d691d1e7470a40a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 19:45:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37062574c1ab44feb868c070a4529456\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0470ca0e75e849138fbebfd3469c2c77 : SUCCESS in 8m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/38bfc8b2c7e14cbd89781ccb2242fc7f : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00203001e804458e8c377238a14b31d4 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e75fd821f4445d089b10e0ca906cadc : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/645d763912a34d7c8cbe74cead5d45ec : SUCCESS in 4m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b081ca5547da487e9e4b8750bd9fc76f : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/625e27302b2e40d187a09f8f8e5cacaa : SUCCESS in 12m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a780a2b89f2749ed98a1149e521fe4c7 : SUCCESS in 13m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8cb5b3d1ffa54d1a94dde10aca6b77aa : SUCCESS in 20m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1c0b370b0cd4250bf5a7fc6a5611c96 : SUCCESS in 13m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4d73f6c6fef44b21834861f65f21bbd6 : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b26b5c204bcf40928a4f789275a37bc8 : SUCCESS in 16m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1ac0908451a64810a768048163166ae1 : SUCCESS in 59m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/393acb5fc62142878465f98ae8b029a7 : SUCCESS in 58m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/79c3a9af4bf54e2e8e5399157c38fc66 : SUCCESS in 1h 03m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/759b8169805d436cb9d501b701c5eff5 : SUCCESS in 1h 03m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3431b8957c3140baa731b4ca5045a91e : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ba08cff7db5c4c4d9b0787b782a50a4e : SUCCESS in 1h 38m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/791f6268e9b443d7b5d888093225727b : FAILURE in 50m 41s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16402ca0ce65497fbbe87da7e40b331f : FAILURE in 50m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/051edf4e2f15437494e4cf9e0d8b658e : SUCCESS in 1h 38m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f713f317672c448a8486bbe47126e76d : SUCCESS in 18m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e9c4a545fec0429ab05cc2c5433c28de : SUCCESS in 13m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5ed8f4cb7574fa0a53813f21a39cff5 : SUCCESS in 16m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bf6a8575e5ae42538d10669043fc8d8f : SUCCESS in 11m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/69675f1176e343f4b382df82566edc8e : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6fce386cc43749d6943b3be796f2e4e5 : FAILURE in 22m 19s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/486b9ee88273416298ff0a8fc352aaca : FAILURE in 29m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/468647b655304ef4ace21f5c4068c4a6 : SUCCESS in 35m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3b3b25f285644549bcea1693df8e2608 : SUCCESS in 20m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/370bc6632e20454d824a12d8d631b407 : SUCCESS in 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cec0a9bcdd52c4a1d21d7bf147b701686f7a6a78","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-16 09:25:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3279aa62d694b1c6555cd218d727474e4c13ffa7","revisions":{"3279aa62d694b1c6555cd218d727474e4c13ffa7":{"kind":"REWORK","_number":1,"created":"2026-02-11 18:04:07.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/24/976524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/976524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/976524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/976524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/976524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/976524/1"}}},"commit":{"parents":[{"commit":"0aceb06f6e718efb1a329d924c2c15394eec669d","subject":"Deprecate kolla-tags and kolla-limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0aceb06f6e718efb1a329d924c2c15394eec669d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-11 17:29:23.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-11 18:03:57.000000000","tz":0},"subject":"Force refresh of facts on bootstrap","message":"Force refresh of facts on bootstrap\n\nRecompute localhost facts after a fresh install\n\nChange-Id: Idac833e94e566e169460980dbe58b4aeac520765\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3279aa62d694b1c6555cd218d727474e4c13ffa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3279aa62d694b1c6555cd218d727474e4c13ffa7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
