)]}'
{"id":"openstack%2Fkayobe~976525","triplet_id":"openstack%2Fkayobe~master~Ic6fd254977697ce7be6916fc89952b16c283ff17","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-21 09:13:18.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-21 15:08:47.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-21 15:08:47.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-04-15 15:45:01.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Ic6fd254977697ce7be6916fc89952b16c283ff17","subject":"Support provisioning hosts without kolla services","status":"MERGED","created":"2026-02-11 18:09:25.000000000","updated":"2026-04-21 15:08:47.000000000","submitted":"2026-04-21 15:08:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976525","meta_rev_id":"b189f45a3b50cab825bfe71ef23cd55a9dd35d6d","_number":976525,"virtual_id_number":976525,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 15:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-17 17:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-17 21:45:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-15 15:45:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-17 17:22:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-21 08:57:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-16 15:16:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-11 19:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-27 09:39:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-04-07 10:03:08.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-04-17 21:45:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"91154c3e7909bc850efd538d6c4d5d5499eec486","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 18:09:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dcf054d23b75651b4118118bc18260c4d3ee309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 19:46:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff0e5f2e3c25422185e98d6218cb7708\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19d642eb87a64d6b9444b2ff14f7f6ef : SUCCESS in 9m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e7f023f1ad4c4d84be2692b529ced1ed : SUCCESS in 7m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a676e3d759db43fdbe97889b29444895 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b03905b9ec42cea669cfdee974b0f7 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0000f6203ead4f28b980f454e8c7d3df : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d07cfe25b78043edb6d2939c72f3e760 : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/faf59a74219a424abeb1d4ad05fc8d80 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d6c9cdaaaaa48d68b0e808e6e17206e : SUCCESS in 11m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cd073a89a63d4ec8be37ad9f8e4cbaac : SUCCESS in 18m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11dad2cc421349e1b4018f628f052602 : SUCCESS in 13m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/42d7dd22e18445c0961e08abdd673794 : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8e2404babfe4d28980b16198d731f0e : SUCCESS in 15m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/10d8d0c32afc41d2a6a08b21b5cc3868 : FAILURE in 10m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6d2d9ef9cb124613b09c457e66045e1a : FAILURE in 16m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3fa54ff30c9940d8ab3e130a6bb3a15b : FAILURE in 11m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3dd0d995e954e6e847f9ddc5526523e : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3aef266bbe5e4b2e8bb4f461ee7eb697 : SUCCESS in 58m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/60a4518528c94e07a98384dd1bc5d422 : SUCCESS in 1h 30m 58s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a9cb114cd83640448031a52455a33f8d : SUCCESS in 1h 35m 02s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e969aaeaaf494017853ff70727a0ae14 : SUCCESS in 1h 29m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b526f13a673b417b88f9c34a8b627a94 : SUCCESS in 1h 35m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5c4fcc21acf54431b22d8ea5e801379b : SUCCESS in 22m 00s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8b4f703380654da4b907c92941406b63 : SUCCESS in 16m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6ea3aa1de094d65b4d665095dc2e7ef : SUCCESS in 20m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/970ac1a74c3f425b9ce695fa5da2bd50 : SUCCESS in 15m 50s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/267bc6d94dc94f929f5014d3722d0505 : SUCCESS in 18m 21s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c8be806c799846d584a18c50163b6965 : SUCCESS in 34m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a6c15cbbd1343448fc59c10b17b6ca3 : SUCCESS in 28m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/671fe4d421fa46558c19f895f1d87c3c : SUCCESS in 27m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/698f381bfac24db0b2cfc8ddb2e281dd : SUCCESS in 19m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dc54ad6c9194bf998573264ca5d6be2 : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":1},{"id":"43cc707a47abff2c5d10d1a8810195afdd3755a4","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-12 12:28:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"22a0dd2b930892648d8698008350282bac6418f3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-02 15:59:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"67c812731841097727117be49b60237e6d53b4e8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-02 16:08:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a686af0717001cd9587ef681f655b8ff47fc6ffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 17:37:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f0aba78336044539eef21202b805ae6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88b35d8d9d2349e5a287d11a36174787 : SUCCESS in 10m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/638ead090d794daf9b5925d32f7ef45d : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3da4d2baf0146a68efc9d7da71b443d : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee235f43d894da5a82c2c45dc96ce3a : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6219e6bdcf454baa93c17458b38e193c : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9f7241f036264ef1a31e7e8cc17d895d : SUCCESS in 7m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/405c5eaf471b4920be349554e16b12dc : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6da77f8cfac3458d9de1597fc1fd9a0a : SUCCESS in 7m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f0db77d9df5a4138911b7b72ff8b905b : FAILURE in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ec68faf544e848a3aafe009665018446 : SUCCESS in 14m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/56cf771a8f2b4020a8ded8fcaaf2a317 : SUCCESS in 20m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64459196072946799c9af6c670de662a : SUCCESS in 6m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/70e43ec9475c4b78be8a2b7aec407a2d : SUCCESS in 12m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1517edf66e04c6aa917e65fb06ccdd8 : SUCCESS in 17m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7a04f7137d1a44cd9857e828207d5af8 : FAILURE in 16m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/be372978dbed4d8092e45e6c6798ab85 : FAILURE in 14m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e1825c601fb24b14b5b595e8c41e199a : FAILURE in 17m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abfe156d174b4a1babad831b8b1cf938 : FAILURE in 13m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b25bd46b379f4f5b957a6469b75086f5 : FAILURE in 14m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/56b39063a380457a92870adc1a2582d9 : FAILURE in 56m 32s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bcb0b38a678a4c32928a5a03b1d15f24 : FAILURE in 51m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7105edb127ab4cf38c19cab795c1627a : FAILURE in 46m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e1e497fafd440699fb7e4edaf5b6396 : FAILURE in 59m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fa24d98742e74dd3862e6ae6f3fba484 : SUCCESS in 10m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c71b24a39c384f409acae7b6790cc79c : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":2},{"id":"0c80a3800d20b7ac2a1a7f70bbd50d4827e3903e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-02 19:06:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"68f3e1e381df617891dc279484fbfb9f81a99a19","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-02 19:16:41.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"77904944d2347639f074831e6b6e9ad0f33522b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-02 19:20:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ba04763e24517ecb92ba6cb22e4a101e99d37d42","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-02 19:21:36.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"a065de4388797c198cf1b52b54b150e5ac09467f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-02 19:23:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"81000379a3660c5150ad7f74e88f88b29339680c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 20:25:18.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ebdd389e4cd45109feb97355c80e542\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/717c2b9e445e4297aad3d06758379f92 : SUCCESS in 8m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/51011a01356c4b2f871d7e767046fae7 : SUCCESS in 5m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4d2a324220f4c52a432be696edbf423 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf874e1442e40bc80a47435adca5fbf : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87ada577094d4a5fb8734d21c8adade6 : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/df9bff5ec5e7483b81540989c8ba1fa3 : SUCCESS in 3m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e75eb72b78f442889667e34b87d0204e : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e011e0f9a67f46e18ba42d8619e43e1c : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e55813e7ab0946d3a54be5951f87d546 : FAILURE in 13m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/26428888bf304f4cafcaddd5c991d6d0 : SUCCESS in 5m 58s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5b9f1e086ab84b53a472c8ba24dc6d97 : SUCCESS in 14m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/437febad99164cd2b9f03b7bc3daec20 : SUCCESS in 15m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dde15706aedb4d18861b6e6e92cdb368 : SUCCESS in 10m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7da4827170ce4665858e1348adc9c079 : SUCCESS in 8m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/50c833403bb44d0fb094935f33e8f2bb : SUCCESS in 56m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2777857524d2483a971eaaebc650eec1 : SUCCESS in 1h 01m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/92bf8d50dc624581b2856d8e0abb281f : SUCCESS in 59m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c783624bc8d4f3a9fe5bece4bd13ffd : SUCCESS in 57m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ca2a2afc96c249ee833c993941ea50e5 : SUCCESS in 56m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/977e4642310a4ee482bb7c3a20d9f043 : FAILURE in 54m 27s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3f882ad7f69541e2ace9ab116214c430 : FAILURE in 52m 13s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b69f0fab0704088829d8228b3a1fef6 : FAILURE in 50m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d80dc87063fb495893cf494ff323b8a6 : FAILURE in 53m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c2143f3bbcfe4aa3989f8b88543eb804 : SUCCESS in 10m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25743b2810ae4602ac38517d7171d5f8 : SUCCESS in 17m 47s","accounts_in_message":[],"_revision_number":7},{"id":"567ce49f107f762f2a9c922f3890067ace33e21f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-04 15:59:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b17db02838f8f61041422c4223bc820868f41ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 17:59:21.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d46b662051b4ebab4e7335094cca12b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b85190baf3f14072bbaf2b1547aff2cc : SUCCESS in 9m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/626eca5a5e2c48159e1d04d673992eff : SUCCESS in 7m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cb61797c5ad464aa6c1f519aa58669b : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b2e156be9f1459a9da29d90b5a5f066 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6feb97d2b824293a4bb810454d4455f : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7cf6f4cd5f604a48937c3705cc454b81 : SUCCESS in 7m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b471af7853e4c0ab89bfc8b683e58d8 : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30a3b08a745148c29e6e3e9fc838ae8f : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0facacbe1b7e49e4acf6266af1e23566 : FAILURE in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b0a2fee01e2c49c18002bfaccb6073e8 : SUCCESS in 11m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3bf1e61a1cc44b2fb26c3967fec64b54 : SUCCESS in 16m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26fa002c0001405fa8f5ebc3e4ddd1a6 : SUCCESS in 15m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/451d95c2386242dfa2a25b63f77f767e : SUCCESS in 12m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94149df4a31a4058a062e185b0e08be2 : SUCCESS in 17m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/12fa46572ccc4d468ec1b9e427264693 : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8daf52ef6e24432db2474997dc016279 : SUCCESS in 1h 08m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3d70468cb3a441dea3a701d9c3a09109 : SUCCESS in 58m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dfdde5b74bb4ff7a43fd29737b581f3 : SUCCESS in 1h 11m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5e1771d681874708908f45ac4fc2a198 : SUCCESS in 1h 07m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2cfc348ab862401d861267fb88dae8c4 : SUCCESS in 1h 38m 51s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dda92ffbd4f947eda54bc9aa1ad22603 : FAILURE in 39m 34s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/998a511981cf4e89af466f36050b86ab : SUCCESS in 1h 46m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecfd6858525d4e37b4d9bfebd956561b : SUCCESS in 1h 56m 40s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b12a49c20d614c44a9bd48bb195e0659 : SUCCESS in 17m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d87f7421b3c8405b8a47b0f6621f729c : SUCCESS in 22m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/296ffd23248744f4acccfbfe049f400a : SUCCESS in 19m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/762e249b44644812b1c181e0dcd4e23f : SUCCESS in 17m 54s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e1dbe4235d42476886128438115f3139 : SUCCESS in 18m 46s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3c55245a62c94c71b79d789b658c4bbb : SUCCESS in 31m 15s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/046937c4c8f6418681f57f85df6de510 : SUCCESS in 31m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85c91098254040eebbde4386a33965d8 : SUCCESS in 30m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c31fa840d57a417daa8d69a9e1a2929f : SUCCESS in 15m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b148a780ec584ac39f550780af814254 : SUCCESS in 14m 34s","accounts_in_message":[],"_revision_number":7},{"id":"e62e11668b63111c6561f82a7bf653b640d3551c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 19:46:42.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2ecb9aa51c047d18aed8d5cb490c62e4fe86f4f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 19:47:28.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9a1e4deae37ffa82b1d4f302e774b11d3a5a3d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 21:21:49.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/946b10ce302949ca865559f4455f5489\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047f4454517d496e95e4d0db5c09199e : SUCCESS in 9m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/97d3cf17171540eda9819225d83e4f21 : SUCCESS in 7m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/247d343ec7fd44499a29b510d7f61bf3 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fbdfb5e00ad431daa6f85aacd02df13 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3479fa4298e14f0f98b8f2f4b51a331d : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/29e798ebc3a74815be82c8fd2c26cc11 : SUCCESS in 6m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73b2812519d1462e9575f8e9a8e7be9c : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e041f82366124526a34bf363bb2b9cd1 : SUCCESS in 7m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae2d8ff304c6490aa9b18be94950689b : SUCCESS in 5m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3bdb7a18e9f1415caaa5fd6427c306c1 : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e34d0ab913fe405cbd5f2802d7d682d5 : SUCCESS in 14m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba5cf8ca036849ee9fe9711b30f5b156 : SUCCESS in 13m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/036ed3ae2aaa4502b769f9302dc27a22 : SUCCESS in 17m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f349ca22e2ce4a79a079c1de626f68fa : SUCCESS in 8m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/033dcced8056433698e0476d15deabf6 : SUCCESS in 56m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/59388b86032446c1a8313222fd7dda4d : SUCCESS in 59m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7f4136a3cc6a476fbe4178e8ae532aa5 : SUCCESS in 58m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a81a569dfe24543b96ff8f59e9cb493 : SUCCESS in 57m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/747abf79faf64bc8b3b13c026495124d : SUCCESS in 56m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f6cff77f24ab48ccbb85e691e60869c1 : SUCCESS in 1h 33m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c933a93c840542ae8104688639a6bc22 : SUCCESS in 1h 27m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ea74483229f4616bd0cdd8663bdc09d : SUCCESS in 1h 27m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3b6adaa03344e4787ed22995127bbbb : SUCCESS in 1h 33m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5027549e693c4fdab5bb72e0cbd469ab : SUCCESS in 23m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3ee3fb2831a74c27a86df8641ec2d2b6 : SUCCESS in 23m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/688abc0782ff45acbc2e54d715bdd7d6 : SUCCESS in 21m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/243f5bf06f4740fca8705464eb136262 : SUCCESS in 9m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b4be6b8846142df8b2b01bbdca2ae42 : FAILURE in 8m 18s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/954b70e573d049368c0af7c7489e0763 : SUCCESS in 16m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64c4de845e214a69bb409f8d26e68533 : SUCCESS in 24m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c52de4882634b63b9368d0cea9d70a4 : SUCCESS in 29m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/498ce01cf88744cba46cb50e0447e8de : SUCCESS in 15m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/330b54b20a914d209c6c75017adbe19c : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":9},{"id":"e67870ccdff91bddb2c05f159a34ed59aa5b574a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 21:24:22.000000000","message":"Topic ukaea/2025.1 removed","accounts_in_message":[],"_revision_number":9},{"id":"9a340e1ad813529b77e41487e13d5137bbed67d2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-23 10:37:22.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":9},{"id":"f215d7c50d3d77aad6e2105f9c851b06d6d1cac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 12:27:13.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60241986e70b4ddcb46045494994c423\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a368124444764ce680861aade3d5fb2f : SUCCESS in 10m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7be4f9a8a6864b5c9da8300f2e04d707 : SUCCESS in 7m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03a71db056f1467687e64aaa87e71b24 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d667c87605c04b169cae770e8ef59ae7 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c6cb151c99c468b9862b910a6b1963c : SUCCESS in 5m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7bb2c7c9c58242fb8d0ae384d6acca6f : SUCCESS in 9m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4a653d5dea64278a53c48d518a20e42 : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c26bcfc069cd4f0a86465e71ba9deca5 : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e570f76b9f3d407c8d954f63a3f5ef78 : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/209b094fd4fb4c91bab401593bf82971 : SUCCESS in 12m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b255a376ccae43f8b6539580e57b8281 : SUCCESS in 9m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99da9fa0668d4605bcd1b2a563958955 : SUCCESS in 14m 18s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/650d2c11e185438992f3ba8a0bec5314 : SUCCESS in 19m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12cb3d3b70a5463cb54e5d15a0d1bfa6 : SUCCESS in 16m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a49a8f2990fa4e56b953bd1c66ebc405 : SUCCESS in 56m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b31d30b9c4784be68f043b4990f26a25 : SUCCESS in 59m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c4de9776a51f4dde8b30c5de283ea2e7 : SUCCESS in 1h 01m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b741bf6bae984b7ea03369ec88f07bca : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f1c3585b914a41f2acd992cfaa99a145 : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/841be421ae62401cbe742a9f5c6b0629 : SUCCESS in 1h 42m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4187532eed0c402e817a8b6f56ba2e7c : SUCCESS in 1h 29m 35s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60b836edb9604c8db789fbfb7ddcad0a : SUCCESS in 1h 33m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c90968809444ff995dabd944701d10d : SUCCESS in 1h 37m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e199aa59bbed4a8cb6b22c19415dee2a : SUCCESS in 18m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ac465a52a75c49cdbdbbcd02c569a123 : SUCCESS in 10m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48f26fc27c65461eb2afa142848b3311 : SUCCESS in 18m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e23239de2c6945c09a39cc942eb8c71a : SUCCESS in 16m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dc4b17a1a3a0467d97b3d89a100d371f : SUCCESS in 16m 21s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/24a972c277b14d0191c5fcb2cd478045 : SUCCESS in 17m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a99632cd5cde4dc287962195c9b4d2ee : SUCCESS in 31m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db1012d9c90a49d38937af1df39e48be : SUCCESS in 28m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c48bad99361443a79df07c6a0c0ffd69 : FAILURE in 14m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5490cd289d9d4e5a9a26a1e5ce24ee7b : SUCCESS in 14m 59s","accounts_in_message":[],"_revision_number":9},{"id":"3566fd91e2e41e6e6bb8fe16d1bd5fe26821be9f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-23 14:01:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b21d03585e2e3ae466787b4f3c4fb7040eda2946","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 15:44:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/132f8339b7e5447c8f3ccfdc3b625027\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f727246e5f4a7394a6f0a6f4427c21 : SUCCESS in 5m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aca91a5c4d7647fd807580a914eab58d : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/017fa6d43394432c84d38759309e5c34 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54810db7d424af3b4ad1d5e594472a8 : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/099de0304cc84ea1875507dcc9817485 : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea6d52b7329441e1854a493b64ca7ab9 : SUCCESS in 4m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c658928a744441fa271242fbc11279e : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/59049d33f5614110b30337e133845301 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/53c76938dc434d8a8f7f8a60123f0dbf : SUCCESS in 13m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25aa3fc9619841588d2cdda126241270 : SUCCESS in 7m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c657be827b3846a98ca724b68a7480e6 : SUCCESS in 19m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d358f4bd61c411ca269e47f36eaef7f : SUCCESS in 15m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ea1ff7ecdaba437dbaf693684f33611b : SUCCESS in 26m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24c3076a6a114f48bacf631b61098b58 : SUCCESS in 9m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0f7d454edb8f4ea0bbb2442ebc0dd7f2 : SUCCESS in 56m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7f8414238da7458bbaa56682514a1f0e : SUCCESS in 59m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1064fa56141c400b92b7b4e16c17824f : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab01c68a924e4fa6a734b50abe1d9676 : SUCCESS in 57m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1290c1b482774577aa7c3a4279c42a95 : SUCCESS in 1h 01m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8ed50712ddf047a4b78ade6aa575bdf8 : SUCCESS in 1h 40m 39s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3ec0bb9777944c6daa928253a67bb12c : SUCCESS in 1h 33m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be200b0444404fa5b579d5bda6e3c269 : SUCCESS in 1h 31m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ea65b8174bd4edda1cafb3481c360f2 : SUCCESS in 1h 41m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0f1ca6f37e9649aa85107ca9194cb814 : SUCCESS in 25m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/51c2767d781c48ac951fd50bb37b0235 : SUCCESS in 23m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f623f2a8a10c40108f5e0d246dbfde07 : SUCCESS in 9m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/69f6e0e1885c43acafad8bc42b6a5fac : SUCCESS in 9m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/02c1e4230e5443e3b77b226710f38ce5 : SUCCESS in 35m 22s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bf8938985a044f81a7edd0d48bf72879 : SUCCESS in 25m 09s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22f1bd32ed0e4b7cb1b5aa2cca687f27 : SUCCESS in 28m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3e06268797f4319bc5705329eea060a : SUCCESS in 29m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/35aa503d180243bc993dd6416a601417 : SUCCESS in 10m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1b12eec9af24cc88f4224f6e072fe8d : SUCCESS in 16m 09s","accounts_in_message":[],"_revision_number":9},{"id":"555388898265a6549509b9bf2168b8c8c637f748","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-26 18:00:28.000000000","message":"Patch Set 9: Review-Priority+1","accounts_in_message":[],"_revision_number":9},{"id":"dc36642963fd62a7b54a2d08b09f112f99eede9a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-27 09:39:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"24e547a69a6d672078c3812829360a3187cc4afe","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-07 10:03:08.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"81fb23eead28f9e95be8df29359104cd3b886ec2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 10:58:46.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"522837659bbda2d2827f3e1b8954ac56cc28bbaa","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 10:59:03.000000000","message":"Patch Set 10: Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"bc92ed909b6e134445aed051da8afa594ad62eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:58:37.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a1f607482604f76902f3e6fce961c6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce460adb11b24e09b69aec287b11ea27 : SUCCESS in 9m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/260d8df799774e0eb3a353a3315f383e : FAILURE in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6eabd65c0dfb402fb6581e398089c5e9 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a09dbe810c4043812068a6b0d3c9d6 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fae7fdec35a64232b5c29dfaa315b862 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7aa1609d967f4b66a5099b45b28a8506 : SUCCESS in 4m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e1eb7cd08cf4f7bac4bdd339965568f : SUCCESS in 2m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/97fdf05a1a204406bcd9b52006ec1aa2 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48b262a31f70481d9f90e81cca857b5f : SUCCESS in 6m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fc38778f4f8a40c5a9b534ef5e3d05c6 : SUCCESS in 8m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/08be6343dd7c4222858fb6d40d3109af : SUCCESS in 16m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/076fa7517db34c17a6c7ea3159b7c10a : SUCCESS in 8m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1cc1d3ca33214a249e539818b61de57c : SUCCESS in 10m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e65a00a7ef924e55861a17fec21d6d61 : SUCCESS in 16m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/eb1a701024f04507aca94f6684cae156 : FAILURE in 43m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/53cbf6fafa8b4c698de834a17d9d29be : FAILURE in 40m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/714ff48b5b53419c9a7309e351832925 : FAILURE in 48m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b64e2dbcd7b446ea7f2dfefe3c2f9e5 : FAILURE in 36m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2a2ed2496e98439c9f7f18dffd99714f : SUCCESS in 58m 53s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3532ce85c225443ebb58d47d8446e3cb : FAILURE in 18m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7b4234bacd86489aacf8a5c26fd43fe5 : FAILURE in 14m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93f21396e10e46b890994c3c5adc47eb : FAILURE in 13m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/990c5b879d5b4265bfdc30cb91663775 : FAILURE in 16m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8ca16830bfc44825b98b152523789053 : FAILURE in 16m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3f470256ca0b4f7d9127068048fa197a : FAILURE in 15m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/614dd34bed364e90b944395bf072b4bc : FAILURE in 10m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1fb12a0dd3e3433fa09d04fe1b3e5228 : FAILURE in 8m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5662dbd5f79a41ca9e2761bf09e6f23b : FAILURE in 17m 19s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4199752ad1974300914b5be9a7b84ec7 : FAILURE in 17m 26s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d339f9d3bef4638b483c31650be48b9 : FAILURE in 16m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a4d4c9136014a31a2467cbad804285a : FAILURE in 15m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b3ce7d9377934d189d95b67c75ec9de7 : SUCCESS in 21m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdc88f7e6988483797778958dd8c7d8f : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":10},{"id":"d035edba53466f439858f18f047f6ff5cce66d67","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 08:13:43.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2563f969ddd94a64637dc7482df86bfa345eeb06","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-14 12:29:55.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"10873b65bd951885ab9bb8a2d64d71683b18e605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 14:08:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f977e62f219c4ce2af26e4a1d2368dd4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6364824082b4fa79d3555ede68225b4 : SUCCESS in 9m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f7139d7652894022a3a9262bb8460e4f : SUCCESS in 7m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97ee07791d374087b71ea7b825406d8e : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c03df6633d8642feb91707b191fcaabd : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f3038dcf6d142eda53a20e9bc795695 : SUCCESS in 4m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/98670b613a9945c99f0887676d377177 : SUCCESS in 7m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/119362299bf14dbfa7179b6b248cd3ba : SUCCESS in 2m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/670d25e5c2074b26a392258e4ff87e53 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/710f92a18a1f448c81acff1eb7c35d32 : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/905d63b72389446489755ebc4a15cc29 : SUCCESS in 14m 40s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bd59aee9052c4d50b20d8b95948d5a4d : SUCCESS in 9m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0aa6588561ca46649e3a633dd6c0df5d : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/60ab37e581014afa82c492dc5ac1bd52 : SUCCESS in 17m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e9f8a92b1274e12b3534c0a76401567 : SUCCESS in 16m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f10e4acdece742a6bce8126446994567 : SUCCESS in 58m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0336764527074ac2bef6d12cdc9e956b : SUCCESS in 1h 02m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/10473112437b4fa2b525c6dca45f2326 : SUCCESS in 1h 00m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcce64ad37ff4a26a9dbc4262a84cae5 : SUCCESS in 1h 00m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5647c22900da4b6892c3c1d1afcb9461 : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e2e1ba43accd4bcebfd915f98c68cd1c : SUCCESS in 1h 36m 22s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/044ef132b47f4357ba3ca16121b06373 : SUCCESS in 1h 32m 56s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0700bd58b3f24741ad998323fee80a91 : SUCCESS in 1h 31m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4967e160339c42439643f54f74a77b67 : SUCCESS in 1h 36m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b7a901f51a984a76a3568f58f56506df : SUCCESS in 22m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e5fd480c81ea4768a93f938a67b809f1 : SUCCESS in 18m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/428c6a661f1448c39c344125f8ceaaba : SUCCESS in 17m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/37a476be9f814462af80c26b1d8d4254 : SUCCESS in 16m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/22876f397f524c8587a5feb3d666e11c : SUCCESS in 36m 29s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a6b606ebc17348cbbaa863e37030ce52 : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8186eca4da50416e899cc610ab803959 : SUCCESS in 33m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d87cd41a417947d3826a8019c24ecc1d : SUCCESS in 28m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5df389fd7d5f40b099c1e0df04e02e35 : SUCCESS in 23m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/789ce4a60ade4083972f8870dc85d327 : SUCCESS in 16m 25s","accounts_in_message":[],"_revision_number":10},{"id":"82fc79e83a8da8c925613be4b572e1cb33afd4e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-15 14:53:55.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"ae72e1f455c0e8396eddb9998481a5f5cc892466","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-15 15:08:10.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":12},{"id":"cdaa805d66c92c89ddae9982a6d00ba787b8206b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-15 15:45:01.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"982dec8868ed12d664ea89aefe099acb5cc5353a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:51:00.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d76261d2671438a856472fc58c3ed03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3dca23f5cdf473da0cc8b239228e664 : SUCCESS in 10m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6ae31d19c812434ca487aa0a929fab5d : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60b2d9f6fb164949a536dc4d1d8fa6d6 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1d24f5ea45b472492ca4a1b395c62b5 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e180b6bc8e374706a63c0ed4e10b0308 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d3af1d44cfe84ec7b73308d3a2ffa972 : SUCCESS in 4m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd8f642ea19e4045ab4877e5a6c01267 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10fe7669daeb4287a13ed022d6538855 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c18416cdcb8f4a45878373fc632a318c : SUCCESS in 12m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bd6b52284e214c9e8f6ac0823dbcb74e : SUCCESS in 13m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b521e8a9d8454eb0aff8ed2f9e0e1d62 : SUCCESS in 25m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaef9d2fabf8419da5981e6376708843 : SUCCESS in 12m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7d9ebd135eae45e19b353875684a1e9e : SUCCESS in 17m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5068fc5727224408a35f4ada89d4aec2 : SUCCESS in 10m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a5146cd611a44b2e857a31075c3a645b : FAILURE in 28m 46s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/75a848ae26ff4857a46675b6f21ed3f5 : FAILURE in 24m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a77eecbe9b3e416dad88b0e734e4b456 : FAILURE in 38m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94adb0f89b1340979b1aa95cd374971e : FAILURE in 25m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3055c619bf124c41930062af3934f9d4 : FAILURE in 26m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b0bb3655ec8744f7b23647d497404e3a : FAILURE in 19m 07s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4da3eba4c9ff4e06a74b28a815d78dc9 : FAILURE in 20m 07s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b1756615c3542e8816195a60dd8a622 : FAILURE in 21m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39700f87859946d1bc92420fd60dbdd8 : FAILURE in 19m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/57b570e6cdea40b18a17be5a46ddf9b5 : SUCCESS in 19m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c5719b60b59a43c98c105c940d0e31fe : SUCCESS in 19m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cb677432aa64a349fa017eddef7ecf3 : SUCCESS in 13m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b6f59b7ee49e42f398cf502abace644b : SUCCESS in 20m 54s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/16fbfe3508c94a4c8d1bd0f1f8893793 : SUCCESS in 41m 01s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/88b8d65ad2ba409ca385e1a42af9092f : SUCCESS in 27m 52s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa6c4a186dd7434e8086e7ee03787a40 : SUCCESS in 31m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f77967191486480f9096cd7f1e03bca3 : SUCCESS in 35m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/195fcf8fddfe4480be1c43d629d2e927 : SUCCESS in 23m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6132150e8d174ba98f2239764cc88390 : SUCCESS in 21m 42s","accounts_in_message":[],"_revision_number":12},{"id":"818d74f261d3dc45a5ba3822b000469241977294","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 15:16:42.000000000","message":"Patch Set 12: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"911e2b6f34ef4371116ef016f61d89c36ad3b780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:57:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc28e7a366348c5be824e2e7a25e54a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c8d287657743dab6ecf49f6c9fddd7 : SUCCESS in 9m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/93906687018246cd8251a6d8daa7f84c : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94f27d46b5614434adc7a4e729fa5e03 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c09d33093a4c13b2234b9a5918e692 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53a5edaef66846c08311ef76b76f3292 : SUCCESS in 5m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c334454c286842c5afcfb0ee69265506 : SUCCESS in 6m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d15c15457843431181b9d4512482f990 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/904d94fb9f8940ca81205ae0a0edb60e : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d62e901761b14194b48b5f9a40381830 : SUCCESS in 12m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2cedc539c74943e584411b1252a5a271 : SUCCESS in 17m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16a6daf33df94af89cc46f8a79576f4c : SUCCESS in 17m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f7c670e6fb54010986856eb7c528c24 : SUCCESS in 19m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7c9c9743ecdb41db96cfdddd9648200c : SUCCESS in 17m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/308c5e0079424958af9ac75053a00d96 : SUCCESS in 21m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0fc70f5c21a847e3860c5797c0ad8910 : SUCCESS in 1h 04m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/361bb8f30bd048ef83a3b0a601e879e4 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5d12d9c6e3fa4a00ae9f04e485ce673f : SUCCESS in 1h 08m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c670cb9f71b240efb949b48b4dc94af0 : SUCCESS in 1h 07m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ee18b7befae74247ad3806326df57d5b : SUCCESS in 1h 07m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c6e6e5e870054fff9fbe5a41233a8fd8 : SUCCESS in 1h 35m 43s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/78788de25a744345908a1211ae7ba900 : SUCCESS in 1h 31m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ece0b8f865dc4f4d8cdd18a9bee082c6 : SUCCESS in 1h 32m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d23881245a794a6cbf5e75049f45cbfe : SUCCESS in 1h 36m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/465ba8631da34fa9b30bc2facc9228d2 : SUCCESS in 27m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ad05eafcff2247d4b631be647e806d71 : SUCCESS in 23m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76363e72f6374bfd87931711fc6c5fa7 : SUCCESS in 23m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3935077a851743df8dc0ba8254126082 : SUCCESS in 22m 12s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/339696cea6d64cb1aede986e6a0426d3 : SUCCESS in 29m 32s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5bed2ab77885477e8362278cd81b2305 : SUCCESS in 38m 26s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50eead1cb9f3422a823eae08dd274ceb : SUCCESS in 31m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/880e2a24b7b74e0c8ccb276afc29ae3e : SUCCESS in 30m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bdbac268423a4a12b36f344a4d9b8eb1 : SUCCESS in 14m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d16006e14a27415187d0a991b698de18 : SUCCESS in 19m 38s","accounts_in_message":[],"_revision_number":12},{"id":"0ab48e4af455e948e7ed8718634d3d8365b1b6eb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-17 17:22:44.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"7cb0d85cd7427b1e6abf59ea82801306d530a9d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 17:23:03.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"97f879c784de0b0aed08ee6e3d30f11e05f51eba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 18:37:00.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a70031439924acdb44974e7fff088bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072843239be74a70b24c57d191a324a5 : SUCCESS in 8m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2a0fc1fc31c6475189598a42334b0a20 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7260026d52c4fbfb438360e4616e8fd : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b5b967c006f46b19a48c501e910c1aa : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22eb31735e7e4c83bd92694b270fa47b : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef4f1ddb6f684c499c6dd957c113a980 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf6deb3c9b8c4179a6ae98e2917bb697 : SUCCESS in 13m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1e8dd24b98464b75b402547117d3a77c : SUCCESS in 13m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7c871e6d821345519b0e05a6794442f3 : SUCCESS in 20m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85e326af029843968ccf4ee228049e29 : SUCCESS in 20m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a1a38ceadbc141ff81c0b13974b2dc6a : SUCCESS in 11m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74f1e5f983984d78ab650e3f388c30d1 : SUCCESS in 11m 26s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/28ca38e377fe446d9e5022cf49c1c0a2 : FAILURE in 29m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4229d584174e4b81bc1f20c0e3f0d408 : FAILURE in 28m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3ffec6e771be4c1281126cfe1faa0e8f : FAILURE in 30m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a597d563fd74734b5d62f600831d8db : FAILURE in 23m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0704ad25e2ba4fd5af8389ecde12eb36 : FAILURE in 22m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5c2427cb3eeb419883cbbd84bd9bd220 : FAILURE in 42m 16s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1358e6088b6d4b269b0216000dfbec5e : FAILURE in 1h 02m 56s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7febe41738c34be4898a9b61ddb87a72 : FAILURE in 1h 02m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc7520c9612049a29234a1f14d66edde : FAILURE in 1h 13m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/45310812c3c8468c8f40ae1c351a78a7 : SUCCESS in 30m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5a0505f5c8984721817baf872502fccc : SUCCESS in 15m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c83c7d79936464f993b91b3bbfd8720 : SUCCESS in 15m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7c24db3de6948dda98ee372f96a77f0 : SUCCESS in 20m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/31d77ad6ccda41a49b7e25b7efe2d7a6 : SUCCESS in 24m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e8ebe397d2fa4f7e8f6c2a3a66cf8ba0 : SUCCESS in 28m 57s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df5bf0023d894d7a83b557106fedd04a : SUCCESS in 18m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6f06d8f717f4c66a3ec97e67d26e43a : SUCCESS in 25m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a8d5809137414548af04172bdfc349bf : SUCCESS in 15m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88e3467efc7d4321ae132c7de749edb4 : SUCCESS in 20m 36s","accounts_in_message":[],"_revision_number":12},{"id":"1a4ad9246e3bfc36c9de8b7780d0a30cd415ba65","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-17 21:45:09.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"86daf301a8c236ba0a8c508c4189ca438d5c4270","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-21 08:57:08.000000000","message":"Patch Set 12: Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"9972d070ff516cc009c48e40a82113c83b20d254","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-21 09:13:18.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"6b5278fde562ec5a8fff5102e48e8b9b7bbcd218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 12:00:39.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8ff602e5e404223bd567f9904d91f0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a884594fb24a218a9db8e9d7ced757 : SUCCESS in 6m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6bb9c61a7ab74b8eb16f4d5c2fccb873 : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45f4a6d0050741909b55e06232b1a006 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06bc015cbc6640ad85fba5f6b94ebd2e : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7c331a9c50c4062baba0f695991134f : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e173445a97ce497782c45ef82fe2d741 : SUCCESS in 7m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c409334b1a84f94b598994e91804fce : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c62d347ad98c4da5b03d69fe86d08c1c : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/de44448300d34974b03ec30e7abca01b : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8bbf5cc0c3284f50a4243200ac594696 : SUCCESS in 12m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8b0e5fedceaa4609a6ffbad408ab573a : SUCCESS in 16m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85d3d6c638ae42389dab4af7211451d0 : SUCCESS in 12m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b58d755ece7b434eba2cfbbab980cdc7 : SUCCESS in 19m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c023cf7e5734a049e2d279af2722592 : SUCCESS in 17m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/40ee7f3b52f44e549ac5b5e00b453b36 : SUCCESS in 49m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/aae3c07bf7934571adbd557f6c6e068a : SUCCESS in 43m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fb3b4b97585c43abbd23aaa24b151ab7 : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db9e1f0deedf4d2785b5ca3ccfa0cd66 : SUCCESS in 1h 25m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/39a19702a9a240f48baea16288dd7075 : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/172a32cfbd174cc0a00169b3f8977091 : SUCCESS in 1h 06m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4d3c581aa17349f093956122cc7567a3 : SUCCESS in 1h 45m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fb90daa0e19470fa97ec85ab409a9df : SUCCESS in 1h 47m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d16c3dc6942d4fe1860db9518cfae17f : SUCCESS in 1h 03m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5ab8921fe86e438e855717145e2762dc : SUCCESS in 24m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/abf0c0b4b0fd4fc999f4e327710f460c : SUCCESS in 19m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f42e5c136a147dc8dc5e152ee4eb7fd : SUCCESS in 14m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/899ba671bac241878730e91f964f204f : SUCCESS in 22m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0cf68c69b45e4f87888354842f618892 : SUCCESS in 30m 47s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/acd5d2f9941545c588733a1c68c8679f : SUCCESS in 31m 10s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73d9703406384e28877c158e8e18d5be : SUCCESS in 18m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83c34a09cdf84f56942fd1553f73bd91 : SUCCESS in 28m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bd833a9c4f3b41c299e95e65a7a0d168 : SUCCESS in 16m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22bc12efb65a4d8da2ea3ec3615dbe4e : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":12},{"id":"bc7651148a731a78cd56382da726c4ded5584407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 12:01:10.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"5b5dc16d18904e1efc2c16c6e1f5ec17fbee46ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:08:47.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7cb641d466b4c99a42a8fb2bf112f3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9e386abeac418bb5c9f2ac5cdd47dd : SUCCESS in 9m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/625362b471404cddb4c5c01d94ac9cd8 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddea71daed0b48f6bd4bbc61a7566388 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60013bcfa172427eb9820bd9cdb7d00d : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/715bc273b7e24d62be68ebce262387c1 : SUCCESS in 4m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e274775b61384c7c96b7482adc15681d : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f41a9a5b28f49f88e19afe19e54757f : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4dd5d731ea342d789fc9af700f0f6b1 : SUCCESS in 11m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/11bc0a44e9a148968d39370fa382cdf4 : SUCCESS in 24m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee43ab3b21cd4d8389d4cdc19122bbde : SUCCESS in 11m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e65ce0f3632845429a81190581a324f3 : SUCCESS in 25m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36d18642074f4c36933157c7628cce79 : SUCCESS in 11m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/36316168c11d4dc3aebbfa6943fb16b5 : SUCCESS in 1h 06m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/87d5d33ee7fe42eda39c83f1cb6240e7 : SUCCESS in 1h 06m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d3af5febeb9646c9b7dd107dde4f3699 : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/478cc3b469b44fb3a59367c6ed349340 : SUCCESS in 1h 07m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/675b0a219ee84f1f88ceed096c4c69c3 : SUCCESS in 1h 08m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a9424d90e84942e2b800501ca4666a6e : SUCCESS in 1h 41m 32s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/be965958d95747f2a8b94f281511fb7b : SUCCESS in 1h 34m 44s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/279eb698b58f4d258d653c112d2e01cb : SUCCESS in 1h 38m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eac3608166f42398e87fd10d21c023e : SUCCESS in 1h 39m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9969377348c0499284f931ccb7c10c58 : SUCCESS in 25m 23s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/aa2565e36e4e41b492d5242be12d717c : SUCCESS in 16m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/635f604c25b440779fff306e859d7ea4 : SUCCESS in 23m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dd5084ed68504c79805afaa6040b833f : SUCCESS in 20m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8f12ca2094be4596aeeee1fd6e6dcdee : SUCCESS in 18m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d14334a5ae7e4e0495bd355ef14e3f3a : SUCCESS in 17m 21s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6114bd0914154140be924b7fdd023dd1 : SUCCESS in 36m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d735d2aa46c43dc9d631f17ea9fa23c : SUCCESS in 17m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/de271ce0063a47d18029285195fc96e5 : SUCCESS in 28m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec27c49103124ecfab5a0913a4a79583 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":12},{"id":"b189f45a3b50cab825bfe71ef23cd55a9dd35d6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:08:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"cde3eba8d0138a0b6c90b3af071aa61fb27949ae","revisions":{"74bc1d8b108d47e2f937b0c3421c220de5f08de8":{"kind":"REWORK","_number":1,"created":"2026-02-11 18:09:25.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/1"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-11 18:09:17.000000000","tz":0},"subject":"Make internal network optional","message":"Make internal network optional\n\nUse case is for deploying hosts that are not passed through\nto kolla.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74bc1d8b108d47e2f937b0c3421c220de5f08de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74bc1d8b108d47e2f937b0c3421c220de5f08de8"}]},"branch":"refs/heads/master"},"f9cccdaae9e17d21de575b083df3f8499c145ce0":{"kind":"REWORK","_number":2,"created":"2026-03-02 15:59:13.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/2"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-02 15:58:46.000000000","tz":0},"subject":"WIP: Make internal network optional","message":"WIP: Make internal network optional\n\nUse case is for deploying hosts that are not passed through\nto kolla.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9cccdaae9e17d21de575b083df3f8499c145ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9cccdaae9e17d21de575b083df3f8499c145ce0"}]},"branch":"refs/heads/master"},"5204a794147e41641486648bce6984c9d0a837cb":{"kind":"REWORK","_number":3,"created":"2026-03-02 19:06:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/3"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-02 19:03:57.000000000","tz":0},"subject":"Make internal network optional","message":"Make internal network optional\n\nUse case is for deploying hosts that are not passed through\nto kolla such as lustre storage nodes. This allows us to\nreuse the usual set of overcloud commands instead of creating\nyet another set of similar ones.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5204a794147e41641486648bce6984c9d0a837cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5204a794147e41641486648bce6984c9d0a837cb"}]},"branch":"refs/heads/master"},"6d06afcad77c566826521ccbe9866cd9bfe133dc":{"kind":"REWORK","_number":4,"created":"2026-03-02 19:16:41.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/4"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-02 19:16:41.000000000","tz":0},"subject":"Make internal network optional","message":"Make internal network optional\n\nUse case is for deploying hosts that are not passed through\nto kolla such as lustre storage nodes. This allows us to\nreuse the usual set of overcloud commands instead of creating\nyet another set of similar ones.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d06afcad77c566826521ccbe9866cd9bfe133dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d06afcad77c566826521ccbe9866cd9bfe133dc"}]},"branch":"refs/heads/master"},"74f98544e32bed83d024819ff476de9fd3a4f995":{"kind":"REWORK","_number":5,"created":"2026-03-02 19:20:18.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/5"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-02 19:20:12.000000000","tz":0},"subject":"Make internal network optional","message":"Make internal network optional\n\nUse case is for deploying hosts that are not passed through\nto kolla such as lustre storage nodes. This allows us to\nreuse the usual set of overcloud commands instead of creating\nyet another set of similar ones.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74f98544e32bed83d024819ff476de9fd3a4f995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74f98544e32bed83d024819ff476de9fd3a4f995"}]},"branch":"refs/heads/master"},"8aec54d2ec2d7ddcdb2a696878b579694c00a152":{"kind":"REWORK","_number":6,"created":"2026-03-02 19:21:36.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/6"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-02 19:21:35.000000000","tz":0},"subject":"Make internal network optional","message":"Make internal network optional\n\nUse case is for deploying hosts that are not passed through\nto kolla such as lustre storage nodes. This allows us to\nreuse the usual set of overcloud commands instead of creating\nyet another set of similar ones.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8aec54d2ec2d7ddcdb2a696878b579694c00a152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8aec54d2ec2d7ddcdb2a696878b579694c00a152"}]},"branch":"refs/heads/master"},"895fc2821c8fd6c3252696a8dabcf463d40f7b3b":{"kind":"REWORK","_number":7,"created":"2026-03-02 19:23:11.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/7"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-02 19:23:06.000000000","tz":0},"subject":"Make internal network optional","message":"Make internal network optional\n\nUse case is for deploying hosts that are not passed through\nto kolla such as lustre storage nodes. This allows us to\nreuse the usual set of overcloud commands instead of creating\nyet another set of similar ones.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/895fc2821c8fd6c3252696a8dabcf463d40f7b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/895fc2821c8fd6c3252696a8dabcf463d40f7b3b"}]},"branch":"refs/heads/master"},"e9bf3395095de83824aa7ca3f13820b32a671a67":{"kind":"REWORK","_number":8,"created":"2026-03-20 19:46:42.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/8"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-20 19:46:36.000000000","tz":0},"subject":"Make internal network optional","message":"Make internal network optional\n\nUse case is for deploying hosts that are not passed through\nto kolla such as lustre storage nodes. This allows us to\nreuse the usual set of overcloud commands instead of creating\nyet another set of similar ones.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9bf3395095de83824aa7ca3f13820b32a671a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9bf3395095de83824aa7ca3f13820b32a671a67"}]},"branch":"refs/heads/master"},"23de6f321b407feb2aa3c1e9a076c35ea0f8ccca":{"kind":"REWORK","_number":9,"created":"2026-03-20 19:47:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/9"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-20 19:47:22.000000000","tz":0},"subject":"Make internal network optional","message":"Make internal network optional\n\nUse case is for deploying hosts that are not passed through\nto kolla such as lustre storage nodes. This allows us to\nreuse the usual set of overcloud commands instead of creating\nyet another set of similar ones.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23de6f321b407feb2aa3c1e9a076c35ea0f8ccca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23de6f321b407feb2aa3c1e9a076c35ea0f8ccca"}]},"branch":"refs/heads/master"},"d298819bc3968e8ba33e4ead569fe081acdd00d5":{"kind":"REWORK","_number":10,"created":"2026-04-07 10:58:46.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/10"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 10:58:00.000000000","tz":0},"subject":"Support provisioning hosts without kolla services","message":"Support provisioning hosts without kolla services\n\nUse case is for deploying hosts that are not passed through\nto kolla such as lustre storage nodes. This allows us to\nreuse the usual set of overcloud commands instead of creating\nyet another set of similar ones.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d298819bc3968e8ba33e4ead569fe081acdd00d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d298819bc3968e8ba33e4ead569fe081acdd00d5"}]},"branch":"refs/heads/master"},"94774ce133f4d7e97f55e58ecaa2b14b6567301f":{"kind":"REWORK","_number":11,"created":"2026-04-15 14:53:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/11"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-15 14:53:47.000000000","tz":60},"subject":"Support provisioning hosts without kolla services","message":"Support provisioning hosts without kolla services\n\nUse case is for deploying hosts that are not passed through\nto kolla such as lustre storage nodes. This allows us to\nreuse the usual set of overcloud commands instead of creating\nyet another set of similar ones.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94774ce133f4d7e97f55e58ecaa2b14b6567301f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94774ce133f4d7e97f55e58ecaa2b14b6567301f"}]},"branch":"refs/heads/master"},"cde3eba8d0138a0b6c90b3af071aa61fb27949ae":{"kind":"REWORK","_number":12,"created":"2026-04-15 15:08:10.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/25/976525/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/976525/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/976525/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/976525/12"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-11-22 14:06:51.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-15 15:08:05.000000000","tz":60},"subject":"Support provisioning hosts without kolla services","message":"Support provisioning hosts without kolla services\n\nUse case is for deploying hosts that are not passed through\nto kolla such as lustre storage nodes. This allows us to\nreuse the usual set of overcloud commands instead of creating\nyet another set of similar ones.\n\nChange-Id: Ic6fd254977697ce7be6916fc89952b16c283ff17\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cde3eba8d0138a0b6c90b3af071aa61fb27949ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cde3eba8d0138a0b6c90b3af071aa61fb27949ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
