)]}'
{"id":"openstack%2Fkayobe~976615","triplet_id":"openstack%2Fkayobe~master~I3d88dec15a4540c85345770029536fceef9c6b32","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-02 16:24:30.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-18 07:23:50.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-02 12:55:41.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I3d88dec15a4540c85345770029536fceef9c6b32","subject":"Template disabled repositories","status":"MERGED","created":"2026-02-12 12:39:07.000000000","updated":"2026-04-02 16:25:59.000000000","submitted":"2026-04-02 16:24:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976615","meta_rev_id":"bbe40c2d183b466c4d07f16b5d3a24a1974311b8","_number":976615,"virtual_id_number":976615,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-02 16:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-03-31 20:31:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-01 07:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-04-01 10:51:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-02 08:44:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-04-01 10:51:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-03-31 20:31:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-12 13:23:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-12 14:27:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-24 18:34:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-03-05 10:10:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-04-01 10:51:05.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"8fd8e76a09313782df82c75177b23427168e473a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-12 12:39:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee7cd772adfb35216032ba93427aa3f21b4ea63b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 13:23:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/016b1700592b40d8a2dbe0edd977f169\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc3aaef0abc412e840f3ceaebe59b56 : SUCCESS in 8m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4010968e80e54991b3892749cf37359b : SUCCESS in 5m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94266f1ddfb34d6ab7aa16cbaa84f3c1 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef953775f2941489896b892a39a696a : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f6cfb16e016468cabc7daf0ee8fac32 : FAILURE in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44ffc3054f6a45e4a574ca2343f4ecc8 : SUCCESS in 4m 26s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/cf6306e8269447b3a6073e791b02eb8e : RETRY_LIMIT in 2m 34s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/eb7f55ac590e4413b5a79ebb706cf6a3 : RETRY_LIMIT in 2m 48s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/8d723e1733c44637a21706dec61f86b4 : RETRY_LIMIT in 4m 00s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce4f9c31c804475f8052c86f7df647a5 : RETRY_LIMIT in 3m 58s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cbd20c0d4e724c5ea9611993460d92a3 : RETRY_LIMIT in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c73d33f3252b97dc73ad784b968ef35c246b1272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-12 14:43:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1b2e6a315103b84828b8cb1768bfbcf830250c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 15:57:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c5a0d55fb7947aa8178b5d3ac5457d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c564b836a2433b8de3c01607c7334d : SUCCESS in 9m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/794e01b00e724ca89d92bfe0c9d5ee77 : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b322b7767c1344a8bf5a20dbc30821a9 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b41dc1f5f65420590b5249091c637da : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9651fae06ad4b0e8ebaee52d1f59b58 : FAILURE in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/029e20f991444921a395069d36d559a4 : SUCCESS in 2m 37s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/2b8b7bad79ea4a4f9f600bf542b4eeb3 : FAILURE in 24m 48s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/c873563f2217436f93cc40f667015840 : FAILURE in 11m 24s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/9a8f1949ba68490f8ae9cfedbb91d986 : FAILURE in 18m 33s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dca88bedb5494e399dee21bd5e2b5148 : SUCCESS in 1h 03m 16s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c9cd791fdc8d43e88fd210c541e190c1 : FAILURE in 9m 13s","accounts_in_message":[],"_revision_number":2},{"id":"35cc3b52dcbd95c1fc91a7d79c87e782426b899d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-12 16:26:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"da8936c8841f3805507e93ddf63871d4145520e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 17:38:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd8b1edf81bd4e98b30471acd042213d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb7416a25af4f988a66b44599fea095 : SUCCESS in 10m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b3db2260bd8b4c5698340ac5fac08af8 : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5541ff3f2164848aef56c64c1a12098 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ccd2eb502949b794d399fcfe989faa : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6506b94334304564a374d42256135734 : FAILURE in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cc05f5fad91459da0cbf4e5c98f069c : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5d867d4e8624d349d6f59fdc2f099c2 : SUCCESS in 2m 50s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/4374ac5a87224046b2e91aaeba1764dc : FAILURE in 15m 58s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/b5d97ee5a32a4248971600bbe11e54ed : FAILURE in 23m 26s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/f0377ebbb4cb4c8a9afc0c1f981f25c8 : FAILURE in 10m 23s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efa7401f44754f8da913a75504bfc10f : SUCCESS in 1h 00m 17s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9c7cb65d95074f0690c39489a5d4b495 : FAILURE in 15m 46s","accounts_in_message":[],"_revision_number":3},{"id":"f19dac536c0913cad707b0f711e428b4ab657519","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-12 18:04:08.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ca064d9b0bed43047c2eec9bd29ead5566ee056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-13 17:24:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"311b6d1d59fb73eb7b42ad4a1bd481f8c700e408","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-13 18:20:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3bfbba8956df59328cd22386a3762cc590d6512e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-13 19:18:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a0b84e9d7a82a5aabc8a4f609681265d5a9e01c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-13 19:25:43.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"53bbfd0bbba8642346213f3de146d7b112d386bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 19:52:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cb39a20a36c4e27bc7a7ce33035c7c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08c39c42493412a94be964a1bc5eaaa : SUCCESS in 8m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/be4489f457a0480fb378042b989064ea : SUCCESS in 7m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ca7563962db4323b98320576b9aed15 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4994575fd6be470aa4225ee3d04fae9f : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adeb61a02d0b4d25bb9e57d4656a0b23 : FAILURE in 2m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acbd762d496b41ebb50c9b66d359a5e9 : SUCCESS in 3m 54s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/d3bc2515bebd4d999a72a5ebdfd0e620 : FAILURE in 10m 42s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/45035a98015341b6b0b3fb11e28c9b97 : FAILURE in 21m 25s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/128b04be83a042dfb41e9c4cce878c19 : FAILURE in 21m 34s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2389a5fd440345cd978620efc5cc9627 : SUCCESS in 24m 54s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0dd4de0a84bc464480f33771f2c48136 : FAILURE in 22m 15s","accounts_in_message":[],"_revision_number":7},{"id":"a8b1dc1e850816b38a7719787ccc1a4c113ad92b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-16 09:27:17.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"09e41cf8cf3740043101a7ea37c3c4b1dbb9e872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 10:18:52.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bbb2fd2938948b58d1b3520a67dcdea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fb4f436711e44c987743697d040a1c3 : SUCCESS in 4m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/150caae5d7364c61965da66fd210678a : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab664a8caa194811ab8fbc1521706f88 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30268cdb75b4d3aa77cf6027a18a048 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da7582d5c98d4a32a5a5e501459c6400 : FAILURE in 2m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e25d821e19c487089040cac7bf12e63 : SUCCESS in 3m 49s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/abf1cdeae945436eb0c492bea0dde7ea : FAILURE in 7m 05s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/3e2cb0b0377448b09e21f48d4fd4bc48 : FAILURE in 10m 55s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/f4720a582c004ab7b86ec91cb57a6751 : FAILURE in 10m 11s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9938534478814304a453be5237bcaaf4 : FAILURE in 43m 34s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f3ee18051429446083c75a57bc1d1a65 : FAILURE in 19m 33s","accounts_in_message":[],"_revision_number":8},{"id":"92ec2a491de208b628118eb9882907b91018c1e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-16 15:13:12.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"fccd4f20a9a301faab47ef8a08a8535e3f1b9646","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-16 15:54:01.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d3a3f70e03205d199736fcf82a61e00cbdde4227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 17:00:36.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9082269c022e4939a87db9781d5fbdd3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b0cc45da4e34fb5b662ee7694e7445d : SUCCESS in 9m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f6603ed5b0f34bc5a2744263d9db7186 : SUCCESS in 7m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d089c711aa543f88c23de03e59ab974 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac1cb68884af4ecdb57a69002ac9d63d : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10e852544de64e8a8d58501e58f87506 : FAILURE in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/016c901ce3d042b39804d412503ef866 : SUCCESS in 4m 08s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/1418cc1da6344f02a4a6d4df9af51f62 : SUCCESS in 27m 40s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/7c14a0a837b643a0823933924049f657 : SUCCESS in 30m 45s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/b90c9a2cc3ed42de848a48b108129fdc : FAILURE in 15m 33s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6e8aebf09a241379c2a4269b003394c : SUCCESS in 1h 04m 01s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/18ea428d3a424e3797ccfab8d977c377 : FAILURE in 16m 38s","accounts_in_message":[],"_revision_number":10},{"id":"4012d9c6dbd7658892696ac632a7e76d7556bf44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-16 17:54:47.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"5a6221ffe12c571ce1104ef8d2a6e2168ff08db9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-16 17:56:25.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"81f4323650a161af698027c72a94273ecf273e4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:06:40.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2f060fc9f55495baeab3479334e4026\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c200ca6eb9420fa5d4592b4c0216ae : SUCCESS in 9m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b0742d128e974fd09fbdf8308e69667a : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b86f478047984539904a2db57a198afa : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea40110cf5c64c12aa217c0b6364fb60 : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46ee5cfe8260412e8b122ab31c3e11c6 : SUCCESS in 3m 59s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/9fb84a14ba7e4812b3df06cd206a023c : SUCCESS in 58m 08s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/8487ec1450b64504981f2e0ce6ff1ab5 : SUCCESS in 1h 06m 21s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/a45372c9f54c4c1b8285065bf26ffb18 : FAILURE in 9m 55s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5a29a7cfb8442b889753d5fe383ca1b : SUCCESS in 57m 48s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/89967c07145f462098452161ceadc128 : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":12},{"id":"624580af84dee1c78ea06ac5feacb421806085e1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-19 14:13:49.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"cc5cd2b3b39e7d24524a6ff4796a13b9f00f4a1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 15:33:01.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bbd5fa15bbc4511a6782637900ad5fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1df362f5acac48aa9e3b411922dc9ed4 : SUCCESS in 9m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/52eadc2d100240d98ca88c45d493541a : SUCCESS in 6m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d48848d2a0b4a9d9de1d167c30c5ec1 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60ac1efcb33a46edab89d1513d02da32 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10ddfff9eb204485bcc867880fb4fba6 : SUCCESS in 2m 43s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/9c55121f6899459d8f7638b399764e4a : SUCCESS in 54m 54s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/fd402243b33c410eba1b58fd8acabef8 : SUCCESS in 1h 02m 39s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/59fe7db60d8744b58fecccda720ea8cd : SUCCESS in 1h 04m 52s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76d50d12daae4a28a82631ca18ae6dfd : FAILURE in 44m 52s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/41dadc54c4514e5e8e1683a6feca655f : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":12},{"id":"25940cc5abf084fc068a977312c61883e74424fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-23 18:58:03.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"fdba6efbe265f8fc05087ce3112c0b53b6923190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 20:05:44.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/115ccf6922e147c39c0eae35dd496b79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c7c86584614c1783033a622a5dc846 : SUCCESS in 10m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c1df36b298ea40b4809c0cf7ea6a4b8e : SUCCESS in 5m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5662f479efad4aa1881b8e2598ca94cd : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979e8114638942a987308e4ef3f77fe7 : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acb2dbd4a2c24feea91a2f99fa905293 : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ebfd76c412c4159a52a2762d09a3265 : SUCCESS in 6m 24s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c3a6cb15e41424c8ccf7bc997efd8f5 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2a9735bf149e46769af14658616592f7 : SUCCESS in 6m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e6668c5772144e53b770ea54325b62c7 : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/927ab9c8ff1a42ab86c8c7db4267d99f : FAILURE in 7m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ab3c1a62a6d4af8a022d7ba64787be5 : FAILURE in 11m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/85e7947cfc3c4e5fa66a958b3f3ad07d : FAILURE in 11m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e5f202b3a3b4b68b694fd6b5167bd86 : FAILURE in 13m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/543f1ecab1d242a89ad8df6a637b89f4 : FAILURE in 9m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4f8afe68c0e84b5abd11d9cb848a48a0 : FAILURE in 10m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8a707e925f234f4f92a5376534ed24cc : FAILURE in 10m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41d108ee7e5f4e5b9aea132e97d73bd8 : FAILURE in 9m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bf237914942c4604b08a65d5e90caac8 : FAILURE in 9m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a029456a54bc4e9fb02dae35f9a9fd77 : FAILURE in 58m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8580dae896d04576805128a9f9f22aa6 : FAILURE in 55m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36c0602598ca4f4c9991698e689c0f13 : FAILURE in 51m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/031e1341f2784e58b4f5ca1b7fd33ba1 : FAILURE in 58m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dc96509a9af04a59a58a1a262f489874 : FAILURE in 6m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c966fad81ea240a1850787479c10eff5 : FAILURE in 6m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50ab3442109846089a73e5b969d74ecc : FAILURE in 17m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/25c937a8393f45ef9de4071653d550d2 : FAILURE in 8m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/32bb5fcbd99b4e7184123686e4e35fef : FAILURE in 42m 05s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bdf1d2cb0c794ca2a4bb879f65ca51ff : FAILURE in 16m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f3a9352e1dd46e692c6352d76b009fb : FAILURE in 29m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f54acce07a4b4f9ebe05dcd0597a6806 : FAILURE in 32m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aa85a54c1a594a09b03d2193cb6f1a03 : FAILURE in 15m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d057875acf454e2dbabaef0e04e9acd6 : FAILURE in 15m 58s","accounts_in_message":[],"_revision_number":13},{"id":"92e4fbb3da7380aa6c40674ddd5193f5bf23e0a6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-24 18:34:44.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"cd1d32150f2f56cfe6f98315371c7c4e80dac094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 20:09:36.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f49af5f5e5b040769c3b6aac651416e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5defd48043f94d23b856424eab19eb1c : SUCCESS in 9m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6108b8e5bdc8485fa131da5425eb66a6 : SUCCESS in 6m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2678734ed9f4670a9c1a6ef2c6f5f66 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/067c9e2c83da4c28b9bd4190cd720ba1 : SUCCESS in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cb1af57c1e44f929e0a6782633ee802 : SUCCESS in 4m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3a2ec2796f854ada81b798406bfa739e : SUCCESS in 3m 59s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9734f0b02da14629b09e97c81ecf15d6 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e995e053f744d5e9c2a666b05308329 : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/22ce836e98634c6db0e75d0d566226f6 : SUCCESS in 10m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1726606fdc8d4cbfb137344afd7fcb80 : SUCCESS in 18m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd780f5b11d5441f8f292a6cbf21d5cc : SUCCESS in 12m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ac345cee157f482392a4828dc2fab28f : SUCCESS in 11m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8254d98379884a17bb82c905f99fa702 : SUCCESS in 18m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b2b84637f6de4fddbcae364f34d7a16b : SUCCESS in 56m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a2b940223bef4c4bbbe4a173db1f62e1 : SUCCESS in 55m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6c42b85761fb4a249561e090f94d331d : SUCCESS in 57m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07c60364ff8c427e96b30f4ee4a579ae : SUCCESS in 56m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/479b199224fb482cb7b34a41e002c2eb : SUCCESS in 55m 14s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f1aab6fca65145d7acee4cbf3a134867 : SUCCESS in 1h 33m 43s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4a44da60834744d8a6b6f61696c49559 : SUCCESS in 1h 28m 58s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c8a36d6cba0432ea9e71305196abd35 : SUCCESS in 1h 23m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95cbb36af7384bd6b1cfeaa215694de1 : SUCCESS in 1h 30m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/611c2961cab84399973d7d76fe4a4505 : FAILURE in 19m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/14b51c8dbfb0403e9055c2ebcbe9f65b : FAILURE in 9m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf9b146883c749cb82733d453893c191 : FAILURE in 8m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e4fd55f0b93425cb8d3a8893d4fc474 : FAILURE in 12m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/102e5cd0db254d4e969628364507a337 : FAILURE in 24m 46s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2c82fe7b4e304fcf8b30bddf3281a723 : FAILURE in 34m 11s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4daba2c578d44419b1c0769bc5cae3a : FAILURE in 27m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76ef446f80bb4c58bdc1d79664e8f60f : FAILURE in 26m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e614117fc8684262b5be8a3bbe58941e : SUCCESS in 14m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5631266514e047c2bccf52380705c749 : SUCCESS in 13m 46s","accounts_in_message":[],"_revision_number":13},{"id":"a76ad5bc27ec337abe653bbd731083b99316a894","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-04 16:00:19.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"7ae9d1cd8a7987112093226f2294c98ac71ada72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 17:53:04.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eb06c42561b44e5a7b95789cefe637b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a71dd4a60c8a428f89e56d41b61c39be : SUCCESS in 8m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/62530864f2e840eaad069ed94ee463ef : SUCCESS in 7m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a029a235a0d4404891b758347fc34524 : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ffb502c2519440f95fd7fdb26ff2ca9 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95266891758540f69dc75c9cc678ca1d : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ba6a070939e44f98bace585ddf8c9414 : SUCCESS in 7m 20s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/93e83e172c1b46ffa9f55b1fbe76e77a : SUCCESS in 7m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/846c3769d0a04cc6b09246221764d63b : SUCCESS in 14m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c1feb6adbe7425db78dd591b9d5a412 : SUCCESS in 12m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/adb522d0397f44c79974ce90c2f83758 : SUCCESS in 19m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8711ac1fd7e041a7a03e97918e7b1fc2 : SUCCESS in 14m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bf033b935b0640d1b8723d3f6d9fcff0 : SUCCESS in 24m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edf786b8314c4a0781303f48bd6793a6 : SUCCESS in 17m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1c8b9b4c921844a683078724556969ad : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/40b5ab8a1cc744cf9587f1b5cd28b62b : SUCCESS in 1h 00m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c6a7b90eb54b4761bcb0d149a3422fd7 : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49d7be2c323a4bb8a9e6b4b931a318ed : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/31130e543c024fd1888a3f3fbddc07ad : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ad59798785944468ada3bce0c5e40ae0 : SUCCESS in 1h 47m 18s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/049412fd96bc42d9816e003e70588499 : SUCCESS in 1h 49m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/829484342b4e462aa0960c5b301932d7 : SUCCESS in 1h 46m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c460ad75e6b434bb216514ae4c4ee00 : SUCCESS in 1h 41m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/87d32ae3adb74e63b680fd21ac78a2b8 : SUCCESS in 12m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5f28fc262a124d8d8db387e63b6bf42e : SUCCESS in 23m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1b23db9bba44b2dadf20c2e7b152bee : SUCCESS in 19m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/091e6c2607034377a02a72fdd085c03b : SUCCESS in 17m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a0c8e67757c54495b035de71b05e0062 : SUCCESS in 19m 05s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/66dd946cea6f448685a683d081a66fb0 : SUCCESS in 26m 11s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce9cf02eebcc4d3fa2413efe1f26070e : SUCCESS in 31m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca2025f8ee824990abc2dba67637f854 : SUCCESS in 28m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1335f0577752459d9600b9008f296253 : SUCCESS in 20m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b16a844a63b48949e02cbe819acd9b1 : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":13},{"id":"aa9848c5a552cd0ad54f1f018740c15bb6c0049f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-05 10:10:03.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"fcb403fc8bb45b4e5e9d58ac2be7c5d7ac57f47c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-05 10:19:10.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"f97880db745e8a918959029d8ae76d4ce9cc73de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-11 09:24:33.000000000","message":"Patch Set 14: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"2e57acfa549e0dc2ff79135d59c436578091583f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-11 09:24:44.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"2994eef43befa060269eb6201811fd043f3e233b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 11:16:12.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7960fbe90c0e4645adf09c0a95c72c5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400f092a229d4d5a8b11f8afc1682253 : SUCCESS in 4m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e1ffbe6b258a462eb1f9b76c260b3d54 : SUCCESS in 6m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/354da412c03b4cdeadab7746a0819dae : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f21efc611c41418a49d7bb42d9240e : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/049250d44fb9472aa942e089f98a1f12 : SUCCESS in 3m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3ae0c23bfffb4a01a57cefef698be18c : SUCCESS in 4m 02s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/52ffb28728fb42c591f514c45fea76be : SUCCESS in 2m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73ad6463168c4a429a8f7f8e8c634360 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5cfb59d9c3e4c8080618677a7511c02 : SUCCESS in 8m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/538074a94c91478eb0ed0d7f4383b17a : SUCCESS in 15m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/442310e339fb46ce9a0ba57da153e929 : SUCCESS in 9m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9036f67c184c4ff6bf6fcc2461eb9742 : SUCCESS in 11m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/362329539db64f2eaf5a745590786ff6 : SUCCESS in 15m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ecc1a5b734594568bd61d329cffee60b : SUCCESS in 56m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/813fc46d93dc4802a4f1f946ee0a05ce : SUCCESS in 57m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/508340813cee42efa3b24abe958387ae : SUCCESS in 58m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b6b6ad258f245229daa47a75600770b : SUCCESS in 58m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/addf604b12ba4a74ae374277d00ae48b : SUCCESS in 58m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8e3b9aeb49df4206ac2b2833d2a20b48 : SUCCESS in 1h 46m 55s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2acc21ecec4348c38deda21d58e8a735 : SUCCESS in 1h 34m 22s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9b77585875845a29932ae577497e4e3 : SUCCESS in 1h 27m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ff66168bcda4fb7911080c9ff8d33b6 : SUCCESS in 1h 36m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0010061856a44e2eb0722b7941d39f59 : SUCCESS in 22m 20s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4290e76fd0bd46648564aff90e0de902 : SUCCESS in 15m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13979d5c0f794050a6f7863e16a224c4 : SUCCESS in 16m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/37d575fcb48f44c782b5c1da09f72eb9 : SUCCESS in 18m 59s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/da24389499164075b0836ca3608609e9 : SUCCESS in 35m 07s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/023933aeb20d441fbac6a398c9122629 : SUCCESS in 18m 55s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c3efe0e65f74785a09837a997f8540c : SUCCESS in 15m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bdfa16db5a9486089485957df0482b8 : SUCCESS in 27m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6d7312dab2ef49c8a627e9ae12ce4b24 : SUCCESS in 19m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da10cd8aba6740979c607e4899548808 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":14},{"id":"e03daf3c83b4364df03aec28f4f9116479762523","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 13:46:52.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"fc10aa733925442227ac429f0264d829b59040e2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 13:47:07.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"81c277179d842eead2568c0cbbf3a088adfa16fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 15:26:50.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d847ce1c2d0e4ea296f42ea30292125f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d39ead1a8d094a63abb9b2f1a3a88a44 : SUCCESS in 9m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/08986fb2eee546c6932723c6816d7e2c : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed593bb49d3e4e85ad27d6aaf635f353 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e2b599a7514245ba3b65f07a60bf05 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5203e123dfc4c5a9b9362b93693d4a5 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bde5ae5b0e7f49d3a20bc2507cc0a501 : SUCCESS in 5m 21s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/113ce7a5740d43ba8f7af016994c954d : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f9098862d3d4fe7b5347caa4c30a3de : SUCCESS in 7m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6737fd995029468cb3e373d87efea535 : SUCCESS in 11m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7b9467ad3de543a0b27b913174f66341 : SUCCESS in 18m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07acec5271354be2893d8f91020dc2df : SUCCESS in 14m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2b6266ac9c4f47f692a51b92a42b3382 : SUCCESS in 11m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48c14ebcc2ed431aa7cc1aefc31a3e95 : SUCCESS in 16m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/da641d192d9a4b37b4474341cda2374e : SUCCESS in 56m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/21f59dd095f84a778680c30596291fe9 : SUCCESS in 56m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b309eb10c79a4213b8d64bfe75dec585 : SUCCESS in 58m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bae55d374cf84ea392ed23e4d0f2a0e8 : SUCCESS in 58m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f047eefbd6c342678aa23f9383afa452 : SUCCESS in 59m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7abd13c86d8c4d93a33ff38725522c19 : SUCCESS in 1h 38m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/96a637b3ff0c42a88d0c6c9eb7ba4667 : SUCCESS in 1h 36m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6744e74ae6054d92b430c0f6013d0f21 : SUCCESS in 1h 32m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54db7cb56c8b44deb24256b49290d33d : SUCCESS in 1h 37m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6b7ce0a482ca48469fba261d6ca630f8 : SUCCESS in 18m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e716ba2eab094b89ba82bb4b36ab4dad : SUCCESS in 10m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3eb679a5142749df8bf9e79778983ce4 : SUCCESS in 21m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/044d8f1751fa4229b24b81d1606837ca : SUCCESS in 15m 45s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bce06a8cf05549beac3a30bc9f43064e : SUCCESS in 28m 09s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f00f9f8830474dab8416f9e26ad4ac11 : SUCCESS in 38m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/953d43cd8e844b9b9b82251cc47602ee : SUCCESS in 16m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92326b6cf2434564bb5fb6cb7bee6f8d : SUCCESS in 28m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4624b6f8bd6845e38532b8f033ff86a7 : SUCCESS in 16m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26660a9b48f142638e0fbed7848302d7 : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":15},{"id":"2632de0ec94435ed1b78f929a52d6933a08002a6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 16:35:25.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"22887a4cefdfa4b2319aa51300c955f8b4dbdcbd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-18 07:23:50.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"e1e75183f0bc6f11988c29c902f5176abbfe4ad8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 17:44:34.000000000","message":"Patch Set 15: Review-Priority+1","accounts_in_message":[],"_revision_number":15},{"id":"643e1318189130f14068c171fa0cb4de0c427219","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-30 19:48:22.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"e4aeb0a8b812d2e7f3b2e61b5a952088053194fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 20:21:53.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d069df8850ec4c1fbb26e70b966ff59d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a95b57e8ae64c068e0b31be3f3b543e : SUCCESS in 9m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/839beea2d6b14968997c2f14fd69dbd8 : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db22143112b4431d8ec09cc849970d33 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840c237dcc8648e4a629767d9878d693 : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9af07f2584b4e33a670969db94bd6f7 : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/347ea279c3c547f8936fa3e2aea38890 : SUCCESS in 9m 18s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d54d45e2c0114b32825e01f663e7456c : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4596dd528dc4f6ea92a0bfd71c8d6bc : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b21491a52d94b3cbbf4245cc3f0ec6f : FAILURE in 5m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5e0a98c4cd4744358c7995a273563e12 : SUCCESS in 10m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f3afd5c803c490e82062f02070b3a5d : SUCCESS in 13m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7c1cb7d853464f4aa312368934193507 : FAILURE in 25m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f630333f20ec4cdaa2eba9187750dc1d : FAILURE in 12m 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ca210835f9c447e58ba97cf7f365efc9 : FAILURE in 13m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9af73d0fb6994242ade56729f188fba1 : FAILURE in 19m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ab83c99d782344c8a967f13c31472a9c : FAILURE in 22m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/618061eb610244ae8264c512f8625a3a : FAILURE in 22m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c3970616a13d4e47bd04715b10543a4a : FAILURE in 12m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4e173b79fbf54ddeb419b0260bb7e5cb : FAILURE in 17m 51s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/681ac455463b48f29dd62324ee64dd01 : FAILURE in 23m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/609f6dafb0a841bf801f1e26523c6cdb : FAILURE in 30m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54c423221b5b4b108140803b6a0393ae : FAILURE in 23m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f731b601daa5459ea824713d0a438c9b : FAILURE in 15m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c2b88db1fc134313b0859382185e632d : FAILURE in 14m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7c42ba80d0d4f5e9ce3bad51f84598a : FAILURE in 11m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/db09ad543037496ba8787e315c001284 : FAILURE in 11m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3308e13efee441768d20e7c404db22e5 : FAILURE in 16m 43s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6fbe7af37a804549a802b16c589a8bcf : FAILURE in 16m 55s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79d45a0dd08447e6a74e45420b9ae979 : FAILURE in 28m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/352ce95c7c3f47d381a153cd0914750e : FAILURE in 12m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d1cf4a12364e43b783a2f3b5e27b17af : SUCCESS in 20m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f6b1f0659cd4e1f8fbf5298705922e3 : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":16},{"id":"fc556b3d392ec03302991546cf88112bbdd0db5f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-31 09:35:54.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b8b604006d4a9c29d475bccf668cc92db12d93d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:46:25.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69b526c9faaa4c0586013753a8595f70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2abc1c6019d1401b83a0d0a1a73dbbe6 : SUCCESS in 5m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aa7fd8b8ba534c6da5eb922962b09202 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f33767f2c3a24172bf02b60d341ff34d : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad406878d7f4cd091abfbcf266f8238 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8160a55b8a14ec99ce68c9fc4b45c68 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db49aa3fa9d0493aab6fbc7f120258bd : SUCCESS in 7m 17s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e8290b22134b4b209e5367aafc25ae37 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d1270677448a492584219af26fd7a524 : SUCCESS in 6m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4605c163104c424782b253370a4cbe16 : SUCCESS in 11m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c80d1e4175874797949d685c1e8d6f93 : SUCCESS in 9m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b3ff884c1164e5ca6d4887ab78ad43f : SUCCESS in 14m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7de737882d36414087cf06684002dbc2 : SUCCESS in 12m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04e4ba75cf5d4f5abe1c4e9b88c66fab : SUCCESS in 17m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d6c7bd539b0a4779add406fd45c140d6 : SUCCESS in 58m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/32c52fa9e09d4a0189988171c396c1c5 : SUCCESS in 59m 21s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/33b5ea015b184557b8cda1f882e247ad : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5fa08cfd4aa486ba53d061f5e2b85f1 : SUCCESS in 58m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ce80c29861524395b3911a0ff09d3ab9 : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/777f531233db4a57892fd98d625b3306 : FAILURE in 3m 50s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3f1ae2da4d1e4d3abe6a4cf2ce12dec3 : FAILURE in 47m 47s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4ada175320d484d80483860f5682ed0 : SUCCESS in 1h 31m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a423ef52c684cf3b511b219aa51c17a : SUCCESS in 2h 09m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2e11f69e0c9447599b2c95229148da4b : SUCCESS in 15m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cfcf9b372d4e4adaa3e2ce6aafbe87e5 : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82abaecf4a1d41d48f76570758174521 : SUCCESS in 19m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ef0bab9072d4b32b206ba303cb24f4f : SUCCESS in 17m 46s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cddd6a8be27b4816b8682b72f171bd8c : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/addfd0b697f74f69aa557132c3c6de7b : FAILURE in 2m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96eb7d3afb614d74a573c76a71efd4ff : SUCCESS in 30m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03d59d19a0cc49218afce6dc31565651 : SUCCESS in 30m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b815989c842b4125bbc6609e6fc34d77 : SUCCESS in 10m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75ed950c13ae4d9f98582926e2fa6d91 : SUCCESS in 16m 28s","accounts_in_message":[],"_revision_number":16},{"id":"13040bc2fc5e8ae5ee70cfab804276cb8f489a90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-31 20:31:25.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"766e89b0a8200016ad7a4d82ddbd9e78e663dcdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 22:09:31.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a7fe7c6cb974e62ad027ff1afdd5056\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d81e5908cd468b8185e0cde5a62d5b : SUCCESS in 7m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/06f34c17c04c45f7b762dab9f1747860 : SUCCESS in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33dc1d2bb2c54296917f85ef388bc437 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109958182be147c5be78213d30a30fb7 : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/737f22aa65d8416dbc143957d5bf5524 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7031baf873b476dbbd32a4413836bc9 : SUCCESS in 4m 13s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c6891336f494f6facc0ba013602c37e : SUCCESS in 3m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1410426fbc4f416eae5178a03aaecbc1 : SUCCESS in 9m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8770dead80d149de88005796ff4c4fbd : SUCCESS in 6m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e492d45b431f416994e79611f5271e52 : SUCCESS in 14m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a5b52f2c342401cade70bf33949db83 : SUCCESS in 14m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/96473098466e4c889f317a3b7f79a9f8 : SUCCESS in 15m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/366ee1e731024273929a93986af672a7 : SUCCESS in 10m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5803fef9dbc1454ca1eba915b0b7e1c1 : SUCCESS in 58m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7f352a9b7c534ab7b52ce0fdb11be3be : SUCCESS in 1h 02m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/816240f9abd44ee69470dc62594bbdb6 : SUCCESS in 1h 01m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/990c375296d148ca9430dd1c4e2935a3 : SUCCESS in 58m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/959db7330a624df58f49f180999bd1ff : SUCCESS in 55m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0342ad46e4b648e79d445022f8efe096 : SUCCESS in 1h 36m 59s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2723a636e2c944708eeef71895365a1f : FAILURE in 4m 11s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/938b52e9ee914f898e05106fd76103c2 : SUCCESS in 1h 27m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c238d50d3494f0ab2dddc994f21982f : SUCCESS in 1h 32m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b51dc4e050a4407cb2b772d3970e132f : SUCCESS in 10m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2db41c5e135c4d47ae49d3fb654c246b : SUCCESS in 11m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2dc7a4e3e1241f1bb1e5fd1ca05110a : SUCCESS in 16m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fe4159de83dd440b8900657abaedbb2d : SUCCESS in 10m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c22319ba2d99407eb4bb86dead1787ec : FAILURE in 4m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d92a48ac89904c0f8afdc883de0c8cc8 : SUCCESS in 19m 24s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2087cb698cc44229991c4cf4d615cae3 : SUCCESS in 30m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c7bd78699e14a888d18f246fdb6e79f : SUCCESS in 22m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2f6eb68e75a84414a0b3887249874d07 : SUCCESS in 9m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a967d7f9da1c493cb3d60673ba67c0bc : SUCCESS in 15m 29s","accounts_in_message":[],"_revision_number":17},{"id":"2360d9bae41688b44e0a3a0f25e26095c243120a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-01 07:49:00.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8014bab28465e7b65078f61b93bbe91a8caa7077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 09:55:32.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4469f374a873415eac4eb617083256e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5544a064fda64c35b7dfbe8438d66631 : SUCCESS in 9m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/78cdfc8ed9c349a2ae9b416a62b05e28 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e4a07ec9e2e4999acbf1d9672b8d24e : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d2ddc069054555b9a4f0e75a085a43 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8161dccb83bb48938de937b473a91c50 : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2b28853df43d4f4ab9af9583a2159c57 : SUCCESS in 6m 10s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f43535d94e1c44d58029b3daa13446f4 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/415552697bfa4be3b326840c71f68f3f : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba9eeff8a5a84aab9ace8d2ee71a9694 : SUCCESS in 6m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dd2a1b22dc8d46a99e02f4231c044ce4 : SUCCESS in 9m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6db338592f240e798f1109c7165598b : SUCCESS in 27m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/557299852b544c46ab80c8dc671ff538 : SUCCESS in 9m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a4e5d18556a498e9f7fa1497e4200d2 : SUCCESS in 9m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6ca9824850004cfb93932a490fd6c086 : SUCCESS in 1h 11m 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3131b2d30429466aafb34199f68dfc81 : SUCCESS in 1h 15m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/40b84f45809c484ebc7d268057eee5b1 : SUCCESS in 1h 20m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/807c7a7292124b3385e0d9b36bc518df : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/611e2b27af9b4a039ac5dacb0619ea9e : SUCCESS in 1h 27m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6bf5dbbebc0b41fb856363230093fc13 : SUCCESS in 2h 01m 51s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f33f2760d6bb4e1580ebaa324c75b7c1 : SUCCESS in 1h 48m 45s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18f1a1332e9f4a7b8be5c1974b9a9881 : SUCCESS in 1h 34m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6c4eac0c09e45d8b36d9a826bbea48c : SUCCESS in 1h 38m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/854afeb4e6d5438982051e5bdd849020 : SUCCESS in 10m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/43964dfc711c40af8a4d26c1cbc1eaab : SUCCESS in 18m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5f2ad7b6f504c83908fb95f7c389ffe : SUCCESS in 16m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/09b76dafe90d40c4958196f7dd6c5faf : SUCCESS in 13m 55s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c3e5517cbd434e34b698afd6611d921e : SUCCESS in 28m 48s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7c44f8d99e8e452cb1438e8afefe685a : SUCCESS in 38m 36s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b77543f5a9624543942e9c14a1eebb1a : SUCCESS in 32m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/843c30b99dbf4df2911902b4955c3b12 : SUCCESS in 22m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a0c007b0dc96465d8ae973a88920f9d7 : SUCCESS in 15m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e34b67f0826490fb8bb0ea30bc624e4 : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":17},{"id":"5901f0e2bad93e85491dd501b9670bed967a7462","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-01 10:51:05.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"a79066cac1c59e9ac09050cad95606a69391f964","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 10:51:23.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"22a43d8f7e277acb240bd79abfb8362eca7937fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 12:37:20.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a534314f10224fbc982495e3ef6f3829\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82c85903cd34a75b3ffd4585eaa2a40 : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/339b884486f54ad38220095a20ae89c3 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdb785c09a847959685211010052ee3 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d00fcc26678a4890a8aac9a9d1763562 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f548821bcdb2423dae3ec0bd4b02f572 : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c1044a13bfd4df2aa3ba5923dfa2f9f : SUCCESS in 15m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d787a23c134a44e882b6f91ab2ed4627 : SUCCESS in 7m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/341bfd857aeb47e0ab581c6a8d919f0e : SUCCESS in 21m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80019f59fda94484bd7fbd3fc4bb20aa : SUCCESS in 11m 18s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a6c709dc1c524643be97bb0246e5c113 : SUCCESS in 19m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb60dce67ef747b390cc2bc3a141c5a0 : SUCCESS in 21m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/642067efa2f1436ca9a6791e8d2d6000 : SUCCESS in 1h 03m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fcf4e0767e594a3abd80ca01504a0757 : SUCCESS in 59m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/46d62771528f46cc81e9e84a1b4835b9 : FAILURE in 9m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c4b11e7d2544ff9bae11ba5247162da : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4335bb699710435db32ae2ef26bb6f6b : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e25e34e6bddd47198fb04499ca7ce36c : SUCCESS in 1h 45m 08s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ab892fc3642d4751ab226fcebef64b5a : SUCCESS in 1h 38m 45s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc7592f0115b433f8c279ad230b0a10e : SUCCESS in 1h 33m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c458174c3f374c45871b5bec77fa6e5b : SUCCESS in 1h 34m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6515fc4101434833b15c35aea39d3b31 : SUCCESS in 25m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ce575c335e2c49c5a85b15471c7df3f8 : SUCCESS in 24m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bd192a2ea9e43aa8043ab1cc18c4e6d : SUCCESS in 9m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6c5e0efc135f442c9c6c65cdbd0f86d0 : SUCCESS in 17m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1cfaad5fe53545c3a0009b63da8f5139 : SUCCESS in 26m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e427551ad91c4d75a21ca1f3a905b3c5 : SUCCESS in 19m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3b531a6df764de1b37279eecc2a7a83 : SUCCESS in 40m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dec44bedb854be3b55fa0fab54580a7 : SUCCESS in 30m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/270652429e934b639f543acf33bdc6c2 : SUCCESS in 21m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbf68f68e3c243f283221d7e26c7479e : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":17},{"id":"0a4f9a040560b4ad71159667c5302c9cca8c2b08","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-01 12:40:46.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"3ee6b30a83c88072d22fcef50814d6672267c7bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 14:21:29.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/322e263bc0314c498967b85a996c2503\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d66bb05916f40d6bdc4bfada76ee086 : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/12f91d7854a64f249f7a225ebe2f0ce4 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69bc6abba36f4a76a38f98e4fc736d66 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb137d45d3944d47b175184bb735b4fb : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20d3d42534a34c41b288dbdcb5c19608 : SUCCESS in 4m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/21b493ad2604450b92e8113fd74c330d : SUCCESS in 4m 27s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28b74bfce49641abbaa74779e3e19b1a : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e74286e186014f7cbfaa29265a547dfd : SUCCESS in 12m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c9c118efa26543c881f67ef3c3fffb58 : SUCCESS in 10m 49s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/373ee8b98d09472b864e7617fa34b2e1 : SUCCESS in 19m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50a722eee41f4527818e883bd68080dd : SUCCESS in 7m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/875ea2da2c3e41ac8d4c53e82583668a : SUCCESS in 18m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30fe2a2cae844bfb9e16abb4dc5be59e : SUCCESS in 17m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6cb03bdd76f341bc886e1695a86d219d : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/df0c8a92348a42ba99f1a80e628716a3 : SUCCESS in 1h 02m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fd631565d75b4ccabf7a6e0f2648790e : FAILURE in 54m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/188df6e82ace4676b79af1219e40f862 : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c9e9e93222a342498814dc89144d352f : FAILURE in 38m 57s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/12813eb0d10946f6acb70df4d9773f60 : FAILURE in 15m 17s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2f338839a8024d22826cdc98c719a0bb : SUCCESS in 1h 39m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17c6132bd9454d489fbd99fd74d26365 : SUCCESS in 1h 39m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d42bfe0ca874848861e27f9e4297eca : FAILURE in 13m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0ff773b8b56042e789b057cc7e7933cd : FAILURE in 9m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/786b4848d3a048fc90d2380b9b539f89 : FAILURE in 17m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26ad7186a3c242eb947c48a122ebac76 : FAILURE in 19m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/733a24311bd145e5961c9fd633b6f83a : FAILURE in 14m 50s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2c66cecb1d4449d2963d9565277b4a83 : FAILURE in 14m 48s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d327d9cb2a2045a9af9018a1764f793f : FAILURE in 18m 32s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d0bfeaa610a4f31a1079205796d32d9 : FAILURE in 26m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dc142315f2043349ebc10ec1064e563 : FAILURE in 12m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3a806bc626d44386994d3d3e0a6670fb : SUCCESS in 16m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f15c980cbe24d23b6a48f023ae57e29 : SUCCESS in 7m 27s","accounts_in_message":[],"_revision_number":17},{"id":"afe73e4557a6d1e90fb2d9df4202286e4a0171cd","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-01 14:43:29.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"3f5d0f250432d121b3db7112d845f653b22c9445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 16:24:38.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92704d34a7a44ece856d7ae0d5eb5d02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c169eaea5225427397f61012511ed456 : SUCCESS in 10m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aa57fd286cdf4114a0aaea615a0d913f : FAILURE in 4m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/949bb735d39f424e8b52c9b660875605 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fff7b5b951ca414191606dc035a0db6d : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/353b4aaf75f64781b984cf71099ca94a : SUCCESS in 2m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/16f564fc8f22430f8065cf962a23cb41 : SUCCESS in 9m 05s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/97258114e8f943f081ae62a068f593fb : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/acd217b482514ea5bcb2cfbd1df25bad : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1767d2aae05547659821d0a281ef9a48 : SUCCESS in 6m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/615d800319d542769f51de598d55f029 : SUCCESS in 20m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/210ede4042164d6dadd32ed03391a544 : SUCCESS in 14m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9f2e9cd36f8d4b6297dbd9de93fc34d8 : SUCCESS in 24m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b95e86fa0d045ac83f5afbed1a4614c : SUCCESS in 16m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/691e04080a5e4dbfb1be7369dcb2a5b5 : SUCCESS in 1h 00m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0d55422183724b99991a152ec4a1cd4b : FAILURE in 38m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d2c1c80df920459bb31a7bdb046a0c8c : SUCCESS in 1h 01m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/946d21ec3b314514bb1d140bef038445 : SUCCESS in 1h 06m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d7b8411b2d54a8fb86f0341d520fbdd : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ac759394dfea47558a7314b7f59da33c : FAILURE in 13m 39s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fc057c2b129941d89c3964bbfd6f10e0 : FAILURE in 14m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c04fda9568f943269687411176979faa : SUCCESS in 1h 39m 57s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7445f1da59da439a8b89124e92c0f255 : FAILURE in 14m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6f42c97541af4531a9c7cba0d99da28c : FAILURE in 10m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/66c8e9f6be6d43d19a095f774203b708 : FAILURE in 9m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7148c6f0884249f7aba8c73739abba98 : FAILURE in 9m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/154862f3659742838cc2f9f6aab294e6 : FAILURE in 16m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/12603766327741bf819b8e024eba6625 : FAILURE in 13m 19s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3d898ef154834b4095a214529ebd77d4 : FAILURE in 13m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8256761f550b49a58725939a299b6942 : FAILURE in 11m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f443c004edd84d6cbdf072044d74abee : FAILURE in 18m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2181d04139104a2f9ffdf4e3440b55f4 : SUCCESS in 22m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80b41bc8b9c7486eaecfce07d75466c2 : SUCCESS in 15m 43s","accounts_in_message":[],"_revision_number":17},{"id":"366fedfe6c8cb7740332986a6ca755c2ce832b20","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-02 08:44:49.000000000","message":"Patch Set 17: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e0ee87123f3f1de204157e18558fda4470abc934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 10:18:45.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcf141431a4c4187a6160c7c75bba061\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b68acef3d1b74f2eace7e11b6728d0be : SUCCESS in 8m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6cf53de3744c4ecd9dededaf08ce0387 : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/018efb4317af44e59c488f0d8b402409 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc8e21ad2fd64466b712ac8ebf45a9f2 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4645bcfcc6bf4843b1ec2db009a21cc1 : SUCCESS in 6m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd10c45959834d6e9b8b87186cbd382f : SUCCESS in 9m 30s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c3bcd2170184aabb421cf624532798e : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5377c91c5b76477da14e9274cc8ba14d : SUCCESS in 5m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8533d4559844ec59367431f75e6125a : SUCCESS in 11m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a4271d7173064972a80a756d24c13c28 : SUCCESS in 9m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d2cd5399dab4860ad10d4926c412097 : SUCCESS in 15m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bc9f6754783c43a9b64a7e65bc8320ae : SUCCESS in 9m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f124407ad854e2fa7befec1fdc0696e : SUCCESS in 21m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/455d92bc23f74554b5a774f446ae9bd1 : SUCCESS in 59m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9190bf2dc1ca4c2aad504e99d795780a : SUCCESS in 55m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/577e6a49dd5d40c38e234899ed203141 : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d97107ba4e74f1bbe9659a162775cd4 : SUCCESS in 55m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3d0d54a98119475b92c45207af4b17db : SUCCESS in 58m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c8ee54a6637445c49047dd1718e68627 : FAILURE in 1h 24m 18s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6ede09f6d1d64a3ba18f46820d5fea70 : SUCCESS in 1h 28m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d49db5bb8f664a7f9519b33831984e2b : SUCCESS in 1h 26m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c6ca603795a4339953564bd6f546243 : SUCCESS in 1h 33m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f774bedd89b24042879f7a01b29aab4c : SUCCESS in 10m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/398bc42c4bae472b88cca06f17cbcedd : SUCCESS in 9m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e084e67a0f6343b79c320ad7069d637d : SUCCESS in 16m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/885219fdb9044f638d158656b0cca9fd : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/21e642e9446442e38ecba0621e040e6c : SUCCESS in 15m 44s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b381b102bed64799b5d4da56a600af42 : SUCCESS in 17m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cad8aa0780214c37a7c3ac2515bb31a2 : SUCCESS in 32m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d2f841d68554d148e1a091a77de1a6e : SUCCESS in 27m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d72bb52a001b4849b6600a9b41a6916d : SUCCESS in 20m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb0c574bc9ae4b84ba5f97ade251f563 : SUCCESS in 16m 21s","accounts_in_message":[],"_revision_number":17},{"id":"5aac76d9911b2fa7129bc3fb8b86065b6e31efc0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-02 12:55:41.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a08d272e1e9b5e2aa164e60772539644c187625e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 14:42:09.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b8ff89eb1084775846e430dd87251e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba27d86248694d9f9e50ce8fb3a9a3be : SUCCESS in 4m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c40dd88c624843d6840decc9b3b678f0 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fc709d90f64409b8c4d51012e0053e7 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/938e5176a7ce40678717f8abd1554392 : SUCCESS in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c17ee9bcc264b9c95a3467b1f6349b6 : SUCCESS in 6m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5afeaa95053447adb3d39581ded859c5 : SUCCESS in 5m 32s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a1f0358f93b0441e902b1190c80e382d : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1aadd17b214458b9a7b4e9c9e2e39c4 : SUCCESS in 12m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e491612740fc495b938f89087d6a7c31 : SUCCESS in 11m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5279f0c449ca4583865ae7f37a8208e9 : SUCCESS in 15m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cab1fb8c354f46c09bc6a225c74ac5a0 : SUCCESS in 14m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6948f28013844215b68dc8f75d82d5ad : SUCCESS in 12m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f201c52ce124760843985999bd1a795 : SUCCESS in 18m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6f50e06a3a0846f08775833bc2e6ba91 : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7ae82ac51c4742278d6267a6ed42d3b4 : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c59e334418104388af3d17583003bb63 : SUCCESS in 1h 04m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53e22d615554459fa49ce1eb30cbc0f0 : SUCCESS in 1h 02m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0260016b46be41809e44ee0592645e79 : SUCCESS in 1h 00m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3b8cdaae72a64895b764b28631ff7561 : SUCCESS in 1h 39m 34s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d296a8f0116147e4a756fe9d7c6c9511 : SUCCESS in 1h 34m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe7c089df97942dc8eb0c2fff6fb0852 : SUCCESS in 1h 31m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fff773e64784faebc4f72eb1d6be31d : SUCCESS in 1h 45m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dfb1b6f2cf284948bf396fc87e693271 : SUCCESS in 17m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8813ec27b5444551aa369f4aa21ce9ad : SUCCESS in 10m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9525472c588f478391eb96949c0ee83b : SUCCESS in 8m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5dd0439c15146c683a81a4b777f9451 : SUCCESS in 20m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d2d42e17ea4445fbba5e3106a37c95be : SUCCESS in 35m 22s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5dfbf3ac235c40c4a69a5724797bd7e8 : SUCCESS in 18m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54cd1ed41cd64baa859ce26b28ef8d5e : SUCCESS in 29m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d7451be13934e0ba2b49fe64d58b5be : SUCCESS in 27m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/94c457cc0c8d4946a258c15365620354 : SUCCESS in 15m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1479331b0ccf469ab6d6221fa89b9414 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":17},{"id":"13ae297d7c1b1e59d1bf9e95bbcec73f8032fc83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 14:42:25.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"c6823fe03dac8c7846a87ce8716d7abb4f38bcf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 16:24:29.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7174824846864898ac4ab2a57761d37c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b08bd3a022a4bdf9845826001f9b167 : SUCCESS in 9m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/533c87803bfd4f61a6796876f36ff163 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b665aa400f74f4f87a65ced750142bf : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f10c2c03863f49758ef518ad990d7ac3 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/abd0d2d94df344acad0b6c9626224da9 : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/906b29ee1afe43a991c3a55b2b2b273d : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1e99f77854d8426b899883be588a90a7 : SUCCESS in 8m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2bb7a8540e734f07b908981d243ba9ef : SUCCESS in 10m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d6534850d8d4183b4f64a1c099c367c : SUCCESS in 16m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/76f370db8fd34e5ca336a896c55508a9 : SUCCESS in 10m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00ef383aebc94d3482734b052a88572a : SUCCESS in 16m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0fdd22ae7bdb4a90abf64ace6fc193d9 : SUCCESS in 59m 45s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/437f2a9890994a198359f2abf9d1e90a : SUCCESS in 57m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5ba8d46c5bf348d3a54727a11cd9db11 : SUCCESS in 1h 00m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4aa90a6b65b43d09f4cbdeb40bce571 : SUCCESS in 58m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0b10d7bbaafc45c095a67f7b73b8b27b : SUCCESS in 1h 00m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e39b317158cd4856b09a1c91c2bf7e87 : SUCCESS in 1h 37m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d43969e01d344ef39868b104f0a44453 : SUCCESS in 1h 32m 07s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12f32fe424ad46c4afe58633be1ade59 : SUCCESS in 1h 31m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76f50dc9139146ef81ef828af3cd99ef : SUCCESS in 1h 41m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b11bd661525346939e49ba572db49416 : SUCCESS in 16m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b08d8e06713746b98552e668bd30b64e : SUCCESS in 11m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51fed2640efc40c2bd24c4349ee5c295 : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/708c9c0be069404ca9b28ac24dfc23c0 : SUCCESS in 16m 12s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dd905e1857474a6d92c8312a48120d7c : SUCCESS in 30m 24s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f0a0044513ab4b369229df059a00f94e : SUCCESS in 18m 44s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb0fd829a14e4df6ab56fa9693fe4559 : SUCCESS in 35m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bedeef40dadd4e1f9169d14c57bd1a36 : SUCCESS in 15m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/315de02e5d17413e97614cc82072c8e4 : SUCCESS in 21m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56ab4c2161784ab1b9cf688623552b17 : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":17},{"id":"402c268e8c3d41aaa43d54a2c2b667f64ec1674c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 16:24:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"bbe40c2d183b466c4d07f16b5d3a24a1974311b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 16:25:59.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bc8e45113a641c995f7f92899fccd07\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/566b1b941f964bfb957d60914394d895 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"229b1efc12d5ff2d60f2225a42e0c657ea6a7924","revisions":{"ed0bc53b9c1120dadb517cb4263cdab709d83ac9":{"kind":"REWORK","_number":1,"created":"2026-02-12 12:39:07.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/1"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:38:45.000000000","tz":0},"subject":"WIP: Fix no package debootstrap","message":"WIP: Fix no package debootstrap\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed0bc53b9c1120dadb517cb4263cdab709d83ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed0bc53b9c1120dadb517cb4263cdab709d83ac9"}]},"branch":"refs/heads/master"},"cb12a10c30298d1c78e0414c59e0886af5d4722b":{"kind":"REWORK","_number":2,"created":"2026-02-12 14:43:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/2"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 14:23:33.000000000","tz":0},"subject":"WIP: Fix no package debootstrap","message":"WIP: Fix no package debootstrap\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb12a10c30298d1c78e0414c59e0886af5d4722b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb12a10c30298d1c78e0414c59e0886af5d4722b"}]},"branch":"refs/heads/master"},"2e1ffbbaadf1a8e61c0bc31840ba399a3c573658":{"kind":"REWORK","_number":3,"created":"2026-02-12 16:26:30.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/3"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 16:26:23.000000000","tz":0},"subject":"WIP: Fix issue installing debootstrap","message":"WIP: Fix issue installing debootstrap\n\nEnables the CRB repository so that we install the zlib-ng\npackage which is a dependency of dpkg.\n\nCloses-Bug: #2141684\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e1ffbbaadf1a8e61c0bc31840ba399a3c573658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e1ffbbaadf1a8e61c0bc31840ba399a3c573658"}]},"branch":"refs/heads/master"},"e26a99e33d2746bbe1c86241cc1c6b24e7b4ff1d":{"kind":"REWORK","_number":4,"created":"2026-02-13 17:24:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/4"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-13 17:24:15.000000000","tz":0},"subject":"CI: Fix issue installing debootstrap","message":"CI: Fix issue installing debootstrap\n\nWorkaround bug in CI where the crb repository has been removed from\nthe standard repo files.\n\nCloses-Bug: #2141684\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e26a99e33d2746bbe1c86241cc1c6b24e7b4ff1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e26a99e33d2746bbe1c86241cc1c6b24e7b4ff1d"}]},"branch":"refs/heads/master"},"857eb4c475fcf1f55087e5a1002c729e76c63c1d":{"kind":"REWORK","_number":5,"created":"2026-02-13 18:20:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/5"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-13 18:20:40.000000000","tz":0},"subject":"CI: Fix issue installing debootstrap","message":"CI: Fix issue installing debootstrap\n\nWorkaround bug in CI where the crb repository has been removed from\nthe standard repo files.\n\nCloses-Bug: #2141684\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/857eb4c475fcf1f55087e5a1002c729e76c63c1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/857eb4c475fcf1f55087e5a1002c729e76c63c1d"}]},"branch":"refs/heads/master"},"ef9b0dcc2e8d1a6b434cc963705eb03cd4594d8d":{"kind":"REWORK","_number":6,"created":"2026-02-13 19:18:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/6"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-13 19:18:15.000000000","tz":0},"subject":"CI: Fix issue installing debootstrap","message":"CI: Fix issue installing debootstrap\n\nWorkaround bug in CI where the crb repository has been removed from\nthe standard repo files.\n\nCloses-Bug: #2141684\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef9b0dcc2e8d1a6b434cc963705eb03cd4594d8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef9b0dcc2e8d1a6b434cc963705eb03cd4594d8d"}]},"branch":"refs/heads/master"},"f256f2151f7f11f51473c8991ebc23e911fcbcf6":{"kind":"REWORK","_number":7,"created":"2026-02-13 19:25:43.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/7"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-13 19:25:36.000000000","tz":0},"subject":"CI: Fix issue installing debootstrap","message":"CI: Fix issue installing debootstrap\n\nWorkaround bug in CI where the crb repository has been removed from\nthe standard repo files.\n\nCloses-Bug: #2141684\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f256f2151f7f11f51473c8991ebc23e911fcbcf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f256f2151f7f11f51473c8991ebc23e911fcbcf6"}]},"branch":"refs/heads/master"},"161682bbd7020dfe3ab9bee103f5c06b0d295982":{"kind":"REWORK","_number":8,"created":"2026-02-16 09:27:17.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/8"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-16 09:27:17.000000000","tz":0},"subject":"CI: Fix issue installing debootstrap","message":"CI: Fix issue installing debootstrap\n\nWorkaround bug in CI where the crb repository has been removed from\nthe standard repo files.\n\nCloses-Bug: #2141684\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/161682bbd7020dfe3ab9bee103f5c06b0d295982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/161682bbd7020dfe3ab9bee103f5c06b0d295982"}]},"branch":"refs/heads/master"},"829143e57ced8733ce2c9eb8680e6996a5113d86":{"kind":"REWORK","_number":9,"created":"2026-02-16 15:13:12.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/9"}}},"commit":{"parents":[{"commit":"eb738b110023f3a257d5f55d9c1c20c481a26dc4","subject":"WIP: Bump stackhpc.openstack to 0.9.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb738b110023f3a257d5f55d9c1c20c481a26dc4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-16 15:13:06.000000000","tz":0},"subject":"Template disabled repositories","message":"Template disabled repositories\n\nIts useful to also template the disabled repositories so that they can\neasily be renabled with dnf config manager.\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/829143e57ced8733ce2c9eb8680e6996a5113d86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/829143e57ced8733ce2c9eb8680e6996a5113d86"}]},"branch":"refs/heads/master"},"110e095be9147e8d7f53a4450f7003a61dc7e415":{"kind":"REWORK","_number":10,"created":"2026-02-16 15:54:01.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/10"}}},"commit":{"parents":[{"commit":"eb738b110023f3a257d5f55d9c1c20c481a26dc4","subject":"WIP: Bump stackhpc.openstack to 0.9.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb738b110023f3a257d5f55d9c1c20c481a26dc4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-16 15:49:47.000000000","tz":0},"subject":"Template disabled repositories","message":"Template disabled repositories\n\nIts useful to also template the disabled repositories so that they can\neasily be renabled with dnf config manager.\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/110e095be9147e8d7f53a4450f7003a61dc7e415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/110e095be9147e8d7f53a4450f7003a61dc7e415"}]},"branch":"refs/heads/master"},"da4cafe5cbf43209833b6e1e874b2c0f4592c691":{"kind":"REWORK","_number":11,"created":"2026-02-16 17:54:47.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/11"}}},"commit":{"parents":[{"commit":"eb738b110023f3a257d5f55d9c1c20c481a26dc4","subject":"WIP: Bump stackhpc.openstack to 0.9.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb738b110023f3a257d5f55d9c1c20c481a26dc4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-16 17:53:09.000000000","tz":0},"subject":"Template disabled repositories","message":"Template disabled repositories\n\nIts useful to also template the disabled repositories so that they can\neasily be renabled with dnf config manager.\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da4cafe5cbf43209833b6e1e874b2c0f4592c691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da4cafe5cbf43209833b6e1e874b2c0f4592c691"}]},"branch":"refs/heads/master"},"a5dfb8e6c4afc1afeb5761bb1b86d55e1d3d5b90":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-02-16 17:56:25.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/12"}}},"commit":{"parents":[{"commit":"0353bf8a3d731081cf8ef76fd452a1b9e19b51a6","subject":"WIP: Bump stackhpc.openstack to 0.9.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0353bf8a3d731081cf8ef76fd452a1b9e19b51a6"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-16 17:56:16.000000000","tz":0},"subject":"Template disabled repositories","message":"Template disabled repositories\n\nIts useful to also template the disabled repositories so that they can\neasily be renabled with dnf config manager.\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5dfb8e6c4afc1afeb5761bb1b86d55e1d3d5b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5dfb8e6c4afc1afeb5761bb1b86d55e1d3d5b90"}]},"branch":"refs/heads/master"},"8301e0e389d066d40340005f78e8c6fef4baeaac":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-02-23 18:58:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/13"}}},"commit":{"parents":[{"commit":"7c70a74d5d6527db204579f447e911d4ac01b861","subject":"Bump stackhpc.openstack to 0.10.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c70a74d5d6527db204579f447e911d4ac01b861"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-23 18:57:51.000000000","tz":0},"subject":"Template disabled repositories","message":"Template disabled repositories\n\nIts useful to also template the disabled repositories so that they can\neasily be renabled with dnf config manager.\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8301e0e389d066d40340005f78e8c6fef4baeaac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8301e0e389d066d40340005f78e8c6fef4baeaac"}]},"branch":"refs/heads/master"},"68593e068ebe400719a66b8ed906503820f20cc4":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2026-03-11 09:24:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/15/976615/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/14"}}},"commit":{"parents":[{"commit":"7c70a74d5d6527db204579f447e911d4ac01b861","subject":"Bump stackhpc.openstack to 0.10.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c70a74d5d6527db204579f447e911d4ac01b861"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-11 09:24:33.000000000","tz":0},"subject":"Template disabled repositories","message":"Template disabled repositories\n\nIt is useful to also template the disabled repositories so that they\ncan easily be enabled with dnf config manager.\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68593e068ebe400719a66b8ed906503820f20cc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68593e068ebe400719a66b8ed906503820f20cc4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"40901368f2c24e40b8dc34588d193502bc1b000e":{"kind":"REWORK","_number":15,"created":"2026-03-17 13:46:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/976615/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/15"}}},"commit":{"parents":[{"commit":"cc33e0e2a38ce4030c82a08f8aa926fe8eb4886c","subject":"Merge \"[firewalld] Fix broken conditional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc33e0e2a38ce4030c82a08f8aa926fe8eb4886c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-17 13:46:02.000000000","tz":60},"subject":"Template disabled repositories","message":"Template disabled repositories\n\nIt is useful to also template the disabled repositories so that they can\neasily be enabled with dnf config manager.\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40901368f2c24e40b8dc34588d193502bc1b000e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40901368f2c24e40b8dc34588d193502bc1b000e"}]},"branch":"refs/heads/master"},"ae440aa5b9c536f1c1dbfd30487ec5f351d8cc27":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-03-30 19:48:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/976615/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/16"}}},"commit":{"parents":[{"commit":"5e1d74ae4b123e3695416ddbb0cdcbc807da4fd7","subject":"Merge \"Drop backwards compatibility for template trusting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e1d74ae4b123e3695416ddbb0cdcbc807da4fd7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-30 19:48:05.000000000","tz":120},"subject":"Template disabled repositories","message":"Template disabled repositories\n\nIt is useful to also template the disabled repositories so that they can\neasily be enabled with dnf config manager.\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae440aa5b9c536f1c1dbfd30487ec5f351d8cc27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae440aa5b9c536f1c1dbfd30487ec5f351d8cc27"}]},"branch":"refs/heads/master"},"229b1efc12d5ff2d60f2225a42e0c657ea6a7924":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-03-31 20:31:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/976615/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/976615/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/976615/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/976615/17"}}},"commit":{"parents":[{"commit":"8b8ac1c293fb48b5c01fc294bfb0f6e2b313737c","subject":"Add nmstate engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b8ac1c293fb48b5c01fc294bfb0f6e2b313737c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-31 20:31:16.000000000","tz":120},"subject":"Template disabled repositories","message":"Template disabled repositories\n\nIt is useful to also template the disabled repositories so that they can\neasily be enabled with dnf config manager.\n\nChange-Id: I3d88dec15a4540c85345770029536fceef9c6b32\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/229b1efc12d5ff2d60f2225a42e0c657ea6a7924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/229b1efc12d5ff2d60f2225a42e0c657ea6a7924"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
