)]}'
{"id":"openstack%2Fkayobe~977013","triplet_id":"openstack%2Fkayobe~master~I47af03d12a66b9a9e3cfc3ab7960633a2752aeef","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I47af03d12a66b9a9e3cfc3ab7960633a2752aeef","subject":"Remove old Dell/iDRAC discovery playbooks","status":"NEW","created":"2026-02-16 17:11:30.000000000","updated":"2026-05-28 10:37:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"82c34beb92acf999a878291a7d674c7cedf10c09","_number":977013,"virtual_id_number":977013,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-28 10:37:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-02-16 19:54:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9dd8c9c78ac42229df6db953e7ff3a14d0032267","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-16 17:11:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1d1d975fdf884fd9e5381f63ecc0ad0a771d21d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-16 18:21:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69bb771d4a12576a0d965b4602f0f3c30d9d49bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 19:54:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/582b5057b6a143a191aeb672a13e0635\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5211d369a143477889ac55e89f2d5bae : SUCCESS in 8m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b44a8a6d6b1f484394dcce0dfd28ab32 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/921e3466f24d4845a9180deb16dc9759 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93721cc176514f95ae0bc057278574db : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b2c138579404edf86dec5faf515f32a : SUCCESS in 3m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c3d27ab85b343419d39892bbf4244d9 : SUCCESS in 5m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a59aa6a9b18347a99bde7cded928ef35 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3c776fe9da4749819418e9de689070c2 : SUCCESS in 12m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/905ee2fcc0e94081a2e3ab41c1a4488d : SUCCESS in 11m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/439d179fb6d1488387a7a59b70bfb816 : SUCCESS in 12m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17965f0233a648a8bf3e9bd70f8b5752 : SUCCESS in 12m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4d21269156174b4bac3946ea1c8f387d : SUCCESS in 24m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58e376f92a1543d893e1ec5f8f8a5649 : SUCCESS in 19m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/546c906e6b7c47cf910c19f2c590fbf8 : SUCCESS in 54m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/55e6357e9f294ca4ac839531421fe650 : SUCCESS in 54m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/768b0ace37ee4e08876f9e39c5ba3d1e : SUCCESS in 58m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac505198f6414c7d9b1594ebc199ed78 : SUCCESS in 55m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/78fe8a335cd34e3f88ce13c2da2dde5a : SUCCESS in 56m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9f64a869f20f4b50a2d86d347d8cdf6d : SUCCESS in 1h 29m 51s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f3f3f3f410384a04aa5dd51b6589d858 : SUCCESS in 1h 25m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b71eabf6815b4114aa72da9c2cb52b11 : SUCCESS in 1h 22m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7cb92d27c1348c7bb2074738c519f53 : SUCCESS in 1h 29m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f5cd37e5612748f8adaf225a27eabdfe : SUCCESS in 22m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/49f5e032426940bea8dcd15f803361ac : SUCCESS in 8m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55e0f86b4b274837ae0baaf2b0e8f5db : SUCCESS in 15m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/858317067cae448a82801b562c4a604a : SUCCESS in 7m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/369919a6bc8941f9b0ea12d6b1100a35 : SUCCESS in 14m 47s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cd7938faefd847ba9d802dda478a5ede : SUCCESS in 16m 06s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28f498b9bc794793b2b8845a224c2538 : SUCCESS in 33m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae5dc1ef75684517bf21750221a0c785 : SUCCESS in 27m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f8915d1bdb104abead5805d22b2eb8df : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9401afbdfb9a47999802a19be7d07023 : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":2},{"id":"155311bbb52e1a619f24e20a0c855db236cd8cf1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-28 08:54:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"82c34beb92acf999a878291a7d674c7cedf10c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 10:37:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d45f80a3d7d40e3920257acd1d1320f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06e4801c07d144f28791f008af7bf140 : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/07135a2e763a420aa88e27df35779574 : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66eaef61ca5e4fd1b4d83a2bbc1c4f8a : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/435d5b705bc24179b7a14e54f8e096e8 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95d21a6d8dd04108a865f18d8d335083 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/408a25a6996348f5906fbd9f96be8d8d : SUCCESS in 8m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcb0fbf292824a3897ebc26ecd540e99 : SUCCESS in 5m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41c962b5174e4be19526bf0539d98750 : SUCCESS in 3m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11303dcfb739431890b0f3f44752c478 : SUCCESS in 13m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5c7333e12004faeac83d9fa715bdfa5 : SUCCESS in 14m 34s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3f6028f5444a439b999eda3b77d4d35c : POST_FAILURE in 15m 58s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/39d6a37ef9b64f2ea9618d8eabd6d794 : SUCCESS in 12m 34s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03fad322bfd141e7984fea7f4ee8ced5 : SUCCESS in 16m 39s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c3dba2acaf5d4994adbd76ba7d745a0a : SUCCESS in 15m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/43cd35ca650b440db9b2547ee894c75f : SUCCESS in 19m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/209e5817da0c42ba9babc5278415ba52 : SUCCESS in 20m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f48bd609cacc40e8987ef0ce700b7bd7 : SUCCESS in 13m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffc33c01a2d44048ac769074ae2ea016 : SUCCESS in 18m 42s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f561f0c20c804505ab297d1296d750d3 : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1bd21046219c4a628baff3f459639412 : SUCCESS in 1h 02m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f22db172eb4b45f1a0f151cc769a9c7a : SUCCESS in 1h 12m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edaa35ef54834dceb14d89a55d5322fe : SUCCESS in 1h 08m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/50cad448fb744112962b5a1848609f9e : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/31244cc398d543b885baee7cfd8eb5be : SUCCESS in 1h 41m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9d41992333748ffb5a4d373d43b459a : SUCCESS in 1h 30m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dd7ad2c88b014993b4d5d3d4bd85c73d : SUCCESS in 15m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/68af2713c54042b5bbdb7ce0cae693df : SUCCESS in 16m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd11f148601347e982b2954e983d610e : SUCCESS in 14m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/365a8415fa4c4453866023c98c1d4eb0 : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c19032e32a3c45489db6c29605d43d1e : SUCCESS in 18m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18b423ef695141ea8fd3580a9df47cdb : SUCCESS in 30m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b8b993126e7a41ab9a5b452619f9c796 : SUCCESS in 26m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2149cdfa992b4201a81ca177dad92bbd : SUCCESS in 16m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"67776e8a2152f835ffa706efaeed1451fc3201ff","revisions":{"de1d8cf7e494b91739619d238bba08ef2d8f1bfa":{"kind":"REWORK","_number":1,"created":"2026-02-16 17:11:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/977013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/977013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/977013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/977013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/977013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/977013/1"}}},"commit":{"parents":[{"commit":"18d022a00f67b7ce4e7784d8a9b2a4c3bb0d3fbe","subject":"Merge \"Bump all Ansible collections and roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18d022a00f67b7ce4e7784d8a9b2a4c3bb0d3fbe"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-16 17:11:19.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-16 17:11:19.000000000","tz":60},"subject":"Remove old Dell/iDRAC discovery playbooks","message":"Remove old Dell/iDRAC discovery playbooks\n\nChange-Id: I47af03d12a66b9a9e3cfc3ab7960633a2752aeef\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de1d8cf7e494b91739619d238bba08ef2d8f1bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de1d8cf7e494b91739619d238bba08ef2d8f1bfa"}]},"branch":"refs/heads/master"},"41ba6b034e993312d2aebaf817f21d4a32ca23a7":{"kind":"REWORK","_number":2,"created":"2026-02-16 18:21:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/977013/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/977013/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/977013/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/977013/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/977013/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/977013/2"}}},"commit":{"parents":[{"commit":"18d022a00f67b7ce4e7784d8a9b2a4c3bb0d3fbe","subject":"Merge \"Bump all Ansible collections and roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18d022a00f67b7ce4e7784d8a9b2a4c3bb0d3fbe"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-16 18:21:34.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-16 18:21:34.000000000","tz":60},"subject":"Remove old Dell/iDRAC discovery playbooks","message":"Remove old Dell/iDRAC discovery playbooks\n\nChange-Id: I47af03d12a66b9a9e3cfc3ab7960633a2752aeef\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41ba6b034e993312d2aebaf817f21d4a32ca23a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41ba6b034e993312d2aebaf817f21d4a32ca23a7"}]},"branch":"refs/heads/master"},"67776e8a2152f835ffa706efaeed1451fc3201ff":{"kind":"REWORK","_number":3,"created":"2026-05-28 08:54:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/13/977013/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/977013/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/977013/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/977013/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/977013/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/977013/3"}}},"commit":{"parents":[{"commit":"86c97bacd214226084e8de5c7d075aaf94e48ec0","subject":"Merge \"CI: disable SLURP jobs on master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86c97bacd214226084e8de5c7d075aaf94e48ec0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-16 18:21:34.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-28 08:53:53.000000000","tz":120},"subject":"Remove old Dell/iDRAC discovery playbooks","message":"Remove old Dell/iDRAC discovery playbooks\n\nChange-Id: I47af03d12a66b9a9e3cfc3ab7960633a2752aeef\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67776e8a2152f835ffa706efaeed1451fc3201ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67776e8a2152f835ffa706efaeed1451fc3201ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
