)]}'
{"id":"openstack%2Fkayobe~977028","triplet_id":"openstack%2Fkayobe~master~I40946f21209090701fb8a2d9ccfaaff573e228ef","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-02 16:24:33.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-01 12:05:44.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-02 09:31:00.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I40946f21209090701fb8a2d9ccfaaff573e228ef","subject":"Adds dnf_repo_state_overrides","status":"MERGED","created":"2026-02-16 19:25:30.000000000","updated":"2026-04-02 16:27:13.000000000","submitted":"2026-04-02 16:24:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977028","meta_rev_id":"20e2090b087f52d17ad154a17538e0bec2dcaf3f","_number":977028,"virtual_id_number":977028,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-02 16:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-01 12:05:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-02 09:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-04-01 10:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-02 09:31:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-04-01 10:53:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-03-31 20:31:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-16 19:47:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-16 20:28:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-24 18:34:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-03-17 17:05:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-04-01 08:19:52.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2026-04-01 10:53:12.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"6e476f42f140dc3fd7a3b45dd0a0bc3d26e0dbfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-16 19:25:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aca61c04447d397a01b1982f929a0f1df915568d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 20:28:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfb73b260d604d08b09dd4459dff4ef6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/101959c7b2054e66a5c3527aa2c4200c : SUCCESS in 4m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c71ad5a28530416c959fcc4a34d43a65 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ada6e3beb1f4de9bddfed6319dfc825 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c2f10fa75e410b91a72066800815fc : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0d92325480542269c3b302c6320f5a3 : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fa68f08c70d49069a217f66fc2edbe0 : SUCCESS in 3m 05s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/da74b6d7b1b74445b4a39070a4b09db9 : SUCCESS in 31m 19s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/615c3e25ec174b77b36944f56111404b : SUCCESS in 32m 53s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/77ef9bac6c7b47a2b83e5414fa819175 : FAILURE in 23m 23s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/784c06515146467ca484af4e599dc66e : SUCCESS in 57m 33s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa88bbb2ca0347548ced939da6574ce6 : FAILURE in 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1cbd8de0c1c9bccb04c95a8b1005c87d262ea8bb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-17 17:45:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"559daec907d20609077df65e2a7c0edebd91f65b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:53:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43cb780c35ce408798671eaff96c4dc4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b89a1c0acc94d9eb8814779383368ab : SUCCESS in 4m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7604f294b8f246ffa630f9aab9b45acc : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/629f6a9142cc427eab1df2130e133481 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f473c0262894a79a6e904caf16dd98d : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/016d60c191c9498da7b1fa6a45e295a3 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64adb5654d9f45e6a6d4fff9846ad9d3 : SUCCESS in 3m 41s\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/f4d441d037b247de850a1e11329579b8 : SUCCESS in 50m 23s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/df43436c93d646f486e2c92ac6a6f840 : SUCCESS in 1h 00m 37s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/5d183f106eae4831b7df29decce42abe : SUCCESS in 37m 27s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55b770d1d5cf4b86bd6f95d2d247a741 : SUCCESS in 35m 56s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e30a23cc4ff5406cb30f4575321e89dc : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7e155b8c51b496b365b7268b36540eedcc810b24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-23 18:58:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"797d686d06044b1b4bab2028be019f5f229a40be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 20:01:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c8bad328e9340bc9221ce3f3f2c06a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869b3b5c72b6455099993404e76c800a : SUCCESS in 7m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/23bf9c8ff5a44081ac4726262776dc0f : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1193beed48f4d90be4db6c7e9ec6aa5 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d0617686554eb786a9940ecf44043c : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/367ab659f8754141ae699b4f9766b2de : SUCCESS in 5m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b97d078cfa7042caabfd6acede5f1e98 : SUCCESS in 8m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbae1610e8e44aecbf3f69ba87fe7b36 : SUCCESS in 4m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/21d535d4f9664e9bbef6fda8d00e7672 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0b74b5e0c4bd46cfa322f7dbb93ff6c2 : SUCCESS in 6m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad93324615e440c685a340e781893879 : SUCCESS in 9m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b3e33490ca154f3cb20a19394790338e : FAILURE in 7m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe18581df65447279483363757a02007 : FAILURE in 12m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/74789dde93e148fca7cbbcbd915da1b3 : FAILURE in 16m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d38b7846a3f4b62945d7fba3e28a325 : FAILURE in 13m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a285f0cc47734f3484014a31ed84930a : FAILURE in 10m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0dffe56009f24c02a83a1149b46acd73 : FAILURE in 9m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/89e1242a865e4dbdae27507c0c9721eb : FAILURE in 9m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b464d4e378144f6b3106906f7b863d2 : FAILURE in 9m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7ca8ac93ed74b26a579d2dd8148b41f : FAILURE in 9m 54s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/36825b88d9174486a25386e06e1d0d0d : FAILURE in 58m 11s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8a3aa5130ccd43bdae8b4f0609962bdf : FAILURE in 55m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84f81bae68b94d3db6b0eaec49dd96f8 : FAILURE in 51m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0eb72c0df23406794a50a3a49f5df5d : FAILURE in 55m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/eed32f25eac14c38a650a1b30c789f2e : FAILURE in 13m 13s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1f971782f6054b22a2a39a59e9042db0 : FAILURE in 15m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33e5cbf87b7d4f31bacf42237465f822 : FAILURE in 12m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3bdb942d63cf4b9dbb8bfe7855b4a2e7 : FAILURE in 7m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0bc7769f5e344de5a51544229b57745d : FAILURE in 25m 24s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ab4634e9989446bab9142e06f97ee944 : FAILURE in 34m 27s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fde4cfb113645a798a5b06a63fb5dee : FAILURE in 15m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b153b74707c46068f543e818852c866 : FAILURE in 40m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d95dafba678e45a3822620b76a7194ce : FAILURE in 14m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5059dcaf768c4ea69d421e166d680ee4 : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":2},{"id":"a4393d3e3318436231244eec09da43f4c4435496","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-24 18:34:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d5e8c328f50bf3092150e651400124ccc1f2f4c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 20:05:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b17529c82e8346a5bf187edd90caf1c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653aa3170c094927b118c7a4e07bb5ee : SUCCESS in 4m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4d634ec27a72436da4b118f8394422f0 : SUCCESS in 7m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/098b52df04494c3eb34a36bf3dbaa043 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8bfd7915a3a4fa1bfab84e8c6ed6e97 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44cfffe2326a47a7b1cab74e8f49c451 : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee6c7b0ff388472d9ebb24209792aaaf : SUCCESS in 6m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/179db2f817454c59b316ae7f73b6560b : SUCCESS in 2m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/09538b189aa9413784ecfdd35bf4bb79 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7aa71e1b800e4525bf534bd4fce3fee0 : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/880f80a2b5e8415ab0fcdc521707920c : SUCCESS in 10m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4945048096c344078b1630253828add8 : SUCCESS in 18m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1e17e6328714756bae83dd7417a12f9 : SUCCESS in 13m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0fc78f0f49fc45679a779cfb44ceb30c : SUCCESS in 11m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00f8259595cc456c90b118e273ddd553 : SUCCESS in 13m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e42e842ae8ab4192ba0cfca0e6267f41 : SUCCESS in 55m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e8bc1f162d2b490b89d4f70504cd7f2b : SUCCESS in 54m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a438cf3ea70c490b8323118ced085709 : SUCCESS in 57m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68aafbd403bc4a7989187432ef801007 : SUCCESS in 54m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b24bac33084f4decb345e7c805029333 : SUCCESS in 53m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f69b4eaf574e4a66954040fd8734a91c : SUCCESS in 1h 29m 29s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c977f5c5074a445a83c20bea6895c980 : SUCCESS in 1h 26m 25s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86b5f94eff584265af51b3dde55f4a9d : SUCCESS in 1h 22m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69bff088cecc49448b60807d870e26fe : SUCCESS in 1h 29m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/66be0d158fd24fa686d9b499c6e60d03 : FAILURE in 14m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f811aff1049d4350914f2327638d6c81 : FAILURE in 8m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c45f07f6ed28486d91d0d6b70ff2e1a8 : FAILURE in 16m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f155b8eaf5b34ed0b99738d05733165d : FAILURE in 14m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2bc901f5775445c699caeb4107ad2949 : FAILURE in 33m 49s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/93c3ed0ae54c45008ba2ea7445c4063e : FAILURE in 23m 21s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01138568598845f3b1891c0938eb6c0e : FAILURE in 25m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6712c83316644272be34d5615b5baec5 : FAILURE in 14m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/92e53e72ea70462e9fd015a6f8b1f74c : SUCCESS in 9m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/164bbcac587e4555ab0cd08345408b95 : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":2},{"id":"de2dcfb2965d2b4748448ec3cd3edff89cc118ed","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-04 16:00:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f38fe309760ec24301eb3a43a4a404cf4883e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 17:54:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9a66b67ae9c4825be7b01e46b3daf2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f172e291b04da59442e4bcb172a81f : SUCCESS in 9m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f6abc99a521949879a255d70919bf018 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/040fa8fac1de4f6fbaa1fcf337c8c1c8 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a6b6ba3849d487da04996d4508da0ab : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7462d3d0126d4b7a87abb59292491f89 : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ee11d878804467299c39ebd2c20804b : SUCCESS in 7m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c1c61d3fd784eed93797e7523c67291 : SUCCESS in 4m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7adf393802ce49efa690dd347abf890c : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f108e3a800944fb7810930958e51a750 : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df3122e1392e4435bd3c70e76e030c6c : SUCCESS in 11m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/810e1e7687bb41d6b5c80d956fdc0a94 : SUCCESS in 12m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b569309324ca4fe2a2b0965e0c34856e : SUCCESS in 14m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8b2fe958ae7f453ba4298a96e2593e94 : SUCCESS in 13m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e04dc9eb3a484da69dc2ce9226ceb293 : SUCCESS in 18m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9b3caf4e7d17473a8286ce544d2d9cb6 : SUCCESS in 1h 01m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f329f95c5fb341299baf266c7c097bfd : SUCCESS in 57m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/877e411137ef4f6a9200615cc0370561 : SUCCESS in 1h 09m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07f773ecb82d4de285af1dcbeec8de8a : SUCCESS in 1h 01m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2bc4191c8e30420c8fb629045d76896b : SUCCESS in 1h 08m 40s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cbeb9919f75e41c296cd3535f266e3a2 : SUCCESS in 1h 49m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/711ef0078fda4693b0299cb827892028 : SUCCESS in 1h 43m 01s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fef2136455da430e8e568ecbf261c4aa : SUCCESS in 1h 39m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0aa5fb79f78b41ab8cc14d9e6e02e916 : SUCCESS in 1h 40m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/98e362ec6d8848acb3e1c835fb87459c : SUCCESS in 19m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0107fccf0bea49ec81576ee1615d3f19 : SUCCESS in 22m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42afcfd7333e49fd97d19ff9397865f5 : SUCCESS in 18m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c7189cee2a64353b0a734ea3b179555 : SUCCESS in 16m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8f86ff684c4147069eba8c50f6753262 : SUCCESS in 20m 05s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1d5140683e204aa1953462ed3078c238 : SUCCESS in 20m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac89d907d045452eb0873b2009184733 : SUCCESS in 30m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7aa1dc03d954feb853e5b00cefe4dad : SUCCESS in 28m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/967722fe2c30448a9bb0650f2b94a28b : SUCCESS in 11m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bc2dcb347d341c2aab7f3584ba0501d : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3a30f3d716d93d75dfcd2fb53a123f1db59d261e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-05 10:19:35.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"49936be9dd4630b74d59a8f0b9b37196933ae76f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 17:05:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5ed29d43b01b16030e00ebd484315f403ae026c9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 17:05:19.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f474ffb7a592dea1a9dc74a0be43ff8993b65713","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 17:05:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1bb14b9f37de87bfda5bf961f60577fb6ba307aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 17:14:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"264f3fc5bdb5d9540551dfc0a8736a2aace3ffab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 17:17:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0133fd4ccfde9b865adc1955166695bd09ce16fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 18:51:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13b008d8626543c4b5d045ada3dc29a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec405b8b3b874a81bb4c1698e548d36c : SUCCESS in 9m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/31cc94c5e9d546e1a5ac8ec40dd77591 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c2fc6e17d29419b987c14e9eb77e9e2 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8002f21c0e9b459285064934e1597936 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a913c98901d48a2b6a0b3e3c0fddb59 : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2eb4ca2b4c2f4ec187120f21c86195ad : SUCCESS in 3m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8542c2116ed94b369863a249bae14f81 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e932bb0d03ef46b282c7d0ed1e5c25d7 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dfd058d20a2744dea8cb037d06bfd5b7 : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b87aacafead94888a160116a4ad38d31 : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/71e7e835336544b29b087ad7fa98488a : SUCCESS in 15m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/097eda26055d46099f11f919a663b54d : SUCCESS in 10m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3806ffc90e454f6c8e4851dcdde14d6d : SUCCESS in 20m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf35fe73a0bf4485a53a4f03e6743943 : SUCCESS in 16m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/eca35974ba184341ae27a6e9e404ea3b : FAILURE in 18m 46s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d66871bbcca3456faab6ecba7c25f74b : SUCCESS in 56m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cc88e1f93cbb47419d2ebebe3bbd5d25 : SUCCESS in 59m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d174b4117684eef817683223ad8e8b4 : FAILURE in 15m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/83e17b2a1c3c4d5fb443aef4b234ffbb : FAILURE in 15m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4bb755ff511e40229e08f061f04ae00f : FAILURE in 19m 22s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2631684a076142ddb8ecd5167e565651 : SUCCESS in 1h 32m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8cad5b071a54729a6c99f71b38369ef : FAILURE in 15m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e17fca9e34cc45c69b030cc0dde66539 : FAILURE in 18m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cbe6f0031a5b427eb3e04139a2f13d25 : SUCCESS in 9m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a8f712504b01419bb21ddb5a9d6405c4 : SUCCESS in 23m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d84c05accc544f1ab6acbfa07e0d3c1 : SUCCESS in 17m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/95edb6706c4a42478ae2dcb95611755b : SUCCESS in 8m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/44100932ef8149dbaa83bf1f477f886d : SUCCESS in 18m 14s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7f2be2ed1ca34e5d91dbfeb0820b0951 : SUCCESS in 19m 53s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/902971afa49e4c998b6599d7b01ae462 : FAILURE in 18m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3d6451e2cde4b3f8b1c6d4acdf4f87b : SUCCESS in 34m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8acf17e9107f45ab8b77890ea113e5c5 : SUCCESS in 9m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbe9c03100404e3cb4e22266ecfed8f5 : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":5},{"id":"8dbfb9fcf270f3da9a8bde5a7eb40c86e9af591d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 20:11:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1d5b57d76106ddcc7947d6cb27ea3a852774925b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 21:49:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/593e3fc00274480dab66a3e8d26fc49f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d3cfc808de34bada6f5e29f4b098856 : SUCCESS in 9m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/75c32a14880348918fa0e64181d21f65 : SUCCESS in 6m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6451358a7fb44bd5a68023b17535ab70 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b74394d7cfdc44039d39d00110a38295 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98a142be72cc475089d28f70957034c7 : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/10f4a8f73f0d42a8b141c04137e0a428 : SUCCESS in 5m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a8ac9c678f24e749ba56ee08dcb1d8a : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c40ed603244949f3a353b53f88372d12 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/760fada844ba43e0a6d56899146714d0 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00bc83f56d664d5699d7bc2885c99475 : SUCCESS in 9m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d0559f996cc34fa38fee52d4d04893e7 : SUCCESS in 9m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3055b63bdf54141be02bfecfef787f3 : SUCCESS in 7m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d39334abea0d46b09f00bc8da9b6b177 : SUCCESS in 19m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed0a25a528464d89a024a671060bc463 : SUCCESS in 15m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ffa728d0ae694872be791daf7f9e5da7 : SUCCESS in 57m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/71c29098ba274bfaaee4d58a70032671 : FAILURE in 7m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/331c7302df2c493f80dc77b863ef763f : SUCCESS in 56m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a289ebbf7f54a349a713fbba788096f : SUCCESS in 56m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9cd2b32bbbad4bbf8edad6c51b029509 : SUCCESS in 54m 21s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/00c81c053aec4a0ebb041d0812fe4293 : SUCCESS in 1h 34m 11s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/181b4f416b4c4e458469410bef48e4da : FAILURE in 8m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a8ec3c00e0e4869bc322f406983429b : SUCCESS in 1h 29m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ced317dcdf641eaaa2b40253472d867 : SUCCESS in 1h 36m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/608c55b905064e62976f21256f24ade7 : SUCCESS in 18m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/78ea2d69787a4500954129da965db85b : SUCCESS in 10m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/deff5ab05e2748b4bf2f6266ee01a8be : SUCCESS in 9m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dba7fc5901794f8faf07ea2c264f09cc : SUCCESS in 15m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fc037f3fe33c4e05ad65fbed8a69a385 : SUCCESS in 16m 45s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dae30fa937b840d19b2bb5d068b752ba : SUCCESS in 17m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2868ef1820d84930b15f17fca821cfe3 : SUCCESS in 28m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88c7cfb40dc34b5184cf2fbdd2365637 : SUCCESS in 27m 34s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/29a95ea8e4f54f56aa28e94655a77846 : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8e8a2ecac38491abda77679c04a1c3f : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":5},{"id":"67f62773dd89af69ef7402f3fa9f10c7b38c5247","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 21:52:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"949244b4e96a026a7000a7fdba1fad4f58dafa43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 23:33:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7f6d340b06b4b6bb5976614c4bc9a7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3415fcd439014ee195211afd6789abe8 : SUCCESS in 8m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b68d1d9cdcbc436891150c266f592175 : SUCCESS in 6m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78997172d156400da64beb918a44bdd6 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4622cd34ec2a46179ee7c71ec7370de1 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ed1ef2bb42304ffcaf8fe970ae379e53 : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/083954aded0044e18bb627e92a27ff71 : SUCCESS in 6m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcb473ffd65d455196daa4b6abd54514 : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f2548c5f1e624285bcebe6bde9d59935 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/62c29967256247e6b34bd188ee649e6c : SUCCESS in 9m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b47a32d1c9e048e5b37cb9d49abc4e1f : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4e53ed41895745689eb6f8e26fc0b59a : SUCCESS in 18m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a835d0f663b9401da2b658dbc2c794fd : SUCCESS in 12m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/64140bf11074460cada98e6235701e25 : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5b95c5b42cc4d7c8451a4dbd8af94e1 : SUCCESS in 17m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/35739a0f475343eba44e54c1092a1cbf : SUCCESS in 58m 46s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5607280527d54a9c8a6f1bba5d632fc4 : SUCCESS in 59m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/493b6653b30c43d58b01df0202eefd2d : SUCCESS in 59m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54a1f9fc7e7e42808a40e213bb6b293c : SUCCESS in 56m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f04ee53b4ba544e2a342dc4d32838b55 : SUCCESS in 58m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/04c3308c2e2c4637a995144a68e2c528 : SUCCESS in 1h 39m 51s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/31e22791da06442ba095732fdf618d3b : SUCCESS in 1h 32m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74dfa069bf5641f59ccf36ce3d80c6e4 : SUCCESS in 1h 30m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c22516bfbb6a4912ae7e4218b860d236 : SUCCESS in 1h 39m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c04099ca606b4ce4af7a1fdde04d00d5 : SUCCESS in 10m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c95e4ea66a27437fafac74320914b7d7 : SUCCESS in 21m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ca6f4fa484e4ea5bacbfdd7c84599a7 : SUCCESS in 17m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c4e57f7757c646f48f39ca96f4973848 : SUCCESS in 19m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ae0cdba28ec34a538cd13691cc98e41a : SUCCESS in 18m 30s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3f0e548fef834cb183b74c6c7cb8157b : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c59a4bccdd424866b151c9848d678272 : SUCCESS in 23m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c7ac19d42544e279f4d18536f8b4ffc : SUCCESS in 28m 29s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f51e0183edac49949b1da7200809fe48 : SUCCESS in 10m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/712e7db6b6fe4e9fafb68df1d050313f : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":5},{"id":"5e1b6d05d5fcbd6d9bceacdd2ae9b4ad3161cfe5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 17:44:24.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"69bff69b4dc462d72f38fff9e2eda2ab4a25166c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-30 19:48:22.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1d03009b2fa45d4f83eff1239da7d4aa907a7d31","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-30 19:49:28.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"396ba1086fdfa3101d43571e4b234798fc91ff24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 20:22:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddda44c63f0b4fb7b9ccad9f8b3bb8fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1be9a6a58224063a028320d76b5b6de : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/68f3ac902ffd434ca9ff15e3a35a95f5 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/470116e296194cdb81a737f1377f16a2 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354a6899ad2c4f9bb2fff742f160aeda : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83e994e86a974b9788774a79905a6494 : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1e9d26333d60451aa667837e2f25a13e : SUCCESS in 6m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adcf4f303a554d139f2ffb893d555c4c : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eae3609eb02b45b1b9abd1af4cfac9f9 : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3249f280ac334ef29d896e2e8fb2feb8 : SUCCESS in 12m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29176577023c498d8866a8d0af6fac50 : FAILURE in 6m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b7cc0093cfd747e599a70296873dfab7 : SUCCESS in 9m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/008c38150b0949669fdeaab424fa8ba5 : SUCCESS in 13m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f0ab19cfb788429dacb23fb3e44f7dd1 : FAILURE in 20m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d84f67ff0d140a1b272db6db8c75ddb : FAILURE in 19m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1978bc88b35d4c97925c4ea9326d0289 : FAILURE in 13m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ea043ee6a55e4c2aab906b0297a5c7ea : FAILURE in 18m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/48e8ed3b1a4141e5ac31fccaac764c13 : FAILURE in 24m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ee3b7e934ae4b23bd026b0c410b84f7 : FAILURE in 17m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/298a1acc396c4b31b5e75783266417f4 : FAILURE in 12m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6704b552dfd24cfa8e9156be0c8fdd3e : FAILURE in 23m 29s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/90cf283a8a4c4575bd2faf1b11add4e0 : FAILURE in 18m 47s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b4b699f6e8c4a6a81300e3475236dc4 : FAILURE in 22m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d103b25632149ed8356ee15541bc5b4 : FAILURE in 16m 04s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b1793d6184e64eb7853a50997746cde2 : FAILURE in 13m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3f55ef26a64742e58b7e4a8d8819e7c6 : FAILURE in 17m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bbb73f331054ae082f6a5ce5673405b : FAILURE in 11m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0ee58a08fe794286859a4d9497c91d2f : FAILURE in 11m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ecd26aaf2d3442bda7d3e0bebc7aa0b1 : FAILURE in 12m 00s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/051f0a2489734c5dba9825a95b5d3892 : FAILURE in 14m 19s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2b2b26872f6474ea0b15187ef7bcc51 : FAILURE in 28m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e06ae9cfb5648e6be5563f15a5a3fbe : FAILURE in 13m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/36240e767a3c462fb8b6e76c9f9b4cc0 : SUCCESS in 18m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f60d9b6581d4c89830b470c6b470fa3 : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":6},{"id":"cd793133d426f2e7a06d78909e1c3607ea916929","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-31 09:35:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"67bc0f1206ab6acc560e53632320e65b1f17f11c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:18:06.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c44561c7154940dbaccc905b6311dc0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79429d4074644f2cad1e8a5fa98721b9 : SUCCESS in 9m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5b074c93ec8a4941a2b87b5154752e95 : SUCCESS in 7m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e59fd4974824fd6a9a3a3308f689d5a : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca1d3d63171c444b867541b7eff8461d : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a40b7ae2c1ac49d49e77629af19d6c42 : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2564af75cc774a3fbeb829a790a85697 : SUCCESS in 7m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ecae9c74d34452f999dc424fae1a8ff : SUCCESS in 2m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2fb016b25ae84cea98f38dde1746d517 : SUCCESS in 3m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/313eed96de044a6c98e68258c926e960 : SUCCESS in 12m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/32b89b720dc049ddbf4d5fbdb8c3574b : SUCCESS in 11m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/82ca54a50b9c45cc873261899d1b3975 : SUCCESS in 15m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6344ce1f89c54e23a2fb8c616598729a : SUCCESS in 14m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6343b7159c0d4937a5be253dcc310695 : SUCCESS in 25m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3104bc7b03914d86b2acc59fed04679e : SUCCESS in 18m 26s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a407b701366643529e5eb4c8043dc483 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0b40643d74654eef891a8b5e334251d6 : SUCCESS in 1h 04m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/304fb1f07c1e453a94abfe7cff21c147 : SUCCESS in 58m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35581c19b9aa471bbfdd86a5d0dfe00b : SUCCESS in 57m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fa8c019964d54111b0e8b9148944d1c5 : SUCCESS in 59m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2f096ee7355a4821b06cf8e5cd12ac52 : SUCCESS in 1h 40m 46s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0a03049bf583404c8c81e901efb0171b : FAILURE in 44m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3230bab7a2154e99b88f0d0180979f59 : SUCCESS in 1h 32m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50973719ba2c444a95cde506f5599be6 : SUCCESS in 1h 33m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6180e7da7dcc4524a089cbc7a801d7ee : SUCCESS in 19m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/44a3c8c6c5f84f1680d40efc8fa35b3f : SUCCESS in 23m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/491bd3acd70741e18b5e0876e4c734ce : SUCCESS in 9m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/435642c0604a470ab68a503396d05632 : SUCCESS in 19m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3a0d42fb38014329b6159b14e5914969 : SUCCESS in 16m 31s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c384a81ea4b44b8e844a0d22fe6afca8 : SUCCESS in 37m 21s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/415bddb8d36142f9a4b55f3e8b459f3f : SUCCESS in 29m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6942a046b9749f4afa00404f97cc0e2 : SUCCESS in 30m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d9ef68e9957849d990efb8ea348469f5 : SUCCESS in 10m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6652f750ba240c8baaec686110bb7b7 : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":6},{"id":"19523d0b8585a29b95cade582b64132af6ced4ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-31 20:31:25.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"8a11b89ab4419c8cf9abb2fb53f2fa6b02e7bf9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 22:15:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69dd9be756884efdb6e13d1071427444\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62d212c320a44d07b1a5a610a71614ac : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8d86082dacb74bd39151044995241524 : SUCCESS in 5m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/787f2f8bc1f24ab9948d71dadde287d6 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feadf63fa6694068a9654c7f6e6528be : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c70e03c8a77477c9e270554a964e2d2 : SUCCESS in 2m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fbfc45151ed94e0386691ad98679637b : SUCCESS in 6m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/922a1d4d7251426897dc81990d6b7e7a : SUCCESS in 2m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2da84cf5eae64416b027fd761cb8aca4 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0635ecdfc9cb402ba5036bed94884b58 : SUCCESS in 12m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04f6a757463f4a35ad2841c43da51ad2 : SUCCESS in 13m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bcc4cd5e356b4aefb2b1adbd1f269055 : SUCCESS in 14m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79b990e822e348609034a980068c1c7d : SUCCESS in 9m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3d76f99c83d4487bbfe1bb6d0a08b14b : SUCCESS in 11m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61691c7ff8f84dfea88f793c741bbd0a : SUCCESS in 14m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8f99334357cc4a44b60b889dc21cc0fd : SUCCESS in 1h 04m 55s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fb1a1962ed53428ea17e9d8a011d1e1b : SUCCESS in 1h 03m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d61d1e364b2a4a5f9b0d2874bf437497 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/687808d753464cefbc3298a2a05f3a35 : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ab987c51970b495e9c4e439c3e25f0fc : SUCCESS in 1h 02m 21s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8b1d54c45ded4ca68556dc88bcb47db6 : SUCCESS in 1h 43m 15s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8452eac7397741c9b7f7c3786f1c0d37 : SUCCESS in 1h 39m 04s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8169413ad824d9c9b37cb2fc6bf0bba : SUCCESS in 1h 34m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53d86d5ec502419cbb50d7af82d1556a : SUCCESS in 1h 41m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/366c32d96fc04753aa96b0a4e5ef273e : SUCCESS in 23m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/97c46ce7b3134832bb0d3b24421e526e : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46610cc9455b496ea61944d3b6c6e361 : SUCCESS in 10m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4740faaa35664b25a31882fff14a9e65 : SUCCESS in 16m 09s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0b56792bdd134daf89af16b7b437bd82 : FAILURE in 3m 19s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5f84db96152f4406840732b8b0aa43ea : SUCCESS in 37m 20s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54cf6309bb9b4be9852cb9269079eb81 : SUCCESS in 16m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f5dde86b47042f1a0593bf185e5c67e : SUCCESS in 33m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/53b698e1b7314627abb3c05bf87ed154 : SUCCESS in 11m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1ce608203a1497d841dc6f71a10cb01 : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":7},{"id":"77204b1905f573fc48868bb8132f2ab6ffe9f6e6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-01 08:19:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"966118cf4df9569403d796eb9b343885a13573c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 10:02:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cf21574e42640faaeb3568d4dc06791\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9caa3c388234227a1fe1124b48a8671 : SUCCESS in 9m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27f388dcd93940a384ceb23903fac81f : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b73ff6a21d642a3875a7172c020540b : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8158aa3bc1ee4b1783fb0f8064efc8a9 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cdbb4fe4edbb4062976503d85a0f4149 : SUCCESS in 2m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c1c9096b34849728d227dc793c9eb91 : SUCCESS in 6m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f168eac816843a58699d01b9d84cc5e : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e7cebd6878054e45b5a72b70c8cdb8b5 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/96e43431392d4408b9161ab796f208f0 : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e47f577637f484587765203dc94e087 : SUCCESS in 11m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3ec5e63e0a4b44198e2c59a7e22a8cac : SUCCESS in 14m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e291ba146e5c4b2ab8a14b129c8a897c : SUCCESS in 15m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5a819021761a474ba6041004e242e9ec : SUCCESS in 10m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c9e886130b84431bcbc4aecffe2f67e : SUCCESS in 16m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3a063ebbee274537b07bf91a6664ba54 : SUCCESS in 59m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8ae911b78948475083420978350221df : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3c5724a24c8148d8823b35dc149747b6 : SUCCESS in 1h 08m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fae6c4a67c04d81b9dee29cb02a2b75 : SUCCESS in 59m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a88a8e36d344416dad2e2476a08660ea : SUCCESS in 1h 00m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/333690111179425bb7b0f853d7ae28f0 : SUCCESS in 1h 37m 25s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e045de6ed7474a549690d5549a55cd1e : SUCCESS in 1h 32m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2332a52521e540ec86191839f0960de1 : SUCCESS in 1h 34m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03d97880dd174269aedc42bd5bc7389e : SUCCESS in 1h 41m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f9942d9b0f944c4d824866ea412f3b30 : SUCCESS in 10m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c5dfe086a5784f52aaa592c20c272d52 : SUCCESS in 23m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c162ed86d01a4ab291b8e148892b3488 : SUCCESS in 19m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7e47fa6e8e494b6c944d2a60ddaa473e : SUCCESS in 16m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0f845060e66540c29d63bd8d2e6c1db4 : SUCCESS in 16m 33s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fc7919d74aed4e19b11e8d27915f3d23 : SUCCESS in 18m 00s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6d1a35db1604a1abb17ba4a62533573 : SUCCESS in 29m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28ca31446de24196ad0455962d276722 : SUCCESS in 14m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/743442280980441ca602a0ac6420be8f : SUCCESS in 9m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f896332993f4a0eae8264a7ed01deff : SUCCESS in 14m 57s","accounts_in_message":[],"_revision_number":7},{"id":"ddaac67289524b3f34dde43e082679a5f82a3be9","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-01 10:53:12.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"41189a59263b2aee0fa1101a57efc634899db7ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 10:53:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"500eacd2057795e82494c99564f0434da1af6785","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-01 12:05:44.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"45a4325b0aec5deb8dbbe9a2ce0c4b1f5d153899","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 12:37:23.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe/+/976615 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"37208648b92f555fce3264d823711f09e3ffeffa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-02 09:31:00.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cbae235106a92bc37b2316201756a4f34c076714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 11:29:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f016f15a050b482b8d3ec4fe14cbadc4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58cce22f0b534e9b8857ee607ffe0228 : SUCCESS in 7m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cbd06fe502cd4b0c90fcdd3d67eba525 : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b225e52caa844288965edc836cafff1f : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0dde056685042b694bd0ced305f8092 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb913e0b7e934d349ea7a2e02c882a9d : SUCCESS in 2m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b83142a654c8465b88ad7cc54a4a687a : SUCCESS in 5m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0da026fd3c54cda84b3ed2930d86bb6 : SUCCESS in 2m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7887af77a03d4dbdb95cbc7d37c87478 : SUCCESS in 3m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/55a62de868184a8c9b527e70095c03c4 : SUCCESS in 5m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a7e1e90ea0f64fbd9784ac55a275f920 : SUCCESS in 6m 19s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/071116ed7d64412591b729cf1bd4df9d : SUCCESS in 9m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6be16b3bfb164605a4d7931865420146 : SUCCESS in 7m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ecf707f237414391abec597a85da94a6 : SUCCESS in 16m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3644bfea01b945b09be5bc01a1daa80a : SUCCESS in 13m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/32b041fac06a4b06aac0a3eb05a53015 : SUCCESS in 1h 12m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/81e0d5a9d2944f7285c587975d4086ba : SUCCESS in 1h 11m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ea14cc29b5a54392b415f0f828be15a9 : SUCCESS in 1h 15m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/754d296a80124acc90ae39b1ac3f167f : SUCCESS in 1h 13m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c55ae90fae684455877515ca0f022690 : SUCCESS in 1h 15m 05s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e9e4052d140b4d8eb5a849f03def0fea : SUCCESS in 1h 56m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ade1c2055f2945b6818b1ea2f137f3bc : SUCCESS in 1h 30m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59da9777d36143479c667221d46c293a : SUCCESS in 1h 47m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b6a6935e0da48ea8c7481c3efcc2d1a : SUCCESS in 1h 55m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/68d3c02a6da042fe9e474cebdf256851 : SUCCESS in 15m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/99324797d985478fa3cb36103c75520b : SUCCESS in 16m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7496598c25ad4116803ea1f56974bfa4 : SUCCESS in 15m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b317018ef226488f893144119db15f44 : SUCCESS in 14m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7354dea96d6d4d94b3aa29511f1b43ea : SUCCESS in 27m 19s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/de6beffbe041494b9a47c1a5e43ea161 : SUCCESS in 17m 11s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e9487514cfa4e87b32e9c9c1c64b666 : SUCCESS in 15m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0019ab17d6149179277bdae330ac3e2 : SUCCESS in 14m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/973239ae7cde4a1eb8450a000c4dfe4d : SUCCESS in 9m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/451275cac30a4ed4920b80a0d5976b20 : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":7},{"id":"6d337ba97f913335ec77bab0defdbef1be6402c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 14:42:29.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ac3883dbe89bf38e7cff9461d8df2b11eba56b2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 16:24:33.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bf11750ee6140dbb891564c6e13d5ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd40192c663246b4834db5bbc80731da : SUCCESS in 10m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5da19c81acc4403b456b68808294ba4 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75cf7db030534ff89798ea1c943bdfad : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06f33bbed8c64a4faa9ed3b8e07a923b : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/949c182fb68d449b969257aa4f169be3 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a4876a9723b4f4aa7981c5ec42378f2 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/56d1b793df5a4247bfb8a40675d1e073 : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a19d2bade0804acab5d3589e61a495b6 : SUCCESS in 6m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f9a1e97d082348fab5cd6cdce93164f2 : SUCCESS in 20m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3eb5ba3c31b24caaad7a5fdbd6cc2ce2 : SUCCESS in 15m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/db9811526b3b4ce9b2fb23c542f46270 : SUCCESS in 23m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afe46993ae6544038919b91dea0eb76f : SUCCESS in 17m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/869e46da791b423cb78d901207862c6a : SUCCESS in 1h 00m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bbb584e4fb6849e6a21e587fd27d3cdc : SUCCESS in 1h 04m 21s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/15ca947552d6431982f266303b04077d : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9dddb7fee834ee99a690478197eca85 : SUCCESS in 1h 00m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/048ad92d175f4c1eb240610325ea7b91 : SUCCESS in 59m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c05c8ba810384ebd8c4aaef25037e9ad : SUCCESS in 1h 35m 33s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/416db68f043a4efa8fd37e38a20dbe23 : SUCCESS in 1h 28m 15s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12926f292936469cb375e4dc41ac75ae : SUCCESS in 1h 32m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b363a77c250401d8467e0c81c09bbe4 : SUCCESS in 1h 34m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ad824e6abc2c4aeca6ea5723f4ec802e : SUCCESS in 24m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c17bfcef96d4418aaa156202c031bf81 : SUCCESS in 18m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b4716a9fb6a4957a74f5d09e2474610 : SUCCESS in 18m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/33513658ff304e12b3f10458ca173453 : SUCCESS in 20m 25s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a1bf30c15e3b4b70a58fa2157d7ecc86 : SUCCESS in 17m 09s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0de4a3c257f84d6eaad77c98f5737493 : SUCCESS in 39m 04s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c61b8998eb9e4a6a91e22d66e92fb44b : SUCCESS in 29m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee201dae8e954c3d9537b1a1dda531c2 : SUCCESS in 30m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cfd0a6f0c8624911bfc575da7ffe2221 : SUCCESS in 17m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1598b86ed2e48df8891d2b532819edb : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":7},{"id":"7136e1c0497a3f589b439f7587c0af529b165f7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 16:24:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"20e2090b087f52d17ad154a17538e0bec2dcaf3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 16:27:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/369baae329ab4019b048f416bfbc903c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c600e3b53e64ffa9835f2f8ab497866 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a7fd8e81e114a2e9e760b0ed2b0f3d4 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5b2a1356676b1449db8d1014fb988eb969e3a988","revisions":{"38a11edef95c729958845980cd325ffbe5e79cb7":{"kind":"REWORK","_number":1,"created":"2026-02-16 19:25:30.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/28/977028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/977028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/977028/1"}}},"commit":{"parents":[{"commit":"a5dfb8e6c4afc1afeb5761bb1b86d55e1d3d5b90","subject":"Template disabled repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5dfb8e6c4afc1afeb5761bb1b86d55e1d3d5b90"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-16 19:25:22.000000000","tz":0},"subject":"Adds dnf_repo_state_overrides","message":"Adds dnf_repo_state_overrides\n\nThis allows you managed the distro provided OS package repositories\nwithout having to define them explitly in dnf_custom_repos.\n\nChange-Id: I40946f21209090701fb8a2d9ccfaaff573e228ef\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38a11edef95c729958845980cd325ffbe5e79cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38a11edef95c729958845980cd325ffbe5e79cb7"}]},"branch":"refs/heads/master"},"9c4dba00dfd71736522f75f748fa37be1f52336d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-23 18:58:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/28/977028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/977028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/977028/2"}}},"commit":{"parents":[{"commit":"8301e0e389d066d40340005f78e8c6fef4baeaac","subject":"Template disabled repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8301e0e389d066d40340005f78e8c6fef4baeaac"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-23 18:57:51.000000000","tz":0},"subject":"Adds dnf_repo_state_overrides","message":"Adds dnf_repo_state_overrides\n\nThis allows you managed the distro provided OS package repositories\nwithout having to define them explitly in dnf_custom_repos.\n\nChange-Id: I40946f21209090701fb8a2d9ccfaaff573e228ef\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c4dba00dfd71736522f75f748fa37be1f52336d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c4dba00dfd71736522f75f748fa37be1f52336d"}]},"branch":"refs/heads/master"},"e58b8a911498b56c76e0414557cdd547fd4ba9df":{"kind":"REWORK","_number":3,"created":"2026-03-17 17:05:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/977028/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/977028/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/977028/3"}}},"commit":{"parents":[{"commit":"40901368f2c24e40b8dc34588d193502bc1b000e","subject":"Template disabled repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40901368f2c24e40b8dc34588d193502bc1b000e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-17 17:05:01.000000000","tz":60},"subject":"Adds dnf_repo_state_overrides","message":"Adds dnf_repo_state_overrides\n\nThis allows you to manage the distribution-provided OS package\nrepositories without having to define them explicitly in\ndnf_custom_repos.\n\nChange-Id: I40946f21209090701fb8a2d9ccfaaff573e228ef\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e58b8a911498b56c76e0414557cdd547fd4ba9df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e58b8a911498b56c76e0414557cdd547fd4ba9df"}]},"branch":"refs/heads/master"},"4d15b05ca66e45a9912fe8535555601b7c2afe0a":{"kind":"REWORK","_number":4,"created":"2026-03-17 17:14:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/977028/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/977028/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/977028/4"}}},"commit":{"parents":[{"commit":"40901368f2c24e40b8dc34588d193502bc1b000e","subject":"Template disabled repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40901368f2c24e40b8dc34588d193502bc1b000e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-17 17:13:45.000000000","tz":60},"subject":"Adds dnf_repo_state_overrides","message":"Adds dnf_repo_state_overrides\n\nThis allows you to manage the distribution-provided OS package\nrepositories without having to define them explicitly in\ndnf_custom_repos.\n\nChange-Id: I40946f21209090701fb8a2d9ccfaaff573e228ef\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d15b05ca66e45a9912fe8535555601b7c2afe0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d15b05ca66e45a9912fe8535555601b7c2afe0a"}]},"branch":"refs/heads/master"},"285f7673a5235a949ad1c2b2944fe124b16a86c8":{"kind":"REWORK","_number":5,"created":"2026-03-17 17:17:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/977028/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/977028/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/977028/5"}}},"commit":{"parents":[{"commit":"40901368f2c24e40b8dc34588d193502bc1b000e","subject":"Template disabled repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40901368f2c24e40b8dc34588d193502bc1b000e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-17 17:17:45.000000000","tz":60},"subject":"Adds dnf_repo_state_overrides","message":"Adds dnf_repo_state_overrides\n\nThis allows you to manage the distribution-provided OS package\nrepositories without having to define them explicitly in\ndnf_custom_repos.\n\nChange-Id: I40946f21209090701fb8a2d9ccfaaff573e228ef\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/285f7673a5235a949ad1c2b2944fe124b16a86c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/285f7673a5235a949ad1c2b2944fe124b16a86c8"}]},"branch":"refs/heads/master"},"b418e46764374c43cb58f060a90ec1ad7d1abe9c":{"kind":"REWORK","_number":6,"created":"2026-03-30 19:48:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/977028/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/977028/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/977028/6"}}},"commit":{"parents":[{"commit":"ae440aa5b9c536f1c1dbfd30487ec5f351d8cc27","subject":"Template disabled repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae440aa5b9c536f1c1dbfd30487ec5f351d8cc27"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-30 19:48:10.000000000","tz":120},"subject":"Adds dnf_repo_state_overrides","message":"Adds dnf_repo_state_overrides\n\nThis allows you to manage the distribution-provided OS package\nrepositories without having to define them explicitly in\ndnf_custom_repos.\n\nChange-Id: I40946f21209090701fb8a2d9ccfaaff573e228ef\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b418e46764374c43cb58f060a90ec1ad7d1abe9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b418e46764374c43cb58f060a90ec1ad7d1abe9c"}]},"branch":"refs/heads/master"},"5b2a1356676b1449db8d1014fb988eb969e3a988":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-03-31 20:31:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/977028/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/977028/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/977028/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/977028/7"}}},"commit":{"parents":[{"commit":"229b1efc12d5ff2d60f2225a42e0c657ea6a7924","subject":"Template disabled repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/229b1efc12d5ff2d60f2225a42e0c657ea6a7924"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 12:27:21.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-31 20:31:16.000000000","tz":120},"subject":"Adds dnf_repo_state_overrides","message":"Adds dnf_repo_state_overrides\n\nThis allows you to manage the distribution-provided OS package\nrepositories without having to define them explicitly in\ndnf_custom_repos.\n\nChange-Id: I40946f21209090701fb8a2d9ccfaaff573e228ef\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b2a1356676b1449db8d1014fb988eb969e3a988"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b2a1356676b1449db8d1014fb988eb969e3a988"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
