)]}'
{"id":"openstack%2Fkayobe~977525","triplet_id":"openstack%2Fkayobe~master~I36217971254bb14c8931bebb2aab578beb411b57","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-02-21 20:58:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I36217971254bb14c8931bebb2aab578beb411b57","subject":"Remove community.network Ansible collection","status":"ABANDONED","created":"2026-02-21 11:45:50.000000000","updated":"2026-02-21 20:58:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cbd25426b252def0aa47f2ad820304aa288de53b","_number":977525,"virtual_id_number":977525,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-21 13:28:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-21 13:28:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08f1cc0164ff3c3c50644baa0d072a7c328d73bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-21 11:45:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3dc69ae724d896103936c8a47db7a124f6834a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-21 13:28:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78ca80fa1e644ddca11411504cb0a68d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a12c366e23b49a3964e793a9b3b6ec2 : SUCCESS in 6m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3560e45e5aeb4c39b1001fd32f22cea0 : SUCCESS in 5m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23a8dce426164de5ab979797a89c2333 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408888fa2e994a0cb14b9339e91e245e : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47db754c7faf426bafbef51c7c7bf0a6 : SUCCESS in 2m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49d4bd89acbe47cca2fc8e8409b6a0b8 : SUCCESS in 5m 36s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/498337bf483f4814a2d239688fd52fe1 : FAILURE in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/38eb1fac5e97417f8e70c3ac9a8807da : SUCCESS in 5m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2d60292c554b4db4b38f949cefafcd0d : SUCCESS in 10m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/61f0b436bb9e4a559952b93ad598161d : SUCCESS in 7m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be1288fb4c314939959c34e0ee80baee : SUCCESS in 14m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8ed6a1639e2642da8f7db2ce0cfae53f : SUCCESS in 15m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8da15e6b5c04410b2161013f36fcb84 : SUCCESS in 17m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3c22210dfd884ba0883f8a076551e215 : SUCCESS in 52m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4c693cf6987a44d3aae7f148fd1798b5 : SUCCESS in 38m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3786de5782de4b04aeb4abb0efda81ea : SUCCESS in 57m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c5662aab85c443bad42e0b922c3f48d : SUCCESS in 1h 13m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c3d820fc65964017a37241b4ce35db51 : SUCCESS in 1h 15m 17s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b57af299452741b1abaaa66238af43a1 : SUCCESS in 1h 31m 58s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bac9761451dd499ea999e68c050b983a : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc78a0fd89b2456794b4bdbd507110c8 : SUCCESS in 1h 21m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f2645fe90d4474aa8ef2cdb44a232b0 : FAILURE in 1h 37m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/58909b571d394fc882e01284cdecf7c9 : SUCCESS in 22m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cbb36c7f7c2e4816a59a4e82f4c8d499 : SUCCESS in 15m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6165909dd4114190894661799b7cedf8 : SUCCESS in 18m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a51fca5b028c4288bca68b1615a24558 : SUCCESS in 12m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7dbce9714192495b83b9a0c9040b1970 : SUCCESS in 23m 54s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cfa0c5fffd5341d2a300060cb9ab63cc : SUCCESS in 25m 22s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c60a667884e94b3388dec2e1df9e81c7 : SUCCESS in 20m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edafb0db227944f39f6bd6663c6ef9d9 : SUCCESS in 32m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9da58a72a56a49ebbf789a763d0958b2 : SUCCESS in 12m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9b04fa61dd048b9bf6944f6f3b10933 : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"cbd25426b252def0aa47f2ad820304aa288de53b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-21 20:58:18.000000000","message":"Abandoned\n\nRequired for Cumulus NCLU","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"847f88b8be1565bcafe93691f823e44358da636a","revisions":{"847f88b8be1565bcafe93691f823e44358da636a":{"kind":"REWORK","_number":1,"created":"2026-02-21 11:45:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/977525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/977525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/977525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/977525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/977525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/977525/1"}}},"commit":{"parents":[{"commit":"e2279353f9e6ff0b18bade8694d1d8b10d91183e","subject":"Merge \"fix (docs): correct passwords view command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2279353f9e6ff0b18bade8694d1d8b10d91183e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-21 11:45:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-21 11:45:30.000000000","tz":60},"subject":"Remove community.network Ansible collection","message":"Remove community.network Ansible collection\n\nThis Ansible collection is deprecated and its source repository has been\narchived [1]. It does not appear to be required for any supported\nphysical network configuration, so let\u0027s remove it.\n\n[1] https://github.com/ansible-collections/community.network\n\nChange-Id: I36217971254bb14c8931bebb2aab578beb411b57\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/847f88b8be1565bcafe93691f823e44358da636a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/847f88b8be1565bcafe93691f823e44358da636a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
