)]}'
{"id":"openstack%2Fkayobe~977845","triplet_id":"openstack%2Fkayobe~master~Ic193993c78a5650f918f9c5c40c4669e50a624a3","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-13 08:26:35.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-07 10:23:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-13 12:59:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic193993c78a5650f918f9c5c40c4669e50a624a3","subject":"Add kayobe_path variable","status":"MERGED","created":"2026-02-24 18:13:18.000000000","updated":"2026-04-13 13:03:41.000000000","submitted":"2026-04-13 12:59:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977845","meta_rev_id":"01c622cdc1b3c12e7c3cb86c05b2ba6748da40f6","_number":977845,"virtual_id_number":977845,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-13 12:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-07 10:23:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-10 08:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-04-07 10:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-10 08:51:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-09 16:15:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 20:21:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-27 17:38:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-21 08:43:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-04-07 10:05:23.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-04-10 08:51:01.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"cf82010a8fe97e9d46aa7da800d06306e1c5ea3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-24 18:13:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dbc3ec99e2b2f36cdec1889880c9667460cb424","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-24 18:22:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"927c96217f47b4aaf35176d43ce36134921f2a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-24 18:31:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"65fc84ea85975acc988cd99478556e3acbd05361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-24 18:36:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f6ec0371d1e68ccffeb0fc2caafd5858bc5c4f28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-24 18:40:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3e17ed70beeaf2011947559be4322e547d9828db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 20:21:20.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2adc1251461c4a36b6739326c1ed8dfe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a43b0b1fd7645569ea7cf5992c979c2 : SUCCESS in 5m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/067b535c854a4b58a430b27946a5ead0 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d24c50d6d8845b482454df4693234f3 : SUCCESS in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4523331f4648b18faff8d32f3b1831 : FAILURE in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a53572c46f644a7a974b1676c07c4fae : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/83efa322290d495a9d4455ca9da27513 : SUCCESS in 6m 27s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/459712a2d07442afa2d342d87910bf57 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a517a6d9290f46c4838f43111f18b129 : SUCCESS in 5m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b9dee1f618174be882fbe30d61c3d124 : SUCCESS in 8m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3625c619773b4cd69704572a182f87f5 : SUCCESS in 18m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b46c25b6ff6f481e8cc42922982af410 : SUCCESS in 16m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0899a5dff89e4d5b94e445cfc2ad16e4 : SUCCESS in 10m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ba7b2968b08466c9d48550e765bee8a : SUCCESS in 13m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/de9f18d944824cd2976347e4984f6914 : SUCCESS in 54m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e6d8064348942818ad82bb37f802f86 : SUCCESS in 59m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/80dd749be3b54002a8dcc3235dd09a2a : SUCCESS in 58m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2bcb5dd43b040c5ab520160698deff1 : SUCCESS in 58m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6ca38871bcc543408bb0fde5034103ca : SUCCESS in 54m 05s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a01ba13d421d49049325bc153b002370 : SUCCESS in 1h 30m 41s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/371ba92e03b24ccb98a309d40bceee5b : SUCCESS in 1h 24m 36s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03ef8fb2c4b948bd93e87020939e300b : SUCCESS in 1h 25m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3368486030ab41009e27685d5bd3a131 : SUCCESS in 1h 32m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/31492f6234bf4340b3a1133b2257bef1 : FAILURE in 8m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a67a720cb47944d8becde7efd2e1b73a : FAILURE in 9m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d9f9407264b4cce88973751166c5201 : FAILURE in 6m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59bb8513008a4f1dbda3cafd06dc7e29 : FAILURE in 11m 27s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4f5a2ef9afd2479f82e62ceed81969e5 : FAILURE in 15m 29s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/de4a1ec405ab4548bf899749fad1daff : FAILURE in 26m 52s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c4f4a1929c1408099aebadd9bfc4407 : FAILURE in 24m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/495758363574430489c1042a0990dc0b : FAILURE in 26m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/afc65b7fcb474018acd04093b9592443 : SUCCESS in 14m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d896a907a7a475cb8e912843fb5fc08 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":5},{"id":"da2a66d13be25bf06bc2b1d89d83956b94f626d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-26 10:37:00.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"af528aef9f4e8b880ab6404b3824d509fcfc1f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 11:37:39.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0572678045ae4af6965646bf499eecc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c53bc7e2a2ab43f5955bc24a467858a4 : SUCCESS in 8m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bd7ae8d24dab4f7584f6bd5bff971d9a : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad0c9c54ba57408fadd62e46b207bf52 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9197599ede554e73a864ee787e884ee7 : FAILURE in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ac892ddd5d74321a476cfdf15d122e0 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff306fd3271f482b9679dbfbd20a26c2 : SUCCESS in 7m 57s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd6d3bade093496e8e806221763b68a2 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f22d16de2cc4439bcb25b50aadc6b02 : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d772c0ddfd904590a142e21875aba7b7 : SUCCESS in 11m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/33ac116f78834047b35103563d549072 : RETRY_LIMIT in 2m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/587701c545c0447eb4dd65844289226b : SUCCESS in 12m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8bb7c06d3fb6461fac50989c4c03a60b : SUCCESS in 19m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fde5f56db834f8ebf10e1b361b66604 : SUCCESS in 15m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f49c89061bc24e18859bf44526b23d03 : RETRY_LIMIT in 3m 24s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/89a90c14f8fb408ebb69be3b2a556605 : RETRY_LIMIT in 3m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/46098f1ef3de464c9461aef0dc1d38dc : RETRY_LIMIT in 3m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4756ace26a7467bb3530e5c1baf302c : SUCCESS in 57m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5b1b4f38d9b24343a03eaea807e1c44c : SUCCESS in 57m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c04366ef78d24b6db08905e7904d7c37 : RETRY_LIMIT in 3m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/75b11e7c6563465ab692cfea93e5ca59 : RETRY_LIMIT in 3m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7baa29fbb94458d9b0b3eacf9f1e510 : FAILURE in 47m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e75b04a96a7b4936b1901e785442685b : FAILURE in 54m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bfacd0ada12a4f10809328072e77088c : RETRY_LIMIT in 3m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/559a767d0606488c8b8e8ab0672fb3bc : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":6},{"id":"6d57882f146f7cf3ceb999c5f681a4df27e27fb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-27 17:22:10.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"be9bc11ecd12bcceed2859842fed4b5e89cfdb7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:26:36.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f13db734efae42be9ce548a489355f68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884c84302ed54fe9b3884a05cc15b0c0 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0ebcb862b1934260ace711c599790dea : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2945257c7b07414b8f4ddf7528ca76ae : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed2071193dcc490cbc9802bcddd309eb : FAILURE in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/026124301e5947c4aeda2cb7deb1f4c3 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8ac85529add6410da5531f9d21669ff8 : SUCCESS in 4m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6027a8b329844dd90c6f90dcc5b7bcb : SUCCESS in 1m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de6dad0354a44dc494854e78ec8cc855 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f02ac0c5b4334c8cb0386b8d0ad5d0ad : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5396d9dbc714e7e95b0ca62c326acdd : SUCCESS in 10m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/863432ab916a45c0b6fff8ea61a2bc04 : SUCCESS in 18m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddd69b2780a14c72a054ea9457e850db : SUCCESS in 6m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/da7feefabad5492d9143f36ae7615939 : SUCCESS in 25m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3fdd02e85f043c4b1ca105934ecd7fd : SUCCESS in 16m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/919315f12a6f4fe78587f6ae637be4da : SUCCESS in 56m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/be25cf4227494e49b3a0649f8c83cc19 : SUCCESS in 54m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a699c3d25afa47f89a44387536914cbd : SUCCESS in 57m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db1bd164a5484f8c9d21db250fa5840c : SUCCESS in 1h 02m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c9d16158f2964b47b664e3fc0a17c96c : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4b6ee2a4f5534fd68e71bf0dac50192c : FAILURE in 53m 07s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cb04c62db4d246a7a718fc011dca4e54 : FAILURE in 58m 45s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b60fdd89e794707aa755ac8b29c3f8c : FAILURE in 54m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1545a8721d1477490cbdfe9d4d931bd : FAILURE in 59m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fb373dbe73a84cadb46be3b955c25f85 : SUCCESS in 9m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/000c4e1d056b4c2fa3f2d3520c95b535 : SUCCESS in 14m 18s","accounts_in_message":[],"_revision_number":7},{"id":"a6f8a1160c482bb4876c39bb3338801ca3867e7f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-04 15:59:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c29b519217b8b1400715b8905ba78bfe9781df5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 17:51:49.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27ca5196a21447b28408d07037755bb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23e7d9db8b6488d8d3718dde8564a79 : SUCCESS in 8m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7832816ef8fc4bca8b9d2982a4876515 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ff026e1f897499db21b576d0f967746 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6116332b675e47f993423198be436773 : FAILURE in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9aded2489fc64ff489452beb0fafae92 : SUCCESS in 6m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24a06d70150d4f638d5c8d3a2475feae : SUCCESS in 4m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db75a634345c48df9e3b0b8ca82094e6 : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/11b837f4def5482082363f43a6d0f4bb : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f459ccc5fd72430cbc94ffbc912b571b : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/efefd698d6394e45abe4671aafcb95b8 : SUCCESS in 6m 58s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/176eecdbbd0541eb91e0c94c207860e2 : SUCCESS in 19m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a62523d12a5d4c5ab0dbc68af3f59b75 : SUCCESS in 16m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/330f57471dd04b8289157325c8d1aa0f : SUCCESS in 19m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a2db75c4b664598af2c423dfd2ecb0a : SUCCESS in 19m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/982d8eb905e64e4e95b4be4e5e2dd2b1 : SUCCESS in 1h 03m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5d4b1b6d090646f68688f307035706c3 : SUCCESS in 1h 00m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/30825db4afab42a2bb07530938a1b2a3 : SUCCESS in 1h 03m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c27e0527c4074a4ba399c62f2aaa8645 : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/01e0ea6d8be44e16b830e5d6c89b829e : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a13a06ad53664d61b047a83ffbce801d : SUCCESS in 1h 49m 01s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fc9e53fde2ff4732b84f4de015fa581a : SUCCESS in 1h 35m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fb3b4c5620148398bf263ec06f3b13b : SUCCESS in 1h 30m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/398b6ba87d9b48d991b2b81613bcad30 : SUCCESS in 1h 39m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8873002fba244a92bdc8e4631bc07712 : SUCCESS in 12m 43s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/aa1de8e553564af19d31cbff05964450 : SUCCESS in 12m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3e61304d8e54a4986afef76e732a66b : SUCCESS in 20m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e18e844fd5c484e9e643c55aa54589e : SUCCESS in 17m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6cc3efc99ffa4057b7af4a7ed92a1f21 : SUCCESS in 27m 04s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/55c5ac0442704b398ed6450b9719c7d4 : SUCCESS in 36m 21s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d0c3add82c943b78b6d74f129ab0c00 : SUCCESS in 34m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f96a7c403fa458498b0b6fabddf824b : SUCCESS in 29m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7a52ae562fbc4a7b9a40630457611530 : SUCCESS in 21m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6d7006cf415419cb502cd1e4889bdaf : SUCCESS in 15m 16s","accounts_in_message":[],"_revision_number":7},{"id":"fc09f010522212cfd7819b3004b15521e907703b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 21:13:09.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"72c62e05ddef002cae207b12111eb9aa70830aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 22:55:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00d626db941a4d40923ba0b91b3310e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1ce54fdae534d27ba2627b63c19c85e : SUCCESS in 10m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ff2e702c61764450a0503bbfdd17eb20 : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e23a209dac0e43d49a137ddb0a0ac7e0 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25c6cd41a3dd4cea9cc20a4835b6d07e : SUCCESS in 6m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1dbc9dffded74574868c5047eec25147 : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3190ca1ad3644d75bf67bf5d15586353 : SUCCESS in 7m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/568d60dc97e84d4eb505c22e94d6ca72 : SUCCESS in 2m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c22b2962e11546afbc2dbeabc6878ef3 : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8782616ddc32474c93317b3bf552fc20 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b48a9d098f094fb0936e6726af53b0c1 : SUCCESS in 5m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/005043a16e19435690bbda795584daa4 : SUCCESS in 19m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05c0ce3decef4ea6adfb1298cf80be93 : SUCCESS in 13m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/63d82066137242f2991b707cb82737df : SUCCESS in 26m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed82c9139b624aecba8a9a9b72b90948 : SUCCESS in 16m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7ec343262cba455fbb9a4d8b5dc2be77 : SUCCESS in 1h 01m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9b61512f4dee4f33a29694ab23e700ce : SUCCESS in 1h 00m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/49a565515e1e4bf5a669a80d3fc934e8 : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/499cffad82ba45c8ac6476bb8538c4f9 : SUCCESS in 1h 00m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6c731515ae5c4ed1a527f03c09640786 : SUCCESS in 1h 01m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d9f2a52c860840c08b4b266328bdfd02 : SUCCESS in 1h 39m 30s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/549cb6f0856a44b1bd352de95b51ebbf : SUCCESS in 1h 41m 34s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8b1a62b2a8a4e3e94f4a8531c9455fd : SUCCESS in 1h 33m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/478063b7c67d4836808d4df14b681bed : SUCCESS in 1h 41m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6922052b77424e4da05b4bdc745d7d5b : SUCCESS in 24m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/02f96c7280c44598a3a79555c9fc92e3 : SUCCESS in 23m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/610281b7afc74fa38db7c6e42a211630 : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/492d4323e0b54b018641013bc2f5a978 : SUCCESS in 16m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dce6acc2ef2046419c0be1e9b5b082c5 : SUCCESS in 15m 36s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2a81be6b755f4cb289aeb9fbe4fb36cf : SUCCESS in 16m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/812978e225bd4c1c8e72c7c6b5fe6bd8 : SUCCESS in 30m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6030b9a5bb464f1eb2d1075ca9bbaa32 : SUCCESS in 28m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a3d1e17c71ca4bd29ca27e5918b44fe7 : SUCCESS in 23m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/725853f9619c47a6ac4c1e3bcbe24352 : SUCCESS in 18m 13s","accounts_in_message":[],"_revision_number":8},{"id":"6cef56e3ed5467649ea4c012c54f766f56476ab7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-21 08:43:17.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5f6ae28a60bb6e14fc4790ff9da7638521a70e0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-21 08:47:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"f2dd76794658e917bedb5d4a2ed62830c032721d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 10:24:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3748fabef4e49b3bcfddcef6139b486\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ef677f782e42d4af3d86488d7a63cd : SUCCESS in 5m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/18b740205d284880811ecc42579c4560 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e19c3cb57b3d45ffbdb53668cfce65bd : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0128c4ce1b940a99c35d3e573290fb9 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21e87c34570b4790b9cad5d4db050bef : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/25e21772533a4c129eeb11a10abc12bf : SUCCESS in 6m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b03e5819ca084ac68b88d8c5543aa4ad : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a8c8031986949c7ad1a6239dc0272f6 : SUCCESS in 3m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5869178c68ad447c9f90843cabfd2172 : SUCCESS in 6m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dc14f83a4ba0497c92efb11168d7b1d9 : SUCCESS in 13m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a6794645dba54f7f8c7ce869b6e122a8 : SUCCESS in 8m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b05191df17944b42916cfd9a0b996daa : SUCCESS in 14m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/424c72fff35242548dd74d37770ad020 : SUCCESS in 11m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42f4553430ef4dccb7cc4dc62cd5a9ef : SUCCESS in 17m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/757b1f7a007749d7be515e4833dbcc77 : SUCCESS in 56m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7777967915de4108b9e4bad6beba9bee : SUCCESS in 57m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/47ff013f964340a7bed7c4f40cc763df : SUCCESS in 59m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a699f7bdedec443a9a22daae824d66bb : SUCCESS in 56m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e808a84626be477fa3f371ba1519b56c : SUCCESS in 57m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c9ee69350dd243b1aa4fcb11fd9b9561 : SUCCESS in 1h 33m 28s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c59699b8fd184c4d96c0c21e6a597ef2 : SUCCESS in 1h 29m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cbb6c8431c045baba639ae41812bd51 : SUCCESS in 1h 27m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56c05bbecc84430dacbdbf60364b5732 : SUCCESS in 1h 34m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f94ce37ddcad4f3fa0396bf358d0545e : SUCCESS in 23m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/75855d6045bd4916a12eab67a007fc63 : SUCCESS in 10m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f511c1f12994661b295fc401d7cc0f0 : SUCCESS in 16m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e6a37fc0a4064e638fc032dc5f9aae22 : SUCCESS in 8m 59s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/136dc02b3f7c46cfba3c6f4336150658 : SUCCESS in 34m 16s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8c42229430424aa0b9d83bc654a0c09a : SUCCESS in 37m 53s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af41ac6e35d24f818b16f5715ad65732 : SUCCESS in 35m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d200ec660524615909df03f3f178b29 : SUCCESS in 34m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cd50f03a78ca46c3945875bfd60685f1 : SUCCESS in 9m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21a1013e7924420d884279405835c3f5 : SUCCESS in 14m 49s","accounts_in_message":[],"_revision_number":9},{"id":"6c05f05bad44a792cdcc914c6cb1bece128fcf59","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-23 10:47:35.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7543f9da5b90d26ed162c40628df38f70a59d91f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-07 10:05:23.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"ec84c0eb777491ce9b2ef58748824ffd800fb2e9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-07 10:23:25.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"073f2e33d251970e3d0da9ad53995f4f9d16f07d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 16:15:08.000000000","message":"Patch Set 9: Review-Priority+1","accounts_in_message":[],"_revision_number":9},{"id":"60b13bc05dcc90f93b2f3855816940be1fe22bfa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-10 08:51:01.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"ed5b232da8066548416fe3b9e6cccf547ad588af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 08:51:22.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"3095071a6b7c83a4ebfb447456a0394ac6f233cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 10:27:24.000000000","message":"Patch Set 9: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe/+/978058 is needed.","accounts_in_message":[],"_revision_number":9},{"id":"e31fb39d031a76c8742f9bb82b6a561af0c443b5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-13 08:26:35.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"100fb3dfd29b618548565017044a2c253b809d63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 10:24:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fbe60dae2d54c44a9fd6294b56e6ed9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d353bd3b4bdd4d68bb56be74caf42b57 : SUCCESS in 9m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fae9a8f4302448ac86f5f804867c170f : SUCCESS in 7m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0196ee1f19104d3dbfcb9e1caf16e664 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889447a574cd46ee98805cdd152b2fee : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25d1031c5e9149ae9985272ce9c4e92f : SUCCESS in 5m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a70f1077b294fc8a5f786e05551a0e5 : SUCCESS in 8m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0a1a10f4f2d4e8989fb45843e0c9121 : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4976d50db4c041ccafb44a39b5ce1888 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/de675e3d85fd4e0bb97c65d0e4acfa96 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/925256382d274784abc0e997da842812 : SUCCESS in 11m 58s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bfd7f6c384ff4bb3a049b25af0689272 : SUCCESS in 25m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69123144935744c69264abee5d4e73bd : SUCCESS in 8m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f8e16a4d0bba41d18b90c1d85972150d : SUCCESS in 21m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8c6a3d893da491bae24a81fe33c0e1b : SUCCESS in 16m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7b620cd1c2d94a1699ee0aa68630f8a3 : SUCCESS in 1h 02m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/893a34e2dc994061958cf4ad21dbab5e : SUCCESS in 1h 01m 30s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d4c16d73146c44ee86d060da5b84d48c : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/506ac6f1c2d646b38ec8a52ad222d52c : SUCCESS in 1h 04m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f087c5154f0145069f383713c092781c : SUCCESS in 1h 00m 58s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/85839999a7624d2fb1a6b3a7c20925de : SUCCESS in 1h 41m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5aad2495c75943639e0aa7fd34d337a9 : SUCCESS in 1h 33m 54s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40cc9a6ff1364c81a07e85e45c626d96 : SUCCESS in 1h 32m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36aa911161324d708c112edfe9aaa23c : SUCCESS in 1h 38m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/141783f2a4524ec08328ae46f3196710 : SUCCESS in 21m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7a092c13420646cea66673a559a8530b : SUCCESS in 24m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd1d806c73b147a7a7e5bebe4a70e506 : SUCCESS in 20m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0aa4614180d2464fb2554adbe644beda : SUCCESS in 18m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ea0f28e22bed47cab0071d0959b54575 : SUCCESS in 29m 52s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9e5ce3dcebe74718b073a25940712733 : SUCCESS in 37m 07s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9229295f1411425580d31f87372fbcfb : SUCCESS in 40m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91be58edfb50437ab7a3e32132a4e895 : SUCCESS in 29m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c42e97fab5c84e04bfc7859d5042e3c0 : SUCCESS in 18m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/488df51a4ed74fdca74d561f98a27279 : SUCCESS in 15m 52s","accounts_in_message":[],"_revision_number":9},{"id":"341d5b111aa4e479395cf2e7793daaf72c208bc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 10:26:46.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"14dd4ff37dc95a7e5c2115ce24800bb74130d721","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 12:59:43.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66fbf79082d34841a55506258492774a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be562ccdc27f45208aa0d7c79a2972cc : SUCCESS in 10m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/883410d95e73426f8d19a2663dc8008f : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b7726228fb458cb17f1b788c8a884d : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2d340e6cb6347bebbb529db40371561 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff54181bcb11455499f77ef827a26aa6 : SUCCESS in 5m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17fcddb11d854b2b9b97cec745414721 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98618a2ddcf5402097345051263fc111 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52ca09cfa6c34ea3abc42c9d464cb5b7 : SUCCESS in 11m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f9fd16e90fca49abbde7c5413587176b : SUCCESS in 17m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d0752498d7f4310a1fa468993437ef4 : SUCCESS in 14m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4e11fe4bd20e4217b9252389300916b5 : SUCCESS in 18m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec89efc346a94503bc18841d075ef387 : SUCCESS in 19m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c47d7309ac9146e9bc4a942a09412391 : SUCCESS in 58m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/97344558169a4c169399fde0d7b3b7dd : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/abb27ee288ca41d4a8f6fd7e01b75d02 : SUCCESS in 1h 01m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a427e175df8b4a30927336a271847a4e : SUCCESS in 1h 04m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8bfa673b755840e0a8917284a8439893 : SUCCESS in 59m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d37f34ad768a47a9a0c823e6d34905a3 : SUCCESS in 1h 36m 40s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9737f265ab6c48cba6497455d1bf5493 : SUCCESS in 1h 35m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c00df861138341f48b85099a01fc078e : SUCCESS in 1h 33m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4d9753811d548939c75511b9316e58f : SUCCESS in 1h 39m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ca69c0deed984f44ae8765b969614cec : SUCCESS in 12m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d345161e52254c75bc66ad128175fd6c : SUCCESS in 19m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5a1bd9d4e584d4ca453bf11b93edfff : SUCCESS in 18m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c55c60545c6842d883a5e19d6a7c3554 : SUCCESS in 20m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ea64cccfc19d4bc5a3a584a5d0cbcd3e : SUCCESS in 17m 40s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cd2cf2f54f1d449d9361d67b58c20dbc : SUCCESS in 29m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6b368d86f174be981805079ffdad9db : SUCCESS in 39m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48c1fa683ae34a3c92a4222df7c799f0 : SUCCESS in 30m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9a69e6c5dbbc4175947f93cdd0c893a1 : SUCCESS in 17m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10fd4d0a2964406b8fca042869195cf0 : SUCCESS in 17m 46s","accounts_in_message":[],"_revision_number":9},{"id":"65bbeaf6acc514e3b0a0d489b6e9aa6b07e88df0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 12:59:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"01c622cdc1b3c12e7c3cb86c05b2ba6748da40f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 13:03:41.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52ad2089216d444788a250ae9b6c79f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd0e16a1054c4690990b6d1bcf6a8771 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/995a8377b1394c04868cbf445010c5f2 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"f039f63d31c879b98fb7ebdd70208111636559d8","revisions":{"e34b42ecad84c08c269d9dcaa78170d79ac3926d":{"kind":"REWORK","_number":1,"created":"2026-02-24 18:13:18.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/45/977845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/977845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/977845/1"}}},"commit":{"parents":[{"commit":"e2279353f9e6ff0b18bade8694d1d8b10d91183e","subject":"Merge \"fix (docs): correct passwords view command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2279353f9e6ff0b18bade8694d1d8b10d91183e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:12:31.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:13:11.000000000","tz":0},"subject":"WIP: Adds kayobe_path variable","message":"WIP: Adds kayobe_path variable\n\nUseful for locating kayobe installion within ansible\n\nChange-Id: Ic193993c78a5650f918f9c5c40c4669e50a624a3\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e34b42ecad84c08c269d9dcaa78170d79ac3926d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e34b42ecad84c08c269d9dcaa78170d79ac3926d"}]},"branch":"refs/heads/master"},"42667d85fdee06065c086cbd9090454ea7e5a770":{"kind":"REWORK","_number":2,"created":"2026-02-24 18:22:38.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/45/977845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/977845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/977845/2"}}},"commit":{"parents":[{"commit":"e2279353f9e6ff0b18bade8694d1d8b10d91183e","subject":"Merge \"fix (docs): correct passwords view command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2279353f9e6ff0b18bade8694d1d8b10d91183e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:12:31.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:22:04.000000000","tz":0},"subject":"WIP: Adds kayobe_path variable","message":"WIP: Adds kayobe_path variable\n\nUseful for locating kayobe installion within ansible\n\nChange-Id: Ic193993c78a5650f918f9c5c40c4669e50a624a3\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42667d85fdee06065c086cbd9090454ea7e5a770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42667d85fdee06065c086cbd9090454ea7e5a770"}]},"branch":"refs/heads/master"},"2f8369498aa2927b54ef779bfc89bfc0d7615cf9":{"kind":"REWORK","_number":3,"created":"2026-02-24 18:31:26.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/45/977845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/977845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/977845/3"}}},"commit":{"parents":[{"commit":"e2279353f9e6ff0b18bade8694d1d8b10d91183e","subject":"Merge \"fix (docs): correct passwords view command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2279353f9e6ff0b18bade8694d1d8b10d91183e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:12:31.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:31:20.000000000","tz":0},"subject":"WIP: Adds kayobe_path variable","message":"WIP: Adds kayobe_path variable\n\nUseful for locating kayobe installion within ansible\n\nChange-Id: Ic193993c78a5650f918f9c5c40c4669e50a624a3\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f8369498aa2927b54ef779bfc89bfc0d7615cf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f8369498aa2927b54ef779bfc89bfc0d7615cf9"}]},"branch":"refs/heads/master"},"2912491cd81ae85cd8c7d9a3db7185faee767223":{"kind":"REWORK","_number":4,"created":"2026-02-24 18:36:35.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/45/977845/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/977845/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/977845/4"}}},"commit":{"parents":[{"commit":"e2279353f9e6ff0b18bade8694d1d8b10d91183e","subject":"Merge \"fix (docs): correct passwords view command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2279353f9e6ff0b18bade8694d1d8b10d91183e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:12:31.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:36:29.000000000","tz":0},"subject":"WIP: Adds kayobe_path variable","message":"WIP: Adds kayobe_path variable\n\nUseful for locating kayobe installion within ansible\n\nChange-Id: Ic193993c78a5650f918f9c5c40c4669e50a624a3\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2912491cd81ae85cd8c7d9a3db7185faee767223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2912491cd81ae85cd8c7d9a3db7185faee767223"}]},"branch":"refs/heads/master"},"155f22344f2decf7cc37e97615729405bf7bc703":{"kind":"REWORK","_number":5,"created":"2026-02-24 18:40:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/45/977845/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/977845/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/977845/5"}}},"commit":{"parents":[{"commit":"e2279353f9e6ff0b18bade8694d1d8b10d91183e","subject":"Merge \"fix (docs): correct passwords view command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2279353f9e6ff0b18bade8694d1d8b10d91183e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:12:31.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:40:28.000000000","tz":0},"subject":"WIP: Adds kayobe_path variable","message":"WIP: Adds kayobe_path variable\n\nUseful for locating kayobe installion within ansible\n\nChange-Id: Ic193993c78a5650f918f9c5c40c4669e50a624a3\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/155f22344f2decf7cc37e97615729405bf7bc703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/155f22344f2decf7cc37e97615729405bf7bc703"}]},"branch":"refs/heads/master"},"8f282a6ec594a804c4d72228c020b3cf6c0a8e9c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2026-02-26 10:37:00.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/45/977845/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/977845/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/977845/6"}}},"commit":{"parents":[{"commit":"97b5f04a41ab98a815c24ad4e21605ff68120972","subject":"Fix lookup plugins for external playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97b5f04a41ab98a815c24ad4e21605ff68120972"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:12:31.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-26 10:36:53.000000000","tz":0},"subject":"Adds kayobe_path variable","message":"Adds kayobe_path variable\n\nUseful for locating kayobe installion within ansible\n\nChange-Id: Ic193993c78a5650f918f9c5c40c4669e50a624a3\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f282a6ec594a804c4d72228c020b3cf6c0a8e9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f282a6ec594a804c4d72228c020b3cf6c0a8e9c"}]},"branch":"refs/heads/master"},"ae9e9df46bc2e85b6c9ec6fe6d7bbbe9d9312b2f":{"kind":"REWORK","_number":7,"created":"2026-02-27 17:22:10.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/45/977845/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/977845/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/977845/7"}}},"commit":{"parents":[{"commit":"29428866ded45d3450860da0e16f24256fa46821","subject":"Fix lookup plugins for external playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/29428866ded45d3450860da0e16f24256fa46821"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:12:31.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-27 17:21:52.000000000","tz":0},"subject":"Adds kayobe_path variable","message":"Adds kayobe_path variable\n\nUseful for locating kayobe installion within ansible\n\nChange-Id: Ic193993c78a5650f918f9c5c40c4669e50a624a3\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae9e9df46bc2e85b6c9ec6fe6d7bbbe9d9312b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae9e9df46bc2e85b6c9ec6fe6d7bbbe9d9312b2f"}]},"branch":"refs/heads/master"},"c3762f45dae2717b8c5a2e61dc97741edb9bdafa":{"kind":"REWORK","_number":8,"created":"2026-03-20 21:13:09.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/45/977845/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/977845/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/977845/8"}}},"commit":{"parents":[{"commit":"29428866ded45d3450860da0e16f24256fa46821","subject":"Fix lookup plugins for external playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/29428866ded45d3450860da0e16f24256fa46821"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:12:31.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-20 21:12:15.000000000","tz":0},"subject":"Adds kayobe_path variable","message":"Adds kayobe_path variable\n\nUseful for locating kayobe installion within ansible\n\nChange-Id: Ic193993c78a5650f918f9c5c40c4669e50a624a3\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c3762f45dae2717b8c5a2e61dc97741edb9bdafa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c3762f45dae2717b8c5a2e61dc97741edb9bdafa"}]},"branch":"refs/heads/master"},"f039f63d31c879b98fb7ebdd70208111636559d8":{"kind":"REWORK","_number":9,"created":"2026-03-21 08:47:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/977845/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/977845/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/977845/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/977845/9"}}},"commit":{"parents":[{"commit":"06f29c109a1d2967f1bef3fe1b880bb080e29f2e","subject":"Fix lookup plugins for external playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06f29c109a1d2967f1bef3fe1b880bb080e29f2e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-24 18:12:31.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-21 08:47:00.000000000","tz":60},"subject":"Add kayobe_path variable","message":"Add kayobe_path variable\n\nUseful for locating Kayobe installation within Ansible.\n\nChange-Id: Ic193993c78a5650f918f9c5c40c4669e50a624a3\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f039f63d31c879b98fb7ebdd70208111636559d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f039f63d31c879b98fb7ebdd70208111636559d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
