)]}'
{"id":"openstack%2Fkayobe~978360","triplet_id":"openstack%2Fkayobe~master~Iccb99cfd1a723b3680b64781488d15e417642522","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-02 14:33:12.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-03 18:16:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-03-02 15:21:47.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Iccb99cfd1a723b3680b64781488d15e417642522","subject":"Fix compatibility with python-ironicclient 6.0.0","status":"MERGED","created":"2026-03-01 20:18:34.000000000","updated":"2026-03-03 21:58:34.000000000","submitted":"2026-03-03 21:57:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978360","meta_rev_id":"d2bc2a7e01bb1c4ff46e70ef9e36e33c0801aaf0","_number":978360,"virtual_id_number":978360,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-03 21:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-03-02 14:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-03-02 14:33:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-03-03 09:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-03-03 09:01:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-02 18:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-01 21:43:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-02 08:36:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-02 14:19:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-03-02 14:29:47.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-03-02 14:31:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"8c29cff468d1a32ec8ea19b4ea71f42c1559772c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-01 20:18:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef19ac7994fd7b78ae1a0513a5eb74a888ee295a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-01 21:43:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc731918acd1473db919f2997a2eb77e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02897a2a5d714b01b137720c40e70849 : SUCCESS in 8m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/689eab55ae1147ecb5e41b9d679e38ef : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec0ea13b6c124d778e9494d7548af7dd : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/534e54bb3c9c4956814e2c2f8dea60ad : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5258939c84ba415b967bb024fda9d369 : SUCCESS in 2m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4c5de78870f34f59bbf51a56a64f9a4a : SUCCESS in 6m 50s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85539e9166ba49bdb703c2759dc66f12 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec44ba0523a34d5e863f8b719cfb452c : SUCCESS in 9m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97005ee4ee6c49e4a644c36b4897dd94 : SUCCESS in 8m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5c698ccef3a8406dba8339c940f73e13 : SUCCESS in 8m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58214c4da57644ca8608230c28c850bb : SUCCESS in 13m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b696c55fa53a4de9972de0828c0a35c6 : SUCCESS in 25m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/092983f2262c46af9f147a235b2c881f : SUCCESS in 8m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5c7a50a8c6d14a84b962485785c8932e : SUCCESS in 55m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2e3b06edb40246358cf4f57ded9efa25 : SUCCESS in 54m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/dae1f228dbde4b8880f54e1de6c3b64e : SUCCESS in 55m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/051d06d861234063b35eda38ef242339 : SUCCESS in 53m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0d030bb02a764c0586f816fb95c225ab : SUCCESS in 1h 11m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e3083cfb40b24b018bd3874e2a4c98b7 : FAILURE in 51m 54s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2e70d269b0604babaf3093ae89a0f406 : FAILURE in 48m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdd1112ec0f7405291e49b82c96f7307 : FAILURE in 56m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0735fe91265e430d878fb7a22b6d3f0c : FAILURE in 1h 20m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/57249bc8e32e4a88bf4621ec5f4245d3 : SUCCESS in 12m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/baf43381af0e47c5b5c32648fdec62d5 : SUCCESS in 13m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1238fec3efd8c85aff3638b2316ef85e4b20b0c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-02 08:07:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7f59b558a6355e2c130153ae6f4725b1331d7d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 09:32:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/432c9c4e11904adfa781b547a033767d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce2bba45cad4e45bb5843727232ddf0 : SUCCESS in 8m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2c2586b342674e3d86246b9b353bfc59 : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b24a9fc28304632bd8031a655543240 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b59002cc8854e779114fc3ce7371d66 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebf27df231154764b99271142e3ba226 : SUCCESS in 3m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/74ad13a9b11d433c8982af7258ea32a4 : SUCCESS in 3m 48s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dee762faf67b407c80091c9872636e4b : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa2dabd3e65b4f9ba7ad89b4ffa64b4f : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f5a86685f0b43438b2dc39359475ea4 : SUCCESS in 6m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6eb2c4181ce74f7f9bd2af7fcef3f21b : SUCCESS in 18m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aedfffd6bf4b4fb29c80122e54a7b32e : SUCCESS in 8m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6a30c381a461459596ff4c566610c6e7 : SUCCESS in 11m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ff7a582a7ee4860a3e31277c034de92 : SUCCESS in 16m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e2efb6ce714c41adb2f2f4b6e38ee98f : SUCCESS in 54m 59s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f2300ccbc3404015b6da9682e1f5d672 : SUCCESS in 53m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/746e9457c445442fb2c6e03ea90f5ac2 : SUCCESS in 57m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc6c741b66c54f58901bf7d2994863e2 : SUCCESS in 53m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3de494c64cc14a0780151f84247f8d33 : SUCCESS in 52m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2b8e750c5e4d4a29a8a43c3cc231f69c : FAILURE in 1h 17m 35s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/37ac7ab5b3f642e69f96cb873786ebc7 : FAILURE in 1h 15m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e501d3e0bc1f474fa787f790a6921c40 : FAILURE in 1h 06m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61a6a180c45c44318decd2524db2ade4 : FAILURE in 1h 08m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6be7ee2d7f7649d98e10152b4eb2d030 : FAILURE in 7m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d72be63c54f4d58abd7661f51b661bd : SUCCESS in 14m 34s","accounts_in_message":[],"_revision_number":2},{"id":"d9dd7c4b92fdc05cf54ddc9ad3b8bde5f539f559","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-02 09:51:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b13d1f0eb98b160df9f69c781e06a1df8fcd68f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-02 09:51:50.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4cc91727235a116a65b1c0af8c162e23cb214bf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 11:16:07.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/978389 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"28e6dbb75d4955a33ccada06cc355863e68189ec","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-02 11:17:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3b77e8ff46a0b26b5413359f0079e6e80236677b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 13:27:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed9fe3e463054710862ef52f8304ff7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8941b1fa3d348cdbda224ef88aa9729 : SUCCESS in 9m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ade9ee4c85974de184faaa98441c2d1f : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f340ad14ab5b45bf9c44b249d2a276e5 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ce54fd454484b54844eb9718c35032e : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60f7537e45c24a64a5e9d0f50d1c36f5 : SUCCESS in 2m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6a8681d9fb58498da58b2623e356cd04 : SUCCESS in 6m 38s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/448dab37ebcf408f94c9ec7ada9be86b : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a3fe5598c5045d29bfff0cd68d21bbf : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/22f16e8a24ef426aaebe413bf643e0b9 : SUCCESS in 11m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/49d9e106f73445de82f53e93451ef8d9 : SUCCESS in 14m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30652fa4e21e4a00b89da8fb5210ac7f : SUCCESS in 14m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3a06df0d60aa428bad108392dfabdc17 : SUCCESS in 12m 02s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90f5ff18473b46018e3c3e743673b166 : SUCCESS in 9m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f283f45d20be47d499c15d2b6eeed307 : SUCCESS in 56m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/56def23c6b46426c94815963cb2ec22e : SUCCESS in 55m 33s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cb5357fb72f54ddaaaf55aed3c0c9edd : SUCCESS in 58m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e652fc72df614fc5b09841cd41bc8e9f : SUCCESS in 57m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0d2038aaed3843239582e620a80118c1 : SUCCESS in 56m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/62d47a8296db4118a0e2a4648db0146f : SUCCESS in 1h 35m 55s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/da712a9669154f3aa2c57850288c7a26 : SUCCESS in 1h 32m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65f228d0cb4747d694962fd93ef46ddb : SUCCESS in 1h 28m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53799e1442664377a9f36d9236dfafe0 : SUCCESS in 1h 36m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1f61aa49a3e94d54bcc7917f9f2b0c51 : SUCCESS in 14m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1c2676e26594aae935ab1cb0bdc3d88 : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":4},{"id":"affc16ef2f04904226b5fb40a15ea2396659e685","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-02 13:29:18.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"abb0e8fa6ff7d8e8f3d5a7d6923dfbe497023929","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-02 14:19:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5cf4be959e76b6b41d1166f46befdaa4aec1715d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-02 14:20:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ecf782235fdd2ff589390c1587dea0f8c8c6951f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-02 14:29:47.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a99043ed2bed707050846bdcb63a55ee9bbe64f6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-02 14:33:12.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"90e504e617de1161b17b69da189fb1bd64decdd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 14:57:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16748240381c4821b7624d54a53cbcdc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/988533812753445e99222c7ac4aaf26c : SUCCESS in 8m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7efb3ff3245245f88ca19616f52c17d4 : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/903e87b653334c7ba7698372d116aea8 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b8c705f0efe4988b697bfbaf455265a : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2af35a76ae9f44f1bb5f2e979f7c831f : SUCCESS in 6m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/922cfeb4405e44ecb96e3fa1d240724f : SUCCESS in 7m 41s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79e5139fb0224edcbac701c998ab899f : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e739fc6c3cf048d0af8f032ca83f7719 : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4fff8a755814bef962f95ed23691f15 : SUCCESS in 9m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/173d53dba0bd46d5a8e842cb14a6ef1d : SUCCESS in 15m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eeb7222c74be414fb177dfbb4779cb10 : SUCCESS in 8m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8152f3c0e467473da1f89f1b05549fda : SUCCESS in 12m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3665f205823d472f8b222519e485c85e : SUCCESS in 10m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b0e4695d05d4457a83f58888440a82bb : SUCCESS in 1h 00m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/62181c885b1b46d190ea70a0b38ff279 : SUCCESS in 58m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/15d9266133ff4090b384675422186088 : SUCCESS in 59m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49023ed6b35f4b5591b65cd7e6cd7ff1 : SUCCESS in 55m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/80a5422d46db40dfba332c816699d41b : SUCCESS in 59m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e4577f59f1814b93839b7a5190c98420 : FAILURE in 54m 33s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/774bab5063eb40e9bb1ea1353055f9bd : FAILURE in 1h 08m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e97612d6cd1e4ade867e01355d635c78 : FAILURE in 1h 08m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec3358949b134854815ea1227720a7f8 : FAILURE in 1h 10m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/35b509f6394a4511af73481566733c30 : SUCCESS in 9m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/245dabd0c1c34d5481f5f86760d9647a : SUCCESS in 16m 11s","accounts_in_message":[],"_revision_number":5},{"id":"f011cde5f4a222d15c56223708785d7b16c6aa41","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-02 15:18:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"746fd7cc76a08ebfa4805154107a4d1fe4437943","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-02 15:21:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ed61dd997fb75a5df081fafef65fe60de47f9889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:27:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19dae2062cd64faabba84a240774bb64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b558520cc1b4efd9d889a635af2a3cc : SUCCESS in 10m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f083d66c79ec4f748cffdc5b97f00606 : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51cf1aa0abee4815a58b5835bc6db2a7 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777583136a09408bac031c39a249c563 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61ff89e871744fafb53aad923089a6dd : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bdcb84082a9e46cca5256daae76193d5 : SUCCESS in 4m 11s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0415fff91ece45319bc1f209b3862b4e : SUCCESS in 6m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8ae27ba80ad4562b4a29b7be5ffb8af : SUCCESS in 13m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/44a93b3677e6496e8e0cf6493798a069 : SUCCESS in 11m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ed89a46c11ef4edd97ec66acef3ee21e : SUCCESS in 18m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29f09c38da054dea83cb79cb83fcc83a : SUCCESS in 8m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/41056aeb5fea410da24d91d8f0352a8e : SUCCESS in 25m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ef3023230014e8b8b7e1d5eb5d499a1 : SUCCESS in 16m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a7a53418eb95422ba680637374e38b8b : SUCCESS in 58m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/43925e5bee1c4525add38b0a88c3d1da : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1c4718d4b077470bb0903032c860e9ad : SUCCESS in 1h 02m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3d19d35100c4a19beb0181b160d2c6f : SUCCESS in 1h 00m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/265121dde8934a3ba567c10f542aef1e : SUCCESS in 1h 02m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/be8aab7b91064d4ea160d1bb69415ec0 : SUCCESS in 1h 51m 25s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a1883aadcc32454eb31ff3484e13b272 : SUCCESS in 1h 36m 44s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c6038f368184832a9cec24272268751 : SUCCESS in 1h 35m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ba887742dc54d81a76bca046507142c : SUCCESS in 1h 41m 29s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/418fb474de5d413984268b38dc288266 : SUCCESS in 14m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50690c41394b41aebc553378bd671a51 : SUCCESS in 8m 25s","accounts_in_message":[],"_revision_number":5},{"id":"8ed0bdf5b2b979bee9d78b0eb1f1a1be635d0d39","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-02 18:31:35.000000000","message":"Patch Set 5: Review-Priority+2","accounts_in_message":[],"_revision_number":5},{"id":"daa9f32bee73d8b9d6cf8556d33be9534dd8f0ca","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-03 09:01:40.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4aebd4eedca80375a19df7a121d1975f60ceb0ad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-03 10:47:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9fe465b7d22b053190eaa86c1432ae8a9b13bebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 11:24:16.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/901100 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"54cc268700afe7c41979ef82692267d012eec4a8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-03 12:19:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ff7deec5e4a07e87f923ac9556f6f4cf23ed8540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 12:46:11.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bedb41517a443b9b04bb83477c42bd2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6f9669945e4b43a57cec4532908b92 : SUCCESS in 8m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/243d89a9540f41e6a9f01842a43402e7 : SUCCESS in 4m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca7c83f633394482aaa1c7414461607b : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59a43b29e7224ad793bb0e7ff2bd1891 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4129ac37c59490a8e0cd62ec2a02800 : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/36a5499ee45748e583f7af5951b662e5 : SUCCESS in 6m 18s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41c6cbc474a847ee80b60623131fc535 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7a442ca2dff4f1dbaa76cc1e5b8e656 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1ff4c029ed60450e95de3b9dc8aec755 : SUCCESS in 11m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2aac6d92ec3747b8aceb515db567d66a : SUCCESS in 17m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07d25f5352564393bb5312b7819dc198 : SUCCESS in 7m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/92e5403d5d4b4f0088e27bca899a591d : SUCCESS in 19m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b111a217c00f420e961f58e96262e082 : SUCCESS in 16m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a3f9e16600874cc6ab4a558e19b0c15f : FAILURE in 17m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cc3be245622b4b40aeb8a25a8ec99400 : FAILURE in 18m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f92cc4fd3ab74e55b591b80077c662ac : FAILURE in 20m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eeb3731d1ae4ab280c6847df4b6e69f : FAILURE in 17m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1f4b9272eadb4cacad5d4999f5a7b518 : FAILURE in 17m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/504cdbfae49b47d5be4ce1a9485f72da : FAILURE in 18m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e8f4533db2514208853e4ae083bac333 : FAILURE in 9m 26s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e55368459e44ab198c2741bfabdd52a : FAILURE in 22m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d911ca95767644298d3be5c9ad9d0f7c : FAILURE in 25m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aedec6614e474c7681acd1d44e03dbd6 : SUCCESS in 9m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da4fd1f834064551ba460f0ef9c251c7 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":5},{"id":"9a36ef3fa9b13ee8174c4a686b4d38caae3968d1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-03 13:23:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"18405a9d566d6e3d96265077dbba98afbdf42998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 13:26:45.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/901100 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"314345323e932711881fd98fe373848ef459fa89","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-03 13:33:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"692163bad3bdf7ebc7c13405760b3bf9a074d0b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 14:32:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e597f00b48940f4907ea6670418e248\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35de99400beb4da888e94d84744c8b97 : SUCCESS in 10m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ed80f82686d44dc827c2aaca8334fad : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7268933d4cf847a3997bdaa78a57ee2b : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645e3c184a6b4adc90f835062f2cc7b9 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67647b59d8e648bda8365e7966b24e36 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f802bf4b88784a668a3b04bdb7b9e977 : SUCCESS in 5m 53s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/63577eba944f4cad8e1c29354796a819 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8c17f0e0a186401a85fbad80ddf689eb : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4b40dbfa28d4eb496d9a40dcfeb852b : SUCCESS in 6m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2fcf73a9a2634daab2b08356bda0b9a1 : SUCCESS in 17m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dba052a2fd5497e9d69992797b04251 : SUCCESS in 12m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/26a7fe3b1f9a48af868a0ca3b174abf0 : SUCCESS in 24m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/357e06808fd14da89b9aeb7a61ac062e : SUCCESS in 17m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cf03a39f74834b34878cdcbb4cff784b : FAILURE in 17m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/14fcc98b6762431aaa84c5193c58f8f3 : FAILURE in 16m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/57eaea8e65aa4fa0b89660c92f4b7d44 : FAILURE in 22m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/374fa6f00c504374b357482e4fc7fe14 : FAILURE in 19m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f814efc9dd424e22ab31f7af45219a06 : FAILURE in 17m 37s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9069aebed21d4c6abaa3d56878d14983 : FAILURE in 41m 11s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e732f01f37a04efdb2c2b97ec750f37a : FAILURE in 58m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0965ee8b3dd74d58a2e696f5012f8353 : FAILURE in 34m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b99d126ef3ef4dd1826ce94c255b70a9 : FAILURE in 33m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/989912782f084397931171b96ac908ae : SUCCESS in 19m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0b740e13ec34254ada61088994a9718 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":5},{"id":"7efa60eafd57235495acfa972fd0e64f8b59ed47","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-03 18:16:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9f9cd8dbaa35786377db1881eef6d95826c9df2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:02:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9549bd7d7fe041bda730eaed099f0597\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69d69db333964919b0322a3217504ca3 : SUCCESS in 11m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99de491e04ef4baf8e3d248ae5c6ea20 : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6edff1221d1c47e6bf747193178b9ba0 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1034f73ca87d4a9eac972ddb251e8f33 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7de26ed12394e5dae57fa7fe596e54b : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4f23a6df86b146848780f424f12e4d86 : SUCCESS in 6m 33s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb882a0994894b7da5beb4903dfa00bf : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/37f5dae5414c456f819de27bd03d69cf : SUCCESS in 6m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4231159c063488e85187cba7ef5ae5c : SUCCESS in 12m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4f03b5641c41404fb7634806c80e075a : SUCCESS in 10m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8ce2bcbf96544549519f019a90d16ce : SUCCESS in 8m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e391158b92c94fda87f9a14baa2e667f : SUCCESS in 12m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c26ab6f2de84399bfedf0f80b662dbc : SUCCESS in 15m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9133ec0ab5e842d9bcf99ca9194ad73e : SUCCESS in 58m 21s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b0f11fc20cd141f29b20399e3245e0b0 : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2b84ad6208ee4b9997fe4a07959ff2b2 : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f67f0f3be73432eb46b2c40286bca59 : SUCCESS in 1h 00m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6b2eff4d2ebb4bf6bbbbea84ba26424b : SUCCESS in 58m 34s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3739e05a1b78470db50e5238854131fa : SUCCESS in 1h 44m 18s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/da89c158b9744a64b196f5c47861f7c7 : SUCCESS in 1h 36m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e15cbed55c7949229dc6407ad849f293 : SUCCESS in 1h 31m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19a72a0203d94759828dab62f0efe028 : SUCCESS in 1h 34m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9b1470d90fdc46e09474c788e787bac9 : SUCCESS in 9m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aede5bcd35c54f3db2620bf3056b8355 : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":5},{"id":"d84bdfceb25d595c9357b17cee70be9cc4bb0714","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 20:14:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6c77a1a24242474a41ccee9a180c8cb7a2783821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 21:57:36.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f73a3ff6ff3542989bab4fdd3c2d0506\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1d6eba3c3b4ed29b5b2b99ec799e54 : SUCCESS in 7m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6e08e1f5f72d44adbb978539ce837403 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6bfc86f6b9f4b269be3cf849c022b55 : SUCCESS in 6m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/317b03e17a1f48d392af85717192a7f1 : SUCCESS in 5m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b186d38f190451cac01df66c1e6b132 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/47a8ed6018d246f1a882d4eabc426a95 : SUCCESS in 13m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b0a1b84f9934fd19a3140fef51231ae : SUCCESS in 11m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d38f01589aa84945906e108c0276f8b9 : SUCCESS in 18m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50a1c5b8537a4bcf8fb9698e1244bc54 : SUCCESS in 12m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7c9c85299b944d759c89c23be8cbd2c2 : SUCCESS in 12m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d2e3fb79dd948679df24e1dc3177afa : SUCCESS in 16m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0f3dcd31df89404b90fffd15aaa2e680 : SUCCESS in 58m 31s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b57cb268a0e44deab47931d57c8b1b86 : SUCCESS in 1h 00m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9d2e9e8a892d45adb24fe2c3e59b6737 : SUCCESS in 59m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/147a336974064918898c3acee0ecac78 : SUCCESS in 56m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/daa87a6c4a484159abf41aee9becf81b : SUCCESS in 56m 40s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/921f79d03d3c4d838719fde70898e335 : SUCCESS in 1h 39m 14s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c071e9b2856d42c5a7eec56d3f9bc905 : SUCCESS in 1h 32m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dbaf93b539b4263a8e58ee3b667922d : SUCCESS in 1h 31m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e11bcede16b4e3e908667e582dffae4 : SUCCESS in 1h 33m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3fd25d3649134dc1a3a1127cf5a4ce74 : SUCCESS in 15m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aad632e69bda4ebe9b7318491d216f95 : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":5},{"id":"4717d2b744c5e56cb84434b8237f886bd40d422e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 21:57:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d2bc2a7e01bb1c4ff46e70ef9e36e33c0801aaf0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 21:58:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9119238fc4464b4981ce625c320f4384\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1006848eafa249efa779a77c0abb8133 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9342f24936b9a30867a64bc2ac4d7fa171f67741","revisions":{"4445f62107255fb51d1712781db1c283880e2654":{"kind":"REWORK","_number":1,"created":"2026-03-01 20:18:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/978360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/978360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/978360/1"}}},"commit":{"parents":[{"commit":"aa230f9d05eea695f2aaa09a348f56a2079e6eda","subject":"CI: Disable seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa230f9d05eea695f2aaa09a348f56a2079e6eda"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-01 20:18:10.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-01 20:18:10.000000000","tz":60},"subject":"[WIP] Fix compatibility with python-ironicclient 6.0.0","message":"[WIP] Fix compatibility with python-ironicclient 6.0.0\n\nThe ipa-images role fails due to a breaking change in release 6.0.0 of\npython-ironicclient [1]. We now need to access the `name` attribute\ninstead of `Name`.\n\n[1] https://review.opendev.org/c/openstack/python-ironicclient/+/973948\n\nChange-Id: Iccb99cfd1a723b3680b64781488d15e417642522\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4445f62107255fb51d1712781db1c283880e2654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4445f62107255fb51d1712781db1c283880e2654"}]},"branch":"refs/heads/master"},"ea9efabac18ac95b1830ec077f4156342a2dabc6":{"kind":"REWORK","_number":2,"created":"2026-03-02 08:07:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/978360/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/978360/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/978360/2"}}},"commit":{"parents":[{"commit":"aa230f9d05eea695f2aaa09a348f56a2079e6eda","subject":"CI: Disable seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa230f9d05eea695f2aaa09a348f56a2079e6eda"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-02 08:06:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-02 08:06:36.000000000","tz":60},"subject":"[WIP] Fix compatibility with python-ironicclient 6.0.0","message":"[WIP] Fix compatibility with python-ironicclient 6.0.0\n\nThe ipa-images role fails due to a breaking change in release 6.0.0 of\npython-ironicclient [1]. We now need to access attributes using lower\ncase.\n\n[1] https://review.opendev.org/c/openstack/python-ironicclient/+/973948\n\nChange-Id: Iccb99cfd1a723b3680b64781488d15e417642522\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea9efabac18ac95b1830ec077f4156342a2dabc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea9efabac18ac95b1830ec077f4156342a2dabc6"}]},"branch":"refs/heads/master"},"b9fc753a7e3aa2283613c6bf126b4316394d9f9d":{"kind":"REWORK","_number":3,"created":"2026-03-02 09:51:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/978360/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/978360/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/978360/3"}}},"commit":{"parents":[{"commit":"aa230f9d05eea695f2aaa09a348f56a2079e6eda","subject":"CI: Disable seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa230f9d05eea695f2aaa09a348f56a2079e6eda"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-02 09:50:04.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-02 09:50:04.000000000","tz":60},"subject":"Fix compatibility with python-ironicclient 6.0.0","message":"Fix compatibility with python-ironicclient 6.0.0\n\nRelease 6.0.0 of python-ironicclient changed the way node attributes are\nnamed [1]. We now need to access attributes using lower case.\n\n[1] https://review.opendev.org/c/openstack/python-ironicclient/+/973948\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/978389\n\nChange-Id: Iccb99cfd1a723b3680b64781488d15e417642522\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9fc753a7e3aa2283613c6bf126b4316394d9f9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9fc753a7e3aa2283613c6bf126b4316394d9f9d"}]},"branch":"refs/heads/master"},"b1bb364a32895798ce2af0852c5a52184604b71c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-03-02 09:51:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/978360/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/978360/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/978360/4"}}},"commit":{"parents":[{"commit":"aa230f9d05eea695f2aaa09a348f56a2079e6eda","subject":"CI: Disable seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa230f9d05eea695f2aaa09a348f56a2079e6eda"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-02 09:51:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-02 09:51:38.000000000","tz":60},"subject":"Fix compatibility with python-ironicclient 6.0.0","message":"Fix compatibility with python-ironicclient 6.0.0\n\nRelease 6.0.0 of python-ironicclient changed the way node attributes are\nnamed in JSON output [1]. We now need to access attributes using lower\ncase.\n\n[1] https://review.opendev.org/c/openstack/python-ironicclient/+/973948\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/978389\n\nChange-Id: Iccb99cfd1a723b3680b64781488d15e417642522\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1bb364a32895798ce2af0852c5a52184604b71c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1bb364a32895798ce2af0852c5a52184604b71c"}]},"branch":"refs/heads/master"},"9342f24936b9a30867a64bc2ac4d7fa171f67741":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-03-02 13:29:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/978360/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/978360/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/978360/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/978360/5"}}},"commit":{"parents":[{"commit":"aa230f9d05eea695f2aaa09a348f56a2079e6eda","subject":"CI: Disable seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa230f9d05eea695f2aaa09a348f56a2079e6eda"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-02 13:29:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-02 13:29:08.000000000","tz":60},"subject":"Fix compatibility with python-ironicclient 6.0.0","message":"Fix compatibility with python-ironicclient 6.0.0\n\nRelease 6.0.0 of python-ironicclient changed the way node attributes are\nnamed in JSON output [1]. We now need to access attributes using lower\ncase.\n\n[1] https://review.opendev.org/c/openstack/python-ironicclient/+/973948\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/901100\n\nChange-Id: Iccb99cfd1a723b3680b64781488d15e417642522\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9342f24936b9a30867a64bc2ac4d7fa171f67741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9342f24936b9a30867a64bc2ac4d7fa171f67741"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
