)]}'
{"id":"openstack%2Fkayobe~978474","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Ibab43c633cf718664e144470e944ef24425695fe","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-25 12:06:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibab43c633cf718664e144470e944ef24425695fe","subject":"Split Python installation from user bootstrap","status":"MERGED","created":"2026-03-02 14:23:27.000000000","updated":"2026-03-25 12:08:02.000000000","submitted":"2026-03-25 12:06:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978474","cherry_pick_of_change":976011,"cherry_pick_of_patch_set":1,"meta_rev_id":"707323bb3b46b5ee2f70cd3900bfee2501cfbbc2","_number":978474,"virtual_id_number":978474,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-25 12:06:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-03-24 18:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-03-23 13:11:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-25 08:31:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-25 08:31:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 14:23:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-03-02 17:13:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-23 09:49:52.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-03-24 18:48:21.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-03-25 08:31:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"1a2a10cdfbca995c751a215ed83f26a671bd858c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-02 14:23:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dcf3f506fdb1b4456a12096af21be8f24294f4f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 17:13:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d95c019397ae4aa7b462ed356e3247e2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aae764b53af547d8b676f6ed4ceb2fab : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39ec120b2f364fdab8dea84692405b4b : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e091e29a1bcd4e5b9f8cdbe5890ed22e : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da56d5a13544017bb1ba6647cce7d3d : FAILURE in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cf206d3559624848b2ddc9e56875e132 : SUCCESS in 7m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4dba21d002c74e17a66118cd56750af8 : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/05c47008e7b74cce8f5909cad092914b : SUCCESS in 10m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dc84d1f2db8c48d19a68bae0ed7435b9 : SUCCESS in 16m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2ba79b7505cb4ef0b49533827df9e1e0 : SUCCESS in 12m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f793029cd534fe491f121acb9697afe : SUCCESS in 12m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/27b98a27bd1b4b0785e7138042851940 : SUCCESS in 13m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f129db13eb9a450e9a3e32858f585e78 : SUCCESS in 21m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db99fd70d7ab4290b4d141df889a686d : SUCCESS in 16m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9286fa3c2a3d4884baaa78f017c8be1d : SUCCESS in 39m 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/518eae3a87554ac78d756d107c7cd36a : SUCCESS in 55m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e7e3d30511c343dfa59b3a2ca465426f : SUCCESS in 1h 10m 09s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f110d273d66b456da33810170d84a9af : SUCCESS in 36m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a3541ae6dde14ed4b91fbafd8a390701 : SUCCESS in 40m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bc0dbeb2fd474a51b563526ff9ec25ba : SUCCESS in 1h 16m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2140acbe8001476782b07b2bee3f87ad : SUCCESS in 1h 03m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/29a9780f0ec14705b362fc2eacbf29bf : SUCCESS in 58m 20s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2a5c57e634ed4b498e21b239e41ec929 : SUCCESS in 2h 30m 06s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/20088d04140a45a1b3755437dfae0909 : SUCCESS in 1h 32m 16s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9df695e66d6c410aadaa85671080f4c4 : SUCCESS in 2h 09m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac11cdadc0ff4447a09d78f5d8ee2ed7 : SUCCESS in 1h 21m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/63c37e36ea654ce38e093b03f4fae85f : SUCCESS in 13m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cb59616bf1104821929ed43155983d0b : SUCCESS in 11m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8fb853cc74e242709f156aeebe8294ec : SUCCESS in 11m 26s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/236cd064e433456d81a9a5f7d24b5fd3 : SUCCESS in 17m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42bc2f362c044c50b4b759ee45adce7b : SUCCESS in 15m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/21ebf61a4eab431f8143b3694e9dcc37 : SUCCESS in 15m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8e76e46259a24163a7f6518729a1fdc3 : SUCCESS in 36m 43s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/89a9795657dc4ee7b501a4dac7d1ed9f : SUCCESS in 45m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5459f4ed71754949bd8e0c79b659c006 : SUCCESS in 34m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/65e464e512c448aa8d1f649ea5543d59 : SUCCESS in 18m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c2d80aa5c0ea46e4a349c0a5eec08fb5 : SUCCESS in 14m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a6cfe999e2747aabe2435a12bbca9d6 : SUCCESS in 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"adbe5e96257df47d940aa3fcaa888831cdd0de82","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-02 18:15:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fac6608dcb8a91a4ac4335224a2615225f66d6a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-04 10:42:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40eb4b9ee23858eacd4b10d67a6403c3705e13a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 13:19:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6995b2896d84e41bf594fee26d31fd9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/280856ea46b04cf18edbaf2763bf96a2 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df0999dfb254bdd82de62a796b80d61 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f68235f7185e40c3b28d377e3cdbb424 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/456437ea5dfd4c2fb50756266f068dea : FAILURE in 6m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a731be0c07fe40a9bdb4e8f5bc7e4abd : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e8b6ab7029d44e0951a94b96a02a973 : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b6e27cc84bd243c1a50d75de6b445d27 : SUCCESS in 10m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d4dae593247b4847a0811e75c2419c0b : SUCCESS in 18m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3a65c0a215194b97b772b314dad6c6cd : SUCCESS in 15m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f1518bc9d1f4911b2ef14746ff42762 : SUCCESS in 13m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8c8846f0acb04366b460c4c558013632 : SUCCESS in 11m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d7bcf6bc28d845e180123b8167a49d92 : SUCCESS in 19m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d64e7e45ca64cbbba862460b4b41a59 : SUCCESS in 8m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6abdb13517994472a37a85d55ded011e : SUCCESS in 37m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b156840ebfb44fc3b3f7d500cf851c59 : SUCCESS in 37m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ea4d2c4b30f44fdab7cf4005ae4cdfe9 : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/acefa37bf4dd4ae1bf88208f14af412e : SUCCESS in 35m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6c5f9bead18044749854c03bdaf10a52 : SUCCESS in 43m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2592856a38e246d2bdc53f53a6252d62 : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b8e72562dc24db4b7450e15ca28e30b : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0b8517f7cbb144f3b3b63538a059737d : SUCCESS in 1h 00m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5569f3f55bdc4efdb769cca02369595a : SUCCESS in 2h 17m 09s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6e9e086de451474fab863e1111e3a8bc : SUCCESS in 2h 25m 41s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fabc6e0d5f742509f15433517d7ce3e : SUCCESS in 2h 31m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fbfaf1b836e4c18bc6fc0c1fba66293 : SUCCESS in 2h 28m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/685251ea5e8e4238a5df981042dc67f6 : SUCCESS in 10m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6b4cc97808534ef3942dded2641afaa9 : SUCCESS in 21m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2669fc23a024487181f687b86b02c757 : SUCCESS in 18m 25s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/2334f9e7a9314d33877b09ed29aa0983 : SUCCESS in 17m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a305db7c41f04736b7e6bf377fe0196e : SUCCESS in 21m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ca4e7a254a184376a26ceaaac75900ed : SUCCESS in 18m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1941b999e31b48f4a60be0804204c569 : SUCCESS in 35m 55s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/443145a65bf246b5b0cc32a55c0be084 : SUCCESS in 37m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd4386f9771f40a5b8789cd4cdbc63c3 : SUCCESS in 30m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4342874bc01442e1ac3b0f83f4d8d5ec : SUCCESS in 9m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/33ae9f39fef54c4f867dca1e2334b210 : SUCCESS in 15m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84cd45d49ed54891b82a74db41ddcdc5 : SUCCESS in 12m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e98b5aa0df9f6bb2fcdc1d8450e77e886b08af56","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-23 09:49:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5740570114412d6f38c34b4a766659cc0aba5ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 12:47:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f22e17cca04ecf80cb0435f542ea91\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beb3bc31e7e74b05aef9539ca6fd9da6 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bab79eb41df4aac9d8e16fbd9935000 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ba46a7fcb074a5cb28ebde89f8506af : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece492225587404aa70f9315c88be323 : SUCCESS in 9m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da7e837e06b74442b9efafbf0bac48c8 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa5ca78cbf6e41c5bf5a0e86f9a27f53 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4bf5ed86af8b4d99b3f5efa5fc9bdab4 : SUCCESS in 12m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ba8d10ef88744e149ced04ffd5ad025c : SUCCESS in 9m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/629cb4b143864d12939a46211b1056e6 : SUCCESS in 13m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/302a33683085439fa9101a78c9191a13 : SUCCESS in 7m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a10374ee768748c782e81e45e23cebda : SUCCESS in 24m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d5bd6046f65545e58b11f553544d5903 : SUCCESS in 19m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef1c9adb4ac54f0f98a4fed11efaf74d : SUCCESS in 14m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a6b8bc63177343cca8d80e1f76e674ad : SUCCESS in 33m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/709d4cb768af4d3fa938082603cc6cba : SUCCESS in 36m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/63a92b4193ac4b10abbdc99dfb918f97 : SUCCESS in 34m 02s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/2bedba54b6324e0db15b33497711edca : SUCCESS in 1h 02m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1954ca0ea4bb4fb8b17a142935823ff1 : SUCCESS in 59m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8faf66b1c2ba4c04bccea81f2c950e6a : SUCCESS in 1h 12m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e9438fbcbcf4fe0a6c0230cc5496923 : SUCCESS in 59m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cf55b2ee43bf482aa9170a7c07940d13 : SUCCESS in 1h 07m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b0ed9b9c38114c4f840298d231b225ff : SUCCESS in 2h 22m 15s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d814742ad2f24f70837c77680b840904 : SUCCESS in 2h 35m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9bdd84fb68049d4aa3e381ed130515c : SUCCESS in 2h 05m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a46471948f74722b8fe4bb1be3cdafe : SUCCESS in 1h 13m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ffa7f0b1a0144481920863fa1908a9eb : SUCCESS in 22m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6bf14649a2584eefa7dcae8484093aa1 : SUCCESS in 16m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/719692e3880a460c992be76ecfccb86d : SUCCESS in 21m 41s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1245eb5599c446058c2737f8d356f21c : SUCCESS in 22m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c8d9eafac67403eb1aa99274aae8f81 : SUCCESS in 16m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d37e387fd154e6d9ae58c0acc7bd234 : SUCCESS in 16m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9f148d5a885a462f98245c124dd47c63 : SUCCESS in 35m 43s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2565f363b4374a73bfc9e5d54be06bbc : SUCCESS in 35m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df2512f9eba6457087ac3d456a0fc54c : SUCCESS in 34m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/92fa9aece26f4f54b0bf26ce55b82a43 : SUCCESS in 14m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ca1321bd8d4749a48a02ed7dd99d3534 : SUCCESS in 8m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/530c4124e7704b98ac4fd97e7c9c702f : SUCCESS in 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ac69a86cc9bb4e5eb9ca8207891b8d4682eb307d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-23 13:11:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d2213102b15ebfce371f7bc4010468a824367e4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-24 18:48:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6ecb9a1a90313f6a3959e3e225150301df1f7ba7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-25 08:31:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56baeda63a9c730fa7d5470dadf653a8bd5d7d67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 08:32:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29f4e1fb2579c08569177edbc0ba49cc3533605d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:06:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5137b0767bc84d07ba5e2cffaf6d7054\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d521290b5112445c91749b5b7d9a0a62 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6bbeae66826421eac61975160b612fb : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7da0072556a435ca0fe84ed7af5bf52 : SUCCESS in 11m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/07fbf4cd2f70410387399dd8a68c1c2a : SUCCESS in 8m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f8affd702434dcbbd05ea3557543f8a : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f04e5c992644e10b505ad79b520d71a : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8c85f9d1a4a14c7ead3c2f2c20d3d44b : SUCCESS in 8m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f544a083af234c958b5a736b0ebd5d7d : SUCCESS in 15m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e952d50fa0d498dbf6554f4dce8d9ee : SUCCESS in 13m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bba53b357d8d4742866681919a5fe1dc : SUCCESS in 12m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/331564c7fcdf4e758a34da5f82e4dc71 : SUCCESS in 20m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e64f9129bdfc4441b2f58e042d401c61 : SUCCESS in 15m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/837b638c553c4ff68699cf463b9015ed : SUCCESS in 31m 59s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/df12c1a77472481c81ce80dd7ec9e321 : SUCCESS in 34m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/34a8a3d5b5b84971be401699fd9fbf15 : SUCCESS in 31m 57s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/895b1c7eb2a64c9f87d9958bf34e61d4 : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/813663f7476b44c1b86d410d72093ad9 : SUCCESS in 36m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/42ed24f065884ccb8c54b64d83d760e5 : SUCCESS in 1h 11m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b968ea585d4f434685a5cbac4ea0c560 : SUCCESS in 58m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dc20cd52af6d4d85b89659799618c924 : SUCCESS in 57m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a77e13ce22ea4c8fb3e38373c26862e4 : SUCCESS in 2h 23m 00s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/327b8c20d8dc4b4da72ce5005e6a1723 : SUCCESS in 2h 27m 15s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/586f3c7d7f974d0882753e83916b6cfb : SUCCESS in 2h 27m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75722557afc54bc7a8d972c63565c4c9 : SUCCESS in 2h 10m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6d64a5492a2b4e5b87bdd82dee25f4ce : SUCCESS in 22m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7f872e6a7419411a9eda2e7c07538c20 : SUCCESS in 10m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a48f76f48c6c4a5d98a37ce72b28f0fc : SUCCESS in 20m 31s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5ceff24180df4dea9746c59bb1074cf3 : SUCCESS in 21m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40807d07e19c4102b9abeaaf18ddf38d : SUCCESS in 18m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/250826c9d6dd4dcdbed065b07a1e24e1 : SUCCESS in 15m 34s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e12a5106bba0428c82f111743240ac55 : SUCCESS in 29m 07s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6bc12a39a629411db4d999f5aa001d6d : SUCCESS in 38m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fba12738096d4a18beed06d098fbef47 : SUCCESS in 31m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5679b261793d4908896442764d280d6f : SUCCESS in 15m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4060b8908fbd4914870d4dddaac3b617 : SUCCESS in 17m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f69fa90e48b4df6b662509d57a85c73 : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5788ff507e8bfac2ebde3d9f57279c0ea3fd0756","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:06:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"707323bb3b46b5ee2f70cd3900bfee2501cfbbc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:08:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/103e92d86be24a50968de7ccfaae0239\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3975bc2958fc4c31a274f47d671fc3ec : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f7a2a22e57f32205c0cfe0f448bb9840f71ada6","revisions":{"1f7a2a22e57f32205c0cfe0f448bb9840f71ada6":{"kind":"REWORK","_number":1,"created":"2026-03-02 14:23:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/74/978474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/978474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/978474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/978474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/978474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/978474/1"}}},"commit":{"parents":[{"commit":"d850105571080d48b81693574c2e2b4d4d0843db","subject":"Bump stackhpc.linux collection to v1.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d850105571080d48b81693574c2e2b4d4d0843db"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-07 21:43:41.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-03-02 14:23:27.000000000","tz":0},"subject":"Split Python installation from user bootstrap","message":"Split Python installation from user bootstrap\n\nThe check for the absence of the Python interpreter cannot run with\nansible_python_interpreter set to /usr/bin/python3.\n\nChange-Id: Ibab43c633cf718664e144470e944ef24425695fe\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 981d63926c421a5a39bc2cf16394a5b668b94aba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f7a2a22e57f32205c0cfe0f448bb9840f71ada6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1f7a2a22e57f32205c0cfe0f448bb9840f71ada6"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
