)]}'
{"id":"openstack%2Fkayobe~978475","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~Ibab43c633cf718664e144470e944ef24425695fe","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-26 16:05:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibab43c633cf718664e144470e944ef24425695fe","subject":"Split Python installation from user bootstrap","status":"MERGED","created":"2026-03-02 14:24:02.000000000","updated":"2026-03-26 16:06:16.000000000","submitted":"2026-03-26 16:05:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978475","cherry_pick_of_change":976011,"cherry_pick_of_patch_set":1,"meta_rev_id":"6638e8fcd84e6d541ab2502bc3518d29fc5d637f","_number":978475,"virtual_id_number":978475,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-26 16:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-03-26 14:12:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-03-25 12:40:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-25 12:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-03-26 14:12:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 14:24:02.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-03-02 16:28:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-23 17:43:55.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-03-25 12:20:10.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-03-26 14:12:36.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"57b000d4005686e45d2f73a8ba8f7bc8088ad6ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-02 14:24:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f4b593a3b0bd866ecd63a7d7e78e487cb17fc405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:28:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b611827306f54dc1a669ea2b912d7285\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a11f1158a8f445db34da9366b1f1945 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce68f903765147fab2226ddb017f5229 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fc8ceec8c1c4b2b856cdbf1ac0d779e : SUCCESS in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/669df344c712439f86105e9e5ccbc8c4 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2c9cb44d8534f6a820f914ca8a76398 : FAILURE in 4m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2715790a5d0949e692ca4534e7630bf0 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9aae6a3b611e43c5af380d01aff3c61a : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2fa9a6da1dc64e7c93e5544b1fe282fb : FAILURE in 6m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/233ae837387f43eeaf2c01b20f16a8f8 : SUCCESS in 1h 09m 42s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e7f4d0c692824cb8b39e5d903b8b09c2 : FAILURE in 10m 38s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d72b777d44eb4d5cb902fa8ab759c7c6 : SUCCESS in 1h 05m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c94480df0b764beb9632e7079f38af34 : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e314c6ff7b364354929b3db33f200f27 : SUCCESS in 19m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a158f66e5c8946a6a1bd4489504b207f : FAILURE in 10m 06s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19f80437bde946eaba77e9c454a0da63 : SUCCESS in 13m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7655170b71c44f259876a4cc25ae4102 : SUCCESS in 1h 55m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7af09bd84c0e4efd8a1c10eb74abd4c3 : SUCCESS in 1h 53m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/af52863611804f059ad2ddc6dd841ce6 : SUCCESS in 25m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41a7a9c8472343589b643967a98bc003 : SUCCESS in 20m 30s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/77bf615b49f94218a9c6198638a61fc0 : SUCCESS in 39m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35461624c7b346c4b020f66ecb5a4486 : SUCCESS in 26m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/20b572c6fbcf44519e47a3474f6ca686 : SUCCESS in 19m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a478a7d22534eb09fbea9dac479da85 : SUCCESS in 15m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ee9cdce35d9f43e1984bb8fed45af389 : SUCCESS in 17m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a75785f3f74546c9a3551d814f6030af : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":1},{"id":"56f5b123a0b5cc7449d565234665df12760b44b7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-02 18:16:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02fb777a620a6763b273f30f78b293c19cb88b8a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-04 10:43:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"927bed67829c73da4b38cb53833fdd4584f9915a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 12:24:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccb6a5bfdf484096a6d0199490e62e01\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb3b384f09354cb3a35c909777d74434 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618c56e652a64582b7249b3d8a97b9cb : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/690d9ebeb17543e4b10de22df3d24d18 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/001acf49a466431d881f7bb1e79c5705 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef69727231c745258207b1759c46bcae : FAILURE in 4m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fbe5387987044febaf6811d5a47c4102 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/61826c050b0e451cb50f52d9bc99661a : SUCCESS in 10m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7de17e0b607b403b8f898d0d2a77a29e : FAILURE in 6m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b0a78a0ffed041e086c38aa65f3d4142 : SUCCESS in 1h 05m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e008bf67194c4495bdae2c607070d38d : FAILURE in 8m 31s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd3951809d474db2915eba5704004034 : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/58cd9cda75634af582e7abd32af9dfdd : SUCCESS in 1h 14m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9f40352ce5b845e9a5df984cb370f306 : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/dbaf9baecc954d909f7366d336e0894d : FAILURE in 9m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc8aa36221604e5ea9cfe681af3ccd67 : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/857eb1931e1c4a9a90e2228e89a6c491 : SUCCESS in 1h 12m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a413650f4fcf494cb865d6f497e3d177 : SUCCESS in 1h 32m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/694a93a5afed455aa3ca1f4c521db649 : SUCCESS in 18m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/501029f308094d07982a56312e122273 : SUCCESS in 17m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/632174c8628749c3955f7575f192ae8a : SUCCESS in 21m 05s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21cf2a7398c94b8084c0d138a75b3138 : SUCCESS in 30m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d72155f9d7c24a84a9c73e944b07c5ae : SUCCESS in 19m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0dde8b973294e0b854defc0a9fdae68 : SUCCESS in 14m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec23b1df55cc457fa6a6f0143361a5f2 : SUCCESS in 21m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34b0d03c47ad4a6b8ebb6e9ab207108e : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1893754c62b115228abfd5b704197a2b2bbc3be4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-23 13:11:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2a163d0b41e7e4fdd6f1e085298724a95664571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 15:18:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f241165028054976b5b1a5a664cf1fdd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6573f44736904f6c9cbe719daa22150a : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c717d245818742e091ed33db6f38a69d : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3d42648924946e987ebcb4249a9622f : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a3dcf4f398c4dccbe8152ad4c255653 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d041226f7e84d52941c83cda50a2544 : FAILURE in 4m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c4ba966c02542ea9a230b199e2f02a7 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/08261b3ede374b0a864577391ca15e65 : SUCCESS in 12m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/35a6538e01b046fd82410afbc30d0def : FAILURE in 7m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/45c9125378394438aacee2d40ece3dce : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/daa5f8854abc4233b7f06fd7ce877665 : FAILURE in 6m 53s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1e7f6a62575431ea836f5dccc385e03 : SUCCESS in 59m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8eb4c1ce36e7451b84f4718e996e6982 : SUCCESS in 1h 03m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/198a1cec8f5d4e29bf3e225a67eb5532 : SUCCESS in 21m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b60e7cc8069a4fb4a57ed46bfde14d26 : FAILURE in 10m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33cac8d780c3461682a50a604ef162c2 : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/52688aa896dd444d96a80ec27d012575 : SUCCESS in 1h 56m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a312c6d3e7046c7be4233466f46e9df : SUCCESS in 56m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/706404f6d81f44108c5c41f43ec5bddb : SUCCESS in 25m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92567f40ab5043bfa3f6410abb153c57 : SUCCESS in 12m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0d98f96b45a14af4acdae4d263bcf75f : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab0ab7dfc5ad4b68b6aa199c16c6e9f3 : SUCCESS in 18m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/94735c38bfaf4bdbb0eb56c38e28ad8d : SUCCESS in 12m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e127b0feaee942f3a101faefb5441faf : SUCCESS in 12m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/47f6e203d10a40daad6358fb4623a26d : SUCCESS in 19m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b20b4ff2f9c4c0788b1ee98387e54fd : SUCCESS in 15m 57s","accounts_in_message":[],"_revision_number":1},{"id":"6f9bc0eb2a0dcd1702151a1c0dfa3e6c94d6259b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-23 17:43:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"001d19d15cbee6bc97f2b2aa7966d9fb0debbfdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 19:58:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebe653f21658470fba1d20d4038c8f2a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c971505c5284f1e90efde96053c4dea : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfd1c3e4284b4452ac80942f1e920495 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7972edb24df04fcb946d829b38900495 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e2a2a2eabc1490faecda09d4406a90a : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd0cb82a416445c3bdf8de7572c0e155 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/033f0604ea1d4385ba691c5aeba11e90 : SUCCESS in 3m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/efcb10b32ceb4b26a2cb5520a562ca09 : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a32e2149adf34320931b3375869dc66b : SUCCESS in 11m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fb3e747c9b09445e8096c92daa2a68da : SUCCESS in 53m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ce0f6c79dfd24f9891a039dc2bf3f996 : FAILURE in 5m 15s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b307e8185de54ef480a593af957819b0 : SUCCESS in 58m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/58ee9c7d918e41bfa47de5413b767584 : SUCCESS in 54m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c9f139fe55b54cb5adae1d918028a3f3 : SUCCESS in 18m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d1a0260db03e4dec859680da0006d1be : FAILURE in 8m 57s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/276dd6f6d9d64c6db7afe4a7d20a3434 : SUCCESS in 17m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a01cea0c5604164b27987334f222ff4 : SUCCESS in 1h 55m 19s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ee18a7d45cf4dddae7c0be6675d1847 : SUCCESS in 1h 32m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c5e22a1534884f868ad6dae984f3c49e : SUCCESS in 23m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20409ab6c802412e96fb666184df6955 : SUCCESS in 19m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bb52d13366314fb7be2fb596f06bab10 : SUCCESS in 26m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c04de1e10344fdcb5e39749a47e5412 : SUCCESS in 32m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/45c0d8802c0b459492e8d613281d572d : SUCCESS in 14m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c3e4c978cf84dd8ac5ac7fc50f67def : SUCCESS in 17m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/40ca29d9db4949bb9d9c4e3be0ef37d3 : SUCCESS in 12m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0dc09190a0c470aacbd4ca1d2dfa4cc : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fe5e6511ecf6e6c3b2a3094cf2bc29d46a75a27d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-25 12:20:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c13fee9689431a4a241f5d1814f4bdfb15fc09bb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-25 12:40:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ac8302797a166fde6f8314cbc91e8fcf015a9738","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-26 14:12:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bde55e063f9bd703d073aa65107f4aa714a05c73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 14:13:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f617316eddb0ee2adc7dfc38186970e21de1b862","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 16:05:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e613ea2f50d4ebea73a80438e8544e5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5548c7e60ac490698d7e490876dc1e8 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa1a387beb0f4ce19bfd0eb5c0d0c991 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b59f14aec046509aeea7e3b626c392 : SUCCESS in 7m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78afc571a309406897d1fc80978ad4c3 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78c118f5d4c947dcb779dd133764610c : SUCCESS in 13m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aaf2cb456df54119983ce7470574149d : SUCCESS in 10m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1131614047e6405ea31998ee335baeed : SUCCESS in 39m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cfc73020cf64debb5aec53d637e740a : SUCCESS in 1h 11m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/806a3dab17f1403798be838a8c4837ae : SUCCESS in 42m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/055ff5e042b54e7ab3533ebb4296ae53 : SUCCESS in 12m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1db6bd36d9da4355b7b5797eb22db6e5 : SUCCESS in 14m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/85758f77cc744500bbe4e2a11ccece5d : SUCCESS in 1h 35m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4f255bd29284bdbbf087a3c82dd5c0b : SUCCESS in 1h 46m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f645b53f0f5e4200916f0753a9e5d998 : SUCCESS in 17m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5aaefb9f86194ba0a67b41e80e1ae8e9 : SUCCESS in 12m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/057ae8155aad481f95347b15e0d8c7f3 : SUCCESS in 37m 04s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b61496d2dac14ca9926990df088bc3b0 : SUCCESS in 30m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2852ac1c7b9d4f67b97eed2ea4152575 : SUCCESS in 23m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/809dace0ecd54014a12f9faa22366cff : SUCCESS in 17m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0ca580e2712b4f8eb7e2847a0da09863 : SUCCESS in 18m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b4feccf92cb48568a5b1a890c4dd59e : SUCCESS in 11m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d37920dc0bb75f09565471260007603b84150de3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 16:05:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6638e8fcd84e6d541ab2502bc3518d29fc5d637f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 16:06:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bb852476a1c4142b0564d73b807e84f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39da83ece1014401818810060e329d2a : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6b2b391cf423b9cb8306340cf2612f1abab6bf2","revisions":{"d6b2b391cf423b9cb8306340cf2612f1abab6bf2":{"kind":"REWORK","_number":1,"created":"2026-03-02 14:24:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/75/978475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/978475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/978475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/978475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/978475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/978475/1"}}},"commit":{"parents":[{"commit":"168c89acc3159c7a1069fe67dddbcff2d8f2af5a","subject":"Bump stackhpc.linux collection to v1.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/168c89acc3159c7a1069fe67dddbcff2d8f2af5a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-07 21:43:41.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-03-02 14:24:02.000000000","tz":0},"subject":"Split Python installation from user bootstrap","message":"Split Python installation from user bootstrap\n\nThe check for the absence of the Python interpreter cannot run with\nansible_python_interpreter set to /usr/bin/python3.\n\nChange-Id: Ibab43c633cf718664e144470e944ef24425695fe\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 981d63926c421a5a39bc2cf16394a5b668b94aba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6b2b391cf423b9cb8306340cf2612f1abab6bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6b2b391cf423b9cb8306340cf2612f1abab6bf2"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
