)]}'
{"id":"openstack%2Fkayobe~978476","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Ib2f9c10433c42f783e1d5f618291ba5cb7be1641","project":"openstack/kayobe","branch":"stable/2025.2","topic":"apt-proxy","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-03-23 09:48:52.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-25 12:06:08.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-24 18:37:05.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-03-25 12:06:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib2f9c10433c42f783e1d5f618291ba5cb7be1641","subject":"Support Python installation through Apt proxy","status":"MERGED","created":"2026-03-02 14:28:23.000000000","updated":"2026-03-25 12:07:32.000000000","submitted":"2026-03-25 12:06:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978476-apt-proxy","meta_rev_id":"caaa95a8187f79b6b0cf5437303f98192a77e769","_number":978476,"virtual_id_number":978476,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-25 12:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-03-23 09:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-03-04 10:39:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-24 18:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-24 18:37:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 14:28:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-03-02 16:14:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-02 18:13:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-03-02 18:13:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"fbdd7761b5825e2cfa91b528068fb3ba187f9c05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-02 14:28:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18d2815f3cdeb85737afc14512f5af87cd326f62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:14:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eafb971587548ad900fcfb084ecd969\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d3f2d96fc446688cb20b782be30aeb : SUCCESS in 10m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/224ff5eef63343c9b98fc30b0065a9eb : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/687d72d1f8c545d59ed26ac3c3c7f6f9 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1bc9980e0d4441195abb19e8385b556 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/47201624bf5d4d8c94b07514f07d5e4a : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/903018e63db149ef9916e704829b8f30 : SUCCESS in 8m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6033f1e1eb52450a90584188b0063626 : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b619d9c46d2940fa94c849e72d4d869e : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bbc8d17497d54842a4191cdeb8d2477a : SUCCESS in 10m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ecc3d3335fe49a783e15f54bcece1d5 : SUCCESS in 8m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/598cf7175ca749a2abbc76b59bc82bd3 : SUCCESS in 8m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa15e7bc0cc2436d929565ecdf8f7f9d : SUCCESS in 10m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1359dd7737f54b279655534f36dbb95e : SUCCESS in 26m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95543570a48644f59be9457fb38dc571 : SUCCESS in 18m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f3d4b1befc3a4c76aa3dd0d5a6bb9e35 : SUCCESS in 58m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/811aab88f9884c4ea701122dee0e01de : SUCCESS in 1h 00m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6917502a7c0a4e5189c240383b0ef6b5 : SUCCESS in 1h 02m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/747291a3925b453aa09d93afe7c521e7 : SUCCESS in 58m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/75aa190d5368429f918eede28fa13cb7 : SUCCESS in 56m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/686075e9dbc847039e25b095964c12ef : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33e5758dea134054b021022936f33c7d : SUCCESS in 1h 22m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cefc3778987f438e899bf59a05d04030 : SUCCESS in 17m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9b60130a54bc4591ad598a6fcee0bbbf : SUCCESS in 23m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5aacba8a0c54f00a4eae838b8fbb2f0 : SUCCESS in 16m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/68e798d736944a5bac57c8f5598d9e03 : SUCCESS in 16m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fba5abb4b3d84502b46bca1003b053da : SUCCESS in 29m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ae42076c3834aad854f2302b35d3f22 : SUCCESS in 15m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/00c304729b8d4c6da8f06bddd8e1a0b7 : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dacfd8c4663413c9dadd7223688b065 : SUCCESS in 13m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e76acba044f10c9696c0c1dda4e4bd70fe30514c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-04 10:39:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"530ec7a7f2e66aae3e02fe0cdaa1dffadf338bae","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-23 09:48:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c95605c82a13d10a5cd22227a0081b3e3746d9df","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-24 18:37:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a1ac85b044a8946afcf810178ecf8fb719a673b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 18:40:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa619554d90bd6615819083b95e751b66a2c993f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 20:05:52.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe/+/978473 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"93c0a368e6c3b760aa89b4a528d12b51120e42eb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-25 07:28:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd6f50ad28d29641dfc0e16b979c75138b8dff8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 09:01:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8745e74da9124cfd8d642c9f806dbd31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70b2517b2d764afa967c63eebe6682dd : SUCCESS in 9m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/54fcbf7a5758438098a5461467fc9ae8 : SUCCESS in 7m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b84503326344a2fbf20779613f0154e : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8052e4b42415472cb4c800769511931c : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a04608758a5847fb82869b8dcb433096 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d348721757a94ca885af6bdd450f09b3 : SUCCESS in 6m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e5549d377f5492f81711b99be589efb : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/028d851e59ab4d5c968ee820d55c051d : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/421817a8637548e4846cc0ac3b3b9446 : SUCCESS in 12m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/adfdb9eb3378471bacd731ef7a8e36e9 : SUCCESS in 11m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/43c6fb009d9c4fe8903f2171ecd3a5a0 : SUCCESS in 18m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ee346c5b1ba4d9f84cec6ecfecd8e03 : SUCCESS in 13m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/52c5dfb8c636409c8604d60557c88f9f : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/664cb44dda3b492f8436fbed5a637a39 : SUCCESS in 15m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/be9f9c0d5a624199b9d0daad21b5eb63 : SUCCESS in 53m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/88e1915d25e0442e8eef25c389466063 : SUCCESS in 54m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/de908523bdd4458589e89c5c5f5bc9ff : SUCCESS in 58m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49e0cda80b6e42cb84b5ed98aeb5f813 : SUCCESS in 59m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3f5bb034731417f9bfb868fc0b3f3a8 : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a65225d10f464c5184bf34e1095e26fb : SUCCESS in 1h 31m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32f852f6e5eb4a7d8800ebd0fcbe708c : SUCCESS in 1h 28m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/da175ea53f8c4a968f2253a757965fb1 : SUCCESS in 17m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6913e0c836ff46ddb6609b6043a6d785 : SUCCESS in 16m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/758f59cab08d46acba66a50489426199 : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c2132a67d6a341afb775b1d9d3dd4c34 : SUCCESS in 20m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a042da0d199a4cff948d1158cf111670 : SUCCESS in 27m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a35dc14c726f479f83808abb35e6fb01 : SUCCESS in 28m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c96e109bb03d4dbe8762f833c787e891 : SUCCESS in 19m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/110f36c31c484253870e248719fab684 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d042bf62b8538b717deda3a3d03e5e84a07f434e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 09:01:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7055e2a5da0b2cd0d6e30a2ca3ebaa5b7aa6f536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:06:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/742e92bbea214583a34adeca79067a86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78098d4eb6ca469caa6b20bfed430da4 : SUCCESS in 9m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/19c76d9198cf4cedaa7a1a490ae69434 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/298de680db85462593074b107fd3dbca : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/903f34088ead47c091aea379694fb3f3 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aaa4356b10e45d88c9cce58d852e2b3 : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aff1226613014677809b1911b3552063 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae200131865c4f7a848404079e73c611 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/93eebb07c45b43e79ee3d9ff2716e8b8 : SUCCESS in 11m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/27774cea70d64808866bb6edd316e28c : SUCCESS in 19m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6273bffc212a4818b0af56c6109ed2c0 : SUCCESS in 13m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9f9bbfc888d8437aa7d3f19f2fa53ad7 : SUCCESS in 11m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c2807c962134d5981e19086c36ee51f : SUCCESS in 19m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9c2a000981f148e8b9c3b8540e75f669 : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5fdce340e5964f9db28d0b15bd03b1b6 : SUCCESS in 1h 00m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/37574256c1c345078e8d4ab99a1e8059 : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0a4f0a698ac4f3ea61cdb727f465654 : SUCCESS in 1h 21m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f20a8273afb84207b095ab8d6e0416a7 : SUCCESS in 59m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4c2523f619684ea1a9674f78668b99b5 : SUCCESS in 1h 31m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4df4e2118de9410ea948a22efe1787aa : SUCCESS in 1h 47m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/473f39b6f9094654b2a40059bc390aec : SUCCESS in 11m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/166eb8fd9bab45078b3c978ade83d597 : SUCCESS in 10m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/154ee92896bb46cc966911849a9d9077 : SUCCESS in 21m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5e646ce5150946a3b7385c652183b222 : SUCCESS in 14m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2972edf5df8b475b8c4cfea9ce36ee6e : SUCCESS in 18m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1671fa4325dc43e1af297c6c45bab7bf : SUCCESS in 15m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/48d6bf9674c74f42b137842908f3cd19 : SUCCESS in 20m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3995dc35ef6d490689d22801221890c4 : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ceb69d9e7c560a4edc5cdcaf548ed76dfa373ec8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:06:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"caaa95a8187f79b6b0cf5437303f98192a77e769","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 12:07:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4986aab90bf94620bb7c532b25ccd2bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb58b74947ff45c6a3912b7307122abd : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c6f5c6b03c441d0aa0a8eed504ed3b9 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"147ff934d0dc8881b093acea99f65bfe9d926d14","revisions":{"147ff934d0dc8881b093acea99f65bfe9d926d14":{"kind":"REWORK","_number":1,"created":"2026-03-02 14:28:23.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/76/978476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/978476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/978476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/978476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/978476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/978476/1"}}},"commit":{"parents":[{"commit":"745f810b3464ace29664ad286b6fe90719c2c01a","subject":"Split Python installation from user bootstrap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/745f810b3464ace29664ad286b6fe90719c2c01a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-17 20:50:32.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-03-02 14:27:49.000000000","tz":180},"subject":"Support Python installation through Apt proxy","message":"Support Python installation through Apt proxy\n\nChange-Id: Ib2f9c10433c42f783e1d5f618291ba5cb7be1641\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 8b36acddab1fc16c35ef86852c6a816da3617546)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/147ff934d0dc8881b093acea99f65bfe9d926d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/147ff934d0dc8881b093acea99f65bfe9d926d14"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
