)]}'
{"id":"openstack%2Fkayobe~978478","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~Ib2f9c10433c42f783e1d5f618291ba5cb7be1641","project":"openstack/kayobe","branch":"stable/2024.2","topic":"apt-proxy","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-26 16:19:18.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-25 12:20:19.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Ib2f9c10433c42f783e1d5f618291ba5cb7be1641","subject":"Support Python installation through Apt proxy","status":"MERGED","created":"2026-03-02 14:29:34.000000000","updated":"2026-03-26 16:20:36.000000000","submitted":"2026-03-26 16:19:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978478-apt-proxy","meta_rev_id":"39f4d661bbc241fd0ebae82b348d6fa233a9c602","_number":978478,"virtual_id_number":978478,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-26 16:19:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-03-26 14:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-03-25 12:41:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-25 12:20:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-03-26 14:12:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 14:29:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-03-02 16:52:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-23 17:43:59.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-03-26 14:12:44.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d882ad952fa414b9ffacbc746a6221c3e5caa71a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-02 14:29:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"613c98c1db3242f797cf09ccd7fab3946ebecf56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:52:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c54c44675cdf4b09b3e411ee028fc4e6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb4ad2196e24458cb527da82a887b195 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b291081d4f7d4f5bb0a1c9e2db627f18 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b02baf6947c34e5ebd87ad2afe85dca6 : SUCCESS in 7m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edde1d230e0a45078346a4007418fd40 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d636593446c24937a3cb49e4d2de4f98 : FAILURE in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d8e30c7349247de906b995f9467882a : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/253ca1565dc84843a4ca91bfce42222b : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c47050570ae42178b4f608c13539f1f : SUCCESS in 8m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aff95545426947f9b7d3d231ec4b251e : FAILURE in 6m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b646392527394b0787a16d3af4b8c9de : SUCCESS in 58m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0d29d5a760994f9abe12630c7dfd6961 : FAILURE in 9m 53s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40c6ce74368d4fa99b33dd6942bc680e : SUCCESS in 36m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d52809e730894408ba916c0d780041f5 : SUCCESS in 1h 09m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/724636c7dd2b4bb5a04de9c33691d08f : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d50964232a2d4df082c55cc75727fa77 : FAILURE in 7m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1612182bac0847eeb98c433533e96774 : SUCCESS in 10m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3614a4d89c3b44b88b5c73a36bd2e991 : SUCCESS in 2h 02m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21faeb8e038f42a3a5a779a676607552 : FAILURE in 16m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f4841b5ba45341b8ab455edcba6d51fb : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33dd1817fbfc4f30a980c4cbb111e35b : SUCCESS in 19m 33s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/187f781687be4e72a761c7c606c5a88b : SUCCESS in 29m 04s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b719e86442184b0a977bd5a2c5a2fc16 : SUCCESS in 30m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/299dad8f2caf43fa997ee5b8ff65d4ea : SUCCESS in 19m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1b1f9cbf8834db8915819011be402ec : SUCCESS in 15m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8b6f9cada3e0430ba1e3b908cd3ea9dc : SUCCESS in 18m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7784e71bb4514bf7b5ba0ffaa6d2074b : SUCCESS in 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"765418130b5db3c25b668f7b3f2af4feba431e6a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-02 18:16:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6c7100d0881ace97f38b989b76d0bfe78ec4cd8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-23 13:11:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0335a629a3e88c0850193afae19b4326976f650f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 15:12:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34c2cb839166443296f2ad2f69817b81\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca753fdeba57462491adf6c97d6f6aaa : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f23073227de4c9c9495c78d4e6d4c2d : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67fa6a865a9846ffb86f7007423a4317 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58977daecdcd41209f3cbc2c4a905c02 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3abd6ff8a20426ea584890c3647f18f : FAILURE in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc13a3a0d9d940738a008bc122a9e007 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f2900318ab243d081976d1a8cd9ee75 : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71eada404be4479eb00da927d0b7420d : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4a5dd77aab0d4355944c72f4761c5cef : FAILURE in 5m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d5396e2c6ee5405d8df768a16ca6b611 : SUCCESS in 54m 18s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9724d6b93ae04f739d47a6ece9c0cca1 : FAILURE in 10m 06s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5a8d3416a954c24a0029a473b10a4a6 : SUCCESS in 1h 06m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/184734f2aa0d4bbeb8215f21496cb134 : SUCCESS in 1h 11m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b84f63ce6a5a44f08329d2c6b58cde30 : SUCCESS in 22m 13s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d85e444c41a6445784b4fc537050d12e : FAILURE in 11m 19s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27fc54b54ba94f60b3b72c152843f952 : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/885ae1ba0cda489bbb9bb8cdb5cea78e : SUCCESS in 1h 54m 33s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac9b9961ea1345c0be9ac8eb6b11d84a : SUCCESS in 1h 43m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6349d898817541c083c8db99097069e5 : SUCCESS in 22m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c13e1113df304787ba652663524963a7 : SUCCESS in 20m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e9eb14e36a2544caacbb2970ef5deb84 : SUCCESS in 37m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/145c2d54d0d143f4a1e8a7301fb46138 : SUCCESS in 19m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fb6e9e08ec3c41608a8a9cb7b28aba19 : SUCCESS in 19m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/216fd6cd52c447828469b39ff63812b0 : SUCCESS in 11m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e25344525bdb4ff18034813888c867d0 : SUCCESS in 19m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/663f7b0902a548ac9a76c0501dcfc8bb : SUCCESS in 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"058d368ceb7f6f689d701245ec3629d94c12b46e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-23 17:43:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"350654732b1a3e29cef436d1dc2dd88c2386356e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 19:30:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/215ebdd32055462eaa57465874ac8f6b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c54affd7feed4b14b44f5cbb048bacae : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5efad84da041a396a1b41ddbd2939d : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c65b793c70c4d4d98d71ea3900c1788 : SUCCESS in 6m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3157f64742d947acbad3a33c0e5a7415 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59922417a6547d3967327e1122ceda0 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10a94025183042cb94c60cd5c9149a60 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1253204a7294e27ad83c75597667222 : SUCCESS in 5m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b09e88dfa604b13aec89b73b8d482de : SUCCESS in 5m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb6b1cc7fde14f7496cb7952af4dd366 : SUCCESS in 9m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c633dcd5870d48ee9e57eb64351f6f3f : SUCCESS in 35m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/00da1c3343db41d6a711c5543fc5a257 : FAILURE in 10m 11s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bfb59b61c0c469db896dbb0340959f7 : SUCCESS in 1h 01m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8b2c4bbf54124239acb3cda9810bee92 : SUCCESS in 1h 13m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b757342b90264fffb61855280715d987 : SUCCESS in 20m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e5d14371d0f64a2f86e29b059d0275fc : FAILURE in 9m 54s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3a101cd5e604e15bf376e4190b75f57 : SUCCESS in 9m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4780f6fbdae642d8876675ba486fa350 : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b44b00a7939f4297b9b831f2df3425f9 : SUCCESS in 1h 39m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1225f67e8cc34096afd1aa7da185f26f : SUCCESS in 24m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e5cb00ed3dd408485d8f6be297ee98c : SUCCESS in 12m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a31d01bc910f4006a94230d6838f5ad7 : SUCCESS in 23m 33s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04f38ad9b4fc43a095d4677aa660aec9 : SUCCESS in 31m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d04e0b1e398b4b898742bb6241b93c0d : SUCCESS in 18m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a7522ab11304cd693c72676756c4794 : SUCCESS in 15m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/895bdfaeb2274e2f93de33a0ada99e43 : SUCCESS in 19m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b960c47811d54b5d905633e911e4639b : SUCCESS in 16m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ac1bfdaed408468ff4dfea083e0bc6f98d0f3f2f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-25 12:20:19.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c3fdf31f50b069ba510a40fd4b0aacb75c5eeeb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-25 12:41:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f598ed538216a32ffe56b116a4e3c3ab1b294dde","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-26 14:12:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a25585f963ca99c57349a9e5ef7ad9605a3cf92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 14:13:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"214d664999e28874cf4675d145cb80feedce7228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 16:19:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58daa0cf17dd4adc9da7b8193b872a6c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d07551c23ea4920b38f55c62f192f86 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c641d666f0c42a9ac1c4144e4747462 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc890cf55a84f96829531813abaedee : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3825ae0df58a4d7b891a349b5d46bcbd : SUCCESS in 4m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc1feaec697e410fa20277c41e813ff3 : SUCCESS in 3m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d67c8c8ef8f4df389b896eb840fdca6 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/64463eb2932048718f89e913cd1dc45d : SUCCESS in 10m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7ca50c097c7c4d1284b09445e4fdaa83 : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44c56927872843599265565d5d27ef7b : SUCCESS in 1h 01m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/de454cee03104316a9b17bb2d47244f1 : SUCCESS in 1h 39m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f21990d156e44d70bf6084fc31298676 : SUCCESS in 21m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca41ea7159d3408ca7006ab6c273ec9f : SUCCESS in 17m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/28dbe99312b140cdb1c2f11211c5bc73 : SUCCESS in 2h 01m 21s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a9c224902d24d729566aafb53966281 : SUCCESS in 1h 39m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/83479353f1784d8a9554a62b1a02411e : SUCCESS in 22m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d1ddae18383487b98d6c8a9a882e5cd : SUCCESS in 20m 34s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/03b5e539950a42a4af5888bf3346340f : SUCCESS in 36m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66fb195a614b4955a989209072577cc7 : SUCCESS in 36m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ce74c88706ba430fb6a3e55933c5edae : SUCCESS in 19m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b04c37b926754d3ea9ecf114bde6460a : SUCCESS in 16m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3e793f2c89e14c8aae2f7f66a9506c30 : SUCCESS in 18m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/795d13b4f5d44c9fb3aabc909fd1b433 : SUCCESS in 17m 10s","accounts_in_message":[],"_revision_number":1},{"id":"05656c6a3a35ff7013dcdd52b69867f7457d9def","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 16:19:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"39f4d661bbc241fd0ebae82b348d6fa233a9c602","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 16:20:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1248b72bb81c4f33951b0dd7ba1d80ef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938a48f9625a4d26a464c0b80a107eef : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6beaadb1aefb4d3db1423fe3518b4951 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa7a01cd3715696f456efbf86dfb7d9126bc8464","revisions":{"fa7a01cd3715696f456efbf86dfb7d9126bc8464":{"kind":"REWORK","_number":1,"created":"2026-03-02 14:29:34.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/78/978478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/978478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/978478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/978478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/978478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/978478/1"}}},"commit":{"parents":[{"commit":"d6b2b391cf423b9cb8306340cf2612f1abab6bf2","subject":"Split Python installation from user bootstrap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6b2b391cf423b9cb8306340cf2612f1abab6bf2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-17 20:50:32.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-03-02 14:29:25.000000000","tz":180},"subject":"Support Python installation through Apt proxy","message":"Support Python installation through Apt proxy\n\nChange-Id: Ib2f9c10433c42f783e1d5f618291ba5cb7be1641\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 8b36acddab1fc16c35ef86852c6a816da3617546)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa7a01cd3715696f456efbf86dfb7d9126bc8464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa7a01cd3715696f456efbf86dfb7d9126bc8464"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
