)]}'
{"id":"openstack%2Fkayobe~978735","triplet_id":"openstack%2Fkayobe~master~I0ccbfd7baa2ae43f06a91f34d8357d91e57317ee","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-03-21 00:43:13.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-20 21:32:37.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-03-21 00:43:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0ccbfd7baa2ae43f06a91f34d8357d91e57317ee","subject":"Drop backwards compatibility for template trusting","status":"MERGED","created":"2026-03-03 11:16:00.000000000","updated":"2026-03-21 00:44:04.000000000","submitted":"2026-03-21 00:43:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978735","meta_rev_id":"a19fa6ec32d958f3cb4c66161b2405d3d7058c0e","_number":978735,"virtual_id_number":978735,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-21 00:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-03-20 19:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-03-20 21:32:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-20 21:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-03-20 19:30:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-20 21:25:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-03-20 19:30:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 11:16:12.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-03-03 12:24:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-03 12:25:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-04 15:59:04.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-03-20 19:30:07.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-03-20 21:25:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"2819d03814b204cf7138f95eea52b821d8932749","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-03 11:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e9ae54ffb1dd914608d42e7d20c8bd572cc9c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 12:25:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35825099dfea4d709eaeff67ec4d7d75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a2ffeb64d64f9a9fb3388dcda0f6f2 : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d63674f01ef94600b6901270376d1438 : SUCCESS in 4m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/802bebe80d1b41e3a945624e28f6f2d3 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a038cae7624fe994498f374526bf06 : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c1101fd1acc43f3b9ba991da75829a5 : SUCCESS in 3m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/01fcd03de8b64e4093cd6c2009ae886d : SUCCESS in 7m 21s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b2992d9114074e17b5bfccd7d21f0631 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81e55583a4094f3795573e33ce63ccfb : SUCCESS in 14m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6dc4f0bbe0bf45c6a28ca683b6df8f4e : SUCCESS in 11m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/282a9de1edbd4a19860147bdbf3bd287 : SUCCESS in 14m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d71c9311913c4f65857bc582ce3b8e3c : SUCCESS in 13m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/28724f4586044b37adc596514cda8fc8 : SUCCESS in 17m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38c690f341254979834e91561a16b118 : SUCCESS in 15m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9b48922759664f24817e16e1bee97c43 : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/641b6d88d8884b45aa18cd19479a439e : SUCCESS in 59m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6411c6000e804c8bb4e439b25e63395e : SUCCESS in 1h 08m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f9ec4e853274739ac78dc67db00f9f0 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4d7ab65e029a40748e007270fb197596 : SUCCESS in 58m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/60a0fb4379784abebbb9cfafe77cc2c7 : FAILURE in 55m 08s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f7e5f27a0846437db24f3ccd5add2cca : FAILURE in 1h 02m 07s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9533766575794a6ab1f8bcfb0dc755c7 : FAILURE in 51m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09b07a77cd964472b6bb4b933c40f6c0 : FAILURE in 56m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aacf0cb3fd24414592fb5c746a97bd19 : SUCCESS in 15m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad2406c07cb549d6ae728d69285e432f : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3b6ccf70c3067234be42c9b16ce4981a905c1370","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-04 15:59:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a71d2379d02b88cac60d57eab5dacc6864281da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 17:49:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be9ea8187dd4f27ad80b465f9ced3f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b55598bff95d4a47b3997d7e06680eff : SUCCESS in 5m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0b56293ed8f64234b17689a0471f5a14 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d840969f594644b05630c4c88726a6 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ae97248fb84c8fa5e95f0fabd220ed : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2dab9279b414124980a6faa10edff27 : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8a422c0f05514649a906abd6acde5796 : SUCCESS in 5m 01s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ffb471ab629c47f3a6495c2dca786670 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b22eeee7738485da59e75cb81e3d89b : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c54d562594f34c99aec13b3ef7d535fb : SUCCESS in 12m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f9a52a61cdf94c1d9ba66e1666241b8b : SUCCESS in 11m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0227415d0bd42cd86b4c203b2a16b1f : SUCCESS in 13m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/43480ac38bb94adc9a25ee4ea36211ee : SUCCESS in 21m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d40bc4a4df7949dda8f9fb6b43f7a1e8 : SUCCESS in 17m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1460abd4830e4673a216aa410be7b31c : SUCCESS in 1h 07m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/89a57dc436d040ef99e7dc21dc26b2f6 : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2560ab6165784917881378bef8397a14 : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad274a12f5d94bed9b98a65ec0e6b21b : SUCCESS in 1h 03m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4172242ef8bb461ca7852f1ad2af803d : SUCCESS in 57m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7192a94a24c542cbad1c960fb34476e7 : SUCCESS in 1h 46m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2640dcbb294249f498156ffad321f9a6 : SUCCESS in 1h 40m 05s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45a480559862457e9f3d46f01e9b6f9e : SUCCESS in 1h 32m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb9ca7bb76b34225af56c96f40f14d33 : SUCCESS in 1h 38m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3e72cc2ae0394720a0934f3d16cfdec0 : SUCCESS in 19m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/417d8d3fceb34bf593330e47ab8890b5 : SUCCESS in 19m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9a37dd6d0cf49d5819bbc116a68b920 : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/65558a38b42c4543a69cd2996ec66a97 : SUCCESS in 17m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/964cd5313ca64579b7cd8b990baa0ba5 : SUCCESS in 29m 25s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a29fea26750b46b2b784e5e358c69969 : SUCCESS in 31m 00s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c541f8db67e4601b3b27246a7005dd6 : SUCCESS in 19m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00f0aa59ebd243dbb2790d2edd6ef5de : SUCCESS in 27m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4e712c4119594bd4b514337912a0c398 : SUCCESS in 16m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59679870000a4f61a92e2e4d83b75cf0 : SUCCESS in 15m 32s","accounts_in_message":[],"_revision_number":1},{"id":"52c386131713cd0e2ce930cb8362166c0e8b3c40","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 19:30:07.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"cc58d80db98a73adc1286f57d9f8990adc313656","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 19:30:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"03aa56b3b428b52c70d20d9f3da44c71243a706c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 19:30:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b179583397a8d222d449bf60639b0bb52cafa37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 21:11:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/316a6cb1cf964ad7ae77079aa57835d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f12494d4799e42a48f5b38da8e7d667a : SUCCESS in 5m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6f42e2f748b4ffbb408c9cf112b4958 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5682249619e04fb7b5818911ded01648 : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f21a41c04fe4921bc8121554eac0a57 : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8e036b8e4405449997bbd55e773098f3 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d8c2daeb562c46a6a9d0182994871ee3 : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d209dde58104bab85bd3090047de93d : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d5299c07b56d4007a9e982cbc49b189b : SUCCESS in 17m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5130769a88914277844d2f551c0c3e5f : SUCCESS in 11m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/da119d9191b94956aff8e02681f1b683 : SUCCESS in 21m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2f7a0758a824446b6503399ac289d0b : SUCCESS in 13m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d152b81bd416420cb40e3690f7ea6812 : SUCCESS in 1h 03m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3aaa0ce7ae294c449f4d6f5ffdea479d : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7260a026acf94c00abd72e7d6eac16ce : SUCCESS in 1h 05m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffb74e5aa1754596bc17a0415e1cf546 : SUCCESS in 57m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3609d7f1ea4344d79d6dcf11649d98b0 : SUCCESS in 57m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c21635370bb94481b0867776125f06ec : FAILURE in 10m 15s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/80d525fff16a4a8b88c15c185f93b3d5 : SUCCESS in 1h 31m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbdda18162604de285c4eab19f451396 : SUCCESS in 1h 33m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/245e061baffd482d970bb112d4922a98 : SUCCESS in 1h 39m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0b7fd319d4ff4edeb59ee3769e9b53e8 : SUCCESS in 17m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c3c07675778a4f83bc3caabc4092bbef : SUCCESS in 19m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/104b988f949a46c0b1bf2f68afbb06c4 : SUCCESS in 17m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0db3624249c649d7a4a42ce1fe2945a1 : SUCCESS in 16m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/256ded442a704e2e896f2d57002b74c3 : SUCCESS in 19m 05s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6bfe8248ed4f42abab3edbc51feaeab0 : SUCCESS in 38m 45s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b71758fba9a04c4d8101a4ae5d888a6a : SUCCESS in 32m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5acf6a1129344242b9b2db9bcdb1e21e : SUCCESS in 30m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0f39a4ef36654d44991bbfcca6a2d82a : SUCCESS in 13m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25566efff7f44e31b6cbf309ec8b5fe3 : SUCCESS in 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"34df4e9d298a76e4853cc8927a344cee9bd23262","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-20 21:25:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf321975d9f4431d8bd4f020dee9e16ec1400a34","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-20 21:32:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8848ffd65e6b41b53f250a65260b66b378182217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:06:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58aeeee49e6b4d118bbef8e2ddadc230\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e050a62a8a74c8a8b32f8565333f80c : SUCCESS in 4m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/df80a388722f40209b7fa8e529735964 : SUCCESS in 7m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f033d3d1d47e4363be93614a5b6b3a9f : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a69e9447b7426dba4244e1ecc4678b : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ce0a37d52a54356a68dcd6044abfac3 : SUCCESS in 2m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/79af95922e8b41c1b80d5083b5687cb5 : SUCCESS in 8m 12s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd5e2f5a52534df9a0097898ad79c4f2 : SUCCESS in 3m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9f7311c4fb7485c8bd8e31481636000 : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/215a74ca5bf44090bac0a8fa14d2639c : SUCCESS in 12m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/155adbe1abb8478385fa3556d4f5096f : SUCCESS in 9m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9633f42adf94441ba92994b273db0349 : SUCCESS in 15m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4c1eb8ec14884d2e8ee1cb0e649fba4f : SUCCESS in 12m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/326d3433cf7b4d509b863be42a216170 : SUCCESS in 9m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d5ade5c196464519b5c21d3f17de0ad9 : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e96822a09e204abb934c66f1a14b2825 : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/01c597cc798c43449ce95fbff6128837 : SUCCESS in 1h 03m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/336a3d9280504894b93a3ce2ba0d17e0 : SUCCESS in 1h 01m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/06a61895e2f54bf4ac703082e4d3a92f : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1e6a0dc84ad349b281a1f10d67ad2080 : SUCCESS in 1h 39m 11s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c52a6fb9ad044b7e978e8b45558b88cc : SUCCESS in 1h 29m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d49db47456d439f9b99745eb3b89970 : SUCCESS in 1h 29m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dd3393524b5413fb7591b839e00db0a : SUCCESS in 1h 36m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c442931f648f4f44a8d85da654c324bb : SUCCESS in 10m 03s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5df0b744dfd24271974ec9c960eb7567 : SUCCESS in 22m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f7a1a82bfa44e83a33156870adb597c : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e505bbb65534afabea977a343b992ac : SUCCESS in 16m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/636bdf60a5a34c41944fe617ceab0d20 : SUCCESS in 15m 52s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f33b0f32cc5f403c92d7d0f2be18de17 : SUCCESS in 17m 45s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/153473cd9a6941f5a1fcd34472e93ae8 : SUCCESS in 16m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a75ece86c90d48efad0e038b8d6a4404 : SUCCESS in 29m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0c03b1a7019c4806bb695eec6ed7aa54 : SUCCESS in 10m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/219d6df616fd44c5aefc712df13ef5dd : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"cddafb5d1e259032858210bec27ee6f2f56c8478","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:06:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbd978cc231c46c4300a101e39fd2952a104b508","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 00:43:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e59fb76fd0c3453bb3d7b2ab67bfea69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98ad5071d1a843e59a535c11bf320d5f : SUCCESS in 4m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5cb29c1127148978eca4b09911e9a0e : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/333b76f418e24d7e975edb631149ee3b : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71bfde71ad9744509284d3624d162957 : SUCCESS in 2m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d064501ac6c64f02a7d2ede757c8007a : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/568e7fadd117446e80bec485f75c84ad : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aec3bf36bfb8445aa8a14b78b975af3b : SUCCESS in 6m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/12369be2c26c49a39c0eb8308327870e : SUCCESS in 9m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccc1b850d494442e885d4a27c75548b1 : SUCCESS in 15m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/411aa8f2dbdd49e7a7b98ad03eb720b7 : SUCCESS in 12m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86bfff2d9ee744b0b995140e07cab0d7 : SUCCESS in 9m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5f675931fc734fb4a0a4f65cc88cb4ce : SUCCESS in 57m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3a586b4d142448b990d5ccc061b31aed : SUCCESS in 57m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/261db1385c174737bb20f1414556af7f : SUCCESS in 1h 00m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/152e9c4d9cc446249c1c8f74228ed6a9 : SUCCESS in 57m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2e43ee8690a44ef8b18e1789e6c71969 : SUCCESS in 58m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/30f5caae6fcd4b579ae8127ee937dbde : SUCCESS in 1h 34m 27s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fd2f842a5d9d4058844e325bcd926e09 : SUCCESS in 1h 30m 30s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f44034a3b0ee4cbc8346dd48e3a85394 : SUCCESS in 1h 29m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/408b6af990e24bbea20573f62ac06f89 : SUCCESS in 1h 35m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b928e66a30fd4ae889daaa269547c7af : SUCCESS in 15m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/85aec364fee24aad86508380b258b8d5 : SUCCESS in 16m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a9e63120f7843f287bfdbdc8327a876 : SUCCESS in 18m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e1d1c3c16b04601bf770fc155852f7f : SUCCESS in 17m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c3c455d69fa54607a713662427f1ccf5 : SUCCESS in 15m 41s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4f32cb88b079472e92e9e859c9701601 : SUCCESS in 17m 21s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71ee181ad858416799a86a5cdc32429e : SUCCESS in 35m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/862fc24bc76a4fc99c6730e77e02b617 : SUCCESS in 22m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5ab10208560649fe8f32211ba89b1f83 : SUCCESS in 10m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48f3c36e9eca4386ab952fbc6709bf39 : SUCCESS in 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9e57aead0461d7ceadb31f8305365d0dcb542ad9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 00:43:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a19fa6ec32d958f3cb4c66161b2405d3d7058c0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 00:44:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c8dbdf0e408431686b6b8fb2b892071\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1368da6b565b45f0a81df9cb8389eccf : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"505da7a072fc7688fa7afd9a0ec68d449e1366c1","revisions":{"505da7a072fc7688fa7afd9a0ec68d449e1366c1":{"kind":"REWORK","_number":1,"created":"2026-03-03 11:16:00.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/35/978735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/978735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/978735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/978735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/978735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/978735/1"}}},"commit":{"parents":[{"commit":"aa230f9d05eea695f2aaa09a348f56a2079e6eda","subject":"CI: Disable seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa230f9d05eea695f2aaa09a348f56a2079e6eda"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-03 11:12:25.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-03 11:13:34.000000000","tz":0},"subject":"Drop backwards compatibility for template trusting","message":"Drop backwards compatibility for template trusting\n\nWith the G cycle, this feature is always supported.\n\nChange-Id: I0ccbfd7baa2ae43f06a91f34d8357d91e57317ee\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/505da7a072fc7688fa7afd9a0ec68d449e1366c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/505da7a072fc7688fa7afd9a0ec68d449e1366c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
