)]}'
{"id":"openstack%2Fkayobe~979998","triplet_id":"openstack%2Fkayobe~master~I47035e2d1625fe7e95f29919d59676ea0cc9ee67","project":"openstack/kayobe","branch":"master","topic":"no-log-sensitive-inspector-rules","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-01 13:11:36.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"34212":{"account":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"last_update":"2026-04-09 12:12:23.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-09 12:12:23.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-01 15:45:47.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-09 12:12:23.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-04-09 08:34:42.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I47035e2d1625fe7e95f29919d59676ea0cc9ee67","subject":"Hide sensitive inspector rules from ansible output","status":"MERGED","created":"2026-03-11 10:31:12.000000000","updated":"2026-04-09 12:13:25.000000000","submitted":"2026-04-09 12:12:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979998-no-log-sensitive-inspector-rules","meta_rev_id":"ed71744aea619b41df9abb81826ca205f7bf9589","_number":979998,"virtual_id_number":979998,"owner":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-09 12:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-01 13:11:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-01 15:45:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-01 08:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-01 13:11:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-01 13:11:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-11 10:35:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2026-03-11 12:30:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-19 09:29:36.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-03-20 12:37:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-30 08:37:25.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-03-30 16:17:58.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"b9d6221f7538b31b51a869493b3f24581d6167da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2026-03-11 10:31:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"578ee32ee3ccdd851ad943c7b3fce0c4f49b4382","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2026-03-11 10:35:48.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7dbd5023b07ddc465f981fde23b4a50f046138fd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-11 10:35:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"324a08f61f90fafde30782161c996d71b1db491d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 12:30:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a54e56eca0ee4278ba302ed43b88c073\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ebffc914dd4b7994872d6378b32197 : SUCCESS in 4m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8749816b02474d65a1346bde778eb732 : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/224c6f911fdc4c0b80c1a6bb13241fc7 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef1d7da442964eb4a3af091b6d55a3e9 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e770a3d549a492c86e5a6f5cea97d10 : SUCCESS in 2m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/16c5a16066e94267a4acfe771efb0737 : SUCCESS in 8m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0e963b13a4f415f98f4122f0bb23f25 : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/533c6d4d9f434119a08b70c92b47f7d4 : SUCCESS in 3m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9b52b7d3ea0242c5a68b4231afa8545b : SUCCESS in 15m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c5f6d643d64147ec99271437400c6f80 : SUCCESS in 11m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8c181804d2474b41b06abb846433eaf0 : SUCCESS in 9m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54fbfd9e1a9548d9b1a2e068ceee77b3 : SUCCESS in 11m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5c6d92682fd44a829c5dc7edc3136175 : SUCCESS in 17m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b3ac3d8b312455b8b5b3f6bf9438bf8 : SUCCESS in 14m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b9bd4af1c67f4fab8a8df6c411137653 : SUCCESS in 55m 31s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/599ab6130660454083b22b1fe79b9de0 : SUCCESS in 1h 03m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/40402d97394e4dddb13c0058917ba8d2 : SUCCESS in 1h 05m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f64d3458eea4a4fba8116840747bcff : SUCCESS in 58m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dd2b994fbd524eb28ae539c3a6307861 : SUCCESS in 1h 04m 57s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4e8318e867c7418c811624355b6f6942 : SUCCESS in 1h 47m 45s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a4870fa3465c4e0c9de8f76543640859 : SUCCESS in 1h 43m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf08d90679654d369abad7abc956f207 : SUCCESS in 1h 28m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35926967a9a1487e93817fdb57b6b5e3 : SUCCESS in 1h 34m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e6300e824b7e40138937382686241008 : SUCCESS in 21m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5c704379f7e34c09990819500b93ca05 : SUCCESS in 10m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc5f57785456487eb5e7dd11b06f5bce : SUCCESS in 15m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/88d30ba6235b4618a4c26f9bd53e16b4 : SUCCESS in 14m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c33ec94b57aa4b46a88f6b3b2a29ab40 : SUCCESS in 30m 27s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a4efb6119ffa4717b6210125b393d1af : SUCCESS in 39m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcdcc012514b477cbf25f3f0100e9395 : SUCCESS in 17m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14fa8ea8cd1841e7ad9e5da63e50428b : SUCCESS in 34m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/11f310cfede24f53a239d7a344d7b636 : SUCCESS in 9m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1ea81627d4b461794c2c4f64efcd660 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":2},{"id":"3742adbf143c795e72d449c3cc470af2a1824c2e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-19 09:29:36.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49d181a2efc421c2777c08f3d68907da31a7104b","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2026-03-19 09:36:54.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"12ea69a2a901a6e09246fd6390268f9d05e69d44","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-19 09:41:57.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d09f297a65cd152132ec94187e58baf228ab7f27","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-19 09:48:58.000000000","message":"Patch Set 2: -Review-Priority\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d4234155c283b75d3e38c150030c191b127cdc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2026-03-19 09:56:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0d27281063abe702e0deec28f779d1c051a109b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 11:46:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7021be925013420a8aa594b8059bd91c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06275fe5a154741b4a6d526ed38dd38 : SUCCESS in 8m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bd7e782b7a534b1fbae53879635d4f74 : SUCCESS in 5m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c92ff559609439da6595b040e1334d1 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f0c06b06a549448df9cb3c8d107f90 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75b4408fb05d4dc8b7e751ca79864046 : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d48735dd2d99486e8af75ef3fb346670 : SUCCESS in 7m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a512d35cc69429aa84e63a932b7e5b8 : SUCCESS in 2m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/20a336828c474cbbb162f20bc598cdd0 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9dabe65940847e987b5fe441ef84e63 : SUCCESS in 13m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/021ead97dcfa4d3f8ed896c18245944e : SUCCESS in 10m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/08b5742d91bd4bbb9fcc1fb264fec29e : SUCCESS in 9m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec4ce401bd4b474daa2fc18c578b37b6 : SUCCESS in 13m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2fedf09ad973479891b8f0effe4efa6a : SUCCESS in 17m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d288afac9f0415eadd577482818d4ec : SUCCESS in 10m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/42ad5fbe12534de78fc650f1c337ff9d : SUCCESS in 56m 55s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0e0368a3ddf24f168aa4b09ca260d016 : SUCCESS in 58m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3dc7744a1d25418b8487e2a27241cd33 : SUCCESS in 57m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fce40125737e485da3a83f547f96c5dc : SUCCESS in 1h 03m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/82f4be46be754052aea971b47d730e6d : SUCCESS in 55m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4026530daba34fab8c9eb16e5a6a227a : SUCCESS in 1h 45m 01s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3239580933ec4b1a8ecf552f5a4c2e32 : SUCCESS in 1h 34m 42s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e9f3957fcc74147acb232f0030a5de1 : SUCCESS in 1h 37m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acb2a2b4e5d142d0b0001f98d1985705 : SUCCESS in 1h 42m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3d870cf4b8884d1aa54271f1a0814e4a : SUCCESS in 11m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/515d019d530545d99ab3b92be0b7b47e : SUCCESS in 22m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e304707c88714fe8a775220002c08189 : SUCCESS in 9m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/68747956eb8542a493a430143ce9c232 : SUCCESS in 17m 55s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d0c7cca425b2468eb2b4ed42e196246e : SUCCESS in 24m 40s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/13b700d8ddf7485bba4c26afdb16fd93 : SUCCESS in 17m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b604e1c67b347408fdd27708da86e46 : SUCCESS in 17m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0935a5bbf1a24073a4f1dd29a113082c : SUCCESS in 22m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a9938f3bf97843b499918f51a54491a7 : SUCCESS in 9m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a84da35d0eb47cfb72e66c011c4083a : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":3},{"id":"04058affbe8412c1abc2b78c80a3b8e03faf0d3d","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2026-03-19 12:03:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"67de211d483175f2b8ee537032c6754ee80828aa","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2026-03-19 12:03:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2c4dc4d5b674c354e747ea9b4a68d9f41fcff9e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 09:38:57.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"604666581035275ab90648d455ef8c0367363e36","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 09:41:14.000000000","message":"Patch Set 3: -Code-Review -Review-Priority\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"453f28cbdd6949ea2ec5258395cd16b69be6aee8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2026-03-20 09:44:47.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"aae12d25b15d7a7aee740d4f4c22fbac305d5a48","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2026-03-20 10:47:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f86a773fcb24fefdccf74ff76b469052fadb18e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 11:50:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9517e048e6445d997ee75392880fb84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdba6e80d8e448e49f63175b7bb3592f : SUCCESS in 8m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4822bdc2851a43d69303596cc7edf795 : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f072a00e1efe4cc5833f912655b3dcc7 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92bd510060b943c7a58beb56daa3b6d9 : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cc65c42313144fba7e22197f9b74510 : SUCCESS in 2m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/017bb9d286b6417f9453d6dc48f27548 : SUCCESS in 4m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de0f34ac1a294c7cb1e1b7b32d86e683 : SUCCESS in 1m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e015d37393fd4499a262de109b8d2bc0 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/639d3449163248a8bef28c8ec31fd136 : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa36016549594ddfa6fbcb6ab516303a : SUCCESS in 10m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1c0e42f49e5549c09eb17e0844aa8db8 : SUCCESS in 9m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de05a507ea01403680de6e8e173f8ff5 : SUCCESS in 6m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b23abee94f7d4f0d9f18048dfe972694 : SUCCESS in 12m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd62238127644cedb1ded8c0c29a34a3 : SUCCESS in 8m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8c0698bc74f24c5398624755e55e91c0 : SUCCESS in 56m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/52c0fb65091d4bac9034c6aa087a25fa : SUCCESS in 57m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/845ceb5814e8497c926f14d0273f9a52 : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fa6136a0bdd41f8a638bb575f914534 : SUCCESS in 1h 14m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b7ce7a8ab6274896a9288aac04ed65d0 : SUCCESS in 58m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/485977c0eec64b1793bbf2c142e31125 : SUCCESS in 1h 34m 59s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/54a12cc0c1794064ac30991a8c20fe82 : SUCCESS in 2h 01m 38s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e67205d48bc949079c0d3aea8f348086 : SUCCESS in 1h 28m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75bd98735f6e4afc955249acb672d517 : SUCCESS in 2h 04m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6f5933411eea4a86a576612dee535ab0 : SUCCESS in 18m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/747b2887044c4adcb0c15fb8d0646a14 : SUCCESS in 18m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1d56cde5b4745248adb33a2d9541c1f : SUCCESS in 16m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0214e576302d426e8a02cc057881404f : SUCCESS in 16m 05s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a549f895aac140458c3355927ad7265a : SUCCESS in 15m 56s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3690a713dca9425db9de129f29daad79 : SUCCESS in 36m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d2beaa923e14c5bbb63d82697c78e1b : SUCCESS in 30m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f9497e5250043309d4c019e9c9da467 : SUCCESS in 28m 34s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e0854a12a5c040d8a098834244046312 : SUCCESS in 9m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e01bc620743143c2ad96fb87eb68bd1a : SUCCESS in 15m 10s","accounts_in_message":[],"_revision_number":4},{"id":"c21a86945fbd0b8d7bbc6a0b0c3a2547f6a4960a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-20 12:20:47.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"bd0e0ed5d048af069c312933f582a6c06f6fb407","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-20 12:37:39.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"698cf5a65c5a1051c6bee57c1744566a3019016e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-30 16:17:58.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"537a173fe4f7a3f363bb7d03ec9131dff6cd0f33","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-31 10:22:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ed88154e2c1b43cce7a5ee053df229a77078161f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-01 08:13:40.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"08e4078e33bb97fa5900b1189deb491db8aa17b0","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-01 08:14:20.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3b28566e3478ad7bdbabc871e8e5e438f182693c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 10:21:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0d63f2da69c48adac33dec4929a2096\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc8bc8b7de34c15b07d5f9e42282f90 : SUCCESS in 7m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/656ee713627e4173bc53113ad7c9da3c : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ec7b3aabccd4e978e98a9730b512050 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f9393d9813c454d87baa80e32c45aba : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/587070c6092a4705850c222ac7a2c59c : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d4d85f86813b4bbb909b2d1908c0745a : SUCCESS in 6m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35dff720c2c942adb9755a63e5909763 : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea104eb002cc41b1ae6185b2a73293b4 : SUCCESS in 3m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca51316de50f4e7389e98dafd773177e : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/294f32d12c2045cdbc6af217e200636c : SUCCESS in 6m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3e47e3bdc5a14f5baab03699288ce4bb : SUCCESS in 14m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddd9036697e1416cb50626a2b4ed9d7a : SUCCESS in 7m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6cc7668cad8d4398b1a7c775f1b6320b : SUCCESS in 10m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/513cf4e1105a41939668961e1ae8263e : SUCCESS in 16m 26s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8f4c7f9aa02449b786a614bd72f434aa : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3ddd1dc62a2842babff77b991786976d : SUCCESS in 1h 18m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/61108e453c5743d8ae2deaee23233624 : SUCCESS in 1h 24m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ac650aea1c14fe093ecf26248a55cdb : SUCCESS in 1h 22m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e9765498de654bfebdb77e7d18a0f7ae : SUCCESS in 1h 29m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e32bd430cf2e477fb402b7e1b47fb6a1 : SUCCESS in 1h 41m 27s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1812568d6af24a1c8f61d9b2fdfcc1c6 : SUCCESS in 1h 58m 25s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd5fdab3e66f4b98b80d5b6c3ca9c977 : SUCCESS in 1h 56m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d31da7b853f4f4a87bdcab17761a8cb : SUCCESS in 2h 06m 40s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c2ac8a2501224421aa17503227716921 : SUCCESS in 10m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cdcafa9fb1ff45a2ad5076de4d37aad9 : SUCCESS in 15m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ccab706c2504e75848379086a4cd94c : SUCCESS in 17m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/324cca9988064e53a945035e52219a44 : SUCCESS in 18m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a14432308d1645a68ece07f884446b5a : SUCCESS in 29m 32s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/eae27ba23f6b4463ac57bd2f1d754fac : SUCCESS in 26m 36s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bec5229a07584b2e9605fc8e5a73da07 : SUCCESS in 17m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07281016d05a4c1faeb6babd2fe43f4c : SUCCESS in 15m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/888d49fc3cd94cd98d2f35b4ec101dda : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4f032018e0d4b32ad2063e8aa37d380 : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":5},{"id":"da2b9b15e4a9845ee475c811798a3ca164c000f0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-01 13:11:36.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d52e5a426dd2f557aca08835567807aad98c2bc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:11:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"82e9d74b80713923425ce4d110814d5a024c4a15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 14:41:34.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7df5d538cea94e7ab3e101ee17d92321\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7355c9f8884e4a31bf6d32b4d7ba0809 : SUCCESS in 9m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/969a431442eb4180963150c6128a704a : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f51bfbefe85e4d73a88cbb9436f70035 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/944d0763cb2447d69ac9e110dbb1154b : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1de43401093c4b369f86f15e17344691 : SUCCESS in 1m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f4071ce36b15456d950478bcc51370f6 : SUCCESS in 7m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1c674f5f3d14b2f84fda816546afbd7 : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f9a1dd8be538495191d02a9860b34fb7 : SUCCESS in 12m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7bb8c5f4a55e469d9b4370b0d7c472fd : SUCCESS in 20m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2879d8082e1348828fe5c71b5246230e : SUCCESS in 13m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b19cbcbe61164687bda0d24fc8c4dc4e : SUCCESS in 10m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f94b7beccef4400a9885fdeed2ab73be : SUCCESS in 16m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f7c42ef8a24d4c31ae689a38065be0e6 : FAILURE in 50m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4f81b598728e470e88cdcdb21785c9cb : FAILURE in 46m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5aaeddd40e124f2ab71161b0dc875405 : FAILURE in 53m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/440af48cb35142f18e079d72ed0c13b1 : FAILURE in 45m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e819a02b979348688dc36ebeba8e50d9 : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c2a03fec744146d4bfec6507881e22c4 : FAILURE in 13m 05s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dfe0c63b405042a7af23cc9f6a20a03e : FAILURE in 49m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c492c5691d447d78d5aa8bedec82365 : FAILURE in 16m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1be38371bb634e3b9d21c0d85caa1f83 : FAILURE in 14m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dc3fa489dfa94108b5b862053e857c47 : FAILURE in 15m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/aa8e8d20718c426ea4ce7c28da854b50 : FAILURE in 13m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1451612f8e884b4288dc65af20180d5e : FAILURE in 7m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/72d00e9d4ab348e79ed6bc24cf134e70 : FAILURE in 14m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/769e2f3ba42a4dab9fdf378bc1e841fd : FAILURE in 12m 09s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9ed7253f338e4be485c5c29284cb811f : FAILURE in 19m 06s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1120f678b11b45d285a1a9004b59ed7b : FAILURE in 14m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0f01d8bd2294bc0b00c6666514e765c : FAILURE in 25m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fa26f54a21d640b2a6a078fbf0ba4acc : SUCCESS in 16m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d488bc0c1acf48379ef7a891dd9dd0d5 : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":5},{"id":"4a50f123b7a44b540b187b4c2f08694d4a0fff71","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-01 14:44:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"054c8fec7e68750fb421163a94afbbb41e5b762e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-01 15:45:47.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0c394d2038551fc062d29eead6045a25e12929ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 16:27:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e097f39d4864425c956aa5b5457db607\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d258dc7fb00a4da0a5485bbaf496e52e : SUCCESS in 8m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bc2e24ea36894a58927cd3993bbf6331 : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d8b88d1dbf547b081ebdb7e3c59d767 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e53022d9bbc48ccb9872781ac934de2 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4974b4da2db4de7b1220e8acbf1e136 : SUCCESS in 5m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3d4cdabb09524421b05db4ea65654c2d : SUCCESS in 6m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e70e1c1dfea474ab06dd9e695a9ce85 : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/91303d557ecb496390adf26dce93479c : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aeb57848a4ba4cca88212bdc591c388f : SUCCESS in 13m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/98dcc692712e4df0a7bdd75d82824de2 : SUCCESS in 10m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b48e758b808f436d8bdf40f59ef2a3fa : SUCCESS in 14m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a10b380ff9774bb6aa04f04a0f69f030 : SUCCESS in 15m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/93e33e10d1d84bda98edadce5b650ce7 : SUCCESS in 16m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0b8d75f00894cc9b78cde6a4f526165 : SUCCESS in 18m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/29f71bef4d5c4e5fb46d127e79dfb8c1 : SUCCESS in 1h 00m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3ce39395d27f49f29f1b8e5697d3b465 : SUCCESS in 58m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f91901b0622248e69d5db94cc01ce1ad : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2db3128ed44409c9d18270ab9436090 : FAILURE in 39m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1980396b406b427aa45b5f10f4ddcb7b : FAILURE in 36m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/42d16a7a558c42028892f651d300fd1a : SUCCESS in 1h 41m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5d3a40bd401744f091446d92d8bead64 : SUCCESS in 1h 39m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49d53fad7db84c268385339693690b21 : FAILURE in 35m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/266d3fdeab3e47b3803d8e1a9a55449e : FAILURE in 15m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3d1de588f4e542f48e8a202c0927cda2 : FAILURE in 10m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/54381a50c7e44916a64971006161d3cf : FAILURE in 17m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af803a6eb8fe4f80a792454e4e365cc4 : FAILURE in 18m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16446ca8331f442989524580ffa9d431 : FAILURE in 14m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5dbbff99d6b949669bcbe1d34ded4094 : FAILURE in 12m 06s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/653d1563e4834edbba2f30d7c20a9354 : FAILURE in 15m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de498a66eea7497099017486a880db39 : FAILURE in 16m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e174f188e38470dbf38442adf41d985 : FAILURE in 14m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ec70e05b554145a5bf17d9bb1ff7531f : SUCCESS in 16m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67cc7d65e65744b29ad1ab74644d9d00 : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":5},{"id":"1341542528960de749a0b6eabfb89041761af332","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 08:29:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f59e8de8ed4b99cf5cb9231abebcfae0dc994003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:11:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851abfda729c4ae08f84dd7666d5c550\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33e9b1c23c0a400298c5d551cacd21ac : SUCCESS in 5m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e5369b733f7a4288a6bccca0ca966372 : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e655e6f54a2d447bb7eecfa01ef4a770 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff07c0ebd7fe4528ab17ab7fe8047e4b : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b5f80af063b4a7ebf4899609c8e4398 : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b97e6479e9c44940b0a3498019218743 : SUCCESS in 4m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92abc54f08e2494086543386d5eea831 : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ef7366b8aa143248e75c27a8eea7c90 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/39df34061f724056a6f7943d93d74679 : SUCCESS in 6m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d28906a69c14a70810ac902ba26fcf8 : SUCCESS in 13m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/82e3b90f58be4ea68be4956947f1d2d9 : SUCCESS in 33m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebf6d4358d064a6fa1980429cfe69a22 : SUCCESS in 11m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2b29f84505c747b7be1bd733c55e5f3f : SUCCESS in 10m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c81b22d1f384dbf992ab414bc6470c6 : SUCCESS in 15m 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/12d2663976e94eacb967e14a241e999b : SUCCESS in 58m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c49c7668eb204159a0726fcaffb2d81b : SUCCESS in 1h 02m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f6cd729ae55c47ad8192cee95cd4d713 : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1e80fada41546e082a5e4363a6adaed : SUCCESS in 1h 00m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e765dd65fa0245abaae581ed1b399ba4 : SUCCESS in 1h 02m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e14ea2ba11a04f989b2aec4d1d3582f2 : SUCCESS in 1h 40m 18s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/982b658d195b4ef188863f6d04dbeb8b : SUCCESS in 1h 35m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ece59017825f4b5aa5a936dfaa9cdc68 : SUCCESS in 1h 28m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f11e3a6fd6941029e519f99cbc50979 : SUCCESS in 1h 35m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/200c177df32449b98419393d3e88467a : SUCCESS in 11m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/23266568af914daaab121a2e351b9e4c : SUCCESS in 17m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b01397f50b534ff086577f3d81fbf8a8 : SUCCESS in 19m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c7b92abd2b34b46ae7bcd2c6087eaa3 : SUCCESS in 14m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/04fc5c5ac9c74d369d6e79c97f0c04e9 : SUCCESS in 27m 34s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4b9318d928004a98a0b5c65aef8b930a : SUCCESS in 18m 09s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee1fb9dd0a504a0aad7f17f6112427c8 : SUCCESS in 16m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5bcb141b709402d9e29796c75d7ed96 : SUCCESS in 15m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/01a578196d1a4e918a4a3ce6eaa855d3 : SUCCESS in 37m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6f4fc4ff4734907810851762c8a4ff5 : SUCCESS in 18m 01s","accounts_in_message":[],"_revision_number":5},{"id":"61480a939981bd044707a9c75386f125846b1d61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 10:11:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"67c67749495db8188f41954712bd04ec3f8dc743","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:14:38.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94e202f42e1d48bb86e030273589ca8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4fe9346e8341ae9f58fff063272f64 : SUCCESS in 10m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/edc28d42f2994bb2afd721e7d824cd28 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7029e1d550f0434482ae965cfbaf88ad : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6db79c4a39ae42c5ba9ddc8afa974ee9 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be5a7558328d415cba74d7b5c6344f05 : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b5ca9d87f98c43999cb0f3ebb88c85db : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a9f928616fc4f899a006aa5ac17d375 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77f8cbf6de35418e99f5af0f6acfaf0d : SUCCESS in 11m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5e828bf79d9249aa8b4c8658b081eeee : SUCCESS in 10m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc014b1d68c6486b8577ed651705d198 : SUCCESS in 14m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/107109cf41024e378eb7574c4bd281ae : SUCCESS in 16m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5deae15e04fc4112850d35fada650e7a : SUCCESS in 16m 30s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/49981e3e94de4f188de65323eb144d2f : SUCCESS in 57m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e58e75c788d54abfa917f856deb39ccd : FAILURE in 37m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/65bf202c8ede4202a5875723e23f3ec7 : FAILURE in 48m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ab56f7a10984f97a94cb72bfd09b82d : FAILURE in 37m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/06010df5e8f1425386d2f07e9fcb6cf8 : FAILURE in 17m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5d5686965f104e449f4d6f0fe7279161 : FAILURE in 36m 57s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0fb533c18b0e4f79bf71647b38c57c6c : FAILURE in 36m 05s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c64796b996d54658bf723452b3fe14fe : FAILURE in 32m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa904a51f30a4503908dda827749bbec : FAILURE in 37m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/15ddab84d5ba4691a77339f09241186d : SUCCESS in 23m 43s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/73469fb5ed0741718b090bda7fb60501 : SUCCESS in 24m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e66a0089ebe43e5849a6631f35d590c : SUCCESS in 18m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6ca8da5f4abb48e5a5d839f3b75bde9d : SUCCESS in 17m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/87a95a53d3004831bb7cd40df69ff89f : FAILURE in 29m 09s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/12aaa8b3cb42410fb0971456606cf5f3 : SUCCESS in 17m 22s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2be545f870f54488ae985f0edbb951b1 : FAILURE in 30m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfdae9457ff3430989360da8ec1be776 : FAILURE in 26m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/333b597298fe49939ef6199bd30777ff : SUCCESS in 21m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea04d9d5219646c98a4638344ca41c01 : SUCCESS in 15m 59s","accounts_in_message":[],"_revision_number":5},{"id":"112f4bb1b315741efff9e004aab68111a5103e52","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 11:25:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cecde31a90ad4aad02901c564e85776fa0cb016a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 12:15:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31a50eca402d4460a3019009af9b4fa1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc55b0860a8471fa1ae60e3b15e5a01 : SUCCESS in 8m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e900aba0c29d4fdc8d25a9cf6b38ff26 : FAILURE in 4m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71ba136a2d6a46af865f26084f59e3a3 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5734bf411edf4cbaac8bcff1a374e8d1 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c45cd30cd15f4fb1bc6785f5475554f1 : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd605b77075f4b62987ab3312033859d : SUCCESS in 5m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b727999441406b8e66ca6dfdeb02fa : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/77cb886fafce49e8879f607e8b391af7 : SUCCESS in 3m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/466c12613726408cb8ba4d278e1eb2c7 : SUCCESS in 12m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2d025bee807646a3bb1430c04f8f913c : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/230c30b3a68f495b9263acdf39b285fa : SUCCESS in 8m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9200d28290042618fb674c8747cd3f0 : SUCCESS in 7m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fc9d7d702cfd4279a8c633b4b5c7ed11 : SUCCESS in 10m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d48a594e6fe4433aa2e48744b8afe513 : SUCCESS in 16m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/26157f4be4264941be3935995ecae494 : FAILURE in 40m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5e73e725a2b444a998cc11ff8e7aa080 : FAILURE in 40m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/de35cda0bff449d6a39d1a5a50663b72 : FAILURE in 49m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ee4d2783fd54817988f4186815b13ae : FAILURE in 39m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fa95a353b9b6411091e50fe411d6e142 : FAILURE in 37m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8f7df00dbc434ea0a6129e321bb5cfbf : FAILURE in 21m 02s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0667584c411f4a788f1d3dbfab8ec500 : FAILURE in 15m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/852206e9596a4cadbf79648146bc905b : FAILURE in 14m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3161d91da564af1b6307f9a391060c8 : FAILURE in 14m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/82057f2a17654ed5ac5553283bf98209 : FAILURE in 18m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cbfcc6f425f54691b0cc62d1e6787e6f : FAILURE in 11m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcbbc125bfdd437c8fbc655be505d563 : FAILURE in 15m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7d81026ff6a4fd78f8fe94a8d8cb1d0 : FAILURE in 12m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4b339af0a47147ef8b6fc17f55096af0 : FAILURE in 13m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c8ef3e67b98c4a98882c54cf205060ec : FAILURE in 16m 28s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc6ee6b539e5435aa3e7dcfa54d8cdb0 : FAILURE in 13m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36d6f32d8fc34a3aaee400050a7ef645 : FAILURE in 17m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d1a6bf1e2f9448a2859fa1a7173b4b98 : SUCCESS in 17m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2237fc2d6174a7db0c6445c8c465677 : SUCCESS in 16m 35s","accounts_in_message":[],"_revision_number":5},{"id":"8b0d18272bea2e9eaa9bae64ff20250e89b262d8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 17:13:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6b3233d9af149762df42ecde7c843a57b98fbf32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 18:56:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bac753f57aa5460ab4de2582bceeed9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21bac9f6ddd74ce4bfde93c5ffd9a682 : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dfa9551045c04e52a57c82bbf548585b : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0f6b73f5c104045b0aa8f03b21bcbd5 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46ee2ceae7db49f19a0236ce9cfa8637 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc99f772a1ef4affb52c5fa1e1a41b12 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1cb5ef20041140d1842d0dadb7dcbf66 : SUCCESS in 4m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a13a8ff3c2840899b1b1df45c529128 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/02b5a7efe5324f53b71b2e98b0133060 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aa41239ebfba46498d8a324911d5588b : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1eccb81a8ef42949dc7fd07dd8eb1cc : SUCCESS in 7m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d4afecb4cc99425e9059261124b0b11f : SUCCESS in 19m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb7127216ab240ccb7a0df117540ed37 : SUCCESS in 14m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/89fdc4b013be4754a7c097e0260b31e3 : FAILURE in 10m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f24454ec767428a9a4d8aa8805c0af2 : SUCCESS in 12m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3ee3f30152c04e2d91f53c56c8912ea5 : SUCCESS in 1h 00m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6f2b0aeeecff41e0973375a210986922 : SUCCESS in 1h 03m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/93385c02ff7a41c69df17478c9c1b8c4 : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1630056f4f604c5bbaabc94efebffef3 : SUCCESS in 1h 00m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f95342cdd05f44f18bf7aeceef79db5f : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/aed6bef6a7b34a3f8f89f3b26bc48c6e : SUCCESS in 1h 38m 01s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/32c5fbc3c15e484b8632189e00cf91b4 : SUCCESS in 1h 34m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69c18f0f63ac4cf79bb5adee56e05354 : SUCCESS in 1h 34m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4d3bc2fd7a347d0a5435696013378a0 : SUCCESS in 1h 34m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3d0aba1af447464d92a7800f63d4e531 : SUCCESS in 12m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/afd1447825d7445b9a6a8487dce4f752 : SUCCESS in 12m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99a6f5d06899452ab0f0701ebb11b3b6 : SUCCESS in 21m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cfffb82966084eeeb7273e8a35b29d6c : SUCCESS in 15m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ebfb45716e3d4c08a1b3d5b39a1d2b2e : SUCCESS in 26m 26s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b0b172ae3b214e05bcd94aa7b3e489a6 : SUCCESS in 21m 41s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00f282e314af4b4ea44757bea11af372 : FAILURE in 15m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c68b45c3c3d4ad4b8b791189f81fa0e : SUCCESS in 30m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/96a5be3a8d494806ad4ee6e143c0ae5a : SUCCESS in 16m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9aef2c136f69415aa84caca9693354b6 : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":5},{"id":"8612dee2895639084a9d45a766a19aaafb87b613","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-08 07:52:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"52ae847bae4fb4e7ac919adf7743e74cc4a17561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:33:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bd2c91578cf4f8e92d70f0417b29b9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6830105ee36248f3ba7b89bdbf00c9e1 : SUCCESS in 12m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a7c32448efd94f56a3de9073a348b663 : SUCCESS in 7m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33eba10d15e244c1a3ae55d8ecf5456b : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc481a8e41942f09fe0e1abff380ed3 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ddbe289bf2904117a106f61903c36480 : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/142aa9b1f7f4473199150dd33c0de03a : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30abfc9040744003b7309342f8234f96 : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cdce4d442dd14fb484e2ae96ad53c714 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d8d27fcfc6743378d7c46e0f00ecdbd : SUCCESS in 6m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c39ee076cb54463790ddd244796f6140 : SUCCESS in 12m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7a62882a09fd45309b8f38b6f67f5a07 : SUCCESS in 17m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5f43aca82954af5b3533dfab7e27e9b : SUCCESS in 18m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7e617651ecd0435bb38ad170882ea87a : SUCCESS in 17m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f80d52eb06245c09f4c80c7249a695a : SUCCESS in 20m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b5d70cde1fbc4de9aed0f4a30a864826 : SUCCESS in 1h 01m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4b8053c1c71a4f8990df234beb7c2024 : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/45a9781eb9ce4d16bdfba4cdab70b287 : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8363df5c97d3468383903c4122ccaa00 : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d6835c1701604094a1ec562bde1cf90c : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/076ec24ebddb44fba642bcef14943d03 : SUCCESS in 1h 38m 13s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dcb277332e904c98bdb53ce03ba662c4 : SUCCESS in 1h 34m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bda1285a147d42a98b9ea4d3dad1aeef : SUCCESS in 1h 30m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6009fe43ee6c4bbf85a2e89795ae8064 : SUCCESS in 1h 38m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/289b27eebcb342a799743c44efc0af9c : SUCCESS in 11m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/be10c28b6224408cab4279490d3dee08 : SUCCESS in 18m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bad02790cf494d759a10d471db46494d : SUCCESS in 21m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f6bc76c1346445f8952fcb6d5692f9c : SUCCESS in 19m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/547d4054b4ec471da3a8b239b3522053 : SUCCESS in 37m 24s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2d4aeaedb36342fe9d086e0fc99eec66 : SUCCESS in 26m 15s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab9736215ab849c5827074300c8ea2f9 : SUCCESS in 33m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fa6abb17ad84dd98dd19f92f754ba8f : SUCCESS in 30m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c34e76e0f871432486ce100666529992 : SUCCESS in 18m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f36e1e0e91d4a50a24e0aa286666217 : SUCCESS in 18m 04s","accounts_in_message":[],"_revision_number":5},{"id":"8f07edd74eb07f62cbbbce262fd4033fad4a508e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:34:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"47106a2730559d4159ef76d8df42769f68413562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 13:02:32.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce49aca9a75d4f2aa0dd5f81fc0f6201\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ccb8e5bf8143c9adf83a10c27f740c : SUCCESS in 9m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4e50879cbb3b41e0aaddbc637fe640b7 : FAILURE in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759a1bcbfb67443289da7585429389ad : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7019c8513e4e4262ae1fe041a83a394e : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57bc0b86ba0149dd9e2c8547568be10a : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d65b3ce965e4b26b84e87d442b4775d : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1334d044e4004848b3b93d389302325e : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e96a6a4831d54ec992e5172c8ae54e25 : SUCCESS in 7m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/686f5446fc00444486c4ade4a1e73467 : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15a7f532ab6b48639c46fe81440a8f38 : SUCCESS in 14m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1c217a0224854e2bb1804034ed3ce6d5 : SUCCESS in 21m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cb7972720d8411ca10daa5472c878b5 : SUCCESS in 17m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1199f881f2c84545b9cc656da95d7ee9 : SUCCESS in 59m 02s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/31b0c002303e4f68834de0bc72f2937d : SUCCESS in 1h 00m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ef791301b5514275ab206ce1c2a32318 : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f48f8b5d512e43cdbed04b21ec07bf60 : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3517fb1b18bc4333910f1d0feb8e3395 : SUCCESS in 1h 00m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c0ce2de095ef40679fa873f9d876aa2c : SUCCESS in 1h 42m 24s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9f54f05efcd743fe857e1126262feef4 : SUCCESS in 1h 46m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c59cbb0e03c4fc0939e6c3882725cde : SUCCESS in 1h 39m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e492f98d528242a58e8fadc54851d431 : SUCCESS in 1h 50m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ae4fb536aceb4aa0ac7e8132bdd13397 : FAILURE in 12m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4fe169ce2bb945f783b3b663cb8c9b9b : FAILURE in 21m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bd42f05c57c46cc88aeca6d9efaf11f : FAILURE in 20m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c0d8051d28e54929b8d156c4543e64e8 : FAILURE in 18m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/34551beae47b42d2bfcc35cb9a42cdb9 : FAILURE in 24m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/08dff8101c2346e88ead814b3a6dc70f : FAILURE in 24m 05s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0020147bbf44270a0966012f14810b7 : FAILURE in 22m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/294e8b807a784d21be4ad63a82a4bcae : FAILURE in 21m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/10408c261ec84fd2adb691ade7d9da87 : SUCCESS in 21m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf377b9773db4b97be7d03a7bf835bcb : SUCCESS in 18m 53s","accounts_in_message":[],"_revision_number":5},{"id":"7d940583bf788808eae57e2aaa65ee6a54707755","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-08 14:28:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3cb7669f77333e484eb0416b4ca61f3f06fb45c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 16:08:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50534b1f2391418f8a369b4003a71ec6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7005c09671184c91aae9e1f5935bcb12 : SUCCESS in 9m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8e1b7eb2aac043c9a923f5b08a5e7b38 : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e77506557e36480dbb8fd26796618939 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12375a77d4874d41a168d4b50442fc2d : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2d5e0e24e21433c9ca5657b148fba07 : SUCCESS in 5m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c8a2b46da9b84080ab9e828f1ca2dbfe : SUCCESS in 8m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b8f30afe2554e3cb8fbb93cea0177ec : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5e9ba24bcde1463e977450361e028606 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a7b49c92a9d846c8ab79481059ded934 : SUCCESS in 13m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aecbb9b32bb5400d8ce481f9d701e089 : SUCCESS in 11m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/195477fb4f2e405c9eaee19fdda4b1a2 : SUCCESS in 9m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92d1c7d1e76a4f7bbe5640eb9fc35924 : SUCCESS in 14m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e561ae70e93746ab97b3116ed64bc567 : SUCCESS in 10m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a45387976644588a1375805b2f39134 : SUCCESS in 18m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/61d858f2521a4e34afa2503f9d7634a2 : SUCCESS in 58m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/73a9d30cde6b4f5e97074adfeacfb31e : SUCCESS in 58m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6c21feedcea34250a512aeaece12c17a : SUCCESS in 1h 00m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5c5b41a568345bca114eb85100d19a8 : SUCCESS in 58m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c534c992b49a46e9ae119e22e7a4eb0d : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/afb79613d74f4381943781a21aa3d063 : SUCCESS in 1h 37m 09s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/81c6ce7f4c2647158a42b516a4ab16a0 : SUCCESS in 1h 29m 35s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/298dc03a2862445d8673bc5041954724 : SUCCESS in 1h 30m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be1b6aba9b5c4ba68e6d885f8cc19838 : SUCCESS in 1h 36m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/65b573060dae4803bf82a410f7ee867f : SUCCESS in 16m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3d586a1cbce54a8d9ed792b8f1b2edcc : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77bae9729ecc406c9c6fa565285af87f : SUCCESS in 19m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0f1f300a5cd84631a1451b84334a1720 : SUCCESS in 18m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1c2e36209f1c4bef930da07b3270bfc8 : SUCCESS in 35m 56s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a2702b6e73b04c4b81accc2b5124fcff : SUCCESS in 27m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e410340093da4b8ba6fcba91eb7be2b3 : SUCCESS in 30m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12416e7f62af4cccb7a0e24a759b7283 : SUCCESS in 15m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d3f1110e200d4919b3e8b1cfb62e87bc : SUCCESS in 10m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d4c6a2b97a54ca7ac8bd126ed6e3d53 : SUCCESS in 16m 57s","accounts_in_message":[],"_revision_number":5},{"id":"e6a3f586f582a4c8a7f92e74d7dab80122a555a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 16:08:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cb5f1c36adf4411512a973ef842e46fabfae4e8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 17:50:07.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b08bab06958430aba1e9d79225a556f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b60dfeccae8445ab630e8d33b8d93dd : SUCCESS in 8m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b7c0aed08e1e42cc844b819d1a709834 : FAILURE in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f1a0288e4de45598bf12e90506de32e : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88df080349a34b96be53aa5c0939bad9 : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a6f3632e7414cdfa40abd9f2ea0103f : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2929274333534f8294a32e950080e3a4 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/52c0e1419a444ceeb1c91e416546bed8 : SUCCESS in 6m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa774c6d07534db3bc74c31f714938ad : SUCCESS in 6m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3d79add84c0e46a8a1f2cfc496f8dcc0 : SUCCESS in 10m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ec3a0c0d6604e17a5e1b12a210f0d6d : SUCCESS in 13m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/36c7f8dc8dd34f559810664f2ed73007 : SUCCESS in 16m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a45522bf3ef4cddb3a3f226830ac5fa : SUCCESS in 9m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5fb4a7bd557d42c09e9362d92b22210b : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/41528a9aa74945378a04c715b4079e9f : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e43e0d54df96441da335f22e8e03f181 : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d990356e91054074b0041233b5613fe4 : SUCCESS in 58m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/be7a0e10972f4ca58937cfe14cadbb95 : SUCCESS in 1h 07m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bcdc73a19e984e87a94d42c049b6ab92 : FAILURE in 21m 31s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/830465ee61b143b29f80813adeb1d0b7 : SUCCESS in 1h 40m 40s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a1fec39455f430c809c505fc8a44229 : SUCCESS in 1h 31m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88846334a13c4a2fa7320152920bf561 : SUCCESS in 1h 37m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d5793afa3e074700afd1dfe9153a646c : FAILURE in 13m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d1dc31a4e9874014829c7e178321cae8 : FAILURE in 15m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b70d3ad8355475ab83b5f19430a1ace : FAILURE in 18m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/24055dca7f204d2cb75d137a55560e98 : FAILURE in 15m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4ddc8a86a84c42cf8a61df938a850877 : FAILURE in 26m 39s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/424d5aa4d4b34e2e991e4a4969d75d54 : FAILURE in 19m 26s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52ec7e00aba74a89ac5b72f8594c99d4 : FAILURE in 19m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c1a1063a96d4e698c6e7f3d240f0d54 : FAILURE in 16m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6869bdff7a8c4deb81b65d8d2e6051ac : SUCCESS in 9m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/989b52d014d64e059a3067361d83b1e7 : SUCCESS in 18m 31s","accounts_in_message":[],"_revision_number":5},{"id":"a9535e2b4480bae7435e396893dd73ed220f44a1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-09 08:34:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"abc2e74658bf771eec3969f3f0262b21dc855919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:21:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/348094b957fc41658d7fe15d0fcf7551\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49d00ea81534c05996fe2d4b6b9c8b8 : SUCCESS in 4m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/452e27a61510460db61993dbd85042a3 : SUCCESS in 6m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ece3c803b6a04075bdad022f64d6c827 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f6784c1ad04ef3bdfeccbc3460c9d1 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6dc39e018f9b4889b1004883cfbc6378 : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71f50620ceb8423db121ae9ca033a7ef : SUCCESS in 7m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/765badb5c2e74fb79b8e4d3c99120a6a : SUCCESS in 2m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a2b67719f004aa8958a77b52cc316d2 : SUCCESS in 3m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4bb9845f0a164dd2813aca4ca6b4e6aa : SUCCESS in 6m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9642e541358c49a5be1720bc550e6e49 : SUCCESS in 12m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fb52707044ea41e5a1dc37f8e3e3190b : SUCCESS in 20m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f5d6f19baee494896e93186a7c892a9 : SUCCESS in 8m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9de9c1fc7fd14fd2b2572f15dad1c4ad : SUCCESS in 18m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/527f90bdf01444cdb77629f722af754a : SUCCESS in 21m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f6f6abc8fede4e99953e89db7d9e78cb : SUCCESS in 56m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e1e22116c2045d2acf687f5966d0afe : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/00ed00458b0b49bf949a5cb75726fb71 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f21fcba3fed49eca2f5610bfffccb97 : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/168ffa39655549fe9c286c6febdee55a : SUCCESS in 1h 04m 54s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0626b48d89684ceea993cfadb61b24d2 : SUCCESS in 1h 40m 09s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fe4a8f54dfc74f918bcf798759140c65 : SUCCESS in 1h 32m 58s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cfb8c880d7b413da5e8d3bb956f6c47 : SUCCESS in 1h 31m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8de833127ddd4af987f481d2067c781a : SUCCESS in 1h 44m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4fcdf1294d8b45559052a8585e2f4781 : SUCCESS in 11m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/83139f17d91942c2b918f479bd15e19d : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edc04d9ac06b4dac997096ce13b405d8 : SUCCESS in 21m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e27f6e52004b4ee9b1310b1814d99578 : SUCCESS in 16m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/93866fd4d92a41c3a672a82d5735c8ce : SUCCESS in 36m 47s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b60cb062c4174554b66a4ff52b7f4c9f : SUCCESS in 26m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/393dcc904b104448acf750918df6a981 : SUCCESS in 28m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12cd268fdda3408cb33036ba69d63208 : SUCCESS in 30m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/644a5306dfb644438e0c94d228c75aef : SUCCESS in 10m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6eb8dd0f53304925bb0dd2c58a762f94 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":5},{"id":"39c60dbabec19b4fbe33e5fb32f45b24343046ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:22:04.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fe0697df7fa8f21cac5e4e32e35d55a3212a912d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:12:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa6a7c696d0c441c99d7cd5dd0f435df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f529f9f46ec049dbb4e791cc6b07bd13 : SUCCESS in 6m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb28fd94cfb24a19a285220b15976a34 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd67c2d43d1247d68190f78c660b71ef : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c91e1109bad54388bac49c18c8aaa29f : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29446aa0706e4d0a89eb182c8e7c39ea : SUCCESS in 3m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/549a410603994a5f982f1e1640b52c60 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f9e05aef3a5348bea7d2c65c78a539a3 : SUCCESS in 7m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae5bff56b8114bdaac9df4a6522ad5c4 : SUCCESS in 8m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/417da80427b74ecb9e4595d226ee8f11 : SUCCESS in 17m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ccc845573ac42a287319a32daa27394 : SUCCESS in 11m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7b5ffe252aa541b598cae9bd568ac676 : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/268f1f6f513c4cd79b0c1b61a1ef7bbf : SUCCESS in 17m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1086fbd76fed4bdfaa723342e45f8afc : SUCCESS in 59m 59s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/89369fb09934499699b6875a6265d053 : SUCCESS in 1h 00m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fbda934837a34a68a6e6ea5b26e97b1a : SUCCESS in 1h 00m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5dca3ef21627469c896bf5decda5441d : SUCCESS in 1h 16m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a13c1575fddf443db5418f954dfe3d9d : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/381a5368b7a5471ebf219fba0499a69d : SUCCESS in 1h 34m 47s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/193c16e1b364409183c136f3cf860a48 : SUCCESS in 1h 39m 46s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ded6f55f3494a2c80622cf381f42e43 : SUCCESS in 1h 49m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea79e5bab2ec48adae169ef8e1a4ea08 : SUCCESS in 1h 35m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ef189bd37bcb491eb14bca1da02293e2 : SUCCESS in 25m 35s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3f9480a49669410780691d63c552d7ca : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebf8c1d67818434aab3d2e32dcc932f2 : SUCCESS in 18m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8e020be8f19540b78f54a88b3b196ae1 : SUCCESS in 16m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/abd685490e0047e3bb3719332845da07 : SUCCESS in 27m 00s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f16398f6d128438eb3dfe8bf7a8984f7 : SUCCESS in 38m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdd42e441f7c4022a324f35d146cea21 : SUCCESS in 15m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a3402f9418a4bcaa37e3adff5159f0e : SUCCESS in 23m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cf90764087cf438ca93552b79243c7fb : SUCCESS in 17m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cfd07178ec34e94a9a1e39cea9466be : SUCCESS in 17m 10s","accounts_in_message":[],"_revision_number":5},{"id":"e2aadba0166b5e30c24895d4b1838d764b7c4f39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:12:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ed71744aea619b41df9abb81826ca205f7bf9589","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:13:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b579faad863f4f419bcde498c53656cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e4399f453342159074ad2aad38fc97 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9595448f120442d2b2654a6059352aef : SUCCESS in 45s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a8635f89bcee9032594fd0da23db32dda8e9c0cf","revisions":{"f24a14f1cc22f7782edf811b1240d9727681a214":{"kind":"REWORK","_number":1,"created":"2026-03-11 10:31:12.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/98/979998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/979998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/979998/1"}}},"commit":{"parents":[{"commit":"fa4f0be487f5a6979964c910f47581529cc6cf66","subject":"Revert \"CI: Disable seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa4f0be487f5a6979964c910f47581529cc6cf66"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-03-11 10:24:21.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-03-11 10:24:21.000000000","tz":0},"subject":"Hide sensitive inspector rules from ansible output","message":"Hide sensitive inspector rules from ansible output\n\nSet no_log: true on inspector rule loop items that are marked\nsensitive. Add the ironic_inspector_sensitive_rule_no_log\nvariable to change the default behaviour to not log\nsensitive inspector rules.\n\nChange-Id: I47035e2d1625fe7e95f29919d59676ea0cc9ee67\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f24a14f1cc22f7782edf811b1240d9727681a214"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f24a14f1cc22f7782edf811b1240d9727681a214"}]},"branch":"refs/heads/master"},"70c23c9a637549b215cbda0215987544850f9fe6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-11 10:35:48.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/98/979998/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/979998/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/979998/2"}}},"commit":{"parents":[{"commit":"fa4f0be487f5a6979964c910f47581529cc6cf66","subject":"Revert \"CI: Disable seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa4f0be487f5a6979964c910f47581529cc6cf66"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-03-11 10:24:21.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-03-11 10:35:24.000000000","tz":0},"subject":"Hide sensitive inspector rules from ansible output","message":"Hide sensitive inspector rules from ansible output\n\nSet no_log: true on inspector rule loop items that are marked\nsensitive. Add the ironic_inspector_sensitive_rule_no_log\nvariable to change the default behaviour to not log\nsensitive inspector rules.\n\nChange-Id: I47035e2d1625fe7e95f29919d59676ea0cc9ee67\nCloses-Bug: #2143889\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70c23c9a637549b215cbda0215987544850f9fe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70c23c9a637549b215cbda0215987544850f9fe6"}]},"branch":"refs/heads/master"},"f27b28aeaa3075ce5975c6551e4b03d97874959b":{"kind":"REWORK","_number":3,"created":"2026-03-19 09:56:51.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/98/979998/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/979998/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/979998/3"}}},"commit":{"parents":[{"commit":"fa4f0be487f5a6979964c910f47581529cc6cf66","subject":"Revert \"CI: Disable seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa4f0be487f5a6979964c910f47581529cc6cf66"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-03-11 10:24:21.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-03-19 09:56:10.000000000","tz":0},"subject":"Hide sensitive inspector rules from ansible output","message":"Hide sensitive inspector rules from ansible output\n\nSet no_log: true on inspector rule loop items that are marked\nsensitive. Add the ironic_inspector_sensitive_rule_no_log\nvariable to change the default behaviour to not log\nsensitive inspector rules.\n\nChange-Id: I47035e2d1625fe7e95f29919d59676ea0cc9ee67\nCloses-Bug: #2143889\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f27b28aeaa3075ce5975c6551e4b03d97874959b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f27b28aeaa3075ce5975c6551e4b03d97874959b"}]},"branch":"refs/heads/master"},"ed567c455f7ef822a026ff13330f7775db4519bd":{"kind":"REWORK","_number":4,"created":"2026-03-20 09:44:47.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/98/979998/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/979998/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/979998/4"}}},"commit":{"parents":[{"commit":"fa4f0be487f5a6979964c910f47581529cc6cf66","subject":"Revert \"CI: Disable seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa4f0be487f5a6979964c910f47581529cc6cf66"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-03-11 10:24:21.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-03-20 09:44:39.000000000","tz":0},"subject":"Hide sensitive inspector rules from ansible output","message":"Hide sensitive inspector rules from ansible output\n\nSet no_log: true on inspector rule loop items that are marked\nsensitive. Add the ironic_inspector_sensitive_rule_no_log\nvariable to change the default behaviour to not log\nsensitive inspector rules.\n\nChange-Id: I47035e2d1625fe7e95f29919d59676ea0cc9ee67\nCloses-Bug: #2143889\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed567c455f7ef822a026ff13330f7775db4519bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed567c455f7ef822a026ff13330f7775db4519bd"}]},"branch":"refs/heads/master"},"a8635f89bcee9032594fd0da23db32dda8e9c0cf":{"kind":"REWORK","_number":5,"created":"2026-04-01 08:13:40.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/98/979998/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/979998/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/979998/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/979998/5"}}},"commit":{"parents":[{"commit":"fa4f0be487f5a6979964c910f47581529cc6cf66","subject":"Revert \"CI: Disable seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa4f0be487f5a6979964c910f47581529cc6cf66"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2026-03-11 10:24:21.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-04-01 08:12:50.000000000","tz":60},"subject":"Hide sensitive inspector rules from ansible output","message":"Hide sensitive inspector rules from ansible output\n\nSet no_log: true on inspector rule loop items that are marked\nsensitive. Add the ironic_inspector_sensitive_rule_no_log\nvariable to change the default behaviour to not log\nsensitive inspector rules.\n\nChange-Id: I47035e2d1625fe7e95f29919d59676ea0cc9ee67\nCloses-Bug: #2143889\nSigned-off-by: Matt Anson \u003cmatta@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8635f89bcee9032594fd0da23db32dda8e9c0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8635f89bcee9032594fd0da23db32dda8e9c0cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
