)]}'
{"id":"openstack%2Fkayobe~980241","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Ied90077ba7104c67ea2623b4de16d56d9f1c1c02","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-03-16 15:21:17.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-13 10:01:48.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"38330":{"account":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"last_update":"2026-04-16 20:14:14.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-16 20:14:14.000000000","reason":"Change was submitted"},"36781":{"account":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"last_update":"2026-04-16 20:14:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied90077ba7104c67ea2623b4de16d56d9f1c1c02","subject":"Add Valkey support","status":"MERGED","created":"2026-03-12 10:27:21.000000000","updated":"2026-04-16 20:15:21.000000000","submitted":"2026-04-16 20:14:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"980241","meta_rev_id":"a7b42e367e2765ad7b79e6c72c1d4fe27ec9ff26","_number":980241,"virtual_id_number":980241,"owner":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-16 20:14:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-16 11:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-03-16 15:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-03-13 10:01:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},{"value":2,"date":"2026-03-16 09:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-16 11:15:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-12 10:43:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-12 11:06:13.000000000","updated_by":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"reviewer":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"state":"REVIEWER"},{"updated":"2026-03-12 13:55:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-13 09:23:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-03-16 09:00:56.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-03-16 13:44:11.000000000","updated_by":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-04-16 11:15:29.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"753ade91ce8967c6a85533aaa3daa387410e6dd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-12 10:27:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb89cc13d932fc3b64500e190ec786682b1d32c9","author":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"date":"2026-03-12 11:06:13.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1872e2d3dcc7f2371331706799ca11b5999d6af6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-12 11:08:32.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c04b260bd74e72a50df280bad8466f12df37330a","author":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"date":"2026-03-12 11:10:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"03c6ff71d5634eb64432203da91f5f454d2c328c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-12 13:18:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cc6a3776d813e123034eefc03e6522f53ddf9b27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 13:55:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd8a45547c3a41c0a90c7018dd23ffad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87d1590d4ffa46c1af74e2255312e173 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58fba435b87e42fa95dded90db5fa6fd : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/569ff1dde7e34c558a05b1b370db7fce : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f142324f10a4ab58fbb73184e6f021b : SUCCESS in 9m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6833cab41ab34b3d935a0e37bb95231d : SUCCESS in 3m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/131b1917c192484d88ed5d2a57500fa8 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c2e513f5d7e2408790f959cf8dc6624a : SUCCESS in 10m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/13e95c65460941308c6afab0c0d33272 : SUCCESS in 12m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0a53519e20fe4009a9a9953d07cdc90b : SUCCESS in 18m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10d96af38df64c4d8cf5510438bc74b4 : SUCCESS in 11m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fee0b1ac6bc04186958ad62cc4157b1b : FAILURE in 14m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e8917c463d16466d9c0bc96fb2524c26 : SUCCESS in 17m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cca1ad5622aa45a6be3aff07a70059af : SUCCESS in 14m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a51cbc80578343fe997a6e5f5027bea3 : SUCCESS in 33m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e1cb29a1dc5433b91a5791756ad2755 : SUCCESS in 58m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a888ae7c554c4f8abd4fa700b0a6b0f9 : SUCCESS in 59m 25s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/52124bde287b4ec48280d00d8dde3b98 : SUCCESS in 1h 07m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/18b90211bba24817bf7ea85add71a419 : SUCCESS in 1h 05m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/283b369a78c54c82989100dcb85e6087 : SUCCESS in 1h 08m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ac2e9e9ed1844bdb3682140db292f9d : SUCCESS in 55m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/914b1701c45e4b27a8294f6ff21343af : SUCCESS in 59m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/116a9485d1354b66ba5b897656eb1ec2 : SUCCESS in 2h 23m 56s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5aadc3a4e6c14b03af35079d35ae9112 : SUCCESS in 2h 20m 23s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59feddf4355440d3a4c528b8a400a1d2 : SUCCESS in 1h 40m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0afcbe62c69845fa8cdee4c53d4ded59 : SUCCESS in 1h 09m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d3dac39911ec4c2d82e8f9008639b2f4 : SUCCESS in 18m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2223994c810644c68d02cb9d55f8af2f : SUCCESS in 16m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/63a6511a8a86469fbf951ba7ef9b04bf : SUCCESS in 10m 40s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fe2e11f1532d40799198780ebf93e7be : SUCCESS in 18m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f66cef627e2241b0bf08e8708a4c402c : SUCCESS in 15m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/04a635541bed428ba558edf217daf98f : SUCCESS in 14m 06s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c89cb5829c134f41839be9bb9fd7e9bd : SUCCESS in 37m 09s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e6eae182e39a4c9ebd51627dcaf7157d : SUCCESS in 36m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc4c72678d7c4645821ef6f0ea75443d : SUCCESS in 24m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/44aef68ea7194775a957c5d77a209dc7 : SUCCESS in 13m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/336b41d27f7e44e9a184feee94879b72 : SUCCESS in 17m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e937df9f9334480380a59df61b4730cf : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":2},{"id":"c3b0c2f9ff7c8612255e42f1ca2039a53786136e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-13 08:51:47.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4dbb5119ba4c266b2f8f5100be912f0ac48b60a1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-13 09:23:50.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a1f8a3b5536af04d2e29853c4c82b001e4b39b32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 09:23:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bad0102c55e0474ead1022506e84dd10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21fd045be2f341e682e86bdae1fbd900 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/057581ec55ea4f4dbc56098473d88717 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c980114870844e80a90fc070d231bb57 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e78b9b4f3a04cdf83b2475529d94c2d : SUCCESS in 9m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0813bf08b708460180dc3e996d8e4c61 : SUCCESS in 2m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19046209b8fe4e8583cc313488063b5f : SUCCESS in 5m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ec7f900e851145e7aacddee53ab8a024 : SUCCESS in 5m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0255eb28e03f42cbb2dd94539acfc5cc : SUCCESS in 9m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/16246570ed09414cbd8d10252b79c3aa : SUCCESS in 12m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdd5debf8bfa4ab2be7a4d0e16347d64 : SUCCESS in 12m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fc2e35ed7183426a80f9b12f3019f36a : SUCCESS in 23m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/621974dd1a0a4780a6da96803f13ffb0 : SUCCESS in 10m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fb8fd6082b645d3ae01f59e17924480 : SUCCESS in 18m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/eba1d5c0aa85460d96de06a06fe515ec : FAILURE in 26m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e82aa65108d84264961e0d836026c38b : FAILURE in 25m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/67472e6d75f549e0820873ceafd2fb4a : FAILURE in 22m 39s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/6656fd72c31b42229352c6e46b1aeb43 : FAILURE in 20m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8036a547c30c49cb98b1dd5a81a79751 : FAILURE in 20m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9ddc731acde54dffabf34edddf06f0ec : FAILURE in 18m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38eee011c10f4223a72230da9cf0fc77 : FAILURE in 9m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/195d7f1adb514e5fa54be09de14e2ed9 : FAILURE in 17m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/16f60466219e4d32b31e276b59cd5b62 : FAILURE in 12m 05s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/440237be6cac40d3907484d2b7740457 : FAILURE in 10m 05s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0734ec6aff374f63915849bd358b974e : FAILURE in 11m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dce437413cd460d8f7ec0d6bb63bfe1 : FAILURE in 13m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/92ec9f5edcee404587a65a8445bd51c2 : SUCCESS in 10m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1d5960b972c04939a4f299240c656fde : SUCCESS in 21m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/52569659bd40418fb6cc0be3eb34f84f : SUCCESS in 20m 35s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/90df64210eca4ab8bff291e10f916277 : SUCCESS in 10m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b6732056e1146e9835792cf18db17d1 : SUCCESS in 18m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3e90e5562e1947d5927070c84f116884 : SUCCESS in 17m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/289d17c1f7b74cb2b5cca05a1ae7b65a : FAILURE in 14m 10s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6c5b0d2783c449dd9cbd946eb1802897 : FAILURE in 7m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d73ab5ca92684bb0ab1ea728cf9e19b2 : FAILURE in 14m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/51c3e31b84844ee3ae81d5cb87051c23 : SUCCESS in 8m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fd890d0493a24e0db8ea34d413bb80f5 : SUCCESS in 17m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2097022c78d349d4bb6f4b0d1ea1a2b0 : SUCCESS in 15m 32s","accounts_in_message":[],"_revision_number":3},{"id":"b42fe00dbd9a1ff656fa18afcafe56537d35daf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-13 09:30:01.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"fdf16aaf82a67043e7d0ea6f03b25b42d62deb2a","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-13 09:30:39.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"48bf4fd71e6e1b6c25f887c5ca51b43e23794340","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-13 10:01:48.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3971afffa6126409f22684ca6b3993bf279d7cbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 10:50:55.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla-ansible/+/973214 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"989329d02c5cb8e80c10b66d757b9af388bcd98e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-16 09:00:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"96ef13baa79565ec1db7bc6c79d44a7f7f614fee","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-16 15:21:17.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ba7e7062fb1bffe1d4d79e85e02e82bc8a4e8680","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-03-16 15:21:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f663177beaa657185fa418ab07a9c82d2704c01c","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-16 16:11:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f0bc66d1c1bb3fd118b694101f747ba87425c3f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 16:56:09.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19199f92b8a146d7a917b83a76ce6a49\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ffab8c8971446ca974e02b1d2dd6195 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2d47f7f7c044b4a5892224890f0c5b : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f851c13982b4fd0941a082318feee22 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656b89a60e0444a59281b64523e80029 : SUCCESS in 8m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af9cc5a3c4b645bc951aa0d590f2ca9d : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79e9cd72e95447fe8288aece29c84012 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5d450ac03c614c51b2ac44cc4ab335d6 : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6b3d9f4b778e41af92d8a2d9f6b748e1 : SUCCESS in 9m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fef38862d31242a28b32dce0b2b40b24 : SUCCESS in 16m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17e77caf7d164ae2afd5daa941849e4b : SUCCESS in 12m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/60b6226d0c8046fabad6c016a181df5e : SUCCESS in 10m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/144ec81869004e5e977512608e41a0ed : SUCCESS in 20m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fca459c032744a10843a2c7dcc56c384 : SUCCESS in 13m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/40a9a268f89045a98a0aa55d55689f8a : FAILURE in 19m 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e550f02050f84527befce8706cd5de32 : FAILURE in 26m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7cd24b2df0f7494ca0b200417957259e : FAILURE in 25m 43s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/220bc91ee563441fba4edc07870c8a73 : FAILURE in 21m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/78541a9dc7f54c26b594d0744d65b015 : FAILURE in 16m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9a4355e2410d472196281fe9a49cc392 : FAILURE in 28m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcc13220022546f49cb60ba1f0ff7bab : FAILURE in 29m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e4b964c43db545da974087d2e537e755 : FAILURE in 18m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d5556ddafa49425ca363aecdb076bb48 : FAILURE in 31m 34s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f0214b406cf5413290be6e664cab0f7f : FAILURE in 30m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9713f3cb720426287b730dba97046a7 : FAILURE in 36m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5b4e097b7fb404dbe0f5f0e09b14ec3 : FAILURE in 35m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0b46441feaf444499cbc1d142a37521c : SUCCESS in 16m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2a7ea347fb4948a9a8d8d63188561294 : SUCCESS in 16m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d3cd174b58bf4eb89165857e817dddb6 : SUCCESS in 15m 32s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/37fe7b96e4aa40baa6fe0140548906f8 : SUCCESS in 18m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/151d87b569374083b5e069bef8541074 : SUCCESS in 20m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/79c71a278d6d447b91b97f6b3aa0380b : SUCCESS in 15m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/03a42dc6506d45eea5be03535e379952 : SUCCESS in 33m 14s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5964cd168e3443999c1de4a44799a971 : SUCCESS in 39m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/794aa058d2054e18878ccb53e8ec7015 : SUCCESS in 33m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aa2c8cdbbec04aefb002dd522192c9ff : SUCCESS in 13m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec2ec6c0b7ce4ec4a378eaabc8c55cd0 : SUCCESS in 18m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a72f84f7a80a43e1a96670334669f67b : SUCCESS in 16m 19s","accounts_in_message":[],"_revision_number":4},{"id":"a39df305422b2ef4d1517134f7140d020de57f56","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-18 13:35:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"190d36f3262546ae138f70c4226a7ed55f95fdaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:00:56.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b896f2548cd945fd9d21a95a95c1151e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a81bdfaa51bc4de3ab5224b8fb508c33 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7b248c6a8e2459890ffe5d0e8f0bf65 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5af1e669675546a2a3673814ce5c85dd : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86c44f62cae4597a001c45b0fce0735 : SUCCESS in 9m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17700f7fd0f64eb3b080914cc6786887 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cf03bac83e3744edabd584f783974148 : SUCCESS in 6m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9f5cbcd4e92943628f5f4c10e6b0f0b8 : SUCCESS in 12m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/383efc2abaa54da0b03c82f3a1de020d : SUCCESS in 14m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7c28d32cf6f3401bb1c430be4dde7b67 : SUCCESS in 14m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a01eafc46ea4be28f3effce71b663d3 : SUCCESS in 12m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/111d2f6f191c44ce84533d3adfe10088 : SUCCESS in 18m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bbf95fbcd7ff4f0d80d20e3d0539af37 : SUCCESS in 21m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff1a7a437d7a4f1ab4e5cee1bb4d9b44 : SUCCESS in 15m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3812a5c03c094378b7250a7cfb796ad0 : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/32b3424e22074842acd8d1fe7596a78a : SUCCESS in 1h 11m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f6774a1423b84afdb7cd54d04630350c : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/81abf0e6552f4a51b1e8606fa9a64fc7 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/15d8ddf45ad24d28b7bbb56e3423fcca : SUCCESS in 37m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/00d7cf679cd4416a9cd6439f37fc13d7 : SUCCESS in 36m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de22261f58384e628e0b74977700217b : SUCCESS in 57m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eb61652a391248bdb8016562b4bf92e8 : SUCCESS in 1h 00m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c1db7cc4cfc64b03aeb4858cdd2a1620 : SUCCESS in 2h 18m 13s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/28042b4528db499f9ddbf4d021213faf : SUCCESS in 1h 21m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5a66d98e6b24cc796458d7eb60796c9 : SUCCESS in 1h 15m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b93deaf3a6e74fcabb05397c2a92793b : SUCCESS in 1h 44m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/368b0856c9d044caa852665a72ec9339 : SUCCESS in 21m 35s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bbb93f6a7c4d45e4ba9c4385687bcf13 : SUCCESS in 9m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3108d03b76db4bfa8d21e0215f8f1239 : SUCCESS in 18m 40s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5ea85e0b109143528c5847026b436705 : SUCCESS in 18m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9baca6cedd6a400ea48b2779dc6d8e62 : SUCCESS in 8m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e369d83c52844b1bfaede2c6505f813 : SUCCESS in 8m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/031469ea05fa47ac9a58108520196953 : SUCCESS in 22m 02s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/54a43c6a342a4e92802749e754edec73 : SUCCESS in 36m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2da4e18748ef49e3ba88abb585374f2e : FAILURE in 19m 00s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d9e54ae355ee465d82620f9beb12cacd : SUCCESS in 8m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f9873acfdd1b45158756433cc5494a00 : SUCCESS in 9m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d93fa0688ee049fba5c665333dd2e2f2 : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":4},{"id":"156be6857760fefff612310f792cf20f249dd9aa","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-19 09:03:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"272c9f3c1aad3186b84be1a8099625085aae4f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 11:28:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efe48886d9974aa19c5b85f9189d94b6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d622e4d91da549ddb6bf5669fd74e2ff : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0cbd64a454a4f6ab2ac282f2b202934 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb199060ced944d09c30902f7ed519cf : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/975813159d804b2a8ba5c73422880e2c : SUCCESS in 10m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17451c2c0fe04d738e816dd232555967 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d72d3188bd5c4fe18a67dbb9c2075a9d : SUCCESS in 13m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a66a1498f464896847e252e51278609 : SUCCESS in 9m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0a9d949455e74e68b701fd4239b5ec0f : SUCCESS in 8m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/56542671b17f4c38a635deda981d5a08 : SUCCESS in 7m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78ba4d54a0a34a92bdee62a348ed4d0c : SUCCESS in 10m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2b1397bdf27d44e2a539323d5ea31dea : SUCCESS in 11m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/183c5a92d38e4910b2d905517855efab : SUCCESS in 17m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fe6e379033d40c080b030d649b536ec : SUCCESS in 14m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5742b9d548e0436693ba27c57ca57177 : SUCCESS in 46m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8e41a66fb1af463193729c2e68f3e58d : SUCCESS in 38m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ef655d33060f442197ec17da94210074 : SUCCESS in 31m 04s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/186fcc724e3b4e65a2ae732e1c659c64 : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c0e36e9e3d014899b80a23c6363c7e2d : SUCCESS in 38m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0f35ea66e94a4d41b4343e002aa0dfb4 : SUCCESS in 1h 15m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b6225ac96174345b1e086a78feaa7a2 : SUCCESS in 57m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/158af4da3ba5458f958d17bdc7c46906 : SUCCESS in 31m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cea82e7c9d994ef395d8fe51d4bd9afa : SUCCESS in 1h 14m 16s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/166478385536459ebbf5082c46680fcb : SUCCESS in 2h 13m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9549a50bfe794f6c8bb7a506a7b87f73 : SUCCESS in 2h 18m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27cedcf863d04d84aa578fc7d8ab1fde : SUCCESS in 1h 58m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fd14b89eb24d4e4f985fc8abb62a6420 : SUCCESS in 22m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b184bfb9f296470b98d26389bd700188 : SUCCESS in 21m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/333ffe62d97b4af58f616280dc0642d1 : SUCCESS in 9m 33s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0f891cfff6f1416c9a96d9d972a16218 : SUCCESS in 16m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e0812ef7d72421298e1edf545b373e7 : SUCCESS in 9m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8eef880e9f444fc78367ad5235f74338 : SUCCESS in 14m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6df57c18ece5460c8104ef6f11f0ff8d : SUCCESS in 22m 09s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/87b0602c1890453ca5c7e846b1f1ba7c : SUCCESS in 44m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b547620449b4cc19b10f04c4a79cc85 : SUCCESS in 21m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/543a4759edf4465192a5c1a55dbce081 : SUCCESS in 8m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/147c4c74a2a74eb38b0ba36bccbd7102 : SUCCESS in 7m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17c272301faf47f9b62ee2f69df203b8 : SUCCESS in 15m 52s","accounts_in_message":[],"_revision_number":4},{"id":"5fab92848d8dd2ce690793b138b4aa0a2f56e404","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 11:15:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"49f272cc8f884fa2c1a9fa49c09567bc2556d27c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 11:15:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"92358e9c1092139c4458d486fbf9365420cb5828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 13:38:51.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb6624b775b042a99b15d23bf0533046\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f084497d6f44eedb59c838c8d5baa26 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f412d6f63c854c1ba53c8fd823b7d21e : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28665ea15ae34793a6985b68fda0ac14 : SUCCESS in 8m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/40bb0d454f8b48e7a78c245a7bccd00c : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc57820b2e254d0ca79d166554408c63 : SUCCESS in 12m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ff60e134624549229037f56dbbd979e2 : SUCCESS in 13m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/63e84f6e0423444db1213eee73808b66 : SUCCESS in 14m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76b65f5a6ee04b7e861390ac54a5dbea : SUCCESS in 12m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca46fd782b29418b812f75b92c0a649d : SUCCESS in 14m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a449a5aea38943348c24ec902e868034 : SUCCESS in 11m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/58e323fee3344950af584bfb4ec19562 : SUCCESS in 18m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d027d737b80f4259b13fc80f705bb128 : SUCCESS in 14m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ca2d7a5a697f48e39006c6a7cca4ac50 : SUCCESS in 33m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8dd28bd24dfc47babdd7eba3550cfc04 : SUCCESS in 33m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/763f9059666b4820a1d5568f9e255448 : SUCCESS in 34m 36s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a63323ebc04f4d86b94e9a18751e5507 : SUCCESS in 34m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/468506be42e843079a6402bbecd14b94 : SUCCESS in 56m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/053b60ca916747de9b872ce5bd894ece : SUCCESS in 1h 22m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ac83f84b972426ca6779ea5b42e3feb : FAILURE in 50m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/217f972f8409488e8e721c72883a7f3a : SUCCESS in 30m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5e33a0abba6c49e580ceb7de47170336 : SUCCESS in 1h 18m 36s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6d18aa2f0eb746e3884a36dcc5e30af6 : SUCCESS in 1h 21m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e667f850908f43cda6e5369b2ef1d105 : SUCCESS in 2h 07m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/caa9dabf85fd42b9a8ca33ace481a5b1 : SUCCESS in 2h 17m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9684c0e6931345d1888d9297a6683fe6 : SUCCESS in 22m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/990b525274874314bb3ab9b3b1a6dfa2 : SUCCESS in 16m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f6608f787be346208c13acfecc28f3f6 : SUCCESS in 10m 08s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5b5a5f60085e4d028692df25b3f15c97 : SUCCESS in 16m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1d3d903d5044348b16c2f4187f5731d : SUCCESS in 9m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/337306b15fd4469ab6391f6748331d54 : SUCCESS in 15m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/560fb967ae3343cba4809426c5e47fb7 : SUCCESS in 21m 33s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bc42e18f33e74b91b5053b2a4be32323 : SUCCESS in 37m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3c5163c3bcf4874bcbb3294c686b253 : SUCCESS in 31m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6ab8e1049912445db50be3fc5455aa22 : SUCCESS in 13m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/59bcaed3ac1f40198a0808b12458b49d : SUCCESS in 15m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c28073689ff641e0bbd3d613449c7718 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":4},{"id":"86e5b38cdf7cd94e57e7b371fb5ff06cd8a1570d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 15:11:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"574e62c96967c1915bef02f58f5a6f4f10afb46e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:39:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2bad2da6d0841c3adc62b94540bb4d6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21054f4be5024e1f8ea504fd3ff7b19e : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7cfad728884d719a3090cf3ae70820 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0678348f6c7b44b1b3dfa3f8030c75a9 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a43be6a04e49338e91c829bc78b5c7 : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b92c08a0a0843a1b4dfdf14a30735cf : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a3dc96e8d4c435db81a03b3f12119c9 : SUCCESS in 7m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/706365f0ddf64d54aa9edf80d6889548 : SUCCESS in 10m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/56c1dc70a41a45e499f261db0a644cfd : SUCCESS in 11m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9f6280a8e6d742d9a93ca13a34060e21 : SUCCESS in 16m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ccfe5dc5a4b4392ab48eb88821eb1eb : SUCCESS in 14m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/74b5b5baee4240acb88db4ba917e2634 : SUCCESS in 23m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/69b0af71217c4a1e90c031954f511b08 : SUCCESS in 19m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca754677dde34db0a75e53e1ce56ca61 : SUCCESS in 8m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9b11450a22d24127bca927dd72bffabd : SUCCESS in 1h 09m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d5e2c421e91c4d798b740ecab9d30fd1 : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2cb5ad78ff9d4fa88874c595e23025bb : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/369bea15e1c1476c9919fa88667f4357 : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/62a6a60f64f542f5b34a2bd50854c982 : SUCCESS in 1h 14m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c1b761b4579e4c3296cc5d25b3b33337 : SUCCESS in 1h 16m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad494041d00d40c58d73f09b8496b24d : SUCCESS in 54m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a18d7fcc82fa4bdc9a02afdac04e7da3 : SUCCESS in 49m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/47599b2c30724cdfbd56a964969e0e20 : SUCCESS in 2h 20m 18s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b5084d4ca6a146a68072ae513e0bc26d : SUCCESS in 2h 18m 36s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0c5326c93ca42c6b9d7211871b5faa2 : SUCCESS in 1h 54m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/258ebf691ea54f8f94f4537f3b143b33 : SUCCESS in 2h 09m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c8a40d7f93c040b38b1599c47a2a9aab : SUCCESS in 22m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/14399f0b8f3545eebee22788dc189808 : SUCCESS in 23m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1b09519f52e44b97af3b31b3a4edc024 : SUCCESS in 18m 50s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/605ecf23ea604010a1ac2c8a26e8caa2 : SUCCESS in 18m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be17c143821545b9af323344daae0696 : SUCCESS in 15m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2af505309ed34236b279f9e6af25ce66 : SUCCESS in 15m 54s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/594e31a2ad204552a21fb31cef44142a : SUCCESS in 36m 25s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6473492fc61c43df9f55c0c653535674 : SUCCESS in 41m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60dad7d07eb14a38b801275be81b69c8 : SUCCESS in 34m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1e6286a19f82482ab9ae0ae4892291bc : SUCCESS in 18m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b1f9d6be3962427cb3c76615815113d8 : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/643d1e87145b4e40b4af43ee2933bbf8 : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":4},{"id":"66abd494f833b1a82ca529498b10756101488241","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:39:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5bf0e66a88636219e811647c3a9279702e06fbd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 20:14:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43b7e614d8544e478faa879bb83495f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a80b69227cdd4dc18d0968e6c00e2f29 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a11cc980a5f4d9bb2163620bc070d11 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5cca17a29cc413d8b87245860908eb1 : SUCCESS in 9m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f63860f4baf4e72841392758fdd2bd9 : SUCCESS in 2m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0816f17aaa054983a92dac96e3f84e73 : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/06b44c833d1e4948b80dc5a77a098c53 : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7725df66094f4ba2bda137f26919d30f : SUCCESS in 8m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1fc941c7826c4c6bbcbe06ced36fd02c : SUCCESS in 18m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d1fc5ca86d946ee90fda571375eb589 : SUCCESS in 12m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b713b9b8480c4b2e965202fb9e127fb0 : SUCCESS in 18m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/07ded9841e764cbcbc415a2a4f513125 : SUCCESS in 23m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4aed7cb981d40678b1170a5f54568f3 : SUCCESS in 8m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dbb454383a5142159753e5cfd97f0266 : SUCCESS in 31m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c7de2da727e14dcf856c204e1ca53143 : SUCCESS in 32m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f5cc154aa1364e908474aed4dbe613bf : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/48cdf65e56474f6c93e89fe9ae11f072 : SUCCESS in 51m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f030007b030e4cd2bd19a8ddcdf7f79a : SUCCESS in 35m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1e69cf0cc31f4957bf6649c2effe47da : SUCCESS in 1h 16m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/767d00c8cd6a490ca38c2d0f0f3a99e6 : SUCCESS in 1h 17m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59fe416a3a0c440ebc2ee9c75933bbae : SUCCESS in 56m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f20178d36843453884f369743804aaa3 : SUCCESS in 2h 25m 28s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5010e701a67941ad874c8e10a0cc4482 : SUCCESS in 2h 26m 31s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35b141f1c5564f9cba611c50d109d8ea : SUCCESS in 2h 08m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0286ee08dd3745499bf389e917541356 : SUCCESS in 1h 49m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/44621d7246c749468d606c0fef78f0e4 : SUCCESS in 17m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/99f1a18fbef040bbb5fe35e4ed6c50ea : SUCCESS in 15m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f3b32f1959a246be927eaac88386b2e2 : SUCCESS in 18m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5197d530ba414621858d83e2ff336b8b : SUCCESS in 18m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38cbf0d0b4e949c8ad238aa499dce44c : SUCCESS in 15m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6bf803c749ff47f0ba4c79f0ed953a12 : SUCCESS in 13m 49s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9681c829ffef45508de512250d6b17b3 : SUCCESS in 22m 00s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5b40d46136b54f64a732dcd25d9fc4e3 : SUCCESS in 37m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bd44feb8f1d4b4a99e6ff67a588cd9e : SUCCESS in 32m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/078f2db3f1064bdd8718fefce7fdc2bd : SUCCESS in 8m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fe99aaae019446e98523bcd07a5b8d0e : SUCCESS in 8m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b416372b2f354acda76e8f2699e09233 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":4},{"id":"ed40262c1a9ab3e71fe8bb548d151b133a6a9b96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 20:14:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a7b42e367e2765ad7b79e6c72c1d4fe27ec9ff26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 20:15:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc672f2eae3e437e80027a6d87178e9e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0947c1e2c882443593b7571be9cd376e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6bda8caca53b62b79cf2b38f939cce4898e5badd","revisions":{"08f007647990892742e3fab65c3766a81e786dc2":{"kind":"REWORK","_number":1,"created":"2026-03-12 10:27:21.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/41/980241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/980241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/980241/1"}}},"commit":{"parents":[{"commit":"5dd2eb9bf99e2e420853ea4ed27f6f5c3fd58b25","subject":"[2025.1] Pin setuptools in tox docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5dd2eb9bf99e2e420853ea4ed27f6f5c3fd58b25"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-12 10:21:03.000000000","tz":0},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-12 10:21:03.000000000","tz":0},"subject":"Add Valkey support.","message":"Add Valkey support.\n\nThis change is a partial backport of a846cccc63 to\nvalkey support for 2025.1\n\nCo-Author: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nChange-Id: Ied90077ba7104c67ea2623b4de16d56d9f1c1c02\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08f007647990892742e3fab65c3766a81e786dc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08f007647990892742e3fab65c3766a81e786dc2"}]},"branch":"refs/heads/stable/2025.1"},"9829788a189d598f455dbaeb1a9a13ed7ec2d40a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-12 11:08:32.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/41/980241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/980241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/980241/2"}}},"commit":{"parents":[{"commit":"5dd2eb9bf99e2e420853ea4ed27f6f5c3fd58b25","subject":"[2025.1] Pin setuptools in tox docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5dd2eb9bf99e2e420853ea4ed27f6f5c3fd58b25"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-12 10:21:03.000000000","tz":0},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-12 11:08:32.000000000","tz":0},"subject":"Add Valkey support.","message":"Add Valkey support.\n\nThis change is a partial backport of a846cccc63 to\nvalkey support for 2025.1\n\nDepends-On: I613d471e5f1e6d563d3853761d8432cba60f3c4c\nCo-Author: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nChange-Id: Ied90077ba7104c67ea2623b4de16d56d9f1c1c02\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9829788a189d598f455dbaeb1a9a13ed7ec2d40a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9829788a189d598f455dbaeb1a9a13ed7ec2d40a"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"e114d257dd531aad9c282a6501f5b66837b6c21f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-13 08:51:47.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/41/980241/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/980241/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/980241/3"}}},"commit":{"parents":[{"commit":"5dd2eb9bf99e2e420853ea4ed27f6f5c3fd58b25","subject":"[2025.1] Pin setuptools in tox docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5dd2eb9bf99e2e420853ea4ed27f6f5c3fd58b25"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-12 10:21:03.000000000","tz":0},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-13 08:51:47.000000000","tz":0},"subject":"Add Valkey support","message":"Add Valkey support\n\nThis change is a partial backport of a846cccc63 to\nvalkey support for 2025.1\n\nDepends-On: I613d471e5f1e6d563d3853761d8432cba60f3c4c\nCo-Author: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nChange-Id: Ied90077ba7104c67ea2623b4de16d56d9f1c1c02\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e114d257dd531aad9c282a6501f5b66837b6c21f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e114d257dd531aad9c282a6501f5b66837b6c21f"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"6bda8caca53b62b79cf2b38f939cce4898e5badd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-03-13 09:30:01.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/41/980241/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/980241/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/980241/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/980241/4"}}},"commit":{"parents":[{"commit":"5dd2eb9bf99e2e420853ea4ed27f6f5c3fd58b25","subject":"[2025.1] Pin setuptools in tox docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5dd2eb9bf99e2e420853ea4ed27f6f5c3fd58b25"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-12 10:21:03.000000000","tz":0},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-13 09:30:01.000000000","tz":0},"subject":"Add Valkey support","message":"Add Valkey support\n\nThis change is a partial backport of a846cccc63 to\nadd valkey support for 2025.1.\n\nDepends-On: I613d471e5f1e6d563d3853761d8432cba60f3c4c\nChange-Id: Ied90077ba7104c67ea2623b4de16d56d9f1c1c02\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6bda8caca53b62b79cf2b38f939cce4898e5badd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6bda8caca53b62b79cf2b38f939cce4898e5badd"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
