)]}'
{"id":"openstack%2Fkayobe~981105","triplet_id":"openstack%2Fkayobe~master~I8fbc2f4203ad006b8f347a02e8ff1ec99b538c9f","project":"openstack/kayobe","branch":"master","topic":"ipa_url_patch","attention_set":{},"removed_from_attention_set":{"38330":{"account":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"last_update":"2026-03-19 08:58:07.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-19 08:58:07.000000000","reason":"Change was abandoned"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-03-18 13:40:18.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I8fbc2f4203ad006b8f347a02e8ff1ec99b538c9f","subject":"Add new subdirectory to upstream IPA URL","status":"ABANDONED","created":"2026-03-18 13:08:24.000000000","updated":"2026-03-19 08:58:07.000000000","total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"b1866a85a49e81c6da35a5643aeac7a0174b1013","_number":981105,"virtual_id_number":981105,"owner":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-18 14:39:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"date":"2026-03-18 15:51:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":-1,"date":"2026-03-18 14:56:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2026-03-18 14:55:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2026-03-18 15:51:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 13:39:24.000000000","updated_by":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-03-18 14:03:11.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-03-18 14:38:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-18 14:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-18 14:44:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"6658765e0084f52ad1f0359d1cad982c38f98474","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-18 13:08:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b52e9dd688819d2e8ab1a82f75d78699f746b82d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-18 13:38:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"830c5f012c2991cff227e0c9feb45cb11a05fbb8","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-18 13:40:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ab183e659a27170cee7b186744bc7c06e6e8217","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 14:03:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c67c851f9fb834bda40f4ab9fefd466d90c0500c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-18 14:38:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6ab59edf875285682620c570137ad6054ccdfc86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:39:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e100e34b6b664d1b86f3dd6585e7c1d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d74b43c14d2745f285d05940899374d8 : SUCCESS in 8m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8c9fd44de4b74051a2375992b679b9d0 : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac7197a01bd7403bafc943307be5dc59 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90c9639c4fe24e4c9f2a4942d8403188 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b60e3de89004ce58eaaaf583c2dae58 : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c5f80a9a0f4b40b094b03ce266907903 : SUCCESS in 7m 02s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef893007ea90433eb6ede2484c229896 : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/531bc4bceb69434aa201582bdddfc23e : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b2e4604ad084a35a7598c8f724e04c3 : SUCCESS in 11m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3ae2f668326245638fd256a084ac7f2a : SUCCESS in 10m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30a3bc13becd4ef0bc6b07f5591937c5 : SUCCESS in 8m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/af209f9d0d764ee89b7618090376e635 : SUCCESS in 27m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64900fad01d64c0bb618aa23761812d4 : SUCCESS in 9m 42s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5a390949b06c43f5917fdb1445623530 : FAILURE in 19m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7bbfe5f28e1e42df97575ab9521fca62 : FAILURE in 17m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d7aa44d0ae664e3ea21fbb018c40abbc : FAILURE in 22m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3a802a10db6429eb054187d4b7ead5c : FAILURE in 17m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/338eb97790e742399a9aa0d03a6b45e4 : FAILURE in 15m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/537229615b234d1fb37745732c5a2aae : FAILURE in 56m 15s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b2b7d382d2894323b71181ffaea627f5 : FAILURE in 56m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9d02a1eeeb74cf0b37b3558e5f305af : FAILURE in 50m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbc22d0de488480691aefc3e5cd3668a : FAILURE in 54m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0248677aa89a4672acce7eed3aef4012 : SUCCESS in 18m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/16dfe8771d8140f68d22f0ceb95a37cb : SUCCESS in 12m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8fbde9c87b44a7eb9353b5f1480063d : SUCCESS in 17m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7bf7fce80ba545668ae5d929bacfe14c : SUCCESS in 14m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bab5e0c245e245fda261c79dacd7117f : SUCCESS in 17m 21s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9bfa6210045f4af5b8b9841f965e5a64 : SUCCESS in 17m 37s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4b22f8d416b4c16ad976e5751ae8b51 : SUCCESS in 28m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/170aaa041d0949ae8ee64466d4f88967 : SUCCESS in 27m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/60af2a45d1ba49de9b7e21635acd30ef : SUCCESS in 14m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8f850647e734275bc465f579ecdefb1 : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":2},{"id":"46195618bc9e1ac899db72077400563c86dfb3ed","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-18 14:44:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0564b50f024d39cbecf29e95de740b241af4dc73","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-18 14:47:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7abb470bc7080767c41730d34b8adcba3981919","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-18 14:55:19.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbe3691e80f73720f3bea983ea9d88235fa216a9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-18 14:56:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0300f8406868212f5489d816056b513b00314c0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-18 14:59:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d433e379c1cbad38c1982a5f4f5b898a63a777b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 15:01:26.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"ebdd51d2e07a75e7bae2496fd8fe58f0b7757526","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 15:06:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf36606a7d6d94a6f9f9d1ee813e5b4054526e07","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 15:11:55.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b032744ab3d9b89b2cc6c9f29e1e6c74e72528f1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-18 15:51:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86c9f12880d71ea869b1445ee825ec736f8b0c29","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:26:50.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26732cff69be42ac96fea5056f9d5891\n\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/761ca1dd163d4b40ae1dc23029de69e0 : SUCCESS in 8m 20s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/9a825b8381a147ed8c1f598fe7efca83 : SUCCESS in 11m 20s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/442d2629a2e64225ad6a1db4285ff8e9 : SUCCESS in 12m 39s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/68310aa27d644efd8d5f12b2b7fec794 : SUCCESS in 18m 59s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/95d2ef8e2e144359918b92adcd0ab470 : FAILURE in 14m 41s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/b21ecf76b8e34f338a34cbe89a96139f : SUCCESS in 11m 16s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/d0c2bba68318432cbc95a549e6cb4dc4 : FAILURE in 19m 12s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/afb51c6711934bf1ab93be128a20bd53 : SUCCESS in 9m 26s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/ef7e498e70724465afd47e12f5ede2fa : SUCCESS in 28m 30s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/0eb6082dd51241da98c02f6692ed01da : SUCCESS in 30m 50s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/86e71c9b48bb45d4a5aba6bef08a8aa4 : FAILURE in 1h 12m 31s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/694de5d4e82d44e2bcfea581f3edd9c8 : SUCCESS in 36m 55s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d2e71a1fdc444f919b84aba2f6f9743b : SUCCESS in 1h 14m 11s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/a9f59bb5546645c392663ea02482924e : SUCCESS in 17m 05s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/eb34d6517a0648f8b1073e93f42e2df8 : SUCCESS in 12m 23s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/b84e9efb6c6b4d7ca480ce4dd971c451 : SUCCESS in 18m 28s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/2b95c1524b7d448aa750d5a5e44ac58f : SUCCESS in 22m 30s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/2f1f21ec0f544419a5088eaea35c0514 : SUCCESS in 9m 15s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/4c032562024f4b7a8cbbfe04cd8c45af : SUCCESS in 20m 39s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/8e53a8bf2d6a45e5931570798c99e9de : POST_FAILURE in 15m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b1866a85a49e81c6da35a5643aeac7a0174b1013","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"date":"2026-03-19 08:58:07.000000000","message":"Abandoned\n\nRedirects for URL working","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0d70518268bfb3fc2af00b9eceac2c44ddde6d08","revisions":{"d570d114a64f56b2a9a82f4ce5db5e128c0ce25a":{"kind":"REWORK","_number":1,"created":"2026-03-18 13:08:24.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/05/981105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/981105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/981105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/981105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/981105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/981105/1"}}},"commit":{"parents":[{"commit":"cc33e0e2a38ce4030c82a08f8aa926fe8eb4886c","subject":"Merge \"[firewalld] Fix broken conditional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc33e0e2a38ce4030c82a08f8aa926fe8eb4886c"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-18 13:05:31.000000000","tz":0},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-18 13:07:23.000000000","tz":0},"subject":"Add new subdirectory to upstream IPA URL","message":"Add new subdirectory to upstream IPA URL\n\nIronic python agent upstream URL returns request\nfailed message due to new openstack subdirectory\nbeing added.\n\nChange-Id: I8fbc2f4203ad006b8f347a02e8ff1ec99b538c9f\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d570d114a64f56b2a9a82f4ce5db5e128c0ce25a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d570d114a64f56b2a9a82f4ce5db5e128c0ce25a"}]},"branch":"refs/heads/master"},"0d70518268bfb3fc2af00b9eceac2c44ddde6d08":{"kind":"REWORK","_number":2,"created":"2026-03-18 13:38:56.000000000","uploader":{"_account_id":38330,"name":"Claudia Watson","email":"claudia@stackhpc.com","username":"claudia-lola"},"ref":"refs/changes/05/981105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/981105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/981105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/981105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/981105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/981105/2"}}},"commit":{"parents":[{"commit":"cc33e0e2a38ce4030c82a08f8aa926fe8eb4886c","subject":"Merge \"[firewalld] Fix broken conditional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc33e0e2a38ce4030c82a08f8aa926fe8eb4886c"}]}],"author":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-18 13:05:31.000000000","tz":0},"committer":{"name":"Claudia Watson","email":"claudia@stackhpc.com","date":"2026-03-18 13:38:50.000000000","tz":0},"subject":"Add new subdirectory to upstream IPA URL","message":"Add new subdirectory to upstream IPA URL\n\nIronic python agent upstream URL returns request\nfailed message due to new openstack subdirectory\nbeing added.\n\nChange-Id: I8fbc2f4203ad006b8f347a02e8ff1ec99b538c9f\nSigned-off-by: Claudia Watson \u003cclaudia@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d70518268bfb3fc2af00b9eceac2c44ddde6d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d70518268bfb3fc2af00b9eceac2c44ddde6d08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
