)]}'
{"id":"openstack%2Fkayobe~982396","triplet_id":"openstack%2Fkayobe~master~Iad7fc73f16ffcbc2ab5f261322bd6777036fd5f6","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-21 09:14:15.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-14 18:56:30.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-10 08:53:35.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Iad7fc73f16ffcbc2ab5f261322bd6777036fd5f6","subject":"Block until timedatectl status show synchronized","status":"MERGED","created":"2026-03-27 11:45:12.000000000","updated":"2026-04-21 15:10:50.000000000","submitted":"2026-04-21 15:08:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982396","meta_rev_id":"9fb39bae94590827e435dda03c8fb6fd70eb488c","_number":982396,"virtual_id_number":982396,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 15:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-14 18:56:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-04-15 10:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-04-15 12:42:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-10 08:55:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-04-15 10:15:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-14 10:42:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 13:23:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-07 21:07:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-04-07 21:15:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-15 10:15:43.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"36a12e9b4290f9c76205b68ba8b34f620df9b4e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-27 11:45:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5e962d19ba3a034ede922ac0fce1777ce0ad1a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 13:23:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1efce5427294d49baabd6ed0f499ed2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2864fdd5d4d348d8b0dffe49181f5c03 : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ff2d3971209947bd97ff3dc676036610 : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dc5ff5d7d4d4a058e8ad664e90e3c43 : SUCCESS in 2m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2890ea95d7de43e0b0be62f0641992b8 : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5c6873038fa420591201fe5aef1f8cf : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/952b833310e145908fc19bf0c02ce99e : SUCCESS in 6m 36s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/62cb3e22e59f4cc0bc9d5e02f0ab61d3 : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5195d77c41b4073adaafd4acf8ec06b : SUCCESS in 12m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/61fb3027f8bd44c3b517c1377675d18b : SUCCESS in 10m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4aff99d7998f49d99ab7df8400f38906 : SUCCESS in 13m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54d2883e632844bb8da3a10117ad089f : SUCCESS in 16m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/928f41a0d7ff4783b98fe6876f510c78 : SUCCESS in 13m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8502044b5e384ad79bb5f8db518e59df : SUCCESS in 16m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cd642a212acc423189384e740e0661de : SUCCESS in 58m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/934e1a3f5d8049ce92e6e318856aaa10 : SUCCESS in 57m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ebd3296d4ca44e88b90080b57c637820 : SUCCESS in 1h 02m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61cff93f948d4cbd8d0b0795d7409d03 : SUCCESS in 57m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ce5a0a6871124e1d89e3ba7449a2414e : SUCCESS in 56m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1858e0b98a6d43a29a5ca2ef54999915 : SUCCESS in 1h 37m 03s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9ec1b584eb1d4e33aa09d43910447c0d : FAILURE in 1h 03m 23s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60d13e378a1b40fca7bce1dcb71259c0 : FAILURE in 1h 00m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14dd819ae8d3485c8518728c843ea5f1 : SUCCESS in 1h 32m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/abf2c5f60a544d54b30ff26587d00b24 : SUCCESS in 16m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/036492b481d04118a20352849f6225ca : SUCCESS in 24m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0a744dd17a04863897608bea41290eb : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4a64795421e848688fc729d7417a7cb7 : SUCCESS in 17m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/12f8a095f4f4478a83be2a4ed1162036 : SUCCESS in 15m 25s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/260d66795a564f03b7e75acd309d168b : SUCCESS in 28m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bc44d22a37f47da9f6beb022d5175f1 : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df97d7bcdffd48b1bb7fe60051d2d32a : SUCCESS in 31m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5544febd8e5a4b49bf72e8532df65e26 : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2641b9b1aa54c2289713ab882addc1f : SUCCESS in 15m 05s","accounts_in_message":[],"_revision_number":1},{"id":"db538e839861c5f9274de389d2214d06e9aaf472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-03-27 15:08:18.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ad5a72c9e09b8be7ecca12993328eba0cc85e3ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 16:45:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/797c4bb5a7b24c6ba6b5ef317ba9131e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e828dc1381946208224efd55876461e : SUCCESS in 4m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/39f3328557fd4ccbb782c1728887c0d8 : SUCCESS in 7m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9dbc92d05224ab881d15adf38707bb4 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53b324c620244d4eaae50c86022d006a : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71e81fa16a2e4da8aab74a6872d1e8f6 : SUCCESS in 2m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a392fa00c6d44c179c8690ad04b6370b : SUCCESS in 6m 52s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/43c73beea2f04b33b308d1635203fb47 : SUCCESS in 3m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dce6e89bd97b40cbb096d3015fd221ab : SUCCESS in 6m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3fb7123efcf24f15a19d7503b3ae79e6 : SUCCESS in 11m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/08323d6560ab4f0da1cc9e2e051f0c2d : SUCCESS in 33m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/175c3ee8244e48b1b8a1832e966acdf7 : SUCCESS in 7m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/137fccb9c04e4bd3b4250b9c864159bc : SUCCESS in 13m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a17eac45701b43b58e00e78d239d0b9f : SUCCESS in 17m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8997e7b6422f40279300e76f369668b5 : SUCCESS in 55m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5682feecf6ad473e9c894c2eab4ccbc2 : SUCCESS in 58m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/91704be6fa29415baea8a4bfbd2fc798 : SUCCESS in 1h 00m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f282f2297fa64131b3a7f84fa6925967 : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/89ba264e4c3a4c3996de0e7bd7ca2d61 : SUCCESS in 1h 01m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fa059cd1e3dd46bb85c83e563a3f600f : SUCCESS in 1h 35m 14s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e090733a108f4e73b587dcba373b4e08 : FAILURE in 17m 11s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40cdcf8a4e894af3a86523b5b44433d9 : SUCCESS in 1h 28m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6a526c552054e2c95b67043e28e0f27 : SUCCESS in 1h 36m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2103c7f19e3740748936181ba4173e47 : SUCCESS in 11m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3da5d4a7d3704640ad7650bbd0b730b3 : SUCCESS in 39m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43f000011a134e81b91bc039c9f59716 : SUCCESS in 9m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2a0654923b994fca8ff96e7e98bbf258 : SUCCESS in 9m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ccda4118624e4dca91bc0799bf330652 : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c451cae92fa340c98bb9be17b1516c37 : SUCCESS in 25m 44s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/820f396e0bfa44ed8546fcca0a8d56a0 : SUCCESS in 16m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/482760cabf0a4f57982a1315c522c318 : FAILURE in 9m 00s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4d40bd824ec14ae49d599253e7fab6fe : SUCCESS in 9m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8522fa961acb4ef58e23fc0f85a4630b : SUCCESS in 12m 34s","accounts_in_message":[],"_revision_number":2},{"id":"93a89eb8b362baf13371efd1ec52664e4b461c5e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 14:38:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"982b05980bdc71864c2e0852c9ff9802721b91b1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-07 21:07:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d5bbf1813b0bd640272faaea65db5be28f0b1f4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-07 21:15:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93ef3f274b352bc5970eee6fa8b3098b10704a2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 13:48:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4c0bac94daf5ba2780f2742b0fc59198ee47bc77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 14:57:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9229267b0b834d69b662fa627af77bdd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ea11096de24830a6a62e3d04407e44 : SUCCESS in 5m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a49d3552d9004a688af5170df913054d : FAILURE in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d907bf2fde4f43199979dd6392834868 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19753ccdda6e4fd4b64d531a0fb927af : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2891ef3064b842b5b67c8caf456ccbc9 : SUCCESS in 2m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a710f3ea83c34903b2117b01630e6cdb : SUCCESS in 7m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c36daeba8e544d62a5151ab1f2c21571 : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/44419906d5cc4da8b3a6452d2873a985 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f6ceea293964f489311b5d22e757797 : SUCCESS in 13m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e51c472dcc4e402c94ab37925ee236c1 : SUCCESS in 9m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/509a5f359e0a4917bd83335acd547c39 : SUCCESS in 15m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d66490f022841bb96017c59a25834d8 : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/482b28b4348b437992396d271e93a2aa : SUCCESS in 10m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8f276d6e5ea4b4c9e10201eb028d0cd : SUCCESS in 14m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c02d967cf11f433aa9adf3e1b93d06e8 : SUCCESS in 58m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/30bf007f71bb4320b279e63bb8430066 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/170e8fb11fb54f2594ec74034d3a8f7b : SUCCESS in 1h 07m 08s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ba5736e70c244499651bfe9c6f7b58c : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1e234999b3d04c6d886ab0dc6c2624c0 : SUCCESS in 1h 03m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/33ad63e50b6f43b7abb0b050c4f73da3 : FAILURE in 10m 51s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4505faed0e2b495c983bc9b98b571f18 : FAILURE in 9m 38s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3023d51fd5ec4c6f9c0fdaf8357af5e6 : FAILURE in 12m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26a28f380b8442ca8b3949a176776802 : FAILURE in 14m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bcab6d1a55364cc8800b3289e28a7e64 : FAILURE in 20m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9a2ab898a1884479978571451a4347be : FAILURE in 20m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00bcfdfc3d1348ba95c9e50624ee2b0e : FAILURE in 12m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d5e8137dd0e14f2c8aac2cf205086bc6 : FAILURE in 8m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d2576c4a57b64f8695afe87427fefa8c : FAILURE in 11m 49s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/521bcea5214447f78944dc5bb9d71880 : FAILURE in 8m 30s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6aeb5a04a4af46a7b8eba458304e633c : FAILURE in 9m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f03abdae90d74eb49d31b77b8cdce3be : FAILURE in 12m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/13e5f8089440468a99107f9305d6da7d : SUCCESS in 10m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8849304c9fd41ddb51b9be156b6d2b7 : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":3},{"id":"8952d4292e8bed6840e0d359cc55cc1204d43035","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-10 08:53:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba5540caee1a6389da0060a438a5b01a45080998","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-10 08:55:06.000000000","message":"Patch Set 3: Backport-Candidate+2","accounts_in_message":[],"_revision_number":3},{"id":"8461e12b19216fa5cc40d407ab88abb59b317635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 10:39:54.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b43e27e29f174199abe7e3acd1a8161a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c12eee11887b47538e9f1356cef531d2 : SUCCESS in 8m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/43ac0cdb91f9497284489120c3aa5ae6 : SUCCESS in 8m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/493cce10ef6b457683abf33a7b4438a5 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6826b263687f451c8bee7f8c6b5afbb6 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08b63ab238024a6598d7fd4b1c3588af : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dfbffc4744534e01bebc59da7c7a2cac : SUCCESS in 7m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e47a24f65c4495a8de3053954a2b54a : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/699556c8cdf44341a194f02c5060f27d : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b49d4f68a9c94f06b388a8fa05bbd251 : SUCCESS in 12m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e4b86531c66b4bae8c3d1f2f2349908c : SUCCESS in 7m 01s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fe23f9ce7a7848cba82b7022a12c166b : SUCCESS in 14m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30f9f56e0732437c9b1800a60d89a326 : SUCCESS in 15m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ab091b55a53a45ea8a1efd93834a346b : SUCCESS in 24m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cff22b86930743f080c643b19fecd9aa : SUCCESS in 17m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9bcd16f8c3d54066b1a326e41b88bf6a : FAILURE in 46m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2dcd06478c054ec3a6797a3f69737b33 : FAILURE in 53m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e9f6c948c5204803927548aa320eb424 : SUCCESS in 58m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89d74a810837447cbedcf2bb25a6ee1c : FAILURE in 53m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/933601abf46f4c10977b3cc7ff7f34cc : FAILURE in 42m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/46afc7600fd34031baa9885edcb7752a : FAILURE in 4m 16s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f5c2c24794ea4f15901911f58b9c731f : SUCCESS in 1h 28m 24s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ea3ee8c0bc14f27aa25fa8859c57bfe : SUCCESS in 1h 31m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1f178088079462892a8e76adba206d3 : SUCCESS in 1h 43m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/05a1f2608df642d19a275cd3c45aa60f : SUCCESS in 10m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3728d60e2660411fba910c5b1d812918 : SUCCESS in 20m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/719f80effe0d46c2b9837e08e1468a02 : SUCCESS in 18m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9b819b80b36a421d8e0b77c0ae003783 : SUCCESS in 19m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/258b0376352d4fe3b8754fbcde926af7 : SUCCESS in 33m 20s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/54e20a15d55b425e987f45a88002d041 : SUCCESS in 31m 28s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6fe4c98870d47b28fd00906b18b0178 : SUCCESS in 29m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/045316980a5047c0990f07ec66c1d099 : SUCCESS in 32m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/545086dbecca42a586aae5e35f1f4951 : SUCCESS in 10m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18f2bf0b65cb4af887cc83145cbfe1b9 : SUCCESS in 19m 05s","accounts_in_message":[],"_revision_number":3},{"id":"5159ea7ed5188c8295efcc23e60f25fcd0b87e40","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-14 10:42:38.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"662f0c2b885bdd2175abc0ab29dc4b187d2bdd2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 12:27:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b17b70e5672c435aa6c21b13151278c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d577e4be614279b12aa4859bf9b4ff : SUCCESS in 9m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7c8a907e638c4b2c9175ebe2290411d1 : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b059f30bfac4576b348a0c08f27dd2c : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55ce23eb92e4faf86bfd75c25baf53e : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8289fb6f35b0469c8f106e7b8a753baf : SUCCESS in 4m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/823004c0856848dc92c82306c3e6ea21 : SUCCESS in 7m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc52527aac044adf87845862d9bdf841 : SUCCESS in 2m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa45fa2fb39f411ba594d6f72a3657d4 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/265864c7bd914f01ab145d7d7109d611 : SUCCESS in 6m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b677764c99a24e71a00f2a935a84b934 : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7dc70fab4b474ae0af16d29916346ab2 : SUCCESS in 25m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1b48bf1bcfe4f5da2ed55d2f7cec631 : SUCCESS in 21m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a94dfb0a5ad84c8f8e01b0f74ee9eca0 : SUCCESS in 10m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ebac92be36749ae9601be20bc4546cd : SUCCESS in 16m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a3fb27196f0b4de7917bd0a5a7b4eb74 : SUCCESS in 1h 01m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2669dcd163464805b70a55c52f138235 : SUCCESS in 45m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d2939a5e2a594b19a3010d98a89a0419 : SUCCESS in 45m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6211a2289bc4fc29bf3e8c62da2edd3 : SUCCESS in 35m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/30bed34b8adb422f85048dfca64d5073 : SUCCESS in 41m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/eac8132847fd40d28d9b32c876642352 : SUCCESS in 1h 39m 23s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d9a079a38501474e8cf6844be5d89676 : SUCCESS in 1h 43m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/113c579ee85f4141b8adebee61c8b8bd : SUCCESS in 1h 30m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42ffad9847854253a3b683182c6866be : SUCCESS in 1h 37m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/388282ee3d304ee5b210922957d5dcd2 : SUCCESS in 18m 51s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b418ce61bb624aab8dd56e3aeb3dd7fd : SUCCESS in 26m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e92b562bcd9345168db4d15f40bd9158 : SUCCESS in 8m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/613e6443563a47a4bbcf67bc0d724bd5 : SUCCESS in 17m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c5f6b1b560a748a288afc19f9d8f8645 : SUCCESS in 17m 27s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/45035989c0624ebe89dd90506e3dbe91 : SUCCESS in 17m 42s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f59e2025ce994c22bc376de76a6db375 : SUCCESS in 31m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10e409a363c244ca85bd1e8655d01e6f : SUCCESS in 28m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4a847ab8986b43a4933a57e4a2db7a49 : SUCCESS in 9m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b07a88ee259e4fae93c9b86ebc033bf2 : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":3},{"id":"cd084aef9f1219a16df306aeeb4001d0c48f6719","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-14 18:56:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1c9a4fb737b0f70a23f7d0a73828adb939c3649e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-15 10:15:43.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"30d1ebfb47e2e43e0b94b1c8c0d9f13b9711d6bb","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-15 12:42:17.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cf522f6c69d5ea389ddc419d2eef516cdde1ae79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 12:42:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b8abfd9fb18784b092f44dea02f7ee01fda38b11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 13:48:27.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dd830bc85db4d329ca430762cc9c554\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2131251a43b4f91b13f1893eb822d08 : SUCCESS in 9m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e43683e4db444da98a3dbb1488e43e0f : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e2ec5bf7ffd4fea83461d75d9cc57f7 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd83c385ca774526a79441ea5e201644 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/171dd9a981af4020843f6dc1b4bab588 : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d81f63ffe6b41c9a330b279c248157b : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bbf8a707a4a34f5a848989123c6badd7 : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa44e857cc88401780757b8e1c1e4f0b : SUCCESS in 11m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d257850ed59845fe9894c57b20df754c : SUCCESS in 17m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7814a5603ca44d2293a9c8e22309551d : SUCCESS in 19m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/008743cd314848759cac2bf2160abb0f : SUCCESS in 12m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7381c140b86247bfbfab3251e1c20288 : SUCCESS in 19m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/42f93e6e714b4e7f921e3f70a7fa5f3c : FAILURE in 26m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a64679394ca54efc924ea44326306110 : FAILURE in 25m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/88ac31dc21b544b2a072762ce743b5d7 : FAILURE in 31m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d2efadd14374cddbde0c6ba498cb1e1 : FAILURE in 20m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/10efc120c0774c41bb56d3b9ecc15183 : FAILURE in 24m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/971ce174ed85417ba36f812b2f6ab49c : FAILURE in 1h 04m 57s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/26b2c1a3a8744d6da790cbb707fbfb60 : FAILURE in 1h 04m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53b5b382d5b042738bb30d51d1363fd6 : FAILURE in 1h 01m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08224e6b091b475197e7cce32a422172 : FAILURE in 1h 01m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c485d979e6ed447ab9977d7bf647d00b : SUCCESS in 19m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ec587e03000e45529f7fab041a60de4f : SUCCESS in 15m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44f75b348c9145f2a696fe2cfa9dcc0f : SUCCESS in 23m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/670e32e5f1e542eab4e079dd35a003f1 : SUCCESS in 22m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4a8d562d0f0946e8b0cd5ada74f667fd : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b3f3fed8149e439e89e19ea0e9338445 : SUCCESS in 31m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88e521ac263f4dc8a4c4130a659287e6 : SUCCESS in 33m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78c6f8acb42c462d869e137961f57f11 : SUCCESS in 32m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d5ed281880c246f4b8aef113489d30ee : SUCCESS in 21m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25b1d98c1b524cb08ce85f20a797aaac : SUCCESS in 20m 59s","accounts_in_message":[],"_revision_number":3},{"id":"59e4db7f039156f473837cf08141b9b873c9272f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 15:14:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5da6b9b2417a9935fe5d969354a2a9555ad258a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:05:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97552a49ff1e41c982c9dd3ca5f691ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01add36d64d24bb987cc81ecc3d8bd4c : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/98a8ca87140249e687d7dcee1e83f957 : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5c1d119cced401d853427935a0c895c : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af569d887a0e4c24807f90ad23822c56 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c9c2be30e934a588b806ecd65afc252 : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b4abf79026f14a44a8c8e9cf345cfc0b : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d2ef97ea48b4877a5f43909b0404aa4 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f11f7419afd14561b47029b68623c5f0 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bbd3a1895c4c402098510683188dd660 : SUCCESS in 13m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afcd04de91e24ab5ab9dbc9e6d61a72f : SUCCESS in 16m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2fef6c972f6b4384b6ddd500d762a39b : SUCCESS in 16m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f06520275d14a6ea001b81ef45238fa : SUCCESS in 17m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e38f8bd8cfad4e0cb4cee8d8d6b313c5 : SUCCESS in 14m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d889f30b1164eae8ff77dffae0b5d36 : SUCCESS in 17m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/40af7d84bfbc4ef580d9413eacd0f61b : SUCCESS in 1h 05m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fe7dfc4dcdac4c759adb51c4775e32ed : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/de08a70361d844129ef2758332ca17b8 : SUCCESS in 1h 09m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/597e9ca3b4db417c83f9912fef3eccc0 : SUCCESS in 1h 10m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dc8ab2a18c9d4a44af7282a64f3fea8a : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cceb93202c3044599553ff163ca7c5fc : SUCCESS in 1h 41m 13s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/37ce8a32ae3c4219b6a11c9442ce0cb7 : SUCCESS in 1h 46m 33s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/530a0b3a7ab9424ea80ef0a84fcc444c : SUCCESS in 1h 31m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/805b6e880b0b45b1a4a9094239dc6132 : SUCCESS in 1h 40m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1023680ea7bb4a739fa02bf8a1f54e4b : SUCCESS in 16m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/29123dd5960e42fa8f877844ee27d4d1 : SUCCESS in 15m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b63b8ce88c1c4c6aa20ab545a19789c9 : SUCCESS in 25m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b1c8263ba55d49c2ad5203d0f172342d : SUCCESS in 19m 51s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6283f584ea5f4bebb372ba9dd24db592 : SUCCESS in 36m 16s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ecf89d70112f46729c416593cbebb270 : SUCCESS in 31m 30s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57e33500a61045228e9d7a7b35eda7c8 : SUCCESS in 31m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2effab53669f4ed7901e32d9ebab6c43 : SUCCESS in 28m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5d4d4c74148746e6a39d14cca2d277d2 : SUCCESS in 21m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88ab310e5046474e87356c98ed72c97c : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":3},{"id":"c31145783ca16a08d558c260aa5eb0115e697385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:05:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e7fed61a4fa3f52bde9e3d9421f1322919eb91ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 18:42:52.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73f10bb45c69413dbc256ba0f5076c71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82ccdb8693bc456299d72f503687b731 : SUCCESS in 4m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b9537b77f1c74d08b5504fb14fe88ba7 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8408d63f8ea34d44aead758ed1f3ac77 : SUCCESS in 2m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d355d60a33d4ae5b3c463d30d11434d : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5875d0d2dc9349e58955dc64381cb815 : SUCCESS in 2m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c8edcbd6e3f45cf8c1d033ab2955fd8 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/834c21e77506476d9744baf93f9c2b83 : SUCCESS in 12m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9dcd2ebe3eb04bbbb33241cb31df591c : SUCCESS in 16m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/26a6e65a0a6e4e6ea6c66006eafb6b2f : SUCCESS in 13m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cd691f236004d30a89300850b1f408a : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e66dbe6122b34464980e9add8893fefd : SUCCESS in 16m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/267a34c071eb4bbf885447ba184bad8d : SUCCESS in 19m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f36d548074fe42fdb7593789851c1375 : SUCCESS in 1h 05m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/787d6463ba3546e294e2ef7a6e4d43e7 : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/23fe6d52e06248368e3d4ca4083060d2 : SUCCESS in 1h 05m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1139b918eb0a40b487b88d5fa4bae4ae : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de70b56778254e23a5ad9699e8799e9a : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f727722618884a4cb9a85b25d628b845 : SUCCESS in 1h 36m 36s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/125d40b4b03d4a50b1c2e9de14be23a9 : SUCCESS in 1h 32m 25s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f20d014ff4514e9f92c45cec5048786e : SUCCESS in 1h 31m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dd8bdad643440df963c29073e60cb0d : FAILURE in 16m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/739c94008e1d469fbf573595f585400f : SUCCESS in 19m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0c67974178204f0fbe75cceea1b3fd81 : SUCCESS in 16m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/002a7ac1f8624e7580bd84fab3402afd : SUCCESS in 14m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/58b7732554224949a064e28486436014 : SUCCESS in 21m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cfb15b492f41481dac631d1eb45e3fee : SUCCESS in 16m 39s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/37e551c7512a4bdfa7ce47e3865e3ea7 : SUCCESS in 17m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b11eccf01b2464e8ae5c40e3ff99d3d : SUCCESS in 15m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/438b911e092448d19f883bcf714be17f : SUCCESS in 23m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/26460b1e348541dd843fd567f5543c8f : SUCCESS in 22m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f192d68e08344f8692fca2de58551de0 : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":3},{"id":"fd51b83057d65879834d529ce4b5a870fafea8e8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-17 08:06:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd138c1c232c2f2ac249a7f3a443239ae8b64be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 09:39:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2a44e79764d4856bd5c99febd551135\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e2c29c93e7c45dc9985fb576a38ef06 : SUCCESS in 9m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ea72cdfecbfa4565a9d65701a6e54b63 : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2f2fe0a9f4540f6bccc267f78b7a4f6 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c43f3303d4664b899f17c18930ddc768 : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13768c5f399743d6bf4bb9b6946c7a1a : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae544be0ca6a4bcca249bc3f4a84af95 : SUCCESS in 5m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbabf64abfdc4cbbaf3091f2d51a863d : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b78ee7ffcd9d4b64893884e65285ffa0 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/93e14642d6e74cc98ffd8f75287b21ef : SUCCESS in 12m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/293d662dd8f54116b6f514068fdf409b : FAILURE in 6m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7a6cbe70731047828c77613709e5d2cb : SUCCESS in 14m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82c9d68727c84dddbdd746f42504f100 : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/efa17c621e0444ab83e692ff1529c820 : SUCCESS in 10m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a191a313c94d4d7aa90d024d858aa9cd : SUCCESS in 17m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5acd686541a449b7b54afbde1e50f487 : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3bfe9921676e4d2a98a59901f592804c : SUCCESS in 59m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6c88f00f78654b9f9a780df62adc7ff4 : SUCCESS in 1h 03m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9911900f2dfe4ea78b339d5c7b499bf4 : SUCCESS in 1h 10m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b120ca84d6e74be8893fb9124c1fbe06 : SUCCESS in 1h 10m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/afac8d5bb2fa41b6b2d10ba62a67a9c3 : FAILURE in 56m 40s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/87861c6d6cd64187b587737b9db59a74 : SUCCESS in 1h 30m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54e00418868b40b69ff0af5ca2e6678b : SUCCESS in 1h 26m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc1b843866d044baae0fe47bcf59cd4b : SUCCESS in 1h 31m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/39a3553929c945dd87408c5216abf804 : SUCCESS in 28m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/373de54f8d1248af88fb91004fa8ef1e : SUCCESS in 14m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0583f29c58b142908d1402b5c62f0d97 : SUCCESS in 24m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ccd9a921e37b45ce8472de13255bd428 : SUCCESS in 21m 25s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7c9d72df491d4d829976c5202b8f1ff8 : SUCCESS in 16m 29s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0580e30c97f24de1badde59c222514d6 : SUCCESS in 18m 15s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba48d60c812c4bb7a05de0d720c7ae6a : SUCCESS in 31m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b16046c83751437b8464379b74e87996 : SUCCESS in 31m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/205020493a784529b2b08498e3da26f7 : SUCCESS in 13m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7242de3b94604f8bb693bf564ac97c28 : SUCCESS in 20m 21s","accounts_in_message":[],"_revision_number":3},{"id":"6f4c664dfb261c73071d780a27da680ce9945148","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-17 10:45:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"539c982739b60dc9d539738378b3b2edcfa26b7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 12:33:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4b57763cb194b18856c81179f0703c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f79b48ed724355839668437fb1a5c8 : SUCCESS in 4m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/04ae35c28e7a42968d83ab4af50fca27 : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4f21cc24b18451c84da49d1759c6bb2 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5502588a694fcb929e8aed091d237d : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a45da76ac7b42e6a3b23dccc10f2fb2 : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/22190e1ce0c84d71864d22ab4205c139 : SUCCESS in 4m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2007fd1ce0d745cc8ab08b7739805a35 : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/db5998c3f48946ab9cebe57c6c2bb731 : SUCCESS in 3m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1755092a78946518d362a3376deb027 : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/089a0e6f44dc431593fba8ff09707081 : SUCCESS in 7m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b88f85e9cf2048cda1d913293c5288aa : SUCCESS in 27m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffbe80486aac4c66b9eda4af7bcbda4e : SUCCESS in 11m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3e7a6df1455a4174bb9ad255ba5c64cc : SUCCESS in 23m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d8eb3a0cf314004b45d7d0c4085e470 : SUCCESS in 13m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0e4912d6ffd241408389ff2a75a9ac2c : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/848a44946eac4628b146c8bba11ac0e6 : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4995e57f91fe47aabe5af4790d2c1217 : SUCCESS in 1h 06m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/767103427dbe46588bbf27cf91c59007 : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e9d75df5e8a341388fdbb0bf151a1989 : SUCCESS in 1h 05m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/aef8d7be87fe4a3fa54c4ed148df50a3 : FAILURE in 17m 08s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bb2774078d2d4182a6d3c260c68eccc6 : SUCCESS in 1h 31m 24s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aab46fb87ced4437947c896887b1979a : SUCCESS in 1h 30m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c26e2aadf9eb4749b3137023639c0e40 : SUCCESS in 1h 46m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bde1bc09698f4fe7882ff796b85bca83 : SUCCESS in 19m 54s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/20eed523604f47948340cc8643c57b44 : SUCCESS in 28m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/401f4a65cb43416c8fb324332a8060eb : SUCCESS in 26m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/38ed167f3b1f456786c4ad8d6eafa518 : SUCCESS in 19m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dd4695469b73464cafff4134d2e13e17 : SUCCESS in 25m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4de3100a580b425eae73e248f008f369 : SUCCESS in 18m 44s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/864f7ac0a89d4b4494c7072c19f4cf59 : SUCCESS in 15m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50c2160ed8bc4b98a45e748adcf63cff : SUCCESS in 33m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/08900fee430641f486af0ed2e227b07d : SUCCESS in 14m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ce7addbde184f3dbf1640aca488ac28 : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":3},{"id":"e574af747a55fedcd07d9af63ee95bd6021a71d6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-17 17:19:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d1f042c3acbc6f70c4c1794c84f07fc73c58781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 18:29:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e0f5bd8ff544619a57b65f0fbe86b12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a99af6c586f24193aea3bdd591d30c77 : SUCCESS in 9m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/df0d7fd785fd4cc7a3f6cff8588e4e5c : SUCCESS in 7m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f04208bd2e154331ba1954683a04e1be : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c178badcb9074551a0316e46a92bc080 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87a1a327adf345ec99c95fc8ebe65756 : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b514d51ae0fa49569c6a2ecf843f7ee6 : SUCCESS in 8m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deff0fdee9ae48cca0df282dd9e52e30 : SUCCESS in 2m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/074b90933518471db8edae8c65610978 : SUCCESS in 3m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/263370e05fcd4162bbe3dc1dc62610fc : SUCCESS in 6m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/79da4f2aae614de29dc4a830e5bdd54c : SUCCESS in 13m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0be78f21e9a74c3390088f02aa0ed963 : SUCCESS in 13m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2630432c17346aea17bd8c27eed2789 : SUCCESS in 16m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/49bad3d7e93a4acbab3b63355e288835 : SUCCESS in 13m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c8d6ed4f59e46578ab93b8fc8a500e0 : SUCCESS in 17m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/81932ae45c4a40c88390dc38d3fa74d5 : FAILURE in 23m 55s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ab76179bb3aa4dc0b4f437e59a0c0c0a : FAILURE in 23m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b86c4c89144740ae90dd6092bc1b71ec : FAILURE in 32m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da9a3e26f3af4d10ade95a908eb45c54 : FAILURE in 23m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5eeb4ebf0884a0e80da0a16eb65a2fb : FAILURE in 22m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/68f8e5024bfa4818a8e82d3ffdb1c2fc : FAILURE in 1h 03m 52s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c4252f76e58c420a91f5442c9911cca7 : FAILURE in 59m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09042fad3cce43f2a926776dac019a27 : FAILURE in 59m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9c1928f4a844d61b39e2565e6fdb999 : FAILURE in 1h 09m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cd9e39c4140b403bb07901fbc783df3f : SUCCESS in 24m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bd10ea879b7344b9a8aea28e19d448f3 : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0828e63ffbf34e36a3d68a34ee248c14 : SUCCESS in 21m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d169be76a1a94463b2bf2f6fc5c1c4d6 : SUCCESS in 18m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ad18a93667b34343be9e7c6afdefd7f1 : SUCCESS in 29m 28s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/01876c9c9d6c4ac68e342b2b5da2cf2f : SUCCESS in 38m 41s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e003ae01c01245b8a34cfc9563c210f6 : SUCCESS in 34m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/973d05ec5f4f47b19f36608d9c27410a : SUCCESS in 16m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/38853302e2a74073b3bf980a6bdbe58e : SUCCESS in 25m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f54bf7c091746e6a446e1923e038af7 : SUCCESS in 23m 07s","accounts_in_message":[],"_revision_number":3},{"id":"362d7b5c421a3f8498385be93ea3aef9f7cd55d7","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-21 08:55:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d93de97c6f4c60e2a9997a2f24b5943f125365bb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-21 09:14:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d6fad02eda648483269d48324ddd45dc1ccd7fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 12:39:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f892df6b5c49c9a2f4fb3a38f94f01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/165aeb1e7f574645883976aab68fb827 : SUCCESS in 5m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/13062a3c51ff46fdb0c39638ce40daa5 : SUCCESS in 7m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51b88368220a4f1a8293a62c2ac8d20c : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1391a1b4a20b475296c5c2b2bf179b8d : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/004cece24be84981a55088ae7f088dfd : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b3d85e0e6c0b4c6da09419450aab2fbb : SUCCESS in 7m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df124603f8cb4c2a9d165d077189bae4 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e50e377d506f48b9bf2bb38b4572a2f3 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/18a0335ac204476ca0de93dd454a4f05 : SUCCESS in 7m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/192491e48de04ad78103ae396daa832e : SUCCESS in 12m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0be865b46e824e01a22f1d8930fffa0d : SUCCESS in 14m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c9dca0a362e48d9a6bd9cdd24d88f98 : SUCCESS in 14m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/81fe56cb7e744b178ae6709859448250 : SUCCESS in 20m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdf8ae50ba4f43bb91ecc7010b004993 : SUCCESS in 18m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/97e332b3feee437299d1b9094f627299 : SUCCESS in 44m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9c0a0c397e1f48eda5d75732b11daa32 : SUCCESS in 44m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fd235a30de014350adc5e98ab614d0ef : SUCCESS in 1h 12m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ceff6b8c7c24123b16ada7a743aaba5 : SUCCESS in 1h 53m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3affef4d20144a22b80e74422b1b0048 : SUCCESS in 44m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7bea2622d642433c8d15fcf51f999d28 : SUCCESS in 1h 41m 02s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/40bfa362a5394517886be1e91268ab0c : SUCCESS in 1h 36m 46s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5410e029f21045ed8642eb44f78b5851 : SUCCESS in 2h 32m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/417e9806ec2a46a9bc42c18e670803ab : SUCCESS in 1h 31m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cab9b6220e5544cc9e6977b8388ad3c8 : SUCCESS in 23m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8e2bd0962222407595d125efd9c38dec : SUCCESS in 24m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcf4de4a299645c7bd1ae05e5f0feea3 : SUCCESS in 14m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d42a983a02b4c11a4930a28ccdcbb7e : SUCCESS in 21m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/17ade48c42ee4c6eacad1389b7199f19 : SUCCESS in 30m 55s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b2db900196804192b0497dd124a9049a : SUCCESS in 33m 11s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6183bff62e6c43b58a9f0d063dfc4e6c : SUCCESS in 28m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dff6e55c5e4848c5be457d94cf036f39 : SUCCESS in 27m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4eaea96a2ecf4498b1e18542f896df93 : SUCCESS in 17m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21787a5f4dcb4827b64a183e838f0b5a : SUCCESS in 14m 15s","accounts_in_message":[],"_revision_number":3},{"id":"70634fb09e1d8f8be292b049032530bfb6c00e20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 12:40:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b946b71fbef9b076a83f459f7022d161c55a359e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:08:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc0e4ada0b774ab1b777e42c2ae27a8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e4b3dfa587d48748f305df79c4b291e : SUCCESS in 9m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2d53b9718ee0424591620d0de3aa5f82 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb69b4ec25434addb5d2ce99bcf30e4a : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4897e296ce4544e7b1b0586c8c88c2a1 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45722c98262448c3a501ba57a998fd86 : SUCCESS in 4m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe546112d9fa4f1581e11070feeabd35 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2bcc23880b6d48bbb3a376d6a7e2cb65 : SUCCESS in 14m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f8e13172b1134306927da0f45076ee7b : SUCCESS in 11m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8c29ceb4b2b346b39326455a9c63a13d : SUCCESS in 21m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b2196099f5c4ba8a35064bb2063d25b : SUCCESS in 20m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2dc6d6b59f56494ea458323a6d980cfc : SUCCESS in 11m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a754ac0be02435680162c97a762d17a : SUCCESS in 10m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b94c196d69be4d29b984a9e55eb52922 : SUCCESS in 1h 06m 02s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0eefe918465b4b588cae3d8d07c88777 : SUCCESS in 59m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d0cc77a48be84dc5ba0bb470ae8d87c5 : SUCCESS in 1h 05m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/decee5dede6142cca7a712558de50561 : SUCCESS in 1h 02m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/57c7cd2a61564613863cee7a03436069 : SUCCESS in 1h 10m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f813cbab55094c93934b5f63710748b7 : SUCCESS in 1h 41m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6ae061ffc92d4c5bbf189a66a9fc1d1e : SUCCESS in 1h 34m 24s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bb4347046fa45b2b6336de2c65f5dca : SUCCESS in 1h 34m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da2afebce8cc4a45a7eb24bcf00d4d0e : SUCCESS in 1h 35m 40s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4e925334cdd64cd4b0fa91a20496bb27 : SUCCESS in 15m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/52c4ee885906420794ed9b50601783b4 : SUCCESS in 14m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a215a52aa17b4df7bf7e4ac5d9a02124 : SUCCESS in 18m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7cb9a3c86a6480e9246469d484fe679 : SUCCESS in 21m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4be3373e13314c8781f630dc3e228850 : SUCCESS in 36m 45s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c65f076f08724a308644aa1eb34170b8 : SUCCESS in 36m 07s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a93e62682174b62a17460bc7b32043f : SUCCESS in 16m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e700d330bbd64d4d9f8409dab1548416 : SUCCESS in 28m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e6bfc1a00a994419a92a07477dbe89ec : SUCCESS in 14m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/592143d01cc440a091e86a00bd5f98e3 : SUCCESS in 21m 07s","accounts_in_message":[],"_revision_number":3},{"id":"5afaa1da07175f9e34a9587708a46bb7d5b02a70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:08:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9fb39bae94590827e435dda03c8fb6fd70eb488c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:10:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9365901990d34588b809fba39510fbb3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2983685fe05d42abbebd895e643d0b0f : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a1e8d64af5e46d2b277f26a4e19d919 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f23ced2b59731c7eb89d9212237e933042f74d7b","revisions":{"f88aa4ae48babdd03135029c41a5404e79ef946b":{"kind":"REWORK","_number":1,"created":"2026-03-27 11:45:12.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/982396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/982396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/982396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/982396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/982396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/982396/1"}}},"commit":{"parents":[{"commit":"5e1d74ae4b123e3695416ddbb0cdcbc807da4fd7","subject":"Merge \"Drop backwards compatibility for template trusting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e1d74ae4b123e3695416ddbb0cdcbc807da4fd7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-27 11:44:11.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-27 11:45:03.000000000","tz":0},"subject":"WIP: Block until timedatectl status show synchronized","message":"WIP: Block until timedatectl status show synchronized\n\nChange-Id: Iad7fc73f16ffcbc2ab5f261322bd6777036fd5f6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f88aa4ae48babdd03135029c41a5404e79ef946b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f88aa4ae48babdd03135029c41a5404e79ef946b"}]},"branch":"refs/heads/master"},"faa07c31de75d67b7bb595f8eb42e36c223163ed":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-27 15:08:18.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/982396/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/982396/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/982396/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/982396/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/982396/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/982396/2"}}},"commit":{"parents":[{"commit":"5e1d74ae4b123e3695416ddbb0cdcbc807da4fd7","subject":"Merge \"Drop backwards compatibility for template trusting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e1d74ae4b123e3695416ddbb0cdcbc807da4fd7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-27 11:44:11.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-27 15:08:18.000000000","tz":0},"subject":"WIP: Block until timedatectl status show synchronized","message":"WIP: Block until timedatectl status show synchronized\n\nChange-Id: Iad7fc73f16ffcbc2ab5f261322bd6777036fd5f6\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/966021\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/faa07c31de75d67b7bb595f8eb42e36c223163ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/faa07c31de75d67b7bb595f8eb42e36c223163ed"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f23ced2b59731c7eb89d9212237e933042f74d7b":{"kind":"REWORK","_number":3,"created":"2026-04-09 13:48:57.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/982396/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/982396/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/982396/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/982396/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/982396/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/982396/3"}}},"commit":{"parents":[{"commit":"5e1d74ae4b123e3695416ddbb0cdcbc807da4fd7","subject":"Merge \"Drop backwards compatibility for template trusting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e1d74ae4b123e3695416ddbb0cdcbc807da4fd7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-27 11:44:11.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-09 13:48:17.000000000","tz":60},"subject":"Block until timedatectl status show synchronized","message":"Block until timedatectl status show synchronized\n\nShould reduce incidents of time sync issues when running kolla\npre-checks.\n\nChange-Id: Iad7fc73f16ffcbc2ab5f261322bd6777036fd5f6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f23ced2b59731c7eb89d9212237e933042f74d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f23ced2b59731c7eb89d9212237e933042f74d7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
