)]}'
{"id":"openstack%2Fkayobe~982399","triplet_id":"openstack%2Fkayobe~master~Ia2102d48e5cd172c404bd5d03064913a3f4b7b31","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-03 07:04:59.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-14 13:34:11.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-04-07 09:46:18.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Ia2102d48e5cd172c404bd5d03064913a3f4b7b31","subject":"Bump all Ansible collections and roles","status":"MERGED","created":"2026-03-27 12:11:50.000000000","updated":"2026-04-14 13:36:15.000000000","submitted":"2026-04-14 13:34:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982399","meta_rev_id":"adbd26d831823b1295e8161a2211da68d6267b1c","_number":982399,"virtual_id_number":982399,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-14 13:34:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-07 10:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-03 07:04:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-07 09:46:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-07 10:59:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-07 10:59:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 12:24:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-03-27 13:56:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-01 13:15:50.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-04-01 13:31:45.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"d68ca45a12cc59f422af7b8d60f89ba59718d0d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-27 12:11:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"481d5d15690a82c73c319ecf31ea2d02c2b78bd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 13:56:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a881e3aefd41430bb22f46e6030c515c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae48884407a642969ff14d3c6b2f0852 : SUCCESS in 5m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f3a0705535714a01954aafb542cf6023 : SUCCESS in 7m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0af69ad66cbe45b7a1570e98f601ce09 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fc758607d9c44e883e29e49acc337a4 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2523b4e78984528a94c02ad6b6b4af3 : SUCCESS in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a16820927a12400ea39ada9042b7021d : SUCCESS in 7m 27s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a45213d850a4286a31350e9b31f9778 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/69190b3d95d0454a80af5e1bc80e7205 : SUCCESS in 13m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc725efb62984ea4acefc7f6506ff8e9 : SUCCESS in 9m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6ad87a4911544c3c9dfa1167e8c70adf : SUCCESS in 15m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfe403c43f6046f080ea4c0201699e46 : SUCCESS in 18m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7a354f3f7af04b9f8092636df15c0505 : SUCCESS in 22m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84a470a948f248a9abcf9ae18b0cceb3 : SUCCESS in 21m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5d330df23e394b71bde6cf72a04a8d20 : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a25282c42bfa4720943fb455b5adf674 : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/26742dedc72943a1849da32bb97ad00d : SUCCESS in 1h 06m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8c9c195d8f74ee9a177cd5adbfa0a27 : SUCCESS in 1h 03m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6ac26c17364c400f88050daac4974c4f : SUCCESS in 59m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/90be9541114a4318acf19cdc432c66a6 : SUCCESS in 1h 40m 48s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/073668a25b7e4cd687fc7849c98dd98d : FAILURE in 1h 09m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05d769ed22c9423896c9fc151c33e2d0 : FAILURE in 1h 03m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1731b13db88d4a01b818d2fb7f5be1c9 : SUCCESS in 1h 43m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e3cd0c15578044c9abc6a4ab3bc8a0e6 : SUCCESS in 29m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/29c24b68710b4f1787b40cb748e71ff8 : SUCCESS in 27m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/475c827562dd463daa7d12945d6fae30 : SUCCESS in 13m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f624378a5c504c29980011640028ec7b : SUCCESS in 21m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0deaf1d1e0384b5abced896993f4087f : SUCCESS in 35m 31s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b27c5ee02ef841dda025f9a5cac290d3 : SUCCESS in 32m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0300494b903845ed85e76b551d95dcd0 : SUCCESS in 35m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f00586e20b84e6f98a75f28c21e7e65 : SUCCESS in 29m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f4b59f501f30421a9d3ba95db2b2875a : SUCCESS in 18m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e11b92297e3405c857d772ae9e16f69 : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2146c5f59e605ed8b4f11c057ea24ce540a1e9df","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-01 13:15:50.000000000","message":"Patch Set 1: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"adf823cdd726a52699e843f47c2219f2c6164e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 14:56:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/408206dd2ccd4d388763914f14a760cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98ad020545224b2fa14fc04da479bf1c : SUCCESS in 9m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/81ea3d911a9549e0805787f9bdba1acf : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/158b85c42629443c8132fb1edb7bc837 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fecbf258b6b64f55a253c60cbe3ce5e1 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fab2e0e0cb424e09b74788f260621a7c : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/716fe80bc67c4c00bdaf126eb184d2ba : SUCCESS in 8m 17s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5035dd05689c4eb1be467e0af050d09f : SUCCESS in 3m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cac6fdc7402f495195433c2183b81d23 : SUCCESS in 12m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c41fd4832ec14ef5abb153f04098a40a : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bc0d6e29a7e243a297608863255414c4 : SUCCESS in 25m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa8ef30888154981b83d6606410f93b6 : SUCCESS in 11m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d8aa89b579764aeeb6365c28a12b85f0 : SUCCESS in 10m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f2e602818a549cbb495749f86dc73e2 : SUCCESS in 10m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0c116b69004d49a893ec3a2f788425bb : FAILURE in 43m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/02d444a2ef05452c9bd44455505a770d : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b1bd7f261c374eb78ab06cb498db09d1 : FAILURE in 57m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa1307dedb4149759ae2de57e0227a1e : FAILURE in 42m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9bb2632fee114899a57f46aa5b73ea90 : SUCCESS in 1h 00m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d3c91118410d4e118505c6d44394c316 : SUCCESS in 1h 38m 40s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/282db643723845c8af4b5c3f259592fe : SUCCESS in 1h 33m 44s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f53f693420a4c43b31dce298ee2932d : SUCCESS in 1h 32m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1446379718fb41f0860077525dedc896 : SUCCESS in 1h 38m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6d149d5a46424a2bb56f92f501adc83e : FAILURE in 17m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/45958cecf22c469b810d918d2ecf433b : FAILURE in 26m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f9210455efc41f2bbf888ea08ba4dd1 : FAILURE in 20m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2797bc6f16b540b6a22fd13ef65f51b4 : FAILURE in 12m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cfe475e5d81e4408874836e450a756ec : FAILURE in 23m 28s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/26913c0d5dba4af5869e65023843b295 : FAILURE in 10m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d431d296b93c41dfbbbf94531b6fe969 : FAILURE in 21m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec1466d902284f4fbd6d8d3e53daca8a : FAILURE in 17m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f62f1883c5984fd6b29feeb48e150432 : SUCCESS in 18m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02319171d12d40d9b25f06047a47cd4e : SUCCESS in 20m 29s","accounts_in_message":[],"_revision_number":1},{"id":"06579a632e9f7a3c2f5bd6a337effc48eb544ace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-02 20:21:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"054f1c694b8ee9f8e68c491d2a7a9025124eaedc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 21:57:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e1a003f8f9a4063abe83ac6c73005bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe6bc4aa0274f389bd3238f7878ccdd : SUCCESS in 7m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1ce2834ca2714996894df428dda5ae43 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/505c534795d44a5ab9ab423db6c7a066 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb0beccdbcf4453b1ab849011689358 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afc019b792c746c19b5df052f4a2342e : SUCCESS in 2m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26c59cf750414b6a84007c5b16a3528d : SUCCESS in 6m 36s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8934677d562d4fb79d31e0cac613eaff : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5bf3b1d04397464e825701cfe426d57f : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25bb05349daf46d198267f36a6188689 : SUCCESS in 6m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/11232895b56845b28dcbc1ae0e0d0286 : SUCCESS in 17m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a83f2fa63f97488e81d05584a8b44aeb : SUCCESS in 11m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4933439c63a64809ae45ff7e2ca87f62 : SUCCESS in 16m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe7811ab9170431fb885b2f3071af6f4 : SUCCESS in 15m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0faa0ab76209487599ea2fc628990a0d : SUCCESS in 1h 01m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/74edad6317844bd5911ed0ec50a2e49a : SUCCESS in 1h 15m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3d54b6b87e7c4eff946c058c1d82aa80 : SUCCESS in 1h 00m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15cb196c6f8b45d5b1c4278700392b2c : SUCCESS in 1h 14m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a4823e301e064f8884a044ce0ea4dec0 : SUCCESS in 59m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c5c241ae8a824929b8aab45fce5768d9 : SUCCESS in 1h 34m 27s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/97286cd5c03c4f67a193c8ebbaf4c0f9 : SUCCESS in 1h 34m 01s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2625e681fae84e4eafe6abe7f4c60190 : SUCCESS in 1h 30m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/becfaf0cef8241cd8eed1784659d42b6 : SUCCESS in 1h 35m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f430eb652ad74cfbbff79318fec994d1 : SUCCESS in 21m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2087a9f996034e0bb6af68e46aa20971 : SUCCESS in 19m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32da6341037b40fd9c68b3e9727a5a59 : SUCCESS in 20m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/811df79d71dc4d7798872937db955f4d : SUCCESS in 15m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cb5d2f0de0654ad78d28165d9950b4e8 : SUCCESS in 16m 07s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9fa2a60be5104fb4b8ca5f020a95e5fe : SUCCESS in 32m 28s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/432ac8744b8c46d4aa02f06b1e11cc2e : SUCCESS in 16m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6e2ac790ae54ff4a824f22c64465955 : SUCCESS in 33m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eabd3fbffc7444ef8487d68947e21b59 : SUCCESS in 26m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47380ea06e454adfbbee9a9be0dc72d5 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8df7d4afe9c0797518b4bb950d91db4c5ee07686","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-03 07:04:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8ed15f330736a0101fd38dfe60627ff3ece05199","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-07 09:46:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8a0c8a355c36b5592c931cf000ac3461119e0315","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 10:59:30.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e3018c5612e51c2bf563bdbedf4a4ee4ff3c32ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:00:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc9d8e947f41c9d22922fa8891e1dfb514e6acc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:53:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a8c54b6b19040a0ad8202433e7ecf94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae55b1e0ff054d38a3c5632009096b2d : SUCCESS in 10m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/74ffb0d4347745dcad546a3c107e8235 : FAILURE in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f7b1ee3841448e3b204b75f991e33f0 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71c0b730e20546e58f58c487f459a2a8 : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3ca2de5b0de146b890840746b6eba11d : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1eb78876f3254ab5ac3ccf91ee4a8d5c : SUCCESS in 13m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/755bf2af082c4ba2aa1a8a90db846ec9 : SUCCESS in 7m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/be27aec361684a0781e2b0367c50fa4b : SUCCESS in 19m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92f53ad36a0d4097b5723b2b6ed5534a : SUCCESS in 11m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/eb3f49e12bf3432bacd5f0c3368dd2f0 : SUCCESS in 11m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f3341efb4d3455b9a58ce7cb74df670 : SUCCESS in 16m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/623ace9d18794d3583aa8265f56f7933 : FAILURE in 38m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/92057d8f9f8e41729bfaa67b62cd2688 : FAILURE in 46m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f7efd228f5da4d5292e165142c5f35d2 : FAILURE in 52m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42559d65e08847a38318d6774bb09056 : FAILURE in 45m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a9bb4946f9ee4ec7b0b6d4e538d2a656 : FAILURE in 39m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ca416d36a62e469dae34d1d3957e7485 : FAILURE in 16m 40s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c09b7cc119c8466cb537d475bf890f0f : FAILURE in 18m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4b833485d83450db1babeac28a4f157 : FAILURE in 35m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd729fae7aa9433ca90f6b6c608df79b : FAILURE in 17m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1c90d1c180f642b49b418787184319e5 : FAILURE in 24m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d5d21924d3ff4a4cbdc16992c72cb558 : FAILURE in 15m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7406bcc1b5a34cb08f6ed7176043c941 : FAILURE in 13m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f96e58f6fdb54153b576ad6f686af8f6 : FAILURE in 11m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/91f5bbc3fb77463c896274f27032eac8 : FAILURE in 21m 32s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5523c02afd5a4adfb06b8945436bc04e : FAILURE in 15m 53s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b6606f3661d4fc78308f55f183b5c56 : FAILURE in 15m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f32f2e52798045ec8a72db6bdb28e2e9 : FAILURE in 15m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16e65331a60e400f94336a81ca83fe13 : SUCCESS in 14m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56a7bb5b0d0c4396bd46c0c7e249afb3 : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":2},{"id":"5e8c6509ddd97b121054528eeba0568fb7cb3c4c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-07 13:26:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a730bc7fc837a127a419a3656e5eef69f8a7ba46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 14:37:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af8e3988a89744fbb9c06d66fbb9cb93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9546a80da3aa464bb1b3107e6edd874a : SUCCESS in 9m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6fafcc8e7ea943d492ecd1353448235f : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a52ef274d7e4d8cac1e332695c13421 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5498e0c589a945e5a5a6f9384556489f : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93f1e543aaef4e9f98255c28fc2ffff7 : SUCCESS in 6m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/83798f5691324610ab9b82246d619ea0 : SUCCESS in 8m 33s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bcea2eb159b44e46aecf743bf6962c90 : SUCCESS in 6m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c785669837149bfa8d4e57ad97fa897 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5a1bae9cd1b474ba95535ff52de6053 : SUCCESS in 12m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/968a0f9918a44adc989673aeb08a8818 : SUCCESS in 14m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e4b81e238d44a3ca1b4be131e0063dc : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/316bf982d02544f0bbd0c170354dbacf : SUCCESS in 17m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7700813a6b54775be767adf02af05d8 : SUCCESS in 18m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dc0c3a2d088e45d08d0fc473a3344897 : FAILURE in 45m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fa104973f136443e8faefaa66d3d0ed3 : FAILURE in 43m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b976e41f995f4b34915a9d50d41a6bc7 : FAILURE in 57m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0888394c4c74b99a73a61e774b97d80 : FAILURE in 41m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b20c54db2e49405c99756a016ef050f5 : FAILURE in 41m 34s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b9f6f0e36e740efab44c0f1e8a393e6 : FAILURE in 14m 44s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/db66940ff76f4c6c905f0d09db993208 : FAILURE in 17m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51d7ec1ed05b4fa5be6b92c65c21b966 : FAILURE in 17m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f860a579909145ee8f826c1534f13e7b : FAILURE in 18m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4f27e2f728794dfdbcd89496fe04a653 : FAILURE in 18m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2782c930441a4fc0bcb67cfbd6935f0f : FAILURE in 25m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b77d1480f3f4b88b4c6455ed1cb76f2 : FAILURE in 19m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c82e0390ba3b4003afb2999076368bba : FAILURE in 19m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9f433362bce14643b3ef7afea8b03621 : FAILURE in 18m 44s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/623bb6d49ab04a5783ccf3f82832bd11 : FAILURE in 13m 04s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19fce291b49d4333b9c2a2a04ffe46d1 : FAILURE in 17m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07628ea85894419fa4326cefccf10869 : FAILURE in 19m 00s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/577d17cf045248e3b607eea886077229 : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d8ef85b3a314cb5aa0345b25a769197 : SUCCESS in 21m 33s","accounts_in_message":[],"_revision_number":2},{"id":"be2767adb20ed0d3a97f028b71a68707c4e996e5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-07 17:10:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58209760c1700003bac1dbcf07684259f7a38218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 18:58:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcd35b6060794e5b9545da6b61e66d54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b5b96152bc40e8916d692e874e5791 : SUCCESS in 9m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3776c29ec2bb4b0390630d51791afe3e : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab61f58e61c54dc3ba48bc4a4a7d2eec : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/065aba31e20c491d98c3e25a7918dd8b : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0bacb542f9d4bef84c54a39407ef008 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1f9b515781424da3b2705884784ea835 : SUCCESS in 7m 20s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/275e7fe278744ee49c87201c86b4a326 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d402e381339408f8c98a4a1cd80afb1 : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4282a3dfad2a499e8efc1b41f8337f75 : SUCCESS in 12m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2326a7d2c8fd4e9291119ee95c0474b8 : SUCCESS in 14m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8e4149d2e474ab1b468f62ee0ef2ad3 : SUCCESS in 12m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c2c412c96698437ebcc250d78d1eeed8 : SUCCESS in 16m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/730fecf7248349bab2507a31e121c8e6 : SUCCESS in 16m 37s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c7a94cfeac494b0baafb8aeb8f3982b4 : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/da2c740316ae44aeb8977ebc36c1717e : FAILURE in 8m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5ce25c93377d4260978cf9d6fa4d9b39 : SUCCESS in 1h 12m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/315dcbe19b584fe896da2a721ad7c430 : SUCCESS in 1h 09m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a8f1321bb0f94c018f1277be56845bb7 : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5621a13e64854952b0d050a940582ae1 : SUCCESS in 1h 45m 19s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/213bedd9e3474dc0b8a3995fa040e37f : SUCCESS in 1h 38m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ee04715ec88482696951f0caa84bae0 : SUCCESS in 1h 38m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/912173cbf6c649e2a274a53aae982e70 : SUCCESS in 1h 44m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/82e2aedb514c4d129b2c4e44fe2bbbdd : SUCCESS in 15m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8405002b27974467a6aa712a52dd3093 : SUCCESS in 16m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6884739f68b4caf8738e8bab9045837 : SUCCESS in 14m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b5daf82b49f04841a9be03211491d908 : SUCCESS in 20m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/afc2039a43ed4678b3a8405485d3cc43 : SUCCESS in 18m 27s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a6ab94edb32a4000852509cb2079de73 : SUCCESS in 18m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/288567e41c9849bf85877d4e466ec979 : SUCCESS in 26m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/405237015cbb462c921f6156347b17c1 : SUCCESS in 19m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e1ac409efcc04895850097d0559096f9 : SUCCESS in 15m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/525de06c57cd46a78ae1c30148f0c0ba : SUCCESS in 22m 31s","accounts_in_message":[],"_revision_number":2},{"id":"114422e4e646713050ea0872e2259418157d1e1a","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-08 07:54:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7819a8c652d815199b04203c3f5b575a4f866b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:32:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/996315adea1b484789944feca20b27d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/561a0a10e215480f84980b8c1234d04e : SUCCESS in 5m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d9a359514b0d40149c4d851a634794c7 : SUCCESS in 7m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a1feb2b5f9e4ec384de50d8c1b131ea : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d37b942f7b04ee98dbb1c82f2fdf897 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/085561c766264808aa2ced73d131f274 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/528611b0276d4e608035ab9a6da876e3 : SUCCESS in 3m 51s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/efa0507c947e4c968a1a77609b54d1f4 : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d29c958834214eaca0843c003ee6f1a8 : SUCCESS in 12m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9472cec6745b41b3815de7af7890ab5e : SUCCESS in 9m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f95908bfedbb43abbf2c7e88dbbb39fa : SUCCESS in 21m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a8d73d230284b43be5aa474735d8f10 : SUCCESS in 20m 18s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bc695f8c397742918bf91222db75d13e : SUCCESS in 11m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2451fd90ee84f9f9b0d6d5c5eb1ccfa : SUCCESS in 18m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c8548abff67f4236bda22e65729395b9 : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1bdc45c7caf44d19a918329d5497b364 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/88aa3a23794a46c693ec84a060a6d477 : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cc5d7189cb1450581fb0e3766519fec : SUCCESS in 1h 04m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eb34a021168d44c09ed8965ccb741818 : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/68979d71d94c460fbbc8290fd90f2c60 : SUCCESS in 1h 36m 26s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f31563bc8c5b4c3183ec86d8ab87068e : SUCCESS in 1h 34m 56s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f8b478857ab4143ae9696a7064a9b9d : SUCCESS in 1h 29m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed51e2ec2fd440f68d60b7209d338628 : SUCCESS in 1h 36m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/abaaecb5877a48f289123d4c5a40c6d1 : SUCCESS in 21m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ffe34cab16ef4375aa97e4ddd44baf34 : SUCCESS in 23m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7aed1540d7084c9baebe53b480f9d17a : SUCCESS in 24m 45s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1edc3739e42e41a9a300fa0d274746f0 : SUCCESS in 23m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/32d477eefb4b4976b64a5c865ee77c2b : SUCCESS in 16m 58s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/839dc23ad98f41df8ee7cd2385976ce8 : SUCCESS in 30m 44s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88b4b68e7c2242e8b199d988ffffd7e4 : SUCCESS in 17m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d1f324d5d224f17993adf4e6340316c : SUCCESS in 17m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3e8bc6912e42467e96fc4bce84b61b00 : SUCCESS in 27m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d43a424f3aae4247bca2b47305ec3813 : SUCCESS in 22m 16s","accounts_in_message":[],"_revision_number":2},{"id":"c85d10a0d8994eb5e0b6a859cccf0acebd35f2e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:32:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"251d804dcd502254e091267d39e4ee9701311837","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 11:12:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f10910ba185499caecd20fa8beb2bf8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf07ef3715414deebbc828ccbc33930f : SUCCESS in 4m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/67d684ef36cc4d399eba2b629d9d4380 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/095cfd665937413f91fdade03e9768ad : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad1aeba9d5a34a7b9b870c9fdbff5359 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6164bea565d049849e19b8729d5930aa : SUCCESS in 3m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23b1eb49e9f64e7a8f4b7b86b0f790b8 : SUCCESS in 6m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5865abdb91c24bcfacf932032382797c : SUCCESS in 6m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fdcbbfc4497942dd81a6b6d52cbf1bcc : SUCCESS in 14m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a509994271b4622a1e6ea462cb5ae02 : SUCCESS in 12m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8c6a78c5ca284045bf123c2470b9cfa0 : SUCCESS in 12m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ca4972f850e4c90af9169d4b6be08b8 : SUCCESS in 17m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e4d28acf7592404596da8469ced41598 : SUCCESS in 1h 09m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1811834216d94ff88a6edccadbe9a152 : FAILURE in 53m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d74004569a7f4c1d8ddc3775cedc766a : SUCCESS in 1h 00m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52894823f67f4975936bf90e42867804 : SUCCESS in 1h 02m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2d966549e47e47b39215a693117a6d36 : SUCCESS in 1h 03m 49s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f4d1375127ed46a6802b642c7a6df64c : SUCCESS in 1h 35m 35s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e66dc5ce31d540e9b8b07aac829e8a8b : SUCCESS in 1h 28m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65de166f04df42559690e47bc39dc7ba : SUCCESS in 1h 28m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef3e9f87122f4f9b9558e1cfd698110c : SUCCESS in 1h 32m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0edd439529c44ddd9c9ce895184f46a5 : SUCCESS in 24m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d0a9841e8cea4f34a26c243a57962942 : SUCCESS in 22m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f9e1e5f6d294597acf06a0c20472dfa : SUCCESS in 15m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/face3bb97c5640fd858b008b95710722 : SUCCESS in 19m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3ff61254705a463eaa74a709dadb7669 : SUCCESS in 38m 05s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/630eaa60c7a6428abf1c04b681fd623d : SUCCESS in 20m 27s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0380f6c0bd684782b09df414ea83fd5c : SUCCESS in 29m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab0d32c62d3d47f5a5e19b72f6daa166 : SUCCESS in 29m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1f9dc4e6216e4330bb47709b1c492842 : SUCCESS in 17m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69cea88ad45a45579311d4ed6182a23f : SUCCESS in 23m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f762f5227fd7ea41087a5e17a285b8091dc927ae","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-08 11:48:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f2cbc0315353bac1b8725a7a05ecfb624eea6855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 13:31:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fb2260985ed400f814d36180c03aa93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f609a04a11b74bb7af20b4897b9c97ae : SUCCESS in 8m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eaf89400643f4f589d300ea5bbfe391b : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9f7af6270c7424684b98e49275890bf : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0969a0f9cc9944ae8453fe20bc30c7bc : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/038921f9b99d423e90d5b3e7268ebba7 : SUCCESS in 4m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd891f0866374bde9a3d2c3f83a8f66c : SUCCESS in 6m 18s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46931979273f49779b554070611f27d2 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/866b05787b2d42d0a9e988d6e68fa71b : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9730845fc8a14d72b30ae5b301a546f3 : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/93521090ab50460ca53cd7d2d74a2ad4 : SUCCESS in 25m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba044e6e96d4403694edeeb7bc6fb25f : SUCCESS in 13m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/82952f19897140ee8e06d07d5de0edd2 : SUCCESS in 11m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b62d92c2cd794f1580e407ef69f642ed : SUCCESS in 11m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a6fe042d045346f4935ef8a3c8b318f8 : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cc1575c02de848aa8d7a61882a5d6b94 : FAILURE in 50m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5576af9c09cc43f7a12f46b273fcee49 : FAILURE in 58m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ff3b947d9ff478ba6e5430b41611578 : SUCCESS in 1h 09m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c26284ff9e484e4e9929bcd40bee31cb : FAILURE in 46m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d5ac26c1233d4b3cb946a5bd47686b32 : FAILURE in 57m 54s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/37f5714ef58d46739d3e3e4cd1bd52c5 : FAILURE in 58m 41s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f12eae91cb7f4c6ebefd9387591d0b59 : SUCCESS in 1h 41m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ee83471492f4bfcb57173f4b1f80244 : SUCCESS in 1h 38m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d78f3f56a63a4c3192512b3a1c1fdddb : FAILURE in 20m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/781d3943788d4acd96749a683f81a092 : FAILURE in 19m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91e2b0cbd8c34484a03ad40260919928 : FAILURE in 18m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3db224d5f44846818f0126b5e500aedb : FAILURE in 22m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e02134b11d4d48fba155e690ad1ffab9 : FAILURE in 26m 43s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fe83213f62c142848ff8ce41573a55f4 : FAILURE in 25m 40s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12681fbe82704968b20e93d881bbcfbf : FAILURE in 18m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c12d1d893b744c787eaeb86a111f314 : FAILURE in 24m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/53f2fe3d87c04d6cbac7fa29946fe492 : SUCCESS in 24m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/430cba48141549e2ba2ffa4d169bc485 : SUCCESS in 15m 06s","accounts_in_message":[],"_revision_number":2},{"id":"58facec0a60cb916d7a048e2aa64d6448397e93f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-08 14:28:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"96ddad38b38de0fff06a2c1096e63891824fba92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 16:12:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d554b5a24e4427e8d4db50ddfd2013b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16643a4337b74355ad4f625f9c82fe0f : SUCCESS in 9m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd28dd4d1f484e97b0025210fd35a8a9 : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/002f214e203349d9a2eefb05a566409c : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b59712a14e5b4a7a8b9aad5b9c57ad26 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f841b0cf12d44f81b8dc2aab31f4492c : SUCCESS in 2m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f00b1e63891443b19b350000e39fbf9f : SUCCESS in 10m 08s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/14536619ec084024a1ff08cdf626a05e : SUCCESS in 7m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dbd81826744a4cbe8c14af7dd9a98f14 : SUCCESS in 13m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9715874118834168a2d011b01297bb00 : SUCCESS in 12m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d32b0bd2ada247a9b6053768f7fe4966 : SUCCESS in 15m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9b1392b67804b3d812083986274aa1f : SUCCESS in 19m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d671bcb29c254193b91293db06b32b4f : SUCCESS in 25m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e90fca75818b43e68ad4e12d1f2fc13a : SUCCESS in 20m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7ab50e4654b94dc0a22e98297eba854b : SUCCESS in 1h 06m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3c09fe4fef0b48619b6583e317fe8314 : SUCCESS in 1h 07m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1d53750860bf42fb9a92cb76dd262a82 : SUCCESS in 1h 09m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a97567d0427c47629b8e2d583e1bd8db : SUCCESS in 1h 07m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f94934d2b624eef961081b1a638a362 : SUCCESS in 1h 09m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/82440af2012845438369d1c6f69c377a : SUCCESS in 1h 38m 55s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b07f43867cd54a95a7c5ac164c1eebbc : SUCCESS in 1h 35m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c13c6b5af5414920a4be616f7d8a4a9e : SUCCESS in 1h 30m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e7b4d3d5b82455d8a1ba7aba4120705 : SUCCESS in 1h 41m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f80f21b92544479a8a7223095859f9bb : SUCCESS in 16m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8852a22845ff442bbbfdbdd491a8a099 : SUCCESS in 29m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf9b4a91d92d4e44a6fcdcba20ebd0a3 : SUCCESS in 21m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de4e4125d3094878b009752ceb9c3bfd : SUCCESS in 20m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2634fd2696f141a7935b09f9fb07e8a8 : SUCCESS in 36m 05s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6bd90aa9023a4fdda90f0847ccbe402a : SUCCESS in 30m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba3422adc1f7461e813c56942026fd20 : SUCCESS in 31m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76db33fb64814b289206fff0dc5f6742 : SUCCESS in 29m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/957a9e92d1a746eaabf113e2d800bb8b : SUCCESS in 20m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7e4baf84a1744d884d4792b5ef195ed : SUCCESS in 22m 50s","accounts_in_message":[],"_revision_number":2},{"id":"c2e68e8a130823d191d24883cbd12b81215296cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 16:12:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6b42b0a5eb122f8a7120cf89b2286141a3b9bbce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 17:59:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2389736b10942afa3656a73e1a58950\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90cc03a6781c4a3c81c4416bb3ddb478 : SUCCESS in 8m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d463bdb326a54012a5e8e9330fcd740e : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d558b577bb44aa58d1b96b0608358ac : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d2a9c15e57f4631a80893efd9391884 : SUCCESS in 5m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9efdd79bd75d4700baaf381aa3d0312c : SUCCESS in 3m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9d953d92cd8469f8d519cacfbba1b90 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f1afdb677fd445f9b0f7f95d1022ecc4 : SUCCESS in 6m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f0952aacd0e7459fa4890d13447fcdd1 : SUCCESS in 19m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cd7f17357614817bcf940a6c6839587 : SUCCESS in 12m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/66e26a7451324a4ab4400a45a6217ea1 : SUCCESS in 26m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c4e20789e5642538313110afa3053c1 : SUCCESS in 20m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/aaaac65936324a1ba1c9b9cf63c794e7 : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7308097644674941a4a1b34c141bbe47 : SUCCESS in 1h 03m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2b6f4f6723a14430a280ac5de89f7a11 : SUCCESS in 1h 02m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95e4770ac1af45c0882e2a6f9f1080da : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0be3621d9f054f23aabc9129c403521f : SUCCESS in 1h 04m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/93ec6acac54543a38b19d21308e0e207 : SUCCESS in 1h 39m 15s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c40c38b9d6924c19b2806fa83761f0da : SUCCESS in 1h 40m 08s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10f42c8faf63427a84687960e0833445 : SUCCESS in 1h 35m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ee256c2f4f848edaa593be3b14764c0 : SUCCESS in 1h 40m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a79031e2428c47b7a905f30817d8decd : FAILURE in 21m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/785323dbfc7948bca1986f12f21891eb : SUCCESS in 16m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fdeff24e2b540669ddbcc9ae3ed139f : FAILURE in 22m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6faf565b634a435894b4d4f45f404c10 : FAILURE in 19m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b8b5ed369f7045c78dff298bdc11cc71 : FAILURE in 16m 37s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/507ab6d1546c4640b8bab4c6e911f649 : FAILURE in 22m 41s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/109810d3cb044a3288c944bd7c7ad040 : FAILURE in 18m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3a9cfebb6a34fd49d17df81164f16d1 : SUCCESS in 31m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2d0aaab4bdbb4187a6ad025b7e29fdc3 : SUCCESS in 19m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/296ae70b1896494d991727b5a343a5fb : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":2},{"id":"4c70602efdec9ce95f37278d37b17859aac52eab","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-09 08:34:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b179a60af7bbfb4b7f9ae9eb17d69edf0e409d10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:26:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03820f2f75834ab397068bc55480e62c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3c7fe7fa2b04ce8ad8374b4c0ba8914 : SUCCESS in 8m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7845d44a84f34442bba6f887b1ea16c3 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fecb7ca0e47649b3b8e48e623c2c0715 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a4e8241e8a3436b9ba9fdc9d60123a0 : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db31f435e4c54086bfdcc59fe064ca5a : SUCCESS in 5m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5559c0c24054492e8a633afb91f276a8 : SUCCESS in 4m 19s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/993766234ebe4b1c92635bb29e8eaad8 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b734fbdcdc274fc685202b89563a991c : SUCCESS in 13m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/714d9092147f4c56b69f85c614cc4b93 : SUCCESS in 14m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c4c27a2248694588a6daa5f5ca93d97e : SUCCESS in 20m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ee591640f2744589020630d191f8c2d : SUCCESS in 23m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/df58e4b8a3004c1c840fcde6b59d0578 : SUCCESS in 24m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bfba805485245ab8e4a5045ef602af6 : SUCCESS in 15m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ade115cc59684bd08296e447a12b6e0b : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2d9fbe932e4e494d86c468d16d5c9cf2 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8b447ce17978444d8d58fa1873e0398f : SUCCESS in 1h 10m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/911248792e3043648cdce81238bb43af : SUCCESS in 1h 09m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de09861d53a44d6d837ca71db09bed09 : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c540c6623cda44998b4f902f18054690 : SUCCESS in 1h 37m 16s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c950cc0e60b24bc68051ee605e3bd63b : SUCCESS in 1h 37m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb80b2ed86b341868c65b0ec196441dd : SUCCESS in 1h 47m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57adfe13b6714605b36437e752c7c669 : SUCCESS in 1h 41m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/91ddfc71ba8a4ec2bfb0adf74ed2bcaa : SUCCESS in 21m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/eeafaca8914b472e9e86f0cc482f33e9 : SUCCESS in 17m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c57e88744a847549b20c6839f69f338 : SUCCESS in 18m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d81a5a5d78314c07b3faa709f8142151 : SUCCESS in 21m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/25c8d87cefe24234a1f6a6582af6019f : SUCCESS in 26m 22s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/967fd36be01341a79691bfaaedee71a8 : SUCCESS in 33m 10s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/739454dd0eb344bfa968ec32e49aabe3 : SUCCESS in 34m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4654dcf4941c4271977d7655d6efc2ab : SUCCESS in 30m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8f155a5582a642408c54b133bc5b4aea : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd3a3f11959d492fac894b80ec2cdf77 : SUCCESS in 21m 30s","accounts_in_message":[],"_revision_number":2},{"id":"6b2552185ff96a45a7d4681e9a764ab3a61443d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:26:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f578414f4f3f4d2339e3551192149d3d5f89b71f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:12:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c74fbe5ef9574756a3f8d4b0bdcfb1ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55139aa98544e89bc42a0662b794db5 : SUCCESS in 10m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9b0697f2fed94641b5b3a5d7bc95df86 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d48c49abdfed4ff786eaf1c213888569 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6727e4328514e1bbf0cdc97c6472ace : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c2b4681b75943af967302fd23f24c9b : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8870fd54b39c4784a43a5300963b5f4c : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/54f6055e4cb143d3b3d28979c59e489a : SUCCESS in 11m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/39a36af7be1947a4a10012011c947ba3 : SUCCESS in 21m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aeb48fc521db4a6385ee052e43cf14ad : SUCCESS in 22m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/07853b54a7a94f49866f6c705297a9f4 : SUCCESS in 23m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e047c4758b444739db657abda220e0b : SUCCESS in 21m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/42885cadee8049efb7a1fa025cdc2152 : SUCCESS in 1h 03m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7e01e25754a94b6f9685bc052441476d : FAILURE in 29m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5175e1679f6f437c81a539361125e99b : SUCCESS in 1h 04m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/876579f7649041d4b509d6511e86ac73 : SUCCESS in 59m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/87bf1c0c47e34fc18d618e0cc2d15658 : SUCCESS in 1h 11m 37s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9c2b4f88eab948f6ae20a3598e51d220 : SUCCESS in 1h 37m 30s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0782da70a0964bae9ac84dcc6a8cb1c8 : SUCCESS in 1h 31m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d69fe39a9c87498b94d6ca7310319f9c : SUCCESS in 1h 35m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f2dc9112dda47d3b4eb9116ecf31832 : SUCCESS in 1h 38m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5b7f4a53a5334509a4ace95a79e09aa6 : SUCCESS in 27m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/98b51ca492d14760a28a1ee59688f1c1 : SUCCESS in 27m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f9fd04c241144be963ab20f5d4201d5 : SUCCESS in 22m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0912e3819cbb4cfb9780d412d655ba93 : SUCCESS in 24m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/da28e8b9614d48aba3b1445f28b28c79 : SUCCESS in 20m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ea87de92b999462d8d0f614adec746e8 : SUCCESS in 37m 24s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79ac4a280fdc48dabe87b35da9ce63d2 : SUCCESS in 32m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/947d0343458145d18ac7d4567c4f1116 : SUCCESS in 32m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cfc42dbd1de544cf9819fca5d02ee6e0 : SUCCESS in 23m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9742b1ca9284fa990c6d76da4c470e9 : SUCCESS in 20m 38s","accounts_in_message":[],"_revision_number":2},{"id":"0d8e2925e247edc06c382ca1646bfebd1c3d30ca","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-09 12:14:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9adffd801bf28f25430dc90cbfba8a61fd0c6204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 13:09:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1fe97152bc94184ae98c601170644ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02abacff50d8419eb6b33e1f6b908fe3 : SUCCESS in 10m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3dd3e3b8575f419e9a9af10d138e213a : FAILURE in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74ac00b228c9414b881e475c76622b6d : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a5b250c712427cbd27d05c75f935a2 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b590ef8c86e4e8fa361e93b950ccafe : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/92ffea8ab31549559da378a009a86555 : SUCCESS in 6m 49s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/62bf4e8b2ba743e08df814f454ca0e08 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/08142d1b9aec4d7eb57743ec8fda3775 : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52c73a62307f4be7bf4df796d6abc45c : SUCCESS in 15m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/281755bc86524e7ba538b79560d88078 : SUCCESS in 16m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19095182a99240f48b268160f0d459d1 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0c62d2aaa67a4e67b748fa6f90fba22b : SUCCESS in 22m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78034bcf13a743aa9ddb33b1843c48d6 : SUCCESS in 16m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7ba7cc9f3ad241909851146e35435187 : FAILURE in 50m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e2b8a5b794dc4bbf8ebdd692cc8097a2 : FAILURE in 46m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7687f819e0aa444eb32cef2532efa5e8 : FAILURE in 54m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1776100125954ee3b8da820e9cd6ca5d : FAILURE in 40m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/baacbd303b324e38afaac9818c5e40ed : FAILURE in 46m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/95b367cf4c4d4938b09248faa4d0b106 : FAILURE in 11m 45s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/390d58940e3e4da6bfdc772b285b48e1 : FAILURE in 12m 24s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19f8006e451e410a8bd03a043b8ffbb3 : FAILURE in 11m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f18c58563f74bb7a8e340222ad3c48a : FAILURE in 19m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9781e3fbd6654f8d89696accf1253bce : FAILURE in 18m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f027a82adb8f493d9a5cdab575e0b7ea : FAILURE in 15m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32485fbac5604ba292bdddd98be86b34 : FAILURE in 16m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba711b1f09de46baaba96fab0d70ce12 : FAILURE in 18m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f49e0f2dd822475cb887bc4b3795e420 : FAILURE in 12m 17s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/210d584404634854b989db2394ec4b3c : FAILURE in 9m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83d371b48b7d4761b75e9d8b195432a5 : FAILURE in 20m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b64363830e7416f9c80e92d751014ac : FAILURE in 11m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/22e6a42ec57f4c71a0cd271f4c15798f : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55af18e445f14ff18b7c74b6ea1f75b2 : SUCCESS in 20m 39s","accounts_in_message":[],"_revision_number":2},{"id":"722a9d342d114774ace3a55daa72c0d1b61c37ec","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 14:10:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdcc941778fa6ecf1dd3ba20d81a196000c818e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 17:39:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73224c621ad64da9855bf230c947c387\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a4e5960a934d3585ba54b899f7eb59 : SUCCESS in 7m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a44c9b87b6724551b9545dd8384e3885 : FAILURE in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/099bed04d02e4034a9dcfb26ebc9602c : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e944da3a0b6e4f7d9b2ceb0f1e9bed38 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9fd0e2e5ba374c1ca27f7fbb15da5e91 : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c204556f52b34f2082a61c8b9adbb7a6 : SUCCESS in 6m 44s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a4965d9301fe4bb28410c0ad54d089c6 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1f9a499a284a4602af8e6b3b194b689c : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/95067963dfc945fda1dcf35e74eb782e : SUCCESS in 14m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/22a92e18029e43f2bfba3d2b056479de : SUCCESS in 21m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/436e26e4638d480ab7949a781f5eb0d4 : SUCCESS in 12m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5065d63ebdd54b50aa3c8879f9448c80 : SUCCESS in 20m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c3cde08c4c04b4b92f62b3674d00afd : SUCCESS in 17m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2ccf74085e8d4ade87ab4f80fba4f65c : SUCCESS in 45m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/66c6081afce249a0871c234218261eba : SUCCESS in 1h 13m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/95571461e0e44472af37629816e73112 : SUCCESS in 1h 18m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df9fa88aad0c489180de55404c215a95 : SUCCESS in 1h 18m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/60a919f9d1dd46398a5194de12552438 : SUCCESS in 1h 25m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/582eadad36bb4cf78f43c2fc0cb659b1 : SUCCESS in 1h 54m 16s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d79b748203aa4644854a6d40aacc821f : FAILURE in 9m 30s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/279ecb02e4c64c5a82ec2b9c3a7938fc : SUCCESS in 2h 47m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b9507b5426a4e0d89b3721bc16a5920 : SUCCESS in 1h 40m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e668ac87f23b46ecb8fb2215829f10e1 : FAILURE in 21m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/64a215719a6c45deb293312ded1087bb : FAILURE in 25m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06a3db49daff4bd79be65ac24a443730 : FAILURE in 19m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a6094c15d914bd2b31d7c86bb97ff44 : SUCCESS in 24m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d01e689c11474a22bc458a27ba24ad48 : FAILURE in 15m 55s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/10a23d5937444603a5bff5fc07b3a7d9 : FAILURE in 29m 17s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b846367755f4ad0b17fb9cffcc10844 : FAILURE in 18m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff92e2684fd34e6384e03f272b7515d0 : FAILURE in 22m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f94743a9c7374fbdb8e7115e9dcc4a97 : SUCCESS in 21m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6508264597d64e16b1b94a195693da79 : SUCCESS in 16m 53s","accounts_in_message":[],"_revision_number":2},{"id":"47ee52304e7065b2987568b2901119fa665b63d9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 17:48:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ae518a278b944d40128f5ebe22f84a175af950a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 19:28:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f255676028549db84872148c339f6ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4687dbec4194670b612bd5ca3375826 : SUCCESS in 8m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/032678300a3143bfa3ba7e56ac88c56d : SUCCESS in 5m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/335ff3fe16b04f839d5899f7774a45c2 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04157343eb4941bcb88da1e66a52072d : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e0f79388e9f463ea804e51c61ba26c8 : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/49974252076b43268876d6ffd174328d : SUCCESS in 7m 16s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e10eb9c981574ce1bad8aae8833f1e85 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ef38e8f36e684dd6bbd5c2c261758c35 : SUCCESS in 13m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b12d88280def4cb188052e5b723ddede : SUCCESS in 11m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/21b9b1e9e292413392f0032b9f8f7314 : SUCCESS in 14m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c6c0e2fc26c4538a4870e36c22a66e6 : SUCCESS in 20m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2a6bb2f8513041459050f79629b56e9b : SUCCESS in 11m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93fd352d664340558e9f959e9c9fdd66 : SUCCESS in 17m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f807bcc37045406eb37faccd1b7ca02e : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d7bcd05cc71e4be287201244d09d27a4 : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/029de462cfff4ffcaab099be7ac4d759 : SUCCESS in 1h 19m 08s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/481d47b80abc4797a44345f86424da62 : SUCCESS in 1h 11m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ed222ba44021424699a3db240135f357 : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1b7445a9e07d4d32baac559a511173c7 : SUCCESS in 1h 34m 28s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ab3860608f5a434691c6a6b972f3f1f4 : SUCCESS in 1h 31m 40s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e13beb0f347e45f093a5cf106f128982 : SUCCESS in 1h 31m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c01958983fc149e0b3208874b75bda90 : SUCCESS in 1h 38m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bc570132041d4bab9c43e4d3feecd5c5 : SUCCESS in 15m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2efcaedb28924e72bf969fe90493446f : SUCCESS in 16m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59907dca4934492aa4b5db1910a56c5f : SUCCESS in 23m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ed4008227d79474081a36ba1cefe2e17 : SUCCESS in 14m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e1041297d765495b957cb6fec5fa84df : SUCCESS in 28m 27s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/55a60f39b65d49709582196d8da22295 : SUCCESS in 20m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f437fcfea49c4b7290a5520da4d11aa9 : SUCCESS in 28m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d5e3f20e3b74a7eab22cccfa7486e18 : SUCCESS in 28m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ab93228801894e3e986c2bde12f7d430 : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4c852ebc3b44626b58fc66b88ad8559 : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":2},{"id":"0c61072743f2bb133be108ccd7a3e10b6ae8fc4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 19:28:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e997bc0a87989077c3b39775dbe18000341a6cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 21:06:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6ae9c8812cc403ea1f9f01346ab527f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f1a22fc20f4cde82d5159c3069863b : SUCCESS in 10m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c676fcebfbee414aaa312d99e5af261c : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53f459b2ccb64c7883d91a06339a5b03 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/765d1ca73aff45f0bf543c0043a45046 : SUCCESS in 2m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92d5f236980e4c2f858bb5c55e73d29d : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00f080332afb472cba5af58797c7f4e0 : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ca7f481276b43d9b462b2c3e4851c57 : SUCCESS in 8m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bd8c3d402a9f4e999eb0be51930781ee : SUCCESS in 17m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1289bf272cd746c9a4dc631e11198b62 : SUCCESS in 18m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/32360b11e7f34e418cddf260a713a9ac : SUCCESS in 18m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af5212ce52ef443abefa7982549a9177 : SUCCESS in 13m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3196b6f45d73475697210e25c512611e : FAILURE in 55m 00s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9591ee8e999d4eb8873ee3f908003223 : FAILURE in 51m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/594fba5c19bf4163b7f7900d49ffa23b : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85f9ba2c8d1a4550b4ab1fc30bfa7d8b : FAILURE in 52m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c9fe937859a46548894e1849a58d1dd : FAILURE in 51m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4c9f635219c248d493dcb95a4d7da372 : FAILURE in 1h 33m 06s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/961e5f7ce8c84ea5921ab7a911397096 : SUCCESS in 1h 35m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/887c023ac1354f0da85f1a00dbb6ba25 : SUCCESS in 1h 30m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5cc26966b2742ae956efff78fb80d71 : SUCCESS in 1h 38m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7b1233ee89d748178c53376eb3201240 : SUCCESS in 26m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/138e6b3de0864c638d348966e22a1406 : SUCCESS in 20m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d74e6f28e7c4b21a756a7a72bc2571c : SUCCESS in 21m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/303389de9a9e4808a63450344a0571f1 : SUCCESS in 20m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a7b8a19b53b8478982760eefa267866a : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7880383743704c2f9814c1b3f7d9a41f : SUCCESS in 35m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/475c2e1e33204804ba539de906eea2c3 : SUCCESS in 20m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5321826a63ba4e64a98f3e2885e8a47b : SUCCESS in 28m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fac91e1ec11144ff9bd785b86b6907c7 : SUCCESS in 17m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b84c713df99c4d1fb672afa22e558462 : SUCCESS in 20m 06s","accounts_in_message":[],"_revision_number":2},{"id":"93e5421f58d3035c9cc0cc74abb3db2756c81e4d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-10 08:48:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abf65dff376219f9085c1d1b8138184292e1127a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 10:26:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/155d903946854d5ba3947623da60fbe0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f515545db2494efdbed969010e903f89 : SUCCESS in 4m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bf8ef5d4f4414797a1fbc03434c13631 : SUCCESS in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b56b8712f2e04bac9bbaee53b0feae3a : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15859a3e2cc64e849bbbd02e5a4e3816 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ff0dd0d29c547929ea17a6b182c1737 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f4b6e6b928c4c20b1a37e4a323e2c62 : SUCCESS in 7m 07s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/77bbce13f1b041958ce426ff70e6ea96 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2ef310cb05d147d29de159382d692164 : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c47d1e58b6ea45fb9684563713370053 : SUCCESS in 7m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3bd497b16d404e1babd35b24f61ba30e : SUCCESS in 15m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6a6a687b19148e7b4750980ed3a7040 : SUCCESS in 15m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ff63fb61f0e8480fa49b046b51c9d763 : SUCCESS in 15m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/379cbe9368ed4bff80d31b1e9d13f008 : SUCCESS in 18m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/63ca4daeb62745d0adc51382a15d0671 : FAILURE in 50m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c252ad9b608149a190a38f1a6702466b : FAILURE in 51m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e171ddd30d9f49358d62197053f24514 : SUCCESS in 1h 17m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64133619650746d59d1b3802aabb3fb0 : FAILURE in 45m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/92ee7df135ce4f8cb7bf2a5310d49266 : FAILURE in 52m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/511dcf0994c648478b4f2d3073d093ee : SUCCESS in 1h 36m 42s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1cb551d58c1e465f808cf9e1d948c034 : FAILURE in 4m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e500f4788e8f40aca0acd30b6b4ad1a6 : SUCCESS in 1h 26m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0018d3c976b04e47b074be1b30e917eb : SUCCESS in 1h 34m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/aaa9bd5008244610ac7f90be47a56e5d : SUCCESS in 16m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a77590c6fd91400c9e03d4f7ae48d7c1 : SUCCESS in 15m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57eff6f123f744229cbe5514cc8c7b50 : SUCCESS in 23m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/02bb5976fca94785b8877b5b250f66a8 : SUCCESS in 14m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f241cae7e3eb4a5eaf8c3f4a2c05afda : SUCCESS in 26m 50s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7e0d9693578842d6981984cca445e157 : FAILURE in 4m 22s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d634f94597b146aca284505b931a2d25 : SUCCESS in 15m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/effae447468741eca323f9acd9d167b4 : SUCCESS in 32m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4fe53181b4ce48ecb1138c888fa3b91d : SUCCESS in 18m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3373e58def6340d38ddd8d31c13a833e : SUCCESS in 20m 53s","accounts_in_message":[],"_revision_number":2},{"id":"8348f1d4573d3d461fe844c7fd7c5c9ff66221a4","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-14 08:07:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18f5848b95369b770a3999083af74ecd5ee1662c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 09:46:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/290b7a8657c34f9a95343ebf6f7bf375\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f75e7290068e45348561010a9ed5a4d2 : SUCCESS in 5m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/61c24b0ae66045919f85c68a049af781 : SUCCESS in 6m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2c7482e65a34996b4ad6a1abfb2bc3e : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3b8148f41640128e7297fba0e5127d : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/324e0ab14ab945ec99e3aa676da88b4e : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b044b1cef6e34068a1a023fc2af20689 : SUCCESS in 6m 54s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae7c9480b6684e3cb380627adc551564 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/22733f0fcf7b48f893c82de0c662a422 : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c14e2965e45148b888eee9e1da60ae14 : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/40adb1c297de4ce9824bf7d5518ae1c2 : SUCCESS in 16m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5502ed18f8d642c58fb5ff90c7fc4816 : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/109756eb973f429a8d10437b2f5cba4a : SUCCESS in 12m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0377c596604547ea91b878320bc17bdd : SUCCESS in 11m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/db12dd1c8bd348e78cd327ee7c730541 : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3e6d000f84374489b49925591fc1979c : SUCCESS in 1h 10m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a428935c70e241e792ed91131adbe823 : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3f54d90d0914a6e97a64bc1f727855a : SUCCESS in 1h 02m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f6d5093dfccf46278286ff374295309f : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3215ebf5ffed4d53a163dfe0ece92578 : SUCCESS in 1h 34m 19s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/78fd31f03cb54a45b70aa59d4cc17cf5 : SUCCESS in 1h 37m 14s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9fe856f91bb40b2a0eb445fb1256f70 : SUCCESS in 1h 30m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6f8a9eb4f6a4122a33faff726078cda : SUCCESS in 1h 35m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/87d1296f8dba45ef9d729a0b0d110ce3 : SUCCESS in 15m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6396d123278e4a7d9e1bef0174c8ca08 : SUCCESS in 20m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd7d1c228f31477199cca08abb400428 : SUCCESS in 13m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2f4b8cb94d9748fcacbf3cb85441450e : SUCCESS in 21m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e286e8961694478ea4d54f6dfbd270b0 : SUCCESS in 18m 25s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e64a6377d01e40aea0ba3a568d004cb7 : SUCCESS in 18m 53s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/407a249880bc44db8aef11a9c994e612 : SUCCESS in 31m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9837d3a2c200471d93d848f6884b3643 : SUCCESS in 15m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a1b3d8b05d5a4a5d9e727ece466a2d89 : SUCCESS in 15m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c14fde56672a41b68fb5ad7be06eb42c : SUCCESS in 20m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b2b0dfdee466d02bd53b69efe3bd9e484d3d7794","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 09:48:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0e56c10541fce4b1105d10364ba6cc20c6efeab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 13:34:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f27a9d29a969463b80116e1ed088f9cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3acfd03f22e3475ea70b80fad7b613a8 : SUCCESS in 9m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c469addd7e234397928ed87e0bb0af4c : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8242bce636f4c10a1dc190a58a7465e : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3722ac9c76e47689b68ae9c22f6ce5c : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/673da2298e5442c0a5f1c3e7e9b7fbcc : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c0cf737f1a5b4498946611163222478f : SUCCESS in 19m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0dab670292e14594886f1d27007794ba : SUCCESS in 11m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/039cc051cbe34777bf723ee63c7437a1 : SUCCESS in 22m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90261e588b5c4ce0a710295f37186175 : SUCCESS in 19m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/401bd1d6084d429d835823b880100a61 : SUCCESS in 23m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db1e5d891b834cd0a4bb1e9623ae25df : SUCCESS in 21m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/15eb00caddf34a719d82038136952b8c : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6b0043d4504841b19bece3bee5cde507 : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/280d2f3bad96433ca8c3af31f56390f2 : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db423026063d4538ab2e5dd4d29cff3d : SUCCESS in 1h 04m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6115b9cb17a34b5f9d79da9db35096dd : SUCCESS in 1h 05m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/00d2688de4b748debff2953b9b5d80ce : SUCCESS in 1h 41m 39s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ee2e14d38fd54849ad74ef892be51814 : SUCCESS in 1h 34m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6094dd766aa4f95b5ed5b51e9405aba : SUCCESS in 1h 28m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57860e0be0514304a60b89b8c023f6d9 : SUCCESS in 1h 43m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/38064b41fb43416fa4dd8ec67a451b2c : SUCCESS in 26m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4cbbc7d7b47246f580fc94ecc1189287 : SUCCESS in 30m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3adffe75f17d493298d50935ff439583 : SUCCESS in 20m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4a82ec8e07ba4d09a9edae3716c546b5 : SUCCESS in 23m 25s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5f7b1ac2a8e24f7b8e8ba1ec4eaca19e : SUCCESS in 26m 52s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8f8c9b5ab40346e0bd6eeb41b5575a56 : SUCCESS in 20m 07s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37c96483c436459ca05e2fea6fd3cfb1 : SUCCESS in 51m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dea67e946b44918aec87c8fb28432a4 : SUCCESS in 16m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3e90e25e90cf411fbdf42c2eaff33d38 : SUCCESS in 27m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53a724adaf0f472da8a909fc2f0c2e2b : SUCCESS in 24m 00s","accounts_in_message":[],"_revision_number":2},{"id":"19ead8886bdb858d3b3df57b457f6b8623d93173","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 13:34:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"adbd26d831823b1295e8161a2211da68d6267b1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 13:36:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4e5f43b8c2042ffbe44b801af9768cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a43c3a0c31b4b62bf78db8910e35be0 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5dd6d12fb73ad1c17bcf70a8ec6a52341972b16","revisions":{"1890f1750884831cea145e01b5f9f55f59a4bd12":{"kind":"REWORK","_number":1,"created":"2026-03-27 12:11:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/982399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/982399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/982399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/982399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/982399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/982399/1"}}},"commit":{"parents":[{"commit":"5e1d74ae4b123e3695416ddbb0cdcbc807da4fd7","subject":"Merge \"Drop backwards compatibility for template trusting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e1d74ae4b123e3695416ddbb0cdcbc807da4fd7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-27 12:11:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-27 12:11:27.000000000","tz":60},"subject":"Bump all Ansible collections and roles","message":"Bump all Ansible collections and roles\n\nChange jriguera.configdrive to use upstream role again.\n\nChange-Id: Ia2102d48e5cd172c404bd5d03064913a3f4b7b31\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1890f1750884831cea145e01b5f9f55f59a4bd12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1890f1750884831cea145e01b5f9f55f59a4bd12"}]},"branch":"refs/heads/master"},"e5dd6d12fb73ad1c17bcf70a8ec6a52341972b16":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-02 20:21:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/982399/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/982399/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/982399/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/982399/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/982399/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/982399/2"}}},"commit":{"parents":[{"commit":"79bffd0d225a6da20ba00130b0a3b1782744b9b4","subject":"Merge \"Adds dnf_repo_state_overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79bffd0d225a6da20ba00130b0a3b1782744b9b4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-04-02 20:20:55.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-04-02 20:20:55.000000000","tz":120},"subject":"Bump all Ansible collections and roles","message":"Bump all Ansible collections and roles\n\nChange jriguera.configdrive to use upstream role again.\n\nChange-Id: Ia2102d48e5cd172c404bd5d03064913a3f4b7b31\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5dd6d12fb73ad1c17bcf70a8ec6a52341972b16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5dd6d12fb73ad1c17bcf70a8ec6a52341972b16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
