)]}'
{"id":"openstack%2Fkayobe~983013","triplet_id":"openstack%2Fkayobe~master~I862556a4a62482904ab61b36f84cf2bd616c5de9","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-04-14 08:05:07.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I862556a4a62482904ab61b36f84cf2bd616c5de9","subject":"Make Ironic Inspector config comments consistent","status":"MERGED","created":"2026-04-01 08:13:40.000000000","updated":"2026-04-14 13:37:21.000000000","submitted":"2026-04-14 13:34:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983013","meta_rev_id":"0944a0ac1a83f6d8190ff155e1b7ec1c91d219b5","_number":983013,"virtual_id_number":983013,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-14 13:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-07 10:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-07 15:17:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-07 11:34:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-07 10:53:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-01 10:00:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-07 10:53:23.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-04-07 15:17:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"837cbc18819b0cbf1dc11f5f4dc687e933a3a57a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-01 08:13:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3f44af273a4369bec71ddc28175e1a55a1c5da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 10:00:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8187894388784a2c95d6056894235131\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c6d2e84fc048b0a573054490d5a991 : SUCCESS in 9m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/68edf5d7f37044f9907e45ca32a8074b : SUCCESS in 5m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d402f3088d8545e6b7d19e2709ef2e61 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa04ca3192d490894da8790be529efc : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/426b4ffd8c0c4a6b8bd330db9ec58f17 : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1869fe3c076841018b12ae5ec767d5df : SUCCESS in 3m 58s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84f4c0ff136c40388b5e3ef8dcf8d180 : SUCCESS in 3m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/734de72aba2849eb800761a00df32a45 : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97b1c5f20ac6449d87a0b6f049824c82 : SUCCESS in 13m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b2d53e95d94349a9a68c8ef5cb983742 : SUCCESS in 9m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0caa1dc2fc6f4f48a24a043f5cc3806b : SUCCESS in 7m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f4ad7bd0d96c4964ac86f6bf3cc08881 : SUCCESS in 10m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ba74986a77f4a60a66d020fe2a55f02 : SUCCESS in 16m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c2992847fc8442a3858132b8ca3ae9ed : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5991f7907c124d47b2f1135763421fe2 : FAILURE in 29m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e3c40794220341d4b9b854981dc51459 : SUCCESS in 1h 06m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a666c7f2a6849368d25e801d7d7a511 : SUCCESS in 1h 02m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/84d064a22fda4ec098666f482f37824a : SUCCESS in 1h 01m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/21f9d16e35924469bdb796110ac7f268 : SUCCESS in 1h 38m 28s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c5720fabd0f944649f5e92d073b1ccf4 : SUCCESS in 1h 42m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c5746bf38e44774b1d7061800facb7c : SUCCESS in 1h 45m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0c9bdebaedc49d9a836e2b8c8ace1bf : SUCCESS in 1h 38m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a290d2679fd049a69a14f0c2ee0f877f : SUCCESS in 19m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f53905ed6a4940f7811590e490f40962 : SUCCESS in 18m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0aa6df5ead16496581f1e8f9f3721de0 : SUCCESS in 9m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a787125b9e25427383d506ac74ac87a0 : SUCCESS in 16m 25s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/99cbc5ecea384cac9d4045877ddc74ec : SUCCESS in 16m 25s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ca019955ec3947b1a59f72a24090f0e8 : SUCCESS in 28m 00s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce9040626282455590e7aa398e181b0c : SUCCESS in 17m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9338587d1c9c490eb42aaeaf22d43548 : SUCCESS in 27m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2326d45639914c5997f077000a15829d : SUCCESS in 16m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55cbb04a62304d5590b05f2c0066ebcf : SUCCESS in 17m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9f85e2790599b24cf9b7abae8d7cc65f59ed041f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-07 09:46:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"029d19acd5b0d46ca11e803871fb7446d316222d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 10:53:23.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"1ea870f58325cdffe52e3a404ab32d17265e19b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 11:28:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e41dfb869ccf4010b9409ba30f35a259\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa66ba1c4b04620b03e656ea5168a88 : SUCCESS in 9m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27db913f6c044872ba86c3f78c4dd4aa : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a9c5a9487164fec9acfc6fea958f630 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e37e6d2e100e456eb1130343b76777b3 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5dd6208cddae41f3ab405817cb2eb75b : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb9f95f2848243f8b42b3146ecddd93e : SUCCESS in 5m 33s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b279c493d49244c2afec2a87a254e966 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9b5d5fbd244f43aeb21a258204437f11 : SUCCESS in 6m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/285550fb019c43859d8f4afe6f5b470e : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2273ea0d8ce84a0c90fcf67fc853de95 : SUCCESS in 21m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0801aa95ab74c0ca991dc5974a0cf9a : SUCCESS in 15m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b85cd50dc65341479023583d002d80d5 : SUCCESS in 24m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cd436f95a904e93badcd0c989e3c16d : SUCCESS in 15m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/27ce7c52bb4f4e9aa1c79a3a892d7091 : SUCCESS in 57m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/79ac1866050e4508b8c2db58a64029b9 : SUCCESS in 58m 21s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/520e1da5f70f46d99631ef5dca47148c : SUCCESS in 59m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e471d22e66d4729a8b106e5089f263e : SUCCESS in 1h 00m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ea83b0d477f4917a3e15d43641e5404 : SUCCESS in 58m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d4e026f42cae4c1ab63329f82b9965a1 : SUCCESS in 1h 40m 15s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/165965d7026f42eaadbc72bb51df30a7 : SUCCESS in 1h 28m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80c40200419b4d5786d0da5ee6c65fba : SUCCESS in 1h 24m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa7405fccb864d2da5ed0716039f89aa : SUCCESS in 1h 32m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/133bcc3387164951b34f5250464d3d06 : SUCCESS in 17m 32s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1f6eff7118aa4ade854055a5796ef681 : SUCCESS in 15m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcfc358542b0422785dbfcb0c8182877 : SUCCESS in 8m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ee3f5d829b2b42e7ba44c9b47bb0716a : SUCCESS in 17m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/35e42fc892ca460abb0390859292b2ab : SUCCESS in 16m 16s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/939743ea168046b9bf46e03e8e3a3e8e : SUCCESS in 17m 14s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6347a59e4724b53be700d5dd80308c0 : SUCCESS in 16m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfdc14e5cca6426fbee54d47a214e998 : SUCCESS in 28m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a9fbc950c42f401383c7d11ec9d6353e : SUCCESS in 9m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f770746deb9f4c7686285564886df975 : SUCCESS in 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2aad8c8c6a0996e425864064f1f9dc49b29c14fc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 11:34:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99da2f0b9c566f7f88a95b9c348f9f4297594d58","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-07 15:17:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ee80fa3e365a6a5edef2bb0fd38b6dd9f758c5ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:34:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7efaa6a86b7b122963019f2f1fc1d208b9de84de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 13:02:36.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe/+/979998 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"c0a4f000a3885132399283406fc986eb982f392e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 16:47:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb7731eb64459786883bb028b7a2bd38aa4f2b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 18:27:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b21129ebb3024108a7878741e2b84b72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09f53ca9b814a41add2dc954f44ece9 : SUCCESS in 9m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8a3fdb9b5c574490a19bc6dc3cfb8dc9 : SUCCESS in 7m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f15f51e629b443c3acba38b0e48a94a2 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3415b9363264eabba77fec7c4715604 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4c42fca96444f16b3eecdca03bc54b2 : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/507fdf1c75ad4cb39881b973383485ee : SUCCESS in 7m 14s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab02e660c0c64529a0c1d6a3082ec49f : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f5c5e5d1bfa4984b2ea1c95184249f0 : SUCCESS in 13m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/54d7b3d14f8346b8b3eee9b1e5eb377b : SUCCESS in 11m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/02d65ed9b87942819c634c8f8d222b04 : SUCCESS in 10m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c128aa0ccfe841ef9b78428ed5996398 : SUCCESS in 15m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/83997a8d0d9a4190b36ecf07f11b0817 : SUCCESS in 10m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dcb1ecb87124f07b748a6bcf651227e : SUCCESS in 16m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b2fda6a983ba433098d9d40f4a3c5380 : SUCCESS in 1h 01m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4e2c1557711942bd9d01fc8cc6fd97bb : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/86a8543e163d417ba3d0fe656e5c695f : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cefa3fb2118f4e7cb985036495249b7f : SUCCESS in 1h 02m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/11a61e9a7aa0464da9227b4b8044c886 : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/38c53a78ea4a4a2da413c02f75fae152 : SUCCESS in 1h 38m 43s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ea0c3c6cb4e64344a54e020e45811c6f : SUCCESS in 1h 36m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65e38a65e9ab4ce4bc428a8487055b1d : SUCCESS in 1h 36m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13b48ba42cdd4ea8a12210b2cfe0c2f6 : SUCCESS in 1h 37m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cebdd6aebaf24a4f8413e214e9704629 : FAILURE in 22m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4668543ec73848339ae2b4c3cf5e5eaa : FAILURE in 12m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e323eeac3cb1495e815bd21dd9a8c2aa : FAILURE in 20m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/53ccc2ae33fd409591b7d6d4b4a910f4 : FAILURE in 17m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7a79ade3bcd942e5b69d6d0955c1e55b : FAILURE in 17m 46s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ed55c484c81d444c94226077626fc57b : FAILURE in 26m 55s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd55460964964074bd5c4ff34d57aaed : FAILURE in 12m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b73005b58de49dc893fc6187bd5e8a4 : SUCCESS in 38m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0118a59f41e64fda88f8a108ecc79de0 : SUCCESS in 21m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73199506f7cc4a5bb5b737813ab85599 : SUCCESS in 14m 18s","accounts_in_message":[],"_revision_number":1},{"id":"14995c04310d5fd852f8ab35cc642db49aacfd5f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-09 12:35:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"061d02b64deee08affea360cfda820117fbd17ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 13:41:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1b1e0cb174a4dbca34234459e2bc319\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7461e2ad24144504b4b2e452fa91d2b9 : SUCCESS in 4m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bba356fbdcbd41618d3595dd2443ab3b : FAILURE in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd1d6d0fd85f49728381048ebd3482dc : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7decb5852954565a92020a8431d6de3 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92012930355a43abb3d52ce534ec67f9 : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/684a165ed99e4768a1d49e396d4ff6d3 : SUCCESS in 4m 16s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0542ad6074764bc8830348575db73540 : SUCCESS in 3m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e0032eb13e84b1388da9b1c55a9c91d : SUCCESS in 7m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/15e41a702ea84339806a2fda2682923a : SUCCESS in 10m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/74c72f5affc94e9c93e26ee49168daba : SUCCESS in 14m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8737d2930df64a1296af6c27adc4d27b : SUCCESS in 11m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a4ea0c5d5b214be1864c71867f44b61b : SUCCESS in 10m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9aead241f4844b20baa3d0845ae71555 : SUCCESS in 15m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fccfcace28164f44bb941f254e16c374 : SUCCESS in 1h 01m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/388ed86b5a144a1494d976e32c0fe4f3 : FAILURE in 43m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cbdcc5d3b85044af93a829d26d03b10d : FAILURE in 51m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06070abd83c14d11806bc86f69d4446f : FAILURE in 38m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/100464a9074d4986804b068f8f277400 : FAILURE in 39m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/847b30e56cf0496f992b32311f7e62c9 : FAILURE in 11m 58s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b79119fa1c05405594ee492d0e126413 : FAILURE in 10m 56s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72215dcb9c114832ae603b5f801f6f6c : FAILURE in 10m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6665d4812df649bd991a5a689c84626d : FAILURE in 10m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/082547e63ee342f99ffba9deaafbd6c4 : FAILURE in 9m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4d8f513a02f84eb8960fe37f2651842e : FAILURE in 14m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b19ff97c07934d6497bf343474193961 : FAILURE in 18m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa883e94f2814de199888a1dc7500fda : FAILURE in 14m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/39eaf244960f4544965a679bdb2ca828 : FAILURE in 12m 28s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/21c10e6158e04015a243ee166992a077 : FAILURE in 12m 55s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b914ef091d114ffda1bbbee535255a03 : FAILURE in 9m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfd8d1c53d9749daaa21f73d2312d58d : FAILURE in 15m 29s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c6ac128ee6164b18b2f72adc0efec2ed : SUCCESS in 14m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbc6bb6efc914018b2cc17ef34e31cd1 : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ee4b30de4f879cdf769b1222196e5095e3ec96a6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-14 08:05:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3d51590166ee378af4f962ca7e6dd68036b1f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 09:53:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6e5941f6a3748188c4fd497d2fbf3fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d7f8d56d5d4399bdae1851ccac50fd : SUCCESS in 4m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/af5e68b19084485f942b0ab0b644894d : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71c36181ad724f78badd7afba86b1179 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2b57e9d9b234630820d3c375bcda150 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d30e8911fee49b39398aebaebd63ee2 : SUCCESS in 2m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38100d905bb14963beed842a304d5973 : SUCCESS in 6m 56s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9696b46f9450427fa7126aabff196d3e : SUCCESS in 3m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2000c1920755451591ddcaeee0e0d9c9 : SUCCESS in 12m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/95a4eecb2b214a8da7bf4cf6ff60dfbe : SUCCESS in 11m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f1a30b09eada4c26bfeb35938951b512 : SUCCESS in 11m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c979c25acc84475998d82f65735bd127 : SUCCESS in 15m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/48d5ea5965d84d5389e37fd8a24140c7 : SUCCESS in 22m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e60a5a240fae41bdb046f994d4b2ae20 : SUCCESS in 17m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a5a5d6c5f6e241b9bb816321d29e4a69 : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fc40d7bcb63d4b5ead268fc816884a11 : SUCCESS in 58m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1fc93738fb064feb95496692bf833d25 : SUCCESS in 1h 09m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74fc3cc0bb60423dad965bbe8206d790 : SUCCESS in 1h 01m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0f2e1d78b7c641a9bfa14f9c31a9ed1f : SUCCESS in 1h 02m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2e12fcc8ff4a4ad7920fd942d09e5f7d : SUCCESS in 1h 46m 39s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/347ac3427e4e4f9987e605bfce1e9220 : SUCCESS in 1h 31m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6d91c2a54574e6bbc92a7f121398627 : SUCCESS in 1h 29m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea6a7010575449a795406c7008d4f761 : SUCCESS in 1h 40m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/502ec15b30f846d0af2299a86777398b : SUCCESS in 19m 43s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e95be12300614d1b86ebeda1b0339326 : SUCCESS in 12m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02bd45a0e3c740ac92463154956bba9a : SUCCESS in 20m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/37c583497c014fa7b0d9ab0df7187d58 : SUCCESS in 19m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1d62296cf58a46a39bf00b14750305be : SUCCESS in 37m 49s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f80ebc3730484d3083db146189ce1f4d : SUCCESS in 26m 20s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35d88da2e9a949bba931b80f6fb4d6f4 : SUCCESS in 34m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d330ab4a11544fb9f9221dfe4ebc87d : SUCCESS in 16m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b540e04e1efc48bb966f3813845cc3df : SUCCESS in 11m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f3ae5034dc44d9b96dc788a78ac2083 : SUCCESS in 16m 49s","accounts_in_message":[],"_revision_number":1},{"id":"21581ef97610faf6852064b30e4b34a1b854c7a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 10:07:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc7f298f086d23268b5fb813cbfaf98983d4f7b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 13:34:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/362837ab963244f0864d535f153da098\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d788f495ebc4ee2998181dc598fee9e : SUCCESS in 10m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7cc566cccc334a049bffcbd2c85752de : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6576ef9ce442486fa2f19885f8bb749f : SUCCESS in 6m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cd028ceb4204c1f82a19d75a532678d : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1aa7864a91444e4b8f914f2b9d11c877 : SUCCESS in 7m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/084ba04a14b54bce8b2f59f581fd5abe : SUCCESS in 15m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5d7d11f3ac114404b3832175dd16ce51 : SUCCESS in 15m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2bdfc6a211234772aceef60920cc5f0d : SUCCESS in 14m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f12521ba2884bfa9605fa888ad3fe4b : SUCCESS in 21m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/461f33f054be48ccb374939351a0b630 : SUCCESS in 13m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f90f7c532e794c358138229e5894cd9e : SUCCESS in 21m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/399c5ec9386a420894d8c02d2e317d60 : SUCCESS in 1h 03m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b4b40a8fb1554f909dc6dfa55620444d : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/90deb874f987403a8d65c98a2c6a3cc4 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3149f16846ea4d91b21a25ed8788f6e6 : SUCCESS in 1h 05m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b56494bc3558469595c8553d651d97fc : SUCCESS in 1h 07m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/08578ac8a1654f12806a431e13a8f5c3 : SUCCESS in 1h 41m 55s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fdbd420f43dc425fbd5f2a4e322b5fbf : SUCCESS in 1h 32m 22s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f5a412618614cbfb19b8570f03aa71f : SUCCESS in 1h 34m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e2626341a77482680061558c4e938cc : SUCCESS in 1h 38m 15s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/eeecc9de832a41df831df2c176f3efcd : SUCCESS in 16m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a43d1de194824fc3a011431207ebecd5 : SUCCESS in 26m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce3f2e596d524ec190df48f0422b297c : SUCCESS in 24m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/42b85c670d4042c6a7bfc46c55884d2c : SUCCESS in 22m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4eac4e7a38a640bd99fe17f026c3dd35 : SUCCESS in 17m 56s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/500561632325439b826c6203897b466a : SUCCESS in 29m 49s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01214620fc074ba085a2a060d8fd6730 : SUCCESS in 21m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7b4a1c5249240ba8fb0442eac2c64f9 : SUCCESS in 31m 34s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9897ba97cd9e44e4b8253014aa1edd74 : SUCCESS in 14m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ce0098fca4d4fae9efbffdd6007a0df : SUCCESS in 18m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3167b6fec015fec58e1fb99462a4c2a35cc11a44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 13:34:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0944a0ac1a83f6d8190ff155e1b7ec1c91d219b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 13:37:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d86c34539a674cf2bd453e1c28203254\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740f04c459a94388bcdd32dc14292479 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7a640e49fc844e885185704dfe88a32c73a704b","revisions":{"d7a640e49fc844e885185704dfe88a32c73a704b":{"kind":"REWORK","_number":1,"created":"2026-04-01 08:13:40.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/13/983013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/13/983013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/983013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/983013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/13/983013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/13/983013/1"}}},"commit":{"parents":[{"commit":"a8635f89bcee9032594fd0da23db32dda8e9c0cf","subject":"Hide sensitive inspector rules from ansible output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8635f89bcee9032594fd0da23db32dda8e9c0cf"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-04-01 08:12:57.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-04-01 08:13:22.000000000","tz":60},"subject":"Make Ironic Inspector config comments consistent","message":"Make Ironic Inspector config comments consistent\n\nChange-Id: I862556a4a62482904ab61b36f84cf2bd616c5de9\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7a640e49fc844e885185704dfe88a32c73a704b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7a640e49fc844e885185704dfe88a32c73a704b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
