)]}'
{"id":"openstack%2Fkayobe~983565","triplet_id":"openstack%2Fkayobe~master~I3ab75766d23ea06852c494806e5230ad20c779ca","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-14 15:22:29.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"hashtags":[],"change_id":"I3ab75766d23ea06852c494806e5230ad20c779ca","subject":"kolla-ansible: add custom collections requirements support","status":"MERGED","created":"2026-04-07 14:04:50.000000000","updated":"2026-05-14 17:15:20.000000000","submitted":"2026-05-14 17:10:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983565","meta_rev_id":"1cbefe3a9e25301f846d700583a8915ea75eb375","_number":983565,"virtual_id_number":983565,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-14 17:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-14 15:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-12 12:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-14 15:22:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-24 10:06:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-07 15:08:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-12 12:20:32.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"9fc26443bbf428399b56948ef3fc7e17c8872b49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 14:04:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6aa8c79ccd024285238dc984ab93d18695584e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 15:08:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d3bb468b44c41b58b6fbed792a1cad4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364a25a37dc54ec4b34c9a1d009da0b6 : SUCCESS in 5m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a4e6b956dd9a43438d01f7d5450d23e2 : SUCCESS in 8m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c1439d120c44a8eae1fb91d04ad3595 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb6b8d162da4210a4081e8913b68f57 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d676606a9e0448dc92dec999d1a0f291 : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9d478be2d92445ccaaf182c47cd7b888 : SUCCESS in 7m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a09efd3dde054e2ab910c14f9788d57a : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/607e1a3bb6d2454c8be70e0cb6c2fa75 : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/deec0ad161324ceb8e3e182d16b61a01 : SUCCESS in 13m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81415d7769eb4c58ae172172eca1bb77 : SUCCESS in 11m 55s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fd2febc0c725438a87b4de8914db5c28 : SUCCESS in 10m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92928a30523445d995c575ba80c23925 : SUCCESS in 15m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e36971da936a40cc9320a74b4a893dfd : SUCCESS in 18m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39cb27250b064e7693079fab4c9541a6 : SUCCESS in 18m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d1dba3ad034c4c16a9eb869d9b9588cb : FAILURE in 19m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c07c95f66b334a228deeb716bdad133a : FAILURE in 41m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3b2e4ce3c5ff4c9d921003935a590bd7 : FAILURE in 47m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70210721ccc6419195b149e1e1959a60 : FAILURE in 38m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f22c7a0ef38946aea10d0cf1603e6fe5 : FAILURE in 38m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b0ad6750dba4c5884cd8bd5ce3e2e69 : FAILURE in 1h 00m 08s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ece5713caf104a259a7e5459b6dc8f11 : FAILURE in 20m 26s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a06daecb801c44bf915f38ecd0e173d6 : FAILURE in 20m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5b81952dd5b4cc6b92ceaf4a05df7d7 : FAILURE in 17m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0d4cd767c7c146cf89c44ccb328ab41f : FAILURE in 9m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a885c9336e5144a69a580c6537d452de : FAILURE in 10m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32cd66b053164f3a922f17267483ab6a : FAILURE in 19m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dfc6fc053107420db219cd73bd271406 : FAILURE in 18m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/35197c4254a341bfb038ab1fa740d83b : FAILURE in 20m 43s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4f306fa945174232a808ed78242acb42 : FAILURE in 19m 01s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5b9c8f2bfdd46a4bba6a645305bbd48 : FAILURE in 17m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffa03680ef7f454499336c75f293e2c3 : FAILURE in 20m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/206a9c18bffb4b17947b16e7b571070f : SUCCESS in 16m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11cf947481bf4789bd16472238feea1a : SUCCESS in 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"954f1b3448a4822f91d974509b6dbbfc734f6b70","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 13:33:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a849bf00e90d55af2349da95c8df33923b03dd9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 14:40:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de743e2d18b944f7828ee0dc81a6b95d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26996860a93d4d0cbd3b20d4faa6baa0 : SUCCESS in 4m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/39db37299c954a5c9658c153c68dafd9 : FAILURE in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5c95b4ed6f6415283e5c19c892fe351 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b308c957c14623998b0dd5950ae14d : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/271438f49a13437bb59f0ac4918e8807 : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/86bba63c41654e149a6f7cc3583fefc8 : SUCCESS in 3m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0acd4c0bcff4614beb98cb09dec8a63 : SUCCESS in 2m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3fd1dd4296c041e198878acf5705df4c : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/06853478e5c74422b56421a7bf9f1f2d : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef106d87b7ac4f2285fe48a8064610fd : SUCCESS in 16m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/26538ab2cabc4dba92f4995a434811cb : SUCCESS in 16m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afe16973b21f453e843f788de7698fe6 : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1e50bddffe844fdda18f5680832c6ee8 : SUCCESS in 10m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62cd1d26b83b4be69e9dd3429acc6dee : SUCCESS in 17m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a51f9ac1e83f49019227faa3828ca4c7 : SUCCESS in 1h 02m 59s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c8caf29aaa9f494986c69502d1c88b71 : FAILURE in 45m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ddcc4559482d4ef4a295d50e3dcaa595 : SUCCESS in 1h 05m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cef7b7f1703451ca303e96c69c020ae : SUCCESS in 1h 01m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d40c9a5216ca4ca4a432bf8781356a89 : FAILURE in 37m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/92bc872de7134e618aabfca24a91e7f0 : FAILURE in 10m 33s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f13a27281bc542adbee8fbda101435a4 : FAILURE in 10m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/698f56af7c8541f9ac512ddeb62593c6 : FAILURE in 9m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d09677fc065a4e0683099fe9491281f7 : FAILURE in 11m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/74850edc0f2c406b8a200a3c0e563bce : FAILURE in 21m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/85ee477543eb4787a619d332528c666f : FAILURE in 9m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36a5cb65d69c4472a9ecd1c72945ad21 : FAILURE in 18m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5a801aa18cf64cf4bb9177f2c49c8889 : FAILURE in 7m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c362eafb1ea54706b69f52d704893ece : FAILURE in 13m 24s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2fd7de14567c4bf19937403edf328a5f : FAILURE in 14m 05s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc001b7427434ffc80a9fd68fdfdb8f7 : FAILURE in 9m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e930202227dd4009b4ebe9965d8afd6a : FAILURE in 12m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/21aace85dd654460b65de647ae74ff77 : SUCCESS in 11m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ead5ecd72fa41f9b5077f1fa9a300a6 : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d6d8e52105f7ae9dc3111b06750e610341e6e350","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-13 08:39:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94752de7024fd2219758fb3b297e16e1054d61ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 10:34:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7c7abc7d0e84501bdb92d6e1fc76c0c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3de6b38bba481190d9f7834ed10a81 : SUCCESS in 4m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aa26e64d5a2e477db5b243a29943ae6f : SUCCESS in 8m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bad6662607434246a31acfade5bcad3d : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cb0d91138a04540b80e2c4a432a654a : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a976c201029d45deb762bc11123c1877 : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f050a702f3554665914f580d4520c92f : SUCCESS in 8m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8503c5a6bee046ed811443871e6e4b14 : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bf1816566e0b4d31a007c87f796ebbfa : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1edfa93e85504b829cf1924bf1c89312 : SUCCESS in 6m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78a19dd8060c484285fc1e18e3697f45 : SUCCESS in 6m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f3b68511cc6e4d3e9d391328970bf2e7 : SUCCESS in 11m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96f1dc33b5024cb0858362a50b4d3442 : SUCCESS in 11m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4ef346d9b8dc4e7fb2004f32d00d7872 : SUCCESS in 11m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15c42ec92f0c41448c7cf0a823ec43df : SUCCESS in 10m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0f5c3506516f4ff98c16f5e9e2fe5e3d : SUCCESS in 59m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a1238101b6f3402994687929af1a30b0 : SUCCESS in 1h 00m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/275d03a059954ba49f3ed45945851d20 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88ae801e58ad4fe584cd70dc11e68cb6 : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cf0db05076584ca8ba80701b8825d957 : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/60738e80bb6a4e729380b5099695e518 : SUCCESS in 1h 40m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/23275a92f9ca44859a7783a7c2bb05e5 : SUCCESS in 1h 33m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe8f4c0805f94e91950358e3802c18a2 : SUCCESS in 1h 33m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9f979479ee14fe6894532270505a7ff : SUCCESS in 1h 38m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/39eefd68a82b404a8dffa54cdf44551a : SUCCESS in 17m 06s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5c61c916f6fa47f1813b1709ca495ea1 : SUCCESS in 13m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05a8fda016c642039d5a772291df7149 : SUCCESS in 19m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/74bd2b4501bb4bdc910281c340aad0f6 : SUCCESS in 16m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e747e5ca932b465a81dbe5f9fc150fce : SUCCESS in 20m 54s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bbfc5ecec049465ea2875161816efcd7 : SUCCESS in 19m 43s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fe6cea30712465ba73396ee92af18cf : SUCCESS in 30m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3db7bce7ad3343bb86522125b6a3f473 : SUCCESS in 31m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fe36d88ea4764832b4707ecbaa5b3453 : SUCCESS in 15m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d9f989c380c4aec89c5fe4b0ed75836 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":1},{"id":"65e3d0bed6fbbdab49602f603d755c17e9738ccf","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-24 10:06:38.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"829c9dadf550b958528b4ed4a051214a9e11e6f5","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-12 12:20:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"876a0bfbfb0b2ef6de032e7d14aed680817b58e7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-14 15:22:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d27a8059549049476d1eb9a9a7df42132483d94a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:23:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f2d569b65d5dd6e769b8ad883cd27aedd55726e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:10:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db9a65ace1094b37b94781abdc396410\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f27993696248b8b36dd8e7c0246aa7 : SUCCESS in 9m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0176cc588f2248a0aad82955f4c0a289 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626d8ae21d8f4fd392eead013d0f06cb : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5074276cd0d14fb2abd040217ae7de77 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c6603a633494b18b87ab8787566c6f3 : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cf77150f11a74885bda6a0c440720378 : SUCCESS in 7m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ebb7bdd793374d6d923f6698404d72c9 : SUCCESS in 13m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cba6b38835e040578b28d1038c82a4ad : SUCCESS in 12m 27s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/153f7c7d516746a48169ee6fcb30aa6d : SUCCESS in 18m 10s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44dcb341b5b844fcbebcb5c4d5546a9e : SUCCESS in 16m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/09c2cdb347c84387bd681ed2bb4d78a4 : SUCCESS in 13m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6171a9691c4743398eafc899067d4500 : SUCCESS in 18m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/18ee81cf103b4a47a133085e3598b575 : SUCCESS in 11m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb52c7e884654a3ba9654833cd427929 : SUCCESS in 20m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d17b0abadc7a4a5dbe6596f0a8964ef8 : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d86c0928811e4a88a247ac9cf84eb309 : SUCCESS in 1h 06m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d0a84c63e1974a1d86fabc43710169f9 : SUCCESS in 1h 09m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ad06fc6bb764468b606a20d0ba0aeb3 : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/34b4fc3845254da8bbdc1621e13d3cf1 : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/92561455560e41d1bb886f5b4c73a10b : SUCCESS in 1h 38m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b27a0b9ea33e43fcb70198ad0974fdb3 : SUCCESS in 1h 32m 26s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a771951a674451294fdb064c4191fa2 : SUCCESS in 1h 34m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afeaa43d590c49fdbf93a9c59b12a825 : SUCCESS in 1h 42m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ace6681ef5a340c182c5276612b6ad67 : SUCCESS in 27m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9a0cf158606c440a815b49630b788af1 : SUCCESS in 14m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4332339c7b6497e9a93a50cd4d5b065 : SUCCESS in 26m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/93fe5ba3096d4863864d8aa18e76263f : SUCCESS in 18m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/471919e8857d428bacf4757e8bb19ef6 : SUCCESS in 30m 07s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0c216f2c3df84b298d1a7258b3ef8ca5 : SUCCESS in 18m 24s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fa63ab5d8f1417a99972b1cb4a9e786 : SUCCESS in 30m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/193584d1538e41f2ac52cbe07d97cc79 : SUCCESS in 36m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/128b3d88f538460a9afa23ab4958f31b : SUCCESS in 14m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9abb85f3080e4e808ed5c32259ead0c5 : SUCCESS in 20m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6a423d4d2f90d695ee4c894a14fe49a3c696fc55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:10:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1cbefe3a9e25301f846d700583a8915ea75eb375","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:15:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc8896586504f32a55c0e901f4dff99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e57fcf28475c4ac3a11178644de5fe93 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4c0059a36f44a299fba8c2afd336b15 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c2e905bbfb058a38b2921fb9392920fe65d662f","revisions":{"7c2e905bbfb058a38b2921fb9392920fe65d662f":{"kind":"REWORK","_number":1,"created":"2026-04-07 14:04:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/65/983565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/983565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/983565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/983565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/983565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/983565/1"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 13:45:37.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 14:04:27.000000000","tz":60},"subject":"kolla-ansible: add custom collections requirements support","message":"kolla-ansible: add custom collections requirements support\n\nAdd support for installing extra Kolla Ansible collections from\nkolla/requirements.yml files discovered in ${KAYOBE_CONFIG_PATH} and\nkayobe_env_search_paths when they exist.\n\nThis supports use cases such as shipping custom filter plugins in a\nprivate collection and using those filters in Kolla configuration\ntemplates.\n\nChange-Id: I3ab75766d23ea06852c494806e5230ad20c779ca\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c2e905bbfb058a38b2921fb9392920fe65d662f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c2e905bbfb058a38b2921fb9392920fe65d662f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
