)]}'
{"id":"openstack%2Fkayobe~983606","triplet_id":"openstack%2Fkayobe~master~I74dd61b4f521fa29eb3cf307863a48a2e9172b22","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-23 11:52:13.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-22 16:01:56.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-04-23 09:56:46.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I74dd61b4f521fa29eb3cf307863a48a2e9172b22","subject":"Add real world networking example","status":"MERGED","created":"2026-04-07 17:12:46.000000000","updated":"2026-04-23 11:53:19.000000000","submitted":"2026-04-23 11:52:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983606","meta_rev_id":"e68019744aaca62cfb83c55bf7f2830ca0020ebc","_number":983606,"virtual_id_number":983606,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-23 11:52:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-22 16:01:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-23 09:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-23 09:56:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-22 09:50:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-07 17:26:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 08:33:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-21 15:48:01.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"7b719c812f8a3181f655175ed2d1f856b4c4fcd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-07 17:12:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b5d2d89bbcb04c979bc8adca8cebbc20fa19a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 17:26:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0c1d6006754452caeef00ffd9f91916\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c74cb9022646db9143e9b9d69de3e8 : FAILURE in 7m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aad28f08d58a4181a8d64eb46136cb47 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f40a029aa1d4232b8aa23205e48fd29 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f6d938327b2812b204e6993babb103c76ea90721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 08:53:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"938f6bb159c05329ca9a0d5d550cc5b25c9d5149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 09:05:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61b79c6c89d347d49d1bbb166f4d0891\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e83cfb9827c0402abbaf7aeefde8d257 : FAILURE in 5m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/271507acaea143d8a2c79ec7ef032a66 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b290214cde94dd5a945646ed41732e7 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"ebd0ab3fdec742ac69f223f62da70a176866d63d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 10:53:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a9250ed823c293a1c4d8a68869d1fd779b99e106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 11:06:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f59b319a50fc4422be4743326be52c30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c14d5e5d5b4d119304b43f6c213cd0 : FAILURE in 7m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/44cf145346794a9aacae739ad93f01d4 : FAILURE in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd13e9beb08342cd986b8aaf7b086b71 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e35bce145aa76d5bce92127664433bb59a21dd83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 11:37:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e9ea89fb6c023d04c8f707c0867b4712998c4c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 11:50:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cef7232194443db90b9bd92ad364673\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2897fab56644ed948c1800b44657a6 : FAILURE in 7m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0cd25cf8f75b4b43b947d56a894b1fa9 : FAILURE in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52f5b4637e64a7fb2d66c98890d2012 : FAILURE in 4m 22s","accounts_in_message":[],"_revision_number":4},{"id":"31e2c8c710e686f268d2b0a49bf42de88973ff95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 12:17:29.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d771c3111816f82817777f7a160c3ec65c58a9f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 12:18:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"92b870b87584e74fc11a4dd359e40fa87ff74120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 12:29:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba483db9561c454db6cfe205433d3189\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb31119e04849f9a0a19b621c821d90 : FAILURE in 6m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0ea6d699cb884d17bf1b1667032c84e9 : FAILURE in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8c60db9cd742f1a7b60a93a1102bcb : FAILURE in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e28862e35df2463cac6a167ed997f705 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":6},{"id":"4c58289c622f9382c0584a4d5af6c4b0de252d87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 13:50:20.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5c431a9a561c73f5ef965cf8fd80dff3a3faedff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 15:36:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d7e27711f0b40deb3d8b6567bfc0670\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94ff0889b0341f784b6ab9f78b0a7c8 : SUCCESS in 5m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/83b8895e7896454fa4606cfeef2ce90b : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a3f47c35bca40dbb06de13552228e22 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b65f0eaa90040829f756dba7a4fa7a3 : FAILURE in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3929abf0d7249988ac0efcaf8cdbdbe : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a5510a74081d48e7971c9fd30719c444 : SUCCESS in 6m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69553d81629e4f8185ae4fa450a7f2a5 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/446bfb388b55484aab633097e648e72f : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/daad5939d4aa4a8aa0729d893348f239 : SUCCESS in 13m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c01218fe0b5144a1a9240691fb5237d6 : SUCCESS in 11m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d21a66bd958e4fdbbad5aeb3a7528668 : SUCCESS in 11m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be40e76e27bb48e99455062d8bd3be33 : SUCCESS in 15m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/eca3c7a2ddbf41b5b9da209665e2e6c0 : SUCCESS in 10m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03841c45dfd54c39af629f1225e65d09 : SUCCESS in 19m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/60700b84eac74eba8f572d3e84107563 : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bd633aa8c0a74697b7583a12cdf24263 : SUCCESS in 1h 06m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8c9b96fbd1014d06ac44b752f42ef98a : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2631aebc71a048a685a915ad07452364 : SUCCESS in 1h 05m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d73d187a06954c65a6a5bafb5288c271 : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0d3a90c2ec2a44bdb098b040b1a7dac3 : SUCCESS in 1h 44m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ae71ff5f5012442da920f7c846b46cb3 : SUCCESS in 1h 29m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/388fc28d620348f69311e0cf91a4d23e : SUCCESS in 1h 30m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68f138137ff748ddb80a0960a50fdc75 : SUCCESS in 1h 41m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b77ac4f1adff477f8cd0b9143e2cb454 : SUCCESS in 11m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fdd3db8001494dd9a1f8b290761047c1 : SUCCESS in 9m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/497c8785b37b4d1c817dda8daaf09f44 : SUCCESS in 9m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a2b349f413234fe2a2e5f320caf33270 : SUCCESS in 17m 46s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2f5bf6e33b18499186716a7cf4664cc8 : SUCCESS in 38m 26s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d3eb8d4dd76c42318c2c0197613e4429 : SUCCESS in 39m 38s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cd777c869f845778f811f2074c2c003 : SUCCESS in 17m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c98f379a58947b582f4c7e253afb9e8 : SUCCESS in 30m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4c5eb8ece2934ac9b60a3e23a0733c92 : SUCCESS in 22m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6162f7b135b4defb1537dc681639476 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":7},{"id":"5593acf2359a15df6f0f42e01e3ae5bfe17b4350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 17:05:30.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"bc08cb648148ff5270fbb2fdb9a612d13e18f1e0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 17:06:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"18380ae6d54b37a2702b5ac9c27974b2579a9d4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 18:25:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e01f3f41060880c22a1b3d3f85b7547eeee1c587","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 18:27:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4f6744f0745c1de37fbaa7f41a7755645683ef8c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 18:43:57.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b30fd45d52f4945366e93c385ecf3d83d532a86e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 20:00:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81f606afd3754db392cb821c86a25d85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2c1fae4fee44fc98d4b709d0a56536 : FAILURE in 6m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/158a399e8a3148aabe6d9bc86a845124 : SUCCESS in 4m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d459caee1a148138bf73b4da3841b21 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d322eb93d444efbdf3ab6d69af4f58 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f3d494bc3a34a2fa00b83f10909220a : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/06386b9c0b8d4d3a873a74a3ded41065 : SUCCESS in 4m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36ffb9ac2c0d4b4085b6c9eb21b98b8e : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b8eeeb9763d4e3d82662a6fb1cd3c9c : SUCCESS in 3m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6377c72fd69d4d418285f24eec3348bd : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa4c857362de4e8597e6b31ed6f8f4ab : SUCCESS in 7m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2d2c76726d7d4c85883160e054fe5c2f : SUCCESS in 19m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/693e47215c88405ea6e26b7c081987f8 : SUCCESS in 13m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/305aa1af5a4f4745ad3cde74588c39ba : SUCCESS in 25m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b43cc4b7fba483086c1709ba49e68fa : SUCCESS in 16m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f7dbc0dcc51146548f72dba0a2994ae9 : SUCCESS in 55m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7480f14bb8614a9b8d654ad54a1880e2 : SUCCESS in 57m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/29db20dd39c4431b862dae08240deef3 : SUCCESS in 58m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77b6c2097e7444298ff91143e24f8106 : SUCCESS in 56m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/43af16beae474c48a7e8c75281a8b7cd : FAILURE in 16m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d8129195900e46608a4c3ac7a4e8f10b : SUCCESS in 1h 33m 34s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7fb3272549ed4e3d9e55094062771ecd : SUCCESS in 1h 29m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc0531b3495045c3ad3eca58350f63ea : SUCCESS in 1h 26m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c44456bc4d3042b4bdfa11ff69474660 : SUCCESS in 1h 31m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e233ebfe08c64a51aa24d48e618c3877 : SUCCESS in 10m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b79f515933534766b783e991fd621178 : SUCCESS in 15m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee9b88fd0cd14b5bad2a3705fe9ca80b : SUCCESS in 17m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/00f3048faf164c90b5d6ed3c2f827f9e : SUCCESS in 13m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c42b506e0bf8493eb915252edfdb9ffc : SUCCESS in 15m 53s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3b0709edc9d54efaa0a70a802ecd9164 : SUCCESS in 27m 19s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26bdb0badfc74e17aa39f0e6351d3f2d : SUCCESS in 28m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f82e9026de634b7aa1aa3f70e7822ebb : SUCCESS in 29m 10s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/621fd14519c94d0aa931521dab55a62c : SUCCESS in 15m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f425ab56acc94d2493da2f2c72616713 : SUCCESS in 9m 38s","accounts_in_message":[],"_revision_number":9},{"id":"6f7996e9fb8ed16e5672089fceba65a628e5a90f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 08:26:51.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"cbecd0220106ab862e59cb5a568dcc140e0fea09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:09:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b216b1a10634a41947edec0f9ae0fad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4b1c90fdf684b7a97efa6c4c010a98e : SUCCESS in 9m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/63ad670a56c24ed899ba5c32c7495beb : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b75d971f819f422880f08265c59eb916 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74072911c854d2594ffcea3ab20e569 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a97c5f1dce124f8e998b9ba3d017ed9b : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/eb3aa0b90ac842a79c08e3cf918e9934 : SUCCESS in 7m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ae8c96d9faa402294155fc5d48108df : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/635f0d4eb17d458cb29074f3f31db4b5 : SUCCESS in 3m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6df4d348063b401ab02123d926d956ba : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/715ed64ac28d416a90de16b0e211e081 : SUCCESS in 12m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/06a33948be3940abbce10518381361c7 : SUCCESS in 9m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db57917231ab4f7cb71669b2a64339a2 : SUCCESS in 15m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/32ff8ff602004630a88d1e208f62db1d : SUCCESS in 17m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b61f705affb04bc8a81db010ae17ff29 : SUCCESS in 17m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/66424eddab6c45e3a872592a57559a40 : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/70c803e9396c46e29bec8f7c3912ac3b : SUCCESS in 1h 00m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a34c036be7a84606834f01bb25f97982 : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e25c6f630c9a439b9b5bd09c6fcd440f : SUCCESS in 1h 02m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2a6909d2d06148dbab4ebd2c63008e85 : SUCCESS in 58m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5ae92f9cb4a041e2a9b97e434dd7566d : SUCCESS in 1h 36m 07s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/40d22e24d1c445c382206d6133ed2d88 : SUCCESS in 1h 35m 11s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5aa6a8341517406d9cb602e8e04c2130 : SUCCESS in 1h 28m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec404adfda0d41808f123387fd32ea62 : SUCCESS in 1h 36m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2747875187dd4e86807ad05b157de373 : SUCCESS in 14m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/19191079f1624065aaa76799106cc6f7 : SUCCESS in 23m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0595fddd0ec843bda5638501cdc59255 : SUCCESS in 19m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c6ff3cdada554b92b0ad10473d8279a8 : SUCCESS in 16m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bfbbe1a524094612af02db16926f2af4 : SUCCESS in 25m 30s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8783cf9c32cd4d43bbbfebf7f757405f : SUCCESS in 17m 02s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2aa357dbbb6d496bac561b79d647d13a : SUCCESS in 29m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5237d43b3c8b4040ae1d651b166c9ade : SUCCESS in 28m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/37630f9f7169477a96e43c6923be1eae : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59b69afee133476390ba183f873d00fb : SUCCESS in 16m 27s","accounts_in_message":[],"_revision_number":10},{"id":"39ab908e27f0cfc31a5d91a069e306cfc91be722","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 14:51:57.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"40dbdeb003ebcc6a94f0be4890ac6c976d388183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 16:30:30.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dca5324754ae4a2c8e0ed41262b56109\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2d4537c20c4637bb6ff608c3caf24a : SUCCESS in 10m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/409a83250b46437eab4cc071907d4cc0 : FAILURE in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3db1b03e08ca4aa3a0c7c86beb4f2c81 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16ed1698b2af46f3ba9285d88ccd368f : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e4821d918d344d1bd7b87096ca8a53a : SUCCESS in 2m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b14a8aefe7a04caaa5def00a3f4a9518 : SUCCESS in 4m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9da13fd3ff6b43938e4a5873c0430be3 : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b3f8bccdad1540f39b65f6e25ed8bb07 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da4df6c7cc2f46f1b2f8c32e8b69c5b6 : SUCCESS in 6m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9691ece257284193959d1b662df44e55 : SUCCESS in 9m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/da8b63a900f9424ebac9b34a29868077 : SUCCESS in 8m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbb5e1fe0ca341dd8c3b0c7c56a7ea8f : SUCCESS in 14m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6f9503ca450148138c94b79872a9e07b : SUCCESS in 21m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0911dfb2a6f34c5bba5ad863cd445d69 : SUCCESS in 18m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fd0f3b055ad148898625840331355522 : SUCCESS in 58m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f33b4af9012d4fbc853cb63dc504cdfd : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1cd059f69f36491698ba9ace936548f3 : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd8a369c2a3c46f48ce2ee5fca742ad6 : SUCCESS in 1h 02m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a584a5a6618f4c27930e8f326cb5a346 : SUCCESS in 1h 03m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9cd4ca2955ef4ecfbbc161fc67333d53 : SUCCESS in 1h 35m 57s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1e1a72ffc06a44c5864536dcedc1893b : SUCCESS in 1h 37m 07s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eec5f14941074f4fb904bd63ded028cc : SUCCESS in 1h 35m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/894b8fe5f0c1407e844285b42fa7ac91 : SUCCESS in 1h 30m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/88e7b691026547e08417a3c1082930b6 : SUCCESS in 21m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/285873b709f8492299aa68da908f7ad2 : SUCCESS in 19m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b26eae8b0a1b4b85a8906db27dcc3dfa : SUCCESS in 18m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c41bd3b6a4514a0f944b22325a6cd845 : SUCCESS in 15m 59s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f5232f3eb8c44e9688680c6720a5b1c2 : SUCCESS in 19m 21s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d43d0f7623c747cc9392606d4e07af3d : SUCCESS in 40m 00s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5aff00a027a407fb594aab51e4ae060 : SUCCESS in 30m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e694013928f479db8b34cef57f8acda : SUCCESS in 15m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f83508ca7e77428db979fba7f1c3da02 : SUCCESS in 17m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc8a94453b03458cacfde21ab08720c9 : SUCCESS in 11m 38s","accounts_in_message":[],"_revision_number":11},{"id":"fc09b7bd4a4718be113cb0111b1903a718759acb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-10 09:46:24.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"bac8faaa984d850256c53e182252b2f80d4846d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 11:25:57.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6f79b95e8a140189fd9a83a79238687\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4688d65520d46cebcdacccb9c5477b9 : SUCCESS in 7m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f432ba11868c4d42bc75f082dac2960c : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b42639ffa304e3bb60caf84003814f4 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c84e188d13b4415bf0a3ac181a0a3fa : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c481ab1c9a3d4ae19ac6db54897e472c : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a9960deda292417eac2d72633e0f53d3 : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/429beaca36234362af81083ccb321e2c : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46b284e0909d4bd4bd1ecbd4a152709f : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a5866f943b14dbea2ecf3cd1bee6fd2 : SUCCESS in 11m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36f80f5a9904496d9157647c95a16e13 : SUCCESS in 10m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2507d6cad16e4682be09655ce47d1945 : SUCCESS in 9m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/790518dda3fc4ca082d2e4a06e16c234 : SUCCESS in 17m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4d6c097260ce448b94a7541ffda91254 : SUCCESS in 24m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/284bf1a0257a4d8fa3539291fd16422b : SUCCESS in 15m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/95010d1412854416bab0f0a845984997 : FAILURE in 47m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4d3eb15ef7a241a3b9ab6abefc6c9e46 : FAILURE in 46m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9242652d0d7445a78e17d7a9d09dbb05 : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b76b6de359ee499c958852745d549f6b : FAILURE in 44m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c04cb144fec04ffc960e051728ebd125 : FAILURE in 44m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/da1d8fcb24fc4c5b98c486db95ce6be2 : SUCCESS in 1h 32m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e48a4e5c60cf47749126bd847b50575a : FAILURE in 4m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6fb84fdeb6449bebba46a5bdc00425d : SUCCESS in 1h 28m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cb6beee3905404a8c990f1a11494622 : SUCCESS in 1h 36m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a6af05253df742328f96ef30b4f96673 : SUCCESS in 24m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5534743b5a384ab49bceedddee13c5d3 : SUCCESS in 20m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/524e4bcee3c34b30b8d24fd4733e35e2 : SUCCESS in 20m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/695daaf814204b4286673d8ada2840ae : SUCCESS in 18m 27s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c914e24f07c849ac9a637981cad7edfa : SUCCESS in 30m 21s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/740fc2ee34a74cf19cf08541571344f1 : SUCCESS in 21m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc82091931fe4b268938a1d8bb6bc589 : SUCCESS in 29m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2025e1546f1346308bca246a0942c45d : SUCCESS in 30m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c0f3d1d7cdfa4d2a94b58914af4aea9c : SUCCESS in 21m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6719cb98cd8549f3b26073cae0be989e : SUCCESS in 19m 25s","accounts_in_message":[],"_revision_number":11},{"id":"8fb6be7c44048af2ba62195bc758225d6afb4b6e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-13 08:29:14.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"cb8ec0052870d6ce69317eae0a545e7ade55c7a5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-13 08:29:30.000000000","message":"Patch Set 11: Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"0db8c377bb097563768a1d87ff5e5c555e7a2a75","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-13 08:33:40.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"6fbad36ad601d63419c6e8e9676a9cbe961b17f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 10:25:49.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/91e0faa4a47846a6a4c009ecb58a6f31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fa54c24d1a41479da42eac369e4857 : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/600357f10a3540b0ba341fd533fa41ee : SUCCESS in 6m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acf213ad0e474a6dbb297a93ef51294f : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4cc6c20f11e4fdea552dd67187cf178 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/246bc0dcc32f482c95913672053fba79 : SUCCESS in 5m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d9588ed1f9ab4bc490fe7633cf55a87f : SUCCESS in 8m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d3eebb96ac24b99ba69e1ccf962fc5b : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3bce168adacb46db8e9e06b869b54b02 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/440965f7f7d1421097b20da0d6f7f002 : SUCCESS in 12m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/adb349d743d34f068801c7c38b5833c4 : SUCCESS in 11m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/290a0a38bdf647b4be6c6965647d172b : SUCCESS in 17m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af1c517c37c547bdb80f32ae5f2bb0ce : SUCCESS in 18m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4b573a698c1e40309519cf22b586d6b1 : SUCCESS in 24m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6edf6e0103664dc3a4f0e5eb8a647c4b : SUCCESS in 17m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/316ae2a0da064e14a24c997c6139f935 : SUCCESS in 59m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c2054e58e16844a4b042ea7a02a4758b : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/260068ddf2e54f26b80fc444bbfcd852 : SUCCESS in 1h 04m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74c70f3bc4d94f4db9fd7e90cdc3e4e8 : SUCCESS in 1h 06m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/abd6d98eeca842bf8319c2be3103d622 : SUCCESS in 59m 34s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d53fb1fd67c44a80999ebc77b4b13b99 : SUCCESS in 1h 35m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c5838e7238c542fb8103e59899a4b971 : SUCCESS in 1h 31m 34s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24b53f3b18ea4661baac890f4650f0ed : SUCCESS in 1h 35m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50c6fd1314f340819711713575cd8bc7 : SUCCESS in 1h 38m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5438f010bb584ba08622c1c3e452f713 : SUCCESS in 19m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c5eb6b2743fe44638a31f7a86ca83dd0 : SUCCESS in 19m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c87f1c84dbe462981657b5f516bcb85 : SUCCESS in 23m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/29dd01ed4aa3443d91e916cba80ff1d2 : SUCCESS in 17m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7057ab8849cf43b9b88744a9055aa73f : SUCCESS in 16m 57s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7ea9ed823e7a46589647d24b43aba6af : SUCCESS in 18m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b49d344338044edb87a679d3f45653fe : FAILURE in 15m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebbe8ac0523b4a25a3b99a21a06b6768 : SUCCESS in 32m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a9ddea4d7a864500b61ee3c9001410c4 : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3384200969543379cf73c4f262dd71a : SUCCESS in 20m 55s","accounts_in_message":[],"_revision_number":11},{"id":"9634d39494ef414e3dea78d062eb90cea57252e0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-13 10:37:13.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e1c150c34dcdacbe94518d2bd29b35de43e7de26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 12:18:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef3df09f237f4859ac98d522b1c7050d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3a7571e6c7a4c909ff893dae96abb1d : SUCCESS in 10m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9ae5fea7fd0344e185120b2b2565ac76 : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/211cad352461417fad071f6d992df34d : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbee582ebcbb4093841f3ddd1be7441a : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12bd2aca2b144bc199b208f471b4d90b : SUCCESS in 7m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb72d21bd5af49be9a708f3382028c19 : SUCCESS in 8m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42f6a445d5384a9e92d4d711fa0bc7aa : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/02fae64f4ac549a985a5b01f375d646d : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e45b855e310494886214967dde4aca0 : SUCCESS in 14m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9d287e5386824b648af5925bdc495e44 : SUCCESS in 13m 16s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/249e2592de3d45f191a354d6c962ea49 : SUCCESS in 20m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd63ee99859b4280bc59ce87bb707ec7 : SUCCESS in 16m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/eead0c5ad0e14478b0475dd6fe0918f1 : SUCCESS in 24m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd2bfd6aca214bfd81c648f294109bf8 : SUCCESS in 20m 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0a9a52e46ffb4e9b9ad0836ffe1ee8d5 : SUCCESS in 59m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/49008f95e6414d75a637388b760a11f6 : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/932acaa8d1f844dfb48d2e320d26f0f7 : SUCCESS in 1h 03m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4fc226b342b49b6a31147fec8967fae : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/595e2202a46c4670a840748a8fe33cfa : SUCCESS in 59m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/68ab0f0f7c4f491eb530299991898db9 : SUCCESS in 1h 37m 14s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/760cf0d7a7924584bd69df51565fbdae : SUCCESS in 1h 38m 02s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6aa85bcc6c95463db1a492fb6a831066 : SUCCESS in 1h 26m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8998071cdd2b4221a4386d802f93767c : SUCCESS in 1h 38m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/665c21c8b5b64aa5b2c5e1f837d8ea2b : SUCCESS in 23m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d06a0716438549fbb3a6e1b1c2752692 : SUCCESS in 19m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b808731f0c045cabbec9df8046600f6 : SUCCESS in 20m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6de4039b1eb44930bfce154fa752d840 : SUCCESS in 16m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6bd62870d8414e10983f5b7cdfb07b7a : SUCCESS in 35m 57s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5cf991d88e70443d9a130d871d89ec1d : SUCCESS in 37m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ce233f7188e4f3da21b95cad7dacf0b : SUCCESS in 36m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c12394eacedc44cf97d6ac36bd772bdb : SUCCESS in 34m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a3795569c3ad4dbcab73104ea5b89756 : SUCCESS in 21m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/809e82fe1d93451bb220ae438a06ef86 : SUCCESS in 18m 25s","accounts_in_message":[],"_revision_number":11},{"id":"8d97ca5a6bbecdca0b58fc176c0448a4ea9d70ad","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-21 15:48:01.000000000","message":"Patch Set 11: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"779cf27e9e59a7fb287e36e0c436e15dfc31f0b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-22 09:50:07.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"6776d1ad792c212bacb19380b987c3d6e25e0766","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-22 09:50:44.000000000","message":"Patch Set 11: Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c70de9c24704b44f2f0d8bd02c73a0b1ed3f509d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 11:30:26.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a529ac165c4d431b85c217b7076cd763\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8705f1cd3ba14ae08fb470d496430489 : SUCCESS in 9m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/47d7bf726e7c40dfa8589a6b97830d31 : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fb5a6e6e46444688509128c6e0d37dd : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/babc17c58e2e4f28be561573134300cd : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c6ad987f038449db68b997e5cbcf78d : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71d469f85cd946b28e6a56b5e2a79f54 : SUCCESS in 4m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/343025e1afda4378a0875456543b55b7 : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/37ede94f9320411e9557f2b00120c632 : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03b527a2b5174671b28e438c17309590 : SUCCESS in 8m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb159e01429c4c94a9e7cb37e42ee749 : SUCCESS in 11m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e7f5791969ec4f7d86febb12c7dee140 : SUCCESS in 16m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b979ef1e8be4d54ba2f50e5b1386bd6 : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/43b6433b425747d598860c4f57f1e38e : SUCCESS in 26m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03ca9dff87b7430ea5eec2c415c1a80d : SUCCESS in 18m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c54670625ca2478997974ceb967c427d : SUCCESS in 1h 13m 17s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ea81996e38a84d4bbbde8426ddbb0431 : SUCCESS in 1h 14m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/dd64f43c578b420bb010006d65a486f6 : SUCCESS in 1h 06m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86ee2ad1f7bb4504a34ba89614e1e0a3 : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ee025213f283498dbeed5d3eeeb88bc1 : SUCCESS in 1h 08m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e93a859fbeda4983b49a3b5a40728909 : SUCCESS in 1h 36m 46s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/36ddccee936f4fb4b519661d4ecafdd5 : SUCCESS in 1h 36m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52390615fb6340248debf2ed2e30ce71 : SUCCESS in 1h 34m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7c394a0f85548708cc738f6498ba634 : SUCCESS in 1h 38m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2526e32370984c25aaf0231a8ddb3dd6 : SUCCESS in 21m 00s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cc622c5bad5741ddba1c786d3e73e373 : SUCCESS in 19m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/035d2d6eb0764fe1a83e49d752543074 : SUCCESS in 25m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e89633e0a3aa4b088e6a34d87c19b17b : SUCCESS in 16m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6854d90f70f04b62a2cd97666f693666 : SUCCESS in 24m 51s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2615d0603e7443a09d1f01b82df0177e : SUCCESS in 26m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16bc7d0cfec94444a0d36abdce897d03 : SUCCESS in 28m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e755829353fc4f16a8dd6044ef385dc1 : SUCCESS in 18m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/679a06dd0d054517bbb441b218d4f4a0 : SUCCESS in 28m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9382c21358ce4bb590bd2c95ad23202a : SUCCESS in 23m 05s","accounts_in_message":[],"_revision_number":12},{"id":"134902d13cce84dfe3f8b0cb6e20c47aa2cdb1a8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-22 16:01:56.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"b7fc7b09ded02ee92b7d5b522253b6303dbff7aa","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-23 09:56:46.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"e583083efd0b0173c53ed2d00ba0e321d8581afc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 09:57:04.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"92fa5a967ebf935d6750778969e05199089c3797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 11:52:13.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b47cd3d84fe48cda4795a35a9062400\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da176883b5bc4d2c90b148ea1effe43c : SUCCESS in 5m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ea453d07f6c540a186bafa54990cd9e7 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bea6fb3a67a4c789825bb107e333b5d : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8776a0202dbb4e239b02e4ea6c317d52 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46b15aa8a06f4b8faad216e640639993 : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a35c34193a1743f3b0cfb8b290e9cc13 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/146d90ab4fde4452843a81d183166f61 : SUCCESS in 7m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7de336758c2e4bd596067285525850d1 : SUCCESS in 11m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aa0cc4b26b884a97ac175f2838a36308 : SUCCESS in 19m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55f332df841a416ea8fec37110db53fd : SUCCESS in 16m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/11fb997261e34c12a98f2ade669d4f4f : SUCCESS in 16m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7972a0c6f32743f7a52b4fef79d1b242 : SUCCESS in 11m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b83186b8f71849839691d931a30fe3f4 : SUCCESS in 1h 12m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2f21670ebfb0451a89491246843dc834 : SUCCESS in 1h 04m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6b556e65412745668174c86ec16e1691 : SUCCESS in 1h 04m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/111dbe3bbea044f2b0fb11060fec4320 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f7fc2506934b431298a0065e685b9cdc : SUCCESS in 1h 17m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2d09e0eb27b547c1b4d10f9509aac395 : SUCCESS in 1h 54m 13s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e1cee88df51f4743bff688384552d64b : SUCCESS in 1h 45m 35s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8030f94f16aa405da540ab014a52a1cc : SUCCESS in 1h 43m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f67682facb3436e8df12e99bc88c7f0 : SUCCESS in 1h 38m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7a3d62bd0d5a4d37a63ca143a0ccca74 : SUCCESS in 21m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/20beb11362a849458da9e8b0e2437234 : SUCCESS in 24m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82e6552ff5e742b79c67abdbfb21be8d : SUCCESS in 18m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9d223b15bcee4a3394898e4b89475aaf : SUCCESS in 20m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f610154218b04ab68029bca8e06dc295 : SUCCESS in 17m 53s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ca0ae782ad2545f5ae8099c564dd5890 : SUCCESS in 18m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84ee63ef67594236a9ca4fbe7eb0b8aa : SUCCESS in 34m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67bf60f358174c52b31aaba3013ed1c2 : SUCCESS in 17m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9c03fceface6406f8227142bd7413871 : SUCCESS in 15m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4d0cb4e2f9845d19461f5fe6ca0945e : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":12},{"id":"760ececdd806b36701b09edd0afe078116db03cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 11:52:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"e68019744aaca62cfb83c55bf7f2830ca0020ebc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 11:53:19.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/107e1eaf5d57402b89903e50948675dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33b7e6da78a84cb59f32c9e1ac8d8106 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dc0960801674e8c8778de6d96ae9a29 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"dae8cd6a9868cd903d2dcd3ad7d2b0681f672ad8","revisions":{"b6cd9fe7f1658e0ccbbdca93a7a9a5d637f1ee37":{"kind":"REWORK","_number":1,"created":"2026-04-07 17:12:46.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/1"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:12:15.000000000","tz":60},"subject":"WIP: Add real world networking example","message":"WIP: Add real world networking example\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6cd9fe7f1658e0ccbbdca93a7a9a5d637f1ee37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6cd9fe7f1658e0ccbbdca93a7a9a5d637f1ee37"}]},"branch":"refs/heads/master"},"4a4886ce9ead89e72c65bd1d5f373e487e77eed3":{"kind":"REWORK","_number":2,"created":"2026-04-08 08:53:30.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/2"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-08 08:53:23.000000000","tz":60},"subject":"WIP: Add real world networking example","message":"WIP: Add real world networking example\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a4886ce9ead89e72c65bd1d5f373e487e77eed3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a4886ce9ead89e72c65bd1d5f373e487e77eed3"}]},"branch":"refs/heads/master"},"9fa9a666b8576601e6828d3ca3750e5402359fd4":{"kind":"REWORK","_number":3,"created":"2026-04-08 10:53:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/3"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-08 10:53:48.000000000","tz":60},"subject":"WIP: Add real world networking example","message":"WIP: Add real world networking example\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fa9a666b8576601e6828d3ca3750e5402359fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fa9a666b8576601e6828d3ca3750e5402359fd4"}]},"branch":"refs/heads/master"},"d281ae2d5e4edb3dc3a6a4cbde0e4760d1f2d02f":{"kind":"REWORK","_number":4,"created":"2026-04-08 11:37:49.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/4"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-08 11:37:39.000000000","tz":60},"subject":"WIP: Add real world networking example","message":"WIP: Add real world networking example\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d281ae2d5e4edb3dc3a6a4cbde0e4760d1f2d02f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d281ae2d5e4edb3dc3a6a4cbde0e4760d1f2d02f"}]},"branch":"refs/heads/master"},"59338357ea5b78a4931b6640858790f2ed5996f9":{"kind":"REWORK","_number":5,"created":"2026-04-08 12:17:29.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/5"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-08 12:17:21.000000000","tz":60},"subject":"WIP: Add real world networking example","message":"WIP: Add real world networking example\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59338357ea5b78a4931b6640858790f2ed5996f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59338357ea5b78a4931b6640858790f2ed5996f9"}]},"branch":"refs/heads/master"},"a88be50eeac1230f9caf20a233966d02f28ecf68":{"kind":"REWORK","_number":6,"created":"2026-04-08 12:18:25.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/6"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-08 12:18:20.000000000","tz":60},"subject":"WIP: Add real world networking example","message":"WIP: Add real world networking example\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a88be50eeac1230f9caf20a233966d02f28ecf68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a88be50eeac1230f9caf20a233966d02f28ecf68"}]},"branch":"refs/heads/master"},"741391d43d87e79486ff56348a0ac8ad4ad5698c":{"kind":"REWORK","_number":7,"created":"2026-04-08 13:50:20.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/7"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-08 13:50:14.000000000","tz":60},"subject":"WIP: Add real world networking example","message":"WIP: Add real world networking example\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/741391d43d87e79486ff56348a0ac8ad4ad5698c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/741391d43d87e79486ff56348a0ac8ad4ad5698c"}]},"branch":"refs/heads/master"},"79c0de0ce5867cf1423a8339be93ce52a9e22426":{"kind":"REWORK","_number":8,"created":"2026-04-08 17:05:30.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/8"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-08 17:01:59.000000000","tz":60},"subject":"Add real world networking example","message":"Add real world networking example\n\nPeople often complain that kayobe networking is complex to configure.\nThis example adds configuration patterns found in a real deployment.\nThe hope is that the same patterns can be applied to new deployments.\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79c0de0ce5867cf1423a8339be93ce52a9e22426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79c0de0ce5867cf1423a8339be93ce52a9e22426"}]},"branch":"refs/heads/master"},"f007271a9b7ac27bb5b08b4906540cf8bb6d044a":{"kind":"REWORK","_number":9,"created":"2026-04-08 18:25:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/9"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-08 18:24:57.000000000","tz":60},"subject":"Add real world networking example","message":"Add real world networking example\n\nPeople often complain that kayobe networking is complex to configure.\nThis example adds configuration patterns found in a real deployment.\nThe hope is that the same patterns can be applied to new deployments.\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f007271a9b7ac27bb5b08b4906540cf8bb6d044a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f007271a9b7ac27bb5b08b4906540cf8bb6d044a"}]},"branch":"refs/heads/master"},"c51826830f4e01737eab105e960cf102674a782e":{"kind":"REWORK","_number":10,"created":"2026-04-09 08:26:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/10"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-09 08:26:44.000000000","tz":60},"subject":"Add real world networking example","message":"Add real world networking example\n\nPeople often complain that kayobe networking is complex to configure.\nThis example adds configuration patterns found in a real deployment.\nThe hope is that the same patterns can be applied to new deployments.\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c51826830f4e01737eab105e960cf102674a782e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c51826830f4e01737eab105e960cf102674a782e"}]},"branch":"refs/heads/master"},"d296e55626051f4057170012c5cd9207daf469aa":{"kind":"REWORK","_number":11,"created":"2026-04-09 14:51:57.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/11"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-09 14:51:47.000000000","tz":60},"subject":"Add real world networking example","message":"Add real world networking example\n\nPeople often complain that kayobe networking is complex to configure.\nThis example adds configuration patterns found in a real deployment.\nThe hope is that the same patterns can be applied to new deployments.\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d296e55626051f4057170012c5cd9207daf469aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d296e55626051f4057170012c5cd9207daf469aa"}]},"branch":"refs/heads/master"},"dae8cd6a9868cd903d2dcd3ad7d2b0681f672ad8":{"kind":"REWORK","_number":12,"created":"2026-04-22 09:50:07.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/06/983606/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/983606/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/983606/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/983606/12"}}},"commit":{"parents":[{"commit":"e4f5c4cfbcfc828f336db7329acb4534e388f177","subject":"CI: Replace use of deprecated ansible-lint option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f5c4cfbcfc828f336db7329acb4534e388f177"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-07 17:08:36.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-22 09:50:00.000000000","tz":60},"subject":"Add real world networking example","message":"Add real world networking example\n\nPeople often complain that kayobe networking is complex to configure.\nThis example adds configuration patterns found in a real deployment.\nThe hope is that the same patterns can be applied to new deployments.\n\nChange-Id: I74dd61b4f521fa29eb3cf307863a48a2e9172b22\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dae8cd6a9868cd903d2dcd3ad7d2b0681f672ad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dae8cd6a9868cd903d2dcd3ad7d2b0681f672ad8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
