)]}'
{"id":"openstack%2Fkayobe~983712","triplet_id":"openstack%2Fkayobe~master~Ie7b194dec24e402949d01da745ccb4743fa6c567","project":"openstack/kayobe","branch":"master","topic":"dockerForInfraVM","attention_set":{},"removed_from_attention_set":{"38288":{"account":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"last_update":"2026-04-21 15:08:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie7b194dec24e402949d01da745ccb4743fa6c567","subject":"Add infra-vms to container-engine group","status":"MERGED","created":"2026-04-08 13:10:16.000000000","updated":"2026-04-21 15:09:49.000000000","submitted":"2026-04-21 15:08:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983712-dockerForInfraVM","meta_rev_id":"dbd175ca809acdcda5416fc4d2fe100680f846c9","_number":983712,"virtual_id_number":983712,"owner":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 15:08:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-17 17:27:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-08 13:35:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-08 13:11:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":2,"date":"2026-04-08 14:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-17 17:27:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-08 13:11:14.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2026-04-08 13:35:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-08 14:29:34.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-04-08 14:52:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-17 17:27:05.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"c2b3241f9516726cbff67ad12841088e3800b63e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-08 13:10:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aeb974937a0810110203612217246b9d25b52256","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2026-04-08 13:11:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"150a86699e894da97f0b61a603bd2392866d0e51","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-08 13:35:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b2f54216322ee308438cc0775cd2b1bf602810b6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-08 14:29:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5c0d87e912a9bd4b5d915038f57c278aebe232e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 14:52:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c9fda2ce5c846a59d28452697a3d016\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d83145058e4924af697e059cd7b303 : SUCCESS in 8m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e0a5492c98e84bcb9d3cce225ca4e286 : SUCCESS in 5m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f9f1b756d2b4ae3930cf9f164574659 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9edd1813984347f4b06d08872cfbea77 : SUCCESS in 5m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/053ac1db3ad64ddba42df8c8822feb84 : SUCCESS in 4m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e5c8c72b40ea4a4c86a244623f62c798 : SUCCESS in 6m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/385107a2405d423a8ba7ee12654655ab : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9ad8e5e1419f48289872c5162b29fae3 : SUCCESS in 3m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09a50603955943308603812918e2647c : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9924fcad909e4d208a9d180f28b36891 : SUCCESS in 8m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1bdf33f9786047a68563e629f7ec648d : SUCCESS in 20m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38612209ec2a42d99fdcd0b50fa5c0b7 : SUCCESS in 16m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/266bcc69d0044579b983d08560e48b78 : SUCCESS in 13m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cff48a25be704201a00391a288940cfb : SUCCESS in 21m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d6fe01473db04c02b2ca703de79a02c7 : SUCCESS in 1h 03m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f695c415b38a4c7c9d69dc6f7fb53f1a : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4c92346c32e64f58bf6b2126858dd4b8 : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6441fca57a844e2c9fa4851865fb2967 : SUCCESS in 1h 00m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4dfe5111bbc0406192bb653566d799f9 : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8742b179feb2427092a202eb3d405f89 : SUCCESS in 1h 40m 40s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2f4093e06ab24266a0fe6c6effe372e5 : SUCCESS in 1h 34m 31s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/200b98f88c7d45b18ac3f402dba7fd51 : SUCCESS in 1h 33m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be109b1a417a426aaa0be2c24a6a0526 : SUCCESS in 1h 38m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8897b3e06631457ea25e571dfde02e29 : SUCCESS in 13m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/953edea57783431dbd92213d118a93ed : SUCCESS in 18m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85650852e5bb427ea97e40d1120836f5 : SUCCESS in 21m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d3bd574116124d45b049aa5066dd5369 : SUCCESS in 11m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2ff6b311329a464886917defd95b79bb : SUCCESS in 26m 04s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6cafab78a14d423780b65497674b0fe1 : SUCCESS in 19m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed38099dcf974a2c9daf4c95f44e35d4 : SUCCESS in 31m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a870fd2515384dc1859b43a6d604c624 : SUCCESS in 27m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c8317fedf0e0457485f7f2b236ee4414 : SUCCESS in 16m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d71ee895f0304d239505702b2c3637f2 : SUCCESS in 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b65bb86bc2adafdb6acbff33d70f912317a84a6a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-17 17:27:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5eee6b1ab2f3015a6ef41c6324b297fbfb10c21c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 17:27:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14b65012c75ad7e98993a4ba5f4399903b4812d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 19:24:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08d5840665834c8681648a91916db5e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29755376979249aa9d0f29984075daf6 : SUCCESS in 5m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b1b2d9636847489782d4d45a72a01a4c : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02edd165a19e4290b1f639bb9fa55a0a : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d55fef758f242048ff786d4fc81b4cd : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39b0401ba6cf4033810e0d22c9459d6d : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/591d6bb738c14f22aa6dc2325ec48c46 : SUCCESS in 3m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28a50b3acc974d1196a47837eb0c1893 : SUCCESS in 6m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4d7296e91254832ab1bfaba21293f8c : SUCCESS in 7m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3d7da799ee9c4280a3b2ffbb5df20529 : SUCCESS in 20m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b78d849c9c642ef90ae448f2f92d513 : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9dedf95d89b2406a830f0d09763e952c : SUCCESS in 11m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f218f09efae442a964595c5602c4b01 : SUCCESS in 11m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b74ac3b312264fec91f9271bd5d6964e : FAILURE in 29m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f13c9d0e86cf4de293fb131a2b18c78f : FAILURE in 19m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ab9dce11e96f4f5dbab8451596f87f4d : FAILURE in 32m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/678a921bd7b640dd8f98bc8b72b8cd81 : FAILURE in 26m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dc35c0851dd04c13b87685ec9a2985a2 : FAILURE in 25m 52s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/964e540d02ee45bbbf5611f4720ead59 : FAILURE in 1h 06m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2c28749f24994b22ad00c209af0f4703 : FAILURE in 1h 09m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eed69d6d4eda4879806dbeb1def6039e : FAILURE in 1h 02m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d4620c9cf2741cdb6045e1655c46751 : FAILURE in 1h 19m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/94e590ce7e5941cf9c01f720040f5f54 : SUCCESS in 20m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f6a14b6f2481439b944f56baad6116fa : SUCCESS in 15m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90fbfae2a1be4da6830677fee85a3636 : SUCCESS in 23m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/da3e9f1c577b43de99c969279a26b010 : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/03462d46a5f548998653aa14f792e99b : FAILURE in 8m 58s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5d4be28e26b64cd1b00eb80d76da7c4d : SUCCESS in 30m 42s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c3499d3215d4262ab8e0152b709426d : SUCCESS in 16m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd3fc6b5e2c840faa522c88ee48d560d : SUCCESS in 14m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f715c3e2aff14739a927de29584d5132 : SUCCESS in 21m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb7728f0a92f47a3834587305c2ba9d9 : SUCCESS in 22m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b0c599488cf78a597d55149bffa66c213411e76d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-21 09:13:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"157ac037795f947f2da6774688c358ede639879f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 11:37:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3b10776d62e44b395c68258231241fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/883eae5b62fc4365b23d4cd736530cec : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f4401970a0ce4496b2d27f5fee3c8663 : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7739307ea6264869bcab5abda5b5966d : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354c94ad931d49788fe0fdbe1f019667 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/966a925f183043d48134a1343942e2f3 : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/78fd78b9c74c48e18af063098e755abc : SUCCESS in 7m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45b35f3cf6b24f60bec1bd661d0609d8 : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/63adfc4fa55b43309f9980e10c1307fa : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f7ff9571f2ef4916ae483c878ce9dfc5 : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/53cbda3128294938ac5d987a842ac881 : SUCCESS in 8m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6f6dc3031b724a53b9ecd6869b67c485 : SUCCESS in 16m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d59d0fd1731493797ca8529921041eb : SUCCESS in 20m 18s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f3aa39e8878b4f659e83eb6281b02278 : SUCCESS in 18m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3027e61b7e0645e291e434642654d6dd : SUCCESS in 19m 11s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fa6395b5b5ba47aaaa37a6f2f545113b : SUCCESS in 1h 03m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ad65cd9cef34465d9cc8bd95461b6e0f : SUCCESS in 1h 01m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4b05d4dacd374e00a414648c4b503011 : SUCCESS in 1h 04m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba2316b1549747409716030832fa8075 : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8da01fc814d44a5585880bf237905ba2 : SUCCESS in 1h 03m 14s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0dd5c2ef0f9c446db707dd78223c9357 : SUCCESS in 1h 36m 26s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c300cb2eddc4415ca3c573f8cb176d95 : SUCCESS in 1h 32m 47s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42eb297fc7c040fcba5e227b2a45c087 : SUCCESS in 1h 27m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3575228022f4fd281da285e7a9f8911 : SUCCESS in 1h 34m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c73fb2a010cb4ab5b498ead5ccecdebb : SUCCESS in 23m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6fcf370d1ab74f14aa22ecc5dffdefd0 : SUCCESS in 14m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07368b8e2b21462fb10418666ad7df5e : SUCCESS in 23m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ef193d6067a4602ae748a99618b9973 : SUCCESS in 20m 45s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1ac2f8372a9242628d7b4e50961c1a5a : SUCCESS in 27m 16s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3865c8510e1f4603b13bbc9bcacaf980 : SUCCESS in 29m 32s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93f346bc602f4c7f87655ecc2a8a22e1 : SUCCESS in 30m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e0f7d79f4b743f2b8e36474539cf1e4 : SUCCESS in 30m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9dea85f80d8146a7bdd5097b7078f765 : SUCCESS in 17m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b5f1a2b1074402686e4ec73b8cc7b3b : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"99f5cd60619a63cf8744b665cc55a06e4ada5f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 11:38:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fae1b615c987ecf4e8edeeaa92574561d93da775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:08:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36464eb388284100bded946f469feb4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864bd3070c534d5c90719acb02a18a13 : SUCCESS in 9m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0c73dadfe6cd4ad18e22318bf7144b5a : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d23699b21d4aa8b8f724c5f145ae80 : SUCCESS in 6m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb2050baa85640d3860b7d9c71d75031 : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c860330a53894dc4bf9cdba63abe0378 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ccfebd692ae2405da7aa37410d9aaa8d : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0cb76d08bc4247178ee5fcef716c474b : SUCCESS in 6m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/786f17e08006461fae0b6c69c9693804 : SUCCESS in 11m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/40220583aeb04ae7a457970e34a5682a : SUCCESS in 14m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2f79be9268f4bed8bc6e9b794f7583e : SUCCESS in 19m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/53c855eeb24b4d08bebf09993f8b7072 : SUCCESS in 11m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/789d6c334e4546adb96bc3fac8ca6c95 : SUCCESS in 10m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7d9d09d4b3b7491e9baed88f16da4c56 : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a5280ee0138b4e929a70e4986594ae4a : SUCCESS in 1h 08m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/15abf439a5e4413991e8ce68fc8dc33d : SUCCESS in 1h 04m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2aaae2e5cb08444fb2b580460215103a : SUCCESS in 1h 02m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f8302bd16df044fc8ac92472932e11b6 : SUCCESS in 1h 06m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2829ca5a095c464f89cd2019c7afa06d : SUCCESS in 1h 43m 20s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1a42b83d27cd4c0e92cc2ea49f7f7181 : SUCCESS in 1h 40m 25s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9237510a996d413b9c042faf3898f7b1 : SUCCESS in 1h 37m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1587da5949ca44b2ab7b7bcca0d34f83 : SUCCESS in 1h 45m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b3c3e9af985c492296786b160086f277 : SUCCESS in 22m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b120c7c17c114ef88b951812fa131d1e : SUCCESS in 21m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d3208980f5c4288909b7adc2e5d2c7f : SUCCESS in 23m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3a59e303bf714377a488fe3ef98ee8d1 : SUCCESS in 24m 26s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/996c8192f3da4269a92eef85cffa80de : SUCCESS in 17m 30s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f16176d0345745eca484aa39badde8ad : SUCCESS in 33m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3c89f7f7cfc4351bf3a6d9302591854 : SUCCESS in 35m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a6a4c80a46544c69cee58da0a6b7e25 : SUCCESS in 32m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/695fda9c94c04cb9a0a7e6e2b729b539 : SUCCESS in 25m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33fb8d1ea7ed4b4f875b86119e000280 : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":1},{"id":"18061e860bfdaf2aafda64603ae30551874ad7e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:08:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dbd175ca809acdcda5416fc4d2fe100680f846c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:09:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57b76fb5a52f4290ab53cea8e9d06c55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05830c729ad4b3f9474cfb93517170d : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd2e77ba8f404cecb8491e018ae54404 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c17def5d4272befa6812926eecc40ff13a6e9800","revisions":{"c17def5d4272befa6812926eecc40ff13a6e9800":{"kind":"REWORK","_number":1,"created":"2026-04-08 13:10:16.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/12/983712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/983712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/983712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/983712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/983712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/983712/1"}}},"commit":{"parents":[{"commit":"290694c0feb523a85b4812c7625a6ec2f6698a45","subject":"Correct reno for kolla tags/limit deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/290694c0feb523a85b4812c7625a6ec2f6698a45"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-08 12:58:54.000000000","tz":0},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-08 13:09:56.000000000","tz":0},"subject":"Add infra-vms to container-engine group","message":"Add infra-vms to container-engine group\n\nChange-Id: Ie7b194dec24e402949d01da745ccb4743fa6c567\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c17def5d4272befa6812926eecc40ff13a6e9800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c17def5d4272befa6812926eecc40ff13a6e9800"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
