)]}'
{"id":"openstack%2Fkayobe~983725","triplet_id":"openstack%2Fkayobe~master~I2cb422998f8dc2585c67a3e7f0b4218bdeacf1be","project":"openstack/kayobe","branch":"master","topic":"addControlHostServiceDeployandDestroy","attention_set":{},"removed_from_attention_set":{"38288":{"account":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"last_update":"2026-04-29 21:12:03.000000000","reason":"Change was submitted"},"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-29 10:06:09.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"hashtags":[],"change_id":"I2cb422998f8dc2585c67a3e7f0b4218bdeacf1be","subject":"Add control host service deploy/destroy skeleton","status":"MERGED","created":"2026-04-08 13:44:30.000000000","updated":"2026-04-29 21:13:03.000000000","submitted":"2026-04-29 21:12:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983725-addControlHostServiceDeployandDestroy","meta_rev_id":"ea971464cd6b1f5698183c708ed47df1ce5a87e6","_number":983725,"virtual_id_number":983725,"owner":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-29 21:12:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-29 10:06:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-29 18:47:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-29 12:48:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-29 12:48:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-29 12:48:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-08 15:44:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-22 10:54:48.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-04-29 09:28:28.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-04-29 10:06:09.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-04-29 18:47:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"c5760aacc40aaba8b8f99725e5d6635e846f93d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-08 13:44:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"390c2e43bfcd116296dd2cc0e355ee7fb8bff994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-08 14:01:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"448e2e204d9960acdfe810922ee8d909492c878c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 15:44:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4be0a0e37d2a4180a276bab492840a47\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6544eca46db34e30a1662209174b2953 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5a9afb35c2b450f8eeac5199dbaa7e7 : SUCCESS in 8m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ce93bd2d0dd84943939394b1583067e8 : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b853659c552d4e9295d89b1f80d73add : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d644374a6845858d22f1b517132b16 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ace6c5e179c488bafc169aa81b7b349 : SUCCESS in 5m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f737794aabd243349d9cf8448d313a47 : SUCCESS in 4m 47s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d03434db9bf42749f7b3da44a524544 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/766019f3147349bc8b4a2b5bbd348150 : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b98f7a4620bb468a8ad702bc0ac4f976 : SUCCESS in 9m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7ee94376df7b4b62a66ae803a8484c95 : SUCCESS in 10m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3acd49eef63e436987af70079b88680e : SUCCESS in 14m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3303861c229a41768fb51f6baa812300 : SUCCESS in 23m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e26565233e5c4e15abf3d456aeb8f131 : SUCCESS in 17m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/37fc7b76624e483495dc915505d2c5d8 : SUCCESS in 1h 02m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a2d6d6733fca49fdb01e1ed3444567f0 : SUCCESS in 1h 02m 04s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8a1bbd01259c45cda5a96ebf9a88edd9 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/554451d5a95f43cab10e267f86ef8ddc : SUCCESS in 1h 05m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d04cea8dab0a41c4a41689beaf4ac2c1 : SUCCESS in 1h 08m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b190cf307134448aa4b8e25636929f3c : FAILURE in 1h 30m 34s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5e438ee6299349bfadb0c7a241ef5938 : SUCCESS in 1h 30m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ac768a97d6d4bde80adfa40632aafe9 : SUCCESS in 1h 29m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88770999bbe44783b884a116e5625661 : SUCCESS in 1h 42m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9f9e3ba5fee04dbaadd0ef2c548efe8a : SUCCESS in 22m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/58e975270d124b658f3f5486ac734255 : SUCCESS in 16m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69825c88f7444052bc3a15fba8760a01 : SUCCESS in 10m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fc4fef73931d416988aaac46a812b3a8 : SUCCESS in 17m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bd24be107f0e48aabf77693f4f9d801a : SUCCESS in 18m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b58c99c552154c32873cc4a415943d39 : SUCCESS in 27m 26s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51d806c7de31458ab2e8c9372525f1b3 : SUCCESS in 16m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ca6257d9aa144ad8f8d0b9763ab4074 : SUCCESS in 15m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/958e35a2233e4353ac0a116ce10db1c8 : SUCCESS in 10m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48ba0c44036e49a2b0cc38730513b330 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":2},{"id":"15ec251c14ee734215201c20a909af750d9bbd7f","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-22 08:37:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6c64c2fcda6b2618e0a366ec501598e67b67ccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 10:20:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfcad7d8974c444f8b78ebdfa28f5380\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/afeb04cd46394642bb18044ded84958b : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca302246722d4a889a1f58cc4dc4ea4c : SUCCESS in 10m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0875553de6034d62ac3d91e5997c1433 : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fd9b1a423a046a29892006fe8f37b0f : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ef5212f6064dde96f05b9f4286ba59 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cccc99e8511a4ec9a0aedd5d515d6b77 : SUCCESS in 5m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bbd1cb8ed7724b9eb0dc6cd740344e2c : SUCCESS in 8m 46s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cfe57ee2fc4f428fa0c8f57adf1943bb : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e2df7fbdb5249d987df3e80e648d5dc : SUCCESS in 8m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25e9723060164215a7f2e88cad81399e : SUCCESS in 5m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bea9eff7433944ada232537d261d8647 : SUCCESS in 13m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f73e4828b224fb59e2d9424b96a9b90 : SUCCESS in 22m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0b169c0508724807af54c600107ba49a : SUCCESS in 10m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/462ca46cf23d4f2b997382362062cbfc : SUCCESS in 16m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/18fa1a8ab4184953be47d869974cc2bd : SUCCESS in 1h 08m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9c15dc8c45304cf5b21e923b7424072e : SUCCESS in 1h 02m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/135c0d860f204130962c685ff8c2a1c9 : SUCCESS in 1h 04m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/399ef033a3bb491bb006c66886ba17da : SUCCESS in 1h 00m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d9af145ee28c430fb2b578c4caefdfe3 : SUCCESS in 58m 17s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/668dc352581342199cff43bc0cedc3c3 : SUCCESS in 1h 34m 43s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0305ac4d230b4dbca5ade79f902fa460 : SUCCESS in 1h 32m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6be1ca570224d009ec69d9beffd6aaa : SUCCESS in 1h 27m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38e81fa34eba4c56b8168bfb6401e938 : SUCCESS in 1h 40m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2eda790a670e422ba9d191c2400722b8 : SUCCESS in 14m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2f19a7a837674388b3e5e08bd2f720c1 : SUCCESS in 23m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2829cded75c4451a76b38d046782e73 : SUCCESS in 14m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e3fcbc355e6f4d868a230e5f022b4ef5 : SUCCESS in 22m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/51d21f397c94474c95df64e49f362d9f : SUCCESS in 24m 52s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4657abdf6fc14ed896c83d12e44dff5b : SUCCESS in 32m 20s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7523826072d7484f9073bfe8a542ae83 : SUCCESS in 30m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b759cd56d41046bf8fc749f468293fd9 : SUCCESS in 32m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d1f58e03512042d9a899760a4acb24db : SUCCESS in 20m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f60a97b05714d23b734dd5c9e376982 : SUCCESS in 23m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6cc07cd6702765dabfb6fbe1766d7323c8573d5d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-22 10:54:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c76a86cdbb5648277f74fd8683a9a6541dd0735e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 07:46:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b8afa52acdabc97b9a3bddadcb3278acb1db0280","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 07:47:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"98768c68aff6e5c709c6ec9f38351cb9e321f04a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 09:28:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2ed4155b31541ea9faf882b43bd2b8f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/821a5fcc8e9941e08b019f2a6a2b6c30 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3313f096283647e1b586288dd0316a76 : SUCCESS in 9m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1f4f59870ce64dce874bbe0d7ca19618 : SUCCESS in 8m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2de0c49929e64b488571095b83e23c2e : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5e2137cb2a4d6f919c4f6440ce4aa6 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9937289150724d9aa2e44b03cecb1d9f : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dbd290017ab6458e9101819a3e5f01c9 : SUCCESS in 6m 43s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c5d7d72475e426cbab37fde138a119a : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/617bda1fdb2f428b956f0eb57519057a : SUCCESS in 13m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc45992eac7242ec9f54b99d678e6732 : SUCCESS in 11m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eb51e11287064d2398120bc20b99e396 : SUCCESS in 24m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3c8361be47e48fb87393ce0a42d3784 : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0594cfce93154b3483c59feb95bd234a : SUCCESS in 15m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5b00a2ff5534c0d933766bd7e5eada1 : SUCCESS in 17m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f8aa6039c1a64d78b3e54e907783f7af : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bcdb0e56ae2b479d9536ab4e12052463 : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9e38da54c39445dc8d4eafb6cf03101d : SUCCESS in 1h 09m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0990b4c9f3724563a9665eebc29dd459 : SUCCESS in 1h 04m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a144aa13f6e94888bc161f781c46d50d : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8b88d15129bf43829891e4f08fbbb30b : SUCCESS in 1h 35m 35s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/12c73a1d546a4ff8885920997836f74b : SUCCESS in 1h 35m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4d7c6004a0d42bc8d8a5f0e8cbe2db2 : SUCCESS in 1h 27m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c66e846acdc046cfbde9afe93f00de46 : SUCCESS in 1h 40m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9897a1420ba44c229263de81c5284a58 : SUCCESS in 22m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f2da8e0519824b479cdd910d63f61365 : SUCCESS in 14m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92a29a53fec54577ae0bd8b0949133fc : SUCCESS in 21m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dba735da931d431a830a055eaf40cad0 : SUCCESS in 22m 26s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3582f809a6b34059832d027016472c83 : SUCCESS in 27m 08s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/29694f4e85004f94be8376327a4042f8 : SUCCESS in 26m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20005fa90a6f4aa3922ab9889b8e42b4 : SUCCESS in 30m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c944058098a84e44ac1d31691b467895 : SUCCESS in 29m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2d702c0629fb4e9cac263e555d2f27cb : SUCCESS in 15m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7735bbc4ea004ca88b2795af03152047 : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":3},{"id":"37bf6e7558bc6046f31ce35dd66571ff6866096e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-29 09:28:28.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"185ec8c2216848e781352ffd84b15728fb889f25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-04-29 09:51:14.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"285b843300b0a5df58e8d978166b5f4631ead7d3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 10:06:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2e9b11236443413bcf9ef962203a7a0d253c490f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 11:30:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81e8a8c35dd64a8b9ed27ed97853c243\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e6f7021e7a044c4292c70e4eb03d90fa : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67809731ffd04ae482fe77480196701f : SUCCESS in 9m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9a08012880e04b3c8384384b072ea4de : SUCCESS in 8m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30f907e5901b4ed5ad1ed10c5f03f539 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a5941a7a7b48c3a144f2bc8c4df3ec : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb970f8f23f641d999c9c11c5072bed7 : SUCCESS in 2m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ecb3329e472649788bb44ca279beaa3a : SUCCESS in 8m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56617facb7844b269731a5d5828ef749 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f31c94fdef7943bcb7431079111ef0ca : SUCCESS in 3m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8945920ab48a4352b60c2fad3ce37356 : SUCCESS in 14m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1ab7f044950140ada7a1f70c88fa08fd : SUCCESS in 6m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/20f5a14347ac48ad8e87b4b0c38b1195 : SUCCESS in 20m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd605bee9d4243ebb463d6b332476b0f : SUCCESS in 19m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f7af4cf57d08465e991db51e219c25cc : SUCCESS in 26m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb609aae61754110acd94fdc4858bf9d : SUCCESS in 13m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1996a00fb56b450eacb50952103fe684 : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fd5ab95b6f3c4098b7b77251cf0848b0 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b7a4fbbb514442778723d905d7e52cc7 : SUCCESS in 1h 06m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03fa4c08f5c84414a713f1cd79d2d994 : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9625d7d6d84b43f4aa2c35a7be2eca5a : SUCCESS in 1h 03m 57s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4fb08b5c6a1a4595bf3107da6033cf87 : SUCCESS in 1h 32m 32s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/085afff48f8b4d7bb0b488fde9c2b9ad : SUCCESS in 1h 33m 01s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ddb336b950c471898d4c8950042708b : SUCCESS in 1h 30m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea67cadf22904623a437c48bc7639e66 : SUCCESS in 1h 37m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/abf3863c4b6f4abca6195e285d4ce387 : SUCCESS in 28m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b7aeabf8f57e4efbb1a6b19134440c60 : SUCCESS in 18m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac5aa87c63a84b9b839bcecef0f3f8d6 : SUCCESS in 22m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5d663283c21460e9ba4399405cc3e66 : SUCCESS in 22m 58s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/08fdce99cbf94a5e8d75b3fcc820a4da : SUCCESS in 35m 47s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b67461d39557458dbf817cccc57ddf7d : SUCCESS in 17m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ede5c88ab8f4ecebeb1e668bb8562bc : SUCCESS in 32m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1e3b476c2e9438796d2afe6d5a29477 : SUCCESS in 15m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/344a9ab392184f1b8d6fa3ce2b6d899d : SUCCESS in 18m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6586e90012324de0987e056c2186226d : SUCCESS in 21m 55s","accounts_in_message":[],"_revision_number":4},{"id":"a4b2c2a9d7c8112ed2d608ae8ed9cd7b7ae64164","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-29 12:48:13.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"daf8597c25aad95162daaf10aa43926ff9169b6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:48:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"571d99255b8dc46202e47192b099d5277a86c0f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 14:27:50.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22c3f41448eb48a4b232dfd0e2239060\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c3cf1b2e8d844bac9118adad3540abbf : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2681de647bc943aaac9033799078c5cb : SUCCESS in 7m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d630cfdcb4e842089870ed813e999ad9 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6913d3bd1ca74ea4b14e4d9df040dc8a : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f0678df7e3249268c0e066f0f726ffc : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a170b838c8f249ce814b57e66ea2a785 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/99f63ab092214e8a9b5f2e6a9c48edba : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c8a42198474b47c392d7d45e743dbe33 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/adfbe10c619f4c58a649a0b800cc1b11 : SUCCESS in 13m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d948be52d35e4745ad1de43c81dbaca5 : SUCCESS in 19m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59aca21f5ab04dfd8a404a553ccce7e8 : SUCCESS in 20m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/64e74c8b5ec348c88cd05e22adbd3deb : SUCCESS in 12m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5408a0f260a94d82ae9fd87ab56fd491 : SUCCESS in 20m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/78320d2227944458af97c2ce608032d5 : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/aff1c1f7334f4d59be5831ee967af4c8 : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4f833b4732c5466c996d240dd533ba8b : SUCCESS in 1h 10m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9a7ecd4035a4ed6b7d4ebafe48059a4 : SUCCESS in 1h 08m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f23f83ab4c2840baa0b3e2cfec5962bc : SUCCESS in 1h 20m 34s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3655ffbcd82241dbb6870bf72049042e : SUCCESS in 1h 38m 30s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/957f624f71904a39814cc9c5d13973b2 : SUCCESS in 1h 34m 38s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37206af5386645a2800b2e291b0e7fa5 : SUCCESS in 1h 34m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4b1c9ec733a47f78587e0e5c437aae3 : SUCCESS in 1h 36m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ba183dbcb2da40e69e6239bc213a9753 : SUCCESS in 29m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6730f4ddc00b4a4d82521ac257f8dcb2 : SUCCESS in 14m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbb6fb2b225c417b9d96d1d82643a2b2 : SUCCESS in 25m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/acac2aabf755465c9db70e4d4f63d8be : SUCCESS in 24m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/76eda1f9c74e4c0c8818468fd9d89a22 : SUCCESS in 36m 31s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0865ac71ccbf41d8b8f187dbf4dd91c4 : SUCCESS in 23m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c99e5adbfc054315ae431fe87a1b9d23 : SUCCESS in 31m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c311a37c5ed84b3e94087736d88f9d1f : FAILURE in 7m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8e3b24a99e19428fadc739f0eb1e82e3 : SUCCESS in 17m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc56dc52999447279f766c98c1155a66 : SUCCESS in 18m 53s","accounts_in_message":[],"_revision_number":4},{"id":"d2e9405eff9328d258858efa24c1d367e9d9bde8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 15:54:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"04287aea5d94635037646c24a6cc9d15c8fa4733","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-29 18:47:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3e00bdf130b8a02b55a335a3e2251720002b0e5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:36:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f370b01fb9644f4a1fc2c60583ddcbe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3e4a744e2e124d85a23facdd96e8f571 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916e1a0802074477ba5a08c8e68e4b28 : SUCCESS in 9m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd7bc43d7e3f4040bc2c0ea40a3cd293 : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77ec14bc050e41efb6b9c787c108d6b2 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec4e4178463437f84d3c3b65dcd8479 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5da6dbc820947f09c9d1d1fa27091b8 : SUCCESS in 5m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54affc53a756457d8845146dc52e3ea6 : SUCCESS in 7m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eafdea7b123345e9bac2dffb2499d033 : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/adbb3994d4ee439b81a8a76a351eb4ef : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e5fef76c4b246b482d3dcf8eeb4f40b : SUCCESS in 14m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/44e55c0de97c451ea61b1e745f2dc7f1 : SUCCESS in 14m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c135f9e006844761b26ecda2644eec7b : SUCCESS in 14m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43ae2ac006f44f069b2e904a84e1baec : SUCCESS in 19m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ebde114b18714347a94a0f896dbff117 : SUCCESS in 12m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2c1f91b7d2f4e2dbeafd33fc0522f10 : SUCCESS in 10m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/789651c0ced24358b8186cef8b3f8bd8 : SUCCESS in 1h 14m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0198e54d1de5472eb1c0ba27ae923af4 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0a92a723497e44c882efe36a4fa1d778 : SUCCESS in 1h 12m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f0df290373e45c5b9aa28c2e0467d35 : SUCCESS in 1h 05m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/66a2b71c58b9454eac7a9b4feafec352 : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e5cf94bb6e024a53883e065912d74187 : SUCCESS in 1h 41m 09s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ff25663f2ecf49a78660b9e47329c801 : SUCCESS in 1h 34m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3150a56ed8e94e68921df30918e66019 : SUCCESS in 1h 31m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ee848be2a624b7c85ead4383dfae26d : SUCCESS in 1h 40m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3b73e2a0c6434722a6080eba25ab02d7 : SUCCESS in 19m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c5656860c4bc47278caaa486b8c0b677 : SUCCESS in 16m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/598abccf90d94dfa960cfaa560b45c1f : SUCCESS in 24m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cd9c7e518934426b8a439de5663cec98 : SUCCESS in 16m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/982d42466e7f40a6ab69a63aea44a5c0 : SUCCESS in 17m 28s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fed02cf621ad4879ae886233abfde0d4 : SUCCESS in 18m 06s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bdefdbbbba24688ae8d982887d4d06e : SUCCESS in 33m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fa21dc8198f47fa936ae4a754c22e74 : SUCCESS in 29m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5ed4c398c91a4d3b841771776a4c2940 : SUCCESS in 20m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1263f6da52940ec8dc3e0cda6e38bcd : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":4},{"id":"66c8d6f37fe988a83bb9a46034482cdfa19b1da8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:36:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"739ecd5d716c5557c925528bda1f39859363604d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 21:12:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/560edebcf3bc4f88aef9abda814eeb57\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0152d5069ab74a1a8a6267b7166004e1 : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ed74d0a7844527a7cd65e00b60f02a : SUCCESS in 8m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1d29cd728842441b929a13a65407b173 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/703d9787560a42c79501de9e619972f0 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f4f063c00cf4f009a7a8696ebe8ebcc : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b383af38392473f8142512cd18b6569 : SUCCESS in 1m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5aef48e3c92b4d6ca880e491856c73e8 : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c18a8a8e8b36438899e539f4ce1a0414 : SUCCESS in 5m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e8462fec82641a0864883bc3e8e5a9b : SUCCESS in 8m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eab0ba45bd8a4106b6d59344568431b9 : SUCCESS in 14m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58da25d8fd104e6e989c6484647fa123 : SUCCESS in 13m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/30c1525232c2441abc6718eb406da5ad : SUCCESS in 11m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e8cbf61c1c641328e9996b1014355aa : SUCCESS in 19m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1d85ae8da02446d781956ca923688561 : SUCCESS in 1h 06m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ac730cce80aa460abd4362387e9df760 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cd20b674bbb24ced98f1d40252272dab : SUCCESS in 1h 03m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ad7755fbd414f75be7bb6b2b9550629 : SUCCESS in 1h 06m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5660a3d78d5b48dba1dd6fd582673fe4 : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5774ff49ea444e5183981e10bf40f8ac : SUCCESS in 1h 34m 54s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6d17464c54884f16a98c6a0cfe9c53b6 : SUCCESS in 1h 27m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7f2fb71c9c84d7fb5250092f1d6fae7 : SUCCESS in 1h 29m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a33b3c1a23944d9890186464163baf8f : SUCCESS in 1h 31m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/977eabf0c0064e0bad0d3dd84932ea2c : SUCCESS in 13m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6b340be062a54696be3b8c1085d02291 : SUCCESS in 14m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b84b8219c7148f3917e567878f30718 : SUCCESS in 22m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/826b0f88118440a19db282d4bc4c7e80 : SUCCESS in 21m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4ccd3b1ad6f3442d8296912b71f0e62c : SUCCESS in 17m 35s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/de1aee12469a4a329526b09240269d6d : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3354ae9872a64142b075eeb218f67121 : SUCCESS in 32m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b23e37e9de3c46a9938948816daf678b : SUCCESS in 27m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/77a83cbcaf5f49b7b5ca6c07eaa79fb3 : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16f5230deb434422b38ef01da2372949 : SUCCESS in 19m 23s","accounts_in_message":[],"_revision_number":4},{"id":"3fd9344f10c1b453a6b0db429c9a5b1c62546eff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 21:12:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ea971464cd6b1f5698183c708ed47df1ce5a87e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 21:13:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69c3496c975a49dc9c4b96cb8087576a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96acb6e6205d4511a03f12cc36d507db : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59aba89f1f114d89bef317401cc0bb0a : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2","revisions":{"e04721d08613552a487b31a49dc72b5376619bdb":{"kind":"REWORK","_number":1,"created":"2026-04-08 13:44:30.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/25/983725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/983725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/983725/1"}}},"commit":{"parents":[{"commit":"a3e5e8a3683724e56ca002a71ee876544c40bc4f","subject":"Add skeleton code for Infra VM Service Destroy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3e5e8a3683724e56ca002a71ee876544c40bc4f"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-08 13:43:00.000000000","tz":0},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-08 13:43:33.000000000","tz":0},"subject":"Add control host service deploy/destroy skeleton","message":"Add control host service deploy/destroy skeleton\n\nAdds no-op classes for future control host functionality\n\nChange-Id: I2cb422998f8dc2585c67a3e7f0b4218bdeacf1be\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e04721d08613552a487b31a49dc72b5376619bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e04721d08613552a487b31a49dc72b5376619bdb"}]},"branch":"refs/heads/master"},"a5c425c04331553f41f236ead2b195e5c9245361":{"kind":"REWORK","_number":2,"created":"2026-04-08 14:01:10.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/25/983725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/983725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/983725/2"}}},"commit":{"parents":[{"commit":"a3e5e8a3683724e56ca002a71ee876544c40bc4f","subject":"Add skeleton code for Infra VM Service Destroy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3e5e8a3683724e56ca002a71ee876544c40bc4f"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-08 13:43:00.000000000","tz":0},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-08 14:00:16.000000000","tz":0},"subject":"Add control host service deploy/destroy skeleton","message":"Add control host service deploy/destroy skeleton\n\nAdds no-op classes for future control host functionality\n\nChange-Id: I2cb422998f8dc2585c67a3e7f0b4218bdeacf1be\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5c425c04331553f41f236ead2b195e5c9245361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5c425c04331553f41f236ead2b195e5c9245361"}]},"branch":"refs/heads/master"},"6ae6d313e709016023324932c8ec8b12a237ceb3":{"kind":"REWORK","_number":3,"created":"2026-04-29 07:46:44.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/25/983725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/983725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/983725/3"}}},"commit":{"parents":[{"commit":"21b23ad817fc70f40f73471fce061f0ef94f92eb","subject":"Add skeleton code for Infra VM Service Destroy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21b23ad817fc70f40f73471fce061f0ef94f92eb"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-08 13:43:00.000000000","tz":0},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 07:46:27.000000000","tz":0},"subject":"Add control host service deploy/destroy skeleton","message":"Add control host service deploy/destroy skeleton\n\nAdds no-op classes for future control host functionality\n\nChange-Id: I2cb422998f8dc2585c67a3e7f0b4218bdeacf1be\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ae6d313e709016023324932c8ec8b12a237ceb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ae6d313e709016023324932c8ec8b12a237ceb3"}]},"branch":"refs/heads/master"},"3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2":{"kind":"REWORK","_number":4,"created":"2026-04-29 09:51:14.000000000","uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/25/983725/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/983725/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/983725/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/983725/4"}}},"commit":{"parents":[{"commit":"87d1546f30e776b56bc58e2f56d0342cd91d5cb3","subject":"Add skeleton code for Infra VM Service Destroy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87d1546f30e776b56bc58e2f56d0342cd91d5cb3"}]}],"author":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-08 13:43:00.000000000","tz":0},"committer":{"name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","date":"2026-04-29 09:50:59.000000000","tz":0},"subject":"Add control host service deploy/destroy skeleton","message":"Add control host service deploy/destroy skeleton\n\nAdds no-op classes for future control host functionality\n\nChange-Id: I2cb422998f8dc2585c67a3e7f0b4218bdeacf1be\nSigned-off-by: Leonie Chamberlin-Medd \u003cleonie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3955ae9330c9f1ef5258f9a3ae9f43b07d8ce6b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
