)]}'
{"id":"openstack%2Fkayobe~983903","triplet_id":"openstack%2Fkayobe~master~If338782e9bba9a13bf4e7d55c1ddb36130683866","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-06-02 12:51:38.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-06-01 21:40:49.000000000","reason":"\u003cGERRIT_ACCOUNT_35345\u003e replied on the change","reason_account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}}},"hashtags":[],"change_id":"If338782e9bba9a13bf4e7d55c1ddb36130683866","subject":"Add context for agents","status":"MERGED","created":"2026-04-09 15:53:43.000000000","updated":"2026-06-02 17:55:30.000000000","submitted":"2026-06-02 17:54:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983903","meta_rev_id":"2f28df03aebf63ed742b4ce9ac84a5fdb0d2e6a9","_number":983903,"virtual_id_number":983903,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 17:54:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-06-02 15:33:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-06-01 21:40:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-06-02 15:33:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-09 17:30:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-22 10:13:29.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"CC"},{"updated":"2026-05-11 23:54:14.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"REVIEWER"},{"updated":"2026-05-13 20:39:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"1b8e0ad3b4da0956154cc66b3e7b8115b7e7a77f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 15:53:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"316ef53d3f19a4c818e5affbcd276893bdc8242b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 15:54:08.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"0d17013fba1718fa278833734bad0bedcd8fd6be","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 15:54:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a28253f4a879c805cc1af0d8a8cf2319c9c10272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 17:30:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2731999d9004cd1be0c5fc97d9b98af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8edf7cf3f4c43c2914795504602266a : SUCCESS in 5m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/06a5a76e6a084fe893450542e31cd681 : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bbc1d8b395b4623b5274d8273294968 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d3a1aa2467b45dd9e19316db6cbd84b : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7ed70705870415f8a1a7a56be4e11a0 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d03936dd80e34857a685990d04f57ee0 : SUCCESS in 3m 51s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5ee34bb877d44becbdbefcfdc0458cf8 : SUCCESS in 8m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/272d58c4fee047ee9198727781c862b8 : SUCCESS in 14m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7161a492543e4dda82c583fba0193e36 : SUCCESS in 13m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2920dfdafe1145d2b1ea638b23275a14 : SUCCESS in 22m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/295ddd1a05b34fb7b150ae1a88fa716d : SUCCESS in 14m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b14c88df2a6d44a5b5bcaa82a7bd63b5 : SUCCESS in 9m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd3b95acfeeb43a2ac867f64262d5e60 : SUCCESS in 18m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/659e330eab7342aca6b9ca97ab547697 : SUCCESS in 56m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c5a31aa9ac1541a5a0551f377d5a1421 : SUCCESS in 59m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/777cffebd16440bb9fd180b7aaab6e68 : SUCCESS in 1h 02m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a23adbca55f64c60a331d76c42bb35dc : SUCCESS in 59m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7bcea15baabb44238e99abed84bf7091 : SUCCESS in 59m 37s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/aba16cd979a348f29de6cf5fea2735c4 : FAILURE in 17m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/31eb258c7f6f42daa87034fde2951410 : SUCCESS in 1h 33m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9767375d98244c0598f9bb2ca886a60c : SUCCESS in 1h 30m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93a63249318a4b219cedc4413a3e13c2 : SUCCESS in 1h 32m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5eb0d3e1932b4103bdc40073d88cca14 : SUCCESS in 18m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/860fa481225e40b3bb65ca68aabae262 : SUCCESS in 13m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cc312a774d4416bae875ecfdfc51f62 : SUCCESS in 19m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/62a5aec1262d4e149922d11f53361d09 : SUCCESS in 15m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b7ccebd192b0479f8130b98c644df8e1 : SUCCESS in 37m 24s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3cac955df6ea4ac88b495be83c31edd8 : SUCCESS in 37m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c6282c384024133aa3f504a2dd5cf8c : SUCCESS in 36m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/599b0242bf0441e5a94dc9abc4393098 : SUCCESS in 29m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16f8bb120e06471bbaf30c565dde29c9 : SUCCESS in 17m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b22e07aa25db44948b4aebff97134c6e : SUCCESS in 20m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8b23a05f373bc33441abfb3f7427ba00b01c758b","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-04-22 10:13:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d6db551e921e245eb2c2d03ed362b83ce488510","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-22 10:47:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c431abff9dbaf2660eb2b73ddca47c3d187b63d","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-04-22 11:04:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"297176dd33202d5c1bf1f753c408b6468a161d0a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-28 09:24:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b6b00487a7b309b9d738ca31bcbca28ce27918ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-28 10:01:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a9aebee34293a703512600465ca0e1632c0d21d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 11:47:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5e863df5af54036bef3874324b91721\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c1ec8f535a04b49bb56e9af6439b4c7 : SUCCESS in 5m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/30e373f78af947fb9e4c4138aebdcbee : FAILURE in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3629b47135d4ff69b5704ead2a03fcc : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9f5542aff8e40769abdab7f2ff86df6 : FAILURE in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/523b76caeab9415599a87d609fb77871 : FAILURE in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7f641a1d070a4d7cb53a0b63d8a9b4b0 : FAILURE in 5m 00s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/feaafcaabf1841ef98fc620ea936f702 : FAILURE in 3m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/973dc74d040e40e9bc3d0bfb16f69430 : FAILURE in 2m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2de8d96b998547e196d5e87007f81e46 : SUCCESS in 11m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7f7ab57081b2423c9a2190537450f70b : SUCCESS in 13m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c41c8cd98975466c9ed476a4bf17a64b : SUCCESS in 16m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e324305066a24762b8edb95074a43d2c : SUCCESS in 10m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6678326b40345078b2caa9c5aa5b2b6 : SUCCESS in 18m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/11e6e89081f64e8d89f3e9f3652b27b9 : SUCCESS in 59m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fbdfd02242b3431ba6f937cc8c838e13 : SUCCESS in 1h 07m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/732c3452fb2c4ad6bead1a468556cdb5 : SUCCESS in 1h 10m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc70bec465be4d259986f5239c91cf91 : SUCCESS in 1h 05m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/339a667b7d8d4c3884c2ac20d2f6c171 : SUCCESS in 1h 14m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c4eb9c4f10104f2d90d4ae48826c0b71 : SUCCESS in 1h 43m 26s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/01fce91754244a47bee128baf6b6de2e : SUCCESS in 1h 37m 17s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fac1d847054c47328488c554fb9f7e4e : SUCCESS in 1h 33m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee2884ca7d8f416e894bfad0859762a3 : SUCCESS in 1h 41m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/265195cb7d4240e4a8e74f28c467c056 : SUCCESS in 14m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b2e718110e4e40de8d735116875c8e0f : SUCCESS in 23m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5fd500247c445a29341d4c0e765807f : SUCCESS in 22m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6a0a43263f1043039b1d6a502f6f4860 : SUCCESS in 16m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/83f640c5dc16433daa803cf0e63feec3 : SUCCESS in 34m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/aa8d211f43f24df18f1efe61d2030c98 : FAILURE in 17m 38s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00923c59ec3b4621bf6cf1ab1faf1e4c : SUCCESS in 30m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/977961edbb7349fb861221f5cd9f7717 : SUCCESS in 30m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b09cd2bfb9fa4b65a66c7887dd468698 : SUCCESS in 17m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/256074afdea84ba2b21e0a76fc27a5e2 : SUCCESS in 21m 38s","accounts_in_message":[],"_revision_number":3},{"id":"39bfcc7876993ec51f455060943df6ad4988a1dd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 15:10:27.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"cece4d857358cab6258fc467862ecb0e80f820eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 16:53:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b55eea425c5a4320b7c9c121a988bb87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9928748d7b82430fa58776061ae10b96 : SUCCESS in 10m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/82e97b788bf445f89ebee56f56ecc299 : SUCCESS in 7m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fa2a30967c34d968fb10a088496b620 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c701227858b749729ac6d3e86086408d : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/066c3529708a484180c0bfb8356548d9 : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/92db014ec5324487875292f7c48f846b : SUCCESS in 4m 31s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b01d62a8f4b74d8f989e19d613cfd92d : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd9f50ac507540f8ae287c6bd92f848a : SUCCESS in 13m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ed6d8a264dbf4b52a06cbe93e2858110 : SUCCESS in 14m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d7b74688335340cf86296541b2769dca : SUCCESS in 21m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc9f14c432a44c248a63af34c250c95a : SUCCESS in 13m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e208db0c9da14420b4fc364145438a63 : SUCCESS in 11m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7ae6a4c404d4fd8860885cd6723fb93 : SUCCESS in 16m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cbac3199b06d48609053c69da31e4c2b : SUCCESS in 1h 03m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/71875dc8d82242d49c720e521d002c62 : SUCCESS in 1h 03m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/19ca7ae104b043648e01fb20a515397d : SUCCESS in 1h 05m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/368538abae624b3e98053be7efb6ad71 : SUCCESS in 1h 08m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b7440078c7424c0a90afff465baa4972 : SUCCESS in 1h 02m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8546705086284192a2d34ef8c15cc044 : SUCCESS in 1h 35m 24s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a778ab86895243a2b1b3181cbda2ab0d : SUCCESS in 1h 30m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98c627831c654779928403bac233133e : SUCCESS in 1h 26m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b4601cdbe7a4b32af20c395d9a3296b : SUCCESS in 1h 37m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/caa219230fcb48e3920f0bf7bb700fd7 : SUCCESS in 25m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/91dfc80e0a8b47a4a2dfb5ab26142d6d : SUCCESS in 17m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a9fe9c3dbb34468867774887a32556e : SUCCESS in 24m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e12513814b944a3a9104623362b2b04b : SUCCESS in 28m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/46e152b37145499e94a808f45a2929c6 : SUCCESS in 32m 48s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1b5cf1d19b4c42818d26bc69d990d805 : SUCCESS in 31m 07s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/300f19e58b8d482d8c3559bca199c36c : SUCCESS in 16m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/650985ac617f43599bcded57dc0797b4 : SUCCESS in 34m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6328eaa006df49f8ad2c9397286dcab3 : SUCCESS in 19m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7b45916938b42ea987f046db7ff1d00 : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":4},{"id":"7c2e2ea8ce5d1d1705a24080afbbeb6564293cba","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-11 16:11:15.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2ad5ce84ea5456050a65073327264223f67034f7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 17:39:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"46a86a93851faf29dad3bbdf66d68f04e8383925","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 17:40:45.000000000","message":"Patch Set 4:\n\n(6 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"cdbd1f8d3f01d55daaa202c9b98d87ef74e81575","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 17:40:45.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"1f7e5d6bf148127bdfeace75ebc4d10028a7c490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 19:13:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab4b8df390c8409da3908cd949e7b292\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0cd45b481c49539963675700736412 : SUCCESS in 9m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0ba307876fd14511bebe0cc11719ba56 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84b67baea6154f24a048fd19d87c752f : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b51e21c0ca457eb1c19cf1dc6ef084 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21753ae4c0cf4348a883544c7edb6b1a : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b5c0874d2b3a4f3181ca26937c11364e : SUCCESS in 7m 05s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b451228889a447f888278d1168fd454f : SUCCESS in 3m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b96f06a9e15e42cfa21ffda94829476d : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b301ed962aa4f578326e830b6788667 : SUCCESS in 13m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4a2692f7f9ed4a7988e3de5c60a855c3 : SUCCESS in 13m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c6e985e2ce34359ba91d1b228b448a4 : FAILURE in 11m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f38a2251e8874b7d9582d35455d61fc3 : SUCCESS in 19m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c93895e90b9b4544a063101377f7dbe8 : SUCCESS in 18m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/599914e6b5544187a94a84c056f2f28a : SUCCESS in 1h 01m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e50deb5ef3f9439aa5cd4582dbdda456 : SUCCESS in 1h 03m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6e86592adfcb4f4aa01c8a8ca6ac8ea2 : SUCCESS in 59m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e3cf04ba3dc4bd89fae5a695b699385 : SUCCESS in 1h 03m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/46484a26a7174372949de5731bf50cec : SUCCESS in 1h 03m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b58bd376b58b459198416d6066372968 : SUCCESS in 1h 31m 30s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/721283daca21487db09df728639560fc : SUCCESS in 1h 27m 38s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f82a3b52a513408bbae0c17631bbb9ea : SUCCESS in 1h 26m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77964bfc5b0d47a8a418c4f61ac2d123 : SUCCESS in 1h 32m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5af055b387354c8ea5c18f17ce5921be : SUCCESS in 26m 51s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/53463d5078614182beff704da76237bb : SUCCESS in 26m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1efc58e4041a4e728e85e442b8c728a8 : SUCCESS in 23m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/78064e84c17d404e9dc91cb20242990e : SUCCESS in 24m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ef9a968b835a4a2590fe97aafdbbbc44 : SUCCESS in 27m 57s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/03e58190421f4929add275c80b93547d : SUCCESS in 37m 14s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e522e97e1f94479a60f5bbb743b3da6 : SUCCESS in 33m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32524ff4183e4e409bf51860a56cef6a : SUCCESS in 29m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/847bfb9a505c402bbf7b36c7d7825fbd : SUCCESS in 25m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7f1563b9c3e415dbdc56d657044e629 : SUCCESS in 21m 33s","accounts_in_message":[],"_revision_number":5},{"id":"15dea1f30f9b4ec8f0561e4ede1e740d05769cea","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-11 23:54:14.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4c6fc730cf5ba1f54066db40e8aae3719e5c0688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-13 18:28:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"8680d2c58dbeaa169272a06503f06a8f95819198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 20:07:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3d96047cad34c40afc66bfff1351d7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49fdb0691dd04f8fbd7a8da32bf8593d : SUCCESS in 5m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5f17e49b0cfc4d86aa4e4f0c6f14a5cc : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/054e1a49fc4141c09c896884fa84f780 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f3d30ab028b4bf0bba94edd55c31229 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eaf9d3a45b1c44e5b4fb62e49239b85c : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c98293ff77164772bfc74bf37225f214 : SUCCESS in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c4a4dbf1c36445481456d2e2b279404 : SUCCESS in 5m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4d9ed4df051457482be47519e08f420 : SUCCESS in 3m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd280826a6404e6c9b2ccf469fb3a042 : SUCCESS in 6m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6426cbe973540f98b6c7d11f9b3e839 : SUCCESS in 12m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7fd54829c5014565bbca4907e368625c : SUCCESS in 13m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eeae33e30e0349518f68db8605170c31 : SUCCESS in 20m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c12ad4eb01f54188901c89092eb62805 : SUCCESS in 11m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01df922e060d4697bb8f03ab2aec3af1 : SUCCESS in 11m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/db9399ca9313430b9b8ea22465c56456 : SUCCESS in 1h 02m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/830ca2b9c89745d9ab47998d5ea92dde : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4a691a332dc749808cf1012ebb4eefc9 : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c02c6f3a9c454207a7971e34f30029ef : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f6e1e2a4c2db4406850647a136f0b6bb : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b3161fe480cb481d9e425aa27f574527 : SUCCESS in 1h 37m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/39978c8e29ad48b4804aff66e6db8c16 : SUCCESS in 1h 30m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1201f2851fe484987b9b6a6560b3ae6 : SUCCESS in 1h 27m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b745b2d69dcd410387a9e59ea9d47dc9 : SUCCESS in 1h 37m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0052a206c3cd4794896d9a548af085fa : SUCCESS in 17m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6a9396319b3e4baba45ed30f65b2e914 : SUCCESS in 15m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87e5c54ba42249aaa11fd91dac9531aa : SUCCESS in 26m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e62593c85c7943009202f3c7dc034de5 : SUCCESS in 13m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bb506898480441bca7cdf6c46f1f7890 : SUCCESS in 35m 55s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4275af4086b24c7b94bdabead3c3ad48 : SUCCESS in 37m 10s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/381a37ecd00748d9acc1720521fd4933 : SUCCESS in 17m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23d97765ebd2481387c9ee5c9292e056 : SUCCESS in 15m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a3d6a255a5a24b148645a90c35efb740 : SUCCESS in 25m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38fff42bf78f40dc982df942459c0175 : SUCCESS in 22m 01s","accounts_in_message":[],"_revision_number":6},{"id":"405fd07cecd7596dcca7c8b3a79f6adcbf135e6b","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-14 08:29:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e272e50ed43ee3eabeed4c9a91319feed17bac84","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-14 08:33:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"03f04dfa1528957d90f61afe33d180c532b1332e","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-14 09:54:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6cc0a6fd8cabe853c06f824179af65fc08fa05f6","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-14 10:04:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4d8d3e02d535cc060b49e0e7ee5ee251a1079777","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-14 10:10:03.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e6fe1befa074a21ba727cecb570e45ecbe156798","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-14 10:15:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ef3ac3468030fff6b1f2f75375e1deaca33fa14f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-14 10:31:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4cba885aa219ffff28162175f4c098d00a675d0f","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-14 11:59:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"213f43a226425893f428adc5a0d9dbce267ac64a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 14:40:13.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"be290773958b81eb95e5b0599b368fe845fcc95c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 14:50:26.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8fe41018d31b9f0496f9b5c99f257cc72e564d50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 14:50:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9ec106f289a3a4f1827a40c7d295e022422790a1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-01 14:53:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9b9215d947cbeaa2e82b6d1a4780190568f60b3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 16:24:54.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebf22c9b64334b3383894e2fcc8ea8db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1bb634edc744cdb220effe3ddb376e : SUCCESS in 5m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6c407f22041489ca75ed2868334f91c : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9c0d8c823c9404b9f0d70fc53ce536c : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6be323f6e77945908bcb8214669e93d1 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e87c1ce8fccb417bbaa9f98bee9e7934 : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/635c2981c71848e5902c7f103d5fb038 : SUCCESS in 4m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/521fcedda5b64adbbe51ec3a29d482de : SUCCESS in 5m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d091a5e2d9943dd80b3f385c860407e : SUCCESS in 3m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dc094f82fc144da5ad96e8c27f36d063 : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4db5b3b5187241488b77390d7d2e6fe8 : SUCCESS in 10m 17s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4ad600974b4644aeaa7559bae3ca9861 : SUCCESS in 10m 57s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/4d3aa084825b4d97a46f53f227831394 : SUCCESS in 10m 53s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86868e3a36ae4f0184b5d835597fd5b1 : SUCCESS in 10m 16s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9fe1c1b7d21645acbfcd6cdfeb1cb99e : SUCCESS in 9m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d139a4305041434cbf77491fe9ddc1e1 : SUCCESS in 12m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0d7e76cff6f4566bb32efcdf50eb7b4 : SUCCESS in 12m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d37e812890c049ad90b601f767088af2 : SUCCESS in 11m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7aa6b82f19e4740aa45f06d8ee3383f : SUCCESS in 9m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c903ebae1064454ea2f443d73b8fcd88 : SUCCESS in 58m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d27b90e6fe944d9ea0d69a9c0171936e : SUCCESS in 54m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fb3989c0e00a48bdac166a9b7de2ab8e : FAILURE in 58m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/773e1a9e811e4f9691942915d0d8013a : SUCCESS in 1h 02m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dcdd76499f1348e1a945370a52327dbb : FAILURE in 20m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c7f081e3caca4781b64a0192177279e7 : SUCCESS in 1h 26m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ca60e10123b44d9a41f0bc887449073 : SUCCESS in 1h 32m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6d6fcd8eab2b4117bb25cefbca7e24d0 : SUCCESS in 22m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ffeb62fec6d14d4f8b8db98070e19e39 : SUCCESS in 20m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f329994134c41f39d7295be35657dc4 : SUCCESS in 23m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a08cfa37685240f58e244837a3206af3 : SUCCESS in 23m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/81088afe3f9843f6af41dd00259bf208 : FAILURE in 28m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5936c240b69843ed926519b38e97085a : SUCCESS in 16m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6f805453d3a842a182ed24fddf359f01 : SUCCESS in 12m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd21d82c7d6641f19a582d33a1a60ac0 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":8},{"id":"a539276c053c2d5ea5f5cef5af5d03fe91550cf4","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-06-01 21:40:49.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"eaaee85275a5998eaf484ea4b1d678fc9161a909","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-02 12:51:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"336379d7d25b440229798c9431e97f378b73c3e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 14:27:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e625f7a2ab948ecbdc811ab7fa8d69d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7467105930d04ad783b5e37e3a879bc1 : SUCCESS in 9m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c8d2bbe483064b7f842784065cb589fa : SUCCESS in 8m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73867552fb364320acdb924a183b0fcb : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a1e992765a475a9ef61e9710778edd : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb3c3238e6514bd28416e633a5b19b0a : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/401d3810b65a47b5a7c1aabdbc084829 : SUCCESS in 8m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dca5fcf0cf34b559b383620102d1296 : SUCCESS in 7m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1841af187de84ef4a0e604d93e56e436 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/791da4e79c50455f8fbcaeb140abbf63 : SUCCESS in 14m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f67a59d0e3345ee8aee89477681e993 : SUCCESS in 12m 53s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c655586fd21b42598cb22c7ed4ca432b : SUCCESS in 24m 02s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/02201fd22909482aaba2b64307335e68 : SUCCESS in 17m 55s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1457b37f510f45a2a87a2bf1944c53d6 : SUCCESS in 10m 30s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0568239d6c7740f1bcb913304be63683 : SUCCESS in 19m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/707a557a0a7c4c7baa3699c4f7f6dbcd : SUCCESS in 14m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bc7463f5642433b99109266fb5ced1e : SUCCESS in 22m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6fc91d783f9e47d69b0caa185b15a49d : SUCCESS in 11m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b0ab2e046d74ea291e1438246977bfe : SUCCESS in 21m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1e2250dc10fc4abb9aa68855ca9c1b6b : SUCCESS in 1h 05m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7128650a05ce432c9cfa8084388a8ead : SUCCESS in 1h 01m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b0efa146878c4a2d8fd9124b0ec44271 : SUCCESS in 57m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac10277e9cdf4ecca9b02accdaa0498d : SUCCESS in 1h 04m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ef28aa7ad5f7467ea1bc2e49c7e07552 : SUCCESS in 1h 04m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5ec6999dbcbf4b79a4edea1f15b22ca7 : SUCCESS in 1h 28m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b0f580c1f134bb687ec4e025bbc56af : SUCCESS in 1h 34m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e0455cf51ef14976b997cf791917f91a : SUCCESS in 27m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a015643cace241888be38852b55a5cc9 : SUCCESS in 21m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c27028edadeb42adb0f0cc8a6df65b23 : SUCCESS in 31m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cce8a6a371494ba6901b3ce959b04b50 : SUCCESS in 24m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cce27ac585244190bb4927cb9b12118b : SUCCESS in 18m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1268a9d773a461cb32dc54a0f4f11e0 : SUCCESS in 19m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/03ea184649494780afc2de84c40db724 : SUCCESS in 21m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75ecf293e5a246919197932d71d6eb98 : SUCCESS in 22m 54s","accounts_in_message":[],"_revision_number":8},{"id":"9e29d9963a88a51d16002abe0f22cb7bdd9fd6cf","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-06-02 15:33:11.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6adc09765354d48803a87a8b731b995108768573","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 15:33:58.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f079f09b8278c6846e38bb0c9cde76e83a53aa13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:54:23.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49b504535ea749c48d73843f93c35594\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4846de2be79a4c66a5254eaca1277fa5 : SUCCESS in 9m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/75f824fbb7b24628a849512780c29b9d : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc60a7841a1e4f0cb0d7d596e4414427 : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/914a8c3554e848748ad35347714c7da1 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27d02998bdd246918cd0660ad39fe06e : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec18694a9da340769090866f2a5d0d78 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c4727f82a77a49e984d364f0b90caa84 : SUCCESS in 13m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca03ce0bbf834e8da77750f0de66b86d : SUCCESS in 8m 58s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fefee52310f34a44952c9844e27b9d2c : SUCCESS in 18m 54s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d60f38421bd34434bac5dce4bd86ea00 : SUCCESS in 10m 49s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a74ac8a45758468f856c1c67067ed87b : SUCCESS in 23m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/191738aa938e46deb82762d23e5295c5 : SUCCESS in 18m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bee9252331444cc4a41d75f1188754bb : SUCCESS in 12m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5d028e2a2bb40fbb6a8cfc1d98003d0 : SUCCESS in 19m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/214e68adab3548f6b4a70435a858615f : SUCCESS in 1h 06m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0832419f439941d193d328bb99757a22 : SUCCESS in 1h 14m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/db4891ad31c6470cab70954683ed4b48 : SUCCESS in 1h 11m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0898b5a547454d76b1f17b0775efe4d4 : SUCCESS in 1h 17m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/166831fad62f47d599fa8ae41822b60b : SUCCESS in 1h 14m 57s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/682e353fa009473e851211f6e09b43b6 : SUCCESS in 1h 52m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ebe4e64d2754e6f8308193b6ced7903 : SUCCESS in 1h 44m 51s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0ef4321d3e83448da80a4f7d6f050c48 : SUCCESS in 27m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/540c92e03c5a4bf8859ddee7de75c2da : SUCCESS in 22m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a862ce77df74112ba9db6aa001cbc07 : SUCCESS in 26m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/06bc6083b7bf4d1787fb92d5122eec63 : SUCCESS in 24m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7400fbc29a7c4ca3b676ef60bddfb562 : SUCCESS in 33m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2790247697b74d70ab1cbf6dc275b7f0 : SUCCESS in 34m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fc8bc74ad46f41c49dfaef67ef9bafa4 : SUCCESS in 22m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ffff96383e443ec99751fbfa72ed9b3 : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":8},{"id":"a337c008ae3a1c452d9ecd1bc5a21445fe893226","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:54:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"2f28df03aebf63ed742b4ce9ac84a5fdb0d2e6a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:55:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7b8de1871e647a08cda83ba02eb5d79\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859283d325b34807aef131388d52d208 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d7396e066c1d093a2ee5fa573cdbc1557e296af4","revisions":{"e460a74544f8c004cf4ed87151129a7ae2301144":{"kind":"REWORK","_number":1,"created":"2026-04-09 15:53:43.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/03/983903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/983903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/983903/1"}}},"commit":{"parents":[{"commit":"323bf9c8df7c8536d36d8526b135700670056ce8","subject":"Merge \"Add custom_etc_hosts_entries variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323bf9c8df7c8536d36d8526b135700670056ce8"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-26 16:18:20.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-09 15:53:15.000000000","tz":0},"subject":"Add context for agents","message":"Add context for agents\n\nChange-Id: If338782e9bba9a13bf4e7d55c1ddb36130683866\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e460a74544f8c004cf4ed87151129a7ae2301144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e460a74544f8c004cf4ed87151129a7ae2301144"}]},"branch":"refs/heads/master"},"e2cf1374a058226eceb1e0f8693871a34b54f714":{"kind":"REWORK","_number":2,"created":"2026-04-28 09:24:04.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/03/983903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/983903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/983903/2"}}},"commit":{"parents":[{"commit":"62fd7413f6401cff2d4f0d2c502e78dc92eb9dac","subject":"Merge \"Add nmstate VLAN QoS map support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62fd7413f6401cff2d4f0d2c502e78dc92eb9dac"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-26 16:18:20.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-28 09:23:48.000000000","tz":60},"subject":"Add context for agents","message":"Add context for agents\n\nChange-Id: If338782e9bba9a13bf4e7d55c1ddb36130683866\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2cf1374a058226eceb1e0f8693871a34b54f714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2cf1374a058226eceb1e0f8693871a34b54f714"}]},"branch":"refs/heads/master"},"e292cf0553351e94d267e82006a468a7bd9725c2":{"kind":"REWORK","_number":3,"created":"2026-04-28 10:01:48.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/03/983903/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/983903/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/983903/3"}}},"commit":{"parents":[{"commit":"62fd7413f6401cff2d4f0d2c502e78dc92eb9dac","subject":"Merge \"Add nmstate VLAN QoS map support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62fd7413f6401cff2d4f0d2c502e78dc92eb9dac"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-26 16:18:20.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-28 10:01:43.000000000","tz":60},"subject":"Add context for agents","message":"Add context for agents\n\nChange-Id: If338782e9bba9a13bf4e7d55c1ddb36130683866\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e292cf0553351e94d267e82006a468a7bd9725c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e292cf0553351e94d267e82006a468a7bd9725c2"}]},"branch":"refs/heads/master"},"e7076a4541cee50dfe5ad7d5f70e68fed4f2fcb0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-04-30 15:10:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/03/983903/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/983903/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/983903/4"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-26 16:18:20.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-30 15:02:12.000000000","tz":60},"subject":"Add context for agents","message":"Add context for agents\n\nChange-Id: If338782e9bba9a13bf4e7d55c1ddb36130683866\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7076a4541cee50dfe5ad7d5f70e68fed4f2fcb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7076a4541cee50dfe5ad7d5f70e68fed4f2fcb0"}]},"branch":"refs/heads/master"},"9a2f23f57b3a777992a03be5eb7e380bee08682e":{"kind":"REWORK","_number":5,"created":"2026-05-11 17:39:25.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/03/983903/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/983903/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/983903/5"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-26 16:18:20.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-11 17:39:19.000000000","tz":60},"subject":"Add context for agents","message":"Add context for agents\n\nChange-Id: If338782e9bba9a13bf4e7d55c1ddb36130683866\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a2f23f57b3a777992a03be5eb7e380bee08682e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a2f23f57b3a777992a03be5eb7e380bee08682e"}]},"branch":"refs/heads/master"},"ebfc523e2d6051004e6ac33c01f4c731a49f6e87":{"kind":"REWORK","_number":6,"created":"2026-05-13 18:28:44.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/03/983903/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/983903/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/983903/6"}}},"commit":{"parents":[{"commit":"461c60de89da5d91dd297ac9b50f92a395f51805","subject":"Merge \"Quote baremetal node create arguments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/461c60de89da5d91dd297ac9b50f92a395f51805"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-26 16:18:20.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-13 18:28:39.000000000","tz":60},"subject":"Add context for agents","message":"Add context for agents\n\nChange-Id: If338782e9bba9a13bf4e7d55c1ddb36130683866\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebfc523e2d6051004e6ac33c01f4c731a49f6e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebfc523e2d6051004e6ac33c01f4c731a49f6e87"}]},"branch":"refs/heads/master"},"f03afaf53d49d82a14075d07ee285b1eb37f2c9a":{"kind":"REWORK","_number":7,"created":"2026-06-01 14:40:13.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/03/983903/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/983903/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/983903/7"}}},"commit":{"parents":[{"commit":"461c60de89da5d91dd297ac9b50f92a395f51805","subject":"Merge \"Quote baremetal node create arguments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/461c60de89da5d91dd297ac9b50f92a395f51805"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-26 16:18:20.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-01 14:40:07.000000000","tz":60},"subject":"Add context for agents","message":"Add context for agents\n\nChange-Id: If338782e9bba9a13bf4e7d55c1ddb36130683866\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f03afaf53d49d82a14075d07ee285b1eb37f2c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f03afaf53d49d82a14075d07ee285b1eb37f2c9a"}]},"branch":"refs/heads/master"},"d7396e066c1d093a2ee5fa573cdbc1557e296af4":{"kind":"REWORK","_number":8,"created":"2026-06-01 14:50:49.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/03/983903/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/983903/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/983903/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/983903/8"}}},"commit":{"parents":[{"commit":"461c60de89da5d91dd297ac9b50f92a395f51805","subject":"Merge \"Quote baremetal node create arguments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/461c60de89da5d91dd297ac9b50f92a395f51805"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-26 16:18:20.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-06-01 14:50:42.000000000","tz":60},"subject":"Add context for agents","message":"Add context for agents\n\nChange-Id: If338782e9bba9a13bf4e7d55c1ddb36130683866\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7396e066c1d093a2ee5fa573cdbc1557e296af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7396e066c1d093a2ee5fa573cdbc1557e296af4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
