)]}'
{"id":"openstack%2Fkayobe~984015","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I3e04902ac1cf129d325d291cafb719fd15a84368","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-13 12:59:31.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-13 12:59:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e04902ac1cf129d325d291cafb719fd15a84368","subject":"[networkd] Fix broken conditional","status":"MERGED","created":"2026-04-10 12:48:43.000000000","updated":"2026-04-13 13:01:41.000000000","submitted":"2026-04-13 12:59:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984015","cherry_pick_of_change":975173,"cherry_pick_of_patch_set":8,"meta_rev_id":"1ec2c2d64e736d45a82a6099fc5c61a684aefab0","_number":984015,"virtual_id_number":984015,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-13 12:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-13 09:46:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-11 00:04:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-13 09:46:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-10 12:48:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-04-10 16:05:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-11 00:03:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-04-13 09:46:04.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"7645bb2c3c00a3ba61fad71dc88e477db4713243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-10 12:48:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f597cff9c626594f7c76ad0eed465a92cfa57024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 16:05:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57f8d6839cd041689352be93556522be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cd8271e76e64f7592a2834d30401c28 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcdec21d850d4c28978423b1ca8202df : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a41a21b4b4ce4283a212405c06b77614 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67b8694fd6524628b314eb45e702d698 : SUCCESS in 10m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b9b5a5c74e5b434480886102f3fa09f7 : SUCCESS in 3m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03e4fe99bfb74618b15a1f3719180d0a : SUCCESS in 6m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f585f73643554da39b31fbf98e62068d : SUCCESS in 8m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/155ae1b7793844699766ff0d05d66576 : SUCCESS in 12m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d9b95832e2d742859362e4eeea1c92b5 : SUCCESS in 17m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a37d6e753454a33a89e14fb33a3edb7 : SUCCESS in 13m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dc4d8eced2914fae912339db3336aec0 : SUCCESS in 15m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2122364701964126ab160625f74d7204 : SUCCESS in 19m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3ae3a17df29404fbe9471c3c72b6217 : SUCCESS in 15m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b2d6311235144c8083b49b44af40bd66 : SUCCESS in 35m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/89d47e14aa81413d8676cf0ff049a910 : FAILURE in 59m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3a2dfbcb968b41d3a5fa186e220c9c40 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/78549dc3be8442bf838b73f91def6b43 : SUCCESS in 1h 07m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4b0ab92c52804b5495d528f095babb1b : SUCCESS in 37m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/27e66e810dba4e4a91eddec33284d7db : SUCCESS in 1h 16m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98024be36e0941a29f861c3b184cc465 : SUCCESS in 1h 02m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/664a9523f2ce419fac04979f5f22a01a : SUCCESS in 47m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7da4dd8270054999a69fb82a231d5b7c : SUCCESS in 2h 22m 12s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a5d14e70c73841a2bf491319c45d1a87 : TIMED_OUT in 3h 06m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4aab9188a6f84869b3169a25f8d0be52 : SUCCESS in 2h 07m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/568ed74ac0bb4b44845937b6b5f5a64f : SUCCESS in 2h 08m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6ed099f10f1042aabda59afcc6ea1228 : SUCCESS in 23m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5314d227c640477d8cdb0707f8bc7d1f : SUCCESS in 20m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7789591270d248fe8a42c042d0cafff7 : SUCCESS in 22m 06s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3bb95de7a00144a183bc1294abca845d : SUCCESS in 21m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f3b6699172e49cbb3c24721a946160c : SUCCESS in 21m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/285bd16978794ffdaf95d783c5469c8e : SUCCESS in 15m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ddc22870e290430bbd8d2ee031c7211c : FAILURE in 4m 57s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b0475277d6d244dfb50a45981eda98ac : SUCCESS in 34m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69f682026ff94045af97c4a59b5b6239 : SUCCESS in 36m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f162c4c8c23f492d85ebb271240fde10 : SUCCESS in 9m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6d1a90d8c3704db2985f2a28b298eb3a : SUCCESS in 16m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a06117a6b2394dc88baea0aabdd516aa : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0a5e8acc4f16362daeaace6da2e4cf1d47ca5b8d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-10 17:18:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e3eb4c6f6149be668107f787d68a222a97c8469","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:02:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17512e61e7d44164b470516464741309\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d63af5d5ee234404b1b970e6f1e8b8f3 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85b2ac0269af49618c907a75e4691bc3 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8e5b1016af74451a98f0ee736c1e9f7 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7fe6bba7e14c21a27d52a1ed7c594f : SUCCESS in 9m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af05b81bf74b421fa6ad0c539d1685d0 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1df8e2e4be5e452d9b7ca2142da0624b : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/20f239afffda464a9230320c03a8a684 : SUCCESS in 9m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/127eeabac1ad4aecbe031da37d6db22f : SUCCESS in 15m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e93fdd459bd84c48af50f94ff23294a5 : SUCCESS in 18m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e004273e0f5c4dd79704406c807cc523 : SUCCESS in 7m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8604631f4dd740bb9b5d995e25baeb74 : SUCCESS in 23m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3bc8403976bf434e9b4480f9f2d43a3b : SUCCESS in 24m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ca1d80f7fa34724b31f652052a9260d : SUCCESS in 10m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e802ba9373a5400aa1bbd83fe467eec9 : SUCCESS in 1h 07m 55s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b9815d8e4055466d8e645f258efb5e77 : SUCCESS in 1h 09m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a4a1d5ddd37c413997da59534fce9ce4 : SUCCESS in 1h 01m 48s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/860afa57b25c47d9a17ac7074b0a44df : SUCCESS in 1h 12m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/eb09163a38d84b319294f54790e8b10e : SUCCESS in 35m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1dc5c7f5e9ac4edcb345442f7330af8a : SUCCESS in 1h 10m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e32c0006f6e41b59223f08e3fa87f1d : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59472c5571d74881b3090673e1536bf7 : SUCCESS in 28m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cee3b2d43e0c4bbda8ad7735657442c3 : SUCCESS in 2h 31m 06s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1361c840d8f9417392c99cedc33ea32f : FAILURE in 4m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29aa360e2fbc40738b2ed8d1fae47903 : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edc232e0395e48b9b67f638462f4974e : SUCCESS in 2h 33m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/262fd03313ab4c03bca61ccaeda0daea : SUCCESS in 21m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/93398367491847a3803479ed427c4048 : SUCCESS in 9m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a09fbfc800aa4315afae8d5e7dc287ef : SUCCESS in 21m 34s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/bcd7449a4e8c4cd2b961be8260fd71bf : SUCCESS in 19m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da4c74b9dfb841cf8ccdc32e5144aa9d : SUCCESS in 18m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e9242a08b39c462a85795d3a075168df : SUCCESS in 19m 18s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1508650a57224ef0a300d5cc1ae5e2e5 : SUCCESS in 24m 07s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/19cb323c028c4030b6afc0434b07909c : FAILURE in 3m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efc22e2e85df41fabc312c335b802ccf : SUCCESS in 36m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/34b6372631984037a2ee43706707577b : SUCCESS in 17m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0de29980c1454a5da49ac925f17b15b7 : SUCCESS in 17m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25bd6fceeea5458e855f962778216bb9 : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":1},{"id":"83335bbbb007effdc8ddf03325673998d1884c88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-11 00:03:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6c37c701d8a9a0c3b7be0a2e02c14d8072a05f38","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-11 00:04:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dd0311e2dc59f4294976679c2174f43444855e97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 02:37:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5db7003d4cbd4cb2b16f77fe9c0eb69c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c728457f3cf84cd58523a6703b578ffa : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a8b09bfc1343f896e6fd9d30273048 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0404f6daeca8493aaf96cf2430e62181 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304373ecb15a4022947fae67685bddf2 : SUCCESS in 9m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/238cb05e026f45fb8420441cac5f883c : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4cc7432f00a49bf83b9772d5089686d : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/722766bf4579464d9dbfe89e125d89b7 : SUCCESS in 10m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d0d8c5b3969147da8c453eb30b064bf0 : SUCCESS in 17m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5f17bfdfed5444bba88fe5c4d08e9c48 : SUCCESS in 14m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd415434c50b486a90296fd9c145a9db : SUCCESS in 12m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0eebc749288b4d58896f53ecfe7e649d : SUCCESS in 14m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7355835dfbbd4ad1bb96b4694eee19d8 : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d598e34e419440108fda88158d280f68 : SUCCESS in 14m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fb04ef480a32430e9c19a489779e03aa : SUCCESS in 45m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/70a9c509752f435d94f74e861d3884d9 : SUCCESS in 32m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0adaa9d058584aba8c6167cd970a23de : SUCCESS in 1h 02m 18s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/5fd14a08a1d14b5d9896b2886e34dfd2 : SUCCESS in 1h 00m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/760bbb82b48c42a1913f8774dd0a19a0 : SUCCESS in 1h 16m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9b696b1b34b64d5eae63b661cbc06da0 : SUCCESS in 1h 13m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac0db9f497f245f6b40791ce9ac5efb2 : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4dfb48f6e8444844b98476045e669712 : SUCCESS in 56m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0e6ebed69a154775ac576f068cc07811 : SUCCESS in 2h 15m 31s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0410523d28ad4609a41b3fffeb514e3c : SUCCESS in 2h 21m 58s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58ba84a72a73408280a0ddfbb25760ce : SUCCESS in 2h 06m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6c3fab9503446d19d0725a7f9c538ae : SUCCESS in 2h 31m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e3be582f740243c4aea0bf4501e46ac3 : SUCCESS in 9m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ce4317ca94984411ac2c34b66f62c6af : SUCCESS in 9m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/80440ff5d35743f688409d2df0a3a9ae : SUCCESS in 18m 38s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fc219a4399e44184b9fdf4b95a6f342a : SUCCESS in 18m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6449ac19a57f481eac1d9f83386f5cf1 : SUCCESS in 14m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1ee3580f451441caaeb7e309fde9275 : SUCCESS in 14m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d7ddef6a43b34c63a40b9b01a26706a7 : SUCCESS in 21m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d7a99af75e494f3889a76a8de0a2e9b7 : SUCCESS in 37m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52c0abe8cabf482e89a113bf2328fb51 : SUCCESS in 29m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/511cc12a87164ba5926f2344a6c10e0a : SUCCESS in 17m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3c0f820516e54470bf74e4746b376cd9 : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/caf1b26ae8fe4a20872682ad1450a85f : SUCCESS in 12m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3402fa26394ae17479ce83492db1eccc45de7ce0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-13 09:46:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91f16a6b040ef2d9420a3730a29bfe230aeaeb5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 09:46:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"949f58e16302fbb606c46f13a3eaa9400b27d223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 12:59:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/168da44b710a4ff99115add2f554fc97\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c1c41e019b84ef0b78237253f9b602e : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e625a87b54734ec58b07c6ad3c980b71 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1874fc49a614ce69acc401f4694f07f : SUCCESS in 9m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8340eef88a244662a912696e375691fe : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe67805d72554936bcd83c0fdca4df66 : SUCCESS in 13m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/427056636e434c529b07f3505369aa83 : SUCCESS in 10m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4edfc99f5e33426a8b2517ad7015ebe8 : SUCCESS in 16m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/33f9b9a691954e5992c395123d6ed9a7 : SUCCESS in 16m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31ae4bb4005e4832adbcfc878d7c1448 : SUCCESS in 13m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ab67483c9730422f915a7fed1957b9eb : SUCCESS in 19m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/79e5522486bf47cb925f068552214dce : SUCCESS in 22m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bff0389e06c4b608295aac6683ba61b : SUCCESS in 14m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fe1e9d0984de4792a2f6d9c9e79dccd4 : SUCCESS in 1h 11m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/52f615ea91df440ebefe3271961b2b27 : SUCCESS in 54m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7e98b2c7b7ef498baccea33c741d3da2 : SUCCESS in 55m 48s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7c55974475d841c3ba665adad88dc9b3 : SUCCESS in 58m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/21a2f063467140858ba70667e4e8a649 : SUCCESS in 35m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/447a2107789340a1b3a724b945d80b23 : SUCCESS in 1h 12m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b54b789d0fb4e73b7511740fbc55bb5 : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ace1796586f4484a63ca39919dcaff4 : SUCCESS in 58m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c3f1e520f24c43bd9e3f61c967d02236 : SUCCESS in 2h 33m 15s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/57f1a0ffbec54dd3870dee7d99ed8913 : SUCCESS in 1h 26m 29s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a130c841d8874894908c1e085e9f281a : SUCCESS in 2h 09m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da80ccb991814016b3405221bff4598a : SUCCESS in 1h 55m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ff8fc374bb59441f91d7b7bbb5fbe10a : SUCCESS in 16m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/21e2fbddb0c346e3abdf63f0304d3757 : SUCCESS in 17m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ab95e654940740dcb1cade7e9cea9423 : SUCCESS in 20m 58s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/60af919758164e1d9d519e499fd209f4 : SUCCESS in 22m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed4f2c9f82f2484d8f881b6da421c4cb : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/21b44906e52746608dc9252bd864b485 : SUCCESS in 17m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/08720d8e80ca434f9dd23476467ed654 : SUCCESS in 35m 55s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/65447a70dded4123bcf2dcd4c9483d18 : SUCCESS in 35m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/758b6e400df14042a44b66e50b54c746 : SUCCESS in 37m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fbaa65dd84c54000a626d3c7047e37e0 : SUCCESS in 19m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/70e428dc546445e9835e7e7630475207 : SUCCESS in 15m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d42d588677c64a348dcb7c5782ddd615 : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ade4d1cbe5e4db74813dc481631cd42a9d5fa5b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 12:59:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1ec2c2d64e736d45a82a6099fc5c61a684aefab0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 13:01:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/725f2dd88df34c65a82b6e1d0c04d83a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fcfa91735504d3f8ef61936284b4859 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5df157b38abc992f4ece634d9473c8dcc024330f","revisions":{"2aa2f5e39cbc09bd0b69181f99ce2844c386caed":{"kind":"REWORK","_number":1,"created":"2026-04-10 12:48:43.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/15/984015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/984015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/984015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/984015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/984015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/984015/1"}}},"commit":{"parents":[{"commit":"4687b7635cfb41c28401325029978dd695e464cd","subject":"Merge \"Support Python installation through Apt proxy\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4687b7635cfb41c28401325029978dd695e464cd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 16:00:32.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-10 12:48:43.000000000","tz":0},"subject":"[networkd] Fix broken conditional","message":"[networkd] Fix broken conditional\n\n    [DEPRECATION WARNING]: Conditional result (False) was derived from\n    value of type \u0027int\u0027 at\n    \u0027/home/ubuntu/kayobe/ansible/roles/network-debian/tasks/main.yml:51:9\u0027.\n    Conditionals must have a boolean result. This feature will be\n    removed from ansible-core version 2.23.\n    Origin: /home/ubuntu/kayobe/ansible/roles/network-debian/tasks/main.yml:51:9\n\n    49   command: \"udevadm trigger --verbose --subsystem-match\u003dnet --action\u003dadd\"\n    50   changed_when: false\n    51   when: network_interfaces | networkd_links | length\n               ^ column 9\n\n    Broken conditionals are currently allowed because the\n    `ALLOW_BROKEN_CONDITIONALS` configuration option is enabled.\n\nTrivialFix\n\nChange-Id: I3e04902ac1cf129d325d291cafb719fd15a84368\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit a46f6484f8b361c322533bfe60b6cfe12aeca85a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2aa2f5e39cbc09bd0b69181f99ce2844c386caed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2aa2f5e39cbc09bd0b69181f99ce2844c386caed"}]},"branch":"refs/heads/stable/2025.1"},"5df157b38abc992f4ece634d9473c8dcc024330f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-11 00:03:46.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/15/984015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/984015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/984015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/984015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/984015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/984015/2"}}},"commit":{"parents":[{"commit":"8c73b2d60d9507f93f813756c0f4f18a4b74a661","subject":"CI: Use main Rocky mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c73b2d60d9507f93f813756c0f4f18a4b74a661"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 16:00:32.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-11 00:03:46.000000000","tz":0},"subject":"[networkd] Fix broken conditional","message":"[networkd] Fix broken conditional\n\n    [DEPRECATION WARNING]: Conditional result (False) was derived from\n    value of type \u0027int\u0027 at\n    \u0027/home/ubuntu/kayobe/ansible/roles/network-debian/tasks/main.yml:51:9\u0027.\n    Conditionals must have a boolean result. This feature will be\n    removed from ansible-core version 2.23.\n    Origin: /home/ubuntu/kayobe/ansible/roles/network-debian/tasks/main.yml:51:9\n\n    49   command: \"udevadm trigger --verbose --subsystem-match\u003dnet --action\u003dadd\"\n    50   changed_when: false\n    51   when: network_interfaces | networkd_links | length\n               ^ column 9\n\n    Broken conditionals are currently allowed because the\n    `ALLOW_BROKEN_CONDITIONALS` configuration option is enabled.\n\nTrivialFix\n\nChange-Id: I3e04902ac1cf129d325d291cafb719fd15a84368\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit a46f6484f8b361c322533bfe60b6cfe12aeca85a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5df157b38abc992f4ece634d9473c8dcc024330f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5df157b38abc992f4ece634d9473c8dcc024330f"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
