)]}'
{"id":"openstack%2Fkayobe~984138","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Ic107bc1e21806cc6549ac86d8ccc83377fe2cfba","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-11 07:58:41.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-11 06:35:17.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Ic107bc1e21806cc6549ac86d8ccc83377fe2cfba","subject":"CI: Use main Rocky mirror","status":"MERGED","created":"2026-04-11 00:01:55.000000000","updated":"2026-04-11 07:59:33.000000000","submitted":"2026-04-11 07:58:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984138","cherry_pick_of_change":983990,"cherry_pick_of_patch_set":6,"meta_rev_id":"05fafd67a22f34dd8fa32dbf0f2728a9c203aaaa","_number":984138,"virtual_id_number":984138,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-11 07:58:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-04-11 00:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-11 06:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-11 06:35:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-11 00:01:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-04-11 00:02:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-04-11 01:28:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c36ab8ade2c445fbb5f34280d80abedf93e05322","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-11 00:01:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4132e04b1b6a34c3306b8814844ace76e0d57527","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-11 00:04:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dbf7b498023e29b62748255f0f34d19bb8ff50ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 01:28:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f8426bb775e49d8993fed8ce40d3ac4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ce70813e8d42f3b9705ad8ef174563 : SUCCESS in 5m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d9d74e900a3d4b7fa602f2586ef5a8f6 : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a391dd2f86a4639920cb95354641c56 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add70176ad9e464caadab091d2d786fb : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ccc461cbd3034675ae1fe16ca7a4ec74 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0417a25aa834f3eb1477563e171974d : SUCCESS in 4m 19s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29f4846969cf45948259eee9bd7d9ca9 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/01a58cf534c844fdacf99ac61ace5f97 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dae1d7d633d841c4bbe81c471ce4e099 : SUCCESS in 13m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/af77270fd9534025a60e71e16a2e014a : SUCCESS in 9m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5998dd8c57d4c2783f5112d6d0f3a4d : SUCCESS in 7m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/be24b64a9c254a129ae7c86841595b56 : SUCCESS in 10m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e3363b21c83439d8ccf294e0dbca5e4 : SUCCESS in 14m 30s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d0fac96757204a928179f3b3c5ef4ee4 : SUCCESS in 52m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/348e70fad6be4fa3aa73ed7c7726b00b : SUCCESS in 56m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e424297c820841489a00721a72521668 : SUCCESS in 56m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ae1734d332441ae9ca2ae0b5a4d93db : SUCCESS in 56m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/27c0d393713d463299431b1cdf07c3b8 : SUCCESS in 55m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e349aaef7f6b49cf9f49265e7e093d3f : SUCCESS in 1h 25m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f20aad703254066b93a65fc38cdbdf5 : SUCCESS in 1h 17m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/24119593957f4cfb9bf8e7cd22806ccd : SUCCESS in 10m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d0ba4068adfb4f75af8c4c53f3889df4 : SUCCESS in 9m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab0933c1e8254fc89cd500a9ed29025d : SUCCESS in 8m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/42556782df104610b865e644e399276a : SUCCESS in 18m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/807a3546d58244f78c468f5bacf3d72e : SUCCESS in 15m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ba571d3d57c43b5b6bd97974373f3e9 : SUCCESS in 22m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/11d747b32eed44aea724aeb6de0bc8c7 : SUCCESS in 9m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ead0b758e2445fbaf2b437238bb0efb : SUCCESS in 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"31eb413b2d3e102210793c3d9f9d8b77cf2e274c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-11 06:35:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de8f22cb8a47d07d67e6f9b0738463cd64553d50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 06:35:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d605dd9c0b5a88aa387e796e6bad589e7fb4064","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 07:58:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c82be8e5c5ff4df8b2156f6e66b8441f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422e2d00d7414f0bb8364487fb855b41 : SUCCESS in 7m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9a6e2e0255ab4e329c3cb21ef79131a3 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef9b4461f314a95b70b57f7af26ed1c : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e3cea2048e14e96a35456f7735cf5ec : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0186f1d390af4639854fb59120fc38ec : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/59575e85fdf74745877ac1c633a69e2a : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00807f4975a84d4d84626e47cc79a84d : SUCCESS in 11m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bee867facaaa40e293468ba439ca4531 : SUCCESS in 15m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6615f21009fd4b0bb1e6e63e126f4963 : SUCCESS in 7m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d56e528d38854cc7924032c9aa3fc97b : SUCCESS in 25m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ebe8e33821e4ab586ddd996c344c68b : SUCCESS in 14m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f707c47a0ff64b26aef4f8976eee70f0 : SUCCESS in 54m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3c5c4dd00aeb474588442984769cb915 : SUCCESS in 56m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4660070cedac4384837c16ce899c0fcf : SUCCESS in 57m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc54b1714cc241d2a0a033b4b008389a : SUCCESS in 1h 14m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/854662af5eca4b02b3c73d8a75bf3991 : SUCCESS in 1h 13m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0035af5f4332415e86c3822e9097c251 : SUCCESS in 1h 22m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65e95213b89e4fe4a761609ebb120a05 : SUCCESS in 1h 21m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/13e1104faf124ed790498b3bd85d2f17 : SUCCESS in 15m 20s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8293b130425b424ea9ee8635a1ec8931 : SUCCESS in 15m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ef4b667f6b4475183fae2a32ee3e738 : SUCCESS in 16m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2ebfad947107450aa35791b3ba21ccfb : SUCCESS in 8m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/989a022b9feb4d22889b5caf8a874a0c : SUCCESS in 16m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/216858d182394a55b65db3a7fc3310a3 : SUCCESS in 27m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e92fbfeb7f5746e0b3eb9cb1af4e27b6 : SUCCESS in 9m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d9cec478068453bb23d1f5ace16b074 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ff4cdba579c5cea68603e0e75ca1121bf4cd2513","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 07:58:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"05fafd67a22f34dd8fa32dbf0f2728a9c203aaaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 07:59:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668ab3c8ea874dbfbbc32980aba8673e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1837ac13cdfe4063b87236ac096ef131 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"831847b7a1ecc2b26cbf2481acdbccd2eb0c9ad9","revisions":{"831847b7a1ecc2b26cbf2481acdbccd2eb0c9ad9":{"kind":"REWORK","_number":1,"created":"2026-04-11 00:01:55.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/38/984138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/984138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/984138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/984138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/984138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/984138/1"}}},"commit":{"parents":[{"commit":"147ff934d0dc8881b093acea99f65bfe9d926d14","subject":"Support Python installation through Apt proxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/147ff934d0dc8881b093acea99f65bfe9d926d14"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-10 09:10:15.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-11 00:01:55.000000000","tz":0},"subject":"CI: Use main Rocky mirror","message":"CI: Use main Rocky mirror\n\nWe are frequently hitting issues downloading the mirrorlist in CI.  This\ncopies the change from kolla-ansible[1] with the hope that it will\nimprove reliability.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/983965\n\nChange-Id: Ic107bc1e21806cc6549ac86d8ccc83377fe2cfba\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit 71dee319de0cc99e7657a7c56684d541bc779ee3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/831847b7a1ecc2b26cbf2481acdbccd2eb0c9ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/831847b7a1ecc2b26cbf2481acdbccd2eb0c9ad9"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
