)]}'
{"id":"openstack%2Fkayobe~984741","triplet_id":"openstack%2Fkayobe~master~I50dd57a8535d45d74d0cc9236f7a49792b962daa","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-16 14:54:53.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-16 11:31:56.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-04-16 14:54:53.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-04-16 12:03:29.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I50dd57a8535d45d74d0cc9236f7a49792b962daa","subject":"[CI] Fix various issues","status":"MERGED","created":"2026-04-15 13:01:50.000000000","updated":"2026-04-16 14:57:21.000000000","submitted":"2026-04-16 14:54:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984741","meta_rev_id":"86d29833b30d6edbdc504f6d1187c39d8e714177","_number":984741,"virtual_id_number":984741,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-16 14:54:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-16 11:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-16 11:31:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2026-04-16 12:03:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-16 11:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2026-04-16 12:03:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-15 13:11:33.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-04-15 13:35:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-15 13:54:50.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-04-15 14:38:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c23b75e77de2cf7e576ffb2822e544cec62d70fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-15 13:01:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"927b0707dafe6ad1d2d482347e219bb1ad530caa","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-04-15 13:11:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83cf0ca562da2f10d47fa06085716263bceb395d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-15 13:35:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a921f539bbc1d67d0e3b60d53cb489408ac6a254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 14:38:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19819acc84a3481fbc75a3554694303a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0c7d74b3384008922c32f95220ac4d : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/875fe23d8804416589d4b6ecb3e3b76f : SUCCESS in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3038d67ccfbe4ed388b2be7a7d88579c : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e323116cd2546a8bfe2ca8c5830e11a : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbdc90a16162423c863c70a26302b3b0 : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d81f053b817847fc85f8ff23c46eb052 : SUCCESS in 4m 22s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff762227d1c5454397d684754bd6a894 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a0babff3747b4228bfde7dd59a6f1837 : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fd1cf42c4d33429ba836f53596b864de : SUCCESS in 12m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b91795d651a348f7a48dab10daea55fc : SUCCESS in 15m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e54db31921e4277919e0c12d6b461ac : SUCCESS in 22m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f8dfdcf208c3486e9fbd751bb39f3525 : SUCCESS in 12m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07d50cfbcdf048b9a9d2b0a1db7c7595 : SUCCESS in 11m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8432ab2c9ae34324bc9ad27ce2154940 : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bfa0c869d1df41ba930adae643f44755 : FAILURE in 1h 28m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6b221e6214384b2eb46945f04c2ebc64 : SUCCESS in 1h 06m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccddde9e2e4c4bbaac5b3986f3646a48 : SUCCESS in 57m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/57d4b08aa41b4aef87c4abf8a27b143d : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/06de6d10174f4ad284a04dd0c69f18e5 : SUCCESS in 1h 35m 36s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7c7abaf42e2041829f4684c2177a585c : SUCCESS in 1h 30m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2ef3cac20214eb6acf983a44c207f2f : SUCCESS in 1h 31m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cd4d934392344108ff02df71fefd25f : SUCCESS in 1h 36m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/92b1f6b8945d4b338f674c3b75ed682f : SUCCESS in 15m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/40f808f629f840f2a0c2f29defc5b30f : SUCCESS in 15m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/430045b504884c10973290e99f1db41c : SUCCESS in 22m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3accfc778e7740a699af88f3421bd5cf : SUCCESS in 22m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f49921dbb20b4abd9d1d160556d66e7e : SUCCESS in 36m 54s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c4a2744d0d0e409a8cb20c8affce5bb9 : SUCCESS in 19m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d0e0063b5414871a5aa0009a2146518 : SUCCESS in 35m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8294a109f2864bd5a2007c84f4bece74 : SUCCESS in 32m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/070ecdcd80a54abfb7ea60951560d992 : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ab01db1c37a43fb870e397726fe48ab : SUCCESS in 19m 15s","accounts_in_message":[],"_revision_number":1},{"id":"679dc0e90185a8ae5f6b352a88f893372b392939","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-15 14:51:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"772806fd4ae346aca10ff94d037acd687af59aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:47:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4b6d5b8f73d4fa38121ff11592e1539\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37ecc6cd85a4a30a4a9585658628968 : SUCCESS in 9m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/849ce07b2c5e4ab5a20c04f80f80f5c0 : SUCCESS in 8m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55827acf1d1145989b8ba6ab259135ec : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76ee74c89604a27a2b5550f9afa945e : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd2133bb7d6d415f834bf7649e67dc64 : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/552c1a5b4b4244c38342a14f9eaaf16b : SUCCESS in 7m 14s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e66e87419c3b48ca871828731d73ea2a : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a484872a90bb45b5b64143c370e0147e : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/340bc07dc8ce42459ebb9fff28fc94e2 : SUCCESS in 17m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9635286c85c74d3db327284c23ee3c89 : SUCCESS in 19m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca9e779e3586467d80140f16d94674d7 : SUCCESS in 21m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/935a7f6bbde945ec915c5246ff881b68 : SUCCESS in 25m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be32c866a4b84e119bbaf80eeafe33ad : SUCCESS in 19m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/afacf557646e472581996b8b490eabf0 : FAILURE in 45m 45s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4b8e45354d6144f7a318f2a63810cc66 : FAILURE in 45m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/70c5a30c390a4fecbb6544b217f89ced : FAILURE in 55m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6448681d521f4aaf9d7a1cbcf9a3ce26 : FAILURE in 42m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f78bc555f6e4ceaab81d8e90a0b4ba0 : FAILURE in 38m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fb584f6d08b045ba8dfd448fa3d0ce74 : FAILURE in 42m 38s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/973b8d112a654e888960e5b83a9ef813 : FAILURE in 34m 26s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25bbc7d878344a42acb72464e12e357e : FAILURE in 32m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/074f9cb7000f410eb0db1f00b805d3d1 : FAILURE in 35m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a9252e8ca9814677afcab2a19d65bf90 : SUCCESS in 23m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/598253a2ee904ecabea7048945b04ca8 : SUCCESS in 20m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88121908b0a2436ab70a5ae7f24238f4 : SUCCESS in 24m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6cf5ab96e022484a9317218c141c027e : SUCCESS in 21m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d2db8b5624b34b95a39e15d27ee68a54 : SUCCESS in 26m 13s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/410d5d7bbe85473a95559295c622f947 : SUCCESS in 43m 24s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf0eab69aa3e4178a05c141c0d6d9747 : SUCCESS in 32m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a6f11487e774a6c9f29c584fe9d86dc : SUCCESS in 35m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a051cb637af24031a688661d4e2da9cb : SUCCESS in 15m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e75ef3237e2e4075a03bd42c35cbd149 : SUCCESS in 22m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6aa967e137ce65649f6c4cb437dcc8f3cfc0cd54","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-15 16:04:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42b3dbff57e53ad39057a334ffd665a5100b9e41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 17:09:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/289b4afe5fe3408ab4b6251494a0b9f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4845d6594f3344febd0dce153a04c024 : SUCCESS in 9m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d53d8327d2d943b39eecaa0941c14c51 : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d02b5369b2f444d956c020bd8699499 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cde1aa2560234564b939786f211b738f : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c5078434e3649328fa212c8581a0113 : FAILURE in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd9b2b6d62da486a9ab38be8f47cc9d3 : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7927112963974d7c8693f1abd482dbe7 : SUCCESS in 6m 48s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3fb11b4ef49f4992a45049e76af4cbed : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/943f918ef6574e47bf7c02e9da5d86cc : SUCCESS in 11m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9f610197f617405eb175570f91537820 : SUCCESS in 12m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bbb2c856c3e04a62b4554fe5038b396b : SUCCESS in 16m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbb43c3508a64ab7a36e85859160dfe1 : SUCCESS in 18m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/61b5293941324560a72e77865989ef2e : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/baa7d9a5a87749aaabe55172077eca01 : SUCCESS in 16m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0adfeb3e757b4d7590d304ea43c45bd8 : SUCCESS in 1h 00m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c64c67b4d8bf42a19fe257e1173d2544 : SUCCESS in 1h 03m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bb91b0b18d2b402ab5719a4dc05711db : SUCCESS in 1h 04m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45050d9284b046a1b9d82571074e3413 : SUCCESS in 1h 03m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/87a931d53148462d9b1ad71b92588d5b : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5de481751ac142c692a9d672013a3584 : FAILURE in 38m 40s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e1fc5e9ffd03460e943d9cbfd1306b21 : FAILURE in 36m 01s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6241616a1e2d4989bcfbb7d47d56719b : FAILURE in 33m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80e279e0200c4dcea2af5217891baf15 : FAILURE in 34m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/de5ab90171e441498d7d92fc3a09dd6e : SUCCESS in 26m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1ed384559b224bb7a282efd2fe089673 : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18d52d628ffd4793919ec81496b0f663 : SUCCESS in 16m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7838a869815445e898fd0c5e363f3bc4 : SUCCESS in 13m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/00982af69010439eaba87ecf81bf5807 : SUCCESS in 27m 58s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2f9a1d02b52946899878988589921c90 : SUCCESS in 30m 12s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a967920f0d24cfea0b5cacb30b6b038 : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82d548af5b4a4276b14f67f52232fa23 : SUCCESS in 27m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/26c0cc40d5434c37b8b21c7ea31b9084 : SUCCESS in 13m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d58c3cb9351d4e879c6ca3235a61f229 : SUCCESS in 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"972afdaada5ef1d0532aa53584f0f1aa55c87fc3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-16 07:56:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"350e939a0e8c73eac0939b8cfce80a2a93f1eed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 09:09:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3ce3c37311d456d91b81b3345f2857f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9d640610ae487e84d767a938d65f7b : SUCCESS in 9m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99ea9c0f74704a7891c7fd3d95b76838 : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbd3afbcbd644e3eb6ff45d02b377118 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be52039f65a41438f4a5983efa54ce4 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c33d1c1db1274e8fa6b09b2f2376c53e : FAILURE in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c9c4439be53463dab3c636f1939383a : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f78797a68467467c9fcc70db496df1e4 : SUCCESS in 5m 39s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4abed95e037e4608b88011daaead585a : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ebc4a151df634e349c59d0331920592b : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ed0e13a9f044ef2b846e01bb2ad2779 : SUCCESS in 14m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eda84e65beed4b219ecd02d091e90b88 : SUCCESS in 16m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa2ce07f15b04a85b700879388103717 : SUCCESS in 19m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8a5666de06bc42a787a453513cda049c : SUCCESS in 17m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b57b3931dd0c4f71b7cf9332779f6981 : SUCCESS in 21m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c2c2c08de26f4827bc1002232163abb7 : FAILURE in 53m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ff2ee3329aa14d4bae4079f670517410 : FAILURE in 47m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/91ccf597788445fcbc2e3f775beab91e : FAILURE in 53m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/727d37c900794bc3ac4e7b8b69498d55 : SUCCESS in 1h 03m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1ac600919af94d289b380c1905a6ca06 : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6b2e5bc76e244db8a465494ef874440d : FAILURE in 56m 45s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c821db4a0a0545fb94fac609df5cd0ac : FAILURE in 54m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d86ccff9f78b44ac9dcc5f506d092ce2 : FAILURE in 51m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/350ba84abf8e4628aeaedb7db0c2da47 : FAILURE in 1h 11m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5a4b171e7c2c4749bd395a6a485c21b8 : SUCCESS in 16m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2fbf13b312174fc8abcd8b4114f76468 : SUCCESS in 16m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce1331e98f444bbb9d7f11062d79afa5 : SUCCESS in 21m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d7cc51225e4434ebb310ca61cf39935 : SUCCESS in 21m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c06273760ab1441989a8b4869da9a161 : SUCCESS in 25m 27s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b89d9e6f6a7f4bbea98c6e604d9a2c60 : SUCCESS in 28m 25s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6257e20cf4614f2e88ccf7e4730e4eae : SUCCESS in 32m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82a103b924a24f0d9153f62599e630ed : SUCCESS in 34m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/488c0713e78a41c297c128f9b5ae8dad : SUCCESS in 27m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3909438d2274b2dbd06776e63a5338e : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":1},{"id":"132cb0787c59e47b3633748a48f28b8abb593e8e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-16 09:16:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e77e40ec2724d20cbd8fe99df94f7a582975ed6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 09:25:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b64d0bb7ec615f4f5ae0eb349f14e70826cf872c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-16 09:28:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c0b72b54bd98272ea1e6f3de65b64c59053b4975","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-16 09:33:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1c6bbb7e21f1bc252f264630da482210e81e1f65","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-16 09:34:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"80e1be81eea6fe36dcb903ed02118d156bf4e692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 11:06:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ebe8ac50f524b47aff5b0e11eba217c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cff80ebb3d2441a97cbb96b0e86ea98 : SUCCESS in 8m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5ee736d3a1fd4c448135fb51280f2557 : SUCCESS in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c487e8f4a9f43379a9c3fadcc7955c5 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/341912a5622a4bc98c7975de8a618256 : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ad99469e0ed4992ab3c322993dd6d71 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f023526e49784523aeda14ab84190a74 : SUCCESS in 6m 30s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d177ce6e06a4c099e41b350234cc7d2 : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f902c681f8954884a42c2e154a474160 : SUCCESS in 12m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7608e3d8e74f40c48db3e48550d3dd6e : SUCCESS in 17m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3c60fed064e144de85d2f0946a06e516 : SUCCESS in 25m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa1c203d06e64e92a24eecc3ab8a45b2 : SUCCESS in 17m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/395c42aefe0c42e9856d311e13fafc94 : SUCCESS in 23m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aeba0c34b0c04c64bc456ce8f8dd9fac : SUCCESS in 22m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/85947bca3a15481893b6c29a114aa03d : SUCCESS in 59m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2a826650f93d40b9b54892397034e36d : SUCCESS in 1h 01m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/69daa05ae821433ab2a9af586a415870 : SUCCESS in 1h 06m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce0f38cc3efb4fb6b9facb8560ce923e : SUCCESS in 1h 02m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8a9cc615f77b4eb6b7bab75efc12bc98 : SUCCESS in 1h 00m 54s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e9ae6fa1a9f841258cf7b6a81869e28f : SUCCESS in 1h 36m 01s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9b675d1fbea74d36806c1e63541e6661 : SUCCESS in 1h 33m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7061f3a9164e446ea07bdfd72c0b0926 : SUCCESS in 1h 31m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2958511167949e8900fce0b762875f0 : SUCCESS in 1h 38m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b167509495134ea2a852e1f52c5a9430 : SUCCESS in 16m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5b76e99ecc464c4c9a231d1a920c5ef3 : SUCCESS in 18m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ae1c231afcd4d5c986a8fcd820d55af : SUCCESS in 20m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ff4a776c09b649dcb610c16beb7dae0f : SUCCESS in 12m 49s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5356c7e699484cd18b43d59c2998655d : SUCCESS in 16m 54s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/93a1364274724b2ca9e0ffff301d12dd : SUCCESS in 36m 15s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9dce143bc4f64589ac74dbe5386fe70c : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b023ef015b548c0a09ba420f521ff5a : SUCCESS in 17m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/96b9641d53104852aef1236de508b516 : SUCCESS in 14m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/422150d3f4914b2da8cfd001070cfe3e : SUCCESS in 14m 53s","accounts_in_message":[],"_revision_number":2},{"id":"2a85853b1d6cdba8421882098797d0a739c1222b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 11:06:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2c7c8c20b7cf7e9c03893102302f3aa8674a7d81","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 11:23:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"169d5ae8da2368e5898bef138ab46a421084b27c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 11:25:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1a9ce19c94a97a9cf84ed1334bc7225431ea5025","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 11:25:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ff8e2c1c089908e4fe4b31e5707704fa7190001a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 11:25:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"72c12b1d702da8b4fb51cb130371d0581528e0a8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-16 11:31:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"072deafda4e953ef588f2a2cdd8dc653738152e7","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-16 12:03:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"383e0e3ecef7dbadfc62b826d6d6f23f1bd8acf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 13:11:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43fad91fee4345cba5dba54edfab0800\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd3206fd4674edf96b48288e7194612 : SUCCESS in 7m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2a37532358b94e92a1744a49228e0de7 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32682840d0124d498599f37b02ce77bf : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aaeee4eb8ce4ef9ab77b32bd4673231 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5be8820bb1e6422c91c8e68f618f61bf : SUCCESS in 2m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/057d44395cf84bc99de3a2e2a411cd27 : SUCCESS in 7m 17s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/04ebb5d5ed4649d0a5e22df933f46038 : SUCCESS in 2m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a14f8159d1e94eff95029eecb2da7d22 : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7bcad491a3964a3b8548449dc0fb0a49 : SUCCESS in 16m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/26f9ce4d5f544934b48f4bbd6e50baab : SUCCESS in 21m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68351bb118414e0e819421000325bbda : SUCCESS in 18m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9a9bd30ca1a8478bbbcea47bf8328461 : SUCCESS in 20m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/570b3fe2cff2469f8b4e13cf7d20724c : SUCCESS in 17m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1468933472394e968772cc87bf302c7f : SUCCESS in 59m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/52aa64a6e6974154aea45d6084bd8acc : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/97cf396ba25645fbabd872b655e28c49 : SUCCESS in 1h 04m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/242a77153f294322a0c7bfaaed5d6da9 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/361b480eff6149c7878f23219771a4a8 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/348ab30ba1d64dcc95d9b932a93b244c : SUCCESS in 1h 40m 33s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/30d715c83fd5450ca5785af19554db05 : SUCCESS in 1h 35m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0199fc38432443939824cdfce73b5783 : SUCCESS in 1h 31m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b52a9985c9e5462090ac68890013eb7c : SUCCESS in 1h 45m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ac6c541d14fd401189b33523d19c5d91 : SUCCESS in 20m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9fcfade095354a61a174ccc4a9b84cb7 : SUCCESS in 22m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ca835755db546ab9abaa183cdb14ffd : SUCCESS in 25m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dfb70d1bd19042e8bff5fb7dbcba3a9e : SUCCESS in 19m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f204bd4d17be4d0289e912c7bac322db : SUCCESS in 36m 35s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ecbfcf2d36a841729751aecca214edf6 : SUCCESS in 29m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d59519d6cad423dbee3f365fd3fea3a : SUCCESS in 31m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d429380804e48baabb54e4fe253ae61 : SUCCESS in 29m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a9a98e0577c9492d97703f3ff8dd4afd : SUCCESS in 22m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85b9f3ee9be8493ab5ca043dd5d2e58e : SUCCESS in 22m 44s","accounts_in_message":[],"_revision_number":3},{"id":"d1b9dbcf1326af659c99104c625e441c8e422ab5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 13:12:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8f4d87651647f0bc2ae851badc2001b039312266","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 14:54:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f3fa391af3146f883043ec69e465f0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c876ee4aa24776ab7a17e93d4a7c11 : SUCCESS in 9m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6cecad18f9e24323aa01fb8505363d3c : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c895ceb0ee874dc2949612f99c364a1e : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0514bd81c8554eb58d42cdd6d7c50bd2 : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2aac29be9c754de9ad587f65a944498b : SUCCESS in 3m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/12e70cc2c18f44ed8915667c6963cd55 : SUCCESS in 11m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36b9511a80844623993abe1799fe61da : SUCCESS in 26m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ea066076d4ee4d65b3d113ccd974cc61 : SUCCESS in 13m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ae102cf619e432b9a655777d7226668 : SUCCESS in 18m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/34432dd792cd4a2ab4997c72a9609c4a : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0e8cd2de6a74a868e8ff027921c2558 : SUCCESS in 17m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/405d1947c53345aaa17b7024f366f410 : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7a93b0e7d75646e0b72cbc8f1611308f : SUCCESS in 1h 09m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/095afd454cf44e668f054565225dd6b7 : SUCCESS in 1h 10m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3f3db02f7ea419b9852579669d92d7f : SUCCESS in 1h 07m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e49456cc832a466f994dc989fb3b5b04 : SUCCESS in 1h 09m 20s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/30c7c55b8be843719864df39ac2543b0 : SUCCESS in 1h 38m 34s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0633fe1d3a634412b982911621eca0ed : SUCCESS in 1h 35m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e0373bcba2a4fbf9d02b0fe14a377a9 : SUCCESS in 1h 31m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1576d6d954e4f16b77024f580a5739e : SUCCESS in 1h 41m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4181ecd1fa3b496ebe3d9f78a69399a7 : SUCCESS in 16m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1a1a30f52d7f4233b133d53dddef2e17 : SUCCESS in 18m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4307c6d93ae47c9962b20227c156b43 : SUCCESS in 27m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5e0dce8a8a854e6ca8061b6b266e5ca9 : SUCCESS in 22m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5d2c9f3f2c30400d865d434cb27c5998 : SUCCESS in 16m 15s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cfc39ae118ab449fbbba2efe24ced3d6 : SUCCESS in 16m 53s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fcc4cdef4684e18af07841265be87bf : SUCCESS in 16m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22bb7d2145c74b9f85b392e39fc406df : SUCCESS in 28m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/41a4d0041e9d4afe8198470e26aeee9d : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff3718f77f6e46dea6dcc085972aea74 : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":3},{"id":"042f7877654131a40b8cfebe5e3bf5c20f80b3d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 14:54:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"86d29833b30d6edbdc504f6d1187c39d8e714177","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 14:57:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b25a083408db498ba6d4ebb671c621a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aaa7cb797614d70aea94097b00feece : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ba91754fedbf9ed0b436c2d6721b4c378098648b","revisions":{"17802711ec6b502a258c46ed3e8134b405322d15":{"kind":"REWORK","_number":1,"created":"2026-04-15 13:01:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/41/984741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/984741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/984741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/984741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/984741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/984741/1"}}},"commit":{"parents":[{"commit":"6ea0d331b2e3f59f3632635ce2de20632d6241fd","subject":"Merge \"Make Ironic Inspector config comments consistent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ea0d331b2e3f59f3632635ce2de20632d6241fd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-15 13:00:31.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-15 13:01:44.000000000","tz":60},"subject":"Update kolla inventory templates for kolla-toolbox rename","message":"Update kolla inventory templates for kolla-toolbox rename\n\nSee: https://review.opendev.org/c/openstack/kolla-ansible/+/984303\nChange-Id: I50dd57a8535d45d74d0cc9236f7a49792b962daa\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17802711ec6b502a258c46ed3e8134b405322d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17802711ec6b502a258c46ed3e8134b405322d15"}]},"branch":"refs/heads/master"},"9c9a8f97e09a73e722e43304ae0974052099ea38":{"kind":"REWORK","_number":2,"created":"2026-04-16 09:25:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/41/984741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/984741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/984741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/984741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/984741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/984741/2"}}},"commit":{"parents":[{"commit":"6ea0d331b2e3f59f3632635ce2de20632d6241fd","subject":"Merge \"Make Ironic Inspector config comments consistent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ea0d331b2e3f59f3632635ce2de20632d6241fd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-15 13:00:31.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-16 09:25:24.000000000","tz":60},"subject":"[CI]: Fix various issues","message":"[CI]: Fix various issues\n\n- Disables the py311 job since we require 3.12[1]. This was\n  brought in due to the changes in the jobs templates for\n  2026.2[2].\n\n- Updates kolla inventory templates for kolla-toolbox rename.\n  Matches the change in kolla-ansible[3].\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/967297\n[2] https://github.com/openstack/openstack-zuul-jobs/commit/06df93a5d9ed3cb815b4c7ffd63dab006fee4e8a\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/984303\n\nChange-Id: I50dd57a8535d45d74d0cc9236f7a49792b962daa\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c9a8f97e09a73e722e43304ae0974052099ea38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c9a8f97e09a73e722e43304ae0974052099ea38"}]},"branch":"refs/heads/master"},"ba91754fedbf9ed0b436c2d6721b4c378098648b":{"kind":"REWORK","_number":3,"created":"2026-04-16 11:25:04.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/41/984741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/984741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/984741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/984741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/984741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/984741/3"}}},"commit":{"parents":[{"commit":"6ea0d331b2e3f59f3632635ce2de20632d6241fd","subject":"Merge \"Make Ironic Inspector config comments consistent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ea0d331b2e3f59f3632635ce2de20632d6241fd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-15 13:00:31.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-16 11:24:55.000000000","tz":60},"subject":"[CI] Fix various issues","message":"[CI] Fix various issues\n\n- Disables the py311 job since we require 3.12[1]. This was\n  brought in due to the changes in the jobs templates for\n  2026.2[2].\n\n- Updates kolla inventory templates for kolla-toolbox rename.\n  Matches the change in kolla-ansible[3].\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/967297\n[2] https://github.com/openstack/openstack-zuul-jobs/commit/06df93a5d9ed3cb815b4c7ffd63dab006fee4e8a\n[3] https://review.opendev.org/c/openstack/kolla-ansible/+/984303\n\nChange-Id: I50dd57a8535d45d74d0cc9236f7a49792b962daa\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba91754fedbf9ed0b436c2d6721b4c378098648b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba91754fedbf9ed0b436c2d6721b4c378098648b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
