)]}'
{"id":"openstack%2Fkayobe~984820","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I87184f8a73798a8fbf90e71865067dfdcd76893d","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"38873":{"account":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"last_update":"2026-04-16 14:34:01.000000000","reason":"\u003cGERRIT_ACCOUNT_38873\u003e replied on the change","reason_account":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"}}},"hashtags":[],"change_id":"I87184f8a73798a8fbf90e71865067dfdcd76893d","subject":"Build Rocky based IPA images","status":"ABANDONED","created":"2026-04-15 16:09:27.000000000","updated":"2026-04-16 14:54:36.000000000","total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"9745ad22c0ef9187d0682e4ff78b9e3cfd14d0fe","_number":984820,"virtual_id_number":984820,"owner":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-15 18:35:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-15 16:30:55.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-04-15 17:48:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-15 18:35:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"950f1c968aeeb07cf75e04d8acf857b0296b3420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-15 16:09:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"878a3be1763b65622ca28c398a845801ed514358","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-15 16:20:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb8c1c3c7cdb1111e1f3c53bfe4bafdeb919a3b9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-15 16:30:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e018fb98a0f4dcffb6f7dcb719b599fc798616a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-15 16:45:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fcddae1c65e78fab4d9418e743fb9fa101e7d738","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 17:48:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a03080307964ea4b1958e2017dbd338\n\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/84b7d15c808e4aa1a502a560cd7ee4ef : SUCCESS in 8m 37s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/df0242154ce146d3840140b7559c1f17 : SUCCESS in 16m 22s (non-voting)\n- kayobe-infra-vm-centos9s https://zuul.opendev.org/t/openstack/build/39743c1abc1a46bcb321d98accde4905 : SUCCESS in 16m 31s (non-voting)\n- kayobe-infra-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/fd374af4dbb3416ebb2b28512a04b064 : SUCCESS in 24m 04s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/e631114bb154431ca36a5ab3afc25856 : SUCCESS in 11m 06s\n- kayobe-infra-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/99be4b2ddfa84cea89e3e559af381b78 : SUCCESS in 22m 27s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/80bcd1f190714efd8d90163d73b41f6d : SUCCESS in 8m 54s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/5e63c7dd5a13484f8d844d8e5b77f557 : SUCCESS in 32m 47s (non-voting)\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9c7d4b1390e549dd96f70f99d4dae7f9 : SUCCESS in 59m 39s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/924815ad4cce40c39cb94cc31002f5f7 : SUCCESS in 10m 59s (non-voting)\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/abafc72e5e024f928dd27968f8833d21 : SUCCESS in 19m 14s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/13176aab202e45d59affc14e601cf342 : SUCCESS in 1h 21m 10s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/f423d1849b7140de9e1638367f340291 : SUCCESS in 13m 37s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/970ff931eb51408d8d2d2d5a773503d4 : SUCCESS in 44m 42s (non-voting)\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/bf01d028ee894604bb9893a4b7edf03f : SUCCESS in 1h 00m 56s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/e40cb2555c134a978ce8e40d530c594c : SUCCESS in 55m 44s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/d52687d8d3424fa29d7f777c578af801 : SUCCESS in 43m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/72dfcf5859f3479c9242c6f0632c3a9c : SUCCESS in 54m 12s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/cd7b01fcb51a457093679ca6da2b4bd6 : SUCCESS in 1h 21m 06s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/937a1e5019964cae93ccf0782ce61b94 : SUCCESS in 42m 11s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/669c7bc45da446eb9b6bdfcc1d952be3 : SUCCESS in 50m 37s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/4712222e857c46b4a6c383071b04f2d5 : SUCCESS in 7m 58s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/67ad26df9a0b430a8e3f5e6ef2a08061 : SUCCESS in 20m 10s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/90409f77d1934c1f923f9d41334479fa : SUCCESS in 18m 14s (non-voting)\n- kayobe-seed-vm-centos9s https://zuul.opendev.org/t/openstack/build/017170abad664e5ba85e4b2a3e9b90e3 : SUCCESS in 17m 27s (non-voting)\n- kayobe-seed-vm-centos9s-cloud-image https://zuul.opendev.org/t/openstack/build/ff6ce232c21247fe9f82adb53a73435f : SUCCESS in 25m 13s (non-voting)\n- kayobe-seed-vm-centos9s-q35 https://zuul.opendev.org/t/openstack/build/76f2724899564e99935a6a8d1712643a : SUCCESS in 16m 19s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/1469fccc530943e3b764bee65dd0ced3 : SUCCESS in 11m 15s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/5943a2bb49294f6bb8e188857dd97f52 : SUCCESS in 8m 20s\n- kayobe-seed-vm-rocky9-cloud-image https://zuul.opendev.org/t/openstack/build/9984f0f630bd4381af08213afefab9de : SUCCESS in 20m 19s\n- kayobe-seed-vm-rocky9-q35 https://zuul.opendev.org/t/openstack/build/8822015a08ac42f1965b7ac6ff6b8436 : SUCCESS in 15m 37s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/ebc26753b01446039c044478227b6b0d : SUCCESS in 22m 00s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/b38f7691888f4b7488b376ff7041f4ea : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":1},{"id":"425cf0e161f8920aa685353c0863fc97ef3f70b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 18:35:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a8d5f468e674385b3f33560174c503b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a7a8589a08f4864828e3f51b7e0c555 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab2c374120a4b9e96d0c5265ebac008 : SUCCESS in 5m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c620d50c30fe4e98be17359ebf637a76 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b680afff64d34d0c8881aecc6958e713 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/084f417520c14fda8efa0d75aff6adc2 : SUCCESS in 2m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f48efd5446243d7b1ff4a8d756242c8 : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/26d2d73ff4ca4e6d91fdd45e0330df69 : SUCCESS in 9m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4fe68dda080a4b15b53f0a55f11f51eb : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7c591570bac2453ba9ddf5068a3de5bd : SUCCESS in 9m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c120652b239d4d8f9ba02eaebb3dc3d4 : SUCCESS in 14m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e674628e0be74fc88d2a06e46a3da368 : SUCCESS in 10m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/39c1155a5cf4451b9d0bd5cb45c47bb0 : SUCCESS in 17m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3b8387dd843b46cd9b3f46b331ba8011 : SUCCESS in 17m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a8bdb6ade4f4930b2749dbccaa6e3b9 : SUCCESS in 13m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/443486a875074aa08a478337143d7993 : SUCCESS in 49m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0992dff35412428c950a460f4f2f2c96 : SUCCESS in 33m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/94156295839e4ebb8968209100962ad3 : SUCCESS in 48m 37s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7e17e599107f4a65bea4fb7da592ae7b : FAILURE in 22m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/518e4229f4044c9da4e579977e2bbb1a : SUCCESS in 54m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c4252b11337a4a3f83be902c9afa3e3c : FAILURE in 23m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/111dd69bf973456c940ea1ac900e12fe : SUCCESS in 30m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/439927d798c64ca2afec97714e82ae86 : FAILURE in 17m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/51ddfdb7629d4f39b674d4a3c568e243 : SUCCESS in 2h 20m 09s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c183bb863a2e41c7ae34a4d2b5e601a1 : FAILURE in 28m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/647d7af5e4c04e4cabbc6bce508d7939 : FAILURE in 29m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da84462faaef4f3b9f8b7df175979eb1 : FAILURE in 24m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9ac446bfc80541d9b3812b73b17e3fef : SUCCESS in 21m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cef770584bfc48aebbc2309d3e024e37 : SUCCESS in 9m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1964792ca9114cefa8d84d98abe2b7e2 : SUCCESS in 21m 10s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/4413fb9cb553418db9cde2ca5af570a6 : SUCCESS in 18m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b6ee5a73893440cacedf661479f0dd0 : SUCCESS in 19m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c24373703e14b11a8804c060ebafaf0 : SUCCESS in 18m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/619faa2f320145cdac604b2f497f47f7 : SUCCESS in 37m 13s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2987effd6c2f4cd29346d0a0186d3b29 : SUCCESS in 21m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a57e1ffa4be943578ca67699366ec8e7 : SUCCESS in 28m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/98a46034eff140adb42d55caa7b29381 : SUCCESS in 15m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1dd064d5d9124ece8761ce3a7da86d7a : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/486ef5285bb34b31b53558f064244d79 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d2cb76d45a93970b4f2ff268bdf7d2df6600c30e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 08:40:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4fcf0498d366e31805a5accf9cba133714056c0c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 10:57:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08be9f5786a41aa5c7e22c257f933e578e879ab6","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-16 14:34:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9745ad22c0ef9187d0682e4ff78b9e3cfd14d0fe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-16 14:54:36.000000000","message":"Abandoned\n\nWill work on overriding distro/release in master and backport that","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a109a0741d7baead729271e0cb8d826be769654","revisions":{"0a109a0741d7baead729271e0cb8d826be769654":{"kind":"REWORK","_number":1,"created":"2026-04-15 16:09:27.000000000","uploader":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"ref":"refs/changes/20/984820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/984820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/984820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/984820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/984820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/984820/1"}}},"commit":{"parents":[{"commit":"5df157b38abc992f4ece634d9473c8dcc024330f","subject":"[networkd] Fix broken conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5df157b38abc992f4ece634d9473c8dcc024330f"}]}],"author":{"name":"Owen Jones","email":"owen@stackhpc.com","date":"2026-03-25 13:09:21.000000000","tz":0},"committer":{"name":"Owen Jones","email":"owen@stackhpc.com","date":"2026-04-15 15:48:59.000000000","tz":60},"subject":"Build Rocky based IPA images","message":"Build Rocky based IPA images\n\nWhen `os_distribution` is set to `rocky` and `ipa_build_rocky` is true,\nIPA images will now be built using Rocky, rather than CentOS Stream.\n\nChange-Id: I87184f8a73798a8fbf90e71865067dfdcd76893d\nSigned-off-by: Owen Jones \u003cowen@stackhpc.com\u003e\n(partial cherry-pick from I9fbfaeb1b88e747de1a7c16934d3c46f7cc82d3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a109a0741d7baead729271e0cb8d826be769654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a109a0741d7baead729271e0cb8d826be769654"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
