)]}'
{"id":"openstack%2Fkayobe~984955","triplet_id":"openstack%2Fkayobe~master~I21aad066282582725c8517baf3f240c674046d16","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-17 08:22:48.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"38873":{"account":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"last_update":"2026-04-17 09:45:11.000000000","reason":"\u003cGERRIT_ACCOUNT_38873\u003e replied on the change","reason_account":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"}}},"hashtags":[],"change_id":"I21aad066282582725c8517baf3f240c674046d16","subject":"Override IPA image distribution/release","status":"MERGED","created":"2026-04-16 14:57:11.000000000","updated":"2026-04-17 13:06:30.000000000","submitted":"2026-04-17 13:05:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"984955","meta_rev_id":"be833a0f90e839955fcaec1ba3a20edb540315a4","_number":984955,"virtual_id_number":984955,"owner":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-17 13:05:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-17 08:08:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-17 08:22:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-17 08:08:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-16 15:15:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-16 16:53:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-17 08:08:16.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"ecb75f48dcc1e0eb02d5e5c58e835ba25b03b72b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-16 14:57:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30212d92a1f9a85afd4ea6f80036b97f3c2747fc","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-16 14:58:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"841b7da727656ee2f31b674b60255a87f2c9415f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-16 15:09:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7a22cf267631c25f02712a4fef0378f1ad6643fe","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-16 15:27:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0fecb795842564f670f6b94b2ab989a3ea24fc51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:53:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b43d5e4409415c9932b0e1a6147d30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7337ac28e8ce4a15b0d3b1797d349ab0 : SUCCESS in 6m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6fa437313e94e8aa506961274eec915 : SUCCESS in 6m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c095603cb7248179b95f70646d1e90a : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e26725751548749f46d1d2a4340c82 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e67b248c82aa4e88a12f5cf7fa770dff : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/487957495da041758c273c310ac8c915 : SUCCESS in 7m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/320b592f926742be8bdeaedca70106ed : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e3307ae8a2c410db78286d24be1ce6e : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5e78d08295ce40f8953835d905db067a : SUCCESS in 7m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4d2b964fafe460f995bda7b96e0ffca : SUCCESS in 22m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fa777abaacff43fcbbfc69b06e0f8a26 : SUCCESS in 19m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/240aabeec66945648016488b4019af2a : SUCCESS in 17m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/43b86e59138a42e5b9e2969fb38058f2 : SUCCESS in 13m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f04c5269afd3425c862cc81c773dfd78 : SUCCESS in 17m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d469d3e18a3f48bf88a36420af10a277 : SUCCESS in 1h 07m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/76ffe103a82c4ceabc0301096b2b75d7 : SUCCESS in 1h 16m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/add11ff3173a4844ae82734a1d6eae78 : SUCCESS in 1h 10m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcd3b983308749f6a9850ad679c2611d : SUCCESS in 1h 09m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5b3b5a924bdf436a8b61a0635b351209 : SUCCESS in 1h 10m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6ceff9852be44b32b41540f10c444edc : SUCCESS in 1h 39m 32s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f712e5cc339d4e39b7b380790fa6a520 : SUCCESS in 1h 37m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f36ff0fb59c54be29410ea35f59465ba : SUCCESS in 1h 37m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7ddd746d4544cc2b84e1952a812b182 : SUCCESS in 1h 40m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9851622ff2634c5886c6d80681a1503a : SUCCESS in 15m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/152d44098bef42728b15d13902ef26cf : SUCCESS in 16m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf8e04e11f184050bd2536b6dee4fe28 : SUCCESS in 22m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9dcf46b413b4444bac635c2ee476d688 : SUCCESS in 15m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/213fe28121464f9894e801fab47bf2ff : SUCCESS in 28m 33s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8a84d45d20da440bb54cf132f4f67cc4 : SUCCESS in 32m 34s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45ec644091a24524be859d311627160e : SUCCESS in 21m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ae10b73881b45ec94a47a779a06fcc4 : SUCCESS in 29m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0c02028079394fc0b31176dfb49db113 : SUCCESS in 23m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc18c0efb78f44329ebd5d4e97a6cb3f : SUCCESS in 23m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ba492cb0b510021cd5b4f60557ee2c2a416ff604","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:05:43.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09af5ba3102640218dcba31152370f51\n\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/9e645b99aa434b3eba3733e8267b0f5a : SUCCESS in 14m 47s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/4a9de93946894174be39fff973268550 : SUCCESS in 18m 48s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/2026568db7184400912d6629b6c6fea6 : SUCCESS in 22m 35s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/17c9c3cb195e49c49b0eec14b0935d2d : SUCCESS in 23m 47s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/8e802c0957e04d6b9b8f387285790918 : SUCCESS in 58m 26s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/e36809815deb4ea5bcd587e3d819605d : RETRY_LIMIT in 2m 59s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/fab33253b0fc4bfa81f6748b0d74acba : SUCCESS in 57m 12s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/5e47613334d24f06852e6f2dd567efbc : SUCCESS in 15m 02s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/1ab1f31adc6a419988df173f87465909 : SUCCESS in 37m 46s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/a353b7363ec0431aab44c387e88a5684 : SUCCESS in 39m 04s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/ab284b51ea9c4c679e87d7f6d0cb7028 : SUCCESS in 47m 08s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84a333bbe461413ba738251327cd9767 : TIMED_OUT in 1h 33m 31s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/87789b9869c346c4928820c11f3f4d66 : TIMED_OUT in 1h 33m 23s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/782a1ea8b927473c9c0ca09206eef769 : SUCCESS in 16m 38s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/d12b570c238644c980831753e42e6b98 : SUCCESS in 29m 28s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/32e21e2a8cce46cda8b1eec889e1d0ab : SUCCESS in 20m 26s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/b330ccda869f4a9e8be80aa099274a44 : SUCCESS in 21m 40s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/b45c9f778c7b4bc2a92f636dd982670a : SUCCESS in 16m 02s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/7bc4bbb5e69d4b8fab8737896b9efcff : SUCCESS in 25m 31s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/41c54a60d68f4b72aae0e615cf2421a8 : SUCCESS in 15m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1da961a7732b0f2dba8c84325670f6a0e6a9e8c6","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-16 17:22:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c876c155b3e6cd3b32f82ea7bfdbc3a1a13b44f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 18:58:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e4235cd803a4d4790e7ca2cabd77bf1\n\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/4d14454654b0411cabcc37f0ccded3d3 : SUCCESS in 12m 55s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/9167664420c44a60bebe950d6d0e764f : SUCCESS in 24m 25s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/581cd369fe304fd2bdc54dac2d115a22 : SUCCESS in 23m 44s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/05b9c286383f4ea295d9e2e0a17213e4 : SUCCESS in 25m 33s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/7aaefa42e9014760942afcb32c660c38 : SUCCESS in 52m 28s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/ba6ff6d3730647b78197108eacae2354 : RETRY_LIMIT in 3m 59s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/e6100bb91b8646b9ad07161d9ca47e67 : SUCCESS in 54m 32s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/158dba3ef3984b7bacbf2c07c8c95221 : SUCCESS in 13m 54s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/04e7ecaf6d9a4d26822458c478bef3c3 : SUCCESS in 1h 06m 33s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/2519544b78be4b1bb77dc9662eef81f9 : SUCCESS in 57m 59s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/b0812633d3984971bbafe81cb6528445 : SUCCESS in 1h 24m 24s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d79e5e6d73fb4e9091680099239ff9e5 : SUCCESS in 1h 13m 35s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b0ef36a6903847958e9b414d5a7f5858 : SUCCESS in 1h 31m 08s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/2679d5c741e14dbe899bbdb6c76d2010 : SUCCESS in 13m 47s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/848aea5935da40aaa02ad461f85b947c : SUCCESS in 26m 19s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/3b9cf41639c94b6dbe400d91b96a9f9e : SUCCESS in 13m 52s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/b430bb421a114a9ab0d0ac4d052648c8 : SUCCESS in 15m 50s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/51a182208dc241389b43b235ca429a89 : SUCCESS in 21m 46s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/3bfed7cb51a542c5b8153876c442962a : SUCCESS in 25m 36s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/dc696effe2f74452aa47985c9ec85cbb : SUCCESS in 19m 59s","accounts_in_message":[],"_revision_number":2},{"id":"bf0effffd9a5f98bda34fa47f031b2dd3f0a0061","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-17 08:08:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2a4207b23b1e37f4cf77e18bb37e29d66f1c596","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 08:08:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6c1966261efae5d7d4bb20ae2d094bce99cb0a82","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-17 08:22:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9c00a43838ead14b5d52852702f9e25f8d161897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 09:44:32.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/602fb06a283e4eff866454875bdecb25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4704d3bc5e374308af78d25521744145 : SUCCESS in 9m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6e1b18ac5f2c4de4bb5394d7a7879694 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/921bec8265ba4a2681bcb0bae4eb7e8d : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca87450e808a4ef9aaaff3e4a89284d8 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ccadd7897b34ab9903b688eba019365 : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ccd42f3517a462ea0cbc037dbcf3d49 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/564a4645435541efa38719a8d01f9fc6 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70b1818cfeeb45ed8d4f2b49bb3e48d4 : SUCCESS in 17m 19s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8ae2da7ee2bc4cd2a38f35ba1df98541 : SUCCESS in 25m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bb42c7fbc184f0f9af99be3105ebf29 : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d8955f32b69e40139ca2695c0a2ee2bd : SUCCESS in 11m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3df935afe9af4efbb3e717e0d580c30b : SUCCESS in 21m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9a874318959f4265ac8d21c5a5657fa2 : SUCCESS in 59m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c9a9f8df27334d91bb084137920336b2 : SUCCESS in 1h 03m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e59091f18e164bd6a9cfede189029f9e : SUCCESS in 1h 04m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b10851a5db9744548342c7fb7fa8aad4 : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6769c22073434fe89b0b54c117cc2e19 : SUCCESS in 1h 02m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7fd97aaed9b947138a0feb7f143ea3b2 : SUCCESS in 1h 35m 23s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/20c744bc58174329bc1b11103f14d508 : SUCCESS in 1h 30m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be6a351a182b4194a04f87dc44eb44d0 : FAILURE in 5m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c78e78bc84894c7da7758deb3e5ae483 : SUCCESS in 1h 33m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/550f755694b949d98f3d4244cb0703d2 : SUCCESS in 28m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/450f348f5ea14d44adc20cf605d68951 : SUCCESS in 19m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/feef602c9bc344d687cf4b410f8c9baf : SUCCESS in 23m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0c647589d441447595e33cce19a1357a : SUCCESS in 21m 05s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/98d8068a0ed74617b499a180ebe5a5e0 : SUCCESS in 16m 37s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0051a56248214138be23c57cdccdb2ee : SUCCESS in 29m 31s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b60d0f975f54f56a18975057d893400 : SUCCESS in 29m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18b97dbb46364b8d9ea40b3adfd904de : SUCCESS in 27m 29s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/723dde90c2964ca4a540151006cba543 : SUCCESS in 18m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/854b9ea1ca3f44f0bcdf22822d64d8d0 : SUCCESS in 21m 46s","accounts_in_message":[],"_revision_number":2},{"id":"ece422c982a84f0bd6d0d00728473129279d94ce","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-17 09:45:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8c4472be311510332e6c6cdfd49ac1bdadc45cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 11:22:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91176e1dbeb64c1f84113ff56b7d739a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3006647d9b04fdb92eea38c89900569 : SUCCESS in 9m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5e58c918a7e14415b4bc8e6adadcbcaa : SUCCESS in 7m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/398715d79aa6422ca256ce0746395490 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed87032637d043c1b905d6e2ab82a4ba : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b433435e9da435bb3bd50596dfeacae : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e66141dc7a6d4a2b8cdfddfe607ce157 : SUCCESS in 3m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa8c330120ae4c1d9e4ef6170ffd520f : SUCCESS in 2m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6e85dc07a6ac42cc975d2bfa9682bd44 : SUCCESS in 3m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/882f05116cd546db9bf064b997a2343b : SUCCESS in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a8b12a3d27814498b21e9d0e6ce892a3 : SUCCESS in 5m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b57cca2d04c64eb48c3e15ddb445a70a : SUCCESS in 13m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cf65c61eb584560b51c485f978593fe : SUCCESS in 18m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/afb06bb278924341a09cb0f4130ae56a : SUCCESS in 10m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5134e326d55d420d857bb3ae4b9a4562 : SUCCESS in 17m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0d904e34940d4a6abf1a124a2c37cee3 : SUCCESS in 59m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ad18a15365c140c0b8993d51a8269267 : SUCCESS in 1h 04m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/42d6565974e04164945a0abe4d0781b1 : SUCCESS in 1h 05m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66127c62f7b4489cb92e912d1d81866b : SUCCESS in 59m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6631a167d5474e07bc4fc7d3661059d6 : SUCCESS in 1h 03m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4c2b46fcf93242aeafb1fefb24d3a229 : SUCCESS in 1h 35m 56s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/646d55d65ef84337a3ace690ffaa08c6 : SUCCESS in 1h 28m 56s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4886084f10754ae381f2653935f51c09 : SUCCESS in 1h 29m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3993c4a1e9a04122b89c7e2f94ac55ec : SUCCESS in 1h 33m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5380ad39730c4d73ab736ff5f3659b03 : SUCCESS in 14m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6e602a7a9e0a4bb19f659caa62adccf6 : SUCCESS in 18m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/929d1635cc96448e8a0864ab84a96e77 : SUCCESS in 12m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/72127523fdb0480bbb3b59826b97c436 : SUCCESS in 19m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d629b515369e4f658e7a984b8894919b : SUCCESS in 17m 54s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b4bb723c04944f0f8cee4973af758b9c : SUCCESS in 24m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b19468141dfe46c99f817168a7e2ce8e : SUCCESS in 17m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f760ea15c261477b883c0e87b5bc27ec : SUCCESS in 29m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a594e3811a384945bbf55862cccde60d : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4da036df82f44d7eb022645a43cc582f : SUCCESS in 12m 14s","accounts_in_message":[],"_revision_number":2},{"id":"7ccadb6fe8fbf747ee824f265262edbe03d67a09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 11:22:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d156ea6a238720443396818fe9361bd8d37b039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 13:05:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91aef3c751aa4921b596d719812118d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da593d118ae4d4e8f02b68cadf44ca9 : SUCCESS in 7m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5e3ab50c7ef942e992a4186f47a20ee3 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a06448866d425589622c20596d4235 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91467d62d4944fecad326a8c5c9ee4d5 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a189bd8e93a404981f10d798947c1d2 : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/284efc6b98f946c499bc0aab6364d1cf : SUCCESS in 3m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68895ab3b28a40c78866411ca2b29d2f : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6157e48419444ba28f812c3e1ef3ac38 : SUCCESS in 8m 55s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5752028b570c4510a72025dd10e10e06 : SUCCESS in 13m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e68e4c7a84e041bf9e9bbf9158910f2c : SUCCESS in 19m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d445ef5d406c4bae8129909f61e1278f : SUCCESS in 25m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a327b1cd86f74692895008dbfe53fc51 : SUCCESS in 16m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/73b0a156cfb248deab3922d59e79ad9c : SUCCESS in 1h 09m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7e4a5fd24fba45da9ce8e036243ba29a : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f44bb7811c074f0f97e13d49b0381762 : SUCCESS in 1h 13m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5840c169eba4e3d873a981fa2f669ed : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/be985cfb862f4292ac975c1ad81948aa : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6c6d91bd6b684dc19295d211616d2fa2 : SUCCESS in 1h 39m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/59938495a05a4682b5183eaf95dc22d8 : SUCCESS in 1h 29m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0f12f2b178745b89783516ec022ff3d : SUCCESS in 1h 35m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05ae582b02de4d2ab9530f5088eb72bc : SUCCESS in 1h 42m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d75d5db49a3e4517b5be1f231e5e24e3 : SUCCESS in 30m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a320c0bc017a41e082952c837656d3f1 : SUCCESS in 29m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41ef83c645b14d3aa385816d4e615c18 : SUCCESS in 14m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0d22cc2cbc754e57b13cf3f2e623e645 : SUCCESS in 22m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/64f3ed947f6a4c759f5452f09554c538 : SUCCESS in 18m 42s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5d6edab7440f45ffa3230ac576579aa7 : SUCCESS in 37m 28s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75bed065f250484c99cf25612ece8f37 : SUCCESS in 28m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7c5965a5d06444c93829a0eee514ac6 : SUCCESS in 18m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0eacf6864c2e4bccb12e12be8db525da : SUCCESS in 13m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71e9f2787771453d80edb2262879090a : SUCCESS in 22m 43s","accounts_in_message":[],"_revision_number":2},{"id":"0703e8f66b33e760f072573e5f2cfefcaf865db0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 13:05:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"be833a0f90e839955fcaec1ba3a20edb540315a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-17 13:06:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa80ce83b414552a4307d2706155f32\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/827ca465ed064c90b8d7ffdd93dc34a0 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7942613819c44e50bde4bdc960579c50 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af4fad31f5c35a41d6c28ac5200ad29a20655234","revisions":{"323d5c8bc3ea3d6d5ab53fd36eef2c36a4c7f375":{"kind":"REWORK","_number":1,"created":"2026-04-16 14:57:11.000000000","uploader":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"ref":"refs/changes/55/984955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/984955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/984955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/984955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/984955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/984955/1"}}},"commit":{"parents":[{"commit":"6ea0d331b2e3f59f3632635ce2de20632d6241fd","subject":"Merge \"Make Ironic Inspector config comments consistent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ea0d331b2e3f59f3632635ce2de20632d6241fd"}]}],"author":{"name":"Owen Jones","email":"owen@stackhpc.com","date":"2026-04-16 14:47:07.000000000","tz":60},"committer":{"name":"Owen Jones","email":"owen@stackhpc.com","date":"2026-04-16 14:52:48.000000000","tz":60},"subject":"Override IPA image distribution/release","message":"Override IPA image distribution/release\n\nEnables overriding the OS distribution and release used to build IPA\nimages using the `ipa_build_distro` and `ipa_build_release` options.\n\nChange-Id: I21aad066282582725c8517baf3f240c674046d16\nSigned-off-by: Owen Jones \u003cowen@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323d5c8bc3ea3d6d5ab53fd36eef2c36a4c7f375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323d5c8bc3ea3d6d5ab53fd36eef2c36a4c7f375"}]},"branch":"refs/heads/master"},"af4fad31f5c35a41d6c28ac5200ad29a20655234":{"kind":"REWORK","_number":2,"created":"2026-04-16 15:09:39.000000000","uploader":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"ref":"refs/changes/55/984955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/984955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/984955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/984955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/984955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/984955/2"}}},"commit":{"parents":[{"commit":"6ea0d331b2e3f59f3632635ce2de20632d6241fd","subject":"Merge \"Make Ironic Inspector config comments consistent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ea0d331b2e3f59f3632635ce2de20632d6241fd"}]}],"author":{"name":"Owen Jones","email":"owen@stackhpc.com","date":"2026-04-16 14:47:07.000000000","tz":60},"committer":{"name":"Owen Jones","email":"owen@stackhpc.com","date":"2026-04-16 15:09:32.000000000","tz":60},"subject":"Override IPA image distribution/release","message":"Override IPA image distribution/release\n\nEnables overriding the OS distribution and release used to build IPA\nimages using the `ipa_build_distro` and `ipa_build_release` options.\n\nChange-Id: I21aad066282582725c8517baf3f240c674046d16\nSigned-off-by: Owen Jones \u003cowen@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af4fad31f5c35a41d6c28ac5200ad29a20655234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af4fad31f5c35a41d6c28ac5200ad29a20655234"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
