)]}'
{"id":"openstack%2Fkayobe~985339","triplet_id":"openstack%2Fkayobe~master~Ib7c8e5451ca8bdcea8c4b4dcb1093680acce0778","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-20 19:49:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7c8e5451ca8bdcea8c4b4dcb1093680acce0778","subject":"kolla-ansible inventory sync","status":"MERGED","created":"2026-04-20 08:27:45.000000000","updated":"2026-04-20 19:50:48.000000000","submitted":"2026-04-20 19:49:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985339","meta_rev_id":"3362c39bf20f444d6d8608dd41355ddf12b22ef9","_number":985339,"virtual_id_number":985339,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-20 19:49:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-20 14:57:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-20 10:05:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-20 10:05:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-20 09:34:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-20 10:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-20 10:05:53.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-04-20 14:57:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"bee843d75a9bc63b7c7cdc23b243f37b8a3f3506","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-20 08:27:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb2c9c31df0e3d0eab0be38bc72c6b6682adcb2a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-20 09:34:55.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"2f578b5321c0c9d520f7c738064929c4f125b963","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-20 09:37:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f71fbce6232778a88cd5706e59555da0f783384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:02:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39cf0f81a3884761b0d242adcbf7a2b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071169bdf1c04ac2ab8b034982755403 : SUCCESS in 8m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7d0cbfb5c5d14119bbafdc923057ae9d : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/510956ae896e4bf59e6143e233950812 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d36b53b055c43898a21c20cf7ab3620 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d511297f14c54bf6b3b986849950caab : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/537f5132befd4c4ab33b57712df02a2b : SUCCESS in 6m 56s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cb55ebd5319842e2b79d1eaa5fca161a : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/52d3cddc1d6e47abb723088486729a69 : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d399f073e5bf4725b2ad7d5e6a65cff0 : SUCCESS in 11m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4806c24b8e2645089d35be5f3cc83787 : SUCCESS in 14m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0a480499e2949699a9d28ed1dd2aad1 : SUCCESS in 13m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b8e2eb4265c14ba7870c88aeaa01e172 : SUCCESS in 11m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2ce67d6d76547ff98200ca1817ec2df : SUCCESS in 16m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0bf42b14d9bc4fdaa9a3b4e0f97fb986 : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/72cda0bd46ba4b979b3d1a1975d1d021 : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/94f70119af0c47a28ffafb27ef717260 : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/002ab077465946e6973fe1d1c05881f6 : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7e4dc63ae2d4b00b7c51db4f79e2253 : SUCCESS in 1h 05m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0735a1766a11463b85d87081bfc0b504 : SUCCESS in 1h 34m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/987c4436f12b4e6193615f5e313253fd : SUCCESS in 1h 34m 03s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08ddca1c9acc497c897370ce2c2785c1 : SUCCESS in 1h 31m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1624f99c26334398b82eb6586f7de685 : SUCCESS in 1h 32m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/adf97ea874d14b6d94d7f315c21b794d : SUCCESS in 28m 06s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1d56dd7d74a94fe1ae6b7e2e3665306c : SUCCESS in 15m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3548b1f7d0064a3fb088c561fece7573 : SUCCESS in 14m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4e0392a8bfb14a7793b9011eee8f19df : SUCCESS in 20m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/eac6c485a1c345e18421442a107e7dea : SUCCESS in 27m 18s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d60737b499ef41b89e571c9d4791b702 : SUCCESS in 27m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/591b78f075b942fcb84b02e889eaad32 : SUCCESS in 16m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a7c8cdc917d4757bda4ace9b49f53e7 : SUCCESS in 15m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ff015b288f8d43dfa2e6dfa3cbab4031 : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01e3979a1c064cee858583ba41140225 : SUCCESS in 20m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5dcda4f53f4025e9cd235664b77b08b3f4c6f362","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-20 10:05:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"153879abfbed3f943ca398e1fc1da157aea0af11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 10:06:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f69efc590fab40d0d2d41287ec8b3852e0e0f40b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 11:52:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddcafa7b594b4fd59c5f23a50e38a8d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9939f14aca83495c96325a2b31a2185c : SUCCESS in 5m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/155fadc6a7c5470d919f40c643c831bc : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2cf7ae6ac0a47f19a8356dc9d4c846e : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38ecbc4c6988498f9edfed896780063d : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a40c483cb76f44e0b550d2abee0911b6 : SUCCESS in 7m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/811ddcfe26f1402a84d9d31c093fa792 : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac80f541da70451da1747c9b95bfa639 : FAILURE in 6m 31s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2d71aab5679644e1b8de770c83c8f1c3 : SUCCESS in 18m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90817705755e4169974da725f9497f98 : SUCCESS in 19m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/73aa6913db0247d5ab96d1b7930c9ea1 : SUCCESS in 18m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/610105584629441691e12beb5f4d2095 : SUCCESS in 22m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3e0d4c435bf641aabaf57ad8bfb7d95d : SUCCESS in 1h 07m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/80838c1f790a4b2a9e15bd0aa79a994c : SUCCESS in 1h 09m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b0fee1a274344d7d909bab971b3c0053 : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4a2c452c85d4e65a3187171c00c4acf : SUCCESS in 1h 04m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7bc2096a24db4b2b877d6210564d24e9 : SUCCESS in 1h 02m 41s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3d8b84d210294dc6af3867b15dc21e24 : SUCCESS in 1h 45m 21s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6962eacddcaa46278aca2d692c7f504c : SUCCESS in 1h 30m 00s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6edde52b3a194cb485975b6486757939 : SUCCESS in 1h 31m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ceb0e4b09d704720a50526e6c3e613b0 : SUCCESS in 1h 33m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fd627ecb3ac84d74a4c8e9f419e363e6 : SUCCESS in 17m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/884564dfd4ac4bffbbe2eba509e62936 : SUCCESS in 21m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f03ec4c5a93540e795662533755c104e : SUCCESS in 23m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/245c896247364ea2ae274b284d40f9c3 : SUCCESS in 22m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6182883d84b346c6a20c8dec4410b0e6 : SUCCESS in 28m 14s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/03d29790af4049e595576b21e40e1186 : SUCCESS in 18m 51s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81ac46d3593948d39c14cf0e469cbd74 : SUCCESS in 32m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ab6a9f238b44057864102604813bef1 : SUCCESS in 29m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a11c9488f85d43cd8e6f6e405ad2f048 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2451eaaf6ec4a4e9275ea9581c20146 : SUCCESS in 21m 01s","accounts_in_message":[],"_revision_number":1},{"id":"586589665b99616297fe40e8b551940376625950","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-20 12:30:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1da07f92f6a78dd60ad4835eaa59e9ed77ed74b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 14:40:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f753c0c0b6ca4ff8957a44929d7eaf28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef786fd14c6a4c7bbd1a7602a24b8534 : SUCCESS in 10m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3bb08a89590743c1b92e73f7322e3c3a : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78ad2a95f97143d7a37705193749c841 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a252d32e5ba54ad4a59a58ded86f2c8d : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/740ce7dbe00847c6961957c86431ffbf : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e646fe3695ea41c9856a80a02468ad27 : SUCCESS in 8m 00s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7763f30fc0334acf98cea24ad1549a35 : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7116c83ac7754657852505e245f0e5a0 : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b785b561ebb499883292ac2401620a7 : SUCCESS in 7m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ec830501328346d497c2c0e426c4aa11 : SUCCESS in 14m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb974e198aed46ed9aa26c6467e9eae2 : SUCCESS in 12m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2a9fbaa5f69048e9a10a036b0d44ce0b : SUCCESS in 18m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee8e4ee57c0d446c9e29a123ddccabdf : SUCCESS in 11m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/47d15f6929b3411dac2f3c9c3af54728 : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fa74f9fd5cda4fb58cbd7cb1a711aaf5 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6c09d0013de14db7979609cdd459ca5d : SUCCESS in 1h 11m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c68b3eb0cfa147f187910bfa5173a805 : SUCCESS in 1h 04m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fb746d9106c44d5fa75476e22c5e43a8 : SUCCESS in 1h 09m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a07b6f809e134b23b90306ec529de17b : SUCCESS in 1h 36m 32s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/36623b1d095742bd97e95c5cd0e54226 : SUCCESS in 1h 37m 33s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/640fd4bab4be447f8bac38e1077d514a : SUCCESS in 1h 31m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76f3c3f861af4307a094d0ce97de56cd : SUCCESS in 1h 37m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/71244a10e98d4edaa271a5466a1eb752 : SUCCESS in 22m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c4792f8f33a24851bb06a2c414657433 : SUCCESS in 19m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e1b51bb31fb4b79a4c7dce95dcbffd5 : SUCCESS in 25m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/652d2e2515344a28bec1a0a053147f9a : SUCCESS in 14m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e05d2033c709454899f1d327bcfdd90a : SUCCESS in 18m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ffb28f439306466881c43a3143e164ce : SUCCESS in 20m 28s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7abb846f8cc42ffad7ea298a4e8cef6 : SUCCESS in 34m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28e3fbb491904ab193ef7c5901c498c0 : SUCCESS in 28m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c67a6cb7b6684c5a9443d2ee1c925f56 : SUCCESS in 22m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f3cc769c84042c9bfbeb0ce986ecff5 : SUCCESS in 20m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6ab2ed754a76d367798ea8eee25532d40c52237e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 14:41:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2cf163409d977689bdc5c236556ec06cb403e1c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-20 14:57:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3a5f4e3d5d2d11d6bf745b7628271b8bc1f055a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 16:15:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00f6fc00579849cc8b85ce5b46a80075\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40823395ad9b459d98835cc0c26e1090 : SUCCESS in 10m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/004267b28c7c4817a371d363a938f88b : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2168a9903da47ea963e4de9d45b2fa8 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4e571a8e428402ea05addccb745693d : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2bc8d0a292fd4cfdbbf4baf3e2c41eeb : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ded8563bf8740ea87218af6499ae3a5 : SUCCESS in 15m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7b88f446b7f48779a6a2143b5a0c5ec : SUCCESS in 12m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e5938c1c71cf4806b3b48400184fe1b2 : SUCCESS in 14m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/958467410ee04eb09ee5895e71cc0764 : SUCCESS in 19m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7b55c3d3ccfe43a2b886ee892605807f : SUCCESS in 27m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8289f14d972b4ca9b7ed35735492ddc8 : SUCCESS in 17m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/78d5bd0373454aeda88a0bd8bc0aa132 : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a3e38e7bcaad4596affc4ee41a3a7690 : SUCCESS in 1h 04m 21s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e332810b7e0741349f2b10a727fcef12 : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3615b4bdf3bc475ca4a165a809ef74b1 : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f7595fed95c34213857213bb76d18500 : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cdb88915581945cfb7e914c913fb43e7 : FAILURE in 17m 17s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/22a9854cb5654f14aa15c15bbeba2d7d : SUCCESS in 1h 32m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5118eedd1434d07af121cab1b271ffd : SUCCESS in 1h 31m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ed15cac98764a2ebf0f7913562d3878 : FAILURE in 6m 40s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4f3fb65643a1430f9c699b9802b6c50f : SUCCESS in 25m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b1c8f61ad33143cd967238f49792c42c : FAILURE in 7m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a62419831fb24b6994eb02f80d9396f4 : SUCCESS in 22m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6b8e3e3d08284b6eb04ec860ad045d5b : SUCCESS in 21m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cde6afee120d43eba610aa17b26cfa3c : SUCCESS in 16m 23s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d12a7f18ecb648cc8d0187f92448aa9f : SUCCESS in 29m 40s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86a9877ef0fc4bbb9fe50174082b6a2b : SUCCESS in 23m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3573de99acbf471ab9b036e70f305570 : SUCCESS in 33m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/498262e8f8504933a6c8ab2eafb18d07 : FAILURE in 5m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2364e2a98d0f4288bffe55b069e04f7a : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"63070bba162fa20aa2805d6e72cb993d5f2d5f3d","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-20 16:16:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"845c496683064e30029efd664d019c6c58de6e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 18:11:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7815496955874dc2b5a49b13135bbec0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278279cc16c745c196bad56563bd9145 : SUCCESS in 5m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3a76d721866e47a6ba8de3aef18375c9 : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c96d1db74f5d49a89715fcb044c60b30 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4353ee6c3004b6ab889904cce9a1390 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac6e1997a4c040f4beb5d76acafef920 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c47b45353bed44939a357c136e621e27 : SUCCESS in 8m 30s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84325234ac4c438bafca28ab4685e3ca : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cbd6cc7e9bf24bca8bf9aac9286f8fd8 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b1225056e4e4e1f86d9627270a396d6 : SUCCESS in 12m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cbd911fd35da418d97a22abebfc5dc93 : SUCCESS in 15m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5db9cbcde85d4ae6b7265cc7ef8c778e : SUCCESS in 19m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9422d55dd6b746b4ad03b730d5f8cf4a : SUCCESS in 12m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/620a0b3ca3f94bb7919c73b05258cfb4 : SUCCESS in 19m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d0c348c1714e4c7f8f1c8461683df4fd : SUCCESS in 1h 03m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1600292f41f540f69f7b54cd7f085300 : SUCCESS in 1h 07m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ba1f409da0f346f481829ba3043d6170 : SUCCESS in 1h 11m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7886efc6ea5e49f1b4815b5be6aa6113 : SUCCESS in 1h 03m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a4b0c687ad3451181d9f4e4942e232b : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dc8d09be3a5740e8b4efdc6d706c3957 : SUCCESS in 1h 34m 46s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6d784a07e9ef4932b9cda2ab51cafcc4 : SUCCESS in 1h 32m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90c2fda365f146cf9d38830f5dd82bcf : SUCCESS in 1h 29m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9392a6317eb1407f91dbed36e6965f80 : SUCCESS in 1h 40m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cbc0ce8f4ca642d2a81e5c627b914b24 : SUCCESS in 15m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/90ff425878084ba7b6355ab5cc4984e1 : SUCCESS in 20m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28476e7bb84844108adc177a3cbdc97a : SUCCESS in 18m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59a9550ba449431bb972887c626fdf28 : SUCCESS in 22m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f4532dbd193f4d0aaec89d1cf30fa650 : SUCCESS in 16m 10s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a7a4cd4ec78540eca3429d0ad37d6f2d : SUCCESS in 29m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63682ab9aa4246858f5b58e9615b1ec9 : SUCCESS in 34m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a046f54df75f4969b88a0c8e00b20c4e : SUCCESS in 30m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e359ba7b3f944c2d95d30f50f227f395 : SUCCESS in 15m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6915a5835eb4501ab11c3a02f3bf35b : SUCCESS in 21m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a370a2ae3a6880200dc9ecc898eca1dfea9d0a51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 18:11:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebc173a9c07be119f2d013bdc040807e99585ca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 19:49:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ff8796e939e4e19a9b0daaee1495488\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c6987cd4714d62ae7b15ae7ef64712 : SUCCESS in 10m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d588b08ab05b43768c6e1a8c55b9e721 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9bdd7fe1ad473080155fec7c101083 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48b8b2d508ac4447b0d97537246a3228 : SUCCESS in 4m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be025fb1d6254bb491015783ab30c4cf : SUCCESS in 6m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2fb7d1a7c1f643d4a8911edf241f1858 : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4596e0f64b44418b9eb5093c434cfcf2 : SUCCESS in 11m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/de3fa68a394445c881ed466e1c76daf1 : SUCCESS in 25m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d01e65d809404279a6de79de02b41303 : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d99a3f7157a447198f0034f0ad0c7f8f : SUCCESS in 17m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/805054a63ad840868db6e5d9cf2d96fe : SUCCESS in 19m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1924884a051d4dc992360e2422517994 : SUCCESS in 1h 01m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6de228a8244e475b9321f30b7be9c61c : SUCCESS in 1h 02m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/acc8a788ac584c479df7d1830ce4be25 : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c15ce414b7ea446ca6f4047277021bd2 : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f92dd2c243c547c19bd91de9ab669b43 : SUCCESS in 59m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a93f2d73310747839cad7aef5fdff33e : SUCCESS in 1h 36m 44s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7e95fbf5714e494999b7dcf177f5986f : SUCCESS in 1h 30m 47s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1b3026043864d48955114e5eb50bff0 : SUCCESS in 1h 26m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c9f71254eed4da8b7e06a4401d4d219 : SUCCESS in 1h 33m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f9e8b57e53db41f493cb91312febf5e5 : SUCCESS in 16m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/32c4a9bdc1f940e7a135d3a332537ab6 : SUCCESS in 21m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f71e1c5a23a4a4fbfdb173c96304f37 : SUCCESS in 24m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8efcb0b965a84746b12ef85a3473903b : SUCCESS in 21m 59s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0bb7b438b6de4b1eb3dcc55662d8186e : SUCCESS in 18m 43s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5c2a62a6fbc645ed8657baca85ffed63 : SUCCESS in 25m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6aac2f501448462a925a119859ba062d : SUCCESS in 29m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1794823821648d5be498501b05e19c6 : SUCCESS in 30m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/04f70b89e1a04212928299a29162f697 : SUCCESS in 26m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7bbc883a0534e4cba997187b6b70292 : SUCCESS in 20m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e7fcf0cca7c41963d9cc59e51c7e3b21875eb08a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 19:49:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3362c39bf20f444d6d8608dd41355ddf12b22ef9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 19:50:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c921f60f01e437cb66938f562ab0d66\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d945f6a0a24d0eb10c01a6b9715c93 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c863900248e95eb6b2c004940e536f3162ec8968","revisions":{"c863900248e95eb6b2c004940e536f3162ec8968":{"kind":"REWORK","_number":1,"created":"2026-04-20 08:27:45.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/39/985339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/985339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/985339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/985339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/985339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/985339/1"}}},"commit":{"parents":[{"commit":"6d34bfdcc1dfece086daeef153c02729f1bd81b4","subject":"Merge \"Override IPA image distribution/release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d34bfdcc1dfece086daeef153c02729f1bd81b4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-20 08:24:12.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-20 08:27:40.000000000","tz":60},"subject":"kolla-ansible inventory sync","message":"kolla-ansible inventory sync\n\nSee:\n\n- https://review.opendev.org/c/openstack/kolla-ansible/+/924575\n- https://review.opendev.org/c/openstack/kolla-ansible/+/959853\n\nChange-Id: Ib7c8e5451ca8bdcea8c4b4dcb1093680acce0778\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c863900248e95eb6b2c004940e536f3162ec8968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c863900248e95eb6b2c004940e536f3162ec8968"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
