)]}'
{"id":"openstack%2Fkayobe~985401","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Id3e610ad466212d3b8dde7a429ea66cc1562b047","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-21 15:07:45.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"hashtags":[],"change_id":"Id3e610ad466212d3b8dde7a429ea66cc1562b047","subject":"Bump stackhpc.openstack to 0.10.1","status":"MERGED","created":"2026-04-20 12:34:19.000000000","updated":"2026-04-21 16:41:08.000000000","submitted":"2026-04-21 16:40:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985401","meta_rev_id":"e28732f047bd7dea5a96b578421f398ff5cdaf29","_number":985401,"virtual_id_number":985401,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 16:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-21 15:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-21 15:07:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-20 12:45:37.000000000","updated_by":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"reviewer":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"state":"CC"},{"updated":"2026-04-20 14:34:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b0710a48347ce69f8f13ba63d4526d3828e763f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-20 12:34:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b3c81131e55580231eeebbef28e97cb7165498f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-20 14:31:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1068e757362155ab4f76bc34d18c70ede6c656d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 14:34:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/157f3b0a7def498c92ff622d289bf805\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2c31fe756b43d0a9d3e5c1a2c4f627 : SUCCESS in 11m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e33eaf411e9e4255b2c02f277a33261a : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0712e0b749646e4b2daea909c6eed38 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b8f3c1e9ee46558d0c34a0d37b1763 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea096dc0a9ae402295cb13b1df2137a5 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50c3a500e2a14a9ea7b15b231512801f : SUCCESS in 6m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc43eebddbb84888ace960d08bfbb353 : SUCCESS in 4m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a37335b96eb44e7d9736c3bf4bcb7255 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f603e2bbc26d4a26a1e1cfb870016b1b : SUCCESS in 13m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1b9e9ddd5704106b29e526097c0a153 : SUCCESS in 11m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/70a15e9bbf3840b29d0407ccd8370902 : SUCCESS in 14m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d25c00d8a9eb49a599da10afe7c98f4d : SUCCESS in 16m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/21a9cea4b987492c8a9094fe7e6bc0ff : SUCCESS in 19m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f711d37ea6e42d3b22ea10f694bcb48 : SUCCESS in 14m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/30c7f77cf17a40e7a1dea69c0ac209ac : SUCCESS in 56m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bce0eea04bbf42d2a066fe9d42b2eeb7 : SUCCESS in 58m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a25c294a44204e69a6c96379de6f7cf8 : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aec7c2eb3e9a46eaace32813603ad973 : SUCCESS in 1h 02m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/068d088c4c8d49cd9b56abc5d5e81d7a : SUCCESS in 1h 02m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3eb8664c76a84b9bb46a52b1e39369a6 : SUCCESS in 1h 27m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc4e1eeb5c2a490ca8b986a91bbf0522 : SUCCESS in 1h 29m 02s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cbfd39951c824c0387e6002379adb77e : SUCCESS in 17m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/baead307bee2426aace1e6c936dbd3a5 : SUCCESS in 15m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61c1feaa2e7a46229063110a155ab809 : SUCCESS in 17m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/513dcd6d50e14dfd966a6e6a8fa6c73f : SUCCESS in 16m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9179f813d36d4a84a6d3b2612f59b55f : SUCCESS in 26m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2f2eab69d7b40cbb77d79c0c07ff927 : FAILURE in 29m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ee6f16b09f044195af3cc19cbf199e3d : SUCCESS in 14m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8e7f1c876d643df8e3ba1a4cb6b12e8 : SUCCESS in 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"adbe004991e3a11b547ec81aac6cd9f90f7976c9","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 16:24:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3574a37ab66400eaf30a6aa01dbe30a\n\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/311d0f4cc4114f1f9196b991d90a36a8 : SUCCESS in 11m 15s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/89a0715948fc47e0a3986bcfaa37ad8e : SUCCESS in 22m 24s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/cf8b610bd77f4ed084448ab7c2f7b82f : SUCCESS in 13m 16s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/bf1968dac8274542bdc54958a9cf6ec5 : SUCCESS in 22m 14s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/47f1d531905b469680ec7eedd369039f : SUCCESS in 50m 19s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/c6bf59d3e15f4cdf8b8b4dc6f0a3bce6 : SUCCESS in 20m 01s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/755f8139f3dc4741b63b20ef61b1ea3f : SUCCESS in 56m 43s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/5227fd8767934f8eb26ff89194173c4d : SUCCESS in 11m 52s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/70936b55b00340578b7ca985fb06dcd1 : SUCCESS in 49m 47s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/2c321e5de4494aafaa3e960664b4c569 : SUCCESS in 1h 18m 35s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/54e78eaa9e2046168635f8e9172653d4 : SUCCESS in 1h 09m 45s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3deffbbbe4b442148198a5e4a71ebaa9 : SUCCESS in 1h 00m 19s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bbe2ca55b371437c91780e874203ba86 : SUCCESS in 1h 21m 37s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/9002928f0160489592c38d87f148537a : SUCCESS in 9m 29s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/6fb13aec1023416298750b37cf879489 : SUCCESS in 20m 33s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/8d2c57fb626e4b2b8fce003fd5445aea : SUCCESS in 11m 15s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/3490c29600304c618ea60e8b914fc33e : SUCCESS in 21m 41s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/b69baccd6c2a4fcb8d1da51e0253e9c5 : SUCCESS in 21m 30s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/b6a4c5c5bb36446ead73a84a47b63f6e : SUCCESS in 23m 37s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/02ffc041b0eb43ffbeec411c2d7fea26 : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"31ddce5b57e86906968bad24df7fbf70e10df9a7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-21 11:16:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e20682978dacee84f91222af309bafc99186ade8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 12:47:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4e527bcec69405db1e9e47feb92f20d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b0f434a5a045c1b40cc7bf43306b0c : SUCCESS in 4m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/af52d54ccaec403dac6be4977167fe03 : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8de56a9c49734b798177148611ff6c24 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389891f90e56482e932db441b0b39738 : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a7122cedd5443daaabf058919ef6597 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a60362590364413987161ddf352c55ee : SUCCESS in 3m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/539e37c0149c4bf9a35ac866d54ce84e : SUCCESS in 2m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2517dc69f64e4cb58b9c1ad6577a6395 : SUCCESS in 3m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/258bf44631704bacb5d4ed725cf23565 : SUCCESS in 14m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/106427ba19b84889b8c2c8a362b09f96 : SUCCESS in 13m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cf738d662bae4db7b3684c07834cb790 : SUCCESS in 14m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c430748865c14ceb9e15e6b8b5d9a053 : SUCCESS in 7m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9d64d32dae1d49ccbdaff441e855f78d : SUCCESS in 11m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0881a7a2c6b04ca998b63eb6e6480e19 : SUCCESS in 15m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/efef9566915c449db96c998c686b8e60 : SUCCESS in 59m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/67f74dac164f4ce8b19acb2197e416b6 : SUCCESS in 57m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bc5d40f7e2ab4cf18e5240e706ccbb11 : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99934b295d9f445e9a14ea1e6f2ba5b5 : SUCCESS in 58m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ddc99d580c7488d863213325731a01e : SUCCESS in 1h 00m 54s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f0995e9b14064369a14bbd2088d5af7d : FAILURE in 1h 29m 57s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13d54b773e7b450f9c7ffff50701339e : SUCCESS in 1h 26m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fc5d1d9108bd4eb18a0aa628b6f7f364 : SUCCESS in 9m 44s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/72eae61043254993b45babeee2949628 : SUCCESS in 10m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7b6fcffc4ab44cb8eb7e93ce584b2ed : SUCCESS in 8m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ab82c58ea3724ca3ae4a29cd8e100483 : SUCCESS in 16m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e42f6be6fd464121805c2c5fe3d228ce : SUCCESS in 16m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a407ba7506be41658a00940a490e1fd2 : SUCCESS in 16m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d55ea55b2852435aacb44ecb56ebfeb8 : SUCCESS in 9m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c447f54eaec4bf684764a38a534f0a8 : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":1},{"id":"357412c510ffac26eb8204ed456917090543dfa5","author":{"_account_id":38873,"name":"Owen Jones","display_name":"Owen Jones","email":"owen@stackhpc.com","username":"owenjones"},"date":"2026-04-21 13:22:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02c832372fe6077b4ee28eeb4d8fc3d6a9dcd3cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 14:56:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cac4107e8011486e977fed8a330a8dda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea37c50155c24be6b1a4c1f10b54d3b3 : SUCCESS in 9m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c3e1316dbfa449b190b5bf7e78b2abf3 : SUCCESS in 6m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c2790689b1c485389f0e1e452b0be0f : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31603d9aa54543408a37ef89b0beb361 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d9f8074cfdc47b6a87cac61505f4c20 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91225cd4d62b4cbebe62e261d3d28af5 : SUCCESS in 7m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a50c7ab5845c43ca9e2bfea34823ac01 : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6678199d64294d24adc55fb8a23169d1 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5e959853d25424983b784068e3b6d9b : SUCCESS in 12m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e86715b0a1854088b052f0d08d22973a : SUCCESS in 9m 40s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1aa3d95e4d304fc68d46c52d4a51ceb1 : SUCCESS in 15m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecb73a9834664883b7127e0b7b2077f2 : SUCCESS in 14m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/17c947e7512d46bd86eb11bef42eb15e : SUCCESS in 13m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8ba2111159a4fa3b9e17b53405be4e5 : SUCCESS in 17m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ca967bc873004181863c3f17b72bbaee : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/868361d52d4b4c0a8e8f1bc328889119 : SUCCESS in 59m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2bf7fe8c4a0a4e5680c5d516bd6d68f5 : SUCCESS in 1h 00m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4995347d65e244a7b5d1ebd1becc3518 : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f86912a82f564013adef688104e4d008 : SUCCESS in 1h 05m 57s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fd029eeecb7e4cb586b9298675b21053 : SUCCESS in 1h 26m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/804e729d3b0542f0b4e97cb3466df844 : SUCCESS in 1h 30m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f4d299269c094bffbae09b501f109a85 : SUCCESS in 15m 43s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a707936142194b40862c6f6c56a9c71f : SUCCESS in 18m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f45707d3ca4a426aa45231ce247039fc : SUCCESS in 16m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4e29fc3b1940468a89c5d409e00cd4bd : SUCCESS in 16m 25s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/94db4364208b41f4b2d399309c694d73 : SUCCESS in 30m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e5f08a5464c473b91379c817cba009a : SUCCESS in 31m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ecb72cb5cc934969918e776dc5442d81 : SUCCESS in 16m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ee7e4aa8bad462fbb46b5b75dea46d0 : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":1},{"id":"288778538a18b9bc390c31f339901f538378c962","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-21 15:07:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ddb863851853c01949546e40d181b1c97ce65029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:08:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21f3826c5cc87c6a94db282a220b99f2a2adbfc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 16:40:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e158a8d03d349dab8fd2ca7341f655a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9978a29bb24c4a859ae2d69e1b42f0b0 : SUCCESS in 5m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a03a7a5ff3bc4df9aa92434c42b4f71d : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbcd71e995be4fc88f4148ad9397d29d : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e696131b14a4461a9a883cf3683ff249 : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff9d6097d2d04d9d8b8253d2f7d677b4 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/65dca8b96b434b1bb479a88ac55b0a31 : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/178c819c30d34ed983ad8f2a2cb98667 : SUCCESS in 6m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/de980d01e1a9486fa0e0c2db56be3796 : SUCCESS in 12m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4588ddee9172421b930be1eeddd68744 : SUCCESS in 13m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88b75a244b0e4927a9e84f377f056c86 : SUCCESS in 12m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/58e6a14937e04da2ad807e8feac875a9 : SUCCESS in 25m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7953526ae3fd4213a1daf347fd89500e : SUCCESS in 19m 54s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c4e0f46d219e4a3ba4e2325831697d6c : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1ca6af16fc544003b831288ff25f043c : SUCCESS in 1h 00m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2f55c0dac0624b4eb6497d88cac8218e : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f78ea1d1c365479687864cfff08f86d7 : SUCCESS in 57m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f17690e8245a4ae88a1111974c20e2bf : SUCCESS in 1h 00m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c6544488b2084d4792e236dd6aab0b03 : SUCCESS in 1h 30m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cca3532578c4922b9744dcf53dbb8ee : SUCCESS in 1h 29m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2914e509961a41e5af974310a29be66f : SUCCESS in 17m 20s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1af5231befbf4f1e901e3c73d0e1b8f1 : SUCCESS in 15m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cad385979064b8382471c3b78df557d : SUCCESS in 9m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e8bf9eb25e9948ca89ec96af4275ee4c : SUCCESS in 13m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/27546971a44447ebb7a77e46764cea85 : SUCCESS in 24m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf4b9339c7d049099b9c21721e26e0bb : SUCCESS in 21m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c2722dfc364f4e6baa35596db1c5829c : SUCCESS in 9m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f8a34d36b094460ac1d22942230f8cc : SUCCESS in 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"38ec3aee7cfd4359dadc05a01b290af029e9919e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 16:40:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e28732f047bd7dea5a96b578421f398ff5cdaf29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 16:41:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3db98c86c0c74b20b8fe6d32fc798a88\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3be2089ef743a088238291d289a0db : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0bc1a68d787492ea317e73810252519 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cb742fb3e6c36689f692247c0fc9dda58657a27","revisions":{"2cb742fb3e6c36689f692247c0fc9dda58657a27":{"kind":"REWORK","_number":1,"created":"2026-04-20 12:34:19.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/01/985401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/985401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/985401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/985401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/985401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/985401/1"}}},"commit":{"parents":[{"commit":"11e2813de0619197586c27d6f6a2b650e0400ff3","subject":"Merge \"Add support for building valkey images\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11e2813de0619197586c27d6f6a2b650e0400ff3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-16 15:04:46.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-20 12:34:12.000000000","tz":60},"subject":"Bump stackhpc.openstack to 0.10.1","message":"Bump stackhpc.openstack to 0.10.1\n\nThis prevents us needing the CRB and EPEL repositories to be enabled\non Rocky hosts when building DIB images.\n\nCloses-Bug: #2141684\nCloses-Bug: #2142501\nChange-Id: Id3e610ad466212d3b8dde7a429ea66cc1562b047\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit 79bee73dcb9fe2b07157c98ef4a6cf812ac9250a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cb742fb3e6c36689f692247c0fc9dda58657a27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cb742fb3e6c36689f692247c0fc9dda58657a27"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
