)]}'
{"id":"openstack%2Fkayobe~985781","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~Iad7fc73f16ffcbc2ab5f261322bd6777036fd5f6","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-22 17:41:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad7fc73f16ffcbc2ab5f261322bd6777036fd5f6","subject":"Block until timedatectl status show synchronized","status":"MERGED","created":"2026-04-22 08:55:27.000000000","updated":"2026-04-22 17:42:09.000000000","submitted":"2026-04-22 17:41:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985781","cherry_pick_of_change":982396,"cherry_pick_of_patch_set":3,"meta_rev_id":"c7f3cb0dee8fe2999fb4c910a5a0dd3f0019c401","_number":985781,"virtual_id_number":985781,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-22 17:41:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-22 15:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-22 16:04:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-22 15:28:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-22 11:06:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-22 16:04:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"cff6c3e454059b37d220ddc7442854c269892a76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-22 08:55:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"00af08bd10c7a8abedc437ac040110dcf03a2a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 11:06:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5abc52202e60483d99c5518b9db2bca8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2f3999c702c49f8b639259fa3c40942 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7402905b4474db9984746313132750b : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/356c72eb8f4345d980809340dab62479 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca51bf7fcfdb4d068415790b094d56a3 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d1e6aaaad04f46aa8ff7611112ddcd : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f26b7882bcd84ed48f5d8cb0d65cea83 : SUCCESS in 2m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/15806f320bba42289a21d7146ef45327 : SUCCESS in 2m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/720323c0cc394976b7d99746f46d481b : SUCCESS in 7m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d57f195455b248728f5f642b5bdd6bde : SUCCESS in 10m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/93abe592db3244c28787d7c733cff3e6 : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7e25315076db4d208fd6a109272e6242 : FAILURE in 9m 36s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10ce441acb26415e9d05e6c8ac6509a1 : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/84a2f905162848f2bd4b313020101e08 : SUCCESS in 47m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1e5dab256211441f8c49264ea3d2cb48 : SUCCESS in 17m 13s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/716e98ed39634c548fdf81d5b6079032 : FAILURE in 10m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/293e489c16304a17a420b979a6f397e8 : SUCCESS in 14m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/20289befece64c1b8ab6908ddf89ebed : SUCCESS in 2h 05m 48s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9f9379ae22b4e3eb20add80c4bd69e2 : SUCCESS in 1h 48m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/381fc3a47a894edfa6a523446430ba2f : SUCCESS in 22m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf0cf81258234e779abfba6b62755906 : SUCCESS in 21m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/aff7bbed957e4c45ba758016287b6254 : SUCCESS in 37m 01s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/facf0b929e8d464f8f7a14ff7b933dbc : FAILURE in 14m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a61188a5654242e1a15629cb947e2854 : SUCCESS in 19m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd5e1ecd3d694632ae8896ba21d6e285 : SUCCESS in 17m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/58876711c8ec4ddb9da9cbf97df21b5f : SUCCESS in 17m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1f59c90187843a6bf0c0a688df1a3b5 : SUCCESS in 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"78ad7ba31b37f68b377279a10bae8ed295e98f35","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-22 11:15:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef8dd5606eb79be25e72b5e144dfa2567d1dc799","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 12:26:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6d9fad3857428fa6a9af9cd32b25ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/051a227ab04e48a1a0b6fe8110820535 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c21aac7a0f42f09ff76a16086f6c31 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a0e3d41054148128e46f4515242312f : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8241d7e7683f4d18ae7f966a4bdc2e27 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d630eef2b80a47bd87b4b3742939ea9b : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/711d3da1873743a090c7109b113634a1 : SUCCESS in 2m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f611b64dd0784f50bab963ea4fff5b75 : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/46a223258b984ace83afc91bcbc7507f : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/545ec3be0eea4b3398da3b1273563075 : SUCCESS in 6m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3e2881fd28e04cddb9b5c007c83a6600 : SUCCESS in 39m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/884aae533a274a3c950df764d78c48ee : FAILURE in 10m 21s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88b17444741a47629da5216ea7b26dc8 : SUCCESS in 34m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8c965573f08347978c22d589e167bb2b : SUCCESS in 58m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6abc0a22b94444369bd67fcb635a0d23 : SUCCESS in 12m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cf1f35a26c844a45a12ab6e01b71c31a : FAILURE in 9m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/846fe23014574e68882b0f6ce5e6228d : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f044abec40a94a1ab72f3a0bedfc049b : SUCCESS in 1h 08m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43b04fbd5698450ca819063397bb943d : SUCCESS in 58m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9ab9dcba7fbe446fb0ec964c776bf211 : SUCCESS in 20m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c2642d3d8934ef49993612e06c4e799 : SUCCESS in 18m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/190cb6083b6f4a40bcf1c34d74087fb3 : SUCCESS in 39m 27s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2f280dbb12b4fd0af25b9825dfd9c71 : SUCCESS in 20m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4ee8b7470af04fda9ca8b13b8ed953f4 : SUCCESS in 13m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83828f758a814664bf137a311d759966 : SUCCESS in 16m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/53017eaa412a428abbd37f19b5bdd719 : SUCCESS in 19m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a73383e44cb74a6c91630fe300a08e4f : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b28d143008b2c3d393489af839d2fadd3f6c761e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-22 15:28:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ecb05a4f462af1d50a11529c1c6280b2d0557150","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 15:29:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72318860002295e86c1c1f55b77535b7886aeaf3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-22 16:04:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8d4f34f6b1da5d68856142297ef6494b18f85b32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 17:41:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f83bb123fe54668aef0d91a1b9bf32c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d083a8473c504f68a41d1bb72aa6d19a : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67c82bd7dd0e4e9db042a551aafe2f6f : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85606418bbfe4d5ab4e60929cad26475 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c410136497f24047b035249061bacb35 : SUCCESS in 2m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2a169f74eca646d69a0af060f3eeafbd : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3c2d36d1a5443f3bc846beab27874a9 : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8a60a69d94b455288e947aa509b350d : SUCCESS in 9m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7adeaa4bd1bb4021b4cce8a163ae8032 : SUCCESS in 1h 04m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa7e4446c52e403684dff234f886ecdc : SUCCESS in 33m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/13bf34712aea467aa00d567b3ace228c : SUCCESS in 41m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/de119db75c70471ab2bbdd44a58105d3 : SUCCESS in 22m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a77ebf273484505b0514f91082ca0e5 : SUCCESS in 18m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2bdbaa48a15a40aab0fd172a123a8d31 : SUCCESS in 2h 06m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fee15d6f9db54bc99d30cb1c0c113395 : SUCCESS in 59m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/de176f0a950848578734c617e6712b7d : SUCCESS in 25m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3fcb9f842504a2693cd38dd1d3cb502 : SUCCESS in 19m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b55c5bc85b314eee9337f27d4bcade1d : SUCCESS in 36m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46be3f8d3cd9436b99debf001c3dea48 : SUCCESS in 20m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6da15620be0e4d3089ef83efe85b95a2 : SUCCESS in 10m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a51c8433a9e4625bfe959c9d968cb3a : SUCCESS in 16m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9afb8ce89cff4756a759fcce1423aec3 : SUCCESS in 13m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/208e1ac15d2a45a9b33edc37c67686c4 : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"724ac6994a461c4ccb9fd3b05006e5c2a8732d11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 17:41:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c7f3cb0dee8fe2999fb4c910a5a0dd3f0019c401","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 17:42:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e3fddf7399a42a1b71c6169bd6619ca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4d54a7627c472492abf7fed640aecc : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/869160748a7f47ef897d6ec8d9c598c7 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4176fbeaefca576cb3156f79d31db62980ddb6e0","revisions":{"4176fbeaefca576cb3156f79d31db62980ddb6e0":{"kind":"REWORK","_number":1,"created":"2026-04-22 08:55:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/81/985781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/81/985781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/985781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/985781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/985781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/81/985781/1"}}},"commit":{"parents":[{"commit":"a7ddf996e3a9b143b25563f46b00db446c621140","subject":"[networkd] Fix broken conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7ddf996e3a9b143b25563f46b00db446c621140"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-03-27 11:44:11.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-22 08:55:27.000000000","tz":0},"subject":"Block until timedatectl status show synchronized","message":"Block until timedatectl status show synchronized\n\nShould reduce incidents of time sync issues when running kolla\npre-checks.\n\nChange-Id: Iad7fc73f16ffcbc2ab5f261322bd6777036fd5f6\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit f23ced2b59731c7eb89d9212237e933042f74d7b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4176fbeaefca576cb3156f79d31db62980ddb6e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4176fbeaefca576cb3156f79d31db62980ddb6e0"}]},"branch":"refs/heads/stable/2024.2","conflicts":{"ours":"a7ddf996e3a9b143b25563f46b00db446c621140","theirs":"f23ced2b59731c7eb89d9212237e933042f74d7b","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
