)]}'
{"id":"openstack%2Fkayobe~985855","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Id882a49fd9bf30df0afbb31fe1c9ea11ffd57606","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-23 14:13:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id882a49fd9bf30df0afbb31fe1c9ea11ffd57606","subject":"[stable-only] Fix key collection in luks role","status":"MERGED","created":"2026-04-22 17:14:34.000000000","updated":"2026-04-23 14:14:53.000000000","submitted":"2026-04-23 14:13:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985855","meta_rev_id":"879596c28e6fb574e4b4842780f941ede0b2797f","_number":985855,"virtual_id_number":985855,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-23 14:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-23 11:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-23 11:41:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-23 11:41:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-23 10:44:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-22 18:45:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-23 11:41:35.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"5f55370c0a92637361281d3379b9288910ada2f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-22 17:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc378753517c77a306375003c584a6c10c9909e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 18:45:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e2473b285214cbaaa5115a5e2983b0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bffebf9d6db940f8a70e59231284ed0b : SUCCESS in 8m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/107f99c4144c4cb0a07e7fa55f8f78ad : SUCCESS in 7m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/801b4a7a023b40c0ad47ea536eab329d : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d183f34da0404f0a9819a14e63e5fdb8 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6d06e44fbc6438d9f9fc9e50a4ae87e : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83412acf0353482cb045c37effaefe94 : SUCCESS in 6m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59f8cadd076e4618b8c2523b1b2c0014 : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10a254e39bae453594a08dd440c750d9 : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c17fa5a79f494cca922bb098dfc76a20 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac51a6e886494e2b93f961accfc14fed : SUCCESS in 11m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a1b4fbecb6b940309696c2e01d23999c : SUCCESS in 9m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40894eb3912e4264a8de821507be1b5c : SUCCESS in 8m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/197e855f345849959f6ed8e6d2539e7c : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/477af38bb1b44d0ea1dfba8e55f6475b : SUCCESS in 8m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/93023d44f8c74761b7f29027f933dd6a : SUCCESS in 1h 05m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1ed6221861c140adbe14f77ef13ad77a : SUCCESS in 58m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/aa1d0368d3c344168152f57c201d785d : SUCCESS in 59m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb7e336c87bb470bab36bb671b9b9be5 : SUCCESS in 59m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d97791b58769478499ec7ffa6b821d8a : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c58f9e433dca4ef89b4b08cc0d50df5f : SUCCESS in 1h 29m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2962376fea048d184744eb3dc5ef5f8 : SUCCESS in 1h 28m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1207ac2af4b9487491c27c53c75faa69 : SUCCESS in 9m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/dc2e87e0528f4db392dc0dcf064bb123 : SUCCESS in 16m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8538933a5f744a87bccdd2b6739f3265 : SUCCESS in 19m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/faab11ad5151461ca7b810cf3da6cd06 : SUCCESS in 16m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7ba78996516742baba376adbbe87009a : SUCCESS in 17m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b728a3d03c684f44ba7109c31332a5be : SUCCESS in 17m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/39b1ea3259ea41b49dba4adfea0c9c7b : SUCCESS in 15m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3dd4e36665e4a7289093f11154c2ee1 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"df7f2802c8fd07d2cdac4902f1436d933c8c032c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-23 10:44:08.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"f50daa19af8150bc34d3559e168bce3daded71ff","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-23 11:41:35.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6cb9e1b63d419b07a2119230b9118bcad48a28c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 11:42:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c7140c1fc07364e726bf289a0a950f07ae1d8b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 14:13:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba7f7dcff52b4153a4c14214e923579f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c96d71b6cf448f29b69b8ba2763988b : SUCCESS in 4m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e20e7770e46843a9945de36ec9e73ec5 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a0fca36540d4463bbc365ed6f197fd1 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/779f67f62889432cbf4272c6ec85e32e : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da9e26b23be54588830ca431d6ece691 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/140a96bf335e4f71a4e727581d070206 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/52bb387e4ea54a40b3756a3830becac5 : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e649ada0d61e45cebd8fd9a1687522ff : SUCCESS in 6m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1905a08980f64a9b8afa3b1134cd089d : SUCCESS in 10m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e62e8509a4d4c978ca3acecdbe7952b : SUCCESS in 12m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/283686ac024b4ee1b574fce0149bae60 : SUCCESS in 10m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a43df85b8cc4887854cf531e4781c52 : SUCCESS in 8m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7dd3952ea7564052a163c6938ce4426c : SUCCESS in 57m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7199a1af07e6404395541e8a0f332b54 : SUCCESS in 56m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/28651586baf34955b11fdc10f26db74a : SUCCESS in 56m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7734ed4d131b49069e52b1f58331df0e : SUCCESS in 59m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/10f15d8156204dd2ae29513aa3088566 : SUCCESS in 58m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/34dad24d6b11476fa30e58c42175cc90 : SUCCESS in 1h 24m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c220e36da98b4214a042d5a9ac9c7162 : SUCCESS in 1h 28m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0b8c83504e804423917073998ed1d117 : SUCCESS in 22m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/695bf332f5f54ce3a46d297171d0cabd : SUCCESS in 9m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f248634de33428d9d328fdc44277429 : SUCCESS in 8m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d8569c69756042d7b0dbc8cce787f8f9 : SUCCESS in 14m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7c72123c6b5c49d3bb632bf5f7eaa830 : SUCCESS in 16m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cf256f374e74ed9b8bb413f7694b402 : SUCCESS in 15m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cd4275defea44ee9be7cfa53d5251c9e : SUCCESS in 19m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3a441770b6a436dbbf80960428d7619 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"41ad17a9f933a788207f0e5e4840b3cac9ff6387","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 14:13:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"879596c28e6fb574e4b4842780f941ede0b2797f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 14:14:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a3f0d1f1d3743a399449bde86390b6c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1356bf7ee6d34c3f94f1987f86fda433 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99c4c3f376f449c68551d9fd806d950e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9179a6fd2a5ee874b498c90f111512d1cfc351e","revisions":{"c9179a6fd2a5ee874b498c90f111512d1cfc351e":{"kind":"REWORK","_number":1,"created":"2026-04-22 17:14:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/55/985855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/985855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/985855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/985855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/985855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/985855/1"}}},"commit":{"parents":[{"commit":"caa9fbb8e7eb77fb9cebe0d6bcc755d88d3b5f5f","subject":"Block until timedatectl status show synchronized","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/caa9fbb8e7eb77fb9cebe0d6bcc755d88d3b5f5f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-22 16:58:46.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-22 17:14:27.000000000","tz":60},"subject":"[stable-only] Fix key collection in luks role","message":"[stable-only] Fix key collection in luks role\n\nThe luks role set bad permissions for the for key directory on the\ncontrol host in versions \u003c 0.4.3[1]. The symptom was that the\nsynchronize module would fail with a permissions error.\n\nThis issue has already fixed on master when the versions of all roles\nwere bumped[2].\n\n[1] https://github.com/stackhpc/ansible-role-luks/commit/b3d4cfbda3fa3b60defe1c5e0be705bf5a337d2c#diff-cceeaaa1206687d1775b9640dd0a6536f606e79c6305ebccfa1d01917463cb1cL34-L67\n[2] https://review.opendev.org/c/openstack/kayobe/+/976059\n\nChange-Id: Id882a49fd9bf30df0afbb31fe1c9ea11ffd57606\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9179a6fd2a5ee874b498c90f111512d1cfc351e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9179a6fd2a5ee874b498c90f111512d1cfc351e"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
