)]}'
{"id":"openstack%2Fkayobe~985936","triplet_id":"openstack%2Fkayobe~master~I963fd697b50694692ce4bcc91b2658ea90ac6cb2","project":"openstack/kayobe","branch":"master","topic":"hwclock-utc","hashtags":[],"change_id":"I963fd697b50694692ce4bcc91b2658ea90ac6cb2","subject":"Set RTC in UTC when forcing synchronization","status":"ABANDONED","created":"2026-04-23 10:34:41.000000000","updated":"2026-04-24 09:47:09.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1ce8c8e01bdebe2cd213739df18d418d896d9fc3","_number":985936,"virtual_id_number":985936,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-23 12:30:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":-1,"date":"2026-04-23 10:39:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-23 12:30:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da8f98e6662ec5da80f259d4069ed790e4635b8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-23 10:34:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82ca205a23d6511ac6a487ba40cb6273a996b98e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-23 10:35:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9293a3e1d8983349bc00e4d4d87e0740c3601fc2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-23 10:39:44.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab98eee32c1837d75b4feee8f190bf19f444a0cc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-23 10:40:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fee611f875b05e92a5edd115537468cd27af1f2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 12:30:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d956fa82423c44c191153175a2e2d8f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d118e33882fb4145952fb26e4eb1c5bb : SUCCESS in 8m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d9c991f99dec4582b4f14c969a0f1c9a : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc9f90d625144d75a518e6e4f2cbcfa7 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/268803e0b0f44445b2970df1e836619d : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca8e4d7a15ad46f9ab6896047f63b089 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/66901fb567064f46acd6486649d16df5 : SUCCESS in 6m 44s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/def55ed9fd73425f8e48c2fcbe9e3e1e : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/49fe783ca37944fa8611d027d4a6f443 : SUCCESS in 13m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4deb819ea4c04aa7852ad71e846b6c0e : SUCCESS in 9m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e72df2828c9f456ca6723b383ec80d10 : SUCCESS in 13m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9431b47b82c64112a34878296f1a1480 : SUCCESS in 19m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/508344ecdc32492d973f58e4917cda27 : SUCCESS in 17m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1943df849124d94bf12b40d55467e54 : SUCCESS in 17m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2ea9d49b78764c36bc0e866d68fcb2f1 : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f6a680e80cf3480b9d73a27539021d8e : SUCCESS in 1h 04m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3c04e4e881ec48b39b48e994a1de6dc5 : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5c058ecb9804c7084b1a7450ed613c5 : SUCCESS in 1h 08m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cf96cb8c5f234a45aeaf7efda87c345b : SUCCESS in 1h 09m 41s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/02ddc899b67f4763bdf2189a05a8ee1a : SUCCESS in 1h 53m 03s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6194a53001b7447ea7141a2c97da0fb5 : SUCCESS in 1h 33m 33s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a134744c62f40f7bd3399b312328b28 : SUCCESS in 1h 32m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58d65baafbfc453fb9530f2a1a066732 : SUCCESS in 1h 42m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/70f7b7c8dff54da5b2d82acc9e2c2ca0 : SUCCESS in 17m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/de3bcc867ad54f29822319cc7536e4ee : SUCCESS in 15m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23fadc4ecd20473181f6e2d7b94954da : SUCCESS in 18m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/748b63f73f01432482075638d061a38e : SUCCESS in 15m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/48cda41057a6489c807cf17d15c684b0 : SUCCESS in 14m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/be1e1080aa2b4e548961fd6b3d63405b : SUCCESS in 16m 42s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7da0ab6df104aea8e5ed519845589b7 : SUCCESS in 29m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a2b866abf13493e877196ef384e0514 : SUCCESS in 35m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a5bab39450ff4d2ea84eda36ef9a4663 : SUCCESS in 14m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6660df9d73db4cf0aa999a744db856e0 : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":2},{"id":"1ce8c8e01bdebe2cd213739df18d418d896d9fc3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-24 09:47:09.000000000","message":"Abandoned\n\nDon\u0027t think this will help. Since UTC is the default in modern versions ...","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"52784f454c112e405482323ec44f6dad7a1435b2","revisions":{"216695d8347878257ddb3d8335ee03329f9e72f3":{"kind":"REWORK","_number":1,"created":"2026-04-23 10:34:41.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/985936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/985936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/985936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/985936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/985936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/985936/1"}}},"commit":{"parents":[{"commit":"1c2e12d7c8e29479584be1a460f29640ebaa21c1","subject":"Merge \"Fix issue with systemd-networkd-wait-online when no IP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c2e12d7c8e29479584be1a460f29640ebaa21c1"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-23 10:18:43.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-23 10:34:35.000000000","tz":60},"subject":"Set RTC in UTC when force synchronization","message":"Set RTC in UTC when force synchronization\n\nThis is to match the behaviour in the mrlesmithjr.chrony role which sets\nthe rtconutc configuration option[1]. The behaviour of hwclock is to use\nlocaltime as a fallback:\n\n    If you specify neither --utc nor --localtime , the default is\n    whichever was specified the last time hwclock was used to set the\n    clock (i.e. hwclock was successfully run with the --set, --systohc,\n    or --adjust options), as recorded in the adjtime file. If the\n    adjtime file doesn\u0027t exist, the default is local time.\n\nThe speculation is that this could explain why timedatectl status\nreports unsynchronized. My understanding is that the kernel will call\nsync_cmos_clock every 11 minutes and if the RTC is out of sychronization\nby more than 16s, it will set STA_UNSYNC. This is then detected by\ntimedatectl, which displays it as unsynchronized. I\u0027m thinking if local\ntime differs from UTC, then STA_UNSYNC could be set on the first call of\nsync_cmos_clock before chronyc has set the time in UTC.\n\n[1] https://github.com/mrlesmithjr/ansible-chrony/blob/e1147d039e940fd85bf1529ef3d804ce8df21f06/templates/etc/chrony/chrony.conf.j2#L18C1-L18C9\n\nChange-Id: I963fd697b50694692ce4bcc91b2658ea90ac6cb2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/216695d8347878257ddb3d8335ee03329f9e72f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/216695d8347878257ddb3d8335ee03329f9e72f3"}]},"branch":"refs/heads/master"},"52784f454c112e405482323ec44f6dad7a1435b2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-23 10:35:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/985936/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/985936/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/985936/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/985936/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/985936/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/985936/2"}}},"commit":{"parents":[{"commit":"1c2e12d7c8e29479584be1a460f29640ebaa21c1","subject":"Merge \"Fix issue with systemd-networkd-wait-online when no IP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c2e12d7c8e29479584be1a460f29640ebaa21c1"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-23 10:18:43.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-23 10:34:54.000000000","tz":60},"subject":"Set RTC in UTC when forcing synchronization","message":"Set RTC in UTC when forcing synchronization\n\nThis is to match the behaviour in the mrlesmithjr.chrony role which sets\nthe rtconutc configuration option[1]. The behaviour of hwclock is to use\nlocaltime as a fallback:\n\n    If you specify neither --utc nor --localtime , the default is\n    whichever was specified the last time hwclock was used to set the\n    clock (i.e. hwclock was successfully run with the --set, --systohc,\n    or --adjust options), as recorded in the adjtime file. If the\n    adjtime file doesn\u0027t exist, the default is local time.\n\nThe speculation is that this could explain why timedatectl status\nreports unsynchronized. My understanding is that the kernel will call\nsync_cmos_clock every 11 minutes and if the RTC is out of sychronization\nby more than 16s, it will set STA_UNSYNC. This is then detected by\ntimedatectl, which displays it as unsynchronized. I\u0027m thinking if local\ntime differs from UTC, then STA_UNSYNC could be set on the first call of\nsync_cmos_clock before chronyc has set the time in UTC.\n\n[1] https://github.com/mrlesmithjr/ansible-chrony/blob/e1147d039e940fd85bf1529ef3d804ce8df21f06/templates/etc/chrony/chrony.conf.j2#L18C1-L18C9\n\nChange-Id: I963fd697b50694692ce4bcc91b2658ea90ac6cb2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52784f454c112e405482323ec44f6dad7a1435b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52784f454c112e405482323ec44f6dad7a1435b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
