)]}'
{"id":"openstack%2Fkayobe~986124","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I036d95cace1fe2083f669837c99e75e15e576032","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-30 14:46:27.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"hashtags":[],"change_id":"I036d95cace1fe2083f669837c99e75e15e576032","subject":"[CI] Use kolla-ansible configure-ephemeral role","status":"MERGED","created":"2026-04-24 16:05:14.000000000","updated":"2026-04-30 17:15:35.000000000","submitted":"2026-04-30 17:13:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986124","cherry_pick_of_change":985775,"cherry_pick_of_patch_set":5,"meta_rev_id":"93aedf667a25209567aee82a738c7c0adc9aa977","_number":986124,"virtual_id_number":986124,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-30 17:13:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-24 16:12:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-24 17:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-30 14:46:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 16:12:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-24 17:51:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-04-24 18:56:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5296d579a10695dfc000839cf0fb43b151826e56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-24 16:05:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e60820319b94db8d48242fc2b4e7775ab1bdcdd7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-24 16:12:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3ea5a7e04eb9b73ea212b09f6c7dfd53ab772b71","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-24 17:51:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c7cc4ab0e58281138bc34b7d9f79aec1a921a22f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 18:56:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/789c18c8b0f84d3d82453c05bd876e0f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb54a4c1d0d344e398a623082d1a8c9c : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd52542cd4548f3b3a427f9a8a0c724 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a07e50975ce24988bc73f60f349cc738 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad12aab92b99424188a8e186591e82e5 : SUCCESS in 9m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a79b094e2304bc1aea28b8b5840eb75 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c16d3e27670d4763b7e0bc92bfa597dc : SUCCESS in 12m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6a525a7b7cd64337b66bb88279f6a3ea : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7cfa2fe99b5b4494814541a96753df53 : SUCCESS in 14m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c20bbb3a576b4174b5dee9e87ab03623 : SUCCESS in 17m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d7c2e2440474300a414f20e66bf0ba0 : SUCCESS in 11m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/97d689f12c54487ea481d7f3a449d0ac : SUCCESS in 17m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6fa9aa1b6a3446be8c0e4de03924527f : SUCCESS in 22m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a46c59ae49346fcaba97aa1e2ec7c4c : SUCCESS in 17m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/957870fd998441a2bc6a740684672fa1 : SUCCESS in 1h 00m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6c95ea8b76ad4414a096db0d38d3933e : SUCCESS in 48m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1313ecf36176458bb2c8c7ee20d036f8 : SUCCESS in 1h 00m 49s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4840e87510294464abbb5167e3d7fed4 : SUCCESS in 59m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/60ca6f67ca9d46ffae0fff86ea7220b3 : SUCCESS in 1h 18m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3d0765164ce1448f89ad7527e02283ce : SUCCESS in 1h 12m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ccdd3eeff0840a5bfdf4da902643c2a : SUCCESS in 58m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5a13cf3d4c94af8aeca507513a3db0d : SUCCESS in 55m 56s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8fe80d0a59d543a690b9e5d6ce32d85e : FAILURE in 2h 45m 01s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/308afe95313b4a1cade8ff11e173ba3e : SUCCESS in 2h 03m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7a04be91328f48dfb9290d11aaeb3de8 : SUCCESS in 15m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b78b724e6ba34c3ea8db5e4de994b498 : SUCCESS in 21m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/905a4eb73a7b4a8598ea943edcc97261 : SUCCESS in 19m 44s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/d64968a6d47e4a889aa3b950133d91b9 : SUCCESS in 18m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2475d708f71e46329e4f63eb9f4fbb2f : SUCCESS in 13m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b80992bb970342c0baadf219f092bb34 : SUCCESS in 14m 49s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a150ed64771a434c9f83479313f01b03 : SUCCESS in 38m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/26c1cfac06b64e52a63c43494a9d1194 : SUCCESS in 12m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1def6ca7a9cc49949c1f6458c2fe5209 : SUCCESS in 16m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5596f00f39ad4991a965bc538d1e0410 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4f75abbbb92e9b8c58669754026b8098c22e6883","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-25 20:17:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d589ce68f1d3af1c816970f88a7b0e06270e3097","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 22:29:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ff9f275cb424f5fbf92cc0b23b26c8d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ade163bcb7d4422aab417f48a8d9a0db : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/157b893d62bc4b33959472c7428dbac9 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b90af471d5fb4f23952ce7c3946e95e7 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1239deb175284d0d9c717fad2f39d5ca : SUCCESS in 9m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a45b94e35d3424b9ce8e36b91a13eeb : SUCCESS in 7m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6874ed21ecaf4f0ca22cba7454d687b6 : SUCCESS in 5m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6a58353c8e304089853e15a3fcdd5606 : SUCCESS in 5m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fed9fae547aa494dad2d2056cd0560c5 : SUCCESS in 8m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c4db40a2fdfd40019662fcbb6ccbf7de : SUCCESS in 14m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/112c37497a6b4e80a47dbd9e1f1d4e76 : SUCCESS in 13m 11s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3a9864d3964b4980a7d6c4ce33f21e90 : SUCCESS in 15m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/aa76e5b8ca8541649304ea1e6f9ec508 : SUCCESS in 11m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f851d6ed3ed64b6d8d2aa23a16dbefbc : SUCCESS in 14m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c0cdbf6e7e3a423a9071378b41501e0b : SUCCESS in 52m 59s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6249684dc801466f809a925b250e7868 : SUCCESS in 52m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ea1e62eed4264d7cb1ddd827893d3ad5 : SUCCESS in 52m 03s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/8a150c08171f4f56b7d6ed93ab78bd8d : SUCCESS in 51m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2e1d59015dbc4cd292fd108d2b0f5afd : SUCCESS in 37m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6ca47edd85e04472900c980229fd2357 : SUCCESS in 33m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74ade2a91327403f8ab106c7fff63844 : SUCCESS in 55m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0b8ae71febc44371bb2a6693ff4ea27f : SUCCESS in 55m 18s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6f4d20ec83f44a40b96b85247cdda259 : SUCCESS in 1h 18m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3a6e0309d814d51876f012a0378e98e : SUCCESS in 2h 10m 40s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/225a9398cbbc44c89d9165d50acf0156 : SUCCESS in 15m 32s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/576f2c6cf1654efa95fdc02a418559fb : SUCCESS in 9m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e6c3e08f4c7f4865b936d146f36b976d : SUCCESS in 23m 23s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/d3b8def7f25644c581c4073fb0dcdd63 : SUCCESS in 9m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2888d01e0044676be0251b8d45a1967 : SUCCESS in 14m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/841e2fcfcee6451f986e4ff95e5e8ce2 : SUCCESS in 15m 04s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/fe2a0c6c9d5b4fd081dc22d175e08b9e : SUCCESS in 21m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e4dcd11a22f648e38f95dd3031596f27 : SUCCESS in 8m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e5a5834c76f04169b9ac75ab9ff23ce3 : SUCCESS in 7m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bea9c6a7ce0c499a857cfe9f798cc946 : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a2d0aa7455e2fd64dcf6ee38ec024d6e729ee256","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 14:46:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eda6df5661378322395c97371afa7dadf3187ada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 14:47:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d0fe57557a37e87ede2ea6cc8b68cea109fb3a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:13:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57f879aef01c4211949d883d00781f0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb9ee96e2ad4a93a37922eb82ff6c92 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57fceac0c1814c26b43ee57aa36902dd : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55afd43d3fe4bea96eac0578d6ffa45 : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76a07a8a87534ccab05170238b1a3711 : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ba67013ebba4177910dec13d3949c77 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/601c92149f434e158ff19364eaa490f1 : SUCCESS in 10m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/961d36b72dc54a7fa322c34686e09afa : SUCCESS in 18m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b23d8521d4f5409098005432dadaaaf9 : SUCCESS in 22m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a34365cd11fd4f86aa74275dfb538f76 : SUCCESS in 8m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2a282e8359eb40818c283324aae580cd : SUCCESS in 12m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1700da7abab84f028113930ae612d4cb : SUCCESS in 20m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27e212d73ea3497c91aaae1405994e0a : SUCCESS in 14m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1157745dd6d14a1ca16497bb66823a51 : SUCCESS in 1h 18m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b301d17eecf24c3b8644494bae34d8f7 : SUCCESS in 33m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5694c14477f44cadbaa977edbaa80f31 : SUCCESS in 32m 28s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/de93364322b444bfb0e00ec2c3cb3758 : SUCCESS in 33m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/eb3aa12df76f4358be889a9d52aa50f1 : SUCCESS in 35m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cffdd37965eb4ae18850a550e86cd517 : SUCCESS in 1h 18m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b7550570c25486b858489879a07e28a : SUCCESS in 59m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c0f453e4d0354b67b2dc7a13f89dbdce : SUCCESS in 55m 17s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/56fca891874b4fbc9e11d75788ce42ec : SUCCESS in 2h 20m 17s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4d1a9036b9d40b1b012799c145c5a10 : SUCCESS in 2h 01m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/717347750da34d4391ed050ecc4b4170 : SUCCESS in 23m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/dbba8b4b173149808e6a6e64229fdc1a : SUCCESS in 17m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6788d2a352f54cc8932cff928ffc2bdc : SUCCESS in 18m 09s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/bdb85ccc3bbd42f7b3974c5e3eacb47d : SUCCESS in 12m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7688308b5c8454f83323c3d04f070bb : SUCCESS in 17m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7bd961b36d0c44a9ba840ccc5e800088 : SUCCESS in 18m 02s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9ee30b524dbf4780a1ee33949d3530fe : SUCCESS in 35m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5510e071ce424c168d56f41691e53952 : SUCCESS in 18m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/447da23793fb4680b3979108ca0f34f6 : SUCCESS in 15m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97fe484056d24ffa88a6c20dce9a4868 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":1},{"id":"051f3f9c321c5fb5d9327c9c980c7def81355d55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:13:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"93aedf667a25209567aee82a738c7c0adc9aa977","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:15:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e48927af45542da959634ff9c840cbd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0400de71b404c99beaeb59e414adc0e : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae20d4ee6afd2722a95dcef7cb87869fa875a6a9","revisions":{"ae20d4ee6afd2722a95dcef7cb87869fa875a6a9":{"kind":"REWORK","_number":1,"created":"2026-04-24 16:05:14.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/24/986124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/986124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/986124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/986124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/986124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/986124/1"}}},"commit":{"parents":[{"commit":"8f05461c806613f1ae06a467dd926230f8c1efca","subject":"Merge \"[CI] Disable NTP checks in CI\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f05461c806613f1ae06a467dd926230f8c1efca"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-22 08:40:37.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:05:14.000000000","tz":0},"subject":"[CI] Use kolla-ansible configure-ephemeral role","message":"[CI] Use kolla-ansible configure-ephemeral role\n\nSometimes we land on a node with a 40G disk. We\u0027ve seen disk space\nrelated failures on the upgrade jobs.\n\nChange-Id: I036d95cace1fe2083f669837c99e75e15e576032\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae20d4ee6afd2722a95dcef7cb87869fa875a6a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae20d4ee6afd2722a95dcef7cb87869fa875a6a9"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"8f05461c806613f1ae06a467dd926230f8c1efca","theirs":"3d6afe204cb03552809402442bff9688cb8506e8","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
