)]}'
{"id":"openstack%2Fkayobe~986130","triplet_id":"openstack%2Fkayobe~master~I3f087d2cbfcb7b73284896d224841657ac80ab51","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-29 16:07:19.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"hashtags":[],"change_id":"I3f087d2cbfcb7b73284896d224841657ac80ab51","subject":"fixup! Skip external connectivity check when behind a proxy","status":"MERGED","created":"2026-04-24 16:15:47.000000000","updated":"2026-04-29 21:18:24.000000000","submitted":"2026-04-29 21:17:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986130","meta_rev_id":"be691f0c270b9f9e585a0352b5be58164069c4ad","_number":986130,"virtual_id_number":986130,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-29 21:17:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-04-29 08:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-04-24 18:13:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-29 08:44:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-27 16:25:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-24 17:58:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-24 18:13:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-29 08:44:53.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"f058001498a28f8df074db12056497c62b0059af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-24 16:15:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c45fbca515f21f4b4d149e5292883760d390c1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 17:58:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf53db7e6df045939f6b78db8624d9b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf6795eaeff402d9c69a193679234d9 : SUCCESS in 7m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb3a80446cbc48f78d1e55452ea0112b : SUCCESS in 5m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3073e5f8b275410f8fecd1aff9244ccd : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca73838a8b8d4461a6365bf216f6c32b : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/510761baf75a4fcb9ef8747c7ae1d9fa : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cf6579d6426c4c058edb0522aa946d1a : SUCCESS in 6m 48s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9bd1eb8cc0c546bf81ee2ccbe1c0316c : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3fa313cb018c4f4e85acd6f6bced88ee : SUCCESS in 6m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/61ce13b5270b4ee4b78a9c3729fc4f20 : SUCCESS in 11m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f0c6daa2f9a141fcaeacc00aa025ad84 : SUCCESS in 21m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22214b9231484c80a51129d553ee884d : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/306f9379290f4f4ba2caf7b18bbb9946 : SUCCESS in 11m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ebb2e4ee964452882985fdb94ef4e64 : SUCCESS in 17m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/95e37c5703e242fe94163b1ce740c09b : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5b64e4bfe58a4a99aed2d9e072335efd : SUCCESS in 1h 08m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e1ec4354e3874252899caafbe97c9143 : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/776e8c59b99d487ba29b014641d91006 : SUCCESS in 1h 02m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/05155b61839a4995a615b89413532bf1 : SUCCESS in 1h 06m 20s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0177fe53eb3a480f89ab277ee858c511 : SUCCESS in 1h 36m 46s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/52410e157c064a4e90f935be59f82586 : SUCCESS in 1h 36m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/129d90c913f34079ba7f33144b2fc577 : SUCCESS in 1h 29m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2176dff8897c48a2b14ee3976e2713f6 : SUCCESS in 1h 41m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a9b4efe135ed492c85545be80b1c54d8 : SUCCESS in 23m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a8dca8de8fce4b7aa2e13dee77ba9d58 : SUCCESS in 14m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f03df33549e434489e55d9939adb54f : SUCCESS in 13m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e6c713fb9fc04a968da97f7ee22ccdad : SUCCESS in 21m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8bfc8f956a4342ae8d953b43b696a019 : SUCCESS in 21m 50s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1d4f06ef93c44b789fdf02cc8aa5a5c8 : SUCCESS in 31m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b38cbd3b26c34a87b2c25881f572282a : SUCCESS in 28m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41a447c33ec24042aac74a4242fc939a : SUCCESS in 28m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8b2a2626c0d1407b84eb38150e897b5c : SUCCESS in 21m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bab2887e441b4c109b74153d6416e540 : SUCCESS in 21m 54s","accounts_in_message":[],"_revision_number":1},{"id":"896b88b0835577ffa2382c40b6dd5a611ebd60fb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-24 18:13:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"deb84a0994aebc66eff58cefaa624eb6cb665735","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:25:13.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"4725e331e29fd48fa8fd698e5a0bbf33f1682a56","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-04-29 08:44:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5da3b8fc3f7747453ea340966ac2d429810c211a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 08:45:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1a9295247b8a3dbdf32ef4d081085a5c9688441","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 10:51:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2390205a92f149589aa67db6202c823d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41cd33dd4ca34807bb6cedca12ce765f : SUCCESS in 10m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/93fb3f9940374262a1a86c8e1c4e11e5 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f98b9d6fe0f346dbb7a22d588f539a10 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/251ad75d65fc4cebb9aee6a01401e2b4 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4375d2bef79a48e59086fbb69ad44dd3 : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/abc1d48844c541f3b12a65db17d09828 : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b326d087b2ea4174b75979aba0374837 : SUCCESS in 11m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/83c6ae3130f14d3f839e926505cf50a8 : SUCCESS in 24m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2262c178e3d041bca01eecff44f5f124 : SUCCESS in 18m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/eea78406462c4e9e8709dceedf96bdbe : SUCCESS in 24m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7ec1743ea7840c586d989f08a963f74 : SUCCESS in 17m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6ea5e458882b476f9a08b224d60d7966 : SUCCESS in 1h 03m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c7ab4e2a9fd44f6999e59781a03f4e25 : SUCCESS in 42m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cf734fe97fa746599b186683322b0edd : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0ef657442834a399b70d7ee82b048b4 : SUCCESS in 1h 35m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c09e31170f584089af9361ab452dd261 : SUCCESS in 1h 25m 14s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7dbe7c40792e4a5dbcc5bd4d00a18048 : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d1f3fae7a30c433282d6401a0df57f10 : SUCCESS in 1h 34m 36s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc498dd3542344a3a7d414c360391e3c : SUCCESS in 1h 48m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38222938dc804bcfaa4d04b7b7f883e8 : SUCCESS in 1h 59m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/31147125558b4960ba00d60caf2b8422 : SUCCESS in 25m 30s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1ed71b647fb844d0b1782ae6e4355d8e : SUCCESS in 22m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/421b6badfc9744ca8717790fdb12a950 : SUCCESS in 22m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9c9a1816c125416c8cb1156431bceb77 : SUCCESS in 22m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/70e88ce5f7094c928a23f0bae6f17116 : FAILURE in 7m 50s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c3cbd4c764234b4cb00020bb2b765b0c : SUCCESS in 40m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0b3d1b99aef41c3976d5afbddc7ac3e : SUCCESS in 30m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/575a6eae696c4f8aa0f6263cc421bb9d : SUCCESS in 28m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6f7fe36ed06c4b62bceaef2a7608c2fe : SUCCESS in 19m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b42416efaf1448e8a106070117da904b : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f0f42451a6f0befc165d29a17c8a007d9681fb66","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 11:08:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80eddb373cab70ac9538701c6cd02586f29597f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 13:05:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86e14379aa8e4dc7972929c184c34a89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cd2b825822141698ee4f687e400ae68 : SUCCESS in 8m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9f76c910f60a410bb70bd89236117447 : SUCCESS in 6m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dad7065f8ad941099d6ed8acdf4878ce : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84fdc45eafe4450782415ae4d7300243 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86054967e2114788a96e046622b2193e : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81a5dd920d4f491885492397e5060e16 : SUCCESS in 5m 54s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d4a767bda074ab7af89456197846a42 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/564a0ce4bb654613aa253becd1d3ddbe : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78c512c846594b10b7fe3791633de030 : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7dde452df4cf4c6caeadd5169a0fcc1a : SUCCESS in 15m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e7d8032ca2c450b967fd800b19bac9c : SUCCESS in 17m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/86b3a908fd95486aba302f5132c99e87 : SUCCESS in 18m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77957cb6eaf4421c964a04a71ad3c9d6 : SUCCESS in 18m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/470e7db46829449a859ad957024b6222 : SUCCESS in 1h 10m 59s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b31d0bb1896e40d8b32e78824e65c6a3 : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d82db5f59ac046a29b28d039fee7c3a7 : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9c28b8a5a0c4270acf6a47a4aeaf46b : SUCCESS in 48m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7b4a536955b844e5abcd6aff7fb0c552 : SUCCESS in 1h 07m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ea952392548b4cc78df8efaa67831c7d : SUCCESS in 1h 39m 19s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/72da39881a17448cb77bd0a7c461327f : FAILURE in 5m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1524a89fd5243a4a8f31090198dc8d7 : FAILURE in 4m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/326fd3c0d9fa4f06a590b0b1e1db21a0 : FAILURE in 1h 51m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/06800905c5764ab3b2a847fda2b59236 : SUCCESS in 23m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/863e4e64a7dc4ea0bcfb0c98633b0380 : SUCCESS in 18m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2becb20cd0b4505b6a5492c87913aba : SUCCESS in 21m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/83b637d0b0d84f5ca8d37b7a887ef859 : SUCCESS in 13m 27s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/540247a9abe34f9b9245bcb9c385a371 : SUCCESS in 15m 18s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7ae7caaa93f54443be05097c2b450a99 : SUCCESS in 29m 01s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/032800904e5b4cfeb2c5bfb0f55a3dce : SUCCESS in 32m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc92195937bb4df08c5ff1a29b150a8b : SUCCESS in 16m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/896554c8cc514ff4a208d782b14b0c74 : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/228776a3241648cb8fd4afb0b2d44f80 : SUCCESS in 22m 24s","accounts_in_message":[],"_revision_number":1},{"id":"420c4a5fbb2e9b120c0e77d6466e00f935394482","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 13:07:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"839fa5dce1324e2403da8d8168c7837232f00812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 14:46:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b61fd8ca938340d0b0ff6d5dff28e20e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c45f3e87d88c465b800651dde2eb41df : SUCCESS in 7m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/71c6ece0f4a1494692e340ad993be7c2 : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/374e8b1827874d89a675768a2ed40edb : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd311d02a0e747c3933b51afbc8b7bdc : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9223cb2c3a834c50a6f34ccd1ac0619c : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/504e55c2897645b38d1c20b325a581bb : SUCCESS in 9m 28s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/71a410f486cb442baae715335cb9de2c : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/61b0dfbb224e4e83b111fb166f8bc64e : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/15ac3b8ee6564aae9eb2f99d9b719b5f : SUCCESS in 12m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/45d4f54853f84c24959ce245f635a759 : SUCCESS in 14m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcf519b8af114fe6bdf8ad35118a053e : SUCCESS in 13m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9ace746b24114dada572b958572b0415 : SUCCESS in 11m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b4eac6907d043ab907beb10937cc5a0 : SUCCESS in 18m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/653c1381d0f14b4689cdc8bc374c4111 : SUCCESS in 1h 02m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a1c67bd3e84e469985bbab02c2c375cc : SUCCESS in 1h 03m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/afc06fca813e49e4830ba62248fb36ea : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c9a6ff7105443a9a82b6b069feb7387 : SUCCESS in 1h 04m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b394634455e04d00a710b5c264642007 : SUCCESS in 1h 16m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f9fc6309590d411ba81b5a3702a1074b : SUCCESS in 1h 37m 02s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/269e9fba0c714be982a2959f08519f6a : SUCCESS in 1h 34m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cfbb9bfcf4b46a3811df879fd459263 : SUCCESS in 1h 30m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c9bade822de41b4a6cb5dcef608e946 : SUCCESS in 1h 34m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1b2f77ca5b614cb19a009b42194a3af6 : SUCCESS in 15m 54s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c05f2606ceec4057add447978862119e : SUCCESS in 15m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/279f3cc50f5a428394d0c2fc06c4c8fb : SUCCESS in 13m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c0f6ef8b0ae74adc8c100a5111100d1c : SUCCESS in 24m 37s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8c993271978344b6b1e883feeb692446 : SUCCESS in 28m 58s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/086ef40911f04fa4b22c824ea930c6bf : SUCCESS in 17m 37s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e9b87af04e047e4bdee318495b44b69 : POST_FAILURE in 7m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb52901c3f8642ceacf7dbc1a2b0f77c : FAILURE in 7m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a974ecd091d340ec8e526380b0d42ec5 : FAILURE in 5m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d50ec354b25e452486cf44ada580f49d : SUCCESS in 22m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e58da5c99db8e07b1ba638a024f11474c68c5355","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 16:07:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3678c9a82714257a098e1ce3f2d16eedc7255e8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:36:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b903f3f1c264b5c997c0eaf81816392\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c95123a02f7f4f989a542066838711b1 : SUCCESS in 10m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/10d61ca58db740dfb257c2f524084d6d : SUCCESS in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a604772fdb5941cfbe01666e2b2d671b : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24e6a3afdcd0492cad88d113be754425 : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87d77fc9d8034bbdaf1820d142c26534 : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c092bf2b4b3049488df109b060ad94cc : SUCCESS in 7m 12s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aacf094758494aa5b72beb1913dcb97a : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0fc81ab4290740c28612569f13cba4a5 : SUCCESS in 12m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/acc1276da2814a12948325c51fb7dabe : SUCCESS in 12m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3b6f3fa845d0413eb276a586d78affda : SUCCESS in 16m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65e8ce9127be4f639aa1343bef7a0c04 : SUCCESS in 12m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/30f8310219404b24a940122fe0505d8e : SUCCESS in 21m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c18587998c9b4fc18554d2524604bf0f : SUCCESS in 19m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c3766447a6954424990c44a7305115d7 : SUCCESS in 1h 08m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1977dc7a76ca460f8560338ee2b8f6b0 : SUCCESS in 1h 09m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/23616d4916e44f04b5ffc36cdfe2e11c : SUCCESS in 1h 08m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63cd89c25b2e45b28b44f367f3b9f879 : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ccc60ffb05b94b25bb2c9f46e0d293a8 : SUCCESS in 1h 01m 41s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3c440cb09ce4456ba315e78d0b941a0c : SUCCESS in 1h 40m 14s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bfb628c8ec6549a7a2041ed189bae497 : SUCCESS in 1h 34m 36s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9882167a33fd42f1ae46c59aa72bc197 : SUCCESS in 1h 30m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa7bffdfa1fa4df4b0fb56463b59273f : SUCCESS in 1h 36m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8335dfaba5154ea09e49fa8ad16dd498 : SUCCESS in 15m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/26721c57b754499f84a89e013f011488 : SUCCESS in 29m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b29f33b5e49d4b4989d94ec42a4353dd : SUCCESS in 18m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4af4e51cd8f7493192971e3adb421232 : SUCCESS in 13m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6e9dcfaba46040368036d5ffa1b4bfe4 : SUCCESS in 27m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/96be3ef67cfc4204aec8072d88f05c55 : SUCCESS in 29m 55s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7c4295d108d4ff499f456bd0e9bfd71 : SUCCESS in 31m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b752c8f88314bdca108e0fcd6855331 : SUCCESS in 27m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f80511e3aabe4a48b40b37c3b049c7ce : SUCCESS in 14m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3bfb1fb255144f59e143b106e893380 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a4c8f211e6c7d4d62e90d7c96e3a9e7d7e71adc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:36:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33be5f015825131a69c0b0454b784117d961c4ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 21:17:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce0d01c3671443ff88f8f7c8f922d227\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb6a38a880c4dea94ea424591acbf12 : SUCCESS in 6m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3c0b4d0ea58e49c18ca40470ae948d86 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1303756975294ce99b3b7160fc2071b8 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/248a2b87445445a8b6c7f2925aea65c5 : SUCCESS in 2m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fcf26b6568094c838f79d01589dd9789 : SUCCESS in 3m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/244f604c109340dc931b959aba80ff49 : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7583d2d093864927af4d2c33a5af003e : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cee0ce7cfe164dbc8355a400c645d038 : SUCCESS in 13m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e09d4dfed374a85a7eb485fe330b2ce : SUCCESS in 20m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/216508a8d5a74488903ca9996971f31b : SUCCESS in 12m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7d9ed7d955842f385998f16823d87a3 : SUCCESS in 11m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1df1465bb26b4cd78f536ea0674f8fb1 : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7e9d4d967bbb46a28ca9f102713afeed : SUCCESS in 1h 07m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8469ea6b37d84a1096feae7d8a15abd2 : SUCCESS in 1h 09m 08s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a0827d05b4a4e81a96747703840dd93 : SUCCESS in 1h 06m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c6458b4fc9d94d10b2a0591d23daef7d : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2857f4623d0443b4adf4769c5bcdd7d9 : SUCCESS in 1h 39m 33s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d63a7a8534f243a8a4dc57b00800952d : SUCCESS in 1h 28m 13s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99b1f017262d43c68b8140095bd52679 : SUCCESS in 1h 28m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ed09d3f0d1a461c9dd82eada27e8cbc : SUCCESS in 1h 39m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/760e5fe019d54d1ca0dbda4c4bd2ef8d : SUCCESS in 21m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3d062a5088c94b47b9827d5620cb39b7 : SUCCESS in 28m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be4d6e924b7e4f198992a7ec63faf59c : SUCCESS in 8m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e98f3b76a3114f978b88a8c2f2153bf8 : SUCCESS in 13m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f51558fde46047c19f67570252118c6a : SUCCESS in 16m 41s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/50da53e9cbc74c43930add87fab239f6 : SUCCESS in 29m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/556d4c8d6f8147e1bdc613b59da25925 : SUCCESS in 29m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a1bf1c4ff2246999200a57ebccfd5c9 : SUCCESS in 28m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/229bc67480104ab6a4d9ac4bfc3484fd : SUCCESS in 13m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0b77dd15f82436eb543cb1f391e5351 : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6539ab1c1da0c210c06050cffda0bfae587bb88c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 21:17:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"be691f0c270b9f9e585a0352b5be58164069c4ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 21:18:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c7c0465912f468ba7db5b34d4d2d570\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca8dd25e8286468a963648efb06de072 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66e4916344f8111a45681a53977c9d0ffb0d6fa9","revisions":{"66e4916344f8111a45681a53977c9d0ffb0d6fa9":{"kind":"REWORK","_number":1,"created":"2026-04-24 16:15:47.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/30/986130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/986130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/986130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/986130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/986130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/986130/1"}}},"commit":{"parents":[{"commit":"378ba9447cbbe9322a8f661f783d032229405c24","subject":"Merge \"[CI] Disable NTP checks in CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/378ba9447cbbe9322a8f661f783d032229405c24"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:14:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-24 16:15:41.000000000","tz":60},"subject":"fixup! Skip external connectivity check when behind a proxy","message":"fixup! Skip external connectivity check when behind a proxy\n\nChange-Id: I3f087d2cbfcb7b73284896d224841657ac80ab51\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66e4916344f8111a45681a53977c9d0ffb0d6fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66e4916344f8111a45681a53977c9d0ffb0d6fa9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
