)]}'
{"id":"openstack%2Fkayobe~986333","triplet_id":"openstack%2Fkayobe~master~Ida1219bd1aee73154e20a5bf0940dc72965c25dc","project":"openstack/kayobe","branch":"master","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-09-19 04:54:30.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-27 12:41:58.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ida1219bd1aee73154e20a5bf0940dc72965c25dc","subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","status":"NEW","created":"2026-04-27 08:25:23.000000000","updated":"2026-09-19 04:54:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c9507e6700646515d1f2f1d035c13a19da2b5b42","_number":986333,"virtual_id_number":986333,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-13 17:45:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-27 08:41:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-27 09:32:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-19 04:54:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"728cffd987c6349d9008c36c24ecedec8898d231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 08:25:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53e9a707d5545252c498f61f75330021d7c4a501","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-27 08:41:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5ae1b1c1607f8aa858de14a588fb26fe02f1267d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 09:32:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6230246b3cc14a8a855688ee88a91f27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ca5471e5b64e54ba82820d3832558a : SUCCESS in 9m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b8dbdd11bc7440318331e7e2df49dbbe : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cf9a4057b4248419fd71a742b2259f3 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2949116ccaa645918280024b019ce120 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e96a9e56ece94f2a9881db4b442a0958 : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cd8bf6676d244376b3076910af5a21f4 : SUCCESS in 6m 49s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d0442f0a122451d9f0d84d3e135947f : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03eef0d3186b4994b1250d6b8b24fc20 : FAILURE in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef0578c871fa4599bbececa278588c90 : SUCCESS in 10m 31s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/119b496090fa45b5b7a78e699f1cd053 : FAILURE in 10m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d404cbae43b4ad4a8641a7a0e128f7d : FAILURE in 17m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7c825546183342c5b582b636fc2df217 : FAILURE in 11m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c07ae8bc5a4c4ff7b80bb5c8e842a757 : FAILURE in 17m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/26fa071c044a4c0986c5c9d1a836d916 : FAILURE in 15m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3031e8358d6044c5b684926e96a04ffd : FAILURE in 15m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/db3789ba7724460aa7ae435ff94532b9 : FAILURE in 15m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dff8427765864316a1f245f9286d971e : FAILURE in 16m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2e5c0955857c4fb0895c6fac31fb34c4 : FAILURE in 14m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c3aef1d2411647bbbe71174635afa634 : FAILURE in 1h 04m 20s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9182115a79594f4aa424a5fa16284f62 : FAILURE in 1h 01m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c109dc1a99294bbba6395597b792c73f : FAILURE in 59m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f537adbf3db74bc195ab0c731b45d98b : FAILURE in 1h 03m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/80d3115419464bd184861daf6db6ccf9 : FAILURE in 21m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/10e4fde053af4a5698d9baf68cdd0587 : FAILURE in 16m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/078bb474a51c4d52809707ff46303bf1 : FAILURE in 17m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0c6f570583fc4274898dc9d283083b5a : FAILURE in 17m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/21fce982a4c34f0e82acfd2948c5cc1b : FAILURE in 45m 47s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f7bdc04f161a47e48c2c1aaaa75f2780 : FAILURE in 28m 34s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba6d6d97d9a441c88e232cae6e3a8279 : FAILURE in 36m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5542224e761d47e089c21a31f3cd3713 : FAILURE in 30m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/11de01e250844a15996e97903dd5dc71 : FAILURE in 15m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/845ddc4d978549c99a3610272e9d5910 : FAILURE in 16m 08s","accounts_in_message":[],"_revision_number":1},{"id":"60d962e950e6d4b9d8d643300e83c2004909f9a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 12:41:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bf64a89c3703f70c7fcf618935e84b1b2d79d0c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 14:25:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70cb7cccc2ae41108231512b6d22dfd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c1a20fdcac4ab1a17815bb3c901ecb : SUCCESS in 8m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/378c6d2088514d1dbb8abf165141642d : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/696661c5dc644589bf7af3d10fbd2678 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a528561d7b534c4ba056a44b1dad2cc4 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/28203e3be7144febbc64d70b29f89fb3 : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bfec67b61b844d4d86677abf5df50a9c : SUCCESS in 6m 53s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f5c91be8bf5144358f708e70453faa80 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d0640b68886948b2972a8633ba79ce2d : SUCCESS in 13m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fad9bbb69ec74355bf0866f97e2aeabe : SUCCESS in 11m 40s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1080288a79c846e5bf587dcbffa7b698 : FAILURE in 23m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc3f2d187e12419eabbefb861d9c9136 : FAILURE in 13m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2d0c8afe3cf14537a560410f091b65ed : FAILURE in 6m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c8c5ec35cc54dd09e3c28dec9d8fe98 : SUCCESS in 12m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e3b4b4c6f50b49a7af94828b72c0a3a1 : SUCCESS in 1h 04m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/08278d76db2546daabbbabc4006d8b54 : FAILURE in 8m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a3b399638f34474da3069e4b1d233140 : FAILURE in 15m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b10412df49284542bf168e7788b0edfd : SUCCESS in 1h 07m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/107d06aed55c4efea1a63fd480d1bae1 : SUCCESS in 1h 09m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cf65f978916b4559800f4458c96f62a2 : SUCCESS in 1h 38m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a9dcc7ec422e47f2a0ea17b713ecc289 : SUCCESS in 1h 33m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18da64fda38a480e9895062964f2ec7f : SUCCESS in 1h 33m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf405d20f88d4ba0a05791b15be4ac75 : SUCCESS in 1h 39m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2e8dc3e151324f6e8bb06b4b2bd3e56e : FAILURE in 16m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a4649e058f8c47eeabeedb2b59375f0e : FAILURE in 21m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26bdefc02bd84b3ca277679a3bb6e386 : FAILURE in 22m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e63e3c17d4e24fcb8e77f942a9c051b5 : FAILURE in 22m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6552b64b778642baad4f03d380370eab : FAILURE in 26m 44s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1f03eb0a2d814afe8c1c47556a4c8f6c : FAILURE in 37m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb8dd565024a40388adab6bf2333f075 : FAILURE in 30m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af1f4eeae146425aa265d0c4845f763d : FAILURE in 35m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/72ed33a4e01e4e6f8994fa1e543238bc : FAILURE in 17m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41f8bad8019946e0aef627a8b49f0e69 : FAILURE in 17m 43s","accounts_in_message":[],"_revision_number":2},{"id":"0970a142ccc8fea9bc3d6166510ecaaa6ebf1734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:02:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"4cab13b14842d342abfce48d7b94c71965f48cac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:30:39.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"019c545c2a69d969e13306d63bfa0f4ebd774b39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:32:05.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f4369476bf679767dfe3d38989693db7e3a66019","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 18:00:55.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"a3adbbaf4f9addb66c4864061644f58b565d8d13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 19:48:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f30ab0b996b4eb5ad917a7f611be6c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad99d4ba03cd4edaa5757679731475a1 : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f7b035183e2d45e08acc7a58903acb60 : FAILURE in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/706f972590a341b8a7ac1bb9396a4343 : FAILURE in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a05359e5afb746c0a42156943224f473 : FAILURE in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/911346b37dea467eb993f33286027aef : FAILURE in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd83a80174604453a5fa19ce526790c5 : FAILURE in 5m 28s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b067401512a84dcca855fb16cf3c342c : FAILURE in 3m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a821d7932aa49a18fab28f08ced604b : FAILURE in 3m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ec3483f5758b48da8f98f369e864ab37 : SUCCESS in 11m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/093024c62c0d4f99970504901e911df5 : SUCCESS in 14m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c77d7f28045141149aea00e60e818197 : SUCCESS in 19m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/476ef2befa1445ae8942f41c473eb193 : SUCCESS in 19m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6a6c9172d264bdfb40acaa7e0b50e22 : SUCCESS in 22m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b3b4b27c944c4b55b5654d1f23985d2b : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d5f403a71c3741c5937b2e73c5fe358f : SUCCESS in 1h 09m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/01f2dabf9ed04c21aed7fe2da7d84240 : SUCCESS in 1h 06m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77863c97377048e385de65722dffabce : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0882d37fc4d94fd8ae5b8a0eb7167d6c : SUCCESS in 1h 03m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d31fa07f4694474aa66402e1ec0e93f4 : SUCCESS in 1h 44m 30s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e240769efc7b49a1b4531c41a8516e79 : SUCCESS in 1h 35m 30s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/733ea30f918b477086b118ff679380bd : SUCCESS in 1h 30m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f66509ca5b624845b577784cd49d3acf : SUCCESS in 1h 44m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e5b011c70a1e4141ae5bff94e81158a5 : SUCCESS in 16m 22s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/46d127951f3e482abc5a5d1f79cbeaae : SUCCESS in 15m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f14c897705284e79ab88cf7e2441755c : SUCCESS in 21m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b4534fa92a554cf0b96a26d42c34dd9b : SUCCESS in 22m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4f5fe1593b624114a76c591bab505bbf : SUCCESS in 27m 56s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d7e37644f7cb401a8b538742fe3dc493 : SUCCESS in 27m 52s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3cd5e1ff4374294a08fe387e72b6ee0 : FAILURE in 35m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35f8ef79706e44539276af2cdbce27bb : SUCCESS in 28m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9e0499cae36341a581448acba91274e3 : SUCCESS in 21m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a6d10d265264890959f28d2a8b30940 : SUCCESS in 20m 07s","accounts_in_message":[],"_revision_number":6},{"id":"93ca5779710fb5f44b19589ab28d9e9ccee78fe5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-28 15:02:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b42ec4b1818a79294d3cb57000ed884328e74728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 16:44:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4750a901e96641bfbf2dcae11980b91d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798dc1a17af94386807bfc6199a20e8f : SUCCESS in 10m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b6eb32f33d1242c591ec296a079619f3 : SUCCESS in 5m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/265abcebec0d471187b374dec99d4c8e : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640336eeca7e44b695a48a90037c6740 : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8e5dfe787824ad5b1399c20c28dacb5 : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/501a00c6ab964c938c5e073cbc98078d : SUCCESS in 7m 10s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17b1c1c53dfd42bc971edc062465ca77 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3178d4ea0ad349a78f701214932c4f5e : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1151219116c0401396db77e9142ea078 : SUCCESS in 9m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a69f4a1502e64bd491dd56776479d922 : SUCCESS in 27m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1fd35a8210a4e6b9f1d3ee05f8d6ad6 : SUCCESS in 18m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2eadcb29471e47b1b0a70172d72d6614 : SUCCESS in 11m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/024baf8f26c745338ea9c43dedddfab2 : SUCCESS in 18m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/98b9ab1363d64d9a82090d2653a53a8b : SUCCESS in 1h 00m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/dd453ec1eb284e3aae2c7c70def2eaca : SUCCESS in 1h 02m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7a8907eb09aa46318753e060181d3041 : SUCCESS in 1h 05m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18458ddd9b054a6f83a514ebebe8d194 : SUCCESS in 1h 00m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4868d203238f409186c2bdb8fdd28228 : SUCCESS in 1h 06m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e1a5f0ffd7f340e9a3ec73fac4688d63 : SUCCESS in 1h 38m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0b7b52c6b5354faca77866c46c2b3110 : SUCCESS in 1h 29m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/158e3932c0bb4d50835717e6fbe15f77 : SUCCESS in 1h 32m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84e2168a054a4093b25a229b9f3e664c : SUCCESS in 1h 39m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0b95b932e85249a0927a0028622b7f59 : SUCCESS in 27m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/03f6d0244afe4c2ba308de313889557f : SUCCESS in 15m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/672007df665740a3a1747eaebf251007 : SUCCESS in 23m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c3b919f386bc479baf7418f460024119 : SUCCESS in 21m 25s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/79ba7dcb42824aa4a293dd7d3e2da937 : SUCCESS in 39m 27s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fcb177b2dba74511a677456f901550b0 : SUCCESS in 37m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92aab59938c54d298528470afda58645 : SUCCESS in 29m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8ce2fa7caba4de98c3bd24cd894e220 : SUCCESS in 28m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/32f81d84254e40348eadbe1aebe1e376 : SUCCESS in 22m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/424f3d729fbe477b81001141e06127dc : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":6},{"id":"c3327dfd25fe89fc8ad6b2ed96a98e48e3079681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 08:41:03.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"b38e41f19fdbdf456cb09db80d0ffb965f74489a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 08:41:43.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"0d379f5789ed566062fdf9bf06c129ffa6629f2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 10:47:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b12c49abebe36eb5e8512a33fde03d72836f6e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:39:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a58c121e82604ee38dff1939f01bcdc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/248c27cbbb7f4c7d8503e61f840f1cd4 : SUCCESS in 9m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6d08516946d540ce84043cb5fcc5539b : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20baf13cb42d4caeb69bb7fe8fd39f88 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeed9883d6a5475e8daa60ca670d16e9 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d02dca0ae3f45009321d9d0731f41cb : SUCCESS in 4m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c1074ade0bc494ab287355c838f3b10 : SUCCESS in 6m 46s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75fa4bebf32148c2aaeef8528cc0e445 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/74fe5a3ea0d9411fa65e9d2ba40ae22f : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f592e63deb7243bb8c0e647f23aefd60 : SUCCESS in 12m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/171890a5fb274c3785c966c9f46f5b1e : SUCCESS in 13m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/764145a12c3345acb49876966ea29e07 : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/db142bbb4f8844448b51a0c680041870 : SUCCESS in 12m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55de672cb1d5466196057fa117852c42 : SUCCESS in 20m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0dfdaad3c05f49e5b6ba3b3551a4bdfa : SUCCESS in 1h 09m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/63494ea63ede487f8a29f4a7a182db6f : SUCCESS in 1h 15m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/97a28cc0d47947b19b442985a47cf972 : SUCCESS in 1h 16m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70a4e141891a433faab4677d4ab5591e : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/101702d520c14b449ecfa1a77a737817 : SUCCESS in 1h 09m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ea523bc99c7c47289545f199ac8374aa : SUCCESS in 1h 39m 38s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c268a77d8257486fb327e1b84bf3e696 : SUCCESS in 1h 34m 23s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aa7c4d6b0c547f0a82e03a67e1774f6 : SUCCESS in 1h 35m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4fb97fe7a8e481eaff45fff65152f05 : SUCCESS in 1h 46m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ccdc550cbe13471a93a755220213455a : SUCCESS in 25m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/92916636130b4eb9bf7cebe30a3e3dc1 : SUCCESS in 15m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0f74165afac47ccb00c5505443aa06a : SUCCESS in 22m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/62d26b8abf744ac0b7d2ac78623d746f : SUCCESS in 22m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/af23de0aa4c845c48e61196d3bd404d8 : SUCCESS in 24m 42s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f59f3b5bee34408f9df2779d20259f50 : SUCCESS in 39m 29s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58e2a22464b04e4295739abb813ece5b : SUCCESS in 28m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2aef87845254a29b5cf2440761b7552 : SUCCESS in 29m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1e2427258a374593aeeb9e28d06a67c1 : SUCCESS in 15m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f6eba49c8614b5ba55e470f51f197d4 : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":9},{"id":"58042731e74228531cc2ac482c3932453c75ad46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 11:56:53.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"754a62fc797165cd71445929e4e75e938eba29c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 13:43:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ea3c1efa2ed4a4db112626eab070088\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c12a51aabb842cdadfc8c71e0a40646 : SUCCESS in 9m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/95e782a6497b497dac8fbe0dd4118be5 : SUCCESS in 7m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bb03aadded248519325dcfe1edee879 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48442e1a3921491b90c8a4c144137da4 : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdd8acdc75a245eea11508c0dce6e603 : SUCCESS in 5m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c58ae624278b409bb2039bc83251048d : SUCCESS in 6m 54s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0623828fc2aa49988f9de4e6241e830a : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/37f3c9379fc74938b3ac49065a922b5d : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/95c80bb66abc4f0d823a0f287464f256 : SUCCESS in 11m 32s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/43a489e4537c497da6940832a49edc7b : SUCCESS in 20m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0aa1a79c2a7949538908f45fd8e0da69 : SUCCESS in 20m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cf201079c59e4ecb9b7621bda7ed530f : SUCCESS in 19m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a7dc6936b0d4b568b1a82fef82a61ad : SUCCESS in 18m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9e658c9cd58f4e66b95a74b450e71b48 : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8c12cf135fad4dc4aeed09da3273051e : SUCCESS in 1h 06m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8cb11e1c511c4e21bcd7b48d8fda13cb : SUCCESS in 1h 11m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1fed183d3cd4fada4d463758413076c : SUCCESS in 1h 06m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e738cca8b6f74e9da26011cee6551c17 : SUCCESS in 1h 05m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0837a97aba874cc78cb686ed63de91c4 : SUCCESS in 1h 42m 36s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/800d3af2856b457c894a20ebe1ec8018 : SUCCESS in 1h 32m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c47bc816423442590abc2d00ef3c71a : SUCCESS in 1h 41m 20s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5d127496eba4b379f58544e76b7e412 : SUCCESS in 1h 39m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e0446d455c404ea683e15fd9a9da4764 : SUCCESS in 16m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/129b0dd985e3407c91a3854f4d2e75f2 : SUCCESS in 16m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88102ba08f474ecdaa49336f397b0b6e : SUCCESS in 19m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/26ce6e57844d4f09a63f49ad9817ee69 : SUCCESS in 20m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2eb4c1056e6c413ca71135f71e774c75 : SUCCESS in 25m 32s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4d135d9dc2c243e198283fc142592452 : SUCCESS in 17m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3518e9ce5e94370a92dc4aca26d6554 : SUCCESS in 30m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10a36a503cd3416c96cca0b13ce24995 : SUCCESS in 28m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/539acd1fd99f4624965eee7f712f99eb : SUCCESS in 16m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b52a5aadec2406989dd13f21464a86a : SUCCESS in 20m 57s","accounts_in_message":[],"_revision_number":10},{"id":"634c5eba91f4b2228f8b217bf8b111fd2af67f3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-13 16:04:52.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"eaf56e2cc3917be591c19b1ddbf0034d0f6137b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 17:45:46.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31acc304e1cd4d999f052c6748f7c91f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ce783162744d00901be1b6209654ab : SUCCESS in 5m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fba8b48dc81e4dac8342042f7b64f655 : SUCCESS in 7m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a21a949f8c0b4a0b9670911cdc49e923 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adf97b7d31994e3abc1a2a98584fb103 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6143cfe190a48f8baea0f514f1800a0 : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9350b1efd4824e33bc1bc3a51de24cba : SUCCESS in 6m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1bb158017da4b8a85190e395f41f8db : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0addcb58d67f4a0f98d3f86ec2134097 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/99b84aaa75af4130b2041156a95c9349 : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a5871fbd97d42be87cabd67274d9fce : SUCCESS in 13m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/28aceb866ae9475c807238d126b27634 : SUCCESS in 17m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44bd696824f940e8a6024f0724ea17a7 : SUCCESS in 18m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f3c06adec7d843feab0c5053a5454382 : SUCCESS in 12m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd963cf2cb094bc9a1a5f26cd42a0e2d : SUCCESS in 18m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/330a35fbd3aa48d39543dd10de8a20f1 : SUCCESS in 1h 04m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/19ce0fe2e2b049909031beba698c4275 : SUCCESS in 1h 04m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cf09c40146464411a1f8af344e239aa5 : SUCCESS in 1h 04m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0d7496fa2624e81970b5d9727d71ec9 : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b6cd0fd40c784a21901f9791d150fb48 : SUCCESS in 1h 09m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fe4db57d3cd24567b64849a4f7363382 : SUCCESS in 1h 39m 16s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ed81b7e889d44bbe9b78f24cf0d9fcd0 : SUCCESS in 1h 30m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba495bb9c4cd4c63bd0ecd78b6506fe6 : SUCCESS in 1h 34m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e71ee6bc1d034fed9e251e0863f88831 : SUCCESS in 1h 38m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1d35848deb034dddb506caf98175b41b : SUCCESS in 21m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0fc191564c1e469592e76b31cb4ec83e : SUCCESS in 19m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c061d9b9e8c64a88b32d31435ad3306c : SUCCESS in 26m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0bf54f86ba1c498fa5afb32da197348b : SUCCESS in 17m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/875c33d9f4bf4a7bbe1dc098e385f853 : SUCCESS in 21m 36s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7d8c008bc5614380824757bf989f03fe : SUCCESS in 23m 44s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8161ce31f55e40c3b01e4d67ca011b16 : SUCCESS in 32m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99e9bed66d04445685121cdb8829272a : SUCCESS in 28m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b5aeb96a90704c7ea0d5a0bfcd914b9c : SUCCESS in 18m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d5e2d1eca644440bb45fe6c644e4f6b : SUCCESS in 22m 40s","accounts_in_message":[],"_revision_number":11},{"id":"c9507e6700646515d1f2f1d035c13a19da2b5b42","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-09-19 04:54:30.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"38eb8c98aa2e193fd3c59b37827a003921aa73d9","revisions":{"d3e4c9389d319832e83771a9186380dc0d90b205":{"kind":"REWORK","_number":1,"created":"2026-04-27 08:25:23.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/1"}}},"commit":{"parents":[{"commit":"6ed5357516e4a38a9d55747e64982ce241f6686e","subject":"Fix reserved variable names \u0027name\u0027 and \u0027port\u0027 in serial console playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ed5357516e4a38a9d55747e64982ce241f6686e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:25:15.000000000","tz":60},"subject":"DNM: Remove ANSIBLE_ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ANSIBLE_ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed.  Remove the temporary\nworkaround environment variable and update the unit tests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3e4c9389d319832e83771a9186380dc0d90b205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3e4c9389d319832e83771a9186380dc0d90b205"}]},"branch":"refs/heads/master"},"8d567ff9545a46c15fd5a0c834e686b055c29603":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2026-04-27 12:41:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/2"}}},"commit":{"parents":[{"commit":"ba2d2b25afc86a722269aa6e242de376659ab080","subject":"Avoid deprecated ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba2d2b25afc86a722269aa6e242de376659ab080"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 12:40:30.000000000","tz":60},"subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed. Remove the\ntemporary workaround environment variable and update the unit\ntests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d567ff9545a46c15fd5a0c834e686b055c29603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d567ff9545a46c15fd5a0c834e686b055c29603"}]},"branch":"refs/heads/master"},"f52baf45c8f8cd00a62b91accbe9ce3baf5059bf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-27 16:02:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/3"}}},"commit":{"parents":[{"commit":"443957985c00dbcce3242581adab0d71757ae634","subject":"Fix non-boolean nmstate zone map conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/443957985c00dbcce3242581adab0d71757ae634"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 16:02:17.000000000","tz":60},"subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed. Remove the\ntemporary workaround environment variable and update the unit\ntests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f52baf45c8f8cd00a62b91accbe9ce3baf5059bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f52baf45c8f8cd00a62b91accbe9ce3baf5059bf"}]},"branch":"refs/heads/master"},"bf5ef0be714e19ce652af3f68373cb146e0c76f5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-04-27 16:30:39.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/4"}}},"commit":{"parents":[{"commit":"a247235c4521e271c98f491f8eb0f8f10e202d3d","subject":"Fix non-boolean nmstate zone map conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a247235c4521e271c98f491f8eb0f8f10e202d3d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 16:30:29.000000000","tz":60},"subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed. Remove the\ntemporary workaround environment variable and update the unit\ntests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf5ef0be714e19ce652af3f68373cb146e0c76f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf5ef0be714e19ce652af3f68373cb146e0c76f5"}]},"branch":"refs/heads/master"},"63faae679277bcda7e43a41b34b076d374de10fb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-04-27 16:32:05.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/5"}}},"commit":{"parents":[{"commit":"e6f475b5eb890377f66bd791b6ba3237b3bf40b7","subject":"Fix non-boolean nmstate zone map conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6f475b5eb890377f66bd791b6ba3237b3bf40b7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 16:31:56.000000000","tz":60},"subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed. Remove the\ntemporary workaround environment variable and update the unit\ntests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/63faae679277bcda7e43a41b34b076d374de10fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/63faae679277bcda7e43a41b34b076d374de10fb"}]},"branch":"refs/heads/master"},"deb18494fe8a146c4698ffc118f765cb085daf4b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-27 18:00:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/6"}}},"commit":{"parents":[{"commit":"17c9c5d6d34e9a67d92027883646e41c7b0215a9","subject":"Fix non-boolean nmstate zone map conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17c9c5d6d34e9a67d92027883646e41c7b0215a9"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 17:59:14.000000000","tz":60},"subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed. Remove the\ntemporary workaround environment variable and update the unit\ntests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/deb18494fe8a146c4698ffc118f765cb085daf4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/deb18494fe8a146c4698ffc118f765cb085daf4b"}]},"branch":"refs/heads/master"},"e42581722c4c11b3a08b3aa80dea2d103c38ef7a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-29 08:41:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/7"}}},"commit":{"parents":[{"commit":"13095e290e0f9183e8b1f2d7e9ce37bcf96ee19f","subject":"Fix non-boolean nmstate zone map conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13095e290e0f9183e8b1f2d7e9ce37bcf96ee19f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 08:41:03.000000000","tz":0},"subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed. Remove the\ntemporary workaround environment variable and update the unit\ntests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e42581722c4c11b3a08b3aa80dea2d103c38ef7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e42581722c4c11b3a08b3aa80dea2d103c38ef7a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"deb18494fe8a146c4698ffc118f765cb085daf4b","theirs":"13095e290e0f9183e8b1f2d7e9ce37bcf96ee19f","contains_conflicts":false}},"2660c8dbe91ee963a7fd5b1ec0259104f94ca9b7":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-04-29 08:41:43.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/8"}}},"commit":{"parents":[{"commit":"25fb848d1bbe0f0e3b1ea7456ea64e2f0ceb95eb","subject":"Fix non-boolean nmstate zone map conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25fb848d1bbe0f0e3b1ea7456ea64e2f0ceb95eb"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 08:41:43.000000000","tz":0},"subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed. Remove the\ntemporary workaround environment variable and update the unit\ntests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2660c8dbe91ee963a7fd5b1ec0259104f94ca9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2660c8dbe91ee963a7fd5b1ec0259104f94ca9b7"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"e42581722c4c11b3a08b3aa80dea2d103c38ef7a","theirs":"25fb848d1bbe0f0e3b1ea7456ea64e2f0ceb95eb","contains_conflicts":false}},"03957dcdd473f22951ef639cf6530fc83ab59c3c":{"kind":"REWORK","_number":9,"created":"2026-04-29 10:47:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/9"}}},"commit":{"parents":[{"commit":"0c8ea8323c79607f12b2d029902c7a1a694efa65","subject":"Add release note for broken conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c8ea8323c79607f12b2d029902c7a1a694efa65"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 10:47:46.000000000","tz":60},"subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed. Remove the\ntemporary workaround environment variable and update the unit\ntests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03957dcdd473f22951ef639cf6530fc83ab59c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03957dcdd473f22951ef639cf6530fc83ab59c3c"}]},"branch":"refs/heads/master"},"80e08064202f27059fe78d521a39d012d52eff9c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-05-11 11:56:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/10"}}},"commit":{"parents":[{"commit":"167e60dd27d71fa0956a66f63787320dd08ce34f","subject":"Add release note for broken conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/167e60dd27d71fa0956a66f63787320dd08ce34f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-11 11:56:53.000000000","tz":0},"subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed. Remove the\ntemporary workaround environment variable and update the unit\ntests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80e08064202f27059fe78d521a39d012d52eff9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80e08064202f27059fe78d521a39d012d52eff9c"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"03957dcdd473f22951ef639cf6530fc83ab59c3c","theirs":"167e60dd27d71fa0956a66f63787320dd08ce34f","contains_conflicts":false}},"38eb8c98aa2e193fd3c59b37827a003921aa73d9":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-05-13 16:04:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/986333/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/986333/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/986333/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/986333/11"}}},"commit":{"parents":[{"commit":"5008cc61da76ccdffd5a61fe22f53082ee581372","subject":"Add release note for broken conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5008cc61da76ccdffd5a61fe22f53082ee581372"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-13 16:04:52.000000000","tz":0},"subject":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround","message":"DNM: Remove ALLOW_BROKEN_CONDITIONALS workaround\n\nThe broken conditionals in kayobe have been fixed. Remove the\ntemporary workaround environment variable and update the unit\ntests accordingly.\n\nFIXME: wait for 2026.2 to give time for people to update their\ncustom playbooks. Add reno.\n\nChange-Id: Ida1219bd1aee73154e20a5bf0940dc72965c25dc\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38eb8c98aa2e193fd3c59b37827a003921aa73d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38eb8c98aa2e193fd3c59b37827a003921aa73d9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"80e08064202f27059fe78d521a39d012d52eff9c","theirs":"5008cc61da76ccdffd5a61fe22f53082ee581372","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
