)]}'
{"id":"openstack%2Fkayobe~986370","triplet_id":"openstack%2Fkayobe~master~Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-13 18:55:16.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-05-13 15:51:03.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-05-13 18:55:16.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-11 15:06:11.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1","subject":"Fix string-typed conditional deprecation warnings","status":"MERGED","created":"2026-04-27 12:41:58.000000000","updated":"2026-05-13 18:56:19.000000000","submitted":"2026-05-13 18:55:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986370","meta_rev_id":"c44fef05533673dd68069be3c8fbc819c9ea849a","_number":986370,"virtual_id_number":986370,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-13 18:55:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-05-11 15:38:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2026-05-13 15:51:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-11 15:06:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-11 15:38:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-13 15:51:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-11 15:04:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-27 14:27:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-29 08:07:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-04-29 18:53:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-07 08:20:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-07 08:20:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-05-07 08:20:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-07 08:20:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-05-07 08:20:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-05-07 08:20:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-05-07 08:20:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"26e11e2f747d79354bd4e5624e858d560e56b67d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 12:41:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"466482fa8d91bdd4a43c3b7858b35dffc303d6c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 14:27:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e2d8bc847bc4577951cd30d250f6b57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221243e9f69b4cbb8fbda822b3f1d7a2 : SUCCESS in 10m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5452dcd45c042c9aabe5915ed6324e7 : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/207780157bed4a259c9791158b85cbfc : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfae7c0da13b426d85b82c1b4674e989 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/306ee6aa17e74c2c85bfe9f2118dc31d : SUCCESS in 5m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0d637fa3231947f8ad8a2e1247676ab3 : SUCCESS in 7m 02s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc3a7cad4f5940bd9ac73f4f2ad3f4dc : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5de4334e05bd49d4b7cbfb91421b45dd : SUCCESS in 6m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bd6325af3a174f248847ff0af248d45c : SUCCESS in 12m 16s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f20bbf209ea4f7c8f1434478802cd00 : SUCCESS in 13m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e85d56530644680b88f542237f4aae0 : SUCCESS in 22m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e35aa5d8b0414d0c8423ba85493cdfbc : SUCCESS in 24m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1234dd7de0ca44d8894ba1f41cb9ac7f : SUCCESS in 19m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e6dae58aad004d22ae17036f9eccfa0a : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6ea9c34cb2a74a19ba6f3583de2baac1 : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/df3aeeff6a2943d1a2dc52cef22284d8 : SUCCESS in 1h 11m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2619d94eda7d46fdb91ea8cb784ce018 : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/62e352ca65b74f19bd6f0727f0178627 : SUCCESS in 1h 04m 57s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c375a77be5ac4735bd423e3815dd3237 : SUCCESS in 1h 40m 38s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/321707b195c644c0be27b978360e3b9c : SUCCESS in 1h 33m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbfd65a6371443518feda74fd06c0301 : SUCCESS in 1h 32m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5aae653df0d4585902ab3108770bfb1 : SUCCESS in 1h 41m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6cdf0979a421482bb52dc90bbd66c2fb : SUCCESS in 28m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e8e4208051d4ae2a39e90c88b8047ae : SUCCESS in 21m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43a536a84a3a402099ad933141174994 : SUCCESS in 14m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2788adefd9f84ced80067737538a9f70 : SUCCESS in 21m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b5e7ec8736d24fedb75e3826a562cadf : SUCCESS in 35m 26s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9fbaf178a54a41c2ba268f643def9a04 : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6db375221e4c4f3594db8b0b488bdd3b : SUCCESS in 18m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ca2268c772342b2a4f7570810558b0e : SUCCESS in 34m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a829ad2641a94928bec9105e9eab5579 : SUCCESS in 15m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/072283eea7684130a34b02eb64dfb23e : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c6373af190d4f10382b22eaa636471ed3f10fd7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:02:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7e34a9b5ad6fa5ad7186dbffb4713455ff0678e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:30:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"433d9701699608d6b8f7159ff213f062b8eeedcd","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:30:49.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"9bcbb854bbc866910be346ea1c9c417aadbb253f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:33:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ad5277873d1316508df6f30e7ae319589f982f5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 18:00:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6108cfe415d773ef89f0d50863a78283a307daef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 19:46:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ded4b9b0de24ba49c6426ef9e5bc90f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe4d86443f644e8a8ab4e26611271cd : SUCCESS in 10m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/779e783a49b64b4bb8e34e4856dcf40c : FAILURE in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6df8dfb696042c9948a98d30924a9f5 : FAILURE in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6cfd1f688684b679d81b514f581dbc7 : FAILURE in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6244ef436b334af384a328d95e179f10 : FAILURE in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6188ed6c053546e9929e0aeea4e40370 : FAILURE in 5m 35s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/390eb37a490a486482a7b9fbb5928c9f : FAILURE in 2m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0db43021162948598e5b3324a021f045 : FAILURE in 2m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ac90e3c4dd04ab88f2c99d548d03805 : SUCCESS in 6m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d62b1ec3928c419db4bb4afc1d818810 : SUCCESS in 13m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/595ceda0f7c0433493189878ba8a8c20 : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c0d656cfad62483281f0245bbcc95027 : SUCCESS in 10m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2da9741e0e564d5b8a70c7db2ccb3269 : SUCCESS in 14m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0481d90f853a4d8ea885f2924921e538 : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7f8a358087374979a107596af144cb18 : SUCCESS in 1h 08m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/973b09f5bee2482c992dac0ad581c142 : SUCCESS in 1h 09m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/985cf5b83b4e4f50a1145738aeb42f4e : SUCCESS in 1h 06m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d40a222bbab4dbb9b2b9f40347dc170 : SUCCESS in 1h 07m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/63d960eed7f548c0af1b219d62890e0b : SUCCESS in 1h 37m 43s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/55208a9cfa5f4c6f828580ecf8bc4a0a : SUCCESS in 1h 36m 29s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8506c649fe0b418b8c61e1d97ba072f5 : SUCCESS in 1h 31m 37s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/487e4d532f78403ca4b4b9bc564c7865 : SUCCESS in 1h 42m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d625ca93ea474a4c8f20b2f01288b65b : SUCCESS in 19m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/86968aa6c3a140f4a21433270e4bf6ac : SUCCESS in 14m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07b27697c6ee4333b9ff9724e93e8a59 : SUCCESS in 26m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1995d50c4a946b2af348b829f4d562e : SUCCESS in 20m 58s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2e1b65b9dec340bf9a897e58d25e6ebe : SUCCESS in 17m 09s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/41d3e3ab1d624ab0b396541387112101 : SUCCESS in 19m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d39c8b13399b47af99a5b26c15194de0 : SUCCESS in 16m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b88c10ac65dc4e7d83cbee051a870529 : SUCCESS in 28m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/05382f769419438cbff537a7506cd69e : SUCCESS in 17m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92ea46348b494f95b011dc635e438d3d : SUCCESS in 21m 26s","accounts_in_message":[],"_revision_number":4},{"id":"d807b05e0e94ed77e1bb1fb2d39d8c418b440ea6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-28 22:36:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a7e523cd0b6dc434e7cd3e0488863df665c654cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 00:17:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e92f9e9e24e48ce93b66e7dd108ffe0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60bd5d5e44684958a92e39c0c41f1e3f : SUCCESS in 9m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/530f287caa7e48e6845eb758533d11d3 : SUCCESS in 7m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ace5e9dcbb34348bd04a57a5dbeb4a3 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25687ddad27f4332baaa8d76e3f4feee : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bd6d1c28cf8412e8cd4742f9c021b66 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/89d3773494734e3abdaa889bda1aff49 : SUCCESS in 6m 25s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/359aae0512cd4b7a8240307039a1a5f3 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ef3d6ab87e424395a42451f161cb91d7 : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d625aac2bec7493894396d79b35f6c6e : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e49e5d347de14f34ade3d81299f2d3d5 : SUCCESS in 16m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fc5342e85b149e9b5b06ca1463a3ac0 : SUCCESS in 13m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/16ff3e683edc42caa897455562b1b6fd : SUCCESS in 24m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dab308f7814a46ebb3165615cfbc65a4 : SUCCESS in 15m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c315dab53ac34001adc449b08f2692a0 : SUCCESS in 1h 03m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f51e5fd93cb347b393090822f7668fd8 : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2f323ab191254b2cb2bfdfdc2162b23b : SUCCESS in 1h 06m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6c9c426836145bd85646754ce225a30 : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8a91e45196bf4694a578836989c31c8a : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/04c0497b0d894601b5ff9f760166f7fc : SUCCESS in 1h 38m 34s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/42c098c67f2d424aa0bc22971c379fb6 : SUCCESS in 1h 29m 04s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bf504a0736b4680b629d0dfe85ea0ed : SUCCESS in 1h 32m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aff16278428d4792862131b11aec3105 : SUCCESS in 1h 36m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1422d921b38241f1936c202295beb6e9 : SUCCESS in 16m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6d7b59a417b746e2bb541c0276aca325 : SUCCESS in 26m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f13b0d17444409392a11490b1d357bc : SUCCESS in 23m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5146110f820f45afbad96d2baf13f65c : SUCCESS in 19m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d664c7551af8416382a5dcfcb892d988 : SUCCESS in 34m 49s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e7640b2aebf7437dbe4a285836d233d8 : SUCCESS in 36m 30s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b960aae613a14d2c901881cc71b4990d : SUCCESS in 35m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06ff1d49ba12446fab4975197261a5ab : SUCCESS in 33m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3b9850fb2ce54107a6b9e57900ff6291 : SUCCESS in 20m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba405de0b1f2445399b67a174d490f9b : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":4},{"id":"db824c632d211cb4864f789f1b896152bbeb78de","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-04-29 08:07:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6e26e68179f0cebf44bcf1411df412e01ff0622","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 08:40:47.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"cd783099e98c13ee334eace716cd7f8f09b9f0c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 08:41:03.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"935cf8f682d600afd53dcbe0bb260bfc49e18046","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 08:41:19.000000000","message":"Patch Set 6:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"1d62f83b744f9f01f270ad0594ca3528625db29b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 08:41:36.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"7be88ae56636cfcc097b907ec4d01ee96ef339cc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 08:42:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8fc9ff68d97dc8efaacbf40fe97b04f064b72957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 10:21:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd13b5c5f49b4188a294ef7cac32163e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e141bbbbdf6a446ab55a929308b9e091 : SUCCESS in 9m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/048a802507b541c7baa23afda4624431 : SUCCESS in 7m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92eb6b92c5674c4ea0a044ec31d0d9e6 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56472b835c0c44b491f79fffc689a6b8 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a50ba02ec3384d54bb5d2b36775edcee : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5a166f1fd4cc438a99aa9ea8164136cd : SUCCESS in 6m 39s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29f6c2d6829b43d0a4b34c77b68818c2 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71886b06b54741cdb4bb48c75ac1d050 : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/47272b92fb144f40b437190818c4fcb1 : SUCCESS in 11m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/69493afd6b6b4bff9534dc537ad05c85 : SUCCESS in 19m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f7234b1401a4e93b2302a89fbfb564c : SUCCESS in 18m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9ed4be72015c4fb5967d2c1238e2a820 : SUCCESS in 25m 02s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/198cfbec260147598490f8be48fcef6e : SUCCESS in 16m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/db343db53cb042cc9b01bf577c1643d1 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fee80d859035442884a466bfad212cae : SUCCESS in 1h 09m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a6506ee1dfb0449d961264ece6a29fa0 : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d6b4f5be432485088adfc66f1661c7a : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7533d2a0215c4eb28d53cb3abd79d680 : SUCCESS in 1h 00m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9fae346f13d04c83a4163d433e6e1f3e : SUCCESS in 1h 37m 57s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4474d65d4ddc4877a7d7cfc6fb3bd78b : SUCCESS in 1h 32m 44s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84cd557fae5b430cbd9db3da7e2956f4 : SUCCESS in 1h 29m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78f6cb757bdd4c508160c8ad47c5687c : SUCCESS in 1h 36m 02s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2f248ff93c5a4a3db64c810161cf33ef : SUCCESS in 22m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f4c99288b917452eacca668a1c41563a : SUCCESS in 27m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5c9b29a25f24b03aeba84b274eb0e89 : SUCCESS in 27m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/205828fdf41d4b3ea3edc6f28b53efdb : SUCCESS in 23m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8a72bb87cb444a81a2621112b6f65633 : SUCCESS in 34m 28s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/09629774f44c41b7a60c91a5a2b20acf : SUCCESS in 38m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a40c4b8703fe4a9c8c9439089d544146 : SUCCESS in 28m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81b89b01126a4e3589947056d1eaece2 : SUCCESS in 27m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/36e0d417bb4949868d150423ad7c2805 : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fde58fd875c4e299bc24a96a4d24df8 : SUCCESS in 19m 52s","accounts_in_message":[],"_revision_number":7},{"id":"9aec4d7b942ae70c199a12714c896d63b250adc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 10:47:52.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"d2f76f3e5c06c5c99e1d2919727acf0bb572eb5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:30:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cd771551bae441ea40d610171c60d00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e086209ce49a4fddaa4c34cdc53e8f86 : SUCCESS in 9m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d35ec452e13248a884f97e3d575a532b : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bfe5f14efb74cb1920e2fb4d8f39fc7 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/103cc539ddc048aca22214c59b847a3b : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a39fbe305b01466083f9318cdae7db91 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/32299be8286f4c2c9e09809cdb98bf54 : SUCCESS in 10m 04s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6269eb650c9e4c74b4e09b5d424dcc28 : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/63d272a6baae400ca6d27fb07ba1b2e5 : SUCCESS in 12m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/474362a37373411b8355023032b519d5 : SUCCESS in 11m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5f0b81fbe1c9461a97711b558f9dedc2 : SUCCESS in 18m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a57cfb2615e4dbb80c13c25c18296d0 : SUCCESS in 18m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/da0f67fdd7444cd5b4bb749ac6ec2755 : SUCCESS in 21m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a580fb725be468888dc043a889a58a7 : SUCCESS in 11m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/673c83f14ac54fd59e1b84aa20c40447 : SUCCESS in 1h 13m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/71b90a6dc97f4566b85faccf77c91092 : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4882b9d05d8044a692fd54dce011acd1 : SUCCESS in 1h 08m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cc3c3c4ce7547f791366f5582378e95 : SUCCESS in 1h 10m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eb4bf95e1bf943028cc8c2eb1c4749a1 : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/46811d4aa45f4c19b8ce3a119993a1be : SUCCESS in 1h 37m 17s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/63ad74bb5e6448258ab9530f997fc475 : SUCCESS in 1h 33m 00s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bc9c252e32247ceb6e99462092d6ac5 : SUCCESS in 1h 36m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4070427f145144d0922f48c4a995e7b0 : SUCCESS in 1h 40m 04s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/228cdca4af294a49a355811762d6f1ee : SUCCESS in 20m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/45b33d713fa84c43889a499b446fd85c : SUCCESS in 24m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e53791e8ac15493c85c1d986587b2e2f : SUCCESS in 22m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e03281689fb4d0793680a1693e09b16 : SUCCESS in 21m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/084b263849d946f2a5c6c398bf6b007e : SUCCESS in 37m 49s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/972e2c589c534eeaa68c7a0353d9f5de : SUCCESS in 18m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e784931a46fa48a8b3aaaf828abb112d : SUCCESS in 15m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d57e4e35fdb543e482badb6b0127e137 : SUCCESS in 16m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3fd678dbebbe4a6dbbeb5a581848ada1 : SUCCESS in 15m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d95a56e7f18e4f26bcbbc7c32f702bc9 : SUCCESS in 22m 28s","accounts_in_message":[],"_revision_number":8},{"id":"7efbc9f70a27883585b44805f27bfc0487953e61","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 17:40:13.000000000","message":"Patch Set 8: Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"57856495b9249a66a2682ac590908721c9200428","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-29 18:53:20.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"e2bf3816264986f3dce729f37e5f180d85564439","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-07 08:20:28.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"93b458c50e0fb409b99bd616ca087540bfeea552","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-05-08 08:29:27.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3dfaba42c7faa665a2131102da9f450ac6e06bfb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-08 12:47:22.000000000","message":"Patch Set 8: Code-Review-1","accounts_in_message":[],"_revision_number":8},{"id":"554a5f9667069bc47f8978370b97979dd1d9692e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 11:56:04.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"b6f7f0a05da7bac090634f225a990731c5164b06","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 11:56:19.000000000","message":"Patch Set 9: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5aa1d59bf6e6dedefdb564bc601bee70e30782cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 11:56:53.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":10},{"id":"54abdee62f7ad8743b5f57208f50f8b1bb4594ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 13:43:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b78364e7a5d4d0ea1123e7db0767b31\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3f0b81f7344832b466dc0e680db84b : SUCCESS in 9m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7a387986915045b3b4b788ee95253a67 : SUCCESS in 6m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2fef07ab9034a959bf5ed4032e67784 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f850459b9d5c4f75b6f7d543e20385ae : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77309e917c6449a9b9ecdb68b9f017c4 : SUCCESS in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ff089bd8da247bf8978350c44a6c0d1 : SUCCESS in 7m 25s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d76ca0035e5e4f0b9e726d057af20c2a : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3249bfb8ae854e6db99c5f9f412c6a01 : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/10569080dad143719fe30bc81e368bc3 : SUCCESS in 12m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/da6f5aca91a746aab8727955396c3263 : SUCCESS in 16m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c619cb50e8d747819adfcaff0fe8b9bd : FAILURE in 18m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dd49b17d312048a4944076f78ea88608 : SUCCESS in 16m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/418aff50a1594d7f9b174f42b4aafda1 : SUCCESS in 18m 54s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3094337f189b4b23b1c2c36b5e5674bb : SUCCESS in 1h 04m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d2582e14073b4fa79e657b055aa72c53 : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/13d80efe74b74ca6aa59769e04c0c1bc : SUCCESS in 1h 07m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c318a97cc3a44e898e7d3940ea4f30f : SUCCESS in 1h 07m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/64b6463de8fc4ca0a8506bd5a0465b11 : SUCCESS in 1h 06m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fc86f4a609354ae3bb91699f48550dac : SUCCESS in 1h 38m 05s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f25468a39fc24cbd955b5f0200a3bea4 : SUCCESS in 1h 37m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de83dd1d90cd432b8046dbd70c426ec7 : SUCCESS in 1h 35m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17075e51501642fd9402aecec9019511 : SUCCESS in 1h 45m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fd2f2de906d249f28fdcdf9ef44023e2 : SUCCESS in 16m 20s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cb0ee299ebf649e596dcfba96c799984 : SUCCESS in 19m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8887dd0e51ef43b08c49af2f9c8d53bd : SUCCESS in 25m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f2cc48e7158d4cf19cce9f90ce5001c6 : SUCCESS in 17m 27s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4af372e635584d16a4fd029893968c25 : SUCCESS in 37m 42s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5ca6a40045864bed92df9b1e89db46bd : SUCCESS in 26m 05s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3832d4cb08c54f089c1d2b3186838c04 : SUCCESS in 31m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb07cb9fed5844a7ac4a7d552e4e6528 : SUCCESS in 30m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1a7a16d2135d412382e3d787720a3313 : SUCCESS in 27m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59cdf94e8c6441558f97a46107b84a82 : SUCCESS in 24m 20s","accounts_in_message":[],"_revision_number":10},{"id":"700048d5e0fef3725e64aa65dcb3706e05141b45","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 15:04:10.000000000","message":"Patch Set 10: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"aaec6c171293204c2ab302908bd444b82382f394","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-11 15:06:11.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d59e71cb32c05b76c2eb5f59843ec4d5f623a6ab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-11 15:38:05.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"551542c7d467e2e458d7c4a225293fc35ca6ea81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 16:54:26.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4f278c2d3c943aca4a24343f8600942\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95360397b36f4315ba854fd285dd2dd4 : SUCCESS in 9m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/06e6aedbebe0477da6a6bffdb575f195 : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bae6e5f19254b618314350a62ef3514 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/800e56ead2af43d5b2a80e6dfeb63bee : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a94b65ff7a8246508aacea3b82e7d1cd : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ca6b08ef826434f99e53063725e9b76 : SUCCESS in 5m 22s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e06cefcedc14414a4ef7e9e0b42238d : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3da00c57948641568c100481d582747f : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c48365d1b42422489b7194c08c3de22 : SUCCESS in 10m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f0ffaa90bc82404ca01dd74ecb984d6d : SUCCESS in 17m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2e61f97c5264ea2b59ae056447401dd : SUCCESS in 21m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/56ce0b3391aa419aa0b0da3e46b08c8a : SUCCESS in 13m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c16ef50456d4d2e92e2624e30ef750d : SUCCESS in 22m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b5ba29ff751045afb009c08cd25fd9f8 : SUCCESS in 1h 10m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d5b18d80d2e343d1b59f833668ec3d24 : SUCCESS in 1h 06m 30s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/152b6199c5464dc2a0ff74ea59648fa2 : SUCCESS in 1h 07m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba1ddc7018c94aaeafcc991c62974698 : SUCCESS in 1h 03m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d79c257dd3c4c7d8e0783894df680ed : SUCCESS in 1h 02m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b4fe6f1e32774061a19107014c593a76 : SUCCESS in 1h 39m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a436bcf5e4ee4f8ca13bade29b08b5d6 : SUCCESS in 1h 40m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/778259fd45724e9a901878689c8706a0 : SUCCESS in 1h 33m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad324366d4a441a78319a872a6aec69a : SUCCESS in 1h 36m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d45148fcd8c445cd8f7d130469b766a9 : SUCCESS in 27m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/92e55b522d684be78f17d70e6ccfa753 : SUCCESS in 28m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca8f2ae418164c76a4e53f4c4c4b8cc7 : SUCCESS in 27m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/36c4fbfd8245419ba08d4ab8e7182ff8 : SUCCESS in 18m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c70c0d4d8d4a4a38b1f602e1009d5dbf : SUCCESS in 36m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fc008e4f6f244922a92d60900be73aa3 : FAILURE in 7m 19s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/268443c3dc984c218e8723fc7a66905e : SUCCESS in 17m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a60b8abb9224d2da29ad25b270a0fc5 : SUCCESS in 29m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/13ad1de5f6364941bae3ae07c3481871 : SUCCESS in 18m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef692c786c484fafbff17662e2ed3b50 : SUCCESS in 20m 45s","accounts_in_message":[],"_revision_number":10},{"id":"38e7c2db14ef4af685569a76015572de259fc406","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 17:34:59.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e30d8eab7bf79de145ac20fe98b10d54b90d51da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 19:09:58.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f7f49d1c0a146e2a33bf4ff0fe02fcc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7cb3fbb60114094ab3afc0feec130d3 : SUCCESS in 8m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/31679d2c88254229854d972420b3c800 : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64ce33bf2b8448bcaa4e323ee2e1769d : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c3ad98a27e14123b8ff1d5b0dce6909 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/469920dafa15458f88bd58fe67cc0749 : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/158744ec12fd46338816cdaa47bc7fa7 : SUCCESS in 9m 48s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d233dd441c8488ba6713500aaccbf2f : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/363a9692fa804569b1241c4a779e9ba2 : SUCCESS in 8m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca765fbac08b4a3995afb342fc920769 : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1d68605704ce485ab3b865b2951ddee1 : SUCCESS in 26m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0bc8d55ceab476ab25264dbcff3957b : SUCCESS in 19m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/556109b614c24d4580b89416b181f529 : SUCCESS in 20m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1237c8dc91f4b0fa2aaee782f57ae5d : SUCCESS in 15m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/94f07ccc609f40278a4f50cf8036efe6 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e4fcf294066c4d63a3c159887d96cca5 : SUCCESS in 1h 02m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/dd66f19c36904a99a7b25abbff8ec32b : SUCCESS in 1h 04m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/013ff38843aa4c5daff77b1343d8b2c2 : SUCCESS in 1h 02m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/82734dff14f2445a8afff508ffd83a71 : SUCCESS in 1h 02m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/daf7b15c4eaf4aaaac36c8ff757eee61 : SUCCESS in 1h 33m 21s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8dfc31ce253f43f091485c13c0e5a71b : SUCCESS in 1h 30m 30s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0f6ccd6a31849609919845288034ad8 : SUCCESS in 1h 28m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af0bf655009f4b78b0548b3f27777bb6 : SUCCESS in 1h 33m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e3bdcd8dd5a841a0856b2fa2f7ecc83f : SUCCESS in 19m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d1c69f446634444591c28980363de949 : SUCCESS in 19m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a6b2507fdd74e6abe23047df49137fb : SUCCESS in 25m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6dc4b9a71d464a77b1b087fe637a055c : SUCCESS in 20m 08s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9b063a347d4a483892223e626a76d227 : SUCCESS in 15m 57s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e1a41b730d364ddaa3d8e1d3311907f3 : SUCCESS in 17m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/257ac0495228457eaa823247c7baae34 : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54ea26e2daa342419305da86a097040b : SUCCESS in 24m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7a3e95a618474f2097e04d6f267cdac3 : SUCCESS in 20m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d491e6bdf4a4fd5876b457e6e5e7d0c : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":10},{"id":"41b2e461937779e509be8c79f82a645d74d13494","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 19:51:41.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1e963d0d7d6ad627ba9c538bececdfbc76cd3be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 21:29:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b95cdee1c694d3e84545ebc2388f381\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a6f3fa80464230a48912995cf0867d : SUCCESS in 9m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2ba4e07d334f4aa1a97765dfb9394c84 : SUCCESS in 4m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/096ab00c59584d4a9cd45046c2df9445 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5774da8b3f0f48a1a29797bc26571da6 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2018cd5b86da466b83126f1d48ca4924 : SUCCESS in 4m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d81567c44730499793a49e7df1b8374c : SUCCESS in 6m 25s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b63177295fe2492e88ef1abf841fba76 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/575f116d93c34a4fbd369b818f6b6ec1 : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00b40ecd2f44423c9c603582861b9e29 : SUCCESS in 11m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/51504154cac744e1b8d8a9581e233bd4 : SUCCESS in 18m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afa86b84f1f94218b0cd5fee7bf6e0a1 : SUCCESS in 18m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7c82a22e4f2e4a29943aab927fa717aa : SUCCESS in 17m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94c87d6d330c452fb3bd4faef8a7d6f6 : SUCCESS in 19m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/20e887ea124a4b158c1c615349a231ba : SUCCESS in 1h 02m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0a89da1c224e4c9c8da70173e641c720 : SUCCESS in 1h 02m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/581ece3272ba48aa9abf1aa517d54c14 : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55c50ddfaadb4ea5a0fce18151656859 : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a8598a7b344c4453a8c73eabc59cdfb6 : SUCCESS in 1h 05m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9a3d6fee029842bb98714b8af039c270 : SUCCESS in 1h 34m 32s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cc01aebcfa6644a2a883db25fb10feea : SUCCESS in 1h 30m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2f21b50b4094e599fa52aad5a4a911c : SUCCESS in 1h 25m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db7633aa60714dc190742cc59c712b20 : SUCCESS in 1h 36m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/18644fce8c9b4053b853f975c6a1440a : SUCCESS in 28m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/88d38e7ccd2948dfbdda6f63b2e4aa0d : SUCCESS in 18m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbc4dae258cc455e8477aeffede48730 : SUCCESS in 24m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/148e2cd67c3e450688370ae4104ef6b4 : SUCCESS in 23m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/28735b83e6d141718c14a1d9a2b4b0d3 : SUCCESS in 16m 19s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e1e61c64bfe84a66ba98e5226c438239 : SUCCESS in 18m 48s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/690c89f5bc8c47cfb2a1d76004e37139 : SUCCESS in 32m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1fe79df91354062af7711b84f79b322 : SUCCESS in 29m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ccb21d2e84f745fe9e61dca1ff043a05 : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e685bb92b58b4abf9ea320a38c9098ca : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":10},{"id":"c7c922f60839e9ff9b4e46e56f1fe880298a3b8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 11:14:45.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"1b2533046a766fec2ba971b140696cea01f6e767","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 13:26:54.000000000","message":"Patch Set 10: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe/+/986330 is needed.","accounts_in_message":[],"_revision_number":10},{"id":"fc58f99d4d8dc197af4514875324c6ec35845f25","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-13 11:42:33.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e943b96300fcc1dcfaf10a943cdd1c5f2b28a108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 13:21:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdbb5800aa2f4276868d5bce4efc1d85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1c965fe93e4d71b3da5ff1bc5e9849 : SUCCESS in 9m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/54d6fb943e80412b9e77a29e9213b088 : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2be0665b0c34f4c854474fa31adf76e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57254f1ccdf84f7185d93f6e7a2a16ec : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ca1c23a2a904d97a7f33c9d6065ed8d : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/22e16d0dd72145b6944d010be591453e : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6886c8ee61504f5c9702bae413f4b5ea : SUCCESS in 5m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea1292db5760498bb2e3bc221474cc11 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/150029220cc043ee9a9b572d2a89b8b2 : SUCCESS in 8m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1aa9732b26ab49ee8137fdcb5d529ed0 : SUCCESS in 8m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7f517897ab06429f83d67f1e768f84ca : SUCCESS in 19m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2231d1c1f84f43099749239eabe85980 : SUCCESS in 17m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e8d5e619fa044e3cb573bc8ca5e53efb : SUCCESS in 24m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b45ced7e41414cb29635e586a169219a : SUCCESS in 13m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c091f9925839437f980c34618f1c4696 : SUCCESS in 1h 04m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2bc3f3881729410085353eab86ab689f : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7d570e05350f4271a5e21df426635a94 : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b54c633b525a467286cf4b79e4773f19 : SUCCESS in 1h 02m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d0efd38106704c79aeda77aa1f8463b1 : SUCCESS in 1h 05m 57s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e8b9082971754b108406f046c97f2740 : SUCCESS in 1h 37m 58s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8cd539b137824a2394158c16ad4d3146 : SUCCESS in 1h 30m 01s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe6dee98fe52400a819c6454be2ca89c : SUCCESS in 1h 30m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbac15f0ed0341938be6768cb76a9a4e : SUCCESS in 1h 37m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f5904a3110bb40c99a92bffa1c19213a : SUCCESS in 17m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c9862228b4bb48e39124fe65394aaf9d : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/444feaadf42a45d1bc22d5cfaed927f6 : SUCCESS in 25m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/464a1acce788492fa4ba75f364048db0 : SUCCESS in 22m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a0c59f0eacc341069174aa476a9797f0 : SUCCESS in 19m 07s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ba2221e57bde4f2886bc55d89cb2f550 : SUCCESS in 17m 04s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d8a7f394a5d4f5099f176a4ffdc0baf : SUCCESS in 28m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da14a89df5564b2fb260c2d4db7b4d5c : SUCCESS in 33m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b25ebef158ab40978d365465d232f67d : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8325ffc512274f2ebe147097cc01276e : SUCCESS in 21m 04s","accounts_in_message":[],"_revision_number":10},{"id":"1d1c473fe155fe3c116af5e243f6028730cb7f04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 13:22:53.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"dce80b4de7c54c70236c6dc03b179ef9694e0318","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 15:13:24.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f260b419b317452c80291e28d0d52caf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a981c120c76e42108cb2c0c1a459d3c5 : SUCCESS in 9m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7ed754bcdcfa4cfc89375bc01e3c6520 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f9f03a812841bfaf6ac0873f554e22 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3f33deaefab407588ca1093d9ef51d3 : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43ce4185982b4a0ebfaf9a565264d2db : SUCCESS in 5m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/13144aec7aad46aead5442fa8be7c615 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c42892f6e74643a09e6f436e8f9e1b4a : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e03a8597e9ef4931b00410c9fee381bf : SUCCESS in 7m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c6e62a15160e4b2ea79a5ec4ae946897 : SUCCESS in 24m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3ee076dfa8b44499d4259f241e11ea8 : SUCCESS in 11m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/14f3df8cab2a4b108b40bca034df4c9f : SUCCESS in 18m 02s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/677e611b49c04eb28a02ffa024b24e76 : SUCCESS in 16m 42s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3e398d1f218e4d96b3d1ddbc724a90e5 : SUCCESS in 1h 07m 24s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6346986ec41f4a3f946d96aa26ca6240 : SUCCESS in 1h 06m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ce6e82ff0b2747909c0f4508f653f77b : FAILURE in 17m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e00ec73e5ec44d41abfdc93fb973b81b : SUCCESS in 1h 09m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/14c52c87412742989a24ac5b56e1ebe4 : SUCCESS in 1h 11m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c2262a364f714a57b5fa238b896d6018 : SUCCESS in 1h 47m 29s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/263264633b724ab8a588ed8b79f06652 : SUCCESS in 1h 44m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2059e46e1354c47858a78376291d559 : SUCCESS in 1h 38m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/484d4800c7cb49c484632e84ae009d47 : SUCCESS in 1h 49m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1b2c0fa6b2e842a7a75b7f399472e199 : SUCCESS in 14m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0f0f85c221534921be55d1d9af0b5ff2 : SUCCESS in 22m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/669c5c63d9c44a38a085579ca7f64ec2 : SUCCESS in 28m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f72be4350f3f4981af1027dd99664dbb : SUCCESS in 22m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/70e935687b254b6fa7ad255ddc014444 : SUCCESS in 15m 39s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d2cc59b6f73348c18e5dac8d9a73bcf1 : SUCCESS in 19m 22s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c574f4431eec4977a4b9c9cf82946039 : SUCCESS in 38m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6608a51f89c46ff8484210921f5324a : SUCCESS in 35m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9958356e0f7749fd8f55bafe894171c9 : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c448860da39548cd8bc743b1c18f8bec : SUCCESS in 22m 46s","accounts_in_message":[],"_revision_number":10},{"id":"c76e5078a1d829ce86a08db8fa7f68bb42394cd8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-13 15:16:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"96cb11aa25f103b91f3b59a284e40be78e67a2c2","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-05-13 15:51:03.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"518a3ea56b0dc1aaedc0d22195a736c452b9f11b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 16:56:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8198b26f09243b5987f7bcce74b96f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11ec3b64f7ac481f9cbf15751a863141 : SUCCESS in 6m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ca660276f7d344e1880d8e056a376025 : SUCCESS in 4m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df88081a9d1c4888bb358260a7726e42 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e952dd877fcb413c8ef01d4aa6c29a4c : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/410ed298469e4ed1a91d68b6d6f78952 : SUCCESS in 2m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c0b42a9c1ffc418b8d4ebfd63006bdfd : SUCCESS in 5m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c424a54379744d79a9ff2318ecab147a : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/983437376b7b4d4783f291f6ac5c5718 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b5951cecd64d4062b43ec4b0353c6b29 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d809e4cd9be443ba64476a184c2764b : SUCCESS in 11m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e1778eb3b48e4772bac4e10b678c6909 : SUCCESS in 17m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/645b62400144413c9f144055f257da2c : SUCCESS in 13m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0f6cc79a8cfd4441ad698b22e859924e : SUCCESS in 14m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0695347c11364ea582cb3923fd3574af : SUCCESS in 9m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c50ba3280990491b935bddba511dda7e : SUCCESS in 58m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/63134a6c53eb46489790f143742cb44c : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a73150b973d249e0bc560f174501f6c7 : SUCCESS in 1h 08m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2456e71bc65d4575aee35accf92bda38 : SUCCESS in 1h 00m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/39857baa37f24db9b3191f604c885e13 : SUCCESS in 1h 03m 58s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e01beb80000b4df785167bde4bee96a5 : SUCCESS in 1h 39m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6893ec5b84254bf69044283539ee3fa9 : SUCCESS in 1h 34m 24s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0f4ed72ea7347f6a485d45c41a024cf : SUCCESS in 1h 32m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3220b5576ffb46bfa5833997f076d602 : SUCCESS in 1h 37m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b28270f78b514c84aab4bca92a7ab744 : SUCCESS in 16m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/05b99a58bff7496abc528965e094cf74 : SUCCESS in 20m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/979610270862439288956cedbcdf38ad : SUCCESS in 25m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/40422c3f793447d5b7da18d70051e95f : SUCCESS in 13m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ad43a157c6d9433ba64a22399593ed9f : SUCCESS in 30m 03s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/db72c545438d4e119ac35602d5e3d903 : SUCCESS in 20m 43s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f95aa213d636459c810391f982fd0053 : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecded0f043db405487c69d95cf7e9e1b : SUCCESS in 28m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2465ae8da66c4135b5e56aad90065547 : SUCCESS in 14m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcbbebc3f8264b0eb41498b2a1363416 : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":10},{"id":"ba9f6f5d04d0c97d78af2653617d79b7c432c2fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 16:57:15.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"eaa4a4a67c3d16cabe06e07ae33d2d966d9b6ade","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 18:55:15.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bfcfdd99a4e4132b5f17dba2dd535f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2674370f084cefbe204f2f4d32496f : SUCCESS in 8m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f1e44e20d8284b929bbcb42de87e7610 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a4e72f79b34beda0f041b3a2875001 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d256a95b39d044f4bee9d51ebcf26ec9 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b879ee11a0744ec8b1f5fb03c9fb475 : SUCCESS in 7m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d210bfaef193446fa553f5eec956fd33 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6929bb73e9ec4ae4b96d46647a4d56d2 : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7861a9d70844484db3e54bf01befa40e : SUCCESS in 10m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/370b104465bc4193829b0f48265c6ccc : SUCCESS in 24m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b999aa2b9ea7456191e18eaad1854a19 : SUCCESS in 21m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/424288640d5e437c80ea41658fa43e51 : SUCCESS in 24m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f8a7debe01e494287094edf9eeca3a1 : SUCCESS in 23m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/28e1460fa0af426680e04e05f3f9c38e : SUCCESS in 1h 06m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f7d9df0e1e0d42f793d1fc912d786021 : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/51a3dacb0fe74c44908dee2aa6b4af34 : SUCCESS in 1h 08m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bc285995a494d95a060626acc8ee663 : SUCCESS in 1h 07m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/34e1819c817c4b1d87eaed0c144694cb : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f7437521b895469d8e32ac4509416347 : SUCCESS in 1h 36m 22s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d25ef15a4e1f49888b3e1261fa4ac774 : SUCCESS in 1h 57m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebbd4937c22340eba8cc3659a4abaae7 : SUCCESS in 1h 55m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bdd1e6a1c9a487ea840316a97bd384c : SUCCESS in 1h 36m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c50c1963afc2419481078445fd7bad49 : SUCCESS in 23m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7cbd7422b3384f138717e32ff94dd2b4 : SUCCESS in 28m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e7df6df90914ac3b8f5e3ae9bc0e5c1 : SUCCESS in 22m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59b034e4c50e4ed99bce6964e45410eb : SUCCESS in 22m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d479ca930cdf4bd196f7730d3ec15938 : SUCCESS in 35m 06s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0d061ff50ea0441b8eb95fbca492763d : SUCCESS in 30m 37s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/795cec623e4c453d99d1853b5f4305eb : SUCCESS in 32m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06f7162f3554465b828d3c7e270e3a0f : SUCCESS in 30m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b1da71382cd44a75b388684e8f5e293c : SUCCESS in 21m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d5a903c309c4d868144596e8c0f70f8 : SUCCESS in 19m 35s","accounts_in_message":[],"_revision_number":10},{"id":"b8f6aaa581b7e258cd72dc5e50501c0eb05ed5e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 18:55:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"c44fef05533673dd68069be3c8fbc819c9ea849a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 18:56:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d05c66012da4543b953cec7118b9f39\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea2176da9bf5425e835fb0faae35c308 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"8bac4abc963e0dc1995d49a69867ba75c096e615","revisions":{"c541cbe6568781e73d940bb219c8d4c94375fd62":{"kind":"REWORK","_number":1,"created":"2026-04-27 12:41:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/70/986370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/986370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/986370/1"}}},"commit":{"parents":[{"commit":"eac11fa07b0f4a5a0c71d4c4eace342750544e85","subject":"Fix deprecated Templar attribute access in plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eac11fa07b0f4a5a0c71d4c4eace342750544e85"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 12:40:30.000000000","tz":60},"subject":"Fix string-typed conditional deprecation warnings","message":"Fix string-typed conditional deprecation warnings\n\nFix conditionals that resolve to non-boolean types, which will be\nremoved in ansible-core 2.23:\n\n- ip-allocation.yml: use \u0027is falsy\u0027 test for net_no_ip filter\n  result which returns None or a string\n- ipa-images/tasks/main.yml: add explicit bool conversion to\n  string-valued conditionals\n- network-connectivity.yml: use \u0027is truthy\u0027 test for net_ip and\n  net_gateway filter results which return strings or None\n\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#noteworthy-plugin-changes\n\nChange-Id: Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c541cbe6568781e73d940bb219c8d4c94375fd62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c541cbe6568781e73d940bb219c8d4c94375fd62"}]},"branch":"refs/heads/master"},"82b3b12bb6e296375d31839e353121ce543ab56a":{"kind":"REWORK","_number":2,"created":"2026-04-27 16:02:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/70/986370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/986370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/986370/2"}}},"commit":{"parents":[{"commit":"2970d255ea2285485d1e365086173226c93aa0ad","subject":"Fix deprecated Templar attribute access in plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2970d255ea2285485d1e365086173226c93aa0ad"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 16:02:17.000000000","tz":60},"subject":"Fix string-typed conditional deprecation warnings","message":"Fix string-typed conditional deprecation warnings\n\nFix conditionals that resolve to non-boolean types, which will be\nremoved in ansible-core 2.23:\n\n- ip-allocation.yml: use \u0027is falsy\u0027 test for net_no_ip filter\n  result which returns None or a string\n- ipa-images/tasks/main.yml: add explicit bool conversion to\n  string-valued conditionals\n- network-connectivity.yml: use \u0027is truthy\u0027 test for net_ip and\n  net_gateway filter results which return strings or None\n- dell-switch-bmp.yml: use \u0027is truthy\u0027 test for list-valued\n  dell_switch_bmp_images conditional\n\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#noteworthy-plugin-changes\n\nChange-Id: Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82b3b12bb6e296375d31839e353121ce543ab56a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82b3b12bb6e296375d31839e353121ce543ab56a"}]},"branch":"refs/heads/master"},"230aabf568897c7cd359bc1c9dab03df2daa3ec3":{"kind":"REWORK","_number":3,"created":"2026-04-27 16:30:39.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/70/986370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/986370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/986370/3"}}},"commit":{"parents":[{"commit":"2970d255ea2285485d1e365086173226c93aa0ad","subject":"Fix deprecated Templar attribute access in plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2970d255ea2285485d1e365086173226c93aa0ad"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 16:30:29.000000000","tz":60},"subject":"Fix string-typed conditional deprecation warnings","message":"Fix string-typed conditional deprecation warnings\n\nFix conditionals that resolve to non-boolean types, which will be\nremoved in ansible-core 2.23:\n\n- ip-allocation.yml: use \u0027is falsy\u0027 test for net_no_ip filter\n  result which returns None or a string\n- ipa-images/tasks/main.yml: add explicit bool conversion to\n  string-valued conditionals\n- network-connectivity.yml: use \u0027is truthy\u0027 test for net_ip and\n  net_gateway filter results which return strings or None\n- dell-switch-bmp.yml: use \u0027is truthy\u0027 test for list-valued\n  dell_switch_bmp_images conditional\n\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#noteworthy-plugin-changes\n\nChange-Id: Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/230aabf568897c7cd359bc1c9dab03df2daa3ec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/230aabf568897c7cd359bc1c9dab03df2daa3ec3"}]},"branch":"refs/heads/master"},"558555c8638acdaa31336287eea83a40f8f48c4a":{"kind":"REWORK","_number":4,"created":"2026-04-27 18:00:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/70/986370/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/986370/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/986370/4"}}},"commit":{"parents":[{"commit":"2970d255ea2285485d1e365086173226c93aa0ad","subject":"Fix deprecated Templar attribute access in plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2970d255ea2285485d1e365086173226c93aa0ad"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 17:59:14.000000000","tz":60},"subject":"Fix string-typed conditional deprecation warnings","message":"Fix string-typed conditional deprecation warnings\n\nFix conditionals that resolve to non-boolean types, which will be\nremoved in ansible-core 2.23:\n\n- ip-allocation.yml: use \u0027is falsy\u0027 test for net_no_ip filter\n  result which returns None or a string\n- ipa-images/tasks/main.yml: add explicit bool conversion to\n  string-valued conditionals\n- network-connectivity.yml: use \u0027is truthy\u0027 test for net_ip and\n  net_gateway filter results which return strings or None\n- dell-switch-bmp.yml: use \u0027is truthy\u0027 test for list-valued\n  dell_switch_bmp_images conditional\n\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#noteworthy-plugin-changes\n\nChange-Id: Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/558555c8638acdaa31336287eea83a40f8f48c4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/558555c8638acdaa31336287eea83a40f8f48c4a"}]},"branch":"refs/heads/master"},"ef490156ae37d4f05893de5de0e3f7a6d9aa3216":{"kind":"REWORK","_number":5,"created":"2026-04-29 08:40:47.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/70/986370/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/986370/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/986370/5"}}},"commit":{"parents":[{"commit":"2970d255ea2285485d1e365086173226c93aa0ad","subject":"Fix deprecated Templar attribute access in plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2970d255ea2285485d1e365086173226c93aa0ad"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 08:40:47.000000000","tz":0},"subject":"Fix string-typed conditional deprecation warnings","message":"Fix string-typed conditional deprecation warnings\n\nFix conditionals that resolve to non-boolean types, which will be\nremoved in ansible-core 2.23:\n\n- ip-allocation.yml: use \u0027is falsy\u0027 test for net_no_ip filter\n  result which returns None or a string\n- ipa-images/tasks/main.yml: add explicit bool conversion to\n  string-valued conditionals\n- network-connectivity.yml: use \u0027is truthy\u0027 test for net_ip and\n  net_gateway filter results which return strings or None\n- dell-switch-bmp.yml: use \u0027is truthy\u0027 test for list-valued\n  dell_switch_bmp_images conditional\n\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#noteworthy-plugin-changes\n\nChange-Id: Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef490156ae37d4f05893de5de0e3f7a6d9aa3216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef490156ae37d4f05893de5de0e3f7a6d9aa3216"}]},"branch":"refs/heads/master"},"141399bd0fbeebf3717e88fca19b5d2e5e95263d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-29 08:41:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/70/986370/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/986370/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/986370/6"}}},"commit":{"parents":[{"commit":"7f824a7034b50696cedc18d87ec446f2e1df420f","subject":"Fix deprecated Templar attribute access in plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f824a7034b50696cedc18d87ec446f2e1df420f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 08:41:03.000000000","tz":0},"subject":"Fix string-typed conditional deprecation warnings","message":"Fix string-typed conditional deprecation warnings\n\nFix conditionals that resolve to non-boolean types, which will be\nremoved in ansible-core 2.23:\n\n- ip-allocation.yml: use \u0027is falsy\u0027 test for net_no_ip filter\n  result which returns None or a string\n- ipa-images/tasks/main.yml: add explicit bool conversion to\n  string-valued conditionals\n- network-connectivity.yml: use \u0027is truthy\u0027 test for net_ip and\n  net_gateway filter results which return strings or None\n- dell-switch-bmp.yml: use \u0027is truthy\u0027 test for list-valued\n  dell_switch_bmp_images conditional\n\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#noteworthy-plugin-changes\n\nChange-Id: Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/141399bd0fbeebf3717e88fca19b5d2e5e95263d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/141399bd0fbeebf3717e88fca19b5d2e5e95263d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"ef490156ae37d4f05893de5de0e3f7a6d9aa3216","theirs":"7f824a7034b50696cedc18d87ec446f2e1df420f","contains_conflicts":false}},"e67f6a004127797ae3652b293c3d08a01ad0748e":{"kind":"REWORK","_number":7,"created":"2026-04-29 08:41:36.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/70/986370/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/986370/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/986370/7"}}},"commit":{"parents":[{"commit":"7f824a7034b50696cedc18d87ec446f2e1df420f","subject":"Fix deprecated Templar attribute access in plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f824a7034b50696cedc18d87ec446f2e1df420f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 08:41:36.000000000","tz":0},"subject":"Fix string-typed conditional deprecation warnings","message":"Fix string-typed conditional deprecation warnings\n\nFix conditionals that resolve to non-boolean types, which will be\nremoved in ansible-core 2.23:\n\n- ip-allocation.yml: use \u0027is falsy\u0027 test for net_no_ip filter\n  result which returns None or a string\n- ipa-images/tasks/main.yml: add explicit bool conversion to\n  string-valued conditionals\n- network-connectivity.yml: use \u0027is truthy\u0027 test for net_ip and\n  net_gateway filter results which return strings or None\n- dell-switch-bmp.yml: use \u0027is truthy\u0027 test for list-valued\n  dell_switch_bmp_images conditional\n\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#noteworthy-plugin-changes\n\nChange-Id: Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e67f6a004127797ae3652b293c3d08a01ad0748e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e67f6a004127797ae3652b293c3d08a01ad0748e"}]},"branch":"refs/heads/master"},"5700658cb90cf986817c2af0d7a9f43e2ebd3bf9":{"kind":"REWORK","_number":8,"created":"2026-04-29 10:47:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/70/986370/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/986370/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/986370/8"}}},"commit":{"parents":[{"commit":"a98ffaf7fbeff9127422ac298e886a5c57ce627f","subject":"Fix deprecated Templar attribute access in plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a98ffaf7fbeff9127422ac298e886a5c57ce627f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 10:47:46.000000000","tz":60},"subject":"Fix string-typed conditional deprecation warnings","message":"Fix string-typed conditional deprecation warnings\n\nFix conditionals that resolve to non-boolean types, which will be\nremoved in ansible-core 2.23:\n\n- ip-allocation.yml: use \u0027is falsy\u0027 test for net_no_ip filter\n  result which returns None or a string\n- ipa-images/tasks/main.yml: add explicit bool conversion to\n  string-valued conditionals\n- network-connectivity.yml: use \u0027is truthy\u0027 test for net_ip and\n  net_gateway filter results which return strings or None\n- dell-switch-bmp.yml: use \u0027is truthy\u0027 test for list-valued\n  dell_switch_bmp_images conditional\n\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#noteworthy-plugin-changes\n\nChange-Id: Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5700658cb90cf986817c2af0d7a9f43e2ebd3bf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5700658cb90cf986817c2af0d7a9f43e2ebd3bf9"}]},"branch":"refs/heads/master"},"3b14861485f5ab29f29afaf228954851a0c7fb7a":{"kind":"REWORK","_number":9,"created":"2026-05-11 11:56:04.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/70/986370/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/986370/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/986370/9"}}},"commit":{"parents":[{"commit":"a98ffaf7fbeff9127422ac298e886a5c57ce627f","subject":"Fix deprecated Templar attribute access in plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a98ffaf7fbeff9127422ac298e886a5c57ce627f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-11 11:56:04.000000000","tz":0},"subject":"Fix string-typed conditional deprecation warnings","message":"Fix string-typed conditional deprecation warnings\n\nFix conditionals that resolve to non-boolean types, which will be\nremoved in ansible-core 2.23:\n\n- ip-allocation.yml: use \u0027is falsy\u0027 test for net_no_ip filter\n  result which returns None or a string\n- ipa-images/tasks/main.yml: add explicit bool conversion to\n  string-valued conditionals\n- network-connectivity.yml: use \u0027is truthy\u0027 test for net_ip and\n  net_gateway filter results which return strings or None\n- dell-switch-bmp.yml: use \u0027is truthy\u0027 test for list-valued\n  dell_switch_bmp_images conditional\n\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#noteworthy-plugin-changes\n\nChange-Id: Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b14861485f5ab29f29afaf228954851a0c7fb7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b14861485f5ab29f29afaf228954851a0c7fb7a"}]},"branch":"refs/heads/master"},"8bac4abc963e0dc1995d49a69867ba75c096e615":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-05-11 11:56:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/70/986370/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/986370/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/986370/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/986370/10"}}},"commit":{"parents":[{"commit":"31252b804931d54c49534b85dd45f823f7688294","subject":"Fix deprecated Templar attribute access in plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31252b804931d54c49534b85dd45f823f7688294"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 08:22:45.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-11 11:56:53.000000000","tz":0},"subject":"Fix string-typed conditional deprecation warnings","message":"Fix string-typed conditional deprecation warnings\n\nFix conditionals that resolve to non-boolean types, which will be\nremoved in ansible-core 2.23:\n\n- ip-allocation.yml: use \u0027is falsy\u0027 test for net_no_ip filter\n  result which returns None or a string\n- ipa-images/tasks/main.yml: add explicit bool conversion to\n  string-valued conditionals\n- network-connectivity.yml: use \u0027is truthy\u0027 test for net_ip and\n  net_gateway filter results which return strings or None\n- dell-switch-bmp.yml: use \u0027is truthy\u0027 test for list-valued\n  dell_switch_bmp_images conditional\n\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nhttps://docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#noteworthy-plugin-changes\n\nChange-Id: Ie8fbc96c4ecb7d18ccc97a2c180ac783584e87d1\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bac4abc963e0dc1995d49a69867ba75c096e615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bac4abc963e0dc1995d49a69867ba75c096e615"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"3b14861485f5ab29f29afaf228954851a0c7fb7a","theirs":"31252b804931d54c49534b85dd45f823f7688294","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
