)]}'
{"id":"openstack%2Fkayobe~986410","triplet_id":"openstack%2Fkayobe~master~I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-14 12:10:52.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-05-14 12:10:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e","subject":"Fix non-boolean nmstate zone map conditionals","status":"MERGED","created":"2026-04-27 16:02:24.000000000","updated":"2026-05-14 12:10:52.000000000","submitted":"2026-05-14 12:10:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986410","meta_rev_id":"9f7e52ade7e1bc22146d258066e8f1f99842a85a","_number":986410,"virtual_id_number":986410,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-14 12:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-13 16:03:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":2,"date":"2026-05-13 16:03:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-05-13 16:03:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-13 16:03:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-05-14 10:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-05-14 10:00:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-13 18:14:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-27 19:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 15:25:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-07 08:24:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-07 08:24:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-05-07 08:24:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-07 08:24:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-05-07 08:24:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-05-07 08:24:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-05-07 08:24:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-05-08 10:24:44.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"REVIEWER"}],"messages":[{"id":"e9f9185ee4a30c72b220255201bd615e7d06e8e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:02:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef2ec6bb7665d7ccbd005ac002d57496aed710b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:30:39.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ddc4c00eb1871ad59b54a41acb7da4b9cbd6218d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 16:32:05.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"523bf50ab9948eaa6ca26bec4d11af7e929ec458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 18:00:55.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"24d69dda9d2e78dc1f21d82b65857efe8350b6ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 19:39:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fccefa977344ff687c7340131a8e63f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc179a75c2c749a89a6e3f81f3aa6924 : SUCCESS in 9m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6296c322c62f447fab6bb046b0d650cf : FAILURE in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/369fd970b5ae4bdfa93b555cbc8f9d83 : FAILURE in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c921abe3c6c43679de4d1e2d6293e66 : FAILURE in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15890bc365f4494a8397a1075eaab4cb : FAILURE in 3m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee8edcd715834d5c904a0f36890327a5 : FAILURE in 5m 39s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/94cbbd7d679345a586e79f049cde0a63 : FAILURE in 3m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e97133110d14408885b360f4278ae5b : FAILURE in 3m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/58091e22d53a4b61b4e480d5a5da3604 : SUCCESS in 11m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e49c140e55754a53a3c4a56f3542f80b : SUCCESS in 20m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccba9a5cf4104eb2b612717ebaa18e98 : SUCCESS in 20m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d34347b809f5488aa4a5a3c9277160ba : SUCCESS in 23m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/594ef7e33fcc4a24807f763cf84d08b6 : SUCCESS in 16m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/50d85d444dca4d7baaee9f735b7e64ca : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d2be9d73aeeb42acb126978c2986b6d6 : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/aa3ceb8e4eca428096db026b91521f9e : SUCCESS in 1h 06m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49104d1f2cb148b9bdaf26823f91f350 : SUCCESS in 1h 07m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/77ac0328a84744c99e937c44d2141745 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/59e47cea29604b0dbd738f2a3336bda1 : SUCCESS in 1h 36m 18s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/70a52b44c250407b91eb4c1150248373 : SUCCESS in 1h 29m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/616e381dea6e456792e0dc8484cdf528 : SUCCESS in 1h 24m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80cdc85edb2c4ba69726181acb1190a1 : SUCCESS in 1h 34m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dd9228786b0c45da9c7cb55ef3455903 : SUCCESS in 22m 03s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/aeda32b535914ca68433f9871e9476fa : SUCCESS in 27m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6a409c62c504d3b97b63a1f93c7a394 : SUCCESS in 21m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/27389b4a68c14e939ac22f65dd494ff1 : SUCCESS in 22m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/112b9768ee4644c7a6ce0cbea8ea3bff : SUCCESS in 26m 27s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fd7129e288b14d4a9877716ada298c28 : SUCCESS in 36m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6cabb550a11f463badc938d38d487d29 : SUCCESS in 32m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72791cb5454043b4b7f5899d437f2133 : SUCCESS in 29m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/23c823f8120c42db8bada1b264f8317c : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/583e5793b8494aeab226639885fb98b3 : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":4},{"id":"63dfae6fc1b16f7d83e5f50a3a07fefb7d38c52e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-28 22:36:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"87e80c3ccfc8a444204fcdc73635c67491a67cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 00:15:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97c074304760404da7b9d89f80466efa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd2c3a62d684476ba3a34e81ddedb38 : SUCCESS in 9m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cf2231ddcef2476689618053bc3ad0a1 : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ffc85bb9af24bf09a9fac6527055187 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7321562079f5423db247a6d52f12dff6 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a963603a0b841a6b09faafabe043c56 : SUCCESS in 3m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/84df521b66364de3a284cf0b27579640 : SUCCESS in 6m 38s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dd888ad91ffc49a2afcb977f64632e5d : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ec13bedebec42229ad8cc4b03ee20a9 : SUCCESS in 11m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5494e19ea4084126a3c67231390774fc : SUCCESS in 10m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cef531457eef48e7a5d8a9d329937a9a : SUCCESS in 15m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ef57896019d4af0b8c2a201d5ccba24 : SUCCESS in 18m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d9dff74603534e6ca33e32e56b074572 : SUCCESS in 19m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/381d4f36b9a146a7b97e6450d4fe9e57 : SUCCESS in 16m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/eb42bc678f4c47509cdffd01f132b067 : SUCCESS in 1h 03m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0add82b44bc74eaab13a78137395e677 : SUCCESS in 1h 06m 32s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b3a25bc34005436f9a7ee3271a8cf8d7 : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5297cfd6a2754433824e57178c7c18a9 : SUCCESS in 1h 03m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fc84eb871f194503865e99eeaef37e01 : SUCCESS in 1h 06m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b3c6a0c4cd348f8bd5e7ce99c4923cc : SUCCESS in 1h 34m 28s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ee8730a75b44423f8cfa84935580ca6b : SUCCESS in 1h 28m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d15987c7ec6b41f3a72735c96c4ea065 : SUCCESS in 1h 25m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4dbe9a55b554db9a91446f71b63bdf7 : SUCCESS in 1h 36m 03s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f3226c901f004af381f443455b94af57 : SUCCESS in 15m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ac48a0a7dc114b58a95f4d8ef67b61a0 : SUCCESS in 14m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3addddc7c6904761875a6e1503035951 : SUCCESS in 21m 45s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a75d17e9835a4523918f222e794b9651 : SUCCESS in 22m 56s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/010348dfeade490bb6a12c62e2b7a087 : SUCCESS in 37m 46s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/85ff19bdc4894c7b84abf19c487d5285 : SUCCESS in 36m 45s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e42147c9fdfd4ec5a3452a2955115b7f : SUCCESS in 29m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/803f3a6084564d42b8e28b7ca7497146 : SUCCESS in 29m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/08c600f45e60490a8d5f7920fe305626 : SUCCESS in 25m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6be757c88e8046649dfcac9caa28528d : SUCCESS in 20m 56s","accounts_in_message":[],"_revision_number":4},{"id":"1a26b50cba0dc2a7e685109e3b5559910df5b597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 08:41:03.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"65d915bd355a0f0803d651d7183df0ff24e1972e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 08:41:43.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3348baa239fbc69d4ad4ee31e3c489d56b17fa6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 10:47:52.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"09c7f2b152034295cfbc38897755f1c4fb308dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:46:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abe4d2391b5f444896428d6d9534eb32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/606eaccb04ba4613ba28305ff97ffc28 : SUCCESS in 7m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/95654f3a1d6e43cc91ad0ea4a686f0e0 : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54f653d4fa504c71b47ad2e4209f6a6c : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb990a4ecec435393a66c13585121fa : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14065e2735d44c5a857cc4c2ab7b8ff7 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f8bccde7aa4f4126b3e7ec8676752c62 : SUCCESS in 6m 17s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/35135ef62d814431b733102ea4327e7d : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bc526378a3754de3be026de251d92bcb : SUCCESS in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dab50173674341d5838a673aed8f8e0c : SUCCESS in 11m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1a3c527f9c1a45c29e58da3e65a2850e : SUCCESS in 21m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c34dda656a8c4876a36cdb3247ea9ff2 : SUCCESS in 16m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c485a50534654c87a42bedebe8d64a41 : SUCCESS in 11m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c612c4211d5e4327b63035179648b8d1 : SUCCESS in 13m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fef54ea0b1b9496d898284b5854b07c5 : SUCCESS in 1h 04m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9d13c09407a845be8e8dbbce04467fe1 : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/86e369c531844096b02dcb18a233b318 : SUCCESS in 1h 13m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d4ac60b5b1f4a8a8d3f217671eb41fe : SUCCESS in 1h 05m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3ee0a9896eb414c85bd84cf0ac8135d : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d6579742decb41349000c45698e67e65 : SUCCESS in 1h 39m 32s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3a9501dfdcc84ee1906670909f35be30 : SUCCESS in 1h 33m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d01fd2cf7d2a4bb6be5566b7e80c7adc : SUCCESS in 1h 27m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84b9b5f964ff4be19a48dc691236e48a : SUCCESS in 1h 54m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9a4c65e6b1414f5d9c69b00e674b6738 : SUCCESS in 15m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/26ab156df07e480995e845c51ffb1540 : SUCCESS in 29m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a2d56bcdc844f6385b53ff6e8e95b5b : SUCCESS in 19m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bd5ffaeed8464cc9962ef66f1e7efc61 : SUCCESS in 20m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cb7d393ea05044d681a0d4d95212223e : SUCCESS in 17m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/14541e393a554e8ea7ed4d76a545a56a : SUCCESS in 20m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/299e8019dcfe4528abde8c4cd524390c : SUCCESS in 35m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/881e3f9e1e20450a8f2bac8d08809722 : SUCCESS in 28m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4acfac2a8a3f4541bbfa1a483307ed06 : SUCCESS in 27m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e63a4a7e7be341d1b32971ca823d930c : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":7},{"id":"bc0bc0dc326a8ce1fbae3f65ed069e86e79c3446","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 14:44:44.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"05f0f1e7e11267d27fa8a3be4b22d88061ff0a70","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-30 15:25:58.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5f27ca1f8ad13c391879bd97defb9c14da6358da","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-07 08:24:48.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"0cfcd4d2a4b1f31290648f66abb77e071f08de96","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-05-08 08:23:09.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"cf12a4ec80b4ee9ea9e5ab907b1877901795adad","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-05-08 10:24:44.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"396aeded15e7464833f8688ea97b9e7280e60d90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 11:56:53.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"edd21410a109d96900dbf9afba638e226ac44efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 13:42:25.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c8ecb5607841758642c24ca27e2450\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f67b4f489b344c0fba678c619eee802e : SUCCESS in 7m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ee2ec046b4b74988b756f7b659443b11 : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bb301b342f048879d50c2915997a257 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1969efb3cbf7460da3dbc78101fc1045 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc967e1ad9fd427d8c551d9fefe1ffa1 : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4fac118af7084101942d5a8fcff76d4a : SUCCESS in 5m 26s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fcb7eebd52094701bcd5dbf2505e8457 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/408e4aea976f4da9bc3164bfe2524f79 : SUCCESS in 7m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d40a7d8508d742d3848ba669647909d2 : SUCCESS in 11m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f501ee7b2e624a31bbabe9125ac0bd1f : SUCCESS in 16m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/970d657196f942e6a5f2c09d3f2e49be : SUCCESS in 20m 58s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6ca4359f7d3f47d0b8f3a0099733ae03 : SUCCESS in 25m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bff4eadb0d94ba2afcae0f43f8c3779 : SUCCESS in 18m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/80bfe9b01e6d4b7bb67d9f619edd7966 : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/af9e6885c3a84ee99db3207becb773fa : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3126b7c086f64721a29ce0ccb8e2c93d : SUCCESS in 1h 08m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d69cf0246f54b5e808479941912a926 : SUCCESS in 1h 07m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/37208cd2ffa94b4ea25e44e03bf9c387 : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6945ba78b5a54d3eb6c315bc95e15375 : SUCCESS in 1h 43m 28s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c58920a581c249e388ae42904000c504 : SUCCESS in 1h 32m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44bb1beb11f841a7a083743c2c8f4251 : SUCCESS in 1h 35m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9243483a4e3c46749d09f5513d50146c : SUCCESS in 1h 40m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/890f69b984bd4ad2ad8a6927400e2a7a : SUCCESS in 17m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e2157db4880047bb95a926839431ea8d : SUCCESS in 16m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db8fb76b94c745a2bd75a1d082399e4c : SUCCESS in 22m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/71a299cd420b4fac9fdde985e5b40401 : SUCCESS in 22m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c1927dcd2dcb42e48fec89ebcb3a26a4 : SUCCESS in 16m 49s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/99433d33ad1a411ebff7a18fc65f5a3b : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/612dfbe3196d4023a7df6c185891780e : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a902e2ba1f564bd0b06f4fe7a2c812da : SUCCESS in 17m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6378095ca4204904972b88f12363b0f1 : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c900c85ab0d4340818fe970d77c4f77 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":8},{"id":"c3b181a94cfe6ee2337ca3e25f432ee6d825877d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-13 16:03:34.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"643e0b31c864e131929406d2a72f5b579612a27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 17:52:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b043855d136f4967956d2ac72c118601\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efeb2dc9287644bbacded0b0bb36c996 : SUCCESS in 9m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d3a80fb251154bef87449468c26907dc : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fce2866ff2e34318a562a3989b99bfae : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65a8410db54646ccb65b66017b39b367 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dfbc07136a24bf389b92c3b5150a796 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a65746bf96df4bd78fff664fc0b60d7b : SUCCESS in 6m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c17f1fe3e7fe43ecbb948c6c1cf23442 : SUCCESS in 5m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea09cc0beccc45379b1de8a6b4e2be31 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3273b986f2ea4bdf8041147ccc1cea94 : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1826269f75c74360aa5a8f71f0c5e4ca : SUCCESS in 11m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8cc883751cf047d4924a3575fc5d3957 : SUCCESS in 24m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf0075026da041a8b758b843d2fdb53a : SUCCESS in 11m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/554395737faa44e89937528b4d35c784 : SUCCESS in 15m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/632906ef74004b8a83cc183a3900d8aa : SUCCESS in 22m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3f9af1d0acba4801926fbc2c09154671 : SUCCESS in 1h 05m 21s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0a3ddd788bbf4776830472b6b9509a48 : SUCCESS in 1h 03m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3758ed4f2bb54d10a380c8ae2b94869a : SUCCESS in 1h 06m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08d13b92177a44f5b2ced29106e15f40 : SUCCESS in 1h 02m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9791db0e760e4639ada3ddcb68fae635 : SUCCESS in 1h 04m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/12159e68d74c434d88d80ff30973aa45 : SUCCESS in 1h 36m 54s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a2838f088b994892bb8761e8741fb16e : SUCCESS in 1h 40m 05s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d54df44b216b40e0b8547771b4ab0ebb : SUCCESS in 1h 29m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6beef0762fa144a5b4439fdf8482c914 : SUCCESS in 1h 47m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/74fef03db8454712825d91a5b37c7068 : SUCCESS in 29m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/48052f099b8b4570a33da3a088a314bf : SUCCESS in 22m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/145d0761c4a64b25811fef42717959a4 : SUCCESS in 23m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0e97eef11b64401a959d60caebb01146 : SUCCESS in 21m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d172e33e1a564c6f890efbc753d8fa92 : SUCCESS in 37m 01s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/01a057c2fe7640be9093b4901fdc9512 : SUCCESS in 25m 01s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7ccfcdd337d43a4b6850b591d847b44 : SUCCESS in 15m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/819efa1a4cf24788a389b2243ad09240 : SUCCESS in 26m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b8163a11e53e47df95d61a90d042535e : SUCCESS in 13m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b05836e85db453597e44af5ce6fa7e7 : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":9},{"id":"e8dfad684290a7b22094fc9abac1fb22a1143700","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-13 18:14:18.000000000","message":"Patch Set 9: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"01de272a161653b470e97fce4176fc7ca5cd0628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 19:52:16.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a265fe9f49d4bad8e9f9ad540140303\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c4e2b68e2354606b90eec61fb99026f : SUCCESS in 10m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/11cd062a9a22497c948dae4685e2ce2b : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8fb6f3219f54d6886526993c12d2395 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750494873ae14002931775ef9d4b80ca : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ffdcd503c1cc41a39505c8f5292b050b : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c98b7f9c9a0444395f32fc807c10f27 : SUCCESS in 6m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53501d7b86d640b390affcbf0fd8ba46 : SUCCESS in 6m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c2fc47c129243b18714bb73600fe93c : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b51899d47cd84bf996f02646d922c45b : SUCCESS in 13m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/79e57583f17c45179d6b456b999e404b : SUCCESS in 11m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0ccfab9358dc4617a95cb1df3fe35100 : SUCCESS in 17m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c0832cadbc844a4b7b5348c231d19c1 : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ec7e3cfefce747ef9460ad3edf129117 : SUCCESS in 19m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6994ec938f3949328def9c9dc792352d : SUCCESS in 19m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2104f35282be4a3da34eace75ada455e : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1f010bf960d643b4b6b5805ffe7f962c : SUCCESS in 1h 03m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/998b96b5a333403c9b73e304aa1e2256 : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48562afa8e554f219f473096e0796b5b : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b248d7d0c0647e99c9361f22af07434 : SUCCESS in 1h 05m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c24fdc4028aa40999a0d111b24c6e8f2 : SUCCESS in 1h 36m 29s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fd23bcb379b54ca7b79b33f0f97cef61 : SUCCESS in 1h 35m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73f94aa897bc4ef7a1bb5335c2f22b28 : SUCCESS in 1h 32m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f853a091da0478ea79dff7b35e1c74b : SUCCESS in 1h 36m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3e7f194cd22042d8ac2ce9e37b6a7b34 : SUCCESS in 18m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c644d831123341cd8d797dc759cf13e0 : SUCCESS in 16m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dd9aabdc205458c87dc05f489e43989 : SUCCESS in 22m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1ea1a4eac6674449976f19f5e6122027 : SUCCESS in 23m 59s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fd31211aef79431eadfca119d7b17133 : SUCCESS in 31m 37s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bef3aa50cf9543f4befcf83d725fd4b4 : SUCCESS in 38m 33s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f081f44ca70d4d43a9f303e6a8565d93 : SUCCESS in 35m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74e5d4cd1dbc41eb97eeff763f58f9e0 : SUCCESS in 27m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/11d017e85ca34d87b098a910345b5d58 : SUCCESS in 15m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93afc2d7e72d4a09b850c3e380b6ea80 : SUCCESS in 13m 44s","accounts_in_message":[],"_revision_number":9},{"id":"209a0f724f2ccfb56f2b4cebf6e7af30112b5a35","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-14 10:00:39.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"e49d0b5c42c77a55e7c62352063f9401f61e41f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 10:18:47.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"61e1e8b5e756768c64e16e3f7273ae9f014a2dd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 12:10:52.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e6a6b2e7e1f457aa335bf7f5f282c4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9239b13464f4e3aaa052b32f2485dcc : SUCCESS in 9m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/847fb7d9dc3242c9990b7c5358d0e472 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e5525711c45484ab01e8ff2f1fe40f7 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab0f5ab9fb634cd0884591b9ee13c0d8 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86b0f9b3e49446d3b68f06e9762ef83f : SUCCESS in 6m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2cad1a06d8b040cabff093c95651aed6 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e522d118c2134cdfac2bf216726162a4 : SUCCESS in 12m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/43227098df8c46d7a81462f441c6c606 : SUCCESS in 11m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/90069a995bca41ee88e9d02040437eea : SUCCESS in 20m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf3b9e3316a44239b68ff0f34902d12e : SUCCESS in 19m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c5089c9a84824131a212653107e8bcde : SUCCESS in 11m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac919249c78b4c8a976383ba0808012e : SUCCESS in 18m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/83fe1d9e4baa436c9fa3b893ece63787 : SUCCESS in 1h 05m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ee69ea7d9e624b5689f76bd0c2ee8c8b : SUCCESS in 1h 04m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/50e3e9bbf6b449ac9cb2f8b1799334fb : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e85a509aeb54be88337e61a340afdd9 : SUCCESS in 1h 02m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5f556236bb145218258ac8590243533 : SUCCESS in 1h 09m 19s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7a67426f741d45918be2dac306b3f5df : SUCCESS in 1h 41m 51s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/964f157975cd45e5a6c71761ae803740 : SUCCESS in 1h 31m 40s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6e58a9cc30e40bd98806f4f781dcd2f : SUCCESS in 1h 35m 57s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/108bcb29d6d9421e8e129f9ffed25833 : SUCCESS in 1h 36m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/808a91ad47244f6897e2ac149ee27b71 : SUCCESS in 21m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/45172c7752dc4fdeb0c5ac19e219a207 : SUCCESS in 23m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dc7e1ba75044b99a0bc4f41692a0503 : SUCCESS in 22m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1dad9ed3a5ec449e991d421d760487dc : SUCCESS in 20m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/72936c4b6c26413f8933bdaf1aea143e : SUCCESS in 16m 03s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dff252758b514c9abe3142fd2854c421 : SUCCESS in 26m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db0ee71a9d5a4e07a52c07d7a1c1a57e : SUCCESS in 28m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49edf65a23094cf7a89448a04c6932f7 : SUCCESS in 27m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ee600d38cfe54c2aa6b75d5ec3bef288 : SUCCESS in 14m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bb7feb087a04bbe9bb43f395d74131b : SUCCESS in 19m 46s","accounts_in_message":[],"_revision_number":9},{"id":"9f7e52ade7e1bc22146d258066e8f1f99842a85a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 12:10:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"5de0ec85fbd09e50f2e809f2cee5d209dcdeefa6","revisions":{"443957985c00dbcce3242581adab0d71757ae634":{"kind":"REWORK","_number":1,"created":"2026-04-27 16:02:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/10/986410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/986410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/986410/1"}}},"commit":{"parents":[{"commit":"e4fb6888cbe816e7b2ba8136a5c98595c0d9414b","subject":"Avoid deprecated ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4fb6888cbe816e7b2ba8136a5c98595c0d9414b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 15:20:05.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 16:02:17.000000000","tz":60},"subject":"Fix non-boolean nmstate zone map conditionals","message":"Fix non-boolean nmstate zone map conditionals\n\nUse explicit boolean tests in the network-nmstate zone map\nloop conditions to avoid ansible-core broken-conditional\nfailures when filters return None or strings.\n\nhttps: //docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nChange-Id: I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/443957985c00dbcce3242581adab0d71757ae634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/443957985c00dbcce3242581adab0d71757ae634"}]},"branch":"refs/heads/master"},"a247235c4521e271c98f491f8eb0f8f10e202d3d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-27 16:30:39.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/10/986410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/986410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/986410/2"}}},"commit":{"parents":[{"commit":"7cba7f60011eea59254eaaca697df18f61f23d0b","subject":"Avoid deprecated ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7cba7f60011eea59254eaaca697df18f61f23d0b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 15:20:05.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 16:30:29.000000000","tz":60},"subject":"Fix non-boolean nmstate zone map conditionals","message":"Fix non-boolean nmstate zone map conditionals\n\nUse explicit boolean tests in the network-nmstate zone map\nloop conditions to avoid ansible-core broken-conditional\nfailures when filters return None or strings.\n\nhttps: //docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nChange-Id: I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a247235c4521e271c98f491f8eb0f8f10e202d3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a247235c4521e271c98f491f8eb0f8f10e202d3d"}]},"branch":"refs/heads/master"},"e6f475b5eb890377f66bd791b6ba3237b3bf40b7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-27 16:32:05.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/10/986410/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/986410/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/986410/3"}}},"commit":{"parents":[{"commit":"db7138d6e25b281514d8662f1841d283c6a0115b","subject":"Avoid deprecated ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db7138d6e25b281514d8662f1841d283c6a0115b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 15:20:05.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 16:31:56.000000000","tz":60},"subject":"Fix non-boolean nmstate zone map conditionals","message":"Fix non-boolean nmstate zone map conditionals\n\nUse explicit boolean tests in the network-nmstate zone map\nloop conditions to avoid ansible-core broken-conditional\nfailures when filters return None or strings.\n\nhttps: //docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nChange-Id: I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6f475b5eb890377f66bd791b6ba3237b3bf40b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6f475b5eb890377f66bd791b6ba3237b3bf40b7"}]},"branch":"refs/heads/master"},"17c9c5d6d34e9a67d92027883646e41c7b0215a9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-04-27 18:00:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/10/986410/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/986410/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/986410/4"}}},"commit":{"parents":[{"commit":"05f83acc7dc02033e4dee3cc06bdcc6fd6d1148d","subject":"Avoid deprecated ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05f83acc7dc02033e4dee3cc06bdcc6fd6d1148d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 15:20:05.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 17:59:14.000000000","tz":60},"subject":"Fix non-boolean nmstate zone map conditionals","message":"Fix non-boolean nmstate zone map conditionals\n\nUse explicit boolean tests in the network-nmstate zone map\nloop conditions to avoid ansible-core broken-conditional\nfailures when filters return None or strings.\n\nhttps: //docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nChange-Id: I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17c9c5d6d34e9a67d92027883646e41c7b0215a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17c9c5d6d34e9a67d92027883646e41c7b0215a9"}]},"branch":"refs/heads/master"},"13095e290e0f9183e8b1f2d7e9ce37bcf96ee19f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-04-29 08:41:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/10/986410/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/986410/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/986410/5"}}},"commit":{"parents":[{"commit":"f73b0a542f508dfc7fdd742c279a1d42f444092b","subject":"Avoid deprecated ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f73b0a542f508dfc7fdd742c279a1d42f444092b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 15:20:05.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 08:41:03.000000000","tz":0},"subject":"Fix non-boolean nmstate zone map conditionals","message":"Fix non-boolean nmstate zone map conditionals\n\nUse explicit boolean tests in the network-nmstate zone map\nloop conditions to avoid ansible-core broken-conditional\nfailures when filters return None or strings.\n\nhttps: //docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nChange-Id: I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13095e290e0f9183e8b1f2d7e9ce37bcf96ee19f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13095e290e0f9183e8b1f2d7e9ce37bcf96ee19f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"17c9c5d6d34e9a67d92027883646e41c7b0215a9","theirs":"f73b0a542f508dfc7fdd742c279a1d42f444092b","contains_conflicts":false}},"25fb848d1bbe0f0e3b1ea7456ea64e2f0ceb95eb":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-29 08:41:43.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/10/986410/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/986410/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/986410/6"}}},"commit":{"parents":[{"commit":"fbb64280737bc7f052f892c30de4a741948fe912","subject":"Avoid deprecated ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbb64280737bc7f052f892c30de4a741948fe912"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 15:20:05.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 08:41:43.000000000","tz":0},"subject":"Fix non-boolean nmstate zone map conditionals","message":"Fix non-boolean nmstate zone map conditionals\n\nUse explicit boolean tests in the network-nmstate zone map\nloop conditions to avoid ansible-core broken-conditional\nfailures when filters return None or strings.\n\nhttps: //docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nChange-Id: I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25fb848d1bbe0f0e3b1ea7456ea64e2f0ceb95eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25fb848d1bbe0f0e3b1ea7456ea64e2f0ceb95eb"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"13095e290e0f9183e8b1f2d7e9ce37bcf96ee19f","theirs":"fbb64280737bc7f052f892c30de4a741948fe912","contains_conflicts":false}},"bc29d2a8354d6d0c9ea4bfcd3252c7efc850c7a7":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-29 10:47:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/10/986410/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/986410/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/986410/7"}}},"commit":{"parents":[{"commit":"ff3a30ccee0dd862244d3aa8de3378cbfa6ff0a3","subject":"Avoid deprecated ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff3a30ccee0dd862244d3aa8de3378cbfa6ff0a3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 15:20:05.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 10:47:46.000000000","tz":60},"subject":"Fix non-boolean nmstate zone map conditionals","message":"Fix non-boolean nmstate zone map conditionals\n\nUse explicit boolean tests in the network-nmstate zone map\nloop conditions to avoid ansible-core broken-conditional\nfailures when filters return None or strings.\n\nhttps: //docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nChange-Id: I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc29d2a8354d6d0c9ea4bfcd3252c7efc850c7a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc29d2a8354d6d0c9ea4bfcd3252c7efc850c7a7"}]},"branch":"refs/heads/master"},"d3fbffd397a737f19e169846f0431b4a3c730371":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-05-11 11:56:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/10/986410/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/986410/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/986410/8"}}},"commit":{"parents":[{"commit":"901a54585ae1e686eee95f9ce1d90f0e971ddaca","subject":"Avoid deprecated ansible_managed variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/901a54585ae1e686eee95f9ce1d90f0e971ddaca"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 15:20:05.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-11 11:56:53.000000000","tz":0},"subject":"Fix non-boolean nmstate zone map conditionals","message":"Fix non-boolean nmstate zone map conditionals\n\nUse explicit boolean tests in the network-nmstate zone map\nloop conditions to avoid ansible-core broken-conditional\nfailures when filters return None or strings.\n\nhttps: //docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nChange-Id: I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3fbffd397a737f19e169846f0431b4a3c730371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3fbffd397a737f19e169846f0431b4a3c730371"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"bc29d2a8354d6d0c9ea4bfcd3252c7efc850c7a7","theirs":"901a54585ae1e686eee95f9ce1d90f0e971ddaca","contains_conflicts":false}},"5de0ec85fbd09e50f2e809f2cee5d209dcdeefa6":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-05-13 16:03:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/10/986410/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/986410/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/986410/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/986410/9"}}},"commit":{"parents":[{"commit":"8bc5c54184a2d79b684dcaec2e43e70a66fb997d","subject":"Fix deprecated play_hosts in network connectivity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bc5c54184a2d79b684dcaec2e43e70a66fb997d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 15:20:05.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-13 16:03:34.000000000","tz":0},"subject":"Fix non-boolean nmstate zone map conditionals","message":"Fix non-boolean nmstate zone map conditionals\n\nUse explicit boolean tests in the network-nmstate zone map\nloop conditions to avoid ansible-core broken-conditional\nfailures when filters return None or strings.\n\nhttps: //docs.ansible.com/ansible-core/devel/porting_guides/porting_guide_core_2.19.html#broken-conditionals\nChange-Id: I3ffd5691603a49361fc7a9f16cfccd5cc0f0415e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5de0ec85fbd09e50f2e809f2cee5d209dcdeefa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5de0ec85fbd09e50f2e809f2cee5d209dcdeefa6"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"d3fbffd397a737f19e169846f0431b4a3c730371","theirs":"8bc5c54184a2d79b684dcaec2e43e70a66fb997d","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
