)]}'
{"id":"openstack%2Fkayobe~986507","triplet_id":"openstack%2Fkayobe~master~I48ccf4fb40036ca25cd00050342fbd0b6e6cfad2","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-14 15:22:09.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"hashtags":[],"change_id":"I48ccf4fb40036ca25cd00050342fbd0b6e6cfad2","subject":"Add kolla_bifrost_inspector_default_node_driver variable","status":"MERGED","created":"2026-04-28 09:54:35.000000000","updated":"2026-05-14 17:10:51.000000000","submitted":"2026-05-14 17:10:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"986507","meta_rev_id":"a782bed95413b37f543f48ddd308e098735cf872","_number":986507,"virtual_id_number":986507,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-14 17:10:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-14 15:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-12 13:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-14 15:22:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-30 14:44:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-28 11:39:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-12 13:01:47.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"81f469f28c1c8523a42234b5f3e815c222f18de8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-28 09:54:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed057154932edf01b977449b3d444973f321c93d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 11:39:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/720bbacfa57b4157bda2ac452d1d1686\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46aa2e87f6e40ab9a14f2454259939b : SUCCESS in 5m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/196f15d2477c488ebf38dce324d5dcba : FAILURE in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a16714222b254fdabae0c40304f8e0ca : FAILURE in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e57d2b97b7a4a28a6f7e1221efb6c8d : FAILURE in 2m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/182fd6f2c6064edaaf4e7a9ca651aadf : FAILURE in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00be3e984e6446c9844963db656cd915 : FAILURE in 4m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f22e5d7b87404e15a8e1e390441e9117 : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4014f9cc9d343179f360fca7538feb2 : FAILURE in 2m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b514396c7ead4e5d99e636963a106b37 : FAILURE in 2m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d1423f8399742faa03ed1001dba17e5 : SUCCESS in 15m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c3f7d655b5ae4b5e90ee0403e76e26f1 : SUCCESS in 25m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e5cb99ef7c948739c23af57c8bb0799 : SUCCESS in 12m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d4d839aa76764bb6b97042f2dcaf1ad3 : SUCCESS in 25m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec5e081c687f4823bfdaf1f843c1c8bc : SUCCESS in 12m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b5fea6f884324842a8ed63cc55e44b63 : SUCCESS in 1h 00m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/980555d9044d4c4885d3518d7371c3ed : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5e27db4f487f44759e96e5512daa926d : SUCCESS in 1h 09m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b490ac684c642e1b2efd2f6d967c2e1 : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3cd975a013a64cd89e5a9b3f03069ed1 : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/451235cae4284a59ac9042c8b35c7ff2 : SUCCESS in 1h 43m 25s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c9fd926489b54b06ba3a320f9e39d9eb : SUCCESS in 1h 32m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7820cb3aca9e45529083d5bd14720569 : SUCCESS in 1h 30m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e452df8a56344f7bb4c85e1206ef634 : SUCCESS in 1h 37m 51s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/de2def35424f4c518f97d26038a993b1 : SUCCESS in 29m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2e5562accb7543428e9c047c2fe157f1 : SUCCESS in 14m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f92446883dd47d09ae92f9b31a73e47 : SUCCESS in 24m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48420e2826024c2abca4ad5b02bc2537 : SUCCESS in 21m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2075e8c0ebec447ebf6a13c24b122a4e : SUCCESS in 15m 29s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5a12ba01624e49f995cb39531c186694 : SUCCESS in 17m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54b1cccccba84988a7421e41ad347709 : SUCCESS in 29m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9f1236338374e42960e3fa1c48c7ab5 : SUCCESS in 29m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1d785907eaa54936880f09ae156b2626 : SUCCESS in 19m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8aea172ff06e4eed9f5c951789704339 : SUCCESS in 19m 36s","accounts_in_message":[],"_revision_number":1},{"id":"26e0782d4646e68a5d06260a48aee517d3100170","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-28 15:01:25.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2df239b81de81fcee8c3187886d244cd55736381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 16:42:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d4daa94d3804930b0af39a9f0fe0d8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590505a51e5e4ad19385b67960db424b : SUCCESS in 8m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/422b6ec91da14e35948af6eac7f33667 : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70114b7df7f944f0880a752e6de0dcce : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33683ea5513d4f939c747779013a9ec6 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95385a54c62942adb7e95c121d02c226 : SUCCESS in 5m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fc3bb33353524032a40220fdffdeb909 : SUCCESS in 7m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/528696c8abd94e8591df45fb7f5ee3a2 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5432aeb0beca47caa442f93862312b42 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1654d85d02bd411babc54204b29463e0 : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b62b8bd24aa442dd875fec966f75e99d : SUCCESS in 7m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/09a06cf557f64527b5480212b4e3985d : SUCCESS in 18m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dbbe7d9e1d74569bf868099be38186c : SUCCESS in 19m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/95187a18b99e474081dd53478fb4f701 : SUCCESS in 13m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65109bd004e1469e8b91520bb6f6d624 : SUCCESS in 20m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6d154434bb67431fa24340aa065b67b3 : SUCCESS in 1h 05m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/476840fecc284ca2ab65b0a6542995f7 : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0bcce3e51c4c4b53906a303b248354d4 : SUCCESS in 1h 05m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a44d22f918a489cba6c9866199a7718 : SUCCESS in 1h 08m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4e25eba34fd146378bfb97f0cebd4be8 : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/10c386f634834ed58578e35e17e94231 : SUCCESS in 1h 39m 11s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c092655aed5f4b19bc49e5a82538b794 : SUCCESS in 1h 35m 05s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f672b387b6442d1a7c12750b2b6af19 : SUCCESS in 1h 36m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4df531c681e412ba7cc79f7ce4fa9fd : SUCCESS in 1h 38m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bc28eba6f1bd467c9f6f508fb0fb8e6c : SUCCESS in 16m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1f6c228266d04fb2b71a6d7ca9303cbe : SUCCESS in 29m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b50b38d911ae40aebbacf76601979464 : SUCCESS in 22m 44s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/82759b14c2ee4e3cabb123f376849883 : SUCCESS in 21m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cba3ce53751f44e4835435cd29340a04 : SUCCESS in 27m 43s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b4a20eaebe62467e8451ca2cbc450bf9 : SUCCESS in 18m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a34382dd01148288f56cce74c57c784 : SUCCESS in 17m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53b113e639fc4a76bb27a4d9cf79155c : SUCCESS in 29m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/96aa49a39c8b4e4585200cfb6d447440 : SUCCESS in 15m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79986ef4993b465a923384b29e4206f9 : SUCCESS in 22m 14s","accounts_in_message":[],"_revision_number":1},{"id":"790daf21ffe1a81fc774f1048d04405e3fab1484","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 11:22:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bbb9e5a1ab841e9fbcf8209b683b80ad2915adc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 13:06:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efbae6bb5ce945acad3818b3ae755586\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c36dffe2e944a12968de490d07e73ec : SUCCESS in 8m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/50faa835d95b41b491f9c5fed297b793 : SUCCESS in 5m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ea344d9b668460ebc1539ad5277c356 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eb7550753bb4bfdbe0b560e8110ea13 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/591337a4c3004a69bb169f615c115214 : SUCCESS in 5m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ec7fa88cf7764628bc17837a7ba9a091 : SUCCESS in 6m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21b591e8f29145f9bf3f11ebbc1ffde7 : SUCCESS in 4m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/832ead5488d94742ab3961061cc8d311 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fe9a563c70f4ca9943931b3181d6ac6 : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86d836050d0748cabfa54d03525a7cef : SUCCESS in 13m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4c92946b445b4fc6b3c0d09915c458ef : SUCCESS in 15m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a657c571273f4ba6a9bca6df33c3c34c : SUCCESS in 12m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1336c1fe4cff441db984ec945a24c82a : SUCCESS in 20m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc5ab30159c1438aae07b836d03c59f9 : SUCCESS in 17m 09s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5584cda9e3b543e1b3580331b0865b7b : SUCCESS in 1h 18m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3f01ae7289374cf1aac518ad329afae4 : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f3381522749c41439915d7c4309e49d9 : SUCCESS in 1h 01m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbb27a5ca16e4a5790dc58379e4c5440 : SUCCESS in 1h 04m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f9fcdef8a2b64ff9ac87e9a3b4e9ca14 : FAILURE in 6m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0c66b278e4af4bc4ba3bd247db498f90 : SUCCESS in 1h 42m 39s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/599a56f986c94739ac59f03044c8032d : SUCCESS in 1h 32m 13s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/404227f009254f84b035d663ef244bbe : SUCCESS in 1h 34m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9884c685005d4542a9ba4d20f415d4d2 : SUCCESS in 1h 40m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/89418ab6dffa43fa82844f67f78a1e76 : SUCCESS in 20m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cdf1d81908a8453384d2b79b951fe352 : SUCCESS in 28m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cd5e31fb32f44b5a95b11f55f7eafc4 : SUCCESS in 22m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4c22b2042d96430f9eaebb8ca4633f15 : SUCCESS in 20m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3721f88a19334531aee7fdac9c570e61 : SUCCESS in 28m 02s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/747ad0bf6df14eab8ba5d187f9d9f238 : SUCCESS in 37m 34s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9fa0f90b66f4d84a1abdd40b1863523 : SUCCESS in 29m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4285f0694ddb442fbe1192cc9eacb6ef : SUCCESS in 28m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ff9de41440d143d295218dd1688c3119 : SUCCESS in 27m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25c95a399fbe43a08b1e18e80c192c67 : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ef59bb0c71764f025bdbff16ba7f5a1206b82282","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 16:06:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86be3d4c7c37564d4248ceace77d5d36b2dd4f43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:36:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ea43396401541f1833b0d7c5a448c14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ebe4fafa8b849f3afffc94d7d06a305 : SUCCESS in 5m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/46068bcb34b24ab6877a1080402107b9 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0943e9530af64bfbaded242dcb8de550 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8183144e95c41c0b5aaef0ca8faf443 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5886ee3c77ee4642a74b66ebe2288d7c : SUCCESS in 6m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/209a12ab8cb04d75ab40ee160fd13237 : SUCCESS in 7m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6261f208895c48ff98ec9732737c6923 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1020140602064cefaff8732698ac6ca4 : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b4f33c8d67a74ec7a4ad9826549cf935 : SUCCESS in 7m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b73fa13320ac47caa260fd6ab64ea8bd : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9886805c8fe44bf7b57d6165805288f2 : SUCCESS in 14m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b634e0c4110e4f1db4b98c3a5c98d4c9 : SUCCESS in 12m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8d94e501c7224c5f9767a3103c66370f : SUCCESS in 18m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7d81ee7dc7d42109ce9947637f63dc2 : SUCCESS in 17m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/52604ec71f84484f8a8d3b5b0d5d4f5b : SUCCESS in 1h 03m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/746266c65df84c0fa8a83a470f207785 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e5bb957bab0e4b6ca90727d24b493d5f : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8214da5136a472eb0a59c548e92cce7 : SUCCESS in 1h 08m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/46befc36d5a84d1fa273ad09bfa4883f : SUCCESS in 1h 11m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4170408562314bbaa045b5581d896832 : SUCCESS in 1h 39m 05s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/73772624956a4c648000c9cf57b8cc69 : SUCCESS in 1h 30m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/663995c061e24765ae8c4ac2690cf7f5 : SUCCESS in 1h 35m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c239c4753f7143d08960eab33b282868 : SUCCESS in 1h 40m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3b974f79fe2f4df1af8aaad0f8ea1547 : SUCCESS in 21m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/07e04ef4ee8e4582ac03770fa39e2c27 : SUCCESS in 25m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ef88027783647fc8cfb32386361f7c9 : SUCCESS in 23m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d471e3822be48d78f87382f8bb1ecef : SUCCESS in 20m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8fa210ba431d423b9c61154f0b79d945 : SUCCESS in 28m 54s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fb2fd3538f0244b5b2ac2bfe39d7560b : SUCCESS in 34m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/561dd338de8448bb81dd7d3ac5bdaf43 : SUCCESS in 32m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c688ae82eeb4b659d11059864dbc0d1 : SUCCESS in 27m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e8da7ecfcbb44d28b3c4021b9e200b0a : SUCCESS in 19m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db06be401972490e8cf90aa68fe7a0b9 : SUCCESS in 18m 01s","accounts_in_message":[],"_revision_number":2},{"id":"e74dd6d977317c60ea587fa4b65a33625db5a4e5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 14:44:25.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"dc17e678699b6c711f1fe24bba05ed2799a8ed9e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-12 13:01:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"99366f214237e73727bd1bb2d66384530c674b9d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-14 15:22:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"078bc2e922c00be5ef639f7a430874c5b840ccd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:22:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d61cffb92472c8cf455ae2b8e149f0f4c631e16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:10:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f33cdc37b4a34bc3aeacfe2647cc8219\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7bfadac97674582b6372675eae3a78f : SUCCESS in 10m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ae5ad1e02f994743918dfee34934bd0b : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/773032d0f0de4aaabbffb0165e588f92 : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d281779006334797aacbf2c541fed321 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/642af0841d764381bc3592f3c1a781b9 : SUCCESS in 7m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4518558a66f9425faa7a0bead672e8f4 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/65660af6d59a46b6947b045a22daa639 : SUCCESS in 12m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/95275795ab24457b8a944a67e756a42b : SUCCESS in 14m 24s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/07ea3e87bd0f4550b42224a4b78053db : SUCCESS in 11m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/639e959a679f4b21903c3465c63aa2b9 : SUCCESS in 16m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3f48ee73d3b9452bafd2c254134e8cf2 : SUCCESS in 13m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8492bd06c424769babc0a9989786c83 : SUCCESS in 22m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3826376eb0cf4bd9a79df396a787f3ec : SUCCESS in 11m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f87da968b9cf47848dbd03c9c93d1796 : SUCCESS in 20m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bcbaac24f97d480bb8a29cca8f082956 : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/df91639bb0884911bad6fca1bdc2b2c4 : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ad973cef05d94ee5837b99bac7885bb8 : SUCCESS in 1h 12m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f819d357d877461b8c24fcb3f58cf990 : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94a501cc0b904fed9d1f55b87a425c98 : SUCCESS in 1h 12m 40s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5864d8018b5f4f59941fee9a7cdcc395 : SUCCESS in 1h 36m 25s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2f4d0844cf474b12a739a502619720ba : SUCCESS in 1h 37m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/970fdc0a46924d1f980f172837ba08cd : SUCCESS in 1h 33m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8f9f24425fd4b168a800acfc5c6b624 : SUCCESS in 1h 44m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/61e61b51957f4dc38e7c53cd3a0648cc : SUCCESS in 24m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/083635c900174a0585bcd075e91124f5 : SUCCESS in 25m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ce897f7a77043fbaddfc561917c685d : SUCCESS in 25m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ac4de0e32f7048a080ac5487ff8ed781 : SUCCESS in 12m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4fef95207b5c4c01bc2938f63a0eab4e : SUCCESS in 29m 40s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a1eba40a17cc45d19de789f2ea268488 : SUCCESS in 17m 45s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04e03b4b87b042128a10a3590c96ddfb : SUCCESS in 15m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c572669bada24161abf38d276d47a61b : SUCCESS in 32m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5122cbbdfcec4b83801d13f7b7a51ba7 : SUCCESS in 22m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fdb8af119ec42f5a30284ad9408906a : SUCCESS in 19m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a782bed95413b37f543f48ddd308e098735cf872","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:10:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ffcf832db86960eb487571ad568160d324fc7a0","revisions":{"1aa9eadb127ab85a468611cb68de412355bf6dc3":{"kind":"REWORK","_number":1,"created":"2026-04-28 09:54:35.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/07/986507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/986507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/986507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/986507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/986507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/986507/1"}}},"commit":{"parents":[{"commit":"62fd7413f6401cff2d4f0d2c502e78dc92eb9dac","subject":"Merge \"Add nmstate VLAN QoS map support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62fd7413f6401cff2d4f0d2c502e78dc92eb9dac"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-28 09:51:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-28 09:54:27.000000000","tz":60},"subject":"Add kolla_bifrost_inspector_default_node_driver variable","message":"Add kolla_bifrost_inspector_default_node_driver variable\n\nExpose the Ironic auto-discovery node driver used by Bifrost via\na new kolla_bifrost_inspector_default_node_driver variable. This\nmaps to inspector_default_node_driver in bifrost.yml, which sets\n[auto_discover] driver in ironic.conf.\n\nThe default is unset, preserving Bifrost\u0027s existing default of\nipmi. Operators can set this to redfish for modern BMC\nimplementations without relying on workarounds.\n\nCloses-Bug: #2137417\nChange-Id: I48ccf4fb40036ca25cd00050342fbd0b6e6cfad2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1aa9eadb127ab85a468611cb68de412355bf6dc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1aa9eadb127ab85a468611cb68de412355bf6dc3"}]},"branch":"refs/heads/master"},"3ffcf832db86960eb487571ad568160d324fc7a0":{"kind":"REWORK","_number":2,"created":"2026-04-29 11:22:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/07/986507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/986507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/986507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/986507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/986507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/986507/2"}}},"commit":{"parents":[{"commit":"38ac2d96cb8c89343b6b1f06fdb096cba12d9f7c","subject":"Split out lint reqs + stop using upper constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38ac2d96cb8c89343b6b1f06fdb096cba12d9f7c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-28 09:51:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 11:22:46.000000000","tz":60},"subject":"Add kolla_bifrost_inspector_default_node_driver variable","message":"Add kolla_bifrost_inspector_default_node_driver variable\n\nExpose the Ironic auto-discovery node driver used by Bifrost via\na new kolla_bifrost_inspector_default_node_driver variable. This\nmaps to inspector_default_node_driver in bifrost.yml, which sets\n[auto_discover] driver in ironic.conf.\n\nThe default is unset, preserving Bifrost\u0027s existing default of\nipmi. Operators can set this to redfish for modern BMC\nimplementations without relying on workarounds.\n\nCloses-Bug: #2137417\nChange-Id: I48ccf4fb40036ca25cd00050342fbd0b6e6cfad2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ffcf832db86960eb487571ad568160d324fc7a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ffcf832db86960eb487571ad568160d324fc7a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
