)]}'
{"id":"openstack%2Fkayobe~986628","triplet_id":"openstack%2Fkayobe~master~I65ea59693acc359fcf4062e9a94a1d73f82c86b0","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-13 18:14:28.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-05-14 12:10:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I65ea59693acc359fcf4062e9a94a1d73f82c86b0","subject":"[CI] Disallow broken conditionals","status":"MERGED","created":"2026-04-29 10:47:52.000000000","updated":"2026-05-14 12:10:56.000000000","submitted":"2026-05-14 12:10:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986628","meta_rev_id":"15ba97f09e43c7a1dfac5b00532f1fae798d7426","_number":986628,"virtual_id_number":986628,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-14 12:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-05-13 16:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-05-13 16:04:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-13 16:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-13 18:14:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-13 18:14:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-29 12:42:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-29 18:49:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-07 08:26:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-07 08:26:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-05-07 08:26:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-07 08:26:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-05-07 08:26:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-05-07 08:26:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-05-07 08:26:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"1bd22da931d6fa15fc8ce435b589048f75d9c826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 10:47:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1eaa1baaeb919492d5080ea443d75c1906f111e5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 11:36:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6eecee7f492e8826aed2e78a142c273e1c93825c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 12:42:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70be4bed103349e2a736f846dbb301a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/626f0089b7604544a2cecaffecd2f517 : SUCCESS in 10m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/36b810d565944fee91b5e68c823e57e8 : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f0901ff4ee74da288afdc7f8735fe28 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fec96c7b28b484eadf5fe4e74a2aab0 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8f379455fb24edf862e628db968e7e4 : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1f4c426f46204b8e96f48ad021d16447 : SUCCESS in 6m 44s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a2dad2283034b4285f40cf8b48eb953 : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7dd4c98b108c49858e03dda4dcb6b8c3 : SUCCESS in 14m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/565b2f4ad114443cb90bf3e33bff7354 : SUCCESS in 10m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/44e5bc3e582d41968450278b21c0dfb2 : SUCCESS in 14m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06f0f7ee88274488a34c341b5781eb95 : SUCCESS in 18m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1d4a7254fd994b87a09f39bc300026a0 : SUCCESS in 25m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6102f67233e4a27b403795dd946bc8c : SUCCESS in 16m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cddd3fc36b194f34822a2778150bd087 : SUCCESS in 1h 15m 55s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/35138368e17a4602b6aaae8fb3eaacab : SUCCESS in 1h 14m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2df8ebc785924edebac1ed401ea3d6f0 : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09ebacce86554938a48fe781a476db2a : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ac40d8d6f0994096995f00c45d384e77 : SUCCESS in 1h 14m 58s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fe1f5cc2fc9b418d9fd453a49f37cdba : SUCCESS in 1h 47m 22s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5fa75417f3bc4028b177ffa5729798a9 : SUCCESS in 1h 46m 04s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e859c9883e704b3783792f32cb17127e : SUCCESS in 1h 32m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/666a3dacb82a4c03a39392facce3b5a6 : SUCCESS in 1h 50m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a79ab99f28984d0586778a74422cba48 : SUCCESS in 22m 51s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5e86bcce7e0d48828cc769ed5eacc714 : SUCCESS in 19m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e4ccc7e09444fefbc307efaa9c2f65f : SUCCESS in 17m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2fc7203249c74cbab00656c068562efe : SUCCESS in 20m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e83965a0541e4feab0654c941c66e336 : SUCCESS in 18m 33s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/969474725b594fed8a836811d62d466d : SUCCESS in 25m 19s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8477f4654c3843f0b9c45d786df4fd65 : SUCCESS in 31m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57c68e606a8e4c31b5ed138c7fb0ee2f : SUCCESS in 28m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2a4f109bf0794a1cad95713de4067999 : SUCCESS in 23m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/488f56fbec05497e8ed85571d63caf3e : SUCCESS in 20m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5baef75efcce47a822d8f1d45a0706a508784400","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 17:39:45.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"f0fd34bf4a76be14955c723abf7b04d38fcdf765","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-29 18:49:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7d4921cd9df14117a43b18c53d85cc41e781d866","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-07 08:26:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8e963cf43b7b08a037eceaaf66c22df3c28c5644","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-05-08 08:23:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"511bbc0f209ca27e76b27891196a201d07b4bfaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-11 11:56:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c8d50e606fc9fc891fda39024c31dbf71c41f238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 13:45:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a693542be7fb4bc8963af256632d5489\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76a1a3be150b4584819ca412af3df7ff : SUCCESS in 10m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c54c700fb2f44e9b979a4e06c4d9ac06 : SUCCESS in 6m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68e5b499afc44b94961d40dbf3dd8c51 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3bd3b4bb05240ea8fadec637252284c : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d917bbd108e04a539f9d27b7af9cba91 : SUCCESS in 4m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/925366a701514f2398f67ec85b42cdce : SUCCESS in 7m 14s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/911b469be86144ebb917e5882cae11de : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d3944c94ce964860b17d33bcc5a70278 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2421a567d08546709e49981012d53795 : SUCCESS in 12m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a04c01eb14574d5e9f39b0d4edb58ce1 : SUCCESS in 25m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/564cdb049307427aa10645a6ed3ec7ec : SUCCESS in 19m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/054dfa1971c24d979ff5386ae23b9ad1 : SUCCESS in 14m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d18806d2c407434b987125b53c423c5b : SUCCESS in 18m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/da6cc4acc9e345a98116250ca57da20b : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7b7cd6391c65459c88385424b2644096 : SUCCESS in 1h 02m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7c00d01ae0a34df994e89cbc59b96dda : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f7aef6b03584572a455be5c8c32b1c7 : SUCCESS in 1h 06m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e2fcd699ba2745d78dd1f822f805fe84 : SUCCESS in 1h 06m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8ce64b50b92f44dabfd6710b94da763e : SUCCESS in 1h 39m 09s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d6dc1570831b4762848f195b309f8315 : SUCCESS in 1h 32m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f52017bc26fb45209189ff27913fbee4 : SUCCESS in 1h 31m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f513629def14e3384c138589926acf9 : SUCCESS in 1h 44m 40s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7fb3e0a86a354a9aaa34b4d349fb9406 : SUCCESS in 24m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bc5a973adafc488e85b81fa2043d9086 : SUCCESS in 21m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9151e40a35cc4c4493648552a2a7d400 : SUCCESS in 24m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/23ed22a8605f4b2987216b6f44af4592 : SUCCESS in 20m 46s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1a59808fd02944b4a4022471af444701 : SUCCESS in 24m 18s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/839a090679e5486ea52fd792efaaad58 : SUCCESS in 27m 13s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1985574114104045aff7e8e16d672d5b : SUCCESS in 28m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45112f521b5c48858fdf794c2533fece : SUCCESS in 29m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2fbec77e475649e2ad59a199c2bc14f1 : SUCCESS in 20m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27d2c86b7e354128a1acda0a31e45fed : SUCCESS in 23m 03s","accounts_in_message":[],"_revision_number":2},{"id":"aa3dbed8676a8c3738b09272610fbb0950bc9b88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-13 16:04:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e9d966995e49fe40dc1ba679b450f9d3af06d081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 17:48:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f9b9ffdcf734153884883b9c4f090c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7340d8cea3a490fa9ca238531cdb333 : SUCCESS in 7m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e6842a7420304fe7b8d1fcaf884d07af : SUCCESS in 7m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efcac30d52dc49d9adbb57fbce975be3 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62134511c9484880a7080509d693da83 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f4e8cc6a1f344bf9470ed26cf051205 : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f6d91a7eae94bf38214becb21bb786f : SUCCESS in 6m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f214cb71ca164bc3a452e5f930a65d4f : SUCCESS in 6m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5c230ff2203f4bc5898c1b3e13566b3d : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c73e83a5da6e49e18686427a7329a8d8 : SUCCESS in 12m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91aff4d5430e4f4ab0d083cd086e3135 : SUCCESS in 7m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fac22c1d994149a6a0c3510cb99adb22 : SUCCESS in 24m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9dd3aab5ad9e4e8b87ca968a4d88351a : SUCCESS in 15m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/409379255f264a9cb27c2427d931f60e : SUCCESS in 19m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ea8e2c2d9a14aa085d1b834379e6110 : SUCCESS in 11m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/da2c6666c51f498bb2be120d3c760890 : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9161f38e1c7a4340b51f892d04662a71 : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b3f4b88da629411ba3ea37060e17d627 : FAILURE in 1h 02m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e9a3ca76ea2443e801ca6ca7e6d8346 : SUCCESS in 1h 05m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d1bd03b542764d14bf52600a89d93d78 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a3dd753a0d7d468bbbe96dde12e30a1b : SUCCESS in 1h 38m 39s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/191410f785974cf39b4a0246e0b426c5 : SUCCESS in 1h 31m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13a7257086874d42b615709fb285ce8b : SUCCESS in 1h 29m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1589ed0b15b4c1ca20b7da2db4a85f7 : SUCCESS in 1h 42m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/58e02dcb60f34bacab6a4bd143d0825b : SUCCESS in 21m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b257e7f41f3e4859b7ce95d79572fa4c : SUCCESS in 23m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82680788ba32459088bdb95fac5ed75b : SUCCESS in 26m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b8c6e00eabf8482d91f2f19c88e0f4b2 : SUCCESS in 13m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/374847ba6a024c82997dd6b308b0586c : SUCCESS in 21m 46s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2cad982e08bf4acf88ac3608bf46ed16 : SUCCESS in 23m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50841913ef0d4c609f324c257442c449 : SUCCESS in 31m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d71abf8b3d04cf3bd9379080f6130df : SUCCESS in 24m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/216cb0e954e54e91a75f6e9a2368470a : SUCCESS in 18m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/668d7ecbc376468bac9021f9468e659a : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":3},{"id":"64f988e9fc693c3af9026587a88730106d4d2422","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-13 18:14:28.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"099e1bcfa8a8f8dace79a65b954cc5a341e372db","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-13 18:14:45.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b8a292313decaae64712b90db7ba18bdecff6409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 19:57:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42e8fb3240114e888f61b6b953faa51e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/226deda289b0446eaf6f39c6cc744068 : SUCCESS in 10m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a62d68f6b5d84a428b129e6c610eacbe : SUCCESS in 7m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6129f90344ca43cdbaf0fec2f2decbbf : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d2e5ed640e94c4f914c8aa65c4ca640 : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fa3a354091249c7b76d20edd9d0a679 : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6b83c0a3545341b2ad4f216eeecc3c15 : SUCCESS in 8m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a76f58ca7e87416aa457b12a17b5103f : SUCCESS in 6m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2fa8345353f14fd692fc4599bc0f7503 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a9523fc585f744fb80a13ca3f4d0ed81 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5d43363467fc49369afbed953f8107f9 : SUCCESS in 11m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a028fab2923a4605a5d22bec35d409d1 : SUCCESS in 25m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a0248a52f2a491592c667f7c192cfa0 : SUCCESS in 18m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/577d12f69d5a46b9a5296e69e2de75f7 : SUCCESS in 12m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/838c0f89247546a888cc323d2d077c5a : SUCCESS in 17m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cf6c63d566454619a87cbf9525675921 : SUCCESS in 1h 10m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a2945adae0854ab7a9663f1b52a612d0 : SUCCESS in 1h 11m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ed951cd4c38447348dd4bc66f9d5e712 : SUCCESS in 1h 09m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6610a59497043069b7ebb936d24c623 : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0f866bd06db44744bb46916ddb076cf3 : SUCCESS in 1h 05m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6be45b43dc944d89a3e437dc27804b5d : SUCCESS in 1h 34m 15s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/20fbd6572e174624967fdcb85f0e52d3 : SUCCESS in 1h 40m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/835f8f55b5274a62aa22a2ea3e68aba2 : SUCCESS in 1h 31m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cf2643934f840f78b511c4f885d1d59 : SUCCESS in 1h 38m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/dc55c9bd9c404890a3de807c9806000c : SUCCESS in 21m 23s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a908da61cc72438f93dd5c1f029cc70d : SUCCESS in 14m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c48caef9be5a46d5af4dfd657ca46561 : SUCCESS in 24m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9b6a9d8d6dce48c5bc1114c1aee3971f : SUCCESS in 20m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1558e52880b34f8593c051cbeecb3042 : SUCCESS in 36m 17s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d562baec95d441528b6c20d89132c8e2 : SUCCESS in 27m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/549998d7efde4b1d997f3f3cdc5078f9 : SUCCESS in 31m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee26569f53454d478f380ecd0ede641d : SUCCESS in 32m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ce8686494ade4dcdbb9bd19c4f66d9f7 : SUCCESS in 22m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb223113db044da9bee125cc8ae8b23e : SUCCESS in 19m 59s","accounts_in_message":[],"_revision_number":3},{"id":"1ebc0a36f770bf11bcf80891927b476f0bb67b5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 10:18:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f8aa51bf44fece4e5363dd5bbcb286931a9e8627","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 12:10:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9d1b0e147344e7cb0e99388f572bce6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7f3f54ac66423582e5b7348d32cbb7 : SUCCESS in 9m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/758241be93dd41188d87d8f6d5ab4fc5 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b7ab6cc3704d11bd0ae0a0c9e32778 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fa6bc315d714c93b904c5ac9be2a064 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/996ec101c174478b9242068c1946165a : SUCCESS in 5m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/152ce7f194854153b4e9f9e03ef460a4 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d8f24dc63674fe18b1ac16bfd0d0bea : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4a9a844b1e304599aa0335e00277183b : SUCCESS in 12m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7b69cb65869b41629c5dd4f77973abdf : SUCCESS in 25m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2be3d6d155642709d3f53c1cea98048 : SUCCESS in 18m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/da77e56dadf64b1584dfbe00157f75df : SUCCESS in 25m 02s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/869647ad1591443d9110c60d73c2213d : SUCCESS in 12m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b0cf9d22c18d48ba98f91f41b9f7cb93 : SUCCESS in 53m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f2e04be3b01a43ce98db1fdd4da4ac2b : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9237e41f166b4f5184cd19e36af13c41 : SUCCESS in 45m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdf5daf0f2ca443eaaca7ea78cc222e9 : SUCCESS in 1h 23m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2490c3f25f144dc0983fa0438554995d : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/06945983879e4cf3ace0bb8430acc865 : SUCCESS in 1h 47m 09s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5087c734e82748d6b77d61576ca44d0d : SUCCESS in 1h 03m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15f64648b1f44e4da72e6b4efb0e0177 : SUCCESS in 1h 29m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b785ccec9922483e90498349f95c9e1f : SUCCESS in 1h 42m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b72a1dead96a4f62abc67c600c0bae87 : SUCCESS in 21m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7897fd2de33944d6a8f74b3099f9f882 : SUCCESS in 24m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfc807b604a343d5984d74374d073a47 : SUCCESS in 25m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9cbb9fa2f3a74a11a60e207aba5097ba : SUCCESS in 20m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/966b61d364b94a26b1881b18b9ebef40 : SUCCESS in 28m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/948424abd64d422b9fa97d5fc809e132 : SUCCESS in 29m 48s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3900a8415c39456cbf742fdf05f160ab : SUCCESS in 32m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42c64604bc1c4783b0b2bdeb0f6d129c : SUCCESS in 31m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5d60eaa7252e439e9f2ee82d5f0c7b8d : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4c0f7a4bc2246368bba5fb537e539d4 : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":3},{"id":"15ba97f09e43c7a1dfac5b00532f1fae798d7426","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 12:10:56.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6abb9842281aee76481169f5116c82d05736ee20","revisions":{"a4c960be3f4c0dcf6e73fd3a4286865937666af2":{"kind":"REWORK","_number":1,"created":"2026-04-29 10:47:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/28/986628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/986628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/986628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/986628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/986628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/986628/1"}}},"commit":{"parents":[{"commit":"bc29d2a8354d6d0c9ea4bfcd3252c7efc850c7a7","subject":"Fix non-boolean nmstate zone map conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc29d2a8354d6d0c9ea4bfcd3252c7efc850c7a7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 10:40:19.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 10:47:46.000000000","tz":60},"subject":"[CI] Disallow broken conditionals","message":"[CI] Disallow broken conditionals\n\nChange-Id: I65ea59693acc359fcf4062e9a94a1d73f82c86b0\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4c960be3f4c0dcf6e73fd3a4286865937666af2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4c960be3f4c0dcf6e73fd3a4286865937666af2"}]},"branch":"refs/heads/master"},"1d0d8d97761c701aa90d63cfcc36c9dbf0560b63":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-11 11:56:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/28/986628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/986628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/986628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/986628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/986628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/986628/2"}}},"commit":{"parents":[{"commit":"d3fbffd397a737f19e169846f0431b4a3c730371","subject":"Fix non-boolean nmstate zone map conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3fbffd397a737f19e169846f0431b4a3c730371"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 10:40:19.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-11 11:56:53.000000000","tz":0},"subject":"[CI] Disallow broken conditionals","message":"[CI] Disallow broken conditionals\n\nChange-Id: I65ea59693acc359fcf4062e9a94a1d73f82c86b0\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d0d8d97761c701aa90d63cfcc36c9dbf0560b63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d0d8d97761c701aa90d63cfcc36c9dbf0560b63"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"a4c960be3f4c0dcf6e73fd3a4286865937666af2","theirs":"d3fbffd397a737f19e169846f0431b4a3c730371","contains_conflicts":false}},"6abb9842281aee76481169f5116c82d05736ee20":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-13 16:04:29.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/28/986628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/986628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/986628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/986628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/986628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/986628/3"}}},"commit":{"parents":[{"commit":"5de0ec85fbd09e50f2e809f2cee5d209dcdeefa6","subject":"Fix non-boolean nmstate zone map conditionals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5de0ec85fbd09e50f2e809f2cee5d209dcdeefa6"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 10:40:19.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-13 16:04:29.000000000","tz":0},"subject":"[CI] Disallow broken conditionals","message":"[CI] Disallow broken conditionals\n\nChange-Id: I65ea59693acc359fcf4062e9a94a1d73f82c86b0\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6abb9842281aee76481169f5116c82d05736ee20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6abb9842281aee76481169f5116c82d05736ee20"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"1d0d8d97761c701aa90d63cfcc36c9dbf0560b63","theirs":"5de0ec85fbd09e50f2e809f2cee5d209dcdeefa6","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
