)]}'
{"id":"openstack%2Fkayobe~986696","triplet_id":"openstack%2Fkayobe~master~Id6373aa5eef515f5d6bd3176f71fe8b35333a009","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-18 11:08:05.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"hashtags":[],"change_id":"Id6373aa5eef515f5d6bd3176f71fe8b35333a009","subject":"Stop gathering facts in trust-store playbook","status":"MERGED","created":"2026-04-29 17:11:21.000000000","updated":"2026-05-18 14:22:29.000000000","submitted":"2026-05-18 14:21:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986696","meta_rev_id":"d0d2b12bcbb4520f46b9d9492d5998e9e3757736","_number":986696,"virtual_id_number":986696,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-18 14:21:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-14 15:22:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-30 15:28:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-12 08:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-14 15:22:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-30 14:44:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-29 19:41:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 15:28:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-12 08:35:59.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"b4c221c6731c4199fa2db33ef761311770fae57d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 17:11:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01a87e42f8e185d24bd8f6371d11fdcb3727410b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:41:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df534577bcc0454a98a95127794dff35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5016e9bcb544707877476a01963bcc5 : SUCCESS in 9m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5857ecc1ec9543f6a4bbc44df114915f : SUCCESS in 6m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6239f02948f242ea862a4ee734b0e585 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d3634abaa254eb7863ee808f22b88f8 : SUCCESS in 6m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d678c51fdeb6409aaf6cc6ff68f6e5f3 : SUCCESS in 5m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3829f155b93a440d8fdf7e138b7c748c : SUCCESS in 6m 42s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c336321fbe1a4f5aa817a4b2f3e0a751 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/229f489eec264fe78b69f29748478a7f : SUCCESS in 12m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/83f7678c0cf44bf789ff8b1d63061058 : SUCCESS in 7m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/29a0a4a1c0f942788fb51625ab7354fe : FAILURE in 9m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea39fbee8b354fc4be8e2f93297eaa48 : FAILURE in 5m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/aaa8436a5a0344c69bb87d1f2f9546a5 : FAILURE in 12m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a21c404be3848e1ba4ceb747521a394 : FAILURE in 9m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/56c736523dc24a18864365c54290fd6c : FAILURE in 7m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b1a2134b59b84c2c88e3a4c41d3b952b : FAILURE in 7m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/15bf182f151a4a72bc20b86112236dbd : FAILURE in 8m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fde261a70f93407e8d6de9f9d1cc69f0 : FAILURE in 8m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a36a24ef353e4bc39f16c47cbd54942d : FAILURE in 8m 15s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e83722594241452d92f834eac8d9d5f0 : SUCCESS in 1h 38m 20s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8c043e1a067c4a67abe5952da0ba57f4 : SUCCESS in 1h 32m 40s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbd25d35576c454fb098086b0a837e8c : SUCCESS in 1h 29m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d88086bb0d4846688aad1c7cc97b6f65 : SUCCESS in 1h 36m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cdb037b7570d419485fdbc506e56551d : FAILURE in 9m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9fc3854509df417ab412cea097fb6af6 : FAILURE in 5m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a5408666b6b47d89fc5dccd1197c6fe : FAILURE in 9m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4f04b5168cf246d7ab09467d40f8b313 : FAILURE in 9m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8b02f4a0e44e40a898ae0b2ae24577bb : SUCCESS in 29m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/32867a810c864514974d7d40984465de : SUCCESS in 28m 36s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ca54457ec5e421181f6d4b6fb6881f7 : SUCCESS in 31m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bdc58372e16412680511e340bba0090 : SUCCESS in 27m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3c011406623840898e4bd6953ecde60c : FAILURE in 8m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61627cf9f8f14d0ca34262dbee594236 : FAILURE in 9m 05s","accounts_in_message":[],"_revision_number":1},{"id":"342e79e4d64a597b9562b7e35293983f3bfee0e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 09:12:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c6f60ce6c5195b9e08125e223a0250512ba9900c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 10:56:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb1ed37bb61e479198821a2f2f680bbc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1919b331a54b119afb0b37fc24b10f : SUCCESS in 8m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e68bb9f966674fc4a6013908fd6773c2 : SUCCESS in 7m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a094c5805e9f415d949753cf55c34488 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9bea96afeb0455b9d30f5b7b3c40421 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/891963ee4832461192df307c7d6add4b : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1648aa2a9af9411fbeb21da8c982f9f2 : SUCCESS in 9m 01s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8173163b31ae4abea727711a936e15d5 : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72cbabb34fc0480fbd7f52fd00ad0a7c : SUCCESS in 14m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70f98ca475ce460ab4f8a982d0f628e9 : SUCCESS in 10m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0276248da4ce423484056768ebb4e33a : SUCCESS in 15m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a510ecaf908340bd81ba9a90f52e59e8 : SUCCESS in 14m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/311f5b1a4e564c37ba098e6af2aa7c00 : SUCCESS in 12m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abd33a5ead254f7e94fc8ff93ae0e512 : SUCCESS in 12m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f83ca122d32c4161a0b03440b8718983 : SUCCESS in 59m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/832c945dcedc4e2586a4dc4aaf1aabed : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9fed02622cda4594bbc6d645d7e154ee : SUCCESS in 1h 05m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d9a4535feee41a2ace9b28f96e9cdae : SUCCESS in 1h 05m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e199619a6bfa45cdb24584bd27321cb2 : SUCCESS in 1h 04m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5e9fa784e51946b6b94972b28729d863 : SUCCESS in 1h 36m 40s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/95970f8af9154901bd4e66c0259f92b1 : SUCCESS in 1h 40m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b8bd3fe5e8644eb82a6e48b6de6e4bd : SUCCESS in 1h 37m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64ea7453ec3d43cbb6ce76a06fc5f760 : SUCCESS in 1h 36m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f86a1f44dbaa405aa6c2eb558364c9fb : SUCCESS in 29m 23s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5969bde0c2074fdaa93e5e9a1383cf1d : SUCCESS in 15m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80ee096a76b2422bad3c0d0c747adb99 : SUCCESS in 16m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3f851b91ea6944888a411206423bfde6 : SUCCESS in 21m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b2bd9c6f8be74caf9c91b2d8014f9cac : SUCCESS in 19m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/08c9fbfb11344d50ac72d7a6c94aa08a : SUCCESS in 22m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57d644ace1914c4cb03123713e461096 : SUCCESS in 27m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2735a71c58a485fb42a145e2f7e536c : SUCCESS in 26m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ac311ce37d734318a937b6a6b115d1cc : SUCCESS in 25m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52a6d5c8d8dc4ea09151dee3566b0e7e : SUCCESS in 17m 54s","accounts_in_message":[],"_revision_number":2},{"id":"5f419f3c8daa733d12ac77dabcd9920cc65e6cdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 11:33:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"77002ddf5e5e97df48d6224154fbd72161c9ade8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 13:13:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec7c3be0f212456c9cf08f622ae3f1b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51246682874d426b97650cbb235ef5a4 : SUCCESS in 9m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b66a705d157149d3a8c5ada14fb542b5 : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb7c6adb89a6450794b9c391ba6e1438 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff00972c9f84b629e04ed7033225174 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3563254f54b64ad980cdf4f1e84dc03b : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ae5b2c8cf1b44878d2bdae0e04858cd : SUCCESS in 6m 19s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/66d64884977d47bc826a6d117dd60b66 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/52118405214543deb96ccdc3edc8b3d0 : SUCCESS in 13m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eb878a29db32451b8711ad01552ac82d : SUCCESS in 13m 31s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e9541d267abe46968a68458a8a82bdba : SUCCESS in 14m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/120fa2a27e3949a39552d45d54e1200b : SUCCESS in 17m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e5aeb22b24434b3fbab8d0c91c5934b7 : SUCCESS in 12m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e04b122935714aa99cdc119985ae0021 : SUCCESS in 21m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/28e27d5e59dd4a6e9f4becb7fd6c1110 : SUCCESS in 1h 05m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c0edc6a81f844eebbe378e17a6f747ea : SUCCESS in 55m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/24995cc6d53e4ab296b91322236f55db : SUCCESS in 1h 03m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/403442432dab4fee891d62d7906785fe : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a9149f778caf4f8ab481e04c7f227453 : SUCCESS in 1h 03m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ddbcd3a6510a4fe587756f046878328e : SUCCESS in 1h 37m 48s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bd564dd275df4f4ca381847586edc512 : SUCCESS in 1h 31m 16s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b9c7fbcbccc4bc1bab9f7762942fe34 : SUCCESS in 1h 31m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/667cd09ba585407896ef6f3cc7d84d44 : SUCCESS in 1h 36m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b776605a5c1b4890a4921a03660c1c67 : SUCCESS in 23m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/200e2a258e984878acb378a40f74455e : SUCCESS in 19m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ff0c47afeda437da0eed68bfd2fa7af : SUCCESS in 22m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/362708e988f94781be7635cc2340932c : SUCCESS in 13m 35s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ed0e30f00ea846729346fd7e3ffc0809 : SUCCESS in 25m 42s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/533284f0fa114823a16a637e5f2e3663 : SUCCESS in 19m 02s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/225e901487f341819783cc4b1ca80be7 : SUCCESS in 34m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/180b5be285054a898bf5870dfe57f88f : SUCCESS in 21m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/80585ee41cf442c7ad1a9b517daf42c8 : SUCCESS in 15m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf96626466c74a76b213a90890fa3110 : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":3},{"id":"7f31e4beb2638d0d5d627c0ef6dbcb57935fadbe","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 14:44:08.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"ff059ab6ff9b543138e8c625a483ddd811c5cefa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-30 15:28:27.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f39a7ef170d5b9841964a5cd7c49c3f653938b4e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-12 08:35:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"57a0ac9e13dbcf3c71bda2f9206ab592ffaeeb03","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-14 15:22:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"61a4279cd3d149059ba9cffa97696b9a1d08deeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:24:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f07dd46c5875a25e4f8d6218272ff60b9f8f5c91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:11:01.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe/+/986779 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"c7fd573ecb27aac634d51410cf9bd5685400ee93","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-18 11:08:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8692b13d5857258ed1c156a4455e46cbd68aa9ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:46:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fca5ac0ebc6c4efd9f98f6a370b10425\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef262b77c4ea43c5a68f1e1018350a80 : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e3a6c2465fc7443781932b9d337c6329 : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8acc32537ca49acb0486b75847319b6 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1db722bdc11d4971a75c726d2ca204ad : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba7b7a7e2a7c4c79adb56a69341bbe09 : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/369fda0b9f2f417bb36c07571d13a64e : SUCCESS in 6m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6504526e57540b78377aab30550427a : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d5df39401e34c28a73981700f080e1a : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e5a186d07db24487b231702b5999e1c2 : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5c1f66c0e52472b994455f0922d963e : SUCCESS in 12m 31s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7afe445afca545019b69e22b3bddc0e1 : SUCCESS in 20m 29s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cad8a445098044c98d8485f014a45f83 : SUCCESS in 14m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/21971abf60e44e698088127838c2fe17 : SUCCESS in 14m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/855e22eb3205463e9fcef41f86c7181e : SUCCESS in 18m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/77738a50077c4b97a518ae28f0d8a981 : SUCCESS in 15m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/459dd9afe3d848f79485cdcea349cd81 : SUCCESS in 21m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dfd861b5aed247c6ae9ad1bcba0d022d : SUCCESS in 59m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b80ed45cc2eb4ec0a86843eddfe26462 : SUCCESS in 1h 00m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0c3b67c5eeeb4a06978bc98a4dcc6059 : SUCCESS in 1h 03m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6029748c5f34488acd31dbdf8e133f0 : SUCCESS in 1h 00m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/def8cf250b224e63a7cad47b9965d263 : SUCCESS in 1h 02m 46s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/eb2b5afd7aa6476b8a923091da352c9c : SUCCESS in 1h 34m 59s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/183b356192764f008157a8967d6ebd39 : SUCCESS in 1h 28m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ba475f8b24a4f0cbc37e5b39d74b464 : SUCCESS in 1h 25m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/600c8ebb895e44c8b1c992d499e73cbd : SUCCESS in 1h 37m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/229b3cbe2e0645fe82f65e88b92a3608 : SUCCESS in 21m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0d462452093948f69a0bbc847f89ebdb : SUCCESS in 23m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed6326d269284419b27cb2fac6274fae : SUCCESS in 25m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e334ec2cc06a4b6d9c01ee3cc8d1119d : SUCCESS in 12m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4cca0c991dff4c9f95f7fef37fcbfb3c : SUCCESS in 34m 42s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/900ffb593fab4810a489d0567a49d32b : SUCCESS in 28m 24s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/897ae897e5b943b28297f46495c79fdb : SUCCESS in 36m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e23fc1a5d04d41fda5677e3575949999 : SUCCESS in 33m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/30f96413a740406e8f1c042b5c5c41f5 : SUCCESS in 18m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0acb80f482b64274b6b914ae019d9a6d : SUCCESS in 20m 23s","accounts_in_message":[],"_revision_number":3},{"id":"8ed0a592b8f31839a9f5248fa73f216f594ce639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 12:46:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"40ca32fbb52929aa8250581855a75655a2026adc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 14:21:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff596e5f2aa148c384c60c1a415f156e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47653f0bf6de476c865a96ab629aa0d8 : SUCCESS in 6m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d015ab88876e410098790e6035ae3e43 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cac12cf1d564e329c1b4beeae230044 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db2598239b264406a2623ba0cf4f4794 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a377991afb34396bef3806a1171ec23 : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/915b442fc04d40adafb6851933ce4e20 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9b2277375f364f759f58b75874ec5849 : SUCCESS in 7m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2fb4478ec6964453b336e72f1e0fb693 : SUCCESS in 10m 51s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d75e23fdcf2f436a9b0993856e79872e : SUCCESS in 13m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77a0b750aeda4012a4706e28cc77a84b : SUCCESS in 12m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b4025c1f09bd41d281017ae8a3e0b47f : SUCCESS in 16m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15533bc7a05f4a24a669d8fa888fc6d7 : SUCCESS in 15m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a48231fa82bc48d1bf9693d92f9cd626 : SUCCESS in 20m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c512a56733c469781acb34b81df2921 : SUCCESS in 12m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b9ce938c5c2849899fa7f21fc7424edd : SUCCESS in 55m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e8103bc0865c47d9aa4c4653c39502f6 : SUCCESS in 1h 02m 04s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/99636b2ff885421d94a8c4898dead691 : SUCCESS in 1h 06m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47fff6504bde4cdbb36628bc8c05bc40 : SUCCESS in 1h 01m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ab4f1ac63b3543e3a4b12e727340dc3c : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/16624ad2eada4c929c03ec91c6ae2eaa : SUCCESS in 1h 34m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/44e228c624194bf9aebe36275e737990 : SUCCESS in 1h 30m 16s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d849ab236124d599a1f223d62c1e1af : SUCCESS in 1h 25m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f40b6b7861046a8a352f7feb5e9907c : SUCCESS in 1h 33m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9411adc3371e40e7af8bf3c2959ada5f : SUCCESS in 16m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a75eeffe879a4c5b9916675c59287d89 : SUCCESS in 19m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a8596b974ff4b18b11d538702b97582 : SUCCESS in 16m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/34611486ec5e4f719d3523e19fdc17e9 : SUCCESS in 15m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a6cb06af44b0469e946b9aacf6dd926d : SUCCESS in 31m 18s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3e52c562c3b14dcaaedfd6d69e55cac9 : SUCCESS in 30m 52s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98cd5761f4e648f089437220cac62bd5 : SUCCESS in 29m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f88083c77634328a3951ed4cdea912c : SUCCESS in 26m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/db23e7d4f21542aaaee5ec1f5f94b5f9 : SUCCESS in 22m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d75326394a674051a67870c9d5dc5dc9 : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":3},{"id":"6d92f3a42081260629dc3ce69cfd3f36f6126301","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 14:21:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d0d2b12bcbb4520f46b9d9492d5998e9e3757736","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 14:22:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76c30d336f66460e80c47c18d6c558fc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e496ca25be484eb7b59a1ed2366956 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4c25c48da33850037b52ca656018f6b6997d5f85","revisions":{"3b82446bba2af658cb1bffc5177bd4eda1c8d861":{"kind":"REWORK","_number":1,"created":"2026-04-29 17:11:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/986696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/986696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/986696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/986696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/986696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/986696/1"}}},"commit":{"parents":[{"commit":"87d1546f30e776b56bc58e2f56d0342cd91d5cb3","subject":"Add skeleton code for Infra VM Service Destroy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87d1546f30e776b56bc58e2f56d0342cd91d5cb3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 17:02:39.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 17:11:11.000000000","tz":60},"subject":"Stop gathering facts in trust-store playbook","message":"Stop gathering facts in trust-store playbook\n\nThe issue was that we were caching these with ansible_user and\nansible_python_interpreter overriden. This caused failures later on. We\nstill want to run this before the kayobe target venv has been added so\nthat we can use the CA for installing software packages.\n\nChange-Id: Id6373aa5eef515f5d6bd3176f71fe8b35333a009\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b82446bba2af658cb1bffc5177bd4eda1c8d861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b82446bba2af658cb1bffc5177bd4eda1c8d861"}]},"branch":"refs/heads/master"},"f3b4c5e58a38e0afaaf78a33e8191b9e8b04e5e8":{"kind":"REWORK","_number":2,"created":"2026-04-30 09:12:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/986696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/986696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/986696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/986696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/986696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/986696/2"}}},"commit":{"parents":[{"commit":"87d1546f30e776b56bc58e2f56d0342cd91d5cb3","subject":"Add skeleton code for Infra VM Service Destroy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87d1546f30e776b56bc58e2f56d0342cd91d5cb3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 17:02:39.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-30 09:11:30.000000000","tz":60},"subject":"Stop gathering facts in trust-store playbook","message":"Stop gathering facts in trust-store playbook\n\nThe issue was that we were caching these with ansible_user and\nansible_python_interpreter overriden. This caused failures later on. We\nstill want to run this before the kayobe target venv has been added so\nthat we can use the CA for installing software packages.\n\nChange-Id: Id6373aa5eef515f5d6bd3176f71fe8b35333a009\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3b4c5e58a38e0afaaf78a33e8191b9e8b04e5e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f3b4c5e58a38e0afaaf78a33e8191b9e8b04e5e8"}]},"branch":"refs/heads/master"},"4c25c48da33850037b52ca656018f6b6997d5f85":{"kind":"REWORK","_number":3,"created":"2026-04-30 11:33:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/96/986696/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/986696/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/986696/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/986696/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/986696/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/986696/3"}}},"commit":{"parents":[{"commit":"3f15fddf9712507831b6ef179147a459b505f6a5","subject":"Adds os_family variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f15fddf9712507831b6ef179147a459b505f6a5"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 17:02:39.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-30 11:32:54.000000000","tz":60},"subject":"Stop gathering facts in trust-store playbook","message":"Stop gathering facts in trust-store playbook\n\nThe issue was that we were caching these with ansible_user and\nansible_python_interpreter overriden. This caused failures later on. We\nstill want to run this before the kayobe target venv has been added so\nthat we can use the CA for installing software packages.\n\nChange-Id: Id6373aa5eef515f5d6bd3176f71fe8b35333a009\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c25c48da33850037b52ca656018f6b6997d5f85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c25c48da33850037b52ca656018f6b6997d5f85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
