)]}'
{"id":"openstack%2Fkayobe~986779","triplet_id":"openstack%2Fkayobe~master~I5a2d5f973a85efc4f561ececf2c83b2981a419b8","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-16 01:19:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a2d5f973a85efc4f561ececf2c83b2981a419b8","subject":"Adds os_family variable","status":"MERGED","created":"2026-04-30 11:33:59.000000000","updated":"2026-05-16 01:20:16.000000000","submitted":"2026-05-16 01:19:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986779","meta_rev_id":"54033dc3f60fe76527cdad09b94753d5273df09e","_number":986779,"virtual_id_number":986779,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-16 01:19:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-14 15:23:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-30 15:27:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-12 08:35:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-14 15:23:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-30 14:44:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-30 13:12:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 15:27:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-12 08:35:04.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"6c12b60e1ca75b4ab0c32a9c64a78220b1e1ab09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 11:33:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7eee647608bb7bee0a07e310e37dbd26086860c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 13:12:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b92f8b32baea4473a541fbb97faa8953\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b030d6135524c55b30e01c78a810755 : SUCCESS in 4m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c44b9b665da44a09b969415dcf610ff : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a10d2ddb34ce4c9885eae1eeb400e401 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed0c27e9c3a6450fb96b01e63992c634 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a99ed675561349889e5e7d6790ae4557 : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ec272fca50834900b3be864698be74ac : SUCCESS in 6m 36s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/53b20dfe31c24ac6819d6a17597bd659 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc9c5d0515c8400e9582fd3704f5dbda : SUCCESS in 6m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d36e927a9a441f8b90c25731ef363be : SUCCESS in 13m 16s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6c1de84a78ed4ca4bd3cb2bb3720f7b6 : SUCCESS in 15m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/751352e8f38a47638111521f699218ac : SUCCESS in 18m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/85eb0077b39d4c259541035f38eed13d : SUCCESS in 12m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b4d0f13599f4ef495e7863b1d241cbf : SUCCESS in 13m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/41f2f4c5b2bc4f1e929848f0dc78d8c1 : SUCCESS in 1h 01m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a7eeb225ca9149739399275c0ad1cf4f : SUCCESS in 1h 05m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a7f2efb171f54fe4951db11fae84a24e : SUCCESS in 1h 05m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2be6167502f3445880b6ce207a47bbb8 : SUCCESS in 1h 02m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/541f1541137d4a108115b210362da150 : SUCCESS in 1h 07m 56s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/aa1da236a2f04770ba03ba476dd94708 : SUCCESS in 1h 35m 19s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ee0c00b0138a47c8a113accba1c0e410 : SUCCESS in 1h 32m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d25750f58e6f43f48db3b60455b92c21 : SUCCESS in 1h 29m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c57cc927b9824d2bbfb1c11b0d549a12 : SUCCESS in 1h 36m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e9c122b620364ac1aae22db1844a2cbc : SUCCESS in 27m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/27c24b97779a49bc92df795d2a0d4230 : SUCCESS in 29m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b57acdf9615d46c09c007be7b2c2b47e : SUCCESS in 21m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/de5f516bb1f2494998d68d938d3bc0f8 : SUCCESS in 21m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f3f02b3dd19d4a4da761b884989d9a9b : SUCCESS in 16m 29s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8f08d89361a1440f972827787c26fcf8 : SUCCESS in 42m 09s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d729964c13c4e04a8c90f50ca817c0b : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61702650218a4892a7dd748f2eb25c38 : SUCCESS in 17m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/66bd3aca1e3f418c9a719484db80dc32 : SUCCESS in 28m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e231acc8e9146dbbd84e8499448f132 : SUCCESS in 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"817c4e39a524a0398ad17067e940778c7c9f035d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 14:44:14.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"fc02afe6bce0d501353e2acb82a877bfa3ff2537","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-30 15:27:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"20e2203a9e99ead98ac67c136f745b7004bff4d9","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-12 08:35:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6dd5975edd7cac944cf6972aafe48b6a9c98c246","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-14 15:23:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c718e38890b12e8b3d926e1e23a41544609414e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:23:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5971583d2b52124a67501444c776425c4495d2fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:10:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34418cfbd8ca4a2a83676700e8cb0875\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/073001b0ff7f4fb8b3be972ab9e144fe : SUCCESS in 10m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c768ccc1195e41a3ae2f102304bc08a1 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a3504393f7408f8750fb5e493a1541 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/adcfae9cdd414cdabcc2bd7f888d7b5b : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b2b7d23d5164734ba3f416ffefd4080 : SUCCESS in 5m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4e056f42670425cb7191c33f52de32e : SUCCESS in 7m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1499dbbfef834289b6d0f6c46bf41ff2 : SUCCESS in 13m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6624d3edc49f44dd9017011a588c59b0 : SUCCESS in 12m 11s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/edb0ae0b85ed46d3a361d7a5c4f8b68f : SUCCESS in 12m 58s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b5a7555296b45808e468a56e7a09fff : SUCCESS in 9m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7c3e3345f4d84acc86e080668790d33f : SUCCESS in 19m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9ea804183cc4fd7b71428ed5e123e8e : SUCCESS in 22m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fdde9c62820e4ab59edd6ddfd7cfd11f : SUCCESS in 25m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2173c861daa47a085132c0f7476eb26 : SUCCESS in 11m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6e49429a436442cba447bc76269e7406 : SUCCESS in 1h 04m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bae98992ae384ab9baf021c6b5d67611 : SUCCESS in 1h 06m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/21574ae54c4f49bc817fddeb391d2d3d : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ff4de233c1042dcb8b2a81f7c0be53c : SUCCESS in 1h 08m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ef840be239647448eb555c20f7a7314 : SUCCESS in 1h 05m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d5732432a108487a8b6e50077a159605 : SUCCESS in 1h 44m 39s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/84e76bf35efa42be9345a402870c578b : SUCCESS in 1h 29m 36s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de1a1be559364d2d9e280f24c0758bfb : SUCCESS in 1h 37m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e669191125eb4506a99730e63f661442 : FAILURE in 1h 25m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d00b5263721d4cc48cc45209878bc532 : SUCCESS in 14m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/512d569b083d4d9e904f87bcad4fd794 : SUCCESS in 23m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/863938a9d8264d8c982bda1b5760a7cf : SUCCESS in 23m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7754dfe420664272babbbb4c48f701ca : SUCCESS in 21m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6b7276e6db6b4b1ba5d4d1e773b0e96b : SUCCESS in 29m 05s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9765f8f2b99f4bada451810fe7ac0f78 : SUCCESS in 29m 57s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6b44ea87a8b4116bac7b9a3701bd15c : SUCCESS in 31m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c0310be564d43faa80b0951a3a998c5 : SUCCESS in 28m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/87cc9f1be1cd41baa8e7b4c163718b46 : SUCCESS in 20m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36e8ec16e0d84c4dbadd78f160705283 : SUCCESS in 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7a999c0a173fd089595b84679fc90969183858c7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-15 22:02:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75e8f9284ca3222500ae93eb51e85adcbdccb225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 23:43:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b53bd75ad7f74a4ba7a615329b3e89d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76f9aa90731248e18b22bedbbc3b1d48 : SUCCESS in 5m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c26f3a9eafc342b6bddb9c8b875eb581 : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a1a9fb05e4e47829539e6aacc7988a7 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/084488c0d1ce48849f0a4a6d7452ce12 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8938a8d622bf4f6cb14b113b5cae00e3 : SUCCESS in 2m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/07c1b2ec41fb4e69b6edf949ceaec5d7 : SUCCESS in 5m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f55e3008e9dd472f8ba1a368f758cc4c : SUCCESS in 5m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61afae9607a44368b4412d6a1cbb1c06 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/adb16b2d161140f2bb4cb2da8184da06 : SUCCESS in 7m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eebc1b6e8e3441f9a8864291598a259a : SUCCESS in 11m 03s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1922251f47564100a811cdf17eb0a390 : SUCCESS in 13m 09s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b463599fa5b04c45bb8f21414edf3163 : SUCCESS in 21m 19s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1b468d4e55a54eaaa69d612ad02ac812 : SUCCESS in 16m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/676670a2ed8145abb8ad7db23380fde8 : SUCCESS in 17m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7a00928fd6b44c81bb8c8928400dd7cc : SUCCESS in 13m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b071d257960c41e9bc515fec48306e67 : SUCCESS in 12m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/aee32f3b5b654db584c49cad33b450ab : SUCCESS in 1h 05m 42s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/67cc09f2b1cd446684a344d36313a7c1 : SUCCESS in 1h 03m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/619cc541c54b49159eaf16ebc5ca3a2c : SUCCESS in 1h 09m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00579e562fcc4659add3e1bcfae2c7d6 : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b58c75593b1a444f8ded182c7136ac03 : SUCCESS in 1h 06m 48s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/407367ff765441c68273484b06623a93 : SUCCESS in 1h 40m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6e4d44d86ee64dce8f25f0935d7e5756 : SUCCESS in 1h 31m 56s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e79fa82c8f247d4862a057ea01c969e : SUCCESS in 1h 27m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2926801cf19489eb4c4e819003ee4fc : SUCCESS in 1h 39m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/02635f14fb684442930eac407ef87d90 : SUCCESS in 28m 06s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a829fea55b75408aa9ab0b5fc296fbcc : SUCCESS in 18m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5baec8c622c54f9ea39ef9f04bfdb5a6 : SUCCESS in 22m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9725068d143541fea2793ce41c1b7fbe : SUCCESS in 16m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d8d37fb43c4a44e29968d6d9c5385f4a : SUCCESS in 29m 00s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4a2aff34ab0c4fa6974759029a468eac : SUCCESS in 33m 20s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e7f2713348c463a8fda6e4a29f0bc94 : SUCCESS in 32m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32e2185244eb48c58e2b563c489f213e : SUCCESS in 19m 34s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/533e945b73c54361943c53df4ac6d39b : SUCCESS in 19m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa38e5a970d04b59b68cdb24e2881d97 : SUCCESS in 16m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ed477e69ac897ecb7930dff98d8826161ca66e17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 23:43:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cb566cc6276f0e7cf471c649294da5fe090b734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 01:19:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1502d39dd9f94b51bf22a444936dcc5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f30463e928640f0a22be41f8fba893b : SUCCESS in 5m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/34d58d0e08ff41b59ce3e044c43596f2 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87f0311659094fd9be776a9c337ffea1 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45a88ea1c049478e9a825802b2443f2d : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df42fec6be5e43ae8a3c84398c1b9c9b : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/639f6670654d4b188d582a3aec62df13 : SUCCESS in 3m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1d2bf434d3c7440ab70a520f5891138c : SUCCESS in 5m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0acc37e078094ed68de96e33671da303 : SUCCESS in 9m 42s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9057a9df60f14e69a2c77c34be2c9631 : SUCCESS in 13m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16ff43740244497f8f5a765a8a4c20e2 : SUCCESS in 18m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a1d3d36e447344158a411f7a712cb224 : SUCCESS in 29m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1e3f8877884478d9a85b12815c2e11b : SUCCESS in 17m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f5ca9d94ed7b44fab588d6fc8248b11a : SUCCESS in 11m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bc47f87da174fdb80fd702a773ea4f6 : SUCCESS in 17m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1e2eab841686487d8337ee3c8fcf62cf : SUCCESS in 59m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/981fca538eeb425987516ab45f169ee5 : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/33f658c017664b0383828bf0fd92414c : SUCCESS in 1h 03m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/537fa3020e8142d987a90f9349a5952d : SUCCESS in 59m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7b84fd1595794fc3b14faa9f89a0d9eb : SUCCESS in 1h 00m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5c6081e67c374840870a975551d435d4 : SUCCESS in 1h 34m 57s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a0e3f24264174b37a81d6095ad971d3a : SUCCESS in 1h 28m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d284f170a354dffb58fb67c123e8340 : SUCCESS in 1h 26m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c8722cfe6e74c9d8190f9f81e39d10d : SUCCESS in 1h 33m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/eed757a78bba4452a1fc3ed5974d1c36 : SUCCESS in 21m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1e779e81998e4cad882322c8709af3ac : SUCCESS in 22m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11c38b55b2e443d4bcff1210aee34b67 : SUCCESS in 22m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2fbfbadc877d44ec856c10c27e204d14 : SUCCESS in 12m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5320cbade2c6401baa150af22bb16bcb : SUCCESS in 16m 03s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dc7642e5fabf4f609a1baf41436124c9 : SUCCESS in 35m 52s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f55bc6d29274030afd7f120ba47f794 : SUCCESS in 35m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/016191d0619f4af4879f8f5f81c40234 : SUCCESS in 33m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e7b4648fa8c9452d85a16797c2a4d5d4 : SUCCESS in 20m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ceaa681b24984c47b56ece375c725772 : SUCCESS in 19m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f9164d1641edbca57f842053bc57baedf5d92845","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 01:19:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"54033dc3f60fe76527cdad09b94753d5273df09e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-16 01:20:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18d14a060b29437db961b32d9c11487f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07df8a1fcd049c6b83fda59246d3d58 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f15fddf9712507831b6ef179147a459b505f6a5","revisions":{"3f15fddf9712507831b6ef179147a459b505f6a5":{"kind":"REWORK","_number":1,"created":"2026-04-30 11:33:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/79/986779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/986779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/986779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/986779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/986779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/986779/1"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-30 11:26:35.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-30 11:32:54.000000000","tz":60},"subject":"Adds os_family variable","message":"Adds os_family variable\n\nEnd goal is to reduce the reliance on facts for when we have the such\ndetails encoded in config. This is a step in that direction.\n\nChange-Id: I5a2d5f973a85efc4f561ececf2c83b2981a419b8\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f15fddf9712507831b6ef179147a459b505f6a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f15fddf9712507831b6ef179147a459b505f6a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
