)]}'
{"id":"openstack%2Fkayobe~986829","triplet_id":"openstack%2Fkayobe~master~I8f7b9905f827d3168333784ea5a570afa04f18a0","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-01 08:34:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8f7b9905f827d3168333784ea5a570afa04f18a0","subject":"Follow up: Cleanup orphans","status":"ABANDONED","created":"2026-04-30 15:31:56.000000000","updated":"2026-05-01 08:34:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8ad8feb4792fd8d9904df9e988f7864eea0b13bf","_number":986829,"virtual_id_number":986829,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-04-30 17:14:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-30 17:14:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14f2807334b488ffaeccd2234cb429faabed9c71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 15:31:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee7564a267ba33f647a27917a42224d3fa76458e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 17:14:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9be37de721f74456ad36bfb3f05e22c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b298f77fcfd4111a293287877d25d02 : SUCCESS in 9m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d597604c36204f6db3077ed1f59ebf95 : FAILURE in 6m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72e85b20f3bc4998884c77260004a274 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd07204764d4c1bab5d0dd15a87e7e4 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dffddeebc4b3440aa96c78a84d4f2a06 : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f46646f317f8434dbdde0f2ba8503ac9 : SUCCESS in 7m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/200ef49f3cba41f184ca7cb9fde574d4 : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/254823e9cc3c4c098863dfe06707d3b0 : SUCCESS in 7m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2ec604bb382d4835a77b942d19d86f52 : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f256f740aa042f0a989041fdd911edc : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/864b21930cbd419dbdbe475b3da47bda : SUCCESS in 24m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a18b073530b444e4b4dfd240ea550a33 : SUCCESS in 19m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7b114286970e4a92b41af6a2553f4dfe : SUCCESS in 10m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77529c9dc866432aa5054b01c7dfda70 : SUCCESS in 18m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/779d69dafb38415e95b986e9f57cbf03 : SUCCESS in 1h 02m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e6b6794cef2c499db1d1a5b8a5dddc46 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/28faf54142c048359d06ad01ff8d3e3e : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c59c4d9a42bd4822ad50efdcc28748af : SUCCESS in 1h 07m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba30d061d6fd48388a42de9f00534f6c : SUCCESS in 1h 05m 40s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/684e7095e9144c4ca9d281ff52429882 : SUCCESS in 1h 36m 48s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0c53f10900de433895248ee87ced2ac8 : SUCCESS in 1h 31m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/979b65fbef4f4096a88e970da51a19e0 : SUCCESS in 1h 30m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/026b81757a8a4460900de458d85ed828 : SUCCESS in 1h 40m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2a93e6707ced4ef4b4ef59438486f9fa : SUCCESS in 22m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2da14cae7aa24a13a3735922daf0d8fb : SUCCESS in 15m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2cf553ac6b840e1a43e6341615e6097 : SUCCESS in 21m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ffe67ba9d6434d248e41d415b64b1f03 : SUCCESS in 23m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/17644da1cea443089e280ce808e814eb : SUCCESS in 31m 23s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6d8a1603c4074e17aabad364935286e2 : SUCCESS in 22m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3352da7b78a4fb990d31021170a596b : SUCCESS in 16m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bff5209756d467e8b5ae84cc6a5da02 : SUCCESS in 20m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f982fae60a74db09e33b21814e41ed4 : SUCCESS in 14m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/649c7cc18f294da6b265069f4ec4429e : SUCCESS in 18m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8ad8feb4792fd8d9904df9e988f7864eea0b13bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-01 08:34:22.000000000","message":"Abandoned\n\nMerged into the patch below","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1fc46e7f23231e5fb8a105367f3daff4d57d766","revisions":{"a1fc46e7f23231e5fb8a105367f3daff4d57d766":{"kind":"REWORK","_number":1,"created":"2026-04-30 15:31:56.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/29/986829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/986829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/986829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/986829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/986829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/986829/1"}}},"commit":{"parents":[{"commit":"52c2e0780c32b492e71bba6b377fab9db2a11887","subject":"Cleanup orphan interfaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52c2e0780c32b492e71bba6b377fab9db2a11887"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-30 15:29:39.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-30 15:31:16.000000000","tz":60},"subject":"Follow up: Cleanup orphans","message":"Follow up: Cleanup orphans\n\nChange-Id: I8f7b9905f827d3168333784ea5a570afa04f18a0\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1fc46e7f23231e5fb8a105367f3daff4d57d766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1fc46e7f23231e5fb8a105367f3daff4d57d766"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
