)]}'
{"id":"openstack%2Fkayobe~986836","triplet_id":"openstack%2Fkayobe~master~Ib0525eb152be6144c5ee93884508512639e735c1","project":"openstack/kayobe","branch":"master","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-30 17:11:14.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib0525eb152be6144c5ee93884508512639e735c1","subject":"Alt: Cleanup orphan interfaces","status":"NEW","created":"2026-04-30 16:39:34.000000000","updated":"2026-05-01 12:40:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":3,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"760e1e33406e32c3822f2955d2cb323a1f10f82d","_number":986836,"virtual_id_number":986836,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-01 12:40:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-04-30 16:39:34.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-30 18:17:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-01 09:31:19.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2026-05-01 09:49:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-01 09:58:02.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2026-05-01 10:51:54.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-01 12:00:34.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"}],"messages":[{"id":"c64fa7d1800f05295f831f84dd8ba5602fee9841","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 16:39:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08375c16182fee13d4deab56fb81517bec3fbc5f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 16:40:48.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9c91a034248864b008c7a047b5a482aa77120e18","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 17:11:14.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c579929f90f10777ae9d258d4c9ae4690e58cc36","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 17:14:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0e7af07ffa23009e10a8c3e9252b8bbde31f35c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 18:17:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15e7df3ca32949589e810b2c37ad383d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f785b02c4c140aaaf22f01fccf543a1 : SUCCESS in 6m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b573469f44004e14ba975d28b7fb5989 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9ae4050bb0040d7ac20783ed9a0033f : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37cd8504f1d74abe931913517e43c110 : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4ddda98a2d1469c9373174cf60257a2 : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/285eb70ae08a4a0f95aa70f366964bb5 : SUCCESS in 3m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f83a54633e1542cf83cd5ac7121f83ff : SUCCESS in 4m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a482c572cd1456e8124c6c1829c56f3 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/206ee23c06a845c0915f4a4bab3663b2 : SUCCESS in 7m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/793efcda01754a539d64667973b8b287 : SUCCESS in 11m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0a15e453926d426fab5ac2b5a52e9bd9 : SUCCESS in 26m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7efac166baea4ca5afa826f218579046 : SUCCESS in 19m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b67069b339e94a728ce2a49e7ac7b2ec : SUCCESS in 25m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/901e1e25221e4b2cb0a4a744452c070b : SUCCESS in 16m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/13aca1c232f142cdb1d9e81ad7cf81b3 : SUCCESS in 1h 05m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/348a30217cd94f70a176bbfd24519ebb : SUCCESS in 1h 06m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9dabcc9162974f358dfa044583c5e1be : FAILURE in 6m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43e965725bff46288da515df539d77a5 : SUCCESS in 59m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4967a080700c41f2b5eb22c53d7ce024 : SUCCESS in 1h 05m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/39266928ffeb4d938cf7010f623788c0 : SUCCESS in 1h 36m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8fd46bce82514e07a618517f128e28ce : SUCCESS in 1h 34m 17s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af63c41ba2754ec7b6d47216251abc49 : SUCCESS in 1h 31m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa9637fd08c543e18186ea857ea04448 : SUCCESS in 1h 36m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3111528b34674efabb958af08236fa25 : SUCCESS in 14m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/633c031a83ea4048904c54fb81145f92 : SUCCESS in 16m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4367dc80a12745a8a4c882341ef0d167 : SUCCESS in 15m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94d040c2c1864bbea490cd40bdc1da37 : SUCCESS in 25m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/06e9972e51cf434bb8ac424c2f010863 : SUCCESS in 36m 06s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/05c59236018f4542a98681b97d39b808 : SUCCESS in 27m 59s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ed5e248e99a4c6da2cb68283f26dff3 : SUCCESS in 26m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f2a1ca8a0fe42b7bdbf0cde35f2c249 : SUCCESS in 30m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fdd6517f635a43a7871aaa50bbe2a257 : SUCCESS in 15m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac4ad6112b154501a3824f5780a98c6e : SUCCESS in 22m 38s","accounts_in_message":[],"_revision_number":1},{"id":"034b0a7093e060dc8c3363beefe896ec13020792","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-01 09:31:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"01cb0da71872829865a46936772983aa82f92ac2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-01 09:49:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2a732662594998b23a4cd04888231930bb1a6917","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-01 09:58:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5aaac7b7a33f3544443b82135c0ac558aef984bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 10:12:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/034ace2c190c4eee9c9164c09dd79dce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31867c64f0234727b1b6d35622d91fd3 : SUCCESS in 8m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/04d808bed0c2470a9da4c90329d94311 : SUCCESS in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1da2c9bdd774472fad6b442b0015ee83 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f809fcea3fb247878706419c5e10c478 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64074e4d2d4948bea28d1ac328b35a0a : SUCCESS in 2m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7b378c7d6b2746199840d40b1b235495 : SUCCESS in 6m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c89bd649172b481ca9b81d9648574b79 : SUCCESS in 2m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bda85649e1cf414286f3532592300e6d : FAILURE in 6m 27s\n- kayobe-overcloud-host-configure-rocky10-rigart https://zuul.opendev.org/t/openstack/build/fed41525393d435795dbe8af5fd22dfe : FAILURE in 6m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb3266e066f84ef690881e8c811238f5 : FAILURE in 6m 53s","accounts_in_message":[],"_revision_number":4},{"id":"52515a7e60d9d5288da1eae12500117831aee46c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-01 10:51:54.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"18348a5c92afc4fdcb50a5db9e69c3e834cb0303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 11:22:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95dece8a881b4bc79410ed85ebc5b762\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57b90b5d74544dfd984873ca627cc778 : SUCCESS in 9m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5a58956d8fad44529739c3856f70ed34 : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81b2ebad2fa546edb3adbf0cef84d199 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad2f1f3847f433498722d2f867c878c : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/790705e4c717448da77672011db9c5df : SUCCESS in 3m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09a62ba9994b4f00a4e2666fc86d0a49 : SUCCESS in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9323f43b64ce44f59ecbe7c01288b76f : SUCCESS in 2m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7c389fa0a356435087b5fcf0ca2b320a : FAILURE in 15m 07s\n- kayobe-overcloud-host-configure-rocky10-rigart https://zuul.opendev.org/t/openstack/build/205b6df8d77c4361bb3dd0fba173aac9 : FAILURE in 5m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7ad2794421d49e483423161d00e7de2 : FAILURE in 25m 26s","accounts_in_message":[],"_revision_number":5},{"id":"5d5b689b769eb79b4c08697fb4db9739c342e0a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-01 12:00:34.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"760e1e33406e32c3822f2955d2cb323a1f10f82d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 12:40:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0c93e95f27447faab65fc4e095964f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae54eeef4fc4cdba8ce6a4a7dd4cb0f : SUCCESS in 5m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7deb668846964d199140163746161ee3 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7436c0e29982445fb82c06c56e41fc89 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d9ce8b09d042e5bdff649db833ce91 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69863d12ef8f409e87419b8bf3c638c2 : SUCCESS in 2m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d4448a3498514e14b352d8e1bf1822dc : SUCCESS in 3m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/528efc07178f4b828d9052dedbd4d02a : SUCCESS in 4m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6fff55585cb5453181d16a4bdee8aad3 : POST_FAILURE in 24m 58s\n- kayobe-overcloud-host-configure-rocky10-rigart https://zuul.opendev.org/t/openstack/build/65b89fc40c044a298bd3af873e4a817d : FAILURE in 6m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d91d54238bc146b391a8d899a6c74609 : FAILURE in 13m 07s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"31156119d46fbd33e6cea4d2ae5932c5c3b48e67","revisions":{"24bae3a4747906efe8ffed946ac28ec220c89306":{"kind":"REWORK","_number":1,"created":"2026-04-30 16:39:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/986836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/986836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/986836/1"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:52:55.000000000","tz":180},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-30 16:39:29.000000000","tz":60},"subject":"Alt: Cleanup orphan interfaces","message":"Alt: Cleanup orphan interfaces\n\nThis change is for mitigating systemd bug [1] that networkd will\nnever delete network interfaces that was created earlier from Bifrost\nprovision. Also it fixes the same issue on RedHat systems where the\norphan interfaces still exists even after reboot because of old\nconfiguration files didn\u0027t removed after configuration.\n\n1. https://github.com/systemd/systemd/issues/6600\n\nChange-Id: Ib0525eb152be6144c5ee93884508512639e735c1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24bae3a4747906efe8ffed946ac28ec220c89306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/24bae3a4747906efe8ffed946ac28ec220c89306"}]},"branch":"refs/heads/master"},"f5c57f2f2d9ded038e2384412aefe7bafde5bbac":{"kind":"REWORK","_number":2,"created":"2026-05-01 09:31:19.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/986836/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/986836/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/986836/2"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:52:55.000000000","tz":180},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-01 09:30:13.000000000","tz":60},"subject":"Alt: Cleanup orphan interfaces","message":"Alt: Cleanup orphan interfaces\n\nCleans up kayobe created interfaces when MichaelRigart.interfaces.  This\nbrings the driver inline with networkd and nmstate, which already\nperform cleanup of kayobe created interfaces.\n\nChange-Id: Ib0525eb152be6144c5ee93884508512639e735c1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5c57f2f2d9ded038e2384412aefe7bafde5bbac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5c57f2f2d9ded038e2384412aefe7bafde5bbac"}]},"branch":"refs/heads/master"},"29b52345a06ae771ee9ac6dfe02432879b52e9eb":{"kind":"REWORK","_number":3,"created":"2026-05-01 09:49:17.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/986836/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/986836/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/986836/3"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:52:55.000000000","tz":180},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-01 09:49:11.000000000","tz":60},"subject":"Alt: Cleanup orphan interfaces","message":"Alt: Cleanup orphan interfaces\n\nCleans up kayobe created interfaces when MichaelRigart.interfaces.  This\nbrings the driver inline with networkd and nmstate, which already\nperform cleanup of kayobe created interfaces.\n\nChange-Id: Ib0525eb152be6144c5ee93884508512639e735c1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/29b52345a06ae771ee9ac6dfe02432879b52e9eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/29b52345a06ae771ee9ac6dfe02432879b52e9eb"}]},"branch":"refs/heads/master"},"fb423275d45947108beb222d32b7eca6bb6ed29d":{"kind":"REWORK","_number":4,"created":"2026-05-01 09:58:02.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/986836/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/986836/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/986836/4"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:52:55.000000000","tz":180},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-01 09:57:57.000000000","tz":60},"subject":"Alt: Cleanup orphan interfaces","message":"Alt: Cleanup orphan interfaces\n\nCleans up kayobe created interfaces when MichaelRigart.interfaces.  This\nbrings the driver inline with networkd and nmstate, which already\nperform cleanup of kayobe created interfaces.\n\nChange-Id: Ib0525eb152be6144c5ee93884508512639e735c1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb423275d45947108beb222d32b7eca6bb6ed29d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb423275d45947108beb222d32b7eca6bb6ed29d"}]},"branch":"refs/heads/master"},"71301f711ca36f0a69ac0a5c00cbc9e3d4a2fef2":{"kind":"REWORK","_number":5,"created":"2026-05-01 10:51:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/986836/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/986836/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/986836/5"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:52:55.000000000","tz":180},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-01 10:51:46.000000000","tz":60},"subject":"Alt: Cleanup orphan interfaces","message":"Alt: Cleanup orphan interfaces\n\nCleans up kayobe created interfaces when MichaelRigart.interfaces.  This\nbrings the driver inline with networkd and nmstate, which already\nperform cleanup of kayobe created interfaces.\n\nChange-Id: Ib0525eb152be6144c5ee93884508512639e735c1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71301f711ca36f0a69ac0a5c00cbc9e3d4a2fef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71301f711ca36f0a69ac0a5c00cbc9e3d4a2fef2"}]},"branch":"refs/heads/master"},"31156119d46fbd33e6cea4d2ae5932c5c3b48e67":{"kind":"REWORK","_number":6,"created":"2026-05-01 12:00:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/36/986836/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/986836/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/986836/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/986836/6"}}},"commit":{"parents":[{"commit":"3af169b943c137ee4a5003515eccf79240a6a572","subject":"Merge \"fixup! Skip external connectivity check when behind a proxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3af169b943c137ee4a5003515eccf79240a6a572"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-04-30 14:52:55.000000000","tz":180},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-01 12:00:27.000000000","tz":60},"subject":"Alt: Cleanup orphan interfaces","message":"Alt: Cleanup orphan interfaces\n\nCleans up kayobe created interfaces when MichaelRigart.interfaces.  This\nbrings the driver inline with networkd and nmstate, which already\nperform cleanup of kayobe created interfaces.\n\nChange-Id: Ib0525eb152be6144c5ee93884508512639e735c1\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31156119d46fbd33e6cea4d2ae5932c5c3b48e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31156119d46fbd33e6cea4d2ae5932c5c3b48e67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
