)]}'
{"id":"openstack%2Fkayobe~987594","triplet_id":"openstack%2Fkayobe~master~Ib42e07912d1fbffe4a82584f26126ff953585879","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-05-20 10:54:39.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-20 09:27:47.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Ib42e07912d1fbffe4a82584f26126ff953585879","subject":"Bump Ansible collections and roles","status":"MERGED","created":"2026-05-07 08:03:50.000000000","updated":"2026-05-20 10:58:03.000000000","submitted":"2026-05-20 10:54:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987594","meta_rev_id":"024116b4a75417065d782764ae23b6858c059d53","_number":987594,"virtual_id_number":987594,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-20 10:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-20 09:13:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2026-05-20 09:18:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-05-20 09:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-20 09:13:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-05-20 09:11:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-07 08:13:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-05-07 09:06:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-20 07:26:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-05-20 07:26:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-05-20 07:26:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-20 07:26:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-05-20 07:26:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-05-20 07:26:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-20 07:26:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"a4ca93b3206ab5996c4a26943b58bf89ef9f71ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-07 08:03:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f5f96374d957091307fd48ceb3b653c935419d8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-07 08:18:27.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e82f68b2d988c390c5e4cde73ab1795371512b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 09:06:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fedd3d743bc438d86f2d859713ce2db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b89a6cbbd50e4a09bd4089955274f23c : SUCCESS in 7m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bb9f6994d0474e059108d2d730089230 : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a721dbfab9047ec802d65f1551792b8 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361813aa1d794f4c949bae8430ca2892 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/795e054e9c5e42adbd15b270c29ecfd0 : SUCCESS in 4m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b6dfaf543bb46dc8a8c63277eacd745 : SUCCESS in 6m 03s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9818460d00cd4e27a3ce983cb3973c70 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dca5605a6c40441eaea713a2e81c4fc2 : SUCCESS in 6m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c98a1a519e4c4e48bf10ed4fc7a248ef : SUCCESS in 5m 58s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/55fb22a28c7d453ea610f59e2052880d : SUCCESS in 13m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b14dc021b2ba4cb4b14e325ec22a4c8f : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5bd54ee506a848179383bad6e9bde92c : SUCCESS in 10m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8944cc3156541818644602ff626793c : SUCCESS in 12m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/abdbe72e6dc44f09bb7e97c0433da0a5 : FAILURE in 21m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5315647cd8a04cc5a6e97963de81bad0 : FAILURE in 20m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b23113ab3ab547598327183b33b00279 : FAILURE in 22m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89794da47c36483fad349640804bf722 : FAILURE in 16m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e111e99d42f145a089b95d43cce20a61 : FAILURE in 16m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/925b84802c904bea9326c6fe3cf480b1 : FAILURE in 1h 00m 40s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/be839a45dd4a47baa6811016fc4bdef2 : FAILURE in 1h 00m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90553e6736fc499bbadd22a764d98564 : FAILURE in 55m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9e6454eea99488f87ca84d6eebc26cb : FAILURE in 1h 00m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f6026307a2934dd4aeb24b9713499096 : SUCCESS in 16m 13s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c122ac16883747e09d5a7403d1b1350f : SUCCESS in 15m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b792c2b06aab42189a49a65845c4e596 : SUCCESS in 22m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c0ef7dd654e4fe5a4ee276b138c4fde : SUCCESS in 21m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/89ec1e8fcaa04e4f8fb43808c7c6da4f : SUCCESS in 19m 46s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/80138f914fbd41d38f4905dcfd074079 : SUCCESS in 21m 27s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a296ab2fb7d1428ebcb6daab7047a7a8 : SUCCESS in 27m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6bc62d4aaa247a1badcaeb7efc40d8d : SUCCESS in 18m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/118dc52f8b934238a68b45e349e2917e : SUCCESS in 20m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce30d3e3b54f44cb990d6ca37d6ef8c5 : SUCCESS in 14m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0cd999170a28ceef48c9c5e446fce26ff6ace38a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-14 21:10:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bf0a9153d423f4d8094799ed59b493c4609ff114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 22:45:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f2d5cbdc14b40dea9cc7d6687070707\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f8d33a86f36449b91256d210d6d7a3e : SUCCESS in 10m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e9c9ca160c7340fe942851e78a91d4d4 : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df242a86beca408d804c6ca806038f32 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a103ae0cfea34ffca3074f0aa4e78d5b : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ff5bb5357b04612ab15070b101b1810 : SUCCESS in 5m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/75c9ff4cb7794c258887f06263692622 : SUCCESS in 8m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6892e74565e74ec296cb127a2b31cb2f : SUCCESS in 6m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a5d201b9c6524d1ba5bdbb2eb740f83a : SUCCESS in 7m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c919c9dac5c34353afa9e061fd0cdc91 : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/af4534f391ba45d8929f7584ec2f6fc9 : SUCCESS in 11m 42s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/927f4e9580914d4f81d702dbf5830718 : SUCCESS in 13m 47s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb346b32c7e147e18ab3b08aeac1b7b2 : SUCCESS in 16m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ccc4a6b1ce1e4ae3bc042a5f273f1064 : SUCCESS in 24m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e06017c734b6400fac570c9b37e17d43 : SUCCESS in 20m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2b424e4452d34a7d99b3ac1d3cba8894 : SUCCESS in 19m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2db44d6405df4948aab72ac0fe6a929d : SUCCESS in 18m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cc262efe6ee343bca12173f2d6f8c69b : SUCCESS in 1h 08m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5cb52f0fa04046cb815e8691fda33fc3 : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/160d00e507be466eb4513ff529f2143e : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8acd1045fb4f4b028e9a19610904ccfe : SUCCESS in 1h 06m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e445274ce3fd4170a86daab36b39c429 : FAILURE in 1h 01m 56s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6cb8d0fe64a7493cbdc5aa6877eacaf9 : SUCCESS in 1h 33m 59s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c1b4c81c05c2464f9d044df939105889 : SUCCESS in 1h 33m 02s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9a2633b37184ec79dae89c05b962fcc : SUCCESS in 1h 28m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42f5d4581d5b4be8b1812e9a11185e4c : SUCCESS in 1h 33m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d90c7d4bd8f14b8dacf933c449cb1b65 : SUCCESS in 27m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b278603b3f2b4904b455fe11b8b9e31c : SUCCESS in 30m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65a739d818ef4d32a9a11fafd1b4adaa : SUCCESS in 26m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fc36b9d6aacf483bacbfae1a1b76df95 : SUCCESS in 17m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/91f57af20dd24319bfa7976291fb2068 : SUCCESS in 35m 30s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/634ef7aaea504c81b6a87ed187bb4f71 : SUCCESS in 40m 06s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e3fa01fc4094080891c2e3be4c875b3 : SUCCESS in 30m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b9b3092660a415e8eec0069948fc33b : SUCCESS in 33m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9b1ee6b762744f3389e7bad1b18cd252 : SUCCESS in 22m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b884124cdaa43e8b16910bbb75eacfd : SUCCESS in 22m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b417600a9802baef34c161410dd3eda382000a5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-19 09:28:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ec4e70cec8a3e8c9b6e95866f27081d3b237c9cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 11:03:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b9a8809a236456e8f604112a8d62286\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38ead964ccef447f949636387a0ec2f1 : SUCCESS in 7m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/baa0cd14f1b14272830e9eee28c08681 : SUCCESS in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3ab4dc93d5c4e1da0e4e8e50fb8a301 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70aeb8763147496ca4c32c9ae0081734 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/582dff956aed48b4a6f638272b4a8aae : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a80ff5f0b9d24329ace38fb3a8184d6a : SUCCESS in 7m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/791b83b06f84457ab3b8c6097b2d5ef6 : SUCCESS in 5m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/961f3b8f3f8f4ec6bbead9f7e213e3cd : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c02e8573f11c4c09932e0e14d1fcf75c : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81dc6225c04c45d4918716ddc4aca504 : SUCCESS in 7m 47s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b5f1d9f412f4486d9de884a151e383db : SUCCESS in 16m 42s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a14a7db9dab44297aeac7dd1bd7f1a31 : SUCCESS in 14m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a30be79d554a4b7b9ab17ccf8baa3c32 : SUCCESS in 20m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf7c646ce7db41d3a9a75c169f3df2b5 : SUCCESS in 18m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8183615fec97423eb457beb93c297aeb : SUCCESS in 12m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0a72395d1894d90821854c20f567eba : SUCCESS in 15m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/046b4fae609f4bfa81e3aab15965d97b : FAILURE in 26m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/19dac67d43914787ae1ed70a93a86b51 : SUCCESS in 1h 02m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3cdee8ed32b24382a89daf936206e64b : FAILURE in 33m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c03cd5810abe45b19612d9fa16cafc3f : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e3bf0c48346e463a9ed0535416a79ac2 : SUCCESS in 1h 02m 45s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/406590b8121d4a8f9a7a205761903be6 : FAILURE in 54m 17s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7ea6b8a464b8458a9516302bca942cfa : FAILURE in 48m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/870e0a96424a47908781dfb7715d79ac : SUCCESS in 1h 26m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/701e88ff5b294465aa2430a20853d8fd : SUCCESS in 1h 33m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/10bc968d69d24163939ab460c326006c : FAILURE in 23m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8efddafc3d9e497891a232da1e4fe876 : SUCCESS in 23m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c9b21ea4243459dba579f8e303e4132 : SUCCESS in 22m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2fdea4e4f6654beb9065627242df9d12 : SUCCESS in 20m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/df428c4496ef481cb05832652e90838f : FAILURE in 15m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7629225a729b4d6f9a82271d05d6e429 : FAILURE in 10m 29s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98d7544fb5034803b5835382f1203217 : SUCCESS in 31m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/474e9e2470d74cbda95fb017df1fc623 : SUCCESS in 30m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c74d591659224b818a9f94a4395a2785 : SUCCESS in 16m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/111e9a7f9f034bce8862f7ec48179fdb : SUCCESS in 20m 18s","accounts_in_message":[],"_revision_number":3},{"id":"1cb3668efdd89ddc504daea84a13e05a116ac4d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-19 20:39:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"46da14ada484a663250edd11a9b07a1403a9a50a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-19 20:50:36.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9f57a6cb8107e6b3e7f8b87f44f194f519279f49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 22:32:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/748077aea9aa412381c42098fd635e05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a5ea50ab6f741eead541a43f4b328dc : SUCCESS in 9m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8e04fb129bf04da2bf9d07a60f6c8997 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ab80ca51cef4b839e6101d9ffdcfbcf : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c64be507d24c5ca509f516da37d7bb : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0d7066540b94751b997bc2ea236d318 : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6dd759424c9c480aa2bc19c69cb2bb70 : SUCCESS in 4m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/632b6f0dd76a4d6fb4fce44d54b1e7fa : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b1f8dbdc24949f7b8bd4c6e5c79faa4 : SUCCESS in 7m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/301c243dd381470a817bcd171202b3f2 : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/45d38388dce142aa89126860a416a451 : SUCCESS in 6m 22s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/50490d0cc42d413b87f3897ed2f3fa02 : SUCCESS in 22m 47s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05aa4f94e25446cfae5dde5af48b4674 : SUCCESS in 8m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/161bb1ed716141e1889acb304d3bfc29 : SUCCESS in 14m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/192889ff700246269eef45d9a06d0ba2 : SUCCESS in 20m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b9df4fe889bf4208910722db01253ebb : SUCCESS in 11m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03569359c664448094e567c88bd7876c : SUCCESS in 18m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/75542f2c2ac348f3b4deac4f99150176 : SUCCESS in 1h 07m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3e4b88ffbc644a21a4344d5607a02146 : SUCCESS in 1h 08m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6b8720e9a97b4de8a64982bcd6c97d78 : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8acb13421874450bd795f4d554b8893 : SUCCESS in 1h 05m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b4eba05b5fce4170b7bb3d6a758561b6 : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8b7f3de6c0d3448faf81670b1916f61d : SUCCESS in 1h 37m 51s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a40d8442314d4143b72d61fbcb8a104d : SUCCESS in 1h 29m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03a069fcaa8b40f4bf8820979d6d7db4 : SUCCESS in 1h 25m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f37fdb796035409b814bac6674c1638a : SUCCESS in 1h 39m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f49bf8e9e9ba408f8560c061532d5935 : SUCCESS in 30m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/856d24f719b047578144dab1f00e6c49 : SUCCESS in 15m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ae115ed03e147e0a154f5e8f97b052f : SUCCESS in 19m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5b1df71ee164406e8e7fa9cd998d3b31 : SUCCESS in 16m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ef10cbbcaced4a7c8f95e0c9f97ce3fb : SUCCESS in 16m 57s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/59ff3a24d98c41eda78134d8c59d3410 : SUCCESS in 26m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b6e9ec3ed8649499e432eefacfbb538 : SUCCESS in 31m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/749d1b1a88b047e6a71a195154b3688c : SUCCESS in 17m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ffa9c5b87ecc4ee59166dcc4cbac2260 : SUCCESS in 28m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4fd6bbb5bad4f7a88711e57f87da694 : SUCCESS in 23m 33s","accounts_in_message":[],"_revision_number":5},{"id":"f26535bf4c7ced7db262bf4022500715c8249539","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-20 09:11:47.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"85993b0e65d353ed7c9c12e1c835885f49c9bfd0","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-20 09:11:56.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"7b51269cad60b453718988d00aa2431683b08cac","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-20 09:13:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f61a1f254b522081478a244562f4bdc0b5636722","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:13:49.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"009a70ea27bff5cbc5a0575449bfdb4e82bc2539","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-05-20 09:18:21.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"74a69eeb4e3e45fa198b98c93dd881abb6ec819e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-20 09:27:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7a5477a706e3f42953ba095136bc719ab942b635","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 10:54:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f8e24bc370d4f9e9e86723207f0285c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8d5eba02f4c43a5b37d86f53981507a : SUCCESS in 9m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d9a354d0ffaf4e359be8805db208efb1 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/270247f8fce5463ab991597408c3833e : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07eafb6f9aa245309fbd7175e9b6583d : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5a0687e6d604703a4e575f9250f43e3 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae3747d35aef4f689ba1597b0b0c7890 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3ba1e8f173e447789113fe97b49b8a06 : SUCCESS in 12m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b8ab957f2a64aae88c85b00aecebb13 : SUCCESS in 12m 39s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/420b0b1b041d45afabe7b72a20d995dc : SUCCESS in 17m 19s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae619d99311541ac9d964c938846132f : SUCCESS in 12m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/63b2f96ff8e8406eba168e0501a20760 : SUCCESS in 25m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27c4f3f32cf64401badd9997fe6bbd7f : SUCCESS in 19m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/11f4a91dd4c94c61ae98df49e8c3effb : SUCCESS in 12m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91d2d309b63e4fa8b78f91ca99fb38ce : SUCCESS in 22m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/51c51ad1220f4ccdb3b589ebeee32577 : SUCCESS in 58m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/103d5c18f2cc441a9635477459f4dc89 : SUCCESS in 1h 03m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8441ad4b2c8c4d30acede27c71ff4205 : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b2b62e726d640ba9f0fdbafdd2b49d3 : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/73f662d2af7d47fba98d1b58c9f66dfb : SUCCESS in 1h 06m 35s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ac61e3c4a26b498cb80ee743e63e58b0 : SUCCESS in 1h 34m 32s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c6f669188768480abcf916c1b67c708f : SUCCESS in 1h 32m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f80294073cd4463e9f0a6fb95a98a86d : SUCCESS in 1h 28m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6034a8e350b4781b0b8900b340b84d4 : SUCCESS in 1h 33m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c887d09dcfde46fb870a794734727a52 : SUCCESS in 28m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/254ce596f9f848f29772e56231ca4f1f : SUCCESS in 23m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54d60b70d28e4e9c9719f65baa862380 : SUCCESS in 18m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cd0dd9b06b2a4e5d89f4f6a06e0af5a8 : SUCCESS in 18m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a2e392283f7740ffbe5ac4b716f2f008 : SUCCESS in 28m 21s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bccf05207bbb416a99c2fcf55cf439c1 : SUCCESS in 17m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65892c4ec8854b8086697ccdf33f51d7 : SUCCESS in 27m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aab0cb272d164e63b45cb94b0ddeca7a : SUCCESS in 15m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1313f16d5cc9455eb12e91eebc2b88f7 : SUCCESS in 15m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec7c65851bb64ca2842791d9c7af883d : SUCCESS in 21m 05s","accounts_in_message":[],"_revision_number":5},{"id":"50787e9bb66a8b26687c894c93b0e533b78fca26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 10:54:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"024116b4a75417065d782764ae23b6858c059d53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 10:58:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72b2fcc28d84414b90e06cccc1504403\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dabb8c2f7b1488586843c9396bb0b97 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"297c74f810fab999e3d2d55853b89622287bc73e","revisions":{"389f1b3b2368ee2c979645818798b7d251fb7e71":{"kind":"REWORK","_number":1,"created":"2026-05-07 08:03:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/94/987594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/987594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/987594/1"}}},"commit":{"parents":[{"commit":"1486c00bdb3e66d88a61e7a8cd70957d6c403774","subject":"Merge \"Extend etc-hosts support to all host groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1486c00bdb3e66d88a61e7a8cd70957d6c403774"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-07 08:03:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-07 08:03:43.000000000","tz":120},"subject":"Bump Ansible collections and roles","message":"Bump Ansible collections and roles\n\nChange-Id: Ib42e07912d1fbffe4a82584f26126ff953585879\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/389f1b3b2368ee2c979645818798b7d251fb7e71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/389f1b3b2368ee2c979645818798b7d251fb7e71"}]},"branch":"refs/heads/master"},"710a4658f06e627ad890d27fc926041cf1482c06":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2026-05-14 21:10:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/94/987594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/987594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/987594/2"}}},"commit":{"parents":[{"commit":"1c5b12b35b48f746c77ffbd31b5536b76972a65d","subject":"Merge \"kolla-ansible: add custom collections requirements support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c5b12b35b48f746c77ffbd31b5536b76972a65d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-14 21:10:21.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-14 21:10:21.000000000","tz":120},"subject":"Bump Ansible collections and roles","message":"Bump Ansible collections and roles\n\nBumps community.docker to 5.2.0 and MichaelRigart.interfaces to v1.17.0.\n\nChange-Id: Ib42e07912d1fbffe4a82584f26126ff953585879\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/710a4658f06e627ad890d27fc926041cf1482c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/710a4658f06e627ad890d27fc926041cf1482c06"}]},"branch":"refs/heads/master"},"6df67b1542303944e9eb639f32a685e96dd97632":{"kind":"REWORK","_number":3,"created":"2026-05-19 09:28:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/94/987594/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/987594/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/987594/3"}}},"commit":{"parents":[{"commit":"b461a8bf3405a2d63884e5cd765e0318180aaad0","subject":"Merge \"Stop gathering facts in trust-store playbook\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b461a8bf3405a2d63884e5cd765e0318180aaad0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-19 09:16:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-19 09:16:12.000000000","tz":120},"subject":"Bump Ansible collections and roles","message":"Bump Ansible collections and roles\n\nBump community.docker to 5.2.0 and MichaelRigart.interfaces to v1.17.0.\nUpdate jriguera.configdrive to the latest commit SHA.\n\nChange-Id: Ib42e07912d1fbffe4a82584f26126ff953585879\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6df67b1542303944e9eb639f32a685e96dd97632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6df67b1542303944e9eb639f32a685e96dd97632"}]},"branch":"refs/heads/master"},"6221f30d6149e640cd7d55d382a053346884167a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-05-19 20:39:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/94/987594/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/987594/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/987594/4"}}},"commit":{"parents":[{"commit":"43294efe8ce9677d56929ae4e79e2c66c25f97f2","subject":"CI: Exclude docker-ce 29.5.1 on CentOS/Rocky Linux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43294efe8ce9677d56929ae4e79e2c66c25f97f2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-19 20:39:38.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-19 20:39:38.000000000","tz":120},"subject":"Bump Ansible collections and roles","message":"Bump Ansible collections and roles\n\nBump community.docker to 5.2.0 and MichaelRigart.interfaces to v1.17.0.\nUpdate jriguera.configdrive to the latest commit SHA.\n\nChange-Id: Ib42e07912d1fbffe4a82584f26126ff953585879\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6221f30d6149e640cd7d55d382a053346884167a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6221f30d6149e640cd7d55d382a053346884167a"}]},"branch":"refs/heads/master"},"297c74f810fab999e3d2d55853b89622287bc73e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-05-19 20:50:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/94/987594/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/987594/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/987594/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/987594/5"}}},"commit":{"parents":[{"commit":"43294efe8ce9677d56929ae4e79e2c66c25f97f2","subject":"CI: Exclude docker-ce 29.5.1 on CentOS/Rocky Linux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43294efe8ce9677d56929ae4e79e2c66c25f97f2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-19 20:50:05.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-19 20:50:05.000000000","tz":120},"subject":"Bump Ansible collections and roles","message":"Bump Ansible collections and roles\n\nBump community.docker to 5.2.0, MichaelRigart.interfaces to v1.17.0 and\nstackhpc.libvirt-vm to v1.16.4. Update jriguera.configdrive to the\nlatest commit SHA.\n\nChange-Id: Ib42e07912d1fbffe4a82584f26126ff953585879\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/297c74f810fab999e3d2d55853b89622287bc73e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/297c74f810fab999e3d2d55853b89622287bc73e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
