)]}'
{"id":"openstack%2Fkayobe~989066","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Ie1fc5c55137e985a21aeb97507c7efe9ec9b5336","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-20 10:55:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1fc5c55137e985a21aeb97507c7efe9ec9b5336","subject":"[CI] Switch to baseurl for all repos","status":"MERGED","created":"2026-05-19 07:32:00.000000000","updated":"2026-05-20 10:57:49.000000000","submitted":"2026-05-20 10:55:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989066","cherry_pick_of_change":988600,"cherry_pick_of_patch_set":1,"meta_rev_id":"2b52ecfdc03762b49e360e457cde11b727fde94e","_number":989066,"virtual_id_number":989066,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-20 10:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-20 08:27:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-20 07:35:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-20 08:27:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-19 07:32:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-05-19 07:33:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:35.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-05-19 08:56:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"538920c558cde0092b3b92ccb86b878bf3b00eb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-19 07:32:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4c9e92d7bc2706f0f150a28af7368207f1cf8e86","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-19 07:33:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"58567a3bfccbee2b05a1d0e96f7e26857d56a33d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 08:56:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c046ad47a4794500ae18e6ff1a3b2b72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bbd3b3af77646828adbceb10f45b613 : SUCCESS in 4m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4cb534e0bbfe4317b80c52954e16d654 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12bd2f89239545a59f08f4a8f1434188 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69fccb918c014666ba03d0a04bad3707 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/062f2c49655e4c9eb08374ac1c253339 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cf40568d4cd48c0a4d562be9980b274 : SUCCESS in 5m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c6bc4f702044500afc6b6c18f69e96f : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/49a693360480403c992f51e3041e2227 : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19afe464342b464da7e1d7b11fa4b0c7 : SUCCESS in 6m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/09dc8ce8d49e45c6a4efdd22d9fea3f5 : SUCCESS in 6m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f0c40ef98134092852f06e57de5d6d1 : SUCCESS in 16m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45af570ccb1142aead4b362c8925abe7 : SUCCESS in 9m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2520554215524dd58e4b4296cf54292b : SUCCESS in 11m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a937551525941638c8a07e60659008f : SUCCESS in 8m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/74a0b2de58554877856198f1e162e145 : SUCCESS in 56m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f6ce3aa3919942e2a41fc213373e276b : SUCCESS in 59m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a7e7d6efb7e14b6d8280386f820e3c9a : SUCCESS in 58m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f12f91e20ab4ca5867a6c1bd1a91ace : SUCCESS in 58m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1da131d77f9a4e3aa9b81e45d4bbc3df : SUCCESS in 55m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4f7538ad07014375afcb947aa230a521 : FAILURE in 48m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8855b12059241999c8a79d799297918 : SUCCESS in 1h 19m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/060eae4b679840a0a210da85e152fddf : FAILURE in 22m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/61369036c9cb4b038a0ccff81db71df8 : SUCCESS in 16m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f650052842b1482bb99a70d170c087ee : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/63a33f38b5834096b54a7fdf782fc75c : SUCCESS in 16m 07s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2abc34629f714e5d988d32a88df972bb : FAILURE in 9m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bdb894c0b0f4e609672cc09511d4eb4 : SUCCESS in 32m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3a01c36e56894b8faeb951367f884bd4 : SUCCESS in 9m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e27bfc161df9461baa6123d892ae7cf3 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a4ec2828b01736bc767d92bd3406e26ce13737a5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-19 09:43:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95bed726062b71367fa14f81383748ac73732896","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-20 07:35:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"86e02fd40b74f73dbf790022830fd1504b3fc6f1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-20 08:27:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"70227dc638c29afc636f23e9dcbf1af117f5246d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:16:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea6196d352a5469d98c28bc2e45b8f52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/366ef775a13d4e41a64334d226391475 : SUCCESS in 10m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/612dc4df632c4561aaa9aa5ce57fa4dd : SUCCESS in 7m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e731cc9f756340f6b49fbfbe42d62e8e : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33f11fdbbaaf44ac951c3075b0bb692e : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3543e248b62a4663aee1a5e332004678 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2698c29d404b4efbb9c10a698add9971 : SUCCESS in 7m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfa2bcee1a9f45cc9c5a0ca959f259c7 : SUCCESS in 5m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d524855fbae404fbaa34eae61579c11 : SUCCESS in 3m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1e70465768944f380010570a729b0ff : SUCCESS in 16m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4a688e4134e4b0c9e55c57d7c757f91 : SUCCESS in 15m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/effbabf68b1d4f41924708e5106de7b4 : SUCCESS in 8m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f283983972a412f99df326f67fe4c9e : SUCCESS in 13m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4bcb7710d8734beabdd65e4debbad277 : SUCCESS in 15m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4256d2815716422d9deee80013dc26b3 : SUCCESS in 8m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c6a55fa3422049298e7dea2b477fbf1d : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a7c89b204c664d80a93bfde7f8f5f409 : SUCCESS in 1h 09m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/eb0ca913c769474a9e675b9762f00a5f : SUCCESS in 1h 06m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e0557dbdb3848ec9f2ce8c4c69805c1 : SUCCESS in 1h 13m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6bc78740f0b34f2c88fb0f4a7e65cc7a : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8b721fd979fc46dd9ba3efe3fd5474ba : SUCCESS in 1h 40m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59e3eef7849242759ec6b95732d08e0f : SUCCESS in 1h 32m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fd4e38f0d6e24474b18989bccc01f01d : SUCCESS in 14m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c014f66d9e874af1861a65b23898c1c6 : SUCCESS in 13m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/add7e5361f4a472099e57f108c1fc1b4 : SUCCESS in 16m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9d0800dacc93481cb9306b876c792e2a : SUCCESS in 14m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ad35cc6311534b858acc18684593121d : SUCCESS in 30m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4084b1eedb6c43db875f282f07c16f02 : SUCCESS in 26m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2daf04d66dfd4463a31ff745049bd79f : SUCCESS in 9m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e4c09200a084dfe890144d545488f7e : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1e8a9a6099108ec6fc7488bfa84ba1f83f9ba32e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 09:17:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7be2ca1b41022196179467a0cd21ecadc8caded1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 10:55:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ad0c08807e34487a2142030f0715c17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4761504aafd49da9dd45d76d65c3612 : SUCCESS in 8m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/891f7929151a4719820b68272d2f52d1 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e22ff1c898f4ad394a0f82c6e74c372 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/606210a706fb4c5990dffd33b2b744b4 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ff9b651843842f9825488398d0c8e11 : SUCCESS in 6m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af945a0cc5064191829fae519bc6c8cb : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a831aa5e0f694139af83074b8eb9310e : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e75bf4341b04e75809bd26d8935dcd4 : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e74ba8bdaf854ab2918df9cb60ac2f8f : SUCCESS in 14m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a2d48f9646043fb9c0f09bd53aa8c7b : SUCCESS in 13m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e398092de3ff42908f319ae0aa7e5ae7 : SUCCESS in 13m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f878852ba24f4bef97beb0caed6ea787 : SUCCESS in 15m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f94be787b5d848b1832cc785c2a6a0c1 : SUCCESS in 56m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9f5d849364c74e32af93cd37953e7f56 : SUCCESS in 56m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fabc54f2e5834fce8728bdd0298287c5 : SUCCESS in 1h 00m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58a708d33c294ce3b6616a2d96ab7284 : SUCCESS in 59m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e75328af76e14fe7a2e0f2aef0cd7879 : SUCCESS in 58m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1e482f085b3241c78a48648c99f3b395 : SUCCESS in 1h 26m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41379cbf184c494ca287741d028c27d1 : SUCCESS in 1h 29m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/06cf46c5555f44ca8cd313bd1ab050a4 : SUCCESS in 9m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/dd2b9bb95d0944f981af7871cabf5df2 : SUCCESS in 10m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96a829e20e5b44938e59def1bdbc58fc : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/37419fe86c624c878516abfe4fadabd2 : SUCCESS in 17m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/45fb4487f4f8464991ca374f1d842d7d : SUCCESS in 16m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14392f5337a04b30b4eb40560ae3f160 : SUCCESS in 27m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/25bef60cba3c48a8b0e1e6deb10eac9d : SUCCESS in 9m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dad73db2caf84d7b8db159d3b64fcb63 : SUCCESS in 14m 57s","accounts_in_message":[],"_revision_number":2},{"id":"2bfa6c3be2dd1e0e9f0628cf6bfa8b317ad5ca05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 10:55:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2b52ecfdc03762b49e360e457cde11b727fde94e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 10:57:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5049990504a44fd99793ee9f286ebb1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2197c874b3041ca908885b7aa6db744 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f5c825ed3ce9ad95377edf6461d8012d0f94cd47","revisions":{"358434aa43b3d64c5c4ad293fa3a88bd7a757044":{"kind":"REWORK","_number":1,"created":"2026-05-19 07:32:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/66/989066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/989066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/989066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/989066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/989066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/989066/1"}}},"commit":{"parents":[{"commit":"f42587a4147436e7c327f8c2cd53ea2d4c03584d","subject":"Quote baremetal node create arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f42587a4147436e7c327f8c2cd53ea2d4c03584d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-14 11:01:39.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-19 07:32:00.000000000","tz":0},"subject":"[CI] Switch to baseurl for all repos","message":"[CI] Switch to baseurl for all repos\n\nWe still seem to be seeing failures with the following signature:\n\n    Failed to download packages: No URLs in mirrorlist\n\nThis extends the previous change[1] to all repos.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/983990\n\nChange-Id: Ie1fc5c55137e985a21aeb97507c7efe9ec9b5336\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit 8d6fcbfcc0e4b369b0f1f955404bbd2e39663274)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/358434aa43b3d64c5c4ad293fa3a88bd7a757044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/358434aa43b3d64c5c4ad293fa3a88bd7a757044"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"ours":"f42587a4147436e7c327f8c2cd53ea2d4c03584d","theirs":"8d6fcbfcc0e4b369b0f1f955404bbd2e39663274","contains_conflicts":false}},"f5c825ed3ce9ad95377edf6461d8012d0f94cd47":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-20 07:35:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/66/989066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/989066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/989066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/989066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/989066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/989066/2"}}},"commit":{"parents":[{"commit":"a666ae1e7d2bc3d0f741ecede9a01142faa41d96","subject":"CI: Exclude docker-ce 29.5.1 on CentOS/Rocky Linux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a666ae1e7d2bc3d0f741ecede9a01142faa41d96"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-14 11:01:39.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-20 07:35:21.000000000","tz":0},"subject":"[CI] Switch to baseurl for all repos","message":"[CI] Switch to baseurl for all repos\n\nWe still seem to be seeing failures with the following signature:\n\n    Failed to download packages: No URLs in mirrorlist\n\nThis extends the previous change[1] to all repos.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/983990\n\nChange-Id: Ie1fc5c55137e985a21aeb97507c7efe9ec9b5336\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit 8d6fcbfcc0e4b369b0f1f955404bbd2e39663274)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5c825ed3ce9ad95377edf6461d8012d0f94cd47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5c825ed3ce9ad95377edf6461d8012d0f94cd47"}]},"branch":"refs/heads/stable/2025.2","description":"Rebase","conflicts":{"ours":"358434aa43b3d64c5c4ad293fa3a88bd7a757044","theirs":"a666ae1e7d2bc3d0f741ecede9a01142faa41d96","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
