)]}'
{"id":"openstack%2Fkayobe~989067","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Ie1fc5c55137e985a21aeb97507c7efe9ec9b5336","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-20 12:31:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1fc5c55137e985a21aeb97507c7efe9ec9b5336","subject":"[CI] Switch to baseurl for all repos","status":"MERGED","created":"2026-05-19 07:32:44.000000000","updated":"2026-05-20 12:33:06.000000000","submitted":"2026-05-20 12:31:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989067","cherry_pick_of_change":988600,"cherry_pick_of_patch_set":1,"meta_rev_id":"9afffa57e86963ad9b7d707ed76ef173535b91a1","_number":989067,"virtual_id_number":989067,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-20 12:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-20 08:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-05-20 07:31:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-20 07:31:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-20 08:27:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-19 07:32:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-05-19 07:33:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-05-19 07:33:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-05-19 09:53:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-19 21:29:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"e7f682bfac042f45d29a743ba194132afe7eadc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-19 07:32:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1516bd2f53d01ec35eae9c16c37e8822ee827caa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-19 07:33:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acebeb2ef9eaa4e6926bb73c8854bfe25de3a475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 09:53:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41aa72d153a944848105c0c7607f79b8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8b3749583b34a7c82a712310894e3c1 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/393adc4052b04d9fa10431615e034040 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1100a152a32741f4a71700ab4bbf6314 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9257452c99114804b9e800d33db1f604 : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bb3353f3b0549fb961da1d0c2a6d988 : SUCCESS in 6m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fbd6d2dc4eb34c42a2faf1ebf612fd19 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0c7827e7e3142cc989e2206d55dbcca : SUCCESS in 10m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/382bbd6818f94ee6a355cc233b9a1aa2 : SUCCESS in 7m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/80c2f6d7b663493983a728f8313fb52e : SUCCESS in 18m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/602d438d363b478b937704bb07b20240 : SUCCESS in 15m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87dbffcd2318427ebff4ce5b6ddf3ea5 : SUCCESS in 8m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bf888b4bf949493e9768c6e66ae6f309 : SUCCESS in 16m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/16eb87b6daaa46abad77b8691808e694 : SUCCESS in 21m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/861ec6b2ebc34bd2add2eaf7b7df6513 : SUCCESS in 9m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/181c4ea48f254057b5d4a80197139957 : SUCCESS in 54m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d9994514cd8d49c1a617adc4107ddb72 : SUCCESS in 51m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/36ffa5255ca5493e9746e10edc079648 : SUCCESS in 53m 59s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/82e595fd1f324b72a87bba5ad80eccdf : SUCCESS in 54m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fe58e65f042b429c8827c2cc0d55c669 : SUCCESS in 35m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d226d459277e482da907b12da24b8cc4 : SUCCESS in 38m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b96ac4ad7f5400a905e9cf8beba846f : SUCCESS in 58m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/34dedee7f48843e4bd9647cae1ffa2e9 : SUCCESS in 31m 37s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8368fe4469684682865ef8a8d9ddc56e : FAILURE in 1h 47m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c41aa7dfcb045dda68f64d2245d4e26 : SUCCESS in 2h 08m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ed4cf0266ef0458fbc5e2f34d5e586d4 : FAILURE in 15m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1db94cafd008450d9b867f535082fbfa : SUCCESS in 12m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/073b860368484732ab58f91406656759 : FAILURE in 11m 25s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/af0982a4d7ad4eacb918a66b111d48be : SUCCESS in 15m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a3b5d94ffad4e39ac09a8a1b6fb74c9 : SUCCESS in 11m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f082d53b38c5492585a203ad9e021c09 : SUCCESS in 8m 23s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/de67a8eac8ea4b24965afe504a75442e : FAILURE in 13m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/95a17a5577e449039c92e01198cce14b : SUCCESS in 9m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3e793a780800461b9868391b5f3d0475 : SUCCESS in 15m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d32a46a86e34102b2faee986a0ae608 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b9503a264d5c817ba2f2dc410dee681c637785fd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-05-19 21:29:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"23bacb97f73369c00ac63802be664fde72b4d05e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-20 07:31:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b1f803e273e338dc2624b90da55bff22accdae6a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-20 08:27:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"413c000156ab6833bb2722ed1e1a91c6dd1fe16f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 10:01:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f05e38e6f08740baa121277fc34063d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7ad727320f34cd4844eada825173fd9 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e06bd6481e47e4ba93de0ca8f1a2af : SUCCESS in 2m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b5b13b0874c4f0f90d22cf170b706a5 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05066524cd9d4f5caea681a5756c97d7 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/001fa744694c4eb79d3880ae6ce3c857 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82633af751d14c3e9fe633872aa77b4d : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f2f4631499554c5c9a0051a8b0e0debd : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1dd5e395707444da8fc21ee7fef8f332 : SUCCESS in 6m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e1e3ae29131d41b7b5634705ac42ae2e : SUCCESS in 19m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7d3bdd9c0240491a88fb6c9c6aec8054 : SUCCESS in 8m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7190dd95197045bd909116d1054024f4 : SUCCESS in 11m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b420f7ecfecd48a68d876c570f927f82 : SUCCESS in 11m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/317b149d41fd4b5eb0e1d83f5ea7d79c : SUCCESS in 19m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6a89f457c2b4aab89a9d8e5ad59f97c : SUCCESS in 7m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/84c0020c574547a393460e5d4fe6e88c : SUCCESS in 33m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/16b7758dff714d9c934a2b012d61b827 : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f9a9c55b03124af8ad6e5983f8b4b7ab : SUCCESS in 1h 02m 43s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/d098751605e04c46aa6fe20e7822e24b : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/14a1d756880e4ef787c253d47790f26e : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b4f7b71ca07c48d98ab9078ccf3f0496 : SUCCESS in 1h 07m 18s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56d98ee39894470ba2ce5647c36e3883 : SUCCESS in 27m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5fbc7890188e43efaebac156a097a8a0 : SUCCESS in 27m 09s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2f48b872b757454d9d09202ee5f7653c : SUCCESS in 2h 24m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76b41ad0472945b18aea00ed6fbd190c : SUCCESS in 2h 05m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/595ef20e68a54493bdd29e41090ddb33 : SUCCESS in 9m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/efa86b0d33f143aa9c0cd8fbceccc52b : SUCCESS in 9m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fd117b57d5034911af3ba39f2f161fd1 : SUCCESS in 9m 44s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/b20cc6acee394874829975d17da11a92 : SUCCESS in 17m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23b15b2abf624096a396836ddbe0c3c4 : SUCCESS in 7m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bcb6677994be4dc5811aba7fa8f3e50c : SUCCESS in 8m 21s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8a33f79464614435aaddaf7e388b9cdb : SUCCESS in 21m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d3491d0e30344802b50384c88b33fc7e : SUCCESS in 9m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4c75fd21dc934940a00a2b5485583dcf : SUCCESS in 8m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a220d0fd6c5643e6ade16af6e16dd255 : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d47ab9e68a3f75eb576fcad48db352af78b43b39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 10:02:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7d729d3ad72b080afb0b1f584d66b4e70fc801b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 12:31:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a56a0b896d9410ca29498b6eeb47db1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c0348d4ec77464c87f5c8daeb86a077 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19db6b50a3674dac919e9ab769a21fa5 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29da23ff51ee427ba65f8b5936708d32 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/683dfbf96abb4f7cafa50f02de2e947e : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/daafd4d0905a43f0a86c4cb494f48328 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1486527b8ee442eeab4a1443f7817c71 : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dc70a35ea24d41ebaf37a5f490bd2ba4 : SUCCESS in 12m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fdda0174a18d4ec38f3b3504e5d2890f : SUCCESS in 13m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/852c6522e2684e4e8b7f93e4fdf49d29 : SUCCESS in 17m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaa003fd839f4116bf704aaeb49229cd : SUCCESS in 6m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bb2273589d914a14b9dbb8537d1bf652 : SUCCESS in 22m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/037941c1fb6544f5bcac867f2920643a : SUCCESS in 11m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b005578ed0ed4b13ab1bf57ae4ad0b66 : SUCCESS in 11m 50s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d4f32c4b65574fa99af5575b9cb76636 : SUCCESS in 34m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/dd4a6e673b5d4c42a794ee33d17408d1 : SUCCESS in 30m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4517b134ce22403a87599f79a6dcbf6e : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/745d9fcb7438459a860c322ef47f4f1d : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e903a47b8be84f76ac516be4fd7e0b38 : SUCCESS in 33m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b2ebffbb2ac44d929b595995baac4efa : SUCCESS in 1h 18m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff2269ef9ed04eeb88ba8c2f8c8e3eb4 : SUCCESS in 1h 01m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0c926ca23cf947ac905b03ff5d8ca277 : SUCCESS in 52m 04s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/651b5fa344ec434abf54364bcadab701 : SUCCESS in 2h 24m 25s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a8e913f7ec84918b22f5add3bca608f : SUCCESS in 1h 10m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e09738e1573f4941a2f11e23596c6056 : SUCCESS in 9m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0320656e7c164c628f4d9108036ef25f : SUCCESS in 10m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b19e51089fb047a38c39a6130731809d : SUCCESS in 18m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/83fedfb24df0417a81ebc4f77c508de7 : SUCCESS in 18m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a374c2ea347f46569c1ab4e3e8bf56ba : SUCCESS in 8m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/28c9730370de4045a0f9109d55416f66 : SUCCESS in 18m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7ca88fc170da49d1b2ccbc854f982674 : SUCCESS in 36m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7ee68904fd9b445ea3d31745622c4b73 : SUCCESS in 18m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/68982be312f04e70b707aeb36a8706d1 : SUCCESS in 8m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46d98f3f4e6f4d52a738b6b55565bdf2 : SUCCESS in 12m 45s","accounts_in_message":[],"_revision_number":2},{"id":"d3055b174cd6cac3057fc639c33649800b0eb602","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 12:31:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9afffa57e86963ad9b7d707ed76ef173535b91a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 12:33:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b3f1c29329244c28045fb90d48c0c6c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db8564cc84b444abde13a51bf29ec98 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"95ae73986e3fd6549a8d8c75685015633c23eeae","revisions":{"2f14b3a163cd8bb8eccb5bce69ddf157b62b397d":{"kind":"REWORK","_number":1,"created":"2026-05-19 07:32:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/67/989067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/989067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/989067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/989067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/989067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/989067/1"}}},"commit":{"parents":[{"commit":"c32ebb8a40498dbd95dafe54ca4766c9c7af469f","subject":"Quote baremetal node create arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c32ebb8a40498dbd95dafe54ca4766c9c7af469f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-14 11:01:39.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-19 07:32:44.000000000","tz":0},"subject":"[CI] Switch to baseurl for all repos","message":"[CI] Switch to baseurl for all repos\n\nWe still seem to be seeing failures with the following signature:\n\n    Failed to download packages: No URLs in mirrorlist\n\nThis extends the previous change[1] to all repos.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/983990\n\nChange-Id: Ie1fc5c55137e985a21aeb97507c7efe9ec9b5336\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit 8d6fcbfcc0e4b369b0f1f955404bbd2e39663274)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f14b3a163cd8bb8eccb5bce69ddf157b62b397d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f14b3a163cd8bb8eccb5bce69ddf157b62b397d"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"c32ebb8a40498dbd95dafe54ca4766c9c7af469f","theirs":"8d6fcbfcc0e4b369b0f1f955404bbd2e39663274","contains_conflicts":false}},"95ae73986e3fd6549a8d8c75685015633c23eeae":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-20 07:31:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/67/989067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/989067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/989067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/989067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/989067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/989067/2"}}},"commit":{"parents":[{"commit":"8feef04d72098305b528a46271f34c051745a813","subject":"CI: Exclude docker-ce 29.5.1 on CentOS/Rocky Linux","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8feef04d72098305b528a46271f34c051745a813"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-14 11:01:39.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-20 07:31:15.000000000","tz":0},"subject":"[CI] Switch to baseurl for all repos","message":"[CI] Switch to baseurl for all repos\n\nWe still seem to be seeing failures with the following signature:\n\n    Failed to download packages: No URLs in mirrorlist\n\nThis extends the previous change[1] to all repos.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/983990\n\nChange-Id: Ie1fc5c55137e985a21aeb97507c7efe9ec9b5336\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n(cherry picked from commit 8d6fcbfcc0e4b369b0f1f955404bbd2e39663274)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95ae73986e3fd6549a8d8c75685015633c23eeae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95ae73986e3fd6549a8d8c75685015633c23eeae"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase","conflicts":{"ours":"2f14b3a163cd8bb8eccb5bce69ddf157b62b397d","theirs":"8feef04d72098305b528a46271f34c051745a813","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
