)]}'
{"id":"openstack%2Fkayobe~989101","triplet_id":"openstack%2Fkayobe~master~I2f1d2fc4e8c116798ae82d9111a7101f99595602","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-19 13:27:23.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I2f1d2fc4e8c116798ae82d9111a7101f99595602","subject":"Use specific version of CentOS Stream with published checksum","status":"ABANDONED","created":"2026-05-19 11:26:45.000000000","updated":"2026-07-30 09:19:59.000000000","total_comment_count":2,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"458dca2d54f9363ddd7d04bf286e45b8c8466eba","_number":989101,"virtual_id_number":989101,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-19 17:40:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-05-19 12:10:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-19 13:04:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83dc16ffe1e8b0bc1a33db72988226b82ffa3d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-19 11:26:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9f2f0de121ca7ab30a70fa152a3527513044a88","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-19 11:30:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74c0df872050a4ff3b75d1c68b5805eb9cde81ee","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-19 11:35:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0540b706ab0a1b3aa80a551ace7967f62b493480","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 12:10:49.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe108fd75ef34360b96e74e7c177d930\n\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/6932471ee65b4834bff83ed8dda14abd : SUCCESS in 10m 57s (non-voting)\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/cef32e7b019a415191dc82ca6cc04b0b : SUCCESS in 15m 09s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/2a22dffc8b734726b8bc8ccc72c8f817 : FAILURE in 23m 26s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/b7d4de975da54feabf333e5440fad63f : SUCCESS in 16m 20s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/48cc6cb33f4644869ab54d42038eba19 : SUCCESS in 23m 32s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/9a53d10d8fed48a8b9b9c296c91821fa : FAILURE in 28m 33s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/d17b277e6838412284f3fb5dd6eb4288 : SUCCESS in 13m 15s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/8b8d121b60d44371a3d3534081536785 : FAILURE in 28m 56s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/e581dc69da1e4961b0312efafd875579 : FAILURE in 10m 46s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/de4ca1c5faf844f594ea8c1e8b6ecfdc : FAILURE in 15m 52s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/89aa2495816644bbadc9506ac847b4a6 : FAILURE in 22m 33s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/954782aaf5d74096827047c9bcbd6eda : FAILURE in 14m 43s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8a6c74490d94ef79d441884f5b4a116 : FAILURE in 23m 15s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/03c6d65543f4460bb34407cbbfc7d01b : FAILURE in 13m 04s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/97118ec2a0674783a4c591ee4f44ecf6 : SUCCESS in 13m 36s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/1eb38f8db48e4ed99d82253ffe933c4f : FAILURE in 23m 31s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/2c4e1052fb6a4ce2b6995c5781676e5d : SUCCESS in 15m 58s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/a5432e567326402ca9b019e22367f70a : SUCCESS in 33m 19s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/6546432637cd4f03bf11f1f32a735e9b : SUCCESS in 19m 11s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/e888e8f7c89e407aa7d15cada5c29a7c : SUCCESS in 28m 17s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/626b3d6545e94562a8b2344491cfd21a : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":1},{"id":"05df32674b65b9103ed20330125eb652217496e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 13:04:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c03d8b7d07f7450cb0bff38e19c7cf28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4adf6c0244d44ad0ae0057db052bf554 : SUCCESS in 10m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cfeb4785e1de445581c0d17f5475b934 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1daf277ba5a74ec59da624063f2a5056 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b724d195f44a2ab6c6d5f0deddb99c : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e16726593b9b4cd9bc0dd63f9a63c1b2 : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b644e2f2db364c00b2bd61a96a3325c2 : SUCCESS in 4m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2def937a60e642b1b332e765e98cbb27 : SUCCESS in 6m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79cbf978477a4dd3a8f2facd414db57c : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/99f890dfbc2b4b7d8208333cb97702c9 : SUCCESS in 7m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/577ed5dc1f3f418494b138f8b509af79 : SUCCESS in 7m 04s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/15f7f80a2b364816bfc444e633043849 : SUCCESS in 13m 32s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b89f9050bc1c4f9f828717172273cbc4 : SUCCESS in 14m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eb47781cca534c88bd1642c4cf16dcff : SUCCESS in 14m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b86d9b9969349f89baaa5de6f66e122 : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/cf9d11f1bfe142819b73077b0322671d : SUCCESS in 13m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4146e4ae8a324d9dbae6612926dd0c64 : SUCCESS in 19m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/256bda33a4044e11b88245b7bf51f8b8 : FAILURE in 24m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4b0bf03820c24f16be4ca18d9bb93bf1 : SUCCESS in 1h 05m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0665a8148f1d4e96aa7c708d94ce84ad : FAILURE in 33m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69f48ec1f28e4505b844f94f5be30e19 : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eee4f6465e5047c1a79de36bd9e3f21d : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e6dc6a3c83ae4472a5bc08ad209ebb98 : FAILURE in 53m 02s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/201116ecb69b45e292887b2175d06b29 : FAILURE in 52m 23s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10cdd10738ea426cae26ee91e078bdfc : SUCCESS in 1h 29m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5505595724a14372923009cd537b7786 : SUCCESS in 1h 34m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/511676492fcd4ae98859fa07179b459e : FAILURE in 26m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/811479df519c4e3b9c6014937b2298ce : SUCCESS in 23m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1552e4e9619c46498a032e5c61a93654 : SUCCESS in 26m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b401bd6c3c7e4934a099f68d03fbbc39 : SUCCESS in 22m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b6b805d8f29944579472a154ab1a84aa : FAILURE in 20m 15s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/94946738dc3647f08c15ddc68cc09dd8 : FAILURE in 12m 05s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f48f66db2a24c0b8a9a4cb08f17ce8c : SUCCESS in 37m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45432d4fbcbc44979c8b03ee4a7dd070 : SUCCESS in 33m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/90417cbc8d6048268a184c3878654ce8 : SUCCESS in 17m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f50c63a5851546f79efaefd7b7f94976 : SUCCESS in 20m 57s","accounts_in_message":[],"_revision_number":1},{"id":"54717348614958ce752762e57179a1ef4f42f111","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-19 13:27:23.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"8124ed7a729c7bf5d2d818e6384bd6b5a96604d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-19 15:57:28.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a6b42785c1fd40c84563a0e2732cf8f3344b0653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 17:40:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b5af414874440cebef0bcaf3a5f9091\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/319f038381df424e87447217fb7dd1c1 : SUCCESS in 8m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb360fb6ebb5422e826c1eee96cdf7ab : SUCCESS in 7m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6249006483174871affc163ddb349b67 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a24b9ba2dc5472e8858f2e14a81affe : SUCCESS in 6m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43e8ca9842f342c2b612fbebbf104ef7 : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d0cfe3880969487abe78b70ea7ce3854 : SUCCESS in 7m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/690906cdff2f43b79929ce0757aad080 : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dd5b398a279d490da353c27297c0eb6a : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb400e1993084109848b17ffff752685 : SUCCESS in 7m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e9358f194e9c4ed6a5d613cdcfaac3df : SUCCESS in 7m 36s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9d7fe796d580484794299064e1d59127 : SUCCESS in 13m 31s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d1accab9f134f418e2a62a5f0714a37 : SUCCESS in 13m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e1d52e7b5b284a8bbeafa2206ab60e5b : SUCCESS in 16m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1f72126c5304735a77b4aaef9d02980 : SUCCESS in 23m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4cccea33a852455d9b4d7c4e300bef8c : SUCCESS in 12m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/299e5fb1d4e14a8bb134678c93885316 : SUCCESS in 10m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5ad0cb2c9da34d26a7d9d7861a8756b8 : FAILURE in 28m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b74cc7af2582405188db5606fb34337f : SUCCESS in 1h 02m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/88363a4497fd45cc9a6e570742187d85 : FAILURE in 31m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fddc7c1fce345729683c17d321404d8 : SUCCESS in 46m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a55d5c64a1754e6590c404f21486d308 : SUCCESS in 1h 05m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4fc1588f6beb4dc6b2124c1306b81527 : FAILURE in 56m 07s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3001b569714d4fb4826f85847eab294d : FAILURE in 37m 14s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9ab494d7b9946079e0e19a0d2fe03b2 : SUCCESS in 1h 30m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83998f66323e4905b6bac5c2fd1f56d2 : SUCCESS in 1h 35m 51s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1e1520ff52724c6d8e04a4211c82bb20 : FAILURE in 17m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2fc3e7c9e3184cab9ca46276d90dfd4e : SUCCESS in 15m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87c4a240b423455eb9934d05521240d1 : SUCCESS in 24m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3ebc19da8f6946aca5011efb8dc944fd : SUCCESS in 14m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3edf75b7b8dc4435a93e4cf704989baf : FAILURE in 18m 48s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ddd1dd75cf8842dc8d4b42021a20d83f : FAILURE in 16m 43s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aeb603f6b5214e2d97f747d8b74629a7 : SUCCESS in 31m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbde7090968a4a53bef7ac3e3aa4a1ab : SUCCESS in 28m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/08f8a617b93d4fbc8300af881e9da6a5 : SUCCESS in 17m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd851f0b5236443d97c24d178e112209 : SUCCESS in 21m 35s","accounts_in_message":[],"_revision_number":2},{"id":"458dca2d54f9363ddd7d04bf286e45b8c8466eba","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-07-30 09:19:59.000000000","message":"Abandoned\n\nChecksum now uploaded","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c85ff7f4e9b7b1dc1f7cf8b114820bc2f7fe220a","revisions":{"87e75aade0b609146c096889ff2bc4c8b11e7cf1":{"kind":"REWORK","_number":1,"created":"2026-05-19 11:26:45.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/01/989101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/989101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/989101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/989101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/989101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/989101/1"}}},"commit":{"parents":[{"commit":"cfb8e893851d3bed36565aa2f6b34a4bfa3c5b2e","subject":"Filter falsey values from IPA build env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfb8e893851d3bed36565aa2f6b34a4bfa3c5b2e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-19 11:20:57.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-19 11:26:36.000000000","tz":60},"subject":"Use specific version of CentOS Stream with published checksum","message":"Use specific version of CentOS Stream with published checksum\n\nWe can revert this once the checksum[1] for the latest image is\nreuploaded. Currently it is a zero byte file.\n\n[1] https://cloud.centos.org/centos/10-stream/x86_64/images/CentOS-Stream-GenericCloud-x86_64-10-latest.x86_64.qcow2.SHA256SUM\n\nChange-Id: I2f1d2fc4e8c116798ae82d9111a7101f99595602\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87e75aade0b609146c096889ff2bc4c8b11e7cf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87e75aade0b609146c096889ff2bc4c8b11e7cf1"}]},"branch":"refs/heads/master"},"c85ff7f4e9b7b1dc1f7cf8b114820bc2f7fe220a":{"kind":"REWORK","_number":2,"created":"2026-05-19 15:57:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/01/989101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/989101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/989101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/989101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/989101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/989101/2"}}},"commit":{"parents":[{"commit":"cfb8e893851d3bed36565aa2f6b34a4bfa3c5b2e","subject":"Filter falsey values from IPA build env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfb8e893851d3bed36565aa2f6b34a4bfa3c5b2e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-19 11:20:57.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-05-19 15:57:28.000000000","tz":0},"subject":"Use specific version of CentOS Stream with published checksum","message":"Use specific version of CentOS Stream with published checksum\n\nWe can revert this once the checksum[1] for the latest image is\nreuploaded. Currently it is a zero byte file.\n\n[1] https://cloud.centos.org/centos/10-stream/x86_64/images/CentOS-Stream-GenericCloud-x86_64-10-latest.x86_64.qcow2.SHA256SUM\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/989169\nChange-Id: I2f1d2fc4e8c116798ae82d9111a7101f99595602\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c85ff7f4e9b7b1dc1f7cf8b114820bc2f7fe220a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c85ff7f4e9b7b1dc1f7cf8b114820bc2f7fe220a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
